)]}'
{"id":"openstack%2Fkayobe~989359","triplet_id":"openstack%2Fkayobe~master~I915561e73108fa0c4c1e977a3b88271f69ff6002","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2026-05-21 16:22:27.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I915561e73108fa0c4c1e977a3b88271f69ff6002","subject":"CI: Replace use of deprecated cp -n option","status":"MERGED","created":"2026-05-20 11:00:07.000000000","updated":"2026-05-21 16:22:27.000000000","submitted":"2026-05-21 16:22:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"989359","meta_rev_id":"8ba179afdd016bd0d15ed682a2d251cdb3c96d87","_number":989359,"virtual_id_number":989359,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-21 16:22:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2026-05-21 14:33:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2026-05-20 18:12:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2026-05-21 14:33:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-20 12:41:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-20 18:12:33.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2026-05-21 14:33:17.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"}],"messages":[{"id":"4d33c1930da508f7d23315dcd45c35334254838a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-05-20 11:00:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f44d12c0e30137aadf96b64c828e62f579c33707","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-20 12:41:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b910e84069564671a1a25953f919af03\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8082f945e7084dbb8144c238e344e775 : SUCCESS in 9m 25s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/7906ac023d6a415f817766660301cff5 : SUCCESS in 5m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f981059df0074c0f92676fb5c8e1cfb8 : SUCCESS in 2m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/36b3af4690044eabbe5c21cf1b452650 : SUCCESS in 4m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/458c1d05fa6340709b3643833c5711b1 : SUCCESS in 5m 52s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/89986c07e1ee4d90b2de57906030f4ec : SUCCESS in 6m 48s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7735c43da1dc46b3bb9be014d6876cab : SUCCESS in 4m 49s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f3a9ab0182824d53a69cb4608535a08a : SUCCESS in 4m 40s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b068ac1586974b11bddb5470833066cb : SUCCESS in 11m 50s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c00ea7f09d9f45e6aeb9c62ba0cdd28c : SUCCESS in 11m 33s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/a73c41d78b75481086d186647e81bddd : SUCCESS in 13m 30s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8b76e1e73edc49bead1e7d39125306ee : SUCCESS in 9m 23s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/33ae7cebba144d63b57b0a1a313899e4 : SUCCESS in 14m 13s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9a8c62c7e6a64f5eb6cf04d61f66f31e : SUCCESS in 11m 45s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/bda6b89cb7fc49adbb1a9ca19111bfb8 : SUCCESS in 11m 47s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cd1674b68119497e81a93ea207bafc8f : SUCCESS in 20m 22s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/ace29bd5d622485eb2e889c2b3bcfab3 : SUCCESS in 1h 02m 38s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/84fcdd5eb86a4e9b94f635611a9a3719 : SUCCESS in 1h 03m 42s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/645d699425674cd49a7075bc028704d4 : SUCCESS in 1h 06m 16s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d1724082bba143588c59e963f486df87 : SUCCESS in 1h 03m 59s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/183e09e4aa914bd9b7e2023b95c5f23b : SUCCESS in 1h 03m 57s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/53e2c3c3920b4b2183bf31f8ae981405 : SUCCESS in 1h 40m 32s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/c7266cf7c77742fa963148b1a51b5d37 : SUCCESS in 1h 28m 33s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/46c49e54f07d4939ad6b6de9eb7c48fd : SUCCESS in 1h 29m 52s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7cc3d4b4ef0a4c93a6853d4928e97d34 : SUCCESS in 1h 31m 09s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/076567dc79474fcd88d3fdc410387651 : SUCCESS in 21m 22s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/c0d733966a7343909538dcf81210f68f : SUCCESS in 15m 20s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e29ba30b62a84e419d0cf0995d5ddca2 : SUCCESS in 24m 23s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/c6f27857740d46bbb89f7b04bdcaab6f : SUCCESS in 13m 49s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/1a2f0e449e83453fb9a92196ea939a98 : SUCCESS in 15m 29s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/8369cb0308df417ab0664a2a6d63fe47 : SUCCESS in 26m 39s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6f3efe6a0952421b9dd8995bff62bcd7 : SUCCESS in 29m 51s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0766f55486a24a339bbe16776e04b79d : SUCCESS in 32m 47s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/e71986be85264c14a5ed667e22fd4c50 : SUCCESS in 18m 56s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3fed4010251248e4bfa063940c7605a9 : SUCCESS in 19m 52s","accounts_in_message":[],"_revision_number":1},{"id":"0dc5772103862ec10231f9454ff00f19502327dc","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-05-20 18:12:33.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d9071043e024e37b0856ca2abfd28db82669021f","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-05-21 14:33:17.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"dabdea6edd401a387ce1694ffe25c2fb4792641f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-21 14:35:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3ec6dcf84b0f9e39db112ec9562de59d3e0be8c4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-21 16:22:27.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a8742a51786b4903acdd866d099b1ea4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da2223ca60974032aca0a43681fd9d36 : SUCCESS in 5m 23s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/0339b342d28a47f58e5244ab454df343 : SUCCESS in 7m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5e78c1bac4e4a5f82e96aceb5e8a5cc : SUCCESS in 5m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1b218129290d488b93b77ae19651a0dd : SUCCESS in 5m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3b3c36d945574712a99a7fff067c2611 : SUCCESS in 6m 00s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6723fa9687594a298f3dce0bbada2bb9 : SUCCESS in 3m 58s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6fae95596a9f407a9746c3532895dea9 : SUCCESS in 12m 16s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e20b9887dc9543efb3a55bd79b02e0aa : SUCCESS in 14m 25s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/db1a04cbf0a44e95a7a75537dbd82cc4 : SUCCESS in 15m 26s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/441b6b95712f4d949feb989bae57f709 : SUCCESS in 10m 14s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/9f9e140d142142ce8313cb28da3b7ff9 : SUCCESS in 16m 09s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3cc8d781b15940af8706a443a12498df : SUCCESS in 13m 13s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/c4392fe59d45410f997fdc499916bd1f : SUCCESS in 16m 10s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/aeb4e73cb3b54665b35b8c1d4d4af940 : SUCCESS in 21m 34s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/9ba6975870ff45bfa7b1dc8dec746bd1 : SUCCESS in 1h 02m 34s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/73eb933bd64444febdf332691dbf0de2 : SUCCESS in 1h 04m 58s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/e5ebed0298f34ceabe11b1ed24008d47 : SUCCESS in 1h 08m 09s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/870ad1fb22dd40deb2dc38885089b742 : SUCCESS in 1h 04m 09s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/69aa7922662644d18a6ec7de198fd3c5 : SUCCESS in 1h 05m 55s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/c0fc0c0dd7f941128869be1fc14aa8ab : SUCCESS in 1h 38m 19s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/d47e1883b24e482c95b677509a9f18d7 : SUCCESS in 1h 32m 14s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4d2b9f632ccd490a8aa6db8c864ad1bd : SUCCESS in 1h 28m 50s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f730326adf564e44a631713dbbc3d704 : SUCCESS in 1h 32m 52s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/c388bef78f6b4a61b4357da87d286d05 : SUCCESS in 28m 19s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/306d0dd22cbe48d4a2468b85448be7e8 : SUCCESS in 21m 31s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a3bdd27d0dab485fb89a3d64a3433dfe : SUCCESS in 13m 17s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/d24ec1aa19bb48ec9097d5b30e9ef11c : SUCCESS in 18m 28s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/ff2ed1a3c369480cbb48423928165ed8 : SUCCESS in 17m 07s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/b4148e9148f04e3cb11165fc36ef8ea6 : SUCCESS in 25m 33s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/02f91529c8bb49ceaa8085ba1a130aab : SUCCESS in 16m 15s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d8d175c671c441efb26c5da9ba86b2a9 : SUCCESS in 30m 27s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/bd79c915578b4c06ab68b2595ef34e9f : SUCCESS in 16m 53s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/003519aede0e4a809e9abdbd4389aebb : SUCCESS in 19m 34s","accounts_in_message":[],"_revision_number":1},{"id":"8ba179afdd016bd0d15ed682a2d251cdb3c96d87","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-21 16:22:27.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"eb5ab490e88c907e032f22c3c5bf953c3d368327","revisions":{"eb5ab490e88c907e032f22c3c5bf953c3d368327":{"kind":"REWORK","_number":1,"created":"2026-05-20 11:00:07.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/59/989359/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/59/989359/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/989359/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/989359/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/989359/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/59/989359/1"}}},"commit":{"parents":[{"commit":"0069fb578ae4f1908ce12b28b6d575fee10c0c9a","subject":"Merge \"Bump Ansible collections and roles\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0069fb578ae4f1908ce12b28b6d575fee10c0c9a"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-05-20 10:59:51.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-05-20 10:59:51.000000000","tz":120},"subject":"CI: Replace use of deprecated cp -n option","message":"CI: Replace use of deprecated cp -n option\n\nThe following warning was visible in CI logs:\n\n    cp: warning: behavior of -n is non-portable and may change in\n    future; use --update\u003dnone instead\n\nChange-Id: I915561e73108fa0c4c1e977a3b88271f69ff6002\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/eb5ab490e88c907e032f22c3c5bf953c3d368327"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/eb5ab490e88c907e032f22c3c5bf953c3d368327"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
