)]}'
{"id":"openstack%2Fkayobe~989606","triplet_id":"openstack%2Fkayobe~stable%2F2025.2~Ib1ce90cd9174a190248e11512308828a2719ba49","project":"openstack/kayobe","branch":"stable/2025.2","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2026-05-22 12:58:28.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib1ce90cd9174a190248e11512308828a2719ba49","subject":"Revert \"CI: Exclude docker-ce 29.5.1 on CentOS/Rocky Linux\"","status":"MERGED","created":"2026-05-21 19:21:50.000000000","updated":"2026-05-22 12:59:39.000000000","submitted":"2026-05-22 12:58:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"989606","cherry_pick_of_change":989487,"cherry_pick_of_patch_set":1,"meta_rev_id":"bc2cd654a1bd170c7a084731781e2d23d007bf27","_number":989606,"virtual_id_number":989606,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-22 12:58:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2026-05-22 11:07:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2026-05-21 22:12:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2026-05-22 11:07:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-21 20:46:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-21 22:12:40.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2026-05-22 11:07:40.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"}],"messages":[{"id":"e490c3d6b9970e907c7d0f4971b260554d181961","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-05-21 19:21:50.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"dd66586761a43d10dec3900db32b265d23e988ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-21 20:46:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b2d419a60b6b4141874992746ddd4cca\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b544188d1e044c4fa705b719036db4a6 : SUCCESS in 9m 20s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/1af847b7bb724f1cacf705e6d1e28887 : SUCCESS in 8m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c3b6fe7f5ce44c169389a4840c44e485 : SUCCESS in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/deb19543be9f4f6ba59d78bf8f5efa6b : SUCCESS in 4m 24s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/861616843446409b93d3d1afad1dfad7 : SUCCESS in 4m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/04523f5eaa6b453e969ec33fb89a21a3 : SUCCESS in 6m 25s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4744f6ee19314a8b97e0cf530dccd6cb : SUCCESS in 5m 47s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/defa837d68ec482e8ce59b2e6caa8201 : SUCCESS in 4m 57s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/195d477fe970423a89d5abf045cf1706 : SUCCESS in 11m 20s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/067c89fecc264346ab0eb9a7ca5bff06 : SUCCESS in 10m 36s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/e5a59d762f5d45ac9c4e54811584b46f : SUCCESS in 9m 03s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/81fb23674df543f088b62cea954944b1 : SUCCESS in 15m 28s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/dc18c405d88f44ce87e94ebcaf0115b9 : SUCCESS in 11m 27s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0c578a8b18164164ba299baa35211f42 : SUCCESS in 15m 59s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/0287be9554344187bc9d82aafe36d630 : SUCCESS in 57m 09s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/014b3508584a4678a1b73fd5195280c4 : SUCCESS in 55m 10s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/6f8ad84111b24641bc92f51504867818 : SUCCESS in 57m 48s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/07b037982b4149b5bb3d9025626f66a3 : SUCCESS in 56m 14s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/dba6aa6aa9504d96bb987d6590a4c085 : SUCCESS in 1h 00m 58s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/2bb6dc93236144dca9a93cb6ec30039d : SUCCESS in 1h 23m 11s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/27525a6340c644c89a7558f2331d7b91 : SUCCESS in 1h 22m 19s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/a49e3f7fee6546deb88f8f22461e0b14 : SUCCESS in 22m 11s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/27690475a1984e628589103de68040b2 : SUCCESS in 17m 49s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a95a74fb1cbd4e4dac82566f07ffbd23 : SUCCESS in 16m 55s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/e5895728f5de49e393b22afba714e714 : SUCCESS in 15m 40s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/18b65dc95e6e49218d2ae41776fbd48f : SUCCESS in 36m 47s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d6444fcfa64743d289ca667147c5849e : SUCCESS in 26m 51s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/6ea3639ccf9744aa8e306ffb686f019d : SUCCESS in 9m 20s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2b47fa97ab0e40ae9f4dc97b5c2c921c : SUCCESS in 16m 54s","accounts_in_message":[],"_revision_number":1},{"id":"5aa09cb101fd734d892872d331f3b344a91e3087","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-05-21 22:12:40.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"5054ba61ec6115168b5bf3d761e81bc6adc31e27","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-05-22 11:07:40.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"79ed287c57d6e7ab9924b48b1fa8fcc866c32f3e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-22 11:08:02.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0b5cb54b7fb03ecfa1450315154c334d6d532b3d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-22 12:58:28.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c9a7bdb0b592491088da7ae7576fa71e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b013b6ae6d747c3acb6ad9a9998e41d : SUCCESS in 9m 09s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/cdb9c1be55b04d1eb0612590f8e0a526 : SUCCESS in 3m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc791bc8fcfb41ca830d6ebba4b1e09c : SUCCESS in 3m 39s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/43b9040a61ab42f68a34d3a22f3bc51d : SUCCESS in 2m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0afc4a60d70d4070bd4cb0b7d15db52f : SUCCESS in 3m 46s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e1735d53070a4718ae22be8a82ff6a3c : SUCCESS in 4m 28s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1c43b7b10d834f01b16098bbab119e30 : SUCCESS in 12m 10s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a3e1187820e7432e927b23300f07d448 : SUCCESS in 7m 10s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/570376c367344c2b99ed1b797e6c1d4b : SUCCESS in 9m 58s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2921f0c2877a416a916886fa6b1b8c2e : SUCCESS in 6m 37s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/3b2591f2d86842eb90936971e567b98a : SUCCESS in 11m 28s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ffd7504bb38f4cc18208e2152684b060 : SUCCESS in 9m 57s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/d0a383ddc9534b878574a6a907e7e599 : SUCCESS in 54m 03s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/19286489757b44faaf5e6dcd8c2b7cfe : SUCCESS in 54m 28s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/a6b4c3463d9946678de3938932d210fd : SUCCESS in 55m 53s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d2e4a5b748a64f2eb5583e1d2d1751e8 : SUCCESS in 54m 10s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/b380790f3a0145b2964791c8d0226bc7 : SUCCESS in 52m 38s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/a5991366020d461d886d83f07a3bfb17 : SUCCESS in 1h 20m 57s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c4a43573f00a459ea5533352a43a291b : SUCCESS in 1h 19m 27s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/76e7c7968cc04b06ae7be51cadcbb0bd : SUCCESS in 9m 18s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/617976a1fa664317a56132e5cc47e87c : SUCCESS in 9m 57s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9fbc65a6507645abb73139dc8ae775de : SUCCESS in 13m 06s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/2aa787477abe4c2abea3dad5775a2f0e : SUCCESS in 15m 36s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/a829f941c35b4a7db529b4ce3566bff3 : SUCCESS in 28m 23s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/87637a4f543e44ed8c921167e4727a59 : SUCCESS in 27m 17s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/1cef3b9c18ee4d4e883fc487a80891c4 : SUCCESS in 9m 03s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c7eb253737bc41b9b22cc5136fe744cc : SUCCESS in 14m 24s","accounts_in_message":[],"_revision_number":1},{"id":"df4f1d2454c4b3b731ea133696e54c2542e9ad40","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-22 12:58:28.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"bc2cd654a1bd170c7a084731781e2d23d007bf27","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-22 12:59:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/10826afbd3364af7b883e942b32e8ddb\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8093f36e3eff4bee97fd4bbeb8cba1a5 : SUCCESS in 52s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7e1920eb96356bc2e0c91e1940deadd221d98869","revisions":{"7e1920eb96356bc2e0c91e1940deadd221d98869":{"kind":"REWORK","_number":1,"created":"2026-05-21 19:21:50.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/06/989606/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/06/989606/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/989606/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/989606/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/989606/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/06/989606/1"}}},"commit":{"parents":[{"commit":"f5c825ed3ce9ad95377edf6461d8012d0f94cd47","subject":"[CI] Switch to baseurl for all repos","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f5c825ed3ce9ad95377edf6461d8012d0f94cd47"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-05-21 07:57:10.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-05-21 19:21:50.000000000","tz":0},"subject":"Revert \"CI: Exclude docker-ce 29.5.1 on CentOS/Rocky Linux\"","message":"Revert \"CI: Exclude docker-ce 29.5.1 on CentOS/Rocky Linux\"\n\nDocker Engine 29.5.2 is out and fixes the regression introduced in 29.5.1 [1].\n\nThis reverts commit 43294efe8ce9677d56929ae4e79e2c66c25f97f2.\n\n[1] https://docs.docker.com/engine/release-notes/29/#2952\n\nCloses-Bug: #2153110\nChange-Id: Ib1ce90cd9174a190248e11512308828a2719ba49\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n(cherry picked from commit 908d87145590ccf3b4ec070cb9767a964a232955)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7e1920eb96356bc2e0c91e1940deadd221d98869"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7e1920eb96356bc2e0c91e1940deadd221d98869"}]},"branch":"refs/heads/stable/2025.2","conflicts":{"ours":"f5c825ed3ce9ad95377edf6461d8012d0f94cd47","theirs":"908d87145590ccf3b4ec070cb9767a964a232955","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
