)]}'
{"id":"openstack%2Fkayobe~989641","triplet_id":"openstack%2Fkayobe~stable%2F2025.2~Ieddb9b48f5e10de26dced947f9e1d7d4276e4448","project":"openstack/kayobe","branch":"stable/2025.2","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2026-05-22 11:54:30.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ieddb9b48f5e10de26dced947f9e1d7d4276e4448","subject":"Fix network connectivity check on modern ansible","status":"MERGED","created":"2026-05-22 08:38:11.000000000","updated":"2026-05-22 11:55:27.000000000","submitted":"2026-05-22 11:54:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"989641","cherry_pick_of_change":986126,"cherry_pick_of_patch_set":2,"meta_rev_id":"5eccd37cfdfc14773cc4301f9873719ba951c71e","_number":989641,"virtual_id_number":989641,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-22 11:54:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2026-05-22 10:26:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2026-05-22 11:04:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-05-22 10:13:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2026-05-22 10:26:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"reviewers":{"REVIEWER":[{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-22 08:38:11.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"CC"},{"updated":"2026-05-22 10:06:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-22 10:18:35.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2026-05-22 10:18:35.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"state":"REVIEWER"},{"updated":"2026-05-22 10:18:35.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2026-05-22 10:18:35.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2026-05-22 10:18:35.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2026-05-22 10:18:35.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2026-05-22 10:18:35.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2026-05-22 11:04:09.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"32c5b4f4e0e4aeb0aedad7cd6bccf7811caf97d5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-05-22 08:38:11.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"cfc0c479c5bdd3bc23c0fb559e9adb7ca023d49a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-22 10:06:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3f36ae8c1956458a99b94dd79777a9f2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0add1f02d9e44e4e99693386747f7ff9 : SUCCESS in 4m 58s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/1e6eacd7e010464a9d56dfc9eee18d91 : SUCCESS in 7m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1017bfc738b24e238cd5c0959f3caa76 : SUCCESS in 4m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/67ffc3ae27af4c57b6ac4824ccba580e : SUCCESS in 4m 07s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e7f7ed565f5540a6896ba6497d2ab9c0 : SUCCESS in 4m 12s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/93d70aa7531941c396777a638bdab66d : SUCCESS in 6m 18s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e162f6c2f94d48a2b7be3d597fb33eef : SUCCESS in 3m 33s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/5c10b067539e45e089fea2428129f2b3 : SUCCESS in 3m 18s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c9d7cf194a7c4912976fc336ff8ba505 : SUCCESS in 6m 19s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/4a46e96e359247a88f87ceec360426f9 : SUCCESS in 10m 42s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/7f4ac38bb70b41b1a85fcd47dc3245a2 : SUCCESS in 8m 49s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/53874abb6ae34c4f964e0d50d9d13e8d : SUCCESS in 12m 36s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/8f79267fc50442febb64bf773149aaeb : SUCCESS in 17m 51s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/74172d2701cc4fbeb3612cfb757f9ee9 : SUCCESS in 16m 01s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/907e9a8d3dbc4396935d7e81d33f1262 : SUCCESS in 55m 49s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/d20f07f928f04fe2ae3a193d73152d56 : SUCCESS in 54m 17s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/e99e79f5e9f8459e929949dba9a55a95 : SUCCESS in 1h 02m 03s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/36d63a49e5c34b0096223effd3f28428 : SUCCESS in 59m 29s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/5dde15ace68e4a54ac5838ee9eef2a3e : SUCCESS in 57m 17s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/83e736023a8140a9a153ef1e8b4eeca3 : SUCCESS in 1h 24m 45s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a78ec00655704b7e89fe4390968c6f3c : SUCCESS in 1h 25m 36s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/cf650f126db049e394e20b4356215ffd : SUCCESS in 21m 17s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/81b17e76508d4ce29359907b7b53de06 : SUCCESS in 21m 13s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/be6e69d534014ae1974b29d06d5c2b38 : SUCCESS in 16m 53s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/2d8666a39ca04a8e9f87fd2f9c8787a0 : SUCCESS in 13m 42s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/fbc3d125011645479bf4b1c93f69bc5d : SUCCESS in 34m 08s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6b13e0fca2ea4732b628e3a4da0bb6cb : SUCCESS in 27m 03s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/4d0e6f5e407a4b10ba6f0a79ebdaf45e : SUCCESS in 15m 10s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/abb14cda192d4fcc9422ee6fe6a2e53f : SUCCESS in 16m 07s","accounts_in_message":[],"_revision_number":1},{"id":"07f2abf0cc9bab3e8911ae14074c19de453af6c4","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-05-22 10:13:49.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b34e7a9b8fc212ba72699a652932c63ea478f130","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-05-22 10:26:47.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"70c54d139b8c76289d0e3f6612ab94b7c03ac546","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-22 10:27:08.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"05e4ef0edff2732b68481a02ca192a05f75103f3","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-05-22 11:04:09.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"393c6a95f2c3216030bf2131af734cb25cf7280f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-22 11:54:29.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7332de42981d4bd3aebe077f48092899\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10a06608510b4b51a9b18523bf484673 : SUCCESS in 5m 08s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/f44f6e79acd549fb9ead9de84ea447a4 : SUCCESS in 6m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d4e74ea514a48fd9b16c026a795b003 : SUCCESS in 4m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/409c43495b5d48c5b0aaf3abd93a157f : SUCCESS in 5m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1f8ecfa1b5524db2a6ea451a14009da7 : SUCCESS in 5m 57s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/5ea8badc0490442e9833825d71a0a60c : SUCCESS in 5m 07s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/dc38535d6dfb47ca84a38928eb4b4612 : SUCCESS in 11m 19s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/822730f15f964225b6388b2fe9dea53b : SUCCESS in 11m 03s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/b10f0a1458fb4592a82a7c6ab292b23b : SUCCESS in 15m 39s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1f863cda1c864811bd330d040e1749cd : SUCCESS in 12m 44s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/6ddb384b3d644cac8866efb546781e34 : SUCCESS in 19m 22s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/31e17d3def584df4870e1fa901d68d11 : SUCCESS in 14m 23s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/887c98364bdd47218ca0ef210a42a89d : SUCCESS in 59m 23s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/352392639d4741409c28968e9f944b1f : SUCCESS in 59m 51s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/52f93d8103a44c7583f3ea0054031918 : SUCCESS in 1h 03m 43s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cf2c39b88b2d4fab889415a44e317d21 : SUCCESS in 58m 23s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/22715cc51ed54cb1a0e62fdfb9262e45 : SUCCESS in 55m 31s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/b178692f5df9433594802a9837eefdd3 : SUCCESS in 1h 26m 48s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/65e44d02f7f34fb1b5de1975de6ff676 : SUCCESS in 1h 20m 00s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/042d251efd054ccb890b81f9f6f5b0b3 : SUCCESS in 15m 52s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/aee190371d1c41edb9511b1449465310 : SUCCESS in 16m 26s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7987779a4ca44b02bb66c3c83d233f3d : SUCCESS in 19m 00s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/3fd8f0cc576a429887535517c79b82c3 : SUCCESS in 15m 26s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/c06d65754c6a4d81972433924725d140 : SUCCESS in 33m 21s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e90952e76cf74673acd0f721f2042640 : SUCCESS in 28m 30s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/debdaf4a21c14386a68efea360e9e637 : SUCCESS in 15m 25s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/890db8aec4e349c880344ec0c56094d9 : SUCCESS in 15m 15s","accounts_in_message":[],"_revision_number":1},{"id":"1d5878265435a8b4910096a85bd800972fc0e800","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-22 11:54:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"5eccd37cfdfc14773cc4301f9873719ba951c71e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-22 11:55:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a6d90f773d6a4cee91b722c0770eccfe\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a497a91c556a4bc5972828ea66065720 : SUCCESS in 47s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/407e27d6993348bfb35cad0541e87a43 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7f9b683494c9146ac7494431b290778728c1bb8d","revisions":{"7f9b683494c9146ac7494431b290778728c1bb8d":{"kind":"REWORK","_number":1,"created":"2026-05-22 08:38:11.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/41/989641/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/41/989641/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/41/989641/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/41/989641/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/41/989641/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/41/989641/1"}}},"commit":{"parents":[{"commit":"f5c825ed3ce9ad95377edf6461d8012d0f94cd47","subject":"[CI] Switch to baseurl for all repos","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f5c825ed3ce9ad95377edf6461d8012d0f94cd47"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-24 15:57:00.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-05-22 08:38:11.000000000","tz":0},"subject":"Fix network connectivity check on modern ansible","message":"Fix network connectivity check on modern ansible\n\nThere is no point using the cached filter here since vars is not an\nexpensive external lookup.\n\nCloses-Bug: #2138614\nChange-Id: Ieddb9b48f5e10de26dced947f9e1d7d4276e4448\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n(cherry picked from commit 47fff684bbf79e4e8dc5d04d1400029f87d94a0b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7f9b683494c9146ac7494431b290778728c1bb8d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7f9b683494c9146ac7494431b290778728c1bb8d"}]},"branch":"refs/heads/stable/2025.2","conflicts":{"ours":"f5c825ed3ce9ad95377edf6461d8012d0f94cd47","theirs":"47fff684bbf79e4e8dc5d04d1400029f87d94a0b","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
