)]}'
{"id":"openstack%2Fkayobe~989837","triplet_id":"openstack%2Fkayobe~stable%2F2025.2~Ie179ac46c28c08f3ed41922485ef2e532a2d6e46","project":"openstack/kayobe","branch":"stable/2025.2","topic":"fix-kolla-ansible-collection-installation-stable/2025.2","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2026-05-27 12:06:54.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie179ac46c28c08f3ed41922485ef2e532a2d6e46","subject":"Fix ansible-collection-kolla not getting installed","status":"MERGED","created":"2026-05-22 22:03:55.000000000","updated":"2026-05-27 12:07:47.000000000","submitted":"2026-05-27 12:06:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"989837-fix-kolla-ansible-collection-installation-stable/2025.2","cherry_pick_of_change":989760,"cherry_pick_of_patch_set":1,"meta_rev_id":"91df17ac81d5b336348b569c9071f554a3aa647a","_number":989837,"virtual_id_number":989837,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-27 12:06:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2026-05-27 09:36:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2026-05-23 09:00:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2026-05-27 09:36:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}],"CC":[{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-22 22:03:55.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"state":"CC"},{"updated":"2026-05-22 23:33:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-23 09:00:03.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2026-05-27 09:36:45.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"}],"messages":[{"id":"6460fa4cdf85138dbe8dc69d3c93ea591cdcf86f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-05-22 22:03:55.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"f2f13335c0d6e841801eae109f93749572dc0d11","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-22 23:33:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/312ab0a0d46248b8af91ae3cb09ec508\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa9e91d0e26b43acacba04fa0ec1b88e : SUCCESS in 9m 03s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/4bc17719f36e44139ca3f85ca1c9f2d3 : SUCCESS in 7m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/19b87f7985ff475bb85711fab271a260 : SUCCESS in 5m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6497e38d27634bb297cbfcea4924d524 : SUCCESS in 5m 28s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/adcfb57f505a4063b3916e5755a295f7 : SUCCESS in 4m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/980f859308a14669b81b2ef0eb30b33c : SUCCESS in 9m 01s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ff95e3ade82042ef8566ec43475914ff : SUCCESS in 5m 32s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6bee985bbf7b4aa78001108ac6631966 : SUCCESS in 4m 44s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1a4eca30933a4af8b7a2f534ae08d91d : SUCCESS in 6m 30s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d6d353bced6143f6813cdf85aecd4874 : SUCCESS in 7m 12s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/5f6f8ecd1a4945b482125d64af64a492 : SUCCESS in 10m 37s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4159bcba07d34d7bb857ad35ae07b4ea : SUCCESS in 14m 25s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/9dfd6ace3a3f48b1ae6c336f8ef2b5df : SUCCESS in 11m 44s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3c5ae98909f744b1b114974c952b5d40 : SUCCESS in 16m 10s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/e14dbc0599e54882b6426804e61a50bb : SUCCESS in 58m 44s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/686e48ea7be846b48e92a1e1fcb063e2 : SUCCESS in 59m 39s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/418fe6b5d1144f439fe54ac89077b620 : SUCCESS in 1h 00m 39s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/daf32017ef12407193c167d6f4ba82d8 : SUCCESS in 1h 00m 11s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/7cc435bd1c71415c8f2c2e59fd3cf12f : SUCCESS in 59m 44s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/ed1a73e8d11e43d1bebf38a31d8ae92c : SUCCESS in 1h 27m 10s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/33c90de9632a440fb0cc0c216361c5a3 : SUCCESS in 1h 18m 01s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/81f599a787674e93a0730a863966a620 : SUCCESS in 9m 05s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/1fac5ab2e25a47749dcc679273f0db2f : SUCCESS in 17m 40s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bcd0af259974465b98244afa693e3116 : SUCCESS in 17m 12s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/bb700facb9ee486dbf82d24644d7f382 : SUCCESS in 15m 43s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/07d72e674d9346d8adc631ee4b8ee7af : SUCCESS in 22m 09s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/befb2890585c494fb1b8958127d1119d : SUCCESS in 33m 57s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/8ff167d54d7542c79bb63311f613a858 : SUCCESS in 9m 52s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/db289220c614424faa5fe0062f01cbf4 : SUCCESS in 14m 39s","accounts_in_message":[],"_revision_number":1},{"id":"28b489ffe516249df701aecf38a352cfec9508da","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-05-23 09:00:03.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"fe149fc6bb3f9773bc5fdbc2974b91e34bde1569","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-05-27 09:36:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"dc798241161a9e0b6126c8d63f28629a9ec731f5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-27 09:37:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c10e7e714faf79229d29bf44edc3540148c590b5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-27 12:06:54.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6b4ce85707334301b63fa10d6b0f03e0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e41800d1442f4a72b9dffe4968293a3c : SUCCESS in 10m 00s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/1fc0c850c0bf4e24a7eee5a5d8e1660c : SUCCESS in 7m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5341b6dca1164529acde055d2ad43e45 : SUCCESS in 4m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/db70e9748aa140d2bbb1e21b098c83f7 : SUCCESS in 5m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e8f0cf52c34e4acb858fcedab4eeecc2 : SUCCESS in 6m 47s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b54fb70b819240b990be3fe2188024a4 : SUCCESS in 4m 42s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c9cdf584225d407eb69bcb28a041c338 : SUCCESS in 12m 10s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/fe805c85715241a59d8e2c35f8b7788a : SUCCESS in 11m 07s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/4aeb2a10cc2f4d839796df340f5690b4 : SUCCESS in 9m 55s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/946900d618cb4eaf8ffbbc274e673f6e : SUCCESS in 13m 25s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/63805e509c5441ed94d5686eb8fb90fb : SUCCESS in 17m 51s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6c05cffe20f345f891572fb41bb5729a : SUCCESS in 17m 32s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/70c89713118242018e3ba514e9ed4328 : SUCCESS in 57m 26s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/6c77c13d8c464f53b07993430d97be00 : SUCCESS in 55m 48s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/169a56dd3eb24843865fe27052d9aa37 : SUCCESS in 59m 17s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/accf6817ab5c491aa1f3f0878ee1d03e : SUCCESS in 55m 54s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/becf904abfae4a26a163dcc956381d13 : SUCCESS in 56m 00s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/b90fb4184bab49e78152e02a74b473e9 : SUCCESS in 1h 23m 16s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f0e066ed72ee4eb6a2a4c68cfea06f0e : SUCCESS in 1h 18m 58s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/d61df159a53544cc975163ceb9319b18 : SUCCESS in 22m 31s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/a9321c76beb04ac3a206ebbbbb45fffc : SUCCESS in 10m 37s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0bf19dfd869840a4a2a0b434405cdd84 : SUCCESS in 14m 36s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/4d3ed4aabf53434a86a162b27aa3dfce : SUCCESS in 15m 13s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/6dd7edce0fbd436d8e53b16b269da4dc : SUCCESS in 16m 44s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b9d4e93cf86b48789f04193b00f1a417 : SUCCESS in 27m 57s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/180c84ec6c604f199860cdbf1e5d2ae0 : SUCCESS in 10m 08s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/eb670398f2584191a774330e5a2a8f2e : SUCCESS in 14m 24s","accounts_in_message":[],"_revision_number":1},{"id":"19d6dca9b3e4cebe08e0c58ab69fa3e539794cf6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-27 12:06:54.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"91df17ac81d5b336348b569c9071f554a3aa647a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-27 12:07:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8545f5c8a62040a08638b392b4e8cc27\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d34ecbce55a04c4e85837ee017789485 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"23b030424b2eaa65e446dbf67c22ee5a32932448","revisions":{"23b030424b2eaa65e446dbf67c22ee5a32932448":{"kind":"REWORK","_number":1,"created":"2026-05-22 22:03:55.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/37/989837/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/37/989837/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/37/989837/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/37/989837/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/37/989837/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/37/989837/1"}}},"commit":{"parents":[{"commit":"9f779bca46e5b585a50c754a6dbefa55ac3a17ef","subject":"Merge \"Revert \"CI: Exclude docker-ce 29.5.1 on CentOS/Rocky Linux\"\" into stable/2025.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9f779bca46e5b585a50c754a6dbefa55ac3a17ef"}]}],"author":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2026-05-22 12:17:10.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-05-22 22:03:55.000000000","tz":0},"subject":"Fix ansible-collection-kolla not getting installed","message":"Fix ansible-collection-kolla not getting installed\n\nWhen ``ansible-galaxy collection install`` is called with two\nrequirements files, only the latter one is registered and dependencies\non the first requirements file are ignored.\n\nFixing by performing two separate ansible-galaxy installation. One for\ncore collections (Defined at file path\n``kolla_ansible_core_requirements_yml``) and another for additional\ncollections (Defined at file path ``kolla_ansible_requirements_yml``).\n\nCloses-Bug: #2153994\nChange-Id: Ie179ac46c28c08f3ed41922485ef2e532a2d6e46\nSigned-off-by: Seunghun Lee \u003cseunghun@stackhpc.com\u003e\n(cherry picked from commit 19fe0d493ea6505b36f2abeddb46f2c5e756fe30)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/23b030424b2eaa65e446dbf67c22ee5a32932448"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/23b030424b2eaa65e446dbf67c22ee5a32932448"}]},"branch":"refs/heads/stable/2025.2","conflicts":{"ours":"9f779bca46e5b585a50c754a6dbefa55ac3a17ef","theirs":"19fe0d493ea6505b36f2abeddb46f2c5e756fe30","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
