)]}'
{"id":"openstack%2Fkayobe~989963","triplet_id":"openstack%2Fkayobe~stable%2F2025.2~If3a92b059063b0aea299a51e76217c0d90a637a5","project":"openstack/kayobe","branch":"stable/2025.2","topic":"backport/nmstate-2025.2","attention_set":{},"removed_from_attention_set":{"35345":{"account":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"last_update":"2026-05-27 14:49:01.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If3a92b059063b0aea299a51e76217c0d90a637a5","subject":"Add nmstate engine","status":"MERGED","created":"2026-05-25 13:31:14.000000000","updated":"2026-05-27 14:51:24.000000000","submitted":"2026-05-27 14:49:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"989963-backport/nmstate-2025.2","meta_rev_id":"a6ccc145e955bbb25ad02e0f83384a4d8817a03f","_number":989963,"virtual_id_number":989963,"owner":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-27 14:49:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2026-05-27 13:20:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2026-05-25 17:04:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2026-05-27 10:13:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2026-05-27 13:20:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-25 14:58:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-25 17:04:48.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2026-05-27 10:13:39.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2026-05-27 13:20:11.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"}],"messages":[{"id":"9b0e9d24628df4e2a1da05350b9ff6e7a5153984","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"date":"2026-05-25 13:31:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"84c49c32677e984b905fbbb40da7cd312e6417a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-25 14:58:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/845c3f5e0c93477f872316866200366c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3db47531790d4f94ba93ee6f15011c49 : SUCCESS in 9m 47s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/0894b01dedb444eabdb4043ca776b02b : SUCCESS in 3m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a5e72c5842c8420dba6f31d480eee3c3 : SUCCESS in 5m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42baa81dafe0483abe1249f61c664f9c : SUCCESS in 3m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/199abb79d9d84a839e3fabf6745901c0 : SUCCESS in 5m 57s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/df9e66d9d1c24c039c694cbdf3bea41d : SUCCESS in 7m 11s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a7aaa14d5d7e4298a79442e2d8408585 : SUCCESS in 6m 42s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a81c33cd5b54493bb5096582f663d556 : SUCCESS in 3m 45s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d07cd7f992294c09b35f0c22623cbfd1 : SUCCESS in 11m 25s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/055a12030e8748f89a9ffc881627c9a3 : SUCCESS in 11m 15s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/de65e9c41e3c4b9182b8a719f2cd2b22 : SUCCESS in 13m 53s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f25458cb5727478e9260360614776994 : SUCCESS in 14m 00s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/cda27a08240b438ea5b32a1790fce435 : SUCCESS in 18m 45s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fb906f1030b44bfe8b7ed08193f9d3dc : SUCCESS in 8m 33s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/b678c1ed51bf414db0f48bbfcbf21068 : SUCCESS in 56m 30s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/f4575154d10e4ca497b52607e7f796f0 : SUCCESS in 56m 11s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/49efcd23ebbb43f8ab2a84b54096e917 : SUCCESS in 57m 54s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b47a6f6a1644488da756bfd699f39948 : SUCCESS in 57m 10s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/882ff27cb05c4163a4d0f1a1aa722d70 : SUCCESS in 56m 39s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/6dafaf16c7ef45b5bfac10d7fa399ef2 : SUCCESS in 1h 24m 32s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/233bb7e28ca644318e7fab2fbc725d0d : SUCCESS in 1h 21m 38s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/866d10bf42064ec3aeb7a35c138eb6f0 : SUCCESS in 11m 27s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/bd48182e41334fc0a1fa18fbff1824a5 : SUCCESS in 21m 59s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b45bafcbbe254fd0b7531c2276cadc31 : SUCCESS in 16m 18s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/4d70b1a7d30041ca87e5cce6eb6f70df : SUCCESS in 14m 05s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/86687e241b5e44d090ee7bf5f487a711 : SUCCESS in 19m 16s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d91dc1b3fed84e649ffb3f66e4af7bde : SUCCESS in 16m 09s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/66a5415e4504472c8382a71b2bf32880 : SUCCESS in 9m 10s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e0e8cf12ec494e2abd073068cc58d16b : SUCCESS in 8m 35s","accounts_in_message":[],"_revision_number":1},{"id":"96cd603c26c21148f466c6a40d68656dfc7dfa6a","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-05-25 17:04:48.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"964831c7fa9111c3e53ec230feba34461bfc5d45","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-05-27 10:13:39.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b962e7e1dafabf8b922f210655cd319bc6f5ff76","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-05-27 13:20:11.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"08fc240958f1d6ca691fd79d161eb85be3870e2b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-27 13:21:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"14afe9e69aa9f2881b830fcbb76adc743bc26e22","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-27 14:49:01.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d4d20a86be494bb79c2504aa9d3d17c0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1afc7952431e46beaece9644313340ad : SUCCESS in 10m 01s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/f6a82db597e94492812fa484c1aeb347 : SUCCESS in 7m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b8e45fa9ba1a4308a4c6ea3e36ad633d : SUCCESS in 6m 08s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/85cb77e0f1d54942a047bb5a2026a407 : SUCCESS in 4m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bb384b62df1f4ac5a12fda77ed02c05c : SUCCESS in 6m 23s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c2a6139f95b44491b1d66e6a9df8c8b6 : SUCCESS in 5m 33s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9938e5b739ce4003b785eb233ce48fb6 : SUCCESS in 13m 08s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/43271085a46449959adf59341824aeb2 : SUCCESS in 11m 16s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/2ece5cc410124469b1f1025a7ee218f5 : SUCCESS in 19m 28s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b3083cdece684dd2b20c2f3a9ff5e8af : SUCCESS in 14m 38s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/de9435f0982f4a99a6c2cf01197fd837 : SUCCESS in 15m 22s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/42f84c1ae65e40679637ea33896855c5 : SUCCESS in 17m 34s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/c0a1ecbba2bf43078a807ca5557b6248 : SUCCESS in 57m 11s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/5d995dd02e58432098f0e78ef9a8231f : SUCCESS in 1h 01m 33s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/2c67ac53167b4fb7b5c9dc4fb80bc5f4 : SUCCESS in 1h 00m 31s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b165134708e74366b649d413d8c6c618 : SUCCESS in 55m 25s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/2aab59c697ef43b1a1034da747d42b8b : SUCCESS in 57m 55s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/6b09ba35f2594a57938563f557cc7724 : SUCCESS in 1h 26m 59s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/73362edecbac4ed0a189f90773222691 : SUCCESS in 1h 24m 31s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/50c84988974c4419b887f7aeb400f996 : SUCCESS in 13m 09s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/5d3a1c58cb0244769d8e46fd868af22f : SUCCESS in 18m 30s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0e0156311b8b42699da648dfa8f32f1c : SUCCESS in 13m 15s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/50f5818fb7564e35b7d89a56d2e71c59 : SUCCESS in 20m 02s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/6ba07fba075c48d680d7975c94c8d3db : SUCCESS in 35m 16s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ee8e11b95b1b42bc9cc40f7ab71d4fdb : SUCCESS in 34m 08s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/16b9aaa9149e4881881ba0c3c6a7d674 : SUCCESS in 12m 53s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/29ce42dc6ee74fceafcf82b9458c10a5 : SUCCESS in 17m 27s","accounts_in_message":[],"_revision_number":1},{"id":"47b97b6864430b8a6aa9061644cf0185a0760910","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-27 14:49:01.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"a6ccc145e955bbb25ad02e0f83384a4d8817a03f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-27 14:51:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4467e449e2c84b1ab0c1a95322eb1561\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/098d421560274d999ce065941e64b082 : SUCCESS in 58s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cf1c2024567145f5891a0e6494485430 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8728b9b34d24d7e4e46017c82bd78078d1105ebb","revisions":{"8728b9b34d24d7e4e46017c82bd78078d1105ebb":{"kind":"REWORK","_number":1,"created":"2026-05-25 13:31:14.000000000","uploader":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"ref":"refs/changes/63/989963/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/63/989963/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/63/989963/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/63/989963/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/63/989963/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/63/989963/1"}}},"commit":{"parents":[{"commit":"9f779bca46e5b585a50c754a6dbefa55ac3a17ef","subject":"Merge \"Revert \"CI: Exclude docker-ce 29.5.1 on CentOS/Rocky Linux\"\" into stable/2025.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9f779bca46e5b585a50c754a6dbefa55ac3a17ef"}]}],"author":{"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","date":"2026-01-26 16:04:14.000000000","tz":60},"committer":{"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","date":"2026-05-25 13:25:53.000000000","tz":120},"subject":"Add nmstate engine","message":"Add nmstate engine\n\nAdd an opt-in nmstate network engine via `network_engine: nmstate`\nfor host network configuration using NetworkManager/libnmstate.\n\nImplement nmstate rendering for Ethernet, VLAN, bond, bridge, routes,\nrouting rules, and structured `*_ethtool_config`. Add OVS patch link\n(veth pair) generation in nmstate rendering.\n\nChange-Id: If3a92b059063b0aea299a51e76217c0d90a637a5\nAssisted-by: various models\nSigned-off-by: Grzegorz Koper \u003cgrzegorzk@stackhpc.com\u003e\n(cherry picked from commit 8b8ac1c293fb48b5c01fc294bfb0f6e2b313737c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8728b9b34d24d7e4e46017c82bd78078d1105ebb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8728b9b34d24d7e4e46017c82bd78078d1105ebb"}]},"branch":"refs/heads/stable/2025.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
