)]}'
{"id":"openstack%2Fkayobe~989965","triplet_id":"openstack%2Fkayobe~stable%2F2025.2~Ie749959774b9c0d25cc9b47250539b5dd75c53e4","project":"openstack/kayobe","branch":"stable/2025.2","topic":"backport/nmstate-2025.2","attention_set":{},"removed_from_attention_set":{"35345":{"account":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"last_update":"2026-05-27 15:11:01.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie749959774b9c0d25cc9b47250539b5dd75c53e4","subject":"Add nmstate VLAN QoS map support","status":"MERGED","created":"2026-05-25 13:31:14.000000000","updated":"2026-05-27 15:12:23.000000000","submitted":"2026-05-27 15:11:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"989965-backport/nmstate-2025.2","meta_rev_id":"5d95b362acb1615718c69d67f21f5215274c066f","_number":989965,"virtual_id_number":989965,"owner":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-27 15:11:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"recommended":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2026-05-25 17:05:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2026-05-27 10:13:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2026-05-27 13:20:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2026-05-27 13:20:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-25 14:57:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-25 17:05:12.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2026-05-27 10:13:47.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2026-05-27 13:20:24.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"}],"messages":[{"id":"5bd5956a04bdc28df5d5adaeeb0d02baa18bc30e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"date":"2026-05-25 13:31:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"48ee778c8f8f2ef2c856657ea5b34fd617c1099e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-25 14:57:54.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e81336cc21b9440ea76eecb17f7d327b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dbfc0e607b50432aae013f3795703afc : SUCCESS in 9m 46s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/ee33fc19da71417fad5e54a937bbd7fb : SUCCESS in 7m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/555fcf3e01634f86b60fdcbcaeecd457 : SUCCESS in 6m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/deee9e2e9db741168a74f8825c587013 : SUCCESS in 2m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b0a0b9c0cb0e46eeacd1b32c6229c7a2 : SUCCESS in 4m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e3f51a911d6f4447ac8ca5467e2cd4b0 : SUCCESS in 6m 02s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ce2de0a66da5411d86bc28545beda7df : SUCCESS in 6m 08s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f6af5a29440a43169e1ef7d853e3d577 : SUCCESS in 5m 04s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/31fb27942c434373aef7b242de04064a : SUCCESS in 11m 59s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/99e6244462db4d4d8290c4f95bbc6195 : SUCCESS in 9m 54s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/1f38e8102537499d8653ff03a725a3d6 : SUCCESS in 10m 49s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/52f0a5c7e1f045dfadcdf7be5ad2835f : SUCCESS in 12m 29s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/f25d7c38175f48c0ba27bedffa541a31 : SUCCESS in 10m 22s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f555860d0eac464baa4c1708b3704dd1 : SUCCESS in 10m 04s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/6f7b0486d5e647b3842779a0502c9cfb : SUCCESS in 56m 49s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/6ed7d4bfddcc4705961e84d442c023c0 : SUCCESS in 56m 07s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/3d7e80a8d38f431483c02be88adaf906 : FAILURE in 56m 14s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6bd5021a05654908b1eefc1818a884d5 : SUCCESS in 55m 23s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/fb9bb14a6b8844788105070e71721b81 : SUCCESS in 54m 54s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/0ed048a7a8b7420487788afecffac2af : SUCCESS in 1h 24m 10s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a38ec0aa79a34d64980f9bc0b3961219 : SUCCESS in 1h 20m 30s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/d1dc90c29ce2407290ca747c4f6e3f7c : SUCCESS in 9m 38s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/1cde42b87b0a432a963e82265a422d35 : SUCCESS in 18m 20s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4ae21f97df8e4d7cbccdad7ee6f0d06a : SUCCESS in 18m 05s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/fcdf02733f6b41679001fac8ca92dff4 : SUCCESS in 15m 47s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/72005ebb224740dba10452383e2cebad : SUCCESS in 22m 06s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0e6885e20cd247bcb4406958f4900373 : SUCCESS in 28m 06s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/fde1a5ea81064c76be758375bdf60d1d : SUCCESS in 10m 31s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/67fb87b40e3341a385de21f6c654f70b : SUCCESS in 7m 34s","accounts_in_message":[],"_revision_number":1},{"id":"02df7772e295efee965a1bea96e1d785ffa7bddf","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-05-25 17:05:12.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"01bf5623f84096195381446c2fe1b3210198db7e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-25 18:29:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9e63ad963a8f4538a3266b8e6fab04d1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c82945db232c4363ba09152fe90f328e : SUCCESS in 7m 30s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/f3be813af955498e970f6e6b4b2cf508 : SUCCESS in 4m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/83e7047b7c4a4ac892c7d120f5cd3230 : SUCCESS in 5m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e1e0f3cdef949efbbfe4faa4dcf1090 : SUCCESS in 2m 39s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8c9b579ca78f418fbaef87c4e6092118 : SUCCESS in 4m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/af2b5f31f06d4d0cbedb539d79056857 : SUCCESS in 8m 26s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/edbc65e842f743fdb773960188fceb10 : SUCCESS in 3m 10s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/64dd832a438a41ce9a7671fe5d74c0e0 : SUCCESS in 6m 25s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b14a5eb55ccb4e03ab47b84b55ddb95f : SUCCESS in 11m 41s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f5df7be1373c4002bbe36eb7b5a35424 : SUCCESS in 6m 22s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/14640403d434498187525b4940d50972 : SUCCESS in 12m 44s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8e5329d1db6e4dd198de1e1cebb1dd54 : SUCCESS in 12m 54s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/101c18f8609f4ed4b2d0a8698a31ed03 : SUCCESS in 9m 05s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2ec7e017989b40e38c5e0accdd5cc94c : SUCCESS in 15m 49s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/be1661a84b104fca96464f089e70ea7b : SUCCESS in 53m 36s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/ad8b0224ac154ec999104d18da0d89ba : SUCCESS in 55m 12s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/076fff3f35ff4cc7b042849ee48dd7b6 : SUCCESS in 57m 59s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bca8e41f9e314279aa829401e25db80a : SUCCESS in 56m 48s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/13c0442fa35940898d093f87659da9e8 : SUCCESS in 57m 30s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/c0dda88cd2fc4bceaa7f33d78573bb52 : SUCCESS in 1h 22m 29s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e8f123a7e3c7442c8a42616e07588afa : SUCCESS in 1h 19m 33s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/e0da44544c71407a882d3ab6ddf0ac8b : SUCCESS in 22m 00s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/b6cf7b43f01f4c98b80dc9618dcaac6b : SUCCESS in 11m 31s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2c24124f3ae44c5daa25b587bb98cc36 : SUCCESS in 16m 35s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/4006540d398c4a47afefd21bc39d2858 : SUCCESS in 9m 53s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/6bc4566ee3154caa9b64b1ae90b977c7 : SUCCESS in 21m 50s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6cdccf462d98480bb1ee6e4fae4a7f6e : SUCCESS in 21m 17s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/dfc8c1bcfa4546d6b4aa34d9cbf21ca9 : SUCCESS in 20m 29s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5af53b43421041b99f4383f7e60eeb95 : SUCCESS in 14m 51s","accounts_in_message":[],"_revision_number":1},{"id":"a774ac6850055109f6557725947c90b8ae8106ff","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-05-27 10:13:47.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"fd93067a004110ff46fd3e1e98a9159ed21888b8","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-05-27 13:20:24.000000000","message":"Patch Set 1: Review-Priority+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4957cebf2d4095f2d88e4cd6f4952442f273cf84","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-27 13:22:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0c35a6d7a4f68847021c36f2fb648eaea0cd8101","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-27 15:11:01.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c573ccdfdea04a03986af4f074fac8d4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c94c7db263646a89c9176e2f6f53f00 : SUCCESS in 10m 36s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/d1565d1a0a4b494c93cc61a00529208d : SUCCESS in 6m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bac062eb4e8c40e9ac8667bb43767ec1 : SUCCESS in 4m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1676109cc45b4df8bfaba6662854bd6c : SUCCESS in 4m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e834816940414919b6d0c21474932636 : SUCCESS in 6m 51s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/10d214530e194b77ace82d124920ef3e : SUCCESS in 6m 42s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/191947428c2942cca5a467018810ce73 : SUCCESS in 14m 48s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/64c1443c8b4347139544f20dc4749fcf : SUCCESS in 12m 13s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/0bbf6690b7ab44f7bba7dac19df420b6 : SUCCESS in 13m 48s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/24997ea0a0b14c4188cf6d95257192b9 : SUCCESS in 15m 40s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/176fd4c9a08643f1aaafa4c6c8626f3a : SUCCESS in 21m 31s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/83e739be0c47432b9e3d40c0f5930bdd : SUCCESS in 17m 00s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/851537e3b65140fe9b5239a348beb4e8 : SUCCESS in 1h 00m 44s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/fe4b290f1b1440ebaf3bedfdd85ba605 : SUCCESS in 1h 00m 00s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/f304f0880a0d42c0bc9ddab784f8ce32 : SUCCESS in 1h 14m 59s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e5f6a04604814ee982c49c70ff362eaf : SUCCESS in 1h 18m 01s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/ded9b9696e6e421da7abadadceebbe96 : SUCCESS in 1h 11m 51s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/0b746399e18b4dd7af5dba33c149a9ca : SUCCESS in 1h 25m 56s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/023f52d9e61844b19d3c218c1670a1b4 : SUCCESS in 1h 43m 50s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/83a0106139824f6bbc06b5fd79c1ac1e : SUCCESS in 19m 30s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/ae0c450ca80d4d1cb4e7fbd00c6a2a5d : SUCCESS in 16m 19s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/743cada15d5d49119b31c4b2bc876e47 : SUCCESS in 17m 41s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/3c24c73a849349548813ea2f528c1eb6 : SUCCESS in 11m 19s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/e554280c68c043f0a6ddbaf90b48ed39 : SUCCESS in 28m 06s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/29cb46bb10184e7892c734c88d67d658 : SUCCESS in 29m 06s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/903c9967581146feb2305f4d470621d2 : SUCCESS in 19m 54s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d4c55675c5004b829770ce0adb829a1a : SUCCESS in 13m 26s","accounts_in_message":[],"_revision_number":1},{"id":"e5a29bcb5929df9d8dc77f1e607ed97c7bab19d5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-27 15:11:01.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"5d95b362acb1615718c69d67f21f5215274c066f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-27 15:12:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7e4af225a912451094621363d9a10450\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4322515f3824e75b01495388454e2be : SUCCESS in 44s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6e2d9d26882e4074b6de1bf69ff723e0 : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a8f451b74a29994b267c558a598d869cbdd872a7","revisions":{"a8f451b74a29994b267c558a598d869cbdd872a7":{"kind":"REWORK","_number":1,"created":"2026-05-25 13:31:14.000000000","uploader":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"ref":"refs/changes/65/989965/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/65/989965/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/65/989965/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/65/989965/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/65/989965/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/65/989965/1"}}},"commit":{"parents":[{"commit":"5b1b36b3cb6c0c416f48c5a9f150bef5cba1e249","subject":"Tweaks to nmstate release note","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5b1b36b3cb6c0c416f48c5a9f150bef5cba1e249"}]}],"author":{"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","date":"2026-04-22 15:58:18.000000000","tz":120},"committer":{"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","date":"2026-05-25 13:28:54.000000000","tz":120},"subject":"Add nmstate VLAN QoS map support","message":"Add nmstate VLAN QoS map support\n\nRender VLAN ingress and egress QoS maps for the nmstate network engine.\n\nAssisted-by: Codex(GPT5.4) and Opus4.7 as an initial reviewer\nChange-Id: Ie749959774b9c0d25cc9b47250539b5dd75c53e4\nSigned-off-by: Grzegorz Koper \u003cgrzegorzk@stackhpc.com\u003e\n(cherry picked from commit 40fca0f7c415247c06f5f05954297b78143bb156)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a8f451b74a29994b267c558a598d869cbdd872a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a8f451b74a29994b267c558a598d869cbdd872a7"}]},"branch":"refs/heads/stable/2025.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
