)]}'
{"id":"openstack%2Fkayobe~990300","triplet_id":"openstack%2Fkayobe~stable%2F2025.1~I266b720156c0d9cf5672d5dcc5a44a2a43888df6","project":"openstack/kayobe","branch":"stable/2025.1","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2026-05-27 21:07:55.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I266b720156c0d9cf5672d5dcc5a44a2a43888df6","subject":"Fix issue with systemd-networkd-wait-online when no IP","status":"MERGED","created":"2026-05-27 14:12:49.000000000","updated":"2026-05-27 21:08:59.000000000","submitted":"2026-05-27 21:07:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"990300","cherry_pick_of_change":975386,"cherry_pick_of_patch_set":3,"meta_rev_id":"09e5938aeeef37743fd7aeee7d6f09f1a5b7208f","_number":990300,"virtual_id_number":990300,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-27 21:07:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2026-05-27 15:48:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2026-05-27 20:36:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2026-05-27 15:48:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-27 14:12:49.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"CC"},{"updated":"2026-05-27 15:48:53.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2026-05-27 16:36:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-27 20:36:41.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"6d98f599d3b909945bff2c247ffbfe564a25fa72","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-05-27 14:12:49.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"1ee9dbce19c17758ff3d1c470dced5bc40a75a8f","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-05-27 15:48:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b77093258fc564304abe10d59b920d9985779a33","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-27 16:36:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a41282170fce402ea51141f73d68d0eb\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8e73062e01164eacb62a45cc0ca197f0 : SUCCESS in 5m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f25e4faed52a46a8b858a7c2f5174813 : SUCCESS in 6m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9f90901a5b834724b654bbfc671c1eca : SUCCESS in 5m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90da31453c8846c982e79d14bdf6bf99 : SUCCESS in 9m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9cb12030097f41699f754a615ac85b20 : SUCCESS in 3m 36s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/0c8d786641604f9c88f388a21168de6a : SUCCESS in 4m 43s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f66b6230363846d0827bbb2b55702712 : SUCCESS in 14m 13s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f0d3bd96270c44fababbf4f4f860a503 : SUCCESS in 13m 12s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/4e9bb15085184a029ccbb96553e41d59 : SUCCESS in 17m 47s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/73d93d33f20b4b6eac55f70fbbe14df1 : SUCCESS in 9m 55s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/819b93ce13dc45b1848f0164f4afee17 : SUCCESS in 12m 25s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/05c6cfca12834bc8863601d88aee6333 : SUCCESS in 23m 37s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/407c936657c04cbc8f915739d3a90b79 : SUCCESS in 19m 04s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8d01343b1a16492a9b4d42753b4d8c9b : SUCCESS in 15m 51s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/816fe128c5c9479981a185f49dd652f6 : SUCCESS in 1h 16m 29s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/6076990c72444e13a09ecd1d24b55322 : SUCCESS in 40m 54s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/1f91bdeeb1ca4092a239273f123aed0c : SUCCESS in 58m 38s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/dd8ba4a5252245c28ea574d25fbd2b01 : SUCCESS in 56m 48s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/43af9f56387a41c7b28db1e8ad302f3a : SUCCESS in 46m 47s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/a9f58da7293d47ef8866559321a2de1b : SUCCESS in 1h 10m 20s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/af8abe1bb0f442e4a6e45ee8bd9b9be3 : SUCCESS in 1h 15m 46s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/80d1fcee66a44c08a288337747df0d6a : SUCCESS in 54m 48s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/8cf65e34612c438dbde7e10ea00ec2f1 : SUCCESS in 1h 31m 37s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/32c72f1417c04af29793ea824af7c883 : SUCCESS in 2h 17m 31s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/cc58f9b7d5854bc1b1ba49a501907c0e : SUCCESS in 18m 50s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/c2d9dd4f63ee4f7783d6dfef2ef558fb : SUCCESS in 13m 41s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/054ea6e646004e63adc3d52a5eb50e1c : SUCCESS in 19m 45s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/0e1e83b0812c4730bbdeb4911c4759c8 : SUCCESS in 19m 01s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b91280614d664ba18bac5e72b2bb09ac : SUCCESS in 17m 53s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/cc9373129e6c4ecab6e314ef49389095 : SUCCESS in 16m 44s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/43b278bf0be24597980c1589f75b25c3 : SUCCESS in 37m 54s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/a38493ff2e4740ac9d7be14ba74971c3 : SUCCESS in 14m 09s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/cc3a5051a3564548bec91b6070305197 : SUCCESS in 15m 34s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/492586a1825844949c5245d97fbca1da : SUCCESS in 14m 40s","accounts_in_message":[],"_revision_number":1},{"id":"324eb67e224e82e193c6491097fa5117d52a8db8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-27 16:37:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a989989d6ddd9d05480f9f4c70d09df979d86483","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-05-27 20:36:41.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"9e8cd0a6d9f0d77031a1cd8c21003f9a23230430","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-27 21:07:55.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4e089062fb3849d1b2d292f6a8c08e67\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a41b13f11264ccaa58e9f075aec5521 : SUCCESS in 2m 31s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0aec373f5a7a4ce88de3b5e04f8a77d9 : SUCCESS in 4m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38d70ba6b5764f48a11d8152f0dbe0d1 : SUCCESS in 10m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c0b53dbf563f49a58460d7a57a944b68 : SUCCESS in 6m 17s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3f5727a6466d48ce9bdfbcf3c1bf98e5 : SUCCESS in 4m 57s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a5278599b68c4d3e8091ba693c50ff7b : SUCCESS in 11m 28s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f6d268f9f37b4b7595db334a7941d2c2 : SUCCESS in 10m 35s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/485557bff3ef4fbfa42cc9f2d3280f8e : SUCCESS in 18m 57s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/3d9202466a914bd78eabef1dc3f9d52b : SUCCESS in 15m 31s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6f70553839ea48cda9c926f8e983a30a : SUCCESS in 12m 56s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/16b07f8a63f64d7b99a81f41c439e297 : SUCCESS in 20m 16s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/3f55bb032b9b46808ea916fafd5862a1 : SUCCESS in 20m 28s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/43cc24b54e4d4d5c8194c5d82a620646 : SUCCESS in 15m 34s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/d570e2f0fd764d62bc0c4979066f5cea : SUCCESS in 33m 11s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/33dfd2b8c1d644acb3e9af1aa2704faf : SUCCESS in 36m 56s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/a941c29e20d347449d51c9c155ae873f : SUCCESS in 54m 30s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/91732ad135144e6fac83c7a2629c3360 : SUCCESS in 1h 04m 46s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/56c46d0e6f944cf7b84607c81d7ff3b7 : SUCCESS in 1h 02m 33s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/d2e644019afd48bea8e9e497466aacc6 : SUCCESS in 1h 12m 20s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bf84b098c09e4719ba207d0d35b48ebd : SUCCESS in 36m 36s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/868ae6b23c944e44b0931975bdde865b : SUCCESS in 52m 55s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/3f174b94676f4965b4fa3d8da9de1450 : SUCCESS in 1h 57m 50s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/82d785413cc34e00b3e2ec31637b9ffc : SUCCESS in 1h 49m 30s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/ce73034bc4724171bc8958501f34fcf5 : SUCCESS in 10m 36s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/f0cefbec895f4381b4345fc59d4ac5ef : SUCCESS in 21m 09s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/0da0b6f3a20a4443a05e3c18f9eb4727 : SUCCESS in 11m 10s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/a0b5ea74e0c94c8ba22cb44c2f3a4987 : SUCCESS in 21m 37s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3b4200c8b56e4d7cacc117cfa984b57e : SUCCESS in 18m 29s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/5e07ff3153ad48a7bf30efa8215e1fee : SUCCESS in 15m 33s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/ab3ff006ab9e478fb9f29e20a3a93e42 : SUCCESS in 39m 25s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/47d2005425a04de7ae2125ddd7bd7cea : SUCCESS in 15m 20s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/fc04a6a25f9344f3be05b9d01948e6a1 : SUCCESS in 15m 41s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e1fdfdfb67304ccfa6e2f247417ebad4 : SUCCESS in 15m 33s","accounts_in_message":[],"_revision_number":1},{"id":"cdb344370823288d09f44ad93cd631a1440c2881","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-27 21:07:55.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"09e5938aeeef37743fd7aeee7d6f09f1a5b7208f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-27 21:08:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4192d3e168bb4d35b019f849ba5978e3\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30915676d42e41e491e1c8210627d5aa : SUCCESS in 52s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/777e38e1d92c44a9be1d0a008843b048 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"56abceab85c2f6ff49f5b6b477b86d3ddca64fcc","revisions":{"56abceab85c2f6ff49f5b6b477b86d3ddca64fcc":{"kind":"REWORK","_number":1,"created":"2026-05-27 14:12:49.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/00/990300/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/00/990300/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/00/990300/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/00/990300/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/00/990300/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/00/990300/1"}}},"commit":{"parents":[{"commit":"32e587c3c77009af1e23c52b757dfd4d5596429a","subject":"Merge \"Fix ansible-collection-kolla not getting installed\" into stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/32e587c3c77009af1e23c52b757dfd4d5596429a"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-02-02 12:08:31.000000000","tz":0},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-05-27 14:12:49.000000000","tz":0},"subject":"Fix issue with systemd-networkd-wait-online when no IP","message":"Fix issue with systemd-networkd-wait-online when no IP\n\nThis sets the RequiredForOnline[1] property to false when no IP\nis configured. This prevents the interface from entering the\ndegraded state[2].\n\n[1] https://www.freedesktop.org/software/systemd/man/latest/systemd.network.html#RequiredForOnline\u003d\n[2] https://github.com/systemd/systemd/issues/575#issuecomment-124286854\n\nCloses-Bug: #2139592\nChange-Id: I266b720156c0d9cf5672d5dcc5a44a2a43888df6\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n(cherry picked from commit cb5925368d9d9e3d5d4b2809c3c4f456b5d1bb26)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/56abceab85c2f6ff49f5b6b477b86d3ddca64fcc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/56abceab85c2f6ff49f5b6b477b86d3ddca64fcc"}]},"branch":"refs/heads/stable/2025.1","conflicts":{"ours":"32e587c3c77009af1e23c52b757dfd4d5596429a","theirs":"cb5925368d9d9e3d5d4b2809c3c4f456b5d1bb26","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
