)]}'
{"id":"openstack%2Fkayobe~990326","triplet_id":"openstack%2Fkayobe~master~Iece6b0d8373356632857d8f7f7d4ef023e54d52b","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2026-05-28 10:54:40.000000000","reason":"\u003cGERRIT_ACCOUNT_28048\u003e replied on the change","reason_account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},"35263":{"account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"last_update":"2026-05-28 09:00:19.000000000","reason":"\u003cGERRIT_ACCOUNT_35263\u003e replied on the change","reason_account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}}},"hashtags":[],"change_id":"Iece6b0d8373356632857d8f7f7d4ef023e54d52b","subject":"Revert \"[release] Use OpenStack 2026.1 release\"","status":"MERGED","created":"2026-05-27 16:17:21.000000000","updated":"2026-05-28 14:12:35.000000000","submitted":"2026-05-28 14:10:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"990326","meta_rev_id":"fb2056f0743a043aaad51fcd4285511b40faf3c2","_number":990326,"virtual_id_number":990326,"owner":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-28 14:10:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"recommended":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":1,"date":"2026-05-27 20:41:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-05-28 07:33:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":2,"date":"2026-05-28 09:00:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":1,"date":"2026-05-28 09:00:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-27 17:59:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-27 20:41:47.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2026-05-28 07:33:42.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2026-05-28 07:33:42.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"}],"messages":[{"id":"712dd0d008aadbd358a529d49d735dfe041755a0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-05-27 16:17:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7fd33af7b6c5458b3f88c36fe805eee506f61350","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-27 17:59:30.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0c8ad31dfba44f1d9f9ee36528429406\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/003670a5cea3460191163232547eb857 : SUCCESS in 5m 07s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/c3983d36ba604b70a4d05bc1dea23dce : SUCCESS in 6m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5fb998a8681a4287951735084c5bbd03 : SUCCESS in 6m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7fa0a7104c546a2819ad33dbde02cd1 : SUCCESS in 2m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/32129d09cddb4515bcf7bce18fc81ad7 : SUCCESS in 4m 05s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/6d8f380204cd42858da7154e7341a3c3 : SUCCESS in 4m 13s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0d73db6d815e4b6e83bfd95383e57580 : SUCCESS in 5m 32s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2321a43b7ab94084a1440f51601656d5 : SUCCESS in 5m 08s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/25a0b1dc16014c7f9675f2d928d92405 : SUCCESS in 13m 03s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/206c57a897b544de93c48ed2e76a49ab : SUCCESS in 8m 05s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/db149cd6bf6d4096808e5460121c9f35 : SUCCESS in 22m 50s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/4dda400160ff47bb9b0bd0fed909a4a7 : SUCCESS in 22m 24s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/11fd00ac05b54de3885916f2258cadd3 : SUCCESS in 16m 37s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/2c84813533114d9a983a6999ad7bc645 : SUCCESS in 16m 04s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/5cc3dea1e9454b7881d32f5719b743e8 : SUCCESS in 24m 21s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bba776afa3694c2493681debd265b693 : SUCCESS in 18m 35s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/bfbe341523384011b08ddc9cab82b83b : SUCCESS in 25m 12s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/42edaf5e307f4a59ba8de7e2eb836390 : SUCCESS in 10m 47s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/4c6e143aaeb1479c8024295f5f3e8ce4 : SUCCESS in 1h 06m 32s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/c3a6bd9c643247a5b8880b4dd1dedbf5 : SUCCESS in 1h 09m 19s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/22e96f9fcb0c43669595622dbf6cc0a8 : FAILURE in 1h 08m 17s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b2156f033c5249a99c5ccd310597f618 : SUCCESS in 1h 01m 56s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/82d79be09ae449629fad782b4db5b238 : SUCCESS in 1h 27m 15s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/6d75236681cd4b1ea74825a834abdf51 : SUCCESS in 1h 40m 18s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e3e34d9f7bc444f1896c085dbeb22b92 : SUCCESS in 1h 31m 48s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/ed5f4b147bbe4eddb6593cff77f13327 : SUCCESS in 14m 33s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/90da0a99ac4741509698c37bb643d499 : SUCCESS in 29m 18s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ccadc3d7bc3741738a5cbab04eebd406 : SUCCESS in 27m 27s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/5c3e43da969541ccbb2b1182a81816c7 : SUCCESS in 15m 02s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/f7ed32236e3b41b8bd80cb4033c87cd3 : SUCCESS in 18m 56s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/da91b2814d3e475e8cd0a6f0660348e5 : SUCCESS in 27m 31s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/793d34f17271436baec0a4b8e70cd42c : SUCCESS in 16m 46s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e6396c3b9a414f72b96a9eaec4a4d0a8 : SUCCESS in 18m 29s","accounts_in_message":[],"_revision_number":1},{"id":"eb9b94846906a150bdb5c9ec577a966dee15bda1","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-05-27 20:41:47.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6beef9d9da4e34f505beaf9afac4d500464e35f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-27 22:17:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/277fdcec84b34cc78940726264841a2d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73f1c51464284ab8b757d5cde440b76b : SUCCESS in 8m 45s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/82d183e9adbd4f8dba432453aaf449e4 : SUCCESS in 6m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c654602933684f3bb3679de1f04bd7ec : SUCCESS in 4m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b4da923f3471490f866b5d6b0a87794e : SUCCESS in 4m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/61b404b97b764bd88bb047aed1288a22 : SUCCESS in 4m 15s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/5add1040b4e54634843b46d14281e833 : SUCCESS in 6m 31s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4e6c4ea8a807432681245a5b1782d389 : SUCCESS in 5m 16s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/29250d71185f4387bfc45f0b6a87ef03 : SUCCESS in 7m 24s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/bb4a28e515754d76bbe4ca75687c61a7 : SUCCESS in 11m 50s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/97f55ca8cf5b4894b0e41885fcf7290a : SUCCESS in 6m 31s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/9f2058e96c4c46bfa98be50923cad41b : SUCCESS in 11m 25s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/701271cfc36545529cbf379cf6065a47 : SUCCESS in 11m 59s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3a513440cc11442a9470ef67a9fe7ef7 : SUCCESS in 14m 32s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/ce202bf4258746129861030f2054a305 : SUCCESS in 18m 06s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/6a46a01489b54828a679fa00563aa614 : SUCCESS in 20m 48s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/aa9f88f94b38460bb84e0090eb2cbee2 : SUCCESS in 13m 13s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/13627cf0a0ba4021bd4dbe37dd3393a0 : SUCCESS in 14m 53s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/46b44b34642442ce9b6078ef165a19a5 : SUCCESS in 11m 34s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/9511c65dcad441c9b947e40234948f65 : SUCCESS in 1h 01m 42s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/3d6f111534da4e6b84b1853f33807eab : SUCCESS in 1h 03m 51s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/71cb57085ae74b3db7058b71c854c0c0 : SUCCESS in 1h 05m 42s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/70a2dda4a735425881c2fb7903447d7f : SUCCESS in 1h 01m 05s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/a529c70d28f44bf4a4188197117a304a : SUCCESS in 1h 03m 48s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/02858c98a474424f90440c2d236d43df : SUCCESS in 1h 34m 08s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/41634976404c4feba21e1c9b16f29581 : SUCCESS in 1h 34m 22s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/2f74da6d0ef447ffb3fbd294d9d9c673 : SUCCESS in 18m 36s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/a7ca2ecc8c4e49a2aab45496ed7ce73b : SUCCESS in 21m 57s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/10f9d968360f483d9e681a17aaca9440 : SUCCESS in 20m 25s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/4a07b6a2e8db49a282067443e1c46ef1 : SUCCESS in 13m 16s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/0f19b3d738144666bbae1a150a84e4c0 : SUCCESS in 17m 24s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3d882ed863ce4788a799b58ed5c91da8 : SUCCESS in 14m 50s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/41433f8cb9b347dfa7924a2f0ac43e83 : SUCCESS in 26m 42s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3d290224578e4420acb26ed0934fdae1 : SUCCESS in 19m 40s","accounts_in_message":[],"_revision_number":1},{"id":"ffae1b7de7d036e60af20fbc2095aca3d548290a","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-05-28 07:33:42.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2dbeb061f98421b4556d4244e84fec8665af4b7e","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-05-28 09:00:19.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d2e4b744cb4fae197eb6e99c96947cd2854df9a0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-28 09:01:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3ccbc72408a4c4f0d59a88dfae93a9c166bcb955","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-28 10:39:57.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3fcc42e514f34cfca3e1358c853811d1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1a4d97b57b6433c95f6fa93f0961d00 : SUCCESS in 5m 37s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/55b4a71a863f4ba0b9e52ab6dbd364fc : SUCCESS in 6m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/99bccad21d0649bba4bab1b913d01eb4 : SUCCESS in 4m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fbea0a7af257444e85e042b10434b51c : SUCCESS in 4m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a3371c7a095e4051a5f177eea6c107a1 : SUCCESS in 4m 43s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/1188233188694a85adfdd510153a12eb : SUCCESS in 5m 02s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f32aefe0f07549a8878afbd35f72e52d : SUCCESS in 12m 44s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/37e9ddd635204cf8a8a22ef308e0b5c5 : SUCCESS in 12m 18s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/12275a841a48416495f9eda9062a7d09 : SUCCESS in 11m 14s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7843db7cc62c44d7b7bd97cb8f89dab3 : SUCCESS in 14m 39s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/f132a3cd990446dbac87876c8b8bcc56 : SUCCESS in 20m 03s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d34df5e153904703a893759d881849e1 : SUCCESS in 16m 48s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/cfab0a8a760d426d98064bcc7ee1b515 : SUCCESS in 16m 03s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/73d6da121c014f3497fd828c17170d02 : SUCCESS in 17m 51s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/e8f51592aa744b54b3b4cefbf85022f0 : SUCCESS in 1h 06m 23s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/ec4b430aa286414ab0a620857fc99f53 : SUCCESS in 1h 05m 57s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/4170c82a1eb2465581f71de1385260d2 : FAILURE in 1h 05m 49s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e10d7bb8c73f4319a609551ff0e1b448 : SUCCESS in 1h 05m 25s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/c002cf21a45a400c806c983118e8d1f6 : SUCCESS in 1h 08m 09s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/c56e0cea9682446f9e51985781e1fb5d : SUCCESS in 1h 37m 31s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8c02354ef3ef42a4be74bc4ed0cf2cc1 : SUCCESS in 1h 34m 22s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/68e9b888a4804ffda1ce9a655b67b6df : SUCCESS in 26m 57s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/b64022531e9e4a60901326a45d474cce : SUCCESS in 15m 06s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/79bfa40ada6f409eaacf61144228a275 : SUCCESS in 22m 18s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/b87d77c4dcce4746971fffd75cd58a82 : SUCCESS in 26m 10s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/5bc0fd2ceab141d880bf35010571cd17 : SUCCESS in 35m 10s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3d3c82e05b214e08ba57ba97496e4f78 : SUCCESS in 23m 38s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/7af4846c320d4352acb14d5667ad7310 : SUCCESS in 15m 21s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/96b3e672849d44bb951bc02851bf6f6e : SUCCESS in 20m 15s","accounts_in_message":[],"_revision_number":1},{"id":"2e0d425a09efd8f591d2f89ff48b85737772f79c","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-05-28 10:54:40.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f34f1e7bd639bbecf7488f07fdcb2a20cc22bbf3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-28 12:33:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/caadb0be98ad407a9633d42a8138a7d7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ddc2dc3a82a4a168cf9baca28ec25fe : SUCCESS in 8m 30s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/f4e2c9e6fc8c4a208c0916c4728c0cf9 : SUCCESS in 6m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5d42cb3c165f45aba58ed30d0cf14f67 : SUCCESS in 5m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/466a08a270e94a37a18c9a0a985fda57 : SUCCESS in 4m 20s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/275f1dedca6446888f5c500a80f86563 : SUCCESS in 2m 25s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/9fe9c3ed99ad41969d44d866dd4a9c3b : SUCCESS in 6m 18s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0e359b1500c5410ab968d50c1b949bd6 : SUCCESS in 5m 53s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/90fafac67131422cb6386a6556b34c5f : SUCCESS in 5m 18s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/45464e4756ba4bbd996fb51e33132df9 : SUCCESS in 11m 55s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b070af159c624bdf9b06db9688bcb6a5 : SUCCESS in 11m 54s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/fd164b6019874ef7be4cdcb3bf97e8eb : SUCCESS in 13m 20s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/73284742ce7a470282a97351dd3e0ae6 : SUCCESS in 17m 21s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8bac5af0bbf348df8442c2eb947e34f6 : SUCCESS in 17m 36s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/20a544d66ead4dfc8c2cf175eda21f99 : SUCCESS in 14m 30s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/23234b04653d48c195ffeabd0f6b0986 : SUCCESS in 20m 01s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1c01c2e8fe7c4bdc9120f72cb83011b7 : SUCCESS in 19m 02s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/7808e9c456e840989f2fa4a14e74c526 : SUCCESS in 19m 11s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f3582f22ccbd46df8049503ee5618e22 : SUCCESS in 17m 04s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/dec4f6c2ffcc446b9bfc922a825c070b : SUCCESS in 1h 04m 01s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/d1f798e6c6c44e3b842e41462a1e801e : SUCCESS in 1h 05m 37s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/509b1b1e0a2840e3815c865d557dad65 : SUCCESS in 1h 07m 11s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/510e10c9b5314304b81ab3e3ccf946a2 : SUCCESS in 1h 01m 24s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/f721a36f95e24d2db7d8141c48cee7db : SUCCESS in 1h 04m 15s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/3f460e404e86408db5504fc1cad38928 : SUCCESS in 1h 35m 37s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/16dbd1fe6af84d6d91a02dbed99e3774 : SUCCESS in 1h 33m 37s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/8b79c07987d04b0eb24f421fc45bd604 : SUCCESS in 24m 44s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/fc55a594fe19489b8b30f6d811bbb8f2 : SUCCESS in 23m 11s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6d6a541fabfe448687f8f94cfe4502df : SUCCESS in 22m 50s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/34e0e8e7410a45bc8c111f9f591a4842 : SUCCESS in 21m 04s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/c89c479c38ef4e5fb7b9803747b84f34 : SUCCESS in 16m 56s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ac1e463a6b1b4f9cb57f65b9fd054204 : SUCCESS in 29m 25s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/c79339f77f514e54900704d4914d105c : SUCCESS in 15m 07s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a4ce18491c37495782a378427fd28c32 : SUCCESS in 19m 53s","accounts_in_message":[],"_revision_number":1},{"id":"60faa09a73095c6853f46f10fcdefc5078c0b1cd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-28 12:34:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"328b2b624555ffa59e77f93919b37b0f9a666747","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-28 14:10:05.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/536abb75bbab4e6d8d9160182584895a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48c894ffb8b74f38919bacf89441ef33 : SUCCESS in 9m 57s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/30aa5f691a9f4e048a4d786267d93e90 : SUCCESS in 6m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3bf8c95d62f24624b1415ffad6f6a5fe : SUCCESS in 3m 54s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/33dd10d0cc624a628120f6a095577ad1 : SUCCESS in 5m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/72adf1892c2a49448d32d18e8fbf8da7 : SUCCESS in 5m 54s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/731dea697e77456eaf99d12546ee7b62 : SUCCESS in 3m 44s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1e14964cd89045d09e9e30e9d9361a41 : SUCCESS in 9m 45s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7b3cba98f0494b8aaa4ef9d52adb7fa6 : SUCCESS in 6m 38s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/00623713975e4599b635c70ad9ca520d : SUCCESS in 12m 15s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/09ee1997861e4150ad13ce96766d022a : SUCCESS in 10m 33s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/ab2411e3a81043b4ae29ed95d82757c3 : SUCCESS in 15m 20s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5c7a22ca63bb4454a3bcd50e8e9e2f81 : SUCCESS in 13m 47s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/33c876537c3443e6b0fff140e64f3bce : SUCCESS in 11m 15s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/df1907ba2fcf46b0a283b46453ba8127 : SUCCESS in 19m 19s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/d8c15489f6eb43a4965f0635438893cf : SUCCESS in 59m 17s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/867a01ec502442d4b9c0b96ab23ca208 : SUCCESS in 1h 02m 05s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/e661edf9a1494db38429f1a3953fd739 : SUCCESS in 1h 09m 02s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cd69f5c33d284ab3b0521160d8690ef7 : SUCCESS in 1h 03m 47s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/3b38de01c9824a1bb5cf3244dc4a0136 : SUCCESS in 1h 06m 22s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/04cc8e69167043f5b2541cabc6de9807 : SUCCESS in 1h 35m 17s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/25a8072cc2654ecba6bb4cf94464da80 : SUCCESS in 1h 35m 19s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/b09c29b1c4044555b8a2217916a5cadb : SUCCESS in 16m 17s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/1605f0826e5346eeb7f598fd0d42a67d : SUCCESS in 23m 32s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/190b8e14379e4951ac236c771c2d18ed : SUCCESS in 23m 54s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/e072f29b52a8476184b551dac6c18349 : SUCCESS in 26m 32s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/7caaae893f0c460c91dd28d756762c2f : SUCCESS in 18m 10s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/394bc8948a724592b88989d74b84bd51 : SUCCESS in 23m 12s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/17657482278c421a8b6ea1520080c5ad : SUCCESS in 16m 14s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/56f01ae380204f008a8eb8822164978c : SUCCESS in 21m 54s","accounts_in_message":[],"_revision_number":1},{"id":"a6b9e723c0829d20c4dc0b723ec643df5af2d9cc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-28 14:10:06.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"fb2056f0743a043aaad51fcd4285511b40faf3c2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-28 14:12:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/851b6a4d693b4f5b86d78f08879b52a7\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ced5785f57f487b850b521fc761713b : SUCCESS in 47s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8bc07a347ed7480f96ba3a978276cc26 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3e3eafce9b7cb4ee6a47a852bc3a7e206da8bb68","revisions":{"3e3eafce9b7cb4ee6a47a852bc3a7e206da8bb68":{"kind":"REWORK","_number":1,"created":"2026-05-27 16:17:21.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/26/990326/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/26/990326/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/26/990326/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/26/990326/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/26/990326/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/26/990326/1"}}},"commit":{"parents":[{"commit":"86c97bacd214226084e8de5c7d075aaf94e48ec0","subject":"Merge \"CI: disable SLURP jobs on master\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/86c97bacd214226084e8de5c7d075aaf94e48ec0"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-05-27 16:16:53.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-05-27 16:17:13.000000000","tz":60},"subject":"Revert \"[release] Use OpenStack 2026.1 release\"","message":"Revert \"[release] Use OpenStack 2026.1 release\"\n\nThis reverts commit 758516fd05d46fec243b0047c3caefc8a92d03a2.\n\nChange-Id: Iece6b0d8373356632857d8f7f7d4ef023e54d52b\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3e3eafce9b7cb4ee6a47a852bc3a7e206da8bb68"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3e3eafce9b7cb4ee6a47a852bc3a7e206da8bb68"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
