)]}'
{"id":"openstack%2Fkayobe~990464","triplet_id":"openstack%2Fkayobe~master~Iecd35ce6ed73e3c536f9f34b27297010ea9c90d8","project":"openstack/kayobe","branch":"master","attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2026-08-29 13:46:52.000000000","reason":"\u003cGERRIT_ACCOUNT_15197\u003e replied on the change","reason_account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}}},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2026-08-29 13:46:52.000000000","reason":"\u003cGERRIT_ACCOUNT_15197\u003e replied on the change","reason_account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}}},"hashtags":[],"change_id":"Iecd35ce6ed73e3c536f9f34b27297010ea9c90d8","subject":"Add support for cisco-ios switches","status":"NEW","created":"2026-05-28 13:21:35.000000000","updated":"2026-08-29 13:46:52.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":4,"unresolved_comment_count":1,"work_in_progress":true,"has_review_started":true,"meta_rev_id":"5b26abf1791139586541def85a4d8854cf2f0fdd","_number":990464,"virtual_id_number":990464,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-05-28 17:38:32.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":-1,"date":"2026-05-28 15:44:40.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"reviewer_updates":[{"updated":"2026-05-28 15:41:04.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2026-05-28 17:38:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3c7460ce8d9459fc05e97a0ddd6dfede72dd20fc","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-05-28 13:21:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"33fd4772fcf5cc3dd0f3f388c9309f71466bda8f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-05-28 13:59:10.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"a054f736b227fcee6602b10ece8e27607e1980bf","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-05-28 14:34:57.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"d8cfb57ee06c54f869d76d2ca19d596bd71cb640","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-05-28 14:35:01.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nThe following files contain Git conflicts:\n* requirements.yml\n","accounts_in_message":[],"_revision_number":4},{"id":"ddddaa159f6ba51c8dc809de186e6a2a94bd884b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-05-28 14:35:19.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"f1b6377260981b3540c079c5a5a2204a886797e8","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-05-28 15:03:08.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"65644bbc0e7f35bc9e68c0c88227cf8c27e4712b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-05-28 15:39:19.000000000","message":"Patch Set 7: Published edit on patch set 6.","accounts_in_message":[],"_revision_number":7},{"id":"549fde4ba9454a30c6d65b90dfa847dbb4cc9cc3","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-05-28 15:39:23.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":7},{"id":"cfb099e8f3f5a8b6a04a41bb34a13c6e5065425c","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-05-28 15:41:04.000000000","message":"Patch Set 7: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"79997e74d87b66e94f754d321f8c661de1accdf1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-05-28 15:44:27.000000000","message":"Patch Set 8: Published edit on patch set 7.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":8},{"id":"c9786d692f6a1743d2e7ae4c03b29c1dfcd8aba2","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-05-28 15:44:31.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":8},{"id":"b48d322d22873a08143a08f82ded18efa0edc141","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-05-28 15:44:40.000000000","message":"Patch Set 8: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"97d95ade61d2b919c690de6679df4208449fc7c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-28 17:38:32.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3dec6090a788408b99fc56c523742153\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6ab5a75a6d14b64a35c169744071fde : SUCCESS in 5m 51s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/d7cb66d73d0e4eb3b11f8e92a3b8e5e9 : SUCCESS in 7m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5a7888ee95234a8fbc21e58497f6a319 : SUCCESS in 5m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/21624ca6ed7a45329095b10d9d798d6c : SUCCESS in 4m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6c375102ec0b4fb0907598836aaa8bb3 : SUCCESS in 4m 34s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/26887cac463a4e778fe7c141163b3ba7 : SUCCESS in 7m 51s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3424e02157aa41f2a07871b2e296e444 : SUCCESS in 6m 45s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7d0c990219a1476889a76332491b7874 : SUCCESS in 6m 39s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c79462ce049c46e1a34da04ac0c5d58f : SUCCESS in 12m 05s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/566905e703e24443964808ffb1379f47 : SUCCESS in 7m 55s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/6eeea991de2f4f1384711c3ce4a951d1 : SUCCESS in 13m 18s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/0fe166fe0316455c8b054742e5b1a87d : SUCCESS in 12m 12s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9d6cee395afe46b2ae681e50fa157490 : SUCCESS in 15m 10s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/35a10d2fe0c04136a6d45a464ae4ce6b : SUCCESS in 18m 30s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/c425fd5f323f4691a3600b81e390e377 : SUCCESS in 17m 06s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a3123c1bb0a740ada1cda0c4ffedacfa : SUCCESS in 14m 53s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/e1d5fe77d5f24a16af23a1b049709454 : SUCCESS in 13m 39s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1efe9e591d0c4a07a8491ccd63ded775 : SUCCESS in 18m 38s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/dda2252d4cbe48459cbe65f4c20c5507 : SUCCESS in 1h 05m 29s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/52341eb3d3204337aad5451188fd20f9 : SUCCESS in 1h 10m 32s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/d3b253fe1ab24d7daa0bd711f4d9f32a : FAILURE in 1h 09m 10s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e264e3bb49cb4ccd959937c1297d0fcb : SUCCESS in 1h 10m 07s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/93d8517292414825be9b31b2c544b120 : SUCCESS in 1h 09m 17s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/a4cbb473be024abc9dfcd10ed78fa6d8 : SUCCESS in 1h 42m 38s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1cc17aa745764317bf1e141fe8367b28 : SUCCESS in 1h 38m 53s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/f7bb5b01c0aa4ca0a7e48df6463b4d54 : SUCCESS in 14m 57s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/03517d04d66743f88b6450c40ae422e6 : SUCCESS in 16m 27s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/87b4df0e9eee42a590c96fde3ef7f145 : SUCCESS in 22m 14s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/4bd6f8c76ced4183a8f7ffc6ca8114b6 : SUCCESS in 16m 56s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/b116808fd6ef491b859d0c83352ec5ff : SUCCESS in 17m 24s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/553043b29694449cb1457c683137e081 : SUCCESS in 30m 57s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/62ce4c561df045b29a01ea6201368ef0 : SUCCESS in 18m 19s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b02e0d612cb845ffacb598587efedf03 : SUCCESS in 23m 38s","accounts_in_message":[],"_revision_number":8},{"id":"5b26abf1791139586541def85a4d8854cf2f0fdd","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-08-29 13:46:52.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"58a920edf9a6114f63554a8a76832fc98ecfbb18","revisions":{"95e86149142894b081b8ab6faff5e63e178033b8":{"kind":"REWORK","_number":1,"created":"2026-05-28 13:21:35.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/64/990464/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/64/990464/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/64/990464/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/64/990464/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/64/990464/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/64/990464/1"}}},"commit":{"parents":[{"commit":"86c97bacd214226084e8de5c7d075aaf94e48ec0","subject":"Merge \"CI: disable SLURP jobs on master\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/86c97bacd214226084e8de5c7d075aaf94e48ec0"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-05-28 13:21:21.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-05-28 13:21:32.000000000","tz":120},"subject":"Add support for cisco-ios switches","message":"Add support for cisco-ios switches\n\nChange-Id: Iecd35ce6ed73e3c536f9f34b27297010ea9c90d8\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/95e86149142894b081b8ab6faff5e63e178033b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/95e86149142894b081b8ab6faff5e63e178033b8"}]},"branch":"refs/heads/master"},"f4587e66d29d5d6b8c56e8c379a8982780d6ce52":{"kind":"REWORK","_number":2,"created":"2026-05-28 13:59:10.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/64/990464/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/64/990464/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/64/990464/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/64/990464/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/64/990464/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/64/990464/2"}}},"commit":{"parents":[{"commit":"86c97bacd214226084e8de5c7d075aaf94e48ec0","subject":"Merge \"CI: disable SLURP jobs on master\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/86c97bacd214226084e8de5c7d075aaf94e48ec0"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-05-28 13:21:21.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-05-28 13:59:10.000000000","tz":0},"subject":"Add support for cisco-ios switches","message":"Add support for cisco-ios switches\n\nChange-Id: Iecd35ce6ed73e3c536f9f34b27297010ea9c90d8\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f4587e66d29d5d6b8c56e8c379a8982780d6ce52"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f4587e66d29d5d6b8c56e8c379a8982780d6ce52"}]},"branch":"refs/heads/master"},"735a9539de113a9168216b0f4fa20f4d37befce0":{"kind":"REWORK","_number":3,"created":"2026-05-28 14:34:57.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/64/990464/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/64/990464/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/64/990464/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/64/990464/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/64/990464/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/64/990464/3"}}},"commit":{"parents":[{"commit":"86c97bacd214226084e8de5c7d075aaf94e48ec0","subject":"Merge \"CI: disable SLURP jobs on master\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/86c97bacd214226084e8de5c7d075aaf94e48ec0"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-05-28 13:21:21.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-05-28 14:34:54.000000000","tz":0},"subject":"Add support for cisco-ios switches","message":"Add support for cisco-ios switches\n\nChange-Id: Iecd35ce6ed73e3c536f9f34b27297010ea9c90d8\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/735a9539de113a9168216b0f4fa20f4d37befce0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/735a9539de113a9168216b0f4fa20f4d37befce0"}]},"branch":"refs/heads/master"},"49673925e0cec187bb7b59a9af2c69a42d345daf":{"kind":"REWORK","_number":4,"created":"2026-05-28 14:35:01.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/64/990464/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/64/990464/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/64/990464/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/64/990464/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/64/990464/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/64/990464/4"}}},"commit":{"parents":[{"commit":"3e3eafce9b7cb4ee6a47a852bc3a7e206da8bb68","subject":"Revert \"[release] Use OpenStack 2026.1 release\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3e3eafce9b7cb4ee6a47a852bc3a7e206da8bb68"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-05-28 13:21:21.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-05-28 14:35:01.000000000","tz":0},"subject":"Add support for cisco-ios switches","message":"Add support for cisco-ios switches\n\nChange-Id: Iecd35ce6ed73e3c536f9f34b27297010ea9c90d8\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/49673925e0cec187bb7b59a9af2c69a42d345daf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/49673925e0cec187bb7b59a9af2c69a42d345daf"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"ours":"735a9539de113a9168216b0f4fa20f4d37befce0","theirs":"3e3eafce9b7cb4ee6a47a852bc3a7e206da8bb68","no_base_reason":"HISTORIC_DATA_WITHOUT_BASE","contains_conflicts":true}},"2d4ff3f72dc699b96fb6369836550360206e90b0":{"kind":"REWORK","_number":5,"created":"2026-05-28 14:35:19.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/64/990464/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/64/990464/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/64/990464/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/64/990464/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/64/990464/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/64/990464/5"}}},"commit":{"parents":[{"commit":"3e3eafce9b7cb4ee6a47a852bc3a7e206da8bb68","subject":"Revert \"[release] Use OpenStack 2026.1 release\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3e3eafce9b7cb4ee6a47a852bc3a7e206da8bb68"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-05-28 13:21:21.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-05-28 14:35:19.000000000","tz":0},"subject":"Add support for cisco-ios switches","message":"Add support for cisco-ios switches\n\nChange-Id: Iecd35ce6ed73e3c536f9f34b27297010ea9c90d8\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2d4ff3f72dc699b96fb6369836550360206e90b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2d4ff3f72dc699b96fb6369836550360206e90b0"}]},"branch":"refs/heads/master"},"f672a50d897c620ae99243f2048356805f433efd":{"kind":"REWORK","_number":6,"created":"2026-05-28 15:03:08.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/64/990464/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/64/990464/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/64/990464/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/64/990464/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/64/990464/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/64/990464/6"}}},"commit":{"parents":[{"commit":"3e3eafce9b7cb4ee6a47a852bc3a7e206da8bb68","subject":"Revert \"[release] Use OpenStack 2026.1 release\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3e3eafce9b7cb4ee6a47a852bc3a7e206da8bb68"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-05-28 13:21:21.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-05-28 15:03:08.000000000","tz":0},"subject":"Add support for cisco-ios switches","message":"Add support for cisco-ios switches\n\nChange-Id: Iecd35ce6ed73e3c536f9f34b27297010ea9c90d8\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f672a50d897c620ae99243f2048356805f433efd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f672a50d897c620ae99243f2048356805f433efd"}]},"branch":"refs/heads/master"},"9ae90e3cfab2ca2e2b955982ed90aef6f33e338b":{"kind":"REWORK","_number":7,"created":"2026-05-28 15:39:19.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/64/990464/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/64/990464/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/64/990464/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/64/990464/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/64/990464/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/64/990464/7"}}},"commit":{"parents":[{"commit":"3e3eafce9b7cb4ee6a47a852bc3a7e206da8bb68","subject":"Revert \"[release] Use OpenStack 2026.1 release\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3e3eafce9b7cb4ee6a47a852bc3a7e206da8bb68"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-05-28 13:21:21.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-05-28 15:39:19.000000000","tz":0},"subject":"Add support for cisco-ios switches","message":"Add support for cisco-ios switches\n\nChange-Id: Iecd35ce6ed73e3c536f9f34b27297010ea9c90d8\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9ae90e3cfab2ca2e2b955982ed90aef6f33e338b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9ae90e3cfab2ca2e2b955982ed90aef6f33e338b"}]},"branch":"refs/heads/master"},"58a920edf9a6114f63554a8a76832fc98ecfbb18":{"kind":"REWORK","_number":8,"created":"2026-05-28 15:44:27.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/64/990464/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/64/990464/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/64/990464/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/64/990464/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/64/990464/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/64/990464/8"}}},"commit":{"parents":[{"commit":"3e3eafce9b7cb4ee6a47a852bc3a7e206da8bb68","subject":"Revert \"[release] Use OpenStack 2026.1 release\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3e3eafce9b7cb4ee6a47a852bc3a7e206da8bb68"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-05-28 13:21:21.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-05-28 15:44:27.000000000","tz":0},"subject":"Add support for cisco-ios switches","message":"Add support for cisco-ios switches\n\nChange-Id: Iecd35ce6ed73e3c536f9f34b27297010ea9c90d8\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/58a920edf9a6114f63554a8a76832fc98ecfbb18"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/58a920edf9a6114f63554a8a76832fc98ecfbb18"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Workflow\u003dMIN"],"failing_atoms":["label:Workflow\u003dMAX"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
