)]}'
{"id":"openstack%2Fkayobe~990474","triplet_id":"openstack%2Fkayobe~master~Ic4a7548e40cdb5c56f395c18b01bec2cfaca36f1","project":"openstack/kayobe","branch":"master","attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2026-09-22 11:59:08.000000000","reason":"\u003cGERRIT_ACCOUNT_23084\u003e replied on the change","reason_account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}}},"removed_from_attention_set":{},"hashtags":[],"change_id":"Ic4a7548e40cdb5c56f395c18b01bec2cfaca36f1","subject":"Add support for extra opts to ssh-keyscan","status":"NEW","created":"2026-05-28 13:54:22.000000000","updated":"2026-09-25 16:43:20.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"742eba8801882e0cbe62d94001b01730200f5a03","_number":990474,"virtual_id_number":990474,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-05-28 15:39:45.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-28 15:39:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-22 11:59:08.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"CC"},{"updated":"2026-09-25 16:43:20.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"CC"}],"messages":[{"id":"1b68ecaf629cf6a5e86058caa68327b4d6a0a82b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-05-28 13:54:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a804cf302177a6e26acc1f17c95e95a1e75c6d40","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-05-28 13:59:14.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"d4b3419c813f2b59900186b886ee5693d9e285ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-28 15:39:45.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8d8094f2a2374154a13f019ef6730a43\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/976d05a402134455992dc80be5fc7729 : SUCCESS in 5m 22s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/e4c78b77f56e415ab7760f8bb15dfb5b : SUCCESS in 4m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8987471d55814fcf8fe1718b56c7861a : SUCCESS in 5m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec7ba71740314bc89c31870899a71f94 : SUCCESS in 3m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/996ec683a1974b048b9fc3b8bd7607dd : SUCCESS in 4m 15s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/1dabbd858f224bcabe6b723b5ee94c3c : SUCCESS in 4m 42s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1467be751ec04e478fc665347302cba6 : SUCCESS in 5m 59s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c09b5eebcea64b72ad098bc87118bdd2 : SUCCESS in 3m 45s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f9f36ef050b348598068c5ca50353beb : SUCCESS in 12m 55s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/29601b1d7a0d4e5d9df8f7517649d4e0 : SUCCESS in 6m 43s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/beb82d9718b94508ab6c6c8d97cadc9f : SUCCESS in 16m 28s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/aa60a1e376b94e658d7e2e2979803a45 : SUCCESS in 12m 08s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/496c459e2e544a0e9e9cc3338fb46233 : SUCCESS in 15m 26s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/f5132d6c75bf4fb5ba4fbe9e5db3676c : SUCCESS in 15m 43s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/66f8b18c3f824954b3dec8b124f290e6 : SUCCESS in 15m 13s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d0771dfa905e47e1bf8e5a3a69ff60bd : SUCCESS in 19m 04s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/ca4c577f4e6542a6bd873c784afe58a3 : SUCCESS in 12m 27s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ae5625a284bb4fa598d207c1ed0fc574 : SUCCESS in 16m 19s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/830b706ca0304fe0a3072e65cdeffbb0 : SUCCESS in 1h 03m 56s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/5b8997fa537d4231a104d836f824ec64 : SUCCESS in 1h 04m 47s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/cd54e188b9874714866c1b470ee0bcb8 : SUCCESS in 1h 07m 17s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e6d52f9d1bd94dc78642f70b5a1d892f : SUCCESS in 1h 00m 20s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/ec45ecf00456406ab508ca983d7e1992 : SUCCESS in 1h 04m 22s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/bbbe355e6a5040d48a4bbb06870e10b1 : SUCCESS in 1h 35m 04s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9a3effa3e6f442ac9d11b788221e73a9 : SUCCESS in 1h 39m 01s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/7ede879ca8494968975477ae9de33e13 : SUCCESS in 20m 57s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/263352c0c5db46718bd52bb670e4a530 : SUCCESS in 17m 22s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1f0b6c2c9ee64797b819e3aa31da5e77 : SUCCESS in 21m 10s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/7631c9024ab6465597205b286c4fff47 : SUCCESS in 21m 10s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/13a37dcec68040929e688f34f56d911d : SUCCESS in 27m 16s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c6735c4099b5485795e125c743c64f5d : SUCCESS in 29m 35s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/7856d9979c4b42bf8335b82190c07639 : SUCCESS in 15m 27s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2f46760647374123946fc652297fd90b : SUCCESS in 20m 27s","accounts_in_message":[],"_revision_number":2},{"id":"3e4f7c0824a4528ff8ec561b30459ec13b64b2d1","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-09-22 11:59:08.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"742eba8801882e0cbe62d94001b01730200f5a03","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-09-25 16:43:20.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"24ae981a6735141908b36a100ec51043ace18b61","revisions":{"bf8f42c745f16458a58ae9b3aa5c72c3baa6e285":{"kind":"REWORK","_number":1,"created":"2026-05-28 13:54:22.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/74/990474/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/74/990474/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/990474/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/990474/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/990474/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/74/990474/1"}}},"commit":{"parents":[{"commit":"95e86149142894b081b8ab6faff5e63e178033b8","subject":"Add support for cisco-ios switches","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/95e86149142894b081b8ab6faff5e63e178033b8"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-05-28 13:52:35.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-05-28 13:54:08.000000000","tz":120},"subject":"Add support for extra opts to ssh-keyscan","message":"Add support for extra opts to ssh-keyscan\n\nChange-Id: Ic4a7548e40cdb5c56f395c18b01bec2cfaca36f1\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bf8f42c745f16458a58ae9b3aa5c72c3baa6e285"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bf8f42c745f16458a58ae9b3aa5c72c3baa6e285"}]},"branch":"refs/heads/master"},"24ae981a6735141908b36a100ec51043ace18b61":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-05-28 13:59:14.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/74/990474/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/74/990474/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/990474/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/990474/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/990474/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/74/990474/2"}}},"commit":{"parents":[{"commit":"f4587e66d29d5d6b8c56e8c379a8982780d6ce52","subject":"Add support for cisco-ios switches","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f4587e66d29d5d6b8c56e8c379a8982780d6ce52"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-05-28 13:52:35.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-05-28 13:59:14.000000000","tz":0},"subject":"Add support for extra opts to ssh-keyscan","message":"Add support for extra opts to ssh-keyscan\n\nChange-Id: Ic4a7548e40cdb5c56f395c18b01bec2cfaca36f1\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/24ae981a6735141908b36a100ec51043ace18b61"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/24ae981a6735141908b36a100ec51043ace18b61"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"ours":"bf8f42c745f16458a58ae9b3aa5c72c3baa6e285","theirs":"f4587e66d29d5d6b8c56e8c379a8982780d6ce52","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
