)]}'
{"id":"openstack%2Fkayobe~990577","triplet_id":"openstack%2Fkayobe~master~I4e78326c960019bb2d4037dda03a0d2d9b2ffab2","project":"openstack/kayobe","branch":"master","hashtags":[],"change_id":"I4e78326c960019bb2d4037dda03a0d2d9b2ffab2","subject":"Limit concurrency of baremetal node registration","status":"ABANDONED","created":"2026-05-28 21:33:44.000000000","updated":"2026-06-02 11:01:39.000000000","total_comment_count":1,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":false,"meta_rev_id":"a1d8a5a424d45e6aa7867c5ee8315a5d67c44ece","_number":990577,"virtual_id_number":990577,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-05-28 23:10:56.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"reviewer_updates":[{"updated":"2026-05-28 23:10:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"62919c8a0aef5341fc09828980b210a93dff2b62","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-05-28 21:33:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7f35bce1245b4ae209185773c1592eb4a75517b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-28 23:10:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a5a258586b4049c295de697e973dd722\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91266f3fda33489ab3ced627c87543ee : SUCCESS in 10m 13s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/f8269d8e83c94d9b82e3805e2cce0095 : SUCCESS in 7m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4b0145665bfd4ffb8a62565375558411 : SUCCESS in 4m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c915ee334eaa4a94a7cefbb026e916dc : SUCCESS in 4m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fe117600cb0240698f8abfc42fc1088e : SUCCESS in 4m 12s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/38672355973e406e837f3b53a71c6301 : SUCCESS in 8m 16s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/20cbe17b4acc4e2797c1c2987cf014df : SUCCESS in 3m 04s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/17bbd9a263e6433b93d95f8afc580d8a : SUCCESS in 6m 18s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/2104dcc87484409296a9798b4ea1dad1 : SUCCESS in 11m 53s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/758f1d427e964e0caa2b18cb60027556 : SUCCESS in 13m 19s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/f0c04625d52e40ffaff65a8f9f8eaf28 : SUCCESS in 10m 52s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/8347dd0dea4645609d588e6a8f799ca5 : SUCCESS in 11m 57s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/33a684fdb2d14a01804bd8c0c93f1eab : SUCCESS in 15m 19s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/111f703e57f7474abee22b262a3534b9 : SUCCESS in 10m 08s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/27d4b18cf39e4ec2ab7fc33d6ed6093c : SUCCESS in 15m 02s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/117b207207cc49f7b4bd9874677ee8f1 : SUCCESS in 15m 29s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/8e05e192955148898e25b973dc494660 : SUCCESS in 19m 15s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/83ce7c7bd1b2442abbcf23a968b6a703 : SUCCESS in 21m 14s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/5994a610517441e2a3786e669d7e45a5 : SUCCESS in 1h 03m 11s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/07ef43e858694fdbb0efb108806e141d : SUCCESS in 1h 03m 12s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/05677690ca214f18932bb768b7915409 : SUCCESS in 1h 07m 50s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2d2c80b88b604b4ca4d3952fb2fdcb14 : SUCCESS in 59m 34s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/01d12dc4d73e4c5784656ebdc53a85be : SUCCESS in 1h 04m 32s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/11a525d341a64d38b57a0796f9f0bcae : SUCCESS in 1h 34m 32s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9ba7598fb1cd431b8e289f749f93107d : SUCCESS in 1h 33m 23s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/9610054449484942ab7e9da35153be35 : SUCCESS in 23m 36s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/b2317f867d0b4f68a2d719d01b38e0ee : SUCCESS in 23m 21s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/61f938b05f6b4943aa6cbbfbbf31a8f3 : SUCCESS in 27m 22s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/5abe4cef703a46f0aa3dd87874b81178 : SUCCESS in 16m 51s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/018ba9164c44465eba389f6bcfb5c1b2 : SUCCESS in 16m 09s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a79778d081824ae8ac7f78eac1fd5d56 : SUCCESS in 28m 24s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/0c27db81afe34ef99d6183467c3458fb : SUCCESS in 14m 39s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0b4c4cf797124dbfaf73a9ee220e6459 : SUCCESS in 22m 58s","accounts_in_message":[],"_revision_number":1},{"id":"0b3eea12292e02ac479a3639ea0f82eaaf6ff08d","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-05-29 06:57:20.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"09f4d326e243884ba500cb17bbed4d59899b93d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-29 08:36:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e8815da922e4418baad5f1cf99d5beae\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e487b4097114d0faee58e5474ec74d3 : SUCCESS in 9m 15s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/1651eec6aff14d8b986d43948a703e57 : SUCCESS in 7m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b81f4a8a77c043b9bc0dbbfc1ccd7168 : SUCCESS in 3m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/85b4745925e746fea2f8783a2007bb2f : SUCCESS in 4m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/cb3dd2fe78794d03875962d4791ccf2e : SUCCESS in 5m 07s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ac66121976454933bf7859856af122de : SUCCESS in 4m 41s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/40a75e4c51e54099b1f231f5ae35cdb2 : SUCCESS in 3m 38s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b8ce7e72bb9c41b3bac201f37822db4b : SUCCESS in 6m 30s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/09fccc4b0a7440a6bc2704fb3db0fd1f : SUCCESS in 14m 40s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/664bdbe548264292bf3881efe3a8a309 : SUCCESS in 11m 52s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/32f6dd46b0c74422ba25fb6635c583da : SUCCESS in 13m 48s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/f268ed0d8fb54555b24434b0cea63eda : SUCCESS in 17m 24s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/61e49ed192344ee5ab96836ff55a7e49 : SUCCESS in 16m 58s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/b66d834141fd43aea7a71a2280b20834 : SUCCESS in 14m 58s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/e770df21ca6b48a693f2117a2ed2345a : SUCCESS in 21m 58s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/34155546d872428ab6d2c78211ae3c97 : SUCCESS in 14m 27s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/5d9b5ad4f6f2423b9614891d3a902625 : SUCCESS in 24m 13s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5a5ae32e02414d37a3674362bba3ea41 : SUCCESS in 23m 31s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/492fc315f4bb442db47085009754c6e9 : SUCCESS in 1h 05m 00s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/1093bc291faf47939e7129567ce92372 : SUCCESS in 1h 06m 29s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/8267da62f0bd4f59b89fcdca33f385d0 : SUCCESS in 1h 09m 31s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/764a94e5ab9547349fc2f75730fe95d1 : SUCCESS in 1h 06m 01s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/2bea884520834151b490e7ec478610b5 : SUCCESS in 1h 07m 39s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/68d8219322a244378c6d351f66c66073 : SUCCESS in 1h 37m 20s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/58bdf80a294542848d998ea709d86174 : SUCCESS in 1h 37m 17s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/cc91589a842344f694f1ad3dce210225 : SUCCESS in 22m 16s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/603689d168894e4fb40c4930fbdbe23f : SUCCESS in 25m 31s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b1c73d1c2ee94058b8959a743b28eb80 : SUCCESS in 16m 55s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/7019a614d0e644a1aced2133c50ae210 : SUCCESS in 16m 28s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/1ed14e918b3d4de48b9ff654ceda659a : SUCCESS in 19m 02s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7d873e2e9d5e486e8592f0f17110b01d : SUCCESS in 30m 11s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/f8184403694b49ea824e815cd20c9613 : SUCCESS in 18m 00s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/eb1259b80b434b60aa3d362644ff7fa6 : SUCCESS in 16m 17s","accounts_in_message":[],"_revision_number":1},{"id":"a1d8a5a424d45e6aa7867c5ee8315a5d67c44ece","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-06-02 11:01:39.000000000","message":"Abandoned\n\nhttps://review.opendev.org/c/openstack/kayobe/+/990736 is fixing the same issue","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1f17afc71bde56ece90541ee125f2b2b196beeb2","revisions":{"1f17afc71bde56ece90541ee125f2b2b196beeb2":{"kind":"REWORK","_number":1,"created":"2026-05-28 21:33:44.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/77/990577/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/77/990577/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/990577/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/990577/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/990577/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/77/990577/1"}}},"commit":{"parents":[{"commit":"3e3eafce9b7cb4ee6a47a852bc3a7e206da8bb68","subject":"Revert \"[release] Use OpenStack 2026.1 release\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3e3eafce9b7cb4ee6a47a852bc3a7e206da8bb68"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-05-28 21:22:26.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-05-28 21:24:36.000000000","tz":120},"subject":"Limit concurrency of baremetal node registration","message":"Limit concurrency of baremetal node registration\n\nWe are seeing frequent CI issues with the following error during\nbaremetal compute node registration:\n\n    [Errno 11] Resource temporarily unavailable: \u0027/home/stack/.cache/python-ironicclient/ironic-api-version.dbm\u0027\n\nThis may be caused by concurrent ironicclient invocations for two\nbaremetal nodes (tk0 and tk1).\n\nAdd throttle to the block to avoid this issue.\n\nChange-Id: I4e78326c960019bb2d4037dda03a0d2d9b2ffab2\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1f17afc71bde56ece90541ee125f2b2b196beeb2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1f17afc71bde56ece90541ee125f2b2b196beeb2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
