)]}'
{"id":"openstack%2Fkayobe~990578","triplet_id":"openstack%2Fkayobe~master~I7619c28574cba3760434302ed6946a9f4baf026c","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2026-06-11 20:32:05.000000000","reason":"Change was abandoned"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2026-06-11 20:32:05.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I7619c28574cba3760434302ed6946a9f4baf026c","subject":"DNM: Reapply \"[release] Use OpenStack 2026.1 release\"","status":"ABANDONED","created":"2026-05-28 21:48:54.000000000","updated":"2026-06-11 20:32:05.000000000","total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"2fd8b6690088cf444462ed7dfd4572ca07b28e81","_number":990578,"virtual_id_number":990578,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-06-02 11:58:41.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-28 23:24:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-02 07:24:06.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"CC"}],"messages":[{"id":"6080ad6b3508b2f184b4c3bd34f10958189aadbf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-05-28 21:48:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5fa46336d6b219021932516ddc35cb7fc07a3127","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-28 23:24:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/28c8b8a4d17c4506b3ef31f8d10e0478\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/801a626c31dc433eb93bddd042333252 : SUCCESS in 4m 59s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/952adcbadfbe424ab7d25c0728000741 : SUCCESS in 6m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ca961a9ee6be4da89050c98b49f11d71 : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a5565fe04474492d9393ec0348112b86 : SUCCESS in 4m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/656949b1db054262af19e2ba8dd2ceed : SUCCESS in 4m 07s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c0028ac63ff0463f975633024ee0f012 : SUCCESS in 7m 03s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5fd14ef5673445d4a6665a6784dd18e5 : SUCCESS in 5m 59s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d958bf8cec594d658092ad4b5bda8bad : SUCCESS in 4m 43s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b2474c3390e34309843f370af60e0e83 : SUCCESS in 6m 18s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a123688637524bfeb1038d9146fe054f : SUCCESS in 11m 53s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/204fa7e58e3840239e504b917304f5f6 : SUCCESS in 11m 13s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/f22e85599acf4c8c9ca27395a8c1e4e9 : SUCCESS in 11m 00s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6b5b2a99bb044058b42c40403cbb67c6 : SUCCESS in 16m 27s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/0a0a62a00b034b8d9240febc5f33e903 : SUCCESS in 15m 00s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/7383dda2d32b42a1bda30eacf3ae61b6 : SUCCESS in 14m 29s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ecf375b2ab864918b92be22ba33ff3db : SUCCESS in 19m 56s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/832d60c950c9454fb87908f0506f11ec : SUCCESS in 11m 17s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7132323167584a678462e5725c781deb : SUCCESS in 18m 14s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/509c6c9f76a84ddaaf3b1335ed3d74c6 : SUCCESS in 1h 01m 51s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/2e60311ecea847f18533fa9320b5f3ea : SUCCESS in 59m 51s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/4fa8940c199442feab2883dc27ca403d : SUCCESS in 1h 04m 59s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/10be1b7b92ee4f4b8ff0206cf5a74c5e : SUCCESS in 1h 02m 23s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/73c0187408cf455fb6d67f8c3e4afc44 : SUCCESS in 58m 59s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/81087403aaa54fb4827e85e9daca9324 : SUCCESS in 1h 34m 23s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/927fdb4aa7d64b069e7bdd27f9d6ef63 : SUCCESS in 1h 33m 23s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/6c295bada7a945cd83da349e1ac32022 : SUCCESS in 19m 14s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/7a7ab7168b7e40bc8d232ed76febd334 : SUCCESS in 19m 59s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/aed3f9d6d10946748706d3908ccf4528 : SUCCESS in 28m 34s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/76b939f82b8f4828866dec6887a341b6 : SUCCESS in 22m 05s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/799cbd9350294e8c812f8dc015b7c6da : SUCCESS in 24m 10s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/232587f4a3a94eb5a981f508489fb5d6 : SUCCESS in 28m 46s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/599936c3fab94110ad1bdaf4fc8c525e : SUCCESS in 19m 14s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a11ebf0a2ade47a0afea6a043a60ac9d : SUCCESS in 20m 04s","accounts_in_message":[],"_revision_number":1},{"id":"6b4028b632a6170bf972929b05c2c4912a0fc603","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-05-29 06:57:16.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a56d35c4c330a5c920ed7eccb550f35691414efa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-29 08:49:20.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2837df7659f44f0da52dfdbeacc00864\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95106a2951a14cd2ae06052e4fd7d99b : SUCCESS in 9m 06s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/b94c8f7cb8094765b95d72c8f1d04009 : SUCCESS in 6m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5e697aa8172d4abab84c1f7e697124e0 : SUCCESS in 5m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/00a49eec734a4471bd7aebbe89321712 : SUCCESS in 2m 56s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4385255410af498594c050aabd6e47e5 : SUCCESS in 3m 19s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/78190c7afc284f91b1b321e13d866a05 : SUCCESS in 6m 48s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f0f156ebeab04acea643b2590cde9a33 : SUCCESS in 6m 23s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/525f73b304fb47d8b5ab282b15fe4b1a : SUCCESS in 5m 32s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/81c5c4d3723a417e85d1bd6cb8075032 : SUCCESS in 19m 06s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e89aeda158e940b691025e48afa56012 : SUCCESS in 7m 45s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/2ef5354d9eac4570907d9a807dc6b268 : SUCCESS in 17m 50s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/559f7976e8ef463f89d3e176d0dbe820 : SUCCESS in 13m 55s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c989a056de584699992b6dfe26665d1d : SUCCESS in 13m 11s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/007ed1f63a9f4a47a565a73ed9a668a3 : SUCCESS in 12m 37s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/87278c1ea39b4472aedb988525d033b5 : SUCCESS in 21m 59s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6423e5c028664cc1b687525e7a648de0 : SUCCESS in 19m 06s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/58a7ec0e1c064dd2b1580db4f9c626d5 : SUCCESS in 15m 04s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4645be03c137435bbf918fc28d3ac19e : SUCCESS in 17m 24s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/bc1f172a7ea04f82a272bb437de00d42 : SUCCESS in 1h 07m 07s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/985e41ba4b6e48789551c9e7bab7bb2d : SUCCESS in 1h 21m 13s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/3567a693de3142beb39d2cdaa86e0645 : FAILURE in 1h 16m 46s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3104f9ad06a44daca1b622cadcef9eb9 : SUCCESS in 1h 08m 55s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/11530eac62ce4f66b0a54bf6b50958c8 : SUCCESS in 1h 21m 24s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/e39390b7386c45c5b83b5e7aa85e0357 : SUCCESS in 1h 46m 35s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b0d3b281eea640d2ac5439b9ac3f2894 : SUCCESS in 1h 50m 56s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/657b9cc545ce47a1b2ebe4a761caa388 : SUCCESS in 17m 32s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/56675d1ee1c9489ea0e7392e8de080a6 : SUCCESS in 17m 54s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bf47927317934551a76dd4ae3a8cc2ab : SUCCESS in 26m 11s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/12203bad53654b1f80d8bea12a45366f : SUCCESS in 23m 18s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/8f500d7bb4de422ea2a29e328e9e6c89 : SUCCESS in 25m 56s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/42de6bf4c50a41e49608e74a41646939 : SUCCESS in 27m 47s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/f284428d30234be8856a5238a814a73d : SUCCESS in 17m 57s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6bc3fd9676fb4b28985df84e1f204c97 : SUCCESS in 16m 28s","accounts_in_message":[],"_revision_number":1},{"id":"4bbbb04155926b2de1dc36490e13fcfbbba16435","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-06-02 06:36:13.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"7e8fda11d935069f5ec2cc70fa145fb634f8a3cf","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-06-02 07:24:06.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2e971b3ca220efdd0d307e541e3fe2490cf6e4eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-02 08:18:32.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ab1a8c4cf0a14e8dad547734023c6c56\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/135860a139384eed9d4e0374a1b2de9d : SUCCESS in 9m 18s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/8401c5ffc3c84662a962e2dbbee5b22d : SUCCESS in 6m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1641e5de5bc7432c86b2d3599a1c9243 : SUCCESS in 4m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a077280f53a46b39f9c4657eaed1a80 : SUCCESS in 4m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2745b782540547e38f1883f059d745fc : SUCCESS in 4m 01s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/8a1978013fb84e339426b97cc4adb1f9 : SUCCESS in 6m 06s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7891a512fea648b39676bb06478f1c3c : SUCCESS in 6m 16s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9d63bea01892460ca7032a6021347b3d : SUCCESS in 5m 14s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/063bbc9e8b744f8881f18ab8faf8c530 : SUCCESS in 15m 48s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/203c87c17e5e4b5cb0779b603d262c4d : SUCCESS in 7m 41s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/0bbb03058a714fcf9275bed2078bb84f : SUCCESS in 13m 01s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/8a8d3620dbca48579fa2ad1a05dd2bef : SUCCESS in 20m 46s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/64812febfef54290aa3050f23ee288f9 : SUCCESS in 12m 24s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/dd995e23ed614dccb84f3a882f3fb081 : SUCCESS in 19m 59s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/a5c40b66421f4601ad0de6e324d12697 : SUCCESS in 18m 16s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/25daacc2f45941bda0c90e85089d9aa3 : SUCCESS in 21m 20s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/61494290256b4d989e7f318c3e980b94 : SUCCESS in 11m 29s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0bca6a9fa2bc406fb516aa6e14562f79 : SUCCESS in 9m 27s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/e34221345f4543e9aeb54d81cd98b6e9 : SUCCESS in 1h 01m 36s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/33eb17e8c5c749f48c235fca328df7b9 : FAILURE in 20m 55s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/8391a34cfe3049baa3580d639c6102c2 : SUCCESS in 1h 08m 54s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/79141e442f3d49dbb80721a8695a9659 : SUCCESS in 1h 10m 00s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/db278e12473846f0a4de3cab418a1248 : SUCCESS in 1h 06m 07s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/43a4a581dafa43b8974d2da10f564deb : SUCCESS in 1h 36m 13s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/92f9a97f171a4255aff948ccad32ea33 : SUCCESS in 1h 41m 01s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/64134de6a689420b8df3328612e0d2c0 : SUCCESS in 22m 46s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/775f5447b200446fbeb0c3da287659e8 : SUCCESS in 16m 58s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0e13ded2aa38450e9a0c90001c86381b : SUCCESS in 27m 46s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/007d6ac2596b49faa3edac861d606cee : SUCCESS in 14m 49s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/3506909bb8e84f1dbbada4bba9be20bc : SUCCESS in 18m 29s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ab23ca77733448159eae54207b46719f : SUCCESS in 29m 33s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/fc875550a38247578ab1dfb6fe995135 : SUCCESS in 12m 48s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6ca6d9a4011b42cfb8f1801534385351 : SUCCESS in 21m 00s","accounts_in_message":[],"_revision_number":2},{"id":"69fcc94c5a88f0ca53e7fe627d06455f3243e785","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-06-02 08:58:19.000000000","message":"Patch Set 2: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3dc08117ec8de4d5843adbc7d18ea343d5426928","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-06-02 10:09:48.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n* Workflow-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"13099b105bb79f321b422bafce93d89b3b0076c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-02 11:58:41.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2690ebf40fb9490c977439d4fd09e697\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bed45065addc4983a025c54e537067df : SUCCESS in 10m 42s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/dbac828f3bf9495f9c3974854359b256 : SUCCESS in 7m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b8c95e5a33b24397ab92d6d67d4de8d0 : SUCCESS in 4m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed4ba446c0fa4aa2abfc888bbf273bd7 : SUCCESS in 2m 45s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c9c4db724f1d4e389fb989af549ef516 : SUCCESS in 2m 37s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/d2285104c2d142a29a84d1503101406a : SUCCESS in 6m 43s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/df5ebf31ee7147f7ae6204e3ee59bf10 : SUCCESS in 3m 17s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/05b0176c75cd4eb39bb02ee607f46f7d : SUCCESS in 6m 35s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/254aec538c0e464487a9393df584b7c5 : SUCCESS in 11m 26s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c41d6192be874d16864f39ef16d848c0 : SUCCESS in 11m 53s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/0a0b2712fcca4d8dbe609878577a0267 : SUCCESS in 12m 40s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/7cb373ea1ce3437c822151f98cd2f140 : SUCCESS in 22m 29s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0bfe92ced60b45388536ceb25d0fd8bf : SUCCESS in 10m 23s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/4f9298302acc49e1bea69d7b10e70c2c : SUCCESS in 18m 27s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/6aa1a5b6d96f40ec8f9b12e973680d1a : SUCCESS in 18m 14s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/78f77c54df1a4e278eb6973a867b85b1 : SUCCESS in 18m 09s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/a6a58270e63a49e99ae9864547a35a63 : SUCCESS in 24m 45s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3cad29b28af54fa387601310108a5655 : SUCCESS in 18m 12s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/4cfc26db898a4f86b0fe25769fda28d9 : SUCCESS in 1h 06m 24s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/00d83add89c94c0697d08d3c8ecdde19 : SUCCESS in 1h 07m 14s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/675768a800dc40e5b08ae471d7879acc : SUCCESS in 1h 03m 27s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/26cfd0b7669a444a941192c25aa462fc : SUCCESS in 1h 08m 47s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/01adca3a22794e929eb36e6c045ff44f : SUCCESS in 1h 04m 33s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/e0b622ed80e24b55917418bde4adab2e : SUCCESS in 1h 43m 23s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/34305ec3c2aa45d398feb6ce260bd206 : SUCCESS in 1h 47m 13s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/7a16ce004ca24c16831a2f7debe2da81 : SUCCESS in 15m 19s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/911f8821bf8147fd8e84e77bbe243901 : SUCCESS in 27m 30s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/24bcd1de53a04cc7afbf0031d14c5381 : SUCCESS in 22m 24s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/55212eeeb19d40dcb4fa1d59f44fb106 : SUCCESS in 24m 56s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/fedd88d5fd80455792b222b737346131 : SUCCESS in 34m 12s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8e098f7223644e50bb1134249f0bf906 : SUCCESS in 28m 32s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/2674d9163f824a608a5efeb89711c6fd : SUCCESS in 23m 48s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e70514722406470e8518131e5d9f3c43 : SUCCESS in 19m 45s","accounts_in_message":[],"_revision_number":3},{"id":"2fd8b6690088cf444462ed7dfd4572ca07b28e81","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-06-11 20:32:05.000000000","message":"Abandoned\n\nTesting only","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"9cb3a7724727b46d383b7e16527bac122cec6c49","revisions":{"386f3e520838053eca240f28ea5ba38f8f18451d":{"kind":"REWORK","_number":1,"created":"2026-05-28 21:48:54.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/78/990578/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/78/990578/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/78/990578/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/78/990578/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/78/990578/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/78/990578/1"}}},"commit":{"parents":[{"commit":"3e3eafce9b7cb4ee6a47a852bc3a7e206da8bb68","subject":"Revert \"[release] Use OpenStack 2026.1 release\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3e3eafce9b7cb4ee6a47a852bc3a7e206da8bb68"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-05-28 21:48:29.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-05-28 21:48:47.000000000","tz":120},"subject":"DNM: Revert to 2026.1 requirements","message":"DNM: Revert to 2026.1 requirements\n\nChange-Id: I7619c28574cba3760434302ed6946a9f4baf026c\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/386f3e520838053eca240f28ea5ba38f8f18451d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/386f3e520838053eca240f28ea5ba38f8f18451d"}]},"branch":"refs/heads/master"},"c90b8c513f2a26d4409e8161f10987eb6ed9f142":{"kind":"REWORK","_number":2,"created":"2026-06-02 06:36:13.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/78/990578/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/78/990578/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/78/990578/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/78/990578/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/78/990578/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/78/990578/2"}}},"commit":{"parents":[{"commit":"3e3eafce9b7cb4ee6a47a852bc3a7e206da8bb68","subject":"Revert \"[release] Use OpenStack 2026.1 release\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3e3eafce9b7cb4ee6a47a852bc3a7e206da8bb68"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-06-02 06:35:42.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-06-02 06:35:42.000000000","tz":120},"subject":"Reapply \"[release] Use OpenStack 2026.1 release\"","message":"Reapply \"[release] Use OpenStack 2026.1 release\"\n\nThis reverts commit 3e3eafce9b7cb4ee6a47a852bc3a7e206da8bb68.\n\nChange-Id: I7619c28574cba3760434302ed6946a9f4baf026c\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c90b8c513f2a26d4409e8161f10987eb6ed9f142"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c90b8c513f2a26d4409e8161f10987eb6ed9f142"}]},"branch":"refs/heads/master"},"9cb3a7724727b46d383b7e16527bac122cec6c49":{"kind":"REWORK","_number":3,"created":"2026-06-02 10:09:48.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/78/990578/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/78/990578/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/78/990578/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/78/990578/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/78/990578/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/78/990578/3"}}},"commit":{"parents":[{"commit":"3e3eafce9b7cb4ee6a47a852bc3a7e206da8bb68","subject":"Revert \"[release] Use OpenStack 2026.1 release\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3e3eafce9b7cb4ee6a47a852bc3a7e206da8bb68"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-06-02 10:09:24.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-06-02 10:09:24.000000000","tz":120},"subject":"DNM: Reapply \"[release] Use OpenStack 2026.1 release\"","message":"DNM: Reapply \"[release] Use OpenStack 2026.1 release\"\n\nThis reverts commit 3e3eafce9b7cb4ee6a47a852bc3a7e206da8bb68.\n\nChange-Id: I7619c28574cba3760434302ed6946a9f4baf026c\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9cb3a7724727b46d383b7e16527bac122cec6c49"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9cb3a7724727b46d383b7e16527bac122cec6c49"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
