)]}'
{"id":"openstack%2Fkayobe~990736","triplet_id":"openstack%2Fkayobe~master~I659847a02b3fdb672c7f9d38280c6145d483d774","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2026-06-02 12:36:06.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I659847a02b3fdb672c7f9d38280c6145d483d774","subject":"Use per-host XDG cache for baremetal registration","status":"MERGED","created":"2026-05-29 17:44:44.000000000","updated":"2026-06-02 12:50:13.000000000","submitted":"2026-06-02 12:36:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"990736","meta_rev_id":"3622e57f4c9a930caf95e4a0d1d84405389b9181","_number":990736,"virtual_id_number":990736,"owner":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-02 12:36:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":2,"date":"2026-06-02 10:54:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-06-02 11:01:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":2,"date":"2026-06-02 10:28:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2026-06-02 10:54:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2026-06-02 08:40:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-29 19:33:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-02 10:28:06.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2026-06-02 10:54:11.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2026-06-02 11:01:00.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"}],"messages":[{"id":"64cb7d9cf0daba081dc96622a9e7a3b612395e0e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-05-29 17:44:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4c6d135acb4558941c480423e169a3258eec3f3c","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-05-29 17:45:00.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"c3a33f05d6e69c68673d0796b0bf89a6f290decd","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-05-29 17:45:34.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9de2d2cbc8719b8b3c543c3847553e2fea249c5d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-29 19:33:45.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d4f1d9193c3746f78b7418550e624853\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4eaea5f7ca0c43e4977505ab1ade4ebc : SUCCESS in 9m 58s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/74f05112a8094f18b39df9a5a3ecacf6 : SUCCESS in 5m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c626832ba1854d74bf1409b1ed4d493b : SUCCESS in 4m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28c395164ace4041b51ea5916475c013 : SUCCESS in 3m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d7d02de5c1224465b2e850a83aacafbe : SUCCESS in 4m 14s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/541b1ae894104cd09b82ce1fcb41f97c : SUCCESS in 7m 51s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/25e6c805b98e49119005b8fc41d7726b : SUCCESS in 6m 41s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4412892e2c364fdca5e9ad1f60855ffb : SUCCESS in 4m 42s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8510b81ea4164f7bbbf892fdb2c1c965 : SUCCESS in 12m 06s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a83ba1ab7b9e4a8a9e7d57a68fcb9d39 : SUCCESS in 11m 15s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/c3c015d2c923495f8feda4401852bb47 : SUCCESS in 11m 14s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/50279043bc214cfd98fd0955abd4453c : SUCCESS in 16m 08s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/07ce57b79faa4fcf9bae7eb7441b28bd : SUCCESS in 16m 54s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/cabe373e632d4f15b4319e5a3e0b8454 : SUCCESS in 14m 21s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/5d1df552026a40c5ba87e5fcad515645 : SUCCESS in 14m 05s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a3a580ea84d64512a2051cc7db5c3f01 : SUCCESS in 19m 18s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/c4653b6ec3d24ecabdf38bba3bc4eae1 : SUCCESS in 25m 27s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e9053a00a9ee44d99cba92f3a5ff3086 : SUCCESS in 19m 24s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/2bc4e60e549c46698354b21dc7dbdaa5 : FAILURE in 1h 00m 35s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/9d691c11eea94ad5be8e861e5a6c3d95 : FAILURE in 1h 06m 24s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/2d5b0a025138413ea2dc68ccfad4b9d8 : FAILURE in 1h 05m 53s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/178a2d8e9b974465900a6d382adf98de : FAILURE in 1h 15m 02s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/6224f2b8663e4da285d96daddb969a43 : FAILURE in 1h 05m 10s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/6cb3994c83f24c00bd0e11fd732441b2 : SUCCESS in 1h 47m 40s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ccdbb1ab715b4d27ac289f6a1e67c9b8 : SUCCESS in 1h 36m 06s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/3bc8606837c345f4adced3248c8f061a : SUCCESS in 21m 41s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/e1e87e30edd345bea6d9ddfd487ebe06 : SUCCESS in 16m 12s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1eae85fa110642aabf7ec64db1fc4f58 : SUCCESS in 21m 47s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/2af48ae7f59b45cdaf76f8ee103915f9 : SUCCESS in 24m 37s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/3c82b40839bc4671bfc3b89d1f0fa32d : SUCCESS in 25m 02s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/00527eebe7614286971188388b624369 : SUCCESS in 26m 55s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/bba0ab422e034148becdf5c49911bedb : SUCCESS in 18m 38s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0bd0756c3b944b25916f8ad9b564c3ee : SUCCESS in 20m 05s","accounts_in_message":[],"_revision_number":1},{"id":"5fe78365389d7313746d3d6389e82236a5cb443b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-06-01 08:35:52.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"1bbcc5e983e0d51320b298fd6d89bf664cce9e2d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-01 10:08:55.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6242731335e54106969f7674a53df8c1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd4753503e2d438196b411dd523cba7a : SUCCESS in 4m 46s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/00a9bdfe6fd448f4b78135e26ab9d904 : SUCCESS in 4m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3f2e61460b0b49a28445686bcd2b5d9f : SUCCESS in 3m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ebf00b87d1f74032b6cc377a0950896e : SUCCESS in 5m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/94b0e668a6864275b7e046ea91f1e593 : SUCCESS in 3m 43s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/3bff0824995e431c83d5f0b058774139 : SUCCESS in 6m 37s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/07be4e06e18a4019849c0199c03eac0b : SUCCESS in 6m 11s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/645d9299e32c4b0187dd8e2b93a4679d : SUCCESS in 3m 39s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5ee9998aaf574b9ba02e3d00db8373f3 : SUCCESS in 6m 09s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/9da63f7cc0574c5ab97ae53de3ca9a55 : SUCCESS in 11m 40s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/e4f2af381f5e4bb3a64a60f360da4f45 : SUCCESS in 10m 42s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/b5627eced5cb4d4c96a4588d473d8bf8 : SUCCESS in 10m 46s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5c7e10c273164fd9b30d48357217e314 : SUCCESS in 16m 42s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/e4c0b53bbb514ee89bfc08a8476d711a : SUCCESS in 10m 34s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/decf0ca906ef42e5bbd0c10ae53f81e5 : SUCCESS in 21m 50s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/110d2928f4234b66953e11a2a7ead787 : SUCCESS in 20m 11s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/a021887ac1a44a829aaa12811ca8a63d : SUCCESS in 18m 08s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f3a69aef6cdc4ad0ac1b4e7d963f5f17 : SUCCESS in 17m 14s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/bd42ab1f09cc47c68a8ada2863abaf03 : SUCCESS in 1h 00m 38s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/37a2b48fccba4b2c9a4b694038c7e394 : SUCCESS in 57m 36s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/5ec5f7c36cc74cafb5849e8a073c9072 : SUCCESS in 59m 31s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/86acbe2f03054b37b43b75431d23efb9 : SUCCESS in 1h 01m 14s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/4d6ae593560947e88299987e4913340a : SUCCESS in 1h 02m 47s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/80285f29be4746068f24f9f6d7255929 : SUCCESS in 1h 28m 08s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9b908907ee0647548d52820459a8a404 : SUCCESS in 1h 31m 56s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/e49184a0a4b44cc9bc2580a6bb67501e : SUCCESS in 23m 36s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/8804cd2ec8fc4a739fd2ef8a811562b9 : SUCCESS in 25m 19s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3bef625c59c54971814be4041d66140b : SUCCESS in 21m 20s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/a99d39edb9f246a2acc088d78f01ec0d : SUCCESS in 20m 08s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/d701c3f2b47746b399b9531f017ec389 : SUCCESS in 18m 16s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/641eb059c73f46e99c7928742d503308 : SUCCESS in 26m 58s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/2b3b1c1a32a047c49bc82e0ebde0f949 : SUCCESS in 15m 52s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e58a9629b3b14f74b9dbad30e7ac3b1d : SUCCESS in 19m 07s","accounts_in_message":[],"_revision_number":2},{"id":"0bf023793a1d89dfc2406df5a6e6c45d6e6d05de","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-06-01 10:15:12.000000000","message":"Patch Set 2:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":2},{"id":"e293e49a7b666323d3f0c526b8cdb3b0581f2a2e","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-06-01 10:15:56.000000000","message":"Patch Set 2: Review-Priority+1","accounts_in_message":[],"_revision_number":2},{"id":"5eff76d922da9a6ca1ae0c850e72c7829b6c03b4","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-06-01 10:17:58.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"473cfe0d1cf9f78cb8d61e225261b7473e7a1764","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-01 10:51:51.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/69ab7782c2e4471b834329932faa6db3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d76ed8330fa429fbcdb21477e91e32d : SUCCESS in 8m 53s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/26401c48572c40c3b5843f7ce29bfb00 : SUCCESS in 5m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6a1ccd2b7c3d49d684f06a4e5f713883 : SUCCESS in 4m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a1d4d0cca754904bb5d8a0342867b19 : SUCCESS in 5m 12s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/34fb7358188a42deb5e143df035edd12 : SUCCESS in 4m 35s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a67a0889944d4faf9d567b79d8103c32 : SUCCESS in 7m 42s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1c0622eab97447058ba198b52620cc1d : SUCCESS in 6m 19s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b965d11f35884a749c5aeac3eab3d872 : SUCCESS in 5m 29s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d15f4f6b8b5f4126870197c585334533 : SUCCESS in 7m 10s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/0d647ab279dc47779bbf535673adb5ef : FAILURE in 11m 10s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/3f1d9279428f42fbb78a8d36f809d7f8 : FAILURE in 21m 58s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/3aef8f7f4145496f8ce2cbd40b39ba17 : FAILURE in 16m 43s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/15381a0ab4384596af3695119a24c8ad : FAILURE in 14m 50s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/27664d57cadb4a338b3bb58d5ab206e0 : FAILURE in 16m 53s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/44b7489059774c5e81cfed4e643be000 : SUCCESS in 13m 29s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/12e9fef8d1cc495abe8d7243f934e564 : SUCCESS in 19m 39s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/c491d94d39634a01baa236ef88def427 : SUCCESS in 25m 06s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/639e1e5cded74c719fb77a98fb72cda2 : FAILURE in 19m 11s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/fc81a9e5130147e685451eca549e3744 : FAILURE in 17m 12s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/04da60ad812e484f9271e64b85f43bc2 : FAILURE in 18m 45s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/001f48e143de44c1a42b533ed5c86134 : FAILURE in 22m 01s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bd9538dda445464ab38b2fb1e8bed993 : FAILURE in 17m 15s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/f3a3a279b2a74e848ff9718282f2b205 : FAILURE in 15m 33s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/9ffb7790023f444e90287850127c27e7 : FAILURE in 14m 41s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3edab1d99a9049d88020b75e18123f4b : FAILURE in 11m 13s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/94560812c0204bdf9face5e6d2083fd0 : FAILURE in 16m 17s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/95f0e4dd93d34733af8fa38edb994692 : FAILURE in 16m 22s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5209a14aa76a49e1b6d131f920f1ff70 : FAILURE in 17m 39s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/cb1f549870e64eae897e064f853e27a1 : FAILURE in 16m 04s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/8fbb9b9e2ddd4bd99739397873eb2e50 : FAILURE in 15m 14s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dc113ed145944afe9be93fc9d99a321f : FAILURE in 21m 03s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/8679f35835c24549a5ce1922a02822d2 : SUCCESS in 14m 31s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2609b9d5eed34972b7aee3cb29b516e0 : SUCCESS in 23m 44s","accounts_in_message":[],"_revision_number":2},{"id":"fb39e9c711c754de694df2c997554330a096d6a0","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-06-01 11:21:16.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"bd6cf827b0a0ccb4f072132c94120352607a8c7d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-01 12:53:53.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/01da21fe7b3741cd8058f7394a3bc528\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b5d913faffa4679a0a901af65927e6e : SUCCESS in 5m 37s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/2ff81c177b29499886e6646ffb760d01 : SUCCESS in 3m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/78875e76435643bca064fd8883e01321 : SUCCESS in 2m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/348a254d1ee2483d95371dd964ca2493 : SUCCESS in 2m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b57bd06ffd6446ce9ce70e4c257fb1ab : SUCCESS in 3m 01s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/5061671d500d4937b0593450060bb4ba : SUCCESS in 7m 11s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a05b1a675adf422c99abc225f5ddfd22 : SUCCESS in 2m 58s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/537850bcf2984fda941078e902187130 : SUCCESS in 3m 19s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/546b1671b01048ada5133f0e4b0af635 : SUCCESS in 14m 13s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6839c65590bb4bc0b647cf4d6379878e : SUCCESS in 12m 39s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/38d4927ce15d4e51aeefe7e5c2dbcd8b : SUCCESS in 11m 57s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/25d0a48d58ef49d8924787c8e9115ec5 : SUCCESS in 22m 21s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f4e2fb5781aa4e97a7c5bc3b742d138b : SUCCESS in 10m 07s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/8f8ef47996b64bb6aab28629e5dac5e9 : SUCCESS in 14m 39s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/626ae853d8f24f2a8786e6101c878265 : SUCCESS in 13m 57s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a66ed76443d54a15a0abade35298a095 : SUCCESS in 18m 35s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/a5648bc69bde4f289ba9b2dbd75c63f1 : SUCCESS in 10m 29s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f8119c31e31b471dbf1e4ba1f1ddec0d : SUCCESS in 19m 02s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/438c7fec9bde4b6bad4c9fb90950f1fa : SUCCESS in 57m 06s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/a7368e6899e74e2db7d2e7e01cbc9bd3 : SUCCESS in 59m 38s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/06ebc1d952fe4718923981c071b5c2b7 : SUCCESS in 59m 22s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/640339e76f0643a592e0f14a074dbe27 : SUCCESS in 1h 01m 09s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/3c6e817b01fc4dc1af9aee1eca9ea88a : SUCCESS in 58m 51s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/4f42ec5a16c24e18a3dbe41d39ecd66a : SUCCESS in 1h 31m 16s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/514c8a112c914c48be6a3ab55e3a65f0 : SUCCESS in 1h 31m 13s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/f115ee354fd147d5aefab952f254949f : SUCCESS in 19m 54s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/24ad99763854443290e75f3a9c02cc61 : SUCCESS in 19m 53s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/43607302dc41405f9d14edfd1ba86212 : SUCCESS in 20m 36s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/ad280dda7fbe476b85655a502f495655 : SUCCESS in 17m 55s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/b4b3b4b941ac454ab297e46043f9cafb : SUCCESS in 26m 26s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4adcd39877cf49d5b5683db04d3c349a : SUCCESS in 29m 51s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/2a403e18875a454fa99d698037a95eb9 : SUCCESS in 13m 47s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/54df1962ec5241838565d341f1303962 : SUCCESS in 17m 25s","accounts_in_message":[],"_revision_number":2},{"id":"6345751b867e8c28bb3d2eb635e6b232377c8652","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-06-01 13:22:01.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8f38978d439221ea99d9b868ecab2bb8be6a92c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-01 14:24:05.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/37c5c4cf310f4e3386f7700821d5ccfc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36c9df85a0be402696348e2a808030ed : SUCCESS in 8m 33s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/07d214c49035487d9b93991783ea481c : SUCCESS in 4m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1576200f3f994ddaad29b68e47ab5569 : SUCCESS in 5m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e27de796af4b4470af73e56ef17676a3 : SUCCESS in 3m 58s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c5f34b9cbd6441dc8cbb9c99d58eb8fa : SUCCESS in 3m 12s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c56d634a88f448388098af098d146a9d : SUCCESS in 9m 19s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d711c5795a634199b4402612570e8b11 : SUCCESS in 3m 45s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/015d1f4dd3054a56afe416b837508bee : SUCCESS in 8m 54s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d4d61ec7fc3340e9b77efe37d64b7ac2 : SUCCESS in 6m 56s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/cfbcc7deba3646ae8c469b538cc9a770 : SUCCESS in 14m 22s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/af6c3e89437e4ed9bc5a0348cd38ba76 : SUCCESS in 11m 10s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/7c6712c14d824a7b82a497380ee4cb3c : SUCCESS in 12m 16s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/50fa0a6e54454afa84da812276a2b371 : SUCCESS in 19m 35s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/990da09f7b0640a1b616bb972ff0b4ea : SUCCESS in 10m 15s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/0a1d494256914c9783190d64cc4113d4 : SUCCESS in 13m 03s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a476ec4d664146da90fbad625e3c4c20 : SUCCESS in 13m 06s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/a92dcc0ee55749198f93ea7a0a2bae0d : SUCCESS in 11m 27s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9f3bfa4fc368429a8939d67dfd9127b6 : SUCCESS in 16m 18s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/97a4320ef49243f4a723ddc406bc1358 : SUCCESS in 56m 45s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/5b0a8b99f9454ebe8deb0ff4ecfcaa5a : SUCCESS in 1h 00m 04s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/35bcf0deda784934b52bdfd4701efdc4 : SUCCESS in 57m 45s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/15f9a1a040dc4e7d8926a493a8be7632 : SUCCESS in 1h 01m 06s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/c89ac83377204b2d83e4c57e12492bdb : SUCCESS in 58m 43s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/16741b36ea6e4e829aee05c570f6beae : FAILURE in 51m 04s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5643661ba69c4a6eb5e34bebd4d8eeab : FAILURE in 55m 09s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/0a9cccf2f6a34f5ba446e3b00a342ff2 : SUCCESS in 19m 53s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/61f4f44c72da4ac68b50be2ee210f499 : SUCCESS in 15m 11s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2e173f0197804da899c525a0c0d99241 : SUCCESS in 15m 22s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/3b504938a4fb4255a731cfebcccae5ff : SUCCESS in 22m 36s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/4b63f6d06bd04249be729bd2f91d9072 : SUCCESS in 27m 59s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/17c8af97d7f842358aaeb8e62adb79de : SUCCESS in 17m 02s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/dcb24e7997174f19bea97b82980ecafa : SUCCESS in 20m 11s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ffc7ba407809468282f2d56683300816 : SUCCESS in 14m 02s","accounts_in_message":[],"_revision_number":2},{"id":"fc4b9605401b286089ecbecc15ffa1b9df8060fe","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-06-01 14:41:20.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a356c46ed73d4198797a17a5dbb1afe868e2be33","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-01 16:22:07.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/839667867c22427a843e89583fff57a5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/12c43aa81bbd44e1bcb283c8b89975d8 : SUCCESS in 8m 58s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/15b78e561da644099e76a77cea52fcf0 : SUCCESS in 7m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/463789533ce54b228194d5dc44df254b : SUCCESS in 4m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a5fc22487867464db3c7577412165a02 : SUCCESS in 4m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/19a99cfd7f9c4ef5b8999b3e44293f46 : SUCCESS in 6m 54s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/096e0d676c8e4266a42d87bc2e7f0749 : SUCCESS in 8m 10s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/078464790e23408b9263ae817a85805e : SUCCESS in 5m 54s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f1299cc60095486faf14d1c0ea4ec689 : SUCCESS in 5m 08s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/03a2e81549274f929abc7aeab85331b5 : SUCCESS in 13m 15s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/201b78c7be994ce8b53aaee8722ef70b : FAILURE in 4m 07s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/3f709e25845d4e99b0fec9f2e9bfe670 : SUCCESS in 21m 39s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/5310f8dc42f84696a72cde0044e43f3c : SUCCESS in 11m 11s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dc1837f0d04a41c1882282cfc0656863 : SUCCESS in 16m 39s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/ac06fbb1339c4d4cb1de6f0b65086b95 : SUCCESS in 16m 08s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/4bd1d3cebf854dd5a51a72a915224821 : SUCCESS in 12m 46s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2cdc550d1f474ba4af0f264b4697b791 : SUCCESS in 20m 25s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/7acbab4d7ff9413e9112385ae3701573 : SUCCESS in 10m 34s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4064d477a2bf478c9d033ffc933f0bc2 : SUCCESS in 18m 38s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/187b9f65b20f4bac8d0a8b0a8fcd99cf : SUCCESS in 58m 46s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/930de521c4f243d196859fc5af1b039b : SUCCESS in 59m 38s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/50eaf26f33a44386b63282b16b4efaa2 : SUCCESS in 1h 03m 34s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f65c793010b44859ab6d9a1d7e0d3f53 : SUCCESS in 1h 02m 54s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/8e117b06f577419fbfe51e33d23baf8a : SUCCESS in 1h 04m 06s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/3105b7dae5dd45a9bd23b69738e88766 : SUCCESS in 1h 30m 27s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/400f2749b4214f42b831971047d735c2 : SUCCESS in 1h 39m 35s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/8fe4c4a4b6a14bb9a5d4a16dc62d5dd6 : SUCCESS in 14m 20s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/5527f7c75e70452d8a7c69a973a346f6 : SUCCESS in 14m 23s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c2511035c3fb4dab83ecbc2266602bf7 : SUCCESS in 22m 56s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/c7bbed48244c4431b236a3ecb93f9b75 : SUCCESS in 22m 03s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/46b0191fa81d47d7852e07857ba1c52e : SUCCESS in 15m 33s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c4ea9e75724e471bb4284658440ca9ba : SUCCESS in 29m 07s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/9ef96f90500446c1a69c33d8e42e2e51 : SUCCESS in 14m 06s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/88321b6ae5e349bcb5531cb2fbdaf13f : SUCCESS in 22m 05s","accounts_in_message":[],"_revision_number":2},{"id":"d6b555f66ad350917cf32905bf102e6da0abb3e1","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-06-01 17:17:06.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"81e067dcd63eab6ab01885aa2f4fb600838f9992","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-01 18:50:50.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c586f311b6e64cc4a551693d49915f6e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8be0e1f936d4052bdc4d4b56093f92c : SUCCESS in 12m 01s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/12f64355722b4b0690427bd2096cc462 : SUCCESS in 4m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/28a139a81f674aa1baf3055eef369182 : SUCCESS in 3m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b43d2fb27744432e83bad421b8b13996 : SUCCESS in 3m 43s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/924b4b13b28042eeb6af33edf95a363b : SUCCESS in 2m 56s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/3c36700c0eda4e9193cda9e0508bc46e : SUCCESS in 8m 57s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/01656b2ddf7e4c608721496cf2d56730 : SUCCESS in 7m 04s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/dab56ce29696429d83aaf1fe2f894856 : SUCCESS in 7m 44s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4c009bb33a4d4200ab188488b92f6396 : SUCCESS in 11m 14s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/078e58e15be444ee845522b103121e0e : SUCCESS in 9m 30s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/96899182c6954f5aad6d9d66db37dadc : SUCCESS in 15m 15s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/89fff96d3bbe4047bf3c0a761a3b6d7d : SUCCESS in 15m 22s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e6067b29cf9048c18dd12466eae4a57c : SUCCESS in 15m 52s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/03d34f5c61dd4bf180cf97e79c7ec87a : SUCCESS in 12m 12s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/1866076df1ff4628a5c3974aba48b132 : SUCCESS in 13m 47s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c39cd21698274c42ad318e248891d477 : SUCCESS in 17m 48s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/d0184c9a4c8e44ce8cc826d203d70392 : SUCCESS in 24m 42s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4eaf4a7b2da84c46b2d6ce03efd4375f : SUCCESS in 22m 55s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/fbde62bc060c434f9bfc66c7f3a47b65 : SUCCESS in 58m 36s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/43a10d0c45e447a7a0f93b8eb09e2567 : SUCCESS in 59m 43s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/962ecc8731234322a84cf4aa71ae5ef0 : SUCCESS in 1h 02m 32s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9c1193d15d904dbcb0c2ba52f931f1d2 : SUCCESS in 1h 03m 54s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/507f4d9567844d98b709aef2632eebb8 : SUCCESS in 1h 04m 46s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/d25faea3944b494baf2ff049597df6d1 : SUCCESS in 1h 30m 17s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7561834e1fef4b3da4e04b934c63556d : SUCCESS in 1h 32m 44s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/11a37cdc787c4f8fa7aa34a4e3a461fb : SUCCESS in 14m 45s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/852c539a672d4b7fa4a9f4599fc13344 : SUCCESS in 16m 20s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e0bf1a8a434a40a2bb582468f0fc3825 : SUCCESS in 21m 35s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/77d0bcd922e542ca98fa4a831ee4ec59 : SUCCESS in 20m 41s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/139ee9c7fbbf4822b0e023a67ec07c0b : SUCCESS in 35m 38s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3faec6525c8442789ab42fd56b33df28 : SUCCESS in 29m 43s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/5a4b6ce6ef3d4c109e94549b544fdea9 : SUCCESS in 18m 42s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b8c16e886fbe4dc3b7d9bb305a5dcb98 : SUCCESS in 23m 54s","accounts_in_message":[],"_revision_number":2},{"id":"3fc5a7c8593e84e235462b64ea5540609f165397","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-06-02 08:40:39.000000000","message":"Uploaded patch set 3: Commit message was updated.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"**changekind:NO_CODE_CHANGE** OR **is:ANY**\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"20295cce9b1c20a6750f350a963a514b16a1f0fb","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-06-02 08:40:59.000000000","message":"Patch Set 3: Review-Priority+2","accounts_in_message":[],"_revision_number":3},{"id":"83b8c3f3a9db1b91729f50ce78dd690873bbba22","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-02 10:19:12.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f590a8c8bdc54ebe85940db999326f8e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71dcbde9a5aa45fba8bbf2c3e43b630d : SUCCESS in 8m 21s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/e4d3457dc62149e1adb0ffe8166d80ce : SUCCESS in 6m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8f1bc54c2d224194bb271ce77a28d9f6 : SUCCESS in 4m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e48bb5cf8c59406186dab97586ef37e7 : SUCCESS in 4m 34s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a0e9b7748a2941cd81c31762f0e674b5 : SUCCESS in 3m 59s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/d431e799412343a9a5ea9f0c7d33b9c2 : SUCCESS in 9m 43s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4a67ea8506844812b0b0f482d1829665 : SUCCESS in 6m 58s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/eba58a7287db407092d74a055d804145 : SUCCESS in 4m 52s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/00ba02e67c914967a9fa0befdf74ee9a : SUCCESS in 8m 46s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/bc55055da66646959fd1ea9cabccf57c : SUCCESS in 13m 58s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/4047fd75cb404d3c8ac57ef1a2ad0af0 : SUCCESS in 12m 06s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/bec5735116414721b9849e4394734888 : SUCCESS in 13m 43s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/950f2a0c86aa42b9b1a65906710402ef : SUCCESS in 15m 45s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/073d124411364d8a8ab107cf260beaa5 : SUCCESS in 14m 49s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/fa14242532cd43b680361664693fb416 : SUCCESS in 13m 03s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9f3585c8b2ab465dbecd9f1bf3252b10 : SUCCESS in 11m 49s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/74d0bc555f814d53aa6562747995f11a : SUCCESS in 18m 40s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bce4ae896b0543be8661a0e306a89b5d : SUCCESS in 17m 21s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/809be92abb134f19a444f41eedbab9b3 : SUCCESS in 57m 04s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/35c1bbaba256457f8f8e9e2a450ee4ce : SUCCESS in 1h 04m 37s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/079bd2de8ba34b8e82d19da94a110f2a : SUCCESS in 1h 00m 36s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f0472172fbf549dd85b1d201ddd64816 : SUCCESS in 1h 05m 51s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/7aaee623bab64fe9909e470d37f769ed : SUCCESS in 1h 05m 50s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/176274d9e0a34b0d93fb146b11223f4d : SUCCESS in 1h 31m 51s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c9ccb4f7431748c9b11fde6adecae706 : SUCCESS in 1h 37m 33s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/5e64a94c8cfc4213bbc70e74a6fc61a1 : SUCCESS in 28m 40s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/21f625b35f8940cebda60434567632be : SUCCESS in 27m 26s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dadc75274a4f45228c4b207ca14b4888 : SUCCESS in 17m 46s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/e6b942b698de437aa2c03a98a90b4e90 : SUCCESS in 21m 14s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/e6ab14d486dc43f7a6cdb28043252423 : SUCCESS in 27m 32s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/45f93e5044c646e6b6927eec757b2ff9 : SUCCESS in 20m 02s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/a74f23186f9c49a9a7b054b5e789d6c9 : SUCCESS in 16m 26s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c8ec10ae83ec447e9bb57c5fc0af9f7e : SUCCESS in 15m 40s","accounts_in_message":[],"_revision_number":3},{"id":"7856aee73853ca5cbfde95cc73dcb0c40cd0ae1a","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-06-02 10:28:06.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"97ba2e12bc0632aa3b12ab51c79eac904e7bafe4","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-06-02 10:54:11.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"908b47640b6bdcda55bbe4464c98c308cb57b66a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-02 10:54:34.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"6ad1a146336120531ef6ca19fe5401ebd87e1069","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-06-02 11:01:00.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"cf5bde678fba1302cd7f29a0dbe5c4b9a3eb242d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-02 12:36:05.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4250735ae9524e75b4e11e4813bd0b46\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f17aa1c6023d465183656f83fa668cb1 : SUCCESS in 5m 19s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/252db4fca7144418851936856e864b0e : SUCCESS in 6m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/777ac71fe57b4a478c99755de4b7e14d : SUCCESS in 2m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bd300a11e7244871b98aecd7a1cabb5c : SUCCESS in 4m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/08387d23d5ea4abf8fa1ee7e34c6ea6c : SUCCESS in 6m 15s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6320135c7c734fa98963c7e52f8eccc6 : SUCCESS in 5m 05s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/eedd0529ec2242acb0aa39c42236b453 : SUCCESS in 12m 11s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/afdc93cd56cc4dbfb97a97a13ce6ed17 : SUCCESS in 7m 40s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/d95886f4ae154424acf473e69769601f : SUCCESS in 12m 09s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/512d6834bda94b2783b4a03c7dbee345 : SUCCESS in 20m 09s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/25788e789c874853b643fc6320b0a00f : SUCCESS in 18m 51s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a8c7892d739e4aa7b744029afd0c0261 : SUCCESS in 19m 01s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/b20c22d5ebb642b898ef8ebb73122e50 : SUCCESS in 12m 33s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1a3cc0a483a84548839b2911e4564b7b : SUCCESS in 19m 23s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/758abbfefec54ccb8d0907f6a56ef315 : SUCCESS in 58m 42s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/95bca19ddb7745638a3d50bcab64fb8c : SUCCESS in 59m 08s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/d64a99bfc5b94097a30fcc967a075255 : SUCCESS in 1h 00m 59s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c3d1892c4ce74891a436dc1b57981360 : SUCCESS in 1h 02m 41s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/a01ad160548c445e9dd1270906863aef : SUCCESS in 1h 05m 10s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/c8fed2c50423449ea2a13c9054b22f2c : SUCCESS in 1h 40m 31s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/165b570d3fe04f0daccfc766435baa7b : SUCCESS in 1h 33m 36s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/a1460b47c84a43f882a9d9c758cb46c4 : SUCCESS in 24m 09s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/a65f5f4cfe6741059d3685829819bb2e : SUCCESS in 27m 55s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/613f9f921ba24f7a9019bfef6f79bcf8 : SUCCESS in 13m 45s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/ca5fbbb538344b9686932c6afcb9da20 : SUCCESS in 17m 11s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/352d5cf2076e4d40a7e4f39e9ba3518f : SUCCESS in 20m 34s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f1d4729a554c4c048b406966f2553796 : SUCCESS in 16m 06s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/53a6788676db43d9b17000670bff3262 : SUCCESS in 15m 03s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2ca0c7994dee4593a1db9f0869f45119 : SUCCESS in 20m 41s","accounts_in_message":[],"_revision_number":3},{"id":"ec18be9cf1369dc3944762659968ae3dec708a33","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-02 12:36:06.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"4b396867549687dfcc13c2a506cc9b2b58291e09","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-02 12:37:30.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e505bde81d5644d58555ec4f7f6e6c78\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6522c4fb85264ccdb046b009a516c661 : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":3},{"id":"3622e57f4c9a930caf95e4a0d1d84405389b9181","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-06-02 12:50:13.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"99fe27b86535a635e042b433a5537503bfc230db","revisions":{"1b752e70db4b037dc549afa89bb953f2ae02ef9b":{"kind":"REWORK","_number":1,"created":"2026-05-29 17:44:44.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/36/990736/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/36/990736/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/36/990736/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/36/990736/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/36/990736/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/36/990736/1"}}},"commit":{"parents":[{"commit":"3e3eafce9b7cb4ee6a47a852bc3a7e206da8bb68","subject":"Revert \"[release] Use OpenStack 2026.1 release\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3e3eafce9b7cb4ee6a47a852bc3a7e206da8bb68"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-05-29 17:43:12.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-05-29 17:44:37.000000000","tz":60},"subject":"Use per-host XDG cache for baremetal registration","message":"Use per-host XDG cache for baremetal registration\n\nFixes:\n\n  [Errno 11] Resource temporarily unavailable: \u0027\u0027/home/stack/.cache/python-ironicclient/ironic-api-version.dbm\n\nChange-Id: I659847a02b3fdb672c7f9d38280c6145d483d774\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1b752e70db4b037dc549afa89bb953f2ae02ef9b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1b752e70db4b037dc549afa89bb953f2ae02ef9b"}]},"branch":"refs/heads/master"},"6e7b6b462c1246c0153bc99011f447a52607dee9":{"kind":"REWORK","_number":2,"created":"2026-06-01 08:35:52.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/36/990736/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/36/990736/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/36/990736/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/36/990736/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/36/990736/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/36/990736/2"}}},"commit":{"parents":[{"commit":"3e3eafce9b7cb4ee6a47a852bc3a7e206da8bb68","subject":"Revert \"[release] Use OpenStack 2026.1 release\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3e3eafce9b7cb4ee6a47a852bc3a7e206da8bb68"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-05-29 17:43:12.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-06-01 08:35:44.000000000","tz":60},"subject":"Use per-host XDG cache for baremetal registration","message":"Use per-host XDG cache for baremetal registration\n\nFixes:\n\n  [Errno 11] Resource temporarily unavailable: \u0027\u0027/home/stack/.cache/python-ironicclient/ironic-api-version.dbm\n\nChange-Id: I659847a02b3fdb672c7f9d38280c6145d483d774\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6e7b6b462c1246c0153bc99011f447a52607dee9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6e7b6b462c1246c0153bc99011f447a52607dee9"}]},"branch":"refs/heads/master"},"99fe27b86535a635e042b433a5537503bfc230db":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2026-06-02 08:40:39.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/36/990736/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/36/990736/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/36/990736/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/36/990736/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/36/990736/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/36/990736/3"}}},"commit":{"parents":[{"commit":"3e3eafce9b7cb4ee6a47a852bc3a7e206da8bb68","subject":"Revert \"[release] Use OpenStack 2026.1 release\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3e3eafce9b7cb4ee6a47a852bc3a7e206da8bb68"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-05-29 17:43:12.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-06-02 08:34:47.000000000","tz":60},"subject":"Use per-host XDG cache for baremetal registration","message":"Use per-host XDG cache for baremetal registration\n\nConcurrent registration of baremetal compute nodes triggers:\n\n  [Errno 11] Resource temporarily unavailable\n\nUsing a seperate cache directory per node prevents these conflicts.\n\nThis is also being tracked as a python-ironicclient bug[1]. We can\nremove this workaround if this issue is resolved within the client.\n\n[1] https://bugs.launchpad.net/python-ironicclient/+bug/2154729\n\nCloses-Bug: #2154810\nChange-Id: I659847a02b3fdb672c7f9d38280c6145d483d774\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/99fe27b86535a635e042b433a5537503bfc230db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/99fe27b86535a635e042b433a5537503bfc230db"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
