)]}'
{"id":"openstack%2Fkayobe~990931","triplet_id":"openstack%2Fkayobe~master~I122af26524197e3d67bbf46a75ff632b836beabe","project":"openstack/kayobe","branch":"master","topic":"customContainersEdits","attention_set":{},"removed_from_attention_set":{"38288":{"account":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"last_update":"2026-06-02 19:02:45.000000000","reason":"Change was submitted"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2026-06-01 14:36:10.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I122af26524197e3d67bbf46a75ff632b836beabe","subject":"Fix naming of custom container variables","status":"MERGED","created":"2026-06-01 14:01:36.000000000","updated":"2026-06-02 19:04:45.000000000","submitted":"2026-06-02 19:02:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"990931-customContainersEdits","meta_rev_id":"a12b7b0c7ea6664456aad58230b41110afee21a4","_number":990931,"virtual_id_number":990931,"owner":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-02 19:02:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2026-06-01 14:54:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2026-06-01 14:40:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-06-01 14:36:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2026-06-01 14:54:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-01 14:25:40.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2026-06-01 14:40:29.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2026-06-01 14:54:40.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2026-06-01 15:00:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f8e55efcc1020fb6b250f71060221b78104b6cae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"date":"2026-06-01 14:01:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"212b120c24b73955f45855cd9439b0664f2db1c1","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-06-01 14:25:40.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"fcb332ef9c13323b7145000d208614f7734545af","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"date":"2026-06-01 14:31:43.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":2},{"id":"7a1ab693970d45750dee6a935ced568af3a826be","author":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"date":"2026-06-01 14:33:03.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"a59a6f311e0b33ba7ad5a1309fbd07be28ed6cd5","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-06-01 14:36:10.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"a59fe5ab1f220f5550d90eb0c8f069e71ba12802","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-06-01 14:40:29.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"d495bcfabb36ca1a5017d077f3860eacf9a28d43","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-06-01 14:54:40.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"cb25285bbcc72f115f761470ff5f3943031d1171","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-01 15:00:35.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3fad9a7789074d11bd05d5ddd75b15a9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1592568e64d94b0f9dcfa7d4f144ce42 : SUCCESS in 10m 53s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/ca19ae4b95d847e1823f2d2e9d402b48 : SUCCESS in 6m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1704cc59f01149fe8aeb7314f6b121f9 : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5cbe473122d4b39bda75dbe2fbbe9f0 : SUCCESS in 5m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6f2664e217c54c6eab7653ed30427e9d : SUCCESS in 2m 35s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/bd8ffa5b01dd4a228197ea07ed549be5 : SUCCESS in 7m 16s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9632296337344c2b939539b60ac6ddc9 : SUCCESS in 5m 47s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b042c781f5cd4a2496e6591bd5329b7a : SUCCESS in 5m 27s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/3e3663ea6b6442dd960760b6c503b885 : SUCCESS in 10m 42s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3a53e0b954644e979a896b07b51a34f6 : FAILURE in 3m 55s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/1bb4a6c7004349ef9083307b9a030922 : FAILURE in 16m 24s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/04a6b09ac90b4b5b913b10f0ab4e27fa : SUCCESS in 21m 31s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ba5c0b8ac0994356b234cb471bdd6516 : FAILURE in 9m 10s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/50f3ad4fc78b4c12802ea389372d7e0f : SUCCESS in 18m 08s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/5050a9fac7694aa2ab663f82d2b6c18d : SUCCESS in 22m 58s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3c0107f65482425e847d5877bd30f0e0 : SUCCESS in 12m 47s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/8c74024fa9f24220a1429ab6cc3b9946 : FAILURE in 11m 11s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2fd3b0ad88104fcdbd381831a595c03f : SUCCESS in 20m 36s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/ce0e9aae90db462e8cac04d09868cba5 : FAILURE in 16m 09s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/58fe077d39ad441eb34994c2f7fdf23f : FAILURE in 15m 56s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/013e7ab71c4a4d74aeb56c98a72c1b8d : FAILURE in 16m 07s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f504783b60704473b12e39a76ced9420 : FAILURE in 15m 50s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/d666e30fd23b4765a20d77998f55435d : FAILURE in 11m 53s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/224616adaf494dcea4c2a5346263463f : FAILURE in 10m 43s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d7696b7e3b444319abb7d1132dc9ce30 : FAILURE in 13m 17s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/2274a08060c14646b24089bb64528c97 : FAILURE in 12m 08s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/a715f71985174ee58ca38b921babb31e : FAILURE in 12m 14s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ee3917db984e40c88900404a441d5f21 : SUCCESS in 22m 07s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/bb5feb05f3a44ee0a91f1625c2336d73 : SUCCESS in 22m 03s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/1cacacbe558442a0a84326d99768f1de : FAILURE in 8m 35s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d1b6026cca444a5091d1459f5950bf10 : FAILURE in 12m 27s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/ae3392373aec4a299336a79290b70fac : SUCCESS in 13m 48s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dfa5e61663ff45948fa2c7bae0545a16 : SUCCESS in 20m 17s","accounts_in_message":[],"_revision_number":2},{"id":"bd8b31fccdd0fb65c26569e9e8eb886c8bf6b535","author":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"date":"2026-06-02 07:22:16.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"750edbd4e9a33dfb9ade6704b65247ce9be9c389","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-02 09:00:24.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e624c7a999a14450869dc2d772a7330e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd6ae186484c4ae48c39230fdbe568b4 : SUCCESS in 8m 53s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/b684ffd6c69d40e6b9ecb8a8437dc4b4 : SUCCESS in 3m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c011924c417047de8d930d553b8816d0 : SUCCESS in 5m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f8e6e77161ec4c0bba860005cbf9a046 : SUCCESS in 5m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1dcb4a588c324447bf21df2f25e0e495 : SUCCESS in 6m 04s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/2d9ec0af702c479ea92ce69d33058412 : SUCCESS in 5m 59s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1bc1d3f2b60f4faa919b71b6d5b221b4 : SUCCESS in 6m 04s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a437221663284d8f954c374133fb0808 : SUCCESS in 5m 06s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/51e9436b9e124c13ab831b5477f980ff : SUCCESS in 8m 18s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1d07ddeb5e7f4dbca447bfd3d2e0dd58 : SUCCESS in 11m 29s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/9768e1202c404962a972a722b64e42ec : SUCCESS in 11m 18s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/f491e3cac15b4fdcbf8eec78375d8ee5 : SUCCESS in 11m 16s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5913dcd5066a4ac18700ed4a742f8298 : SUCCESS in 18m 32s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/8df15c58265e483084468546b8e5017c : SUCCESS in 15m 17s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/59cfb807e4f54912a8005feacbd65d90 : SUCCESS in 19m 04s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c8df05d4fea1413e983807296c0e3cae : SUCCESS in 11m 54s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/c441d506899349aab4114b5e0693b8d1 : SUCCESS in 23m 58s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a37546a2205d4ec5b84fe861c15b8473 : SUCCESS in 17m 43s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/c0f81c764dc142c783dceee3ba7b3312 : SUCCESS in 1h 03m 32s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/1a81b9b12ed54d80bc123764cf6cc762 : SUCCESS in 1h 02m 40s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/981d5b5613a74613961952e94626395c : FAILURE in 1h 00m 33s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fb05837fdaf441448a70527081f7a8f6 : SUCCESS in 58m 00s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/4115e4805d8c4232877b52ae88b9739c : SUCCESS in 1h 05m 51s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/b15b1722c3ac4b6ebffee389a441fc65 : SUCCESS in 1h 30m 19s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fe5d3b98eabf4902a082cd169a69b5b6 : SUCCESS in 1h 35m 14s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/57f766e687f64c179f352519a29db366 : SUCCESS in 28m 23s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/b58bbc372bdf48cd9d16384aa6fce721 : SUCCESS in 23m 27s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1076aeac4fcc46118b97a0333437db56 : SUCCESS in 17m 43s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/cb00bfebde2c440cb01fbeedd1feffc1 : SUCCESS in 17m 18s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/9e4a46050c0a4378aa86a7b9ae0bb3e8 : SUCCESS in 35m 20s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/286342884e094e27ac87704bcbac0939 : SUCCESS in 15m 54s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/82afb4250db84339b2a2bb70d4811f2c : SUCCESS in 16m 05s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d5a2364d726a44efbee1212af4b0b78f : SUCCESS in 22m 54s","accounts_in_message":[],"_revision_number":2},{"id":"1d1fe85e20a6e536a88e4e55d93e3ccccde14603","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-06-02 12:51:30.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"622ec89801417c7ed3bc2321bd25573d0af0dcbe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-02 14:25:50.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6299aa1a15de4d19a2617ca5bfd3b65f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22395614636747f4a5967a0e7d7df827 : SUCCESS in 10m 40s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/b35fc03bade24d86aa6840903c0f3831 : SUCCESS in 8m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f7632d47247d495a9d77c37c3e9c4a89 : SUCCESS in 5m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58ff0ad95d7940999f29a313b84e31e5 : SUCCESS in 6m 05s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e89400c7562c4b01a5689284fbe079ec : SUCCESS in 4m 14s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f8dd04a401774359af793a28c53df896 : SUCCESS in 9m 09s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/45e23a6c610c41d88fd96923da304afd : SUCCESS in 6m 36s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ba32824709d347e39b03a95abc75e7e5 : SUCCESS in 6m 52s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/94015107d5174c8888b4e860ab76934e : SUCCESS in 13m 48s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/cf5be524ccb74d74bcc0c996319eb661 : SUCCESS in 13m 18s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/abfb4b8c1617477783743842992011c2 : SUCCESS in 13m 41s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/b215085b039f41f19514ea5380006e96 : SUCCESS in 22m 58s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8af1485d8cd0426a826f1edff6421b1f : SUCCESS in 18m 45s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/64107581306149fbb228cc20d0544964 : SUCCESS in 16m 45s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/aee122c6b83344a98d094eddad225877 : SUCCESS in 15m 00s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d8c48c8b783b469a81e83c5f4200ab76 : SUCCESS in 23m 16s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/93aff36218274ccc9c2db3f0b1f8f117 : SUCCESS in 23m 48s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b351c09d29334a18ab952978e67d0c4d : SUCCESS in 20m 57s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/b792ee0c83754e0485974dd30fa4ece3 : SUCCESS in 59m 13s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/5a4b826eb1d642d787f8329c0b85b0df : SUCCESS in 1h 00m 06s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/8084937c229549f799c728c3fe67c5dc : FAILURE in 16m 27s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/610f1173f1aa434bb30f10e2e4e340bd : SUCCESS in 1h 02m 57s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/8afa2ab03eff4ea198f79e7fbd2a5675 : SUCCESS in 1h 04m 36s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/6dc5c26c5e5a45b0be41630e4aac96f2 : SUCCESS in 1h 30m 10s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/68177cb4f4dd40a594701e9dc7f5e885 : SUCCESS in 1h 32m 32s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/7a00a574fd2548898dde562e1e0ee061 : SUCCESS in 25m 11s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/46ef730d1c3e4bddaea0a8e4afac3b28 : SUCCESS in 21m 48s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/345d2489e6d6443197a2915049b1816b : SUCCESS in 25m 36s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/6cd135825f2a4fd5987dd118ae24a07d : SUCCESS in 25m 43s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/0edeee04274447d6b25faec4d295b6fa : SUCCESS in 28m 36s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fbdf7a1283344d6f8af46233cd240658 : SUCCESS in 30m 33s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/8866ff83d4c44411a2bf533d28433b74 : SUCCESS in 13m 43s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/23e2a53e458b4d5ea4bb61005945a939 : SUCCESS in 18m 04s","accounts_in_message":[],"_revision_number":2},{"id":"f7dce2d3d86ce335e739d5af913f617d15d2e77b","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-06-02 15:33:37.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9379325b07fed299d27f720629b98f4438e90587","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-02 17:16:16.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2884dda0bdd346978cf98df684a3bd44\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2c01a73f16c408db670c9e92c1c698f : SUCCESS in 11m 20s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/b6075c601c984162aee5488e1e91ffbc : SUCCESS in 6m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c99a46d7dc364e2da2f57efb4ea6762f : SUCCESS in 5m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5db669840d1b47949227d69378f7223a : SUCCESS in 3m 43s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0dbffd60707740e3bd1213d77ddd15eb : SUCCESS in 4m 54s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a41fd2d6984440d0a2eb56b35f5dab26 : SUCCESS in 7m 48s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/302aa3d0de4445c48e9d56fb406cb1a8 : SUCCESS in 4m 33s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c8e697f530a643a2aaf0962fdd1d4b59 : SUCCESS in 7m 12s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0ffcb47c25cc4e76baaf1df1994086fb : SUCCESS in 14m 50s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/734b863286c845c688ff3c09908c2d79 : SUCCESS in 13m 15s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/10d49e65ea1746a6bd6291fe226f3dfc : SUCCESS in 13m 29s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/e0a5bc05b71240faa6d65ca494aee1e7 : SUCCESS in 18m 07s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3d9e53b8c4fd4a39abc45a34c9ace35a : SUCCESS in 9m 56s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/c635ce66ce3048d2a8afaac4c7663297 : SUCCESS in 18m 59s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/0af2243fedb845caa3c5a5d062902f0a : SUCCESS in 19m 45s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0848100ab2bd4fd6b3b9745e4715eb4d : SUCCESS in 14m 46s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/03c37ae289654fd4a6fbb998ff80b6d6 : SUCCESS in 21m 53s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/28e2d7f7b227433fa8367da6716aa84f : SUCCESS in 19m 49s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/16f4f8451d0c4c8daba9feda08280a55 : SUCCESS in 58m 33s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/b7299ce22470474c81e762bbbe197b8c : SUCCESS in 1h 03m 51s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/cb85a1d4cb2b48eabcf8beb8cf4ad7a6 : SUCCESS in 1h 04m 25s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2b86832e0b2a493896177e842ed8198a : SUCCESS in 1h 02m 04s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/222db5c40b03413886ad26e1bebdeac7 : SUCCESS in 1h 06m 29s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/2fa8def0506a4002b4ea5ea4671f1bb5 : SUCCESS in 1h 31m 41s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6db4042c9ec94c6c990f10312f42aaf2 : SUCCESS in 1h 34m 56s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/c696e9644a2f4149b2ecb167d1ecfdf1 : SUCCESS in 22m 46s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/ee121789bb1b4d3494764e75598e4532 : SUCCESS in 24m 46s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d00253fabc81440d9360740583752dd0 : SUCCESS in 14m 04s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/0a4f796aa027475eab292cbe526c13ea : SUCCESS in 17m 41s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/f3b788d00e2245e8ab4f2e010e2bf201 : SUCCESS in 29m 27s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3a846e6088ae41fd8e32c3939c1a7d93 : SUCCESS in 35m 40s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/9f26794fae824ad696de238e303b648b : SUCCESS in 15m 01s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2c3d11b54d0442428309685fe5fdd7dc : SUCCESS in 19m 44s","accounts_in_message":[],"_revision_number":2},{"id":"119f325378ee40277454a4ad3b82b030acf5395d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-02 17:17:04.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"8a61832b23cf2a290a09b63fbbee96b4f9373db8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-02 19:02:45.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cba56c4557b8417e86b7ebe6a99c948b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c91fd9ae8d5e45bfb2804b006c8fa782 : SUCCESS in 9m 57s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/41d1712e09d544c4a85619e37578ee3a : SUCCESS in 7m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f59500ed48e14beaae53c65ae3b604a0 : SUCCESS in 3m 43s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f796d0c78cc642a6a159b4301c609e4f : SUCCESS in 4m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a57e05b1155f40ddac77cef0960e8028 : SUCCESS in 3m 51s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/884d05e830ed4dbda44fee3ab9b89ca1 : SUCCESS in 4m 33s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/fbb01929fddb49e2813a5602fbccad3a : SUCCESS in 10m 26s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/caca7d5e447c48e2814d215b972a8063 : SUCCESS in 12m 22s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/297b2317c6e04695b25712539c7f7a5b : SUCCESS in 23m 17s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/17c7d8d60c8d43eb9ce0280b2195cb3a : SUCCESS in 13m 28s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/7118e602ec4d46248a7d643ecab421a5 : SUCCESS in 14m 33s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8b35b040573a4cedbcfde527e7c87986 : SUCCESS in 20m 24s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/84b7a91baca24872ab56a32a3b32e372 : SUCCESS in 14m 03s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8a05b8d4b63c416b9a7dfe2646f5bdb4 : SUCCESS in 17m 51s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/19e9097a52d3407496c4ec8c8e98c300 : SUCCESS in 57m 47s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/457137f2960544f3bb8ee77841e30507 : SUCCESS in 1h 00m 34s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/ac1d5815a7cc43eb87cbe088a0618874 : SUCCESS in 58m 03s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f663735eb27a4088a7201c3065d01e01 : SUCCESS in 1h 03m 05s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/77f951c5ba404694b5a6db0586b3a30e : SUCCESS in 1h 05m 49s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/df53ad658cc34de4b48833639dc96e3c : SUCCESS in 1h 43m 32s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dca11a5958314b8eb92eba3b1ed3d647 : SUCCESS in 1h 44m 42s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/3a55300ee7c94a08afa95d58c8411cbe : SUCCESS in 17m 41s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/fad282b124c847c4a75cf62dbec9dc7e : SUCCESS in 15m 05s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6f93bf0f58584e6f8f1ebb84bb615889 : SUCCESS in 15m 12s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/edc6d45afa134ca581efcbae95695cd8 : SUCCESS in 20m 36s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/d7ec8b00b3a348628d9024845fb60b17 : SUCCESS in 21m 22s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d01d09a3d4a24edd80bc6dfae8714878 : SUCCESS in 34m 59s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/06e24c5c47b941a883909782a2a1bef7 : SUCCESS in 19m 23s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d2da288b882a458cb97738b397fae6c5 : SUCCESS in 21m 07s","accounts_in_message":[],"_revision_number":2},{"id":"d5295302ab6c2811c07a5dad2d1f1d1ea45b806e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-02 19:02:45.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"a12b7b0c7ea6664456aad58230b41110afee21a4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-02 19:04:45.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4256425dd0584a93b3ad7b57f8dc0b77\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53c0794af3264b81b9d03968bc1c34b7 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"444db636625b3ffae1b9cf03045463787fa485a0","revisions":{"c55d4b61f133eeb0dbb37e20b2be12e2d7b1ab62":{"kind":"REWORK","_number":1,"created":"2026-06-01 14:01:36.000000000","uploader":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"ref":"refs/changes/31/990931/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/31/990931/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/31/990931/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/31/990931/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/31/990931/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/31/990931/1"}}},"commit":{"parents":[{"commit":"3e3eafce9b7cb4ee6a47a852bc3a7e206da8bb68","subject":"Revert \"[release] Use OpenStack 2026.1 release\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3e3eafce9b7cb4ee6a47a852bc3a7e206da8bb68"}]}],"author":{"name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","date":"2026-06-01 13:38:13.000000000","tz":60},"committer":{"name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","date":"2026-06-01 13:45:16.000000000","tz":60},"subject":"Fix naming of custom container variables","message":"Fix naming of custom container variables\n\nUpdates the custom container example in ansible-control.yml\nto the correct variable, and updates a description in\ninfra-vms.yml to reference the correct container.\n\nCloses-Bug: #2154741\nChange-Id: I122af26524197e3d67bbf46a75ff632b836beabe\nSigned-off-by: Leonie Chamberlin-Medd \u003cleonie@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c55d4b61f133eeb0dbb37e20b2be12e2d7b1ab62"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c55d4b61f133eeb0dbb37e20b2be12e2d7b1ab62"}]},"branch":"refs/heads/master"},"444db636625b3ffae1b9cf03045463787fa485a0":{"kind":"REWORK","_number":2,"created":"2026-06-01 14:31:43.000000000","uploader":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"ref":"refs/changes/31/990931/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/31/990931/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/31/990931/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/31/990931/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/31/990931/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/31/990931/2"}}},"commit":{"parents":[{"commit":"3e3eafce9b7cb4ee6a47a852bc3a7e206da8bb68","subject":"Revert \"[release] Use OpenStack 2026.1 release\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3e3eafce9b7cb4ee6a47a852bc3a7e206da8bb68"}]}],"author":{"name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","date":"2026-06-01 13:38:13.000000000","tz":60},"committer":{"name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","date":"2026-06-01 14:31:29.000000000","tz":60},"subject":"Fix naming of custom container variables","message":"Fix naming of custom container variables\n\nUpdates the custom container example in ansible-control.yml\nto the correct variable, and updates a description in\ninfra-vms.yml to reference the correct container.\n\nCloses-Bug: #2154741\nChange-Id: I122af26524197e3d67bbf46a75ff632b836beabe\nSigned-off-by: Leonie Chamberlin-Medd \u003cleonie@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/444db636625b3ffae1b9cf03045463787fa485a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/444db636625b3ffae1b9cf03045463787fa485a0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
