)]}'
{"id":"openstack%2Fkayobe~991480","triplet_id":"openstack%2Fkayobe~stable%2F2025.2~I3618a2b1050b69a031a5d09e30269fcfbddeff8d","project":"openstack/kayobe","branch":"stable/2025.2","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2026-06-04 17:00:52.000000000","reason":"Change was submitted"},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2026-06-04 17:00:52.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2026-06-04 17:00:52.000000000","reason":"Change was submitted"},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2026-06-04 17:00:52.000000000","reason":"Change was submitted"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2026-06-04 13:03:58.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I3618a2b1050b69a031a5d09e30269fcfbddeff8d","subject":"Adds support for custom watcher configuration files","status":"MERGED","created":"2026-06-03 17:18:02.000000000","updated":"2026-06-04 17:02:00.000000000","submitted":"2026-06-04 17:00:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"991480","meta_rev_id":"a865c91e19be5b4f3f9ca9a4d79b3879c56e47ee","_number":991480,"virtual_id_number":991480,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-04 17:00:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"recommended":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":1,"date":"2026-06-03 17:28:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-06-04 13:03:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":2,"date":"2026-06-04 14:19:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":1,"date":"2026-06-04 14:19:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-03 17:18:02.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2026-06-03 17:28:04.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2026-06-03 17:28:04.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2026-06-03 17:28:04.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2026-06-03 18:49:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-04 14:19:21.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"}],"messages":[{"id":"153e71627413dfe6c9c6a56bd92a6747ef9fd2c2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-06-03 17:18:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"503929d6be0fe96373da2650eaae8228adc0dfa2","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-06-03 17:28:09.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"c22d1a5ca99a059348605b07a9190853521e2a81","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-03 18:49:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7ef5975115c14869bf2e14638f34089c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02c0678167834c62bfe72faa3bd5fd5e : SUCCESS in 9m 09s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/2e1cdbeb2e724939957dfa20cb59ab11 : SUCCESS in 7m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d4084311d1294c0d836629832deeba6e : SUCCESS in 6m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a5c6f28dbb441c89a1a283377c0e404 : SUCCESS in 3m 06s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/811ab32745a640468ab4550ae3d8485a : SUCCESS in 5m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e421476b65094cf38b2d2fbfc0fd0daa : SUCCESS in 8m 18s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5ed27bc95b3b4148be152d9d16a53c15 : SUCCESS in 7m 59s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/fb3588fdf0e84d9dabaab71f7c917641 : SUCCESS in 4m 11s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/bc991a7ad129420dbe3401027de91b82 : SUCCESS in 7m 42s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/058ee9ce60794ff5aa0a74be9c2039e7 : SUCCESS in 11m 58s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/50e535233eef4176850267e5a79fe7a3 : SUCCESS in 9m 25s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e6985e40b7a74d5ab63efe435f02aa09 : SUCCESS in 16m 37s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/3d22727c49b24bf082a69121561ab9a1 : SUCCESS in 10m 49s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2fc856fcfa874afbbccfe23e9fd85654 : SUCCESS in 9m 36s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/b5af689ff0894ab09dcecffb5dbcc792 : SUCCESS in 56m 33s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/6730bda1e4f64d1b8584fcdb0e6e150b : SUCCESS in 59m 26s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/f3068b50e1354e1a85b94cad60d8a274 : SUCCESS in 57m 05s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d6cac6dd837449e09d2480e979c3d0ab : SUCCESS in 1h 00m 12s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/b3e60c605422438eb3d2a18212f7362d : SUCCESS in 1h 00m 49s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/b23881b24887494f99f40f5f4f501539 : SUCCESS in 1h 24m 39s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b3ebd8e9ecde47ca8eeb2ba684c8ce73 : SUCCESS in 1h 26m 54s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/52a2b160589f4bc0a84500fcfaf45775 : SUCCESS in 11m 12s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/bb8fb9abde36457989b036d5c38bf349 : SUCCESS in 11m 21s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0a79d534de7b49d098165d1cefb3ad77 : SUCCESS in 18m 02s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/8d4789f7d80c4b71adf6c63c7b1b0f02 : SUCCESS in 16m 58s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/32f717999c794655b4dfb37edf36b016 : SUCCESS in 19m 29s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/90d78d059df145edb9b83694fe8b595b : SUCCESS in 29m 20s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/c9c2abd6d3a74558a714507a67e45f96 : SUCCESS in 14m 32s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b78f06dd0a954256bd4d549aab9474b1 : SUCCESS in 14m 32s","accounts_in_message":[],"_revision_number":1},{"id":"4dc7414cfbf45795eb3c2e028b893154292e5f50","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-06-04 13:03:58.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"bc22a3a1d4c9c26a45708c6fe90e306241ab4cfc","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-06-04 14:19:21.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c943b630572f891af6c57b4d4e57128b09c20be6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-04 14:19:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4604830d2762faf614842bc2dfbff56e403cb404","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-04 17:00:52.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8836e9466c89493fbd717f623f109344\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69e22ae28d6a44cb8ef6577244a4c4b0 : SUCCESS in 9m 52s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/c88b3e107c614ecfb8de5beee4a52689 : SUCCESS in 8m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8efc096e1274497a9683ea2797796297 : SUCCESS in 4m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7e074f14716d402399c8ffb8b269bfee : SUCCESS in 2m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4f65d49d8b734c93aefa030f665f2049 : SUCCESS in 8m 03s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/95416ba8d3e4475484eca3c243984dc3 : SUCCESS in 5m 25s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/06a7d3828f4c40639574225c44c2e794 : SUCCESS in 16m 10s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/917e8c4b76ae4bfaa5f007b416a51b6e : SUCCESS in 11m 40s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/1a8c2197f6964e2882c42c35dd37b898 : SUCCESS in 18m 09s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/60c68487100a41bc9d70fd20ae43ea24 : SUCCESS in 8m 38s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/4a268b29fd554725b74298c7e546f836 : SUCCESS in 14m 57s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1db66d95d003431ca7c89714bdf22fcd : SUCCESS in 11m 36s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/02912296e6cd4df78383df36c07e4f01 : SUCCESS in 53m 36s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/f743237cfabc4738b6fac72d846230fb : SUCCESS in 54m 29s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/145e5eefc57d40b2bc41e8ca74ba9b18 : SUCCESS in 53m 41s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f805b257d1394b61b5d038d6609ae7bb : SUCCESS in 57m 22s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/5beb0ede08dd4a93a40a9a92a1ee5d0f : SUCCESS in 57m 35s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/71b02768c1534193a81302ba417db379 : SUCCESS in 1h 20m 27s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5ef13fac20fe414b87294c95f2e212c5 : SUCCESS in 1h 22m 16s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/6e5bc4e1de0f44e89486c54152f3c3db : SUCCESS in 22m 48s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/67bcdd7881af4b79b2c30abc1ffb95f8 : SUCCESS in 15m 49s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f269588e3e3d47ea87e74a0ca39baf2a : SUCCESS in 11m 58s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/610b5c41d45f44dea924d4ca2bff38dd : SUCCESS in 16m 38s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/7ecec46709e24208bfaff7c800cc749e : SUCCESS in 24m 12s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5740422ebde24853935885841ff38aab : SUCCESS in 30m 48s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/a336b27367cc44a4b21de1e0ec3b83cf : SUCCESS in 10m 23s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7254b8025236424aa06ebdd5c3236197 : SUCCESS in 13m 26s","accounts_in_message":[],"_revision_number":1},{"id":"d3ec1a63d710fcba616d9d6cba4597cb15469189","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-04 17:00:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"a865c91e19be5b4f3f9ca9a4d79b3879c56e47ee","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-04 17:02:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b4cc71eb2e9f4565b5664384527509dc\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2fd85d99a40481e9abf661a653ffb66 : SUCCESS in 53s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3b5c5792411b4c298b9999f1c2547549 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5ea954450a421d6d97a83e3a6b8c04e20f11d8d7","revisions":{"5ea954450a421d6d97a83e3a6b8c04e20f11d8d7":{"kind":"REWORK","_number":1,"created":"2026-06-03 17:18:02.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/80/991480/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/80/991480/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/80/991480/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/80/991480/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/80/991480/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/80/991480/1"}}},"commit":{"parents":[{"commit":"f1cc355af18c6032994bb67f86c0d324789d92ba","subject":"Merge \"Fix issue with systemd-networkd-wait-online when no IP\" into stable/2025.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f1cc355af18c6032994bb67f86c0d324789d92ba"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-23 11:24:42.000000000","tz":60},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2026-06-03 17:17:08.000000000","tz":180},"subject":"Adds support for custom watcher configuration files","message":"Adds support for custom watcher configuration files\n\nCloses-Bug: #2148457\nChange-Id: I3618a2b1050b69a031a5d09e30269fcfbddeff8d\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n(cherry picked from commit 31ba46b4ce0d00bbf3ad501c0ed385efedd98641)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5ea954450a421d6d97a83e3a6b8c04e20f11d8d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5ea954450a421d6d97a83e3a6b8c04e20f11d8d7"}]},"branch":"refs/heads/stable/2025.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
