)]}'
{"id":"openstack%2Fkayobe~991505","triplet_id":"openstack%2Fkayobe~master~Iad22b28c39a023d26f89b753af73dd730d000cea","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2026-06-04 10:18:25.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iad22b28c39a023d26f89b753af73dd730d000cea","subject":"CI: Stop using image from Docker Hub","status":"MERGED","created":"2026-06-03 20:59:07.000000000","updated":"2026-06-04 10:20:39.000000000","submitted":"2026-06-04 10:18:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"991505","meta_rev_id":"7bc9f4859f9b71c290911d80b5dbadd283335330","_number":991505,"virtual_id_number":991505,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-04 10:18:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2026-06-04 08:14:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2026-06-04 10:15:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-06-04 08:30:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-06-04 08:30:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"}],"reviewers":{"REVIEWER":[{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-03 22:49:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-04 08:08:11.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2026-06-04 08:08:11.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"state":"REVIEWER"},{"updated":"2026-06-04 08:08:11.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2026-06-04 08:08:11.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2026-06-04 08:08:11.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2026-06-04 08:08:11.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2026-06-04 08:08:11.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2026-06-04 10:15:01.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"9117ec16b8eaffaff07e962a1de2d9bb25ea9f3f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-06-03 20:59:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"32cc6843cfecec9be427681d071241454a548a63","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-03 22:49:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d1f995d160454b11add092594f438fa7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5859449af4b4fc89a93f93cbf52ee06 : SUCCESS in 10m 46s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/7ffe8193c4a14fb8b86afede2f612d04 : SUCCESS in 7m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7c930f9927ac4b6ea65dbf4acb25b789 : SUCCESS in 4m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/81865984dfb7438f9902dff82b1fba16 : SUCCESS in 6m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/92810e04ce06480fb4ea10111e7fc2b1 : SUCCESS in 4m 46s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/3e0e0ae5804349abaad2e5a79dae078a : SUCCESS in 7m 22s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9bea89b3e45946cd9fb00637259e47e9 : SUCCESS in 3m 09s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/cbb42b6bba4a4d8f969048fc0c42d66c : SUCCESS in 4m 02s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0151bb07faad4eefad760b61374340b9 : SUCCESS in 15m 04s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/838221db62e148528cfdcf2d9333eb40 : SUCCESS in 7m 13s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/94a7ee4a40204a57a847cf8459603ad2 : SUCCESS in 15m 23s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/49cf0e7bebfd4101a683d069cd8ffbe5 : SUCCESS in 15m 28s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2777067ef726440a91962913414cf1b1 : SUCCESS in 17m 53s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/aafc4ad8605247a7887ca1c44e85d8e7 : SUCCESS in 14m 19s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/81e9f9023adf458aac9386ef08e863b6 : SUCCESS in 14m 44s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e97ad4fc2c3c40eeb346df99dbc338d4 : SUCCESS in 19m 02s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/61f1ede1e870450baf1725013c1fe4f7 : SUCCESS in 13m 20s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a3d3ef844a844f4e8fedb156f431d654 : SUCCESS in 12m 47s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/9ec9b7e481c045b588c64bbf2f8c31f7 : SUCCESS in 59m 22s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/89e90c932adb4afba68f5b945e15c968 : SUCCESS in 57m 20s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/df216ae7d8e84d70b6a0d9e915d804ea : SUCCESS in 59m 10s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1ac1614dc14e4086a5d360fbfb4a94ea : SUCCESS in 1h 02m 51s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/b2ceb0b1f3cb4944b81fdab3dd851cac : SUCCESS in 58m 59s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/cd5daf4fdab245a888cf59dea6e1ee95 : SUCCESS in 1h 43m 55s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ce826bdf5f274396aa270d5d7a9aeb72 : SUCCESS in 1h 43m 24s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/a8f8139bd0e24018ac12ffa36f0565a8 : SUCCESS in 20m 31s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/a89446e0477b4ba88b4d672e433cc9f5 : SUCCESS in 21m 32s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/829eded4657943da9e0591570e8a896b : SUCCESS in 17m 59s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/c7705b3660a8406dae09b1329c52a697 : SUCCESS in 21m 40s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/5b4fce8d568f4517a56ab9ea0aa48e76 : SUCCESS in 23m 19s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/995191bf4f764b1d84c1c17a6160d039 : SUCCESS in 21m 01s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/78bd3c4dd89f4f8286fd5c3da31a5e56 : SUCCESS in 25m 50s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9ad23e63767642e5a0fdb31ee9dc1b81 : SUCCESS in 20m 24s","accounts_in_message":[],"_revision_number":1},{"id":"0bc3eeb2ef1d21d3b24c5a9e1c8b19149e82305f","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-06-04 08:14:34.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"7331a4399aff8f017e5019d30e6efb63388ea07d","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-06-04 08:30:50.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5b934a69846d096c2f47dc538308d7f64e6994b6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-04 08:31:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e1f4cf0062a272688c3399dee821b3dbec8290d0","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-06-04 10:15:01.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"a6e8dcaf82d18c0910abd28812bc785c2b40877c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-04 10:18:25.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3bd0d06586044f85beb59ae1275f38db\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/61512703d5d64528aa91dd6901fd56cb : SUCCESS in 9m 47s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/27da2eb7ce29427da5d5b0b13df41b38 : SUCCESS in 6m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f2b5ab36d0bd4c25aa9334d0c4f317fe : SUCCESS in 5m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/afc0eeb16e13464881278d046f7917a7 : SUCCESS in 7m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/35a88536a01a43c1a329347e56316222 : SUCCESS in 6m 48s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/19538badddf24d40aa447ad55d7cf3c5 : SUCCESS in 3m 18s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d4da105415ca4229a77440b9ceebaec5 : SUCCESS in 6m 43s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/aca53bab24c247049168cbb25f02ea4e : SUCCESS in 11m 32s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/b9d208c4addb4b4fb3847c7a3650c3c2 : SUCCESS in 14m 19s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b32c4294df3d44a8bcd4fc7b3f9a813f : SUCCESS in 17m 32s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/430edd417d874c068451c2462774ea03 : SUCCESS in 22m 37s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8b198653b79444db8723b24660a7d1aa : SUCCESS in 20m 32s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/b1f963fa61dc4a03bd7e37696e3bb3d5 : SUCCESS in 12m 58s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f9b9d4e2f00c4872ba6b8f2ebbf709d8 : SUCCESS in 17m 04s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/c19651a325914ec6a9615ac93dc56e3d : SUCCESS in 53m 57s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/3d788b9730734c81baa187180806a0e6 : SUCCESS in 1h 00m 31s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/c12dbb7618134794b69f844877cb9420 : SUCCESS in 1h 01m 35s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6c5cd67616ed4833ab3018f7cb749cb3 : SUCCESS in 1h 02m 37s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/b8b6f615ac074545a4829fd867b9cdfd : SUCCESS in 1h 02m 23s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/c5d9776014b743d7af64ccd434e82cf2 : SUCCESS in 1h 43m 02s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6d74465aca3e4744ba9443a18d5f4ee0 : SUCCESS in 1h 46m 42s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/94974acc519c4b8db9eef7ef29065b65 : SUCCESS in 16m 06s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/21420a4faeb346aa9f919bbff27466ef : SUCCESS in 16m 22s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4a752af9b1b04b3cb8089401df583fd2 : SUCCESS in 23m 39s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/b8bee74ad81343018a36e551ff4e5024 : SUCCESS in 22m 33s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/d440f3e5d6cb4b1984252625890a59bc : SUCCESS in 43m 25s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/07ebd29ecf5748c7a4dad047d3f5a8f8 : SUCCESS in 34m 49s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/51c727eabcdf4cfc8f488c7d9e3d9f1c : SUCCESS in 14m 06s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8178e200cf0c42409a408c067aea6c2d : SUCCESS in 19m 22s","accounts_in_message":[],"_revision_number":1},{"id":"0d4db7d36fad0e97546dd95c2e09dc368b64f69b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-04 10:18:25.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"7bc9f4859f9b71c290911d80b5dbadd283335330","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-04 10:20:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3d34599c5ef04e6ab1876b83188c93fe\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/844714fb5554413d8b7fb81c74b980eb : SUCCESS in 1m 18s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d564feffded2fbb42efbe2626e618e3f232b0791","revisions":{"d564feffded2fbb42efbe2626e618e3f232b0791":{"kind":"REWORK","_number":1,"created":"2026-06-03 20:59:07.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/05/991505/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/05/991505/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/991505/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/991505/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/991505/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/05/991505/1"}}},"commit":{"parents":[{"commit":"919ccfbc7ca7d260b2f1a56fc5ef161f486bd870","subject":"Merge \"Fix naming of custom container variables\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/919ccfbc7ca7d260b2f1a56fc5ef161f486bd870"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-06-03 20:58:01.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-06-03 20:58:01.000000000","tz":120},"subject":"CI: Stop using image from Docker Hub","message":"CI: Stop using image from Docker Hub\n\nSome CI jobs are hitting rate limits.\n\nChange-Id: Iad22b28c39a023d26f89b753af73dd730d000cea\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d564feffded2fbb42efbe2626e618e3f232b0791"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d564feffded2fbb42efbe2626e618e3f232b0791"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
