)]}'
{"id":"openstack%2Fkayobe~991507","triplet_id":"openstack%2Fkayobe~master~I8a1f505e17fa282d9eb3c76cb703b84b322b3dbb","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2026-06-04 08:16:10.000000000","reason":"\u003cGERRIT_ACCOUNT_28048\u003e replied on the change","reason_account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2026-06-04 10:10:51.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I8a1f505e17fa282d9eb3c76cb703b84b322b3dbb","subject":"Replace `which` with `command`","status":"MERGED","created":"2026-06-03 21:17:19.000000000","updated":"2026-06-04 10:11:52.000000000","submitted":"2026-06-04 10:10:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"991507","meta_rev_id":"f8dad0868964b75c583a30fb763fcb42aef51cc5","_number":991507,"virtual_id_number":991507,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-04 10:10:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2026-06-04 08:16:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2026-06-03 23:43:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-06-04 07:59:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2026-06-04 08:16:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-03 23:02:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-03 23:43:36.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2026-06-04 07:59:58.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2026-06-04 08:03:11.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"}],"messages":[{"id":"ddab67122af26d0c68fe3719f38ef247a2f273e5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-06-03 21:17:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c7190c3d3ae452564ed7e39c906e66b717a95418","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-03 23:02:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c5a1fd4692aa4e499f920440fe0b318e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb0327af54b745b8aad86584d9e294e3 : SUCCESS in 9m 41s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/49c2b90b2c504999b819e634343532fd : SUCCESS in 6m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5548b73451e54f3e817190cf11dbe529 : SUCCESS in 4m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/18f56e0acd454299bd408e35df543b6b : SUCCESS in 4m 21s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e47fd1d1207146ef85999a554ce8ed25 : SUCCESS in 4m 22s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/0daa2d4cb4944151b107516cdf1c221e : SUCCESS in 5m 12s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aeefc8a14d5b4b518a2a2e135087bb5f : SUCCESS in 6m 22s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/891d072f5fce4139bd4eb800160f5ef3 : SUCCESS in 4m 38s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1ae7b7a333c5442b814cad0df3efa1a3 : SUCCESS in 11m 37s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/0e1fe5a182e24001a26fb42c079f4ea2 : SUCCESS in 6m 34s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/463598da62904130b549525d731a705b : SUCCESS in 17m 31s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/ed020171885c4415b146c40c44633d84 : SUCCESS in 15m 53s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/72ade16451c84c2a91ba6bd99a46a0f5 : SUCCESS in 10m 46s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/6595640f3c9a4448aa767404d7cccb97 : SUCCESS in 10m 26s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/1f6b0153a4f84010b017481d998241d6 : SUCCESS in 14m 20s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9581c767211448e1a8f7b9c7850ccdf3 : SUCCESS in 18m 10s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/d4abe44fd7224949ba44caa45cc296c2 : SUCCESS in 11m 37s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9673c2cd55714be287ddab111ad6ec75 : SUCCESS in 10m 52s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/0273db0634e34cd9a44bc9a633fdbd42 : SUCCESS in 59m 08s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/9fb5c13ccc144ad8aeec1089f21c2437 : SUCCESS in 1h 01m 15s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/d8883f121dcf4132acbdf52a39145c25 : SUCCESS in 58m 02s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/728cf5fc60694f42b01a59a2fb3cdc67 : SUCCESS in 1h 00m 59s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/e6787b01beb24c61ad3a5b1ed7a0b97d : SUCCESS in 1h 02m 08s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/6b905359190d4de29b58f53cae3685ff : SUCCESS in 1h 43m 27s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9cb225455c6e4089afe2a3a46fdf90b9 : SUCCESS in 1h 44m 12s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/faa71a901f864ce09a43f2270ac9abbd : SUCCESS in 27m 42s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/54b7869ebdfb4690857a69ec604db5bc : SUCCESS in 27m 06s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8343aab6bf7c48499eb76937466cbfaf : SUCCESS in 23m 14s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/c5bfa279e98847b585d17a780a58c1a2 : SUCCESS in 21m 22s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/5f75a6f3fbdf489db8e3cb86631ad024 : SUCCESS in 34m 41s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/87c2dcd42bf1498abec28f3267dbdd5b : SUCCESS in 40m 06s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/181225b3befe4e79aeb7814bbaf48419 : SUCCESS in 13m 37s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/39fcc269d93a422f876cb6a264fa5bcb : SUCCESS in 22m 07s","accounts_in_message":[],"_revision_number":1},{"id":"fdacfbfaa15bc85012c6c8c8e024115bcc30fc61","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-06-03 23:43:36.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"1fab746f00a8be05ba9d04418903e0d99937e7cb","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-06-04 07:59:58.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"9e8e912b558d543b43cd49c59466905d9f21592d","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-06-04 08:16:10.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"794e0cb54b94ef3cfaaadb8777db7f8bc7a22da3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-04 08:16:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a054022f47e2408a27fcb96e62d356d2cef0c0de","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-04 10:10:50.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/16a94def2c584e209de09d6f913b96d2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18b195e00ab246ffab287c2accdd7d03 : SUCCESS in 9m 29s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/086d03d7deb44403bb0d7626f01ae297 : SUCCESS in 6m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ece8288b043f4426866dfafded63964b : SUCCESS in 5m 00s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d23b1f357c644d63aa9ed5a5e75bee27 : SUCCESS in 4m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4a298fa9594b49c7ae1ea893a7dcb072 : SUCCESS in 4m 28s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e57fc2a80f2e4a2a9c0a3e94bc40b515 : SUCCESS in 4m 49s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/cdc08b1f0fec44968c78471648823b81 : SUCCESS in 13m 43s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c3b279fcad5f49099d0f16a44876040e : SUCCESS in 11m 19s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/3d1fb447cde1413681e5edaa808baea1 : SUCCESS in 24m 21s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/09777bc2a6bd4ce49e131534aaaba1ad : SUCCESS in 11m 21s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/b5672a2193a04e2daba7b74aaf097119 : SUCCESS in 13m 12s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/acea844beeb7490c954a76fd25bb6c35 : SUCCESS in 17m 54s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/6d7dbe5e0aed462e9d2b50543b425c2c : SUCCESS in 24m 17s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a00055666dec42f79d3de5f635e47e5f : SUCCESS in 15m 38s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/90fd53c7ca1d4e4396dd75aeecd331cb : SUCCESS in 58m 35s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/e6d632575dbf494eb108323ca5b235dd : SUCCESS in 1h 03m 41s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/88d41ad5757f4cc7b29041f1b90291f8 : SUCCESS in 1h 06m 29s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3c5a2afaa48241a58287e04e9a06b515 : SUCCESS in 1h 05m 17s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/0afccd8bf5b8488dbcbcce0d35765cee : SUCCESS in 1h 08m 52s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/7b535516ee8c467caf20d507f3165b61 : SUCCESS in 1h 48m 32s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/386399fedea34100bef9b377b7561e3d : SUCCESS in 1h 53m 50s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/576d44f11ba743c68e2be122817d6fec : SUCCESS in 15m 05s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/0750feec02894aae978b997763c48e5d : SUCCESS in 22m 49s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/38a7fbb2df3840ef99b78cf114c545b9 : SUCCESS in 23m 13s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/c46cd472c6a64db6b07695e5afc5a92c : SUCCESS in 23m 02s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/d878224e53ac40f4a84175388b6e239a : SUCCESS in 21m 05s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b26f29ec75d842d18471dacf39646c36 : SUCCESS in 41m 05s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/14855c17b48d4752bf484f03a84ec5b4 : SUCCESS in 18m 29s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fc60ee887ba24980b295c5db0ffd312e : SUCCESS in 15m 21s","accounts_in_message":[],"_revision_number":1},{"id":"7647e46fbeda91898ad763db49efcded5ed9a7e2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-04 10:10:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"f8dad0868964b75c583a30fb763fcb42aef51cc5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-04 10:11:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ff40ee75b02446738daf971296dd39b7\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd1b31ec281e40f79d2f685315b9aef7 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0e1fab73cd9b45b5e317f8fddb729d1a6896656a","revisions":{"0e1fab73cd9b45b5e317f8fddb729d1a6896656a":{"kind":"REWORK","_number":1,"created":"2026-06-03 21:17:19.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/07/991507/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/07/991507/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/07/991507/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/07/991507/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/07/991507/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/07/991507/1"}}},"commit":{"parents":[{"commit":"919ccfbc7ca7d260b2f1a56fc5ef161f486bd870","subject":"Merge \"Fix naming of custom container variables\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/919ccfbc7ca7d260b2f1a56fc5ef161f486bd870"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-06-03 21:15:58.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-06-03 21:15:58.000000000","tz":120},"subject":"Replace `which` with `command`","message":"Replace `which` with `command`\n\n`command` is a builtin command in any POSIX-compatible shell, while\n`which` is an extra package which can be absent.\n\nChange-Id: I8a1f505e17fa282d9eb3c76cb703b84b322b3dbb\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0e1fab73cd9b45b5e317f8fddb729d1a6896656a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0e1fab73cd9b45b5e317f8fddb729d1a6896656a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
