)]}'
{"id":"openstack%2Fkayobe~991575","triplet_id":"openstack%2Fkayobe~master~Icc517339b5ca1b66e750c0623f99451402da7cca","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2026-07-13 14:47:06.000000000","reason":"\u003cGERRIT_ACCOUNT_28048\u003e replied on the change","reason_account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},"17669":{"account":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"last_update":"2026-07-13 15:23:22.000000000","reason":"\u003cGERRIT_ACCOUNT_17669\u003e replied on the change","reason_account":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"}},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2026-07-13 19:04:20.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"35263":{"account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"last_update":"2026-09-10 09:56:13.000000000","reason":"\u003cGERRIT_ACCOUNT_35263\u003e replied on the change","reason_account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}}},"hashtags":[],"change_id":"Icc517339b5ca1b66e750c0623f99451402da7cca","subject":"Never reboot without explicit permission","status":"MERGED","created":"2026-06-04 11:04:44.000000000","updated":"2026-09-10 15:22:21.000000000","submitted":"2026-09-10 15:21:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":29,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"991575","meta_rev_id":"243c60573ae32afbf6e36707ed492baca15e0e9c","_number":991575,"virtual_id_number":991575,"owner":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-10 15:20:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2026-07-13 14:47:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":2,"date":"2026-07-13 15:23:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":1,"date":"2026-07-13 19:04:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2026-09-03 08:43:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2026-07-13 14:47:06.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"}],"reviewers":{"REVIEWER":[{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-04 11:11:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-07-06 10:22:48.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2026-07-06 10:22:48.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2026-07-06 10:22:48.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2026-07-06 10:22:48.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2026-07-06 10:22:48.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"state":"REVIEWER"},{"updated":"2026-07-06 10:22:48.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2026-07-06 10:22:48.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2026-07-13 10:01:15.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"ddce1af37849a10cf694155c6963a5ca893eccc3","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-06-04 11:04:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e71a63dd139c02ff81b5c933bd0c9a34ed32312d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-04 11:11:51.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/51c7eb8369544389ae6171ee84f5793d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/096c157a4f9144928fbda5bd3b208995 : ERROR Project github.com/stackhpc/ansible-collection-linux does not have the default branch master in 5s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/3a2994e44d22465581d2009a7832f7ce : ERROR Project github.com/stackhpc/ansible-collection-linux does not have the default branch master in 6s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5fdb39f0d3814a9593b2af856e6886d1 : ERROR Project github.com/stackhpc/ansible-collection-linux does not have the default branch master in 6s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e3322ac3cd8b48e99937072ae76369b6 : ERROR Project github.com/stackhpc/ansible-collection-linux does not have the default branch master in 6s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4caa9af0a1274d3aa85e6b43d0318474 : ERROR Project github.com/stackhpc/ansible-collection-linux does not have the default branch master in 4s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/97cf91ad885f401083fb73c2d57fd6ac : ERROR Project github.com/stackhpc/ansible-collection-linux does not have the default branch master in 8s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bd7e15ac8d7b4d54b2a9d0e4a36522b6 : ERROR Project github.com/stackhpc/ansible-collection-linux does not have the default branch master in 4s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/742cb405985d46eb858f3542068a30be : ERROR Project github.com/stackhpc/ansible-collection-linux does not have the default branch master in 6s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e8ccb745829f4fd0b9699ce24f357124 : ERROR Project github.com/stackhpc/ansible-collection-linux does not have the default branch master in 4s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f7cac4ca7a23464e86aed79f987584a9 : ERROR Project github.com/stackhpc/ansible-collection-linux does not have the default branch master in 8s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/07f2ee46a63d474489181ab057550878 : ERROR Project github.com/stackhpc/ansible-collection-linux does not have the default branch master in 11s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/e1a0de045a294b699d3e2bcd590b6bba : ERROR Project github.com/stackhpc/ansible-collection-linux does not have the default branch master in 11s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/679b5e88fbaf45fea5bd2fa0d9d9edd7 : ERROR Project github.com/stackhpc/ansible-collection-linux does not have the default branch master in 14s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/f02f7547506641a1aa562f99c37b70b0 : ERROR Project github.com/stackhpc/ansible-collection-linux does not have the default branch master in 12s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/04278eead3d44bf68d9c6dad354faafa : ERROR Project github.com/stackhpc/ansible-collection-linux does not have the default branch master in 13s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3b560d0abc8b4445b531e8e6188dd184 : ERROR Project github.com/stackhpc/ansible-collection-linux does not have the default branch master in 19s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/05067c5311eb4167861560a0d74cb370 : ERROR Project github.com/stackhpc/ansible-collection-linux does not have the default branch master in 12s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e6c0a667c7084379af881c1639792232 : ERROR Project github.com/stackhpc/ansible-collection-linux does not have the default branch master in 11s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/a95efa74c226404ab33e60bcdd0b1189 : ERROR Project github.com/stackhpc/ansible-collection-linux does not have the default branch master in 19s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/d16fbe3647534a118ad1b7fc4e9dfcb9 : ERROR Project github.com/stackhpc/ansible-collection-linux does not have the default branch master in 20s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/2040501a922c40968e34ecb199cc93ca : ERROR Project github.com/stackhpc/ansible-collection-linux does not have the default branch master in 18s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e459ea55cb374cc5a6d303e6ce0f0898 : ERROR Project github.com/stackhpc/ansible-collection-linux does not have the default branch master in 15s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/d8903a5929ea4aa0b97c700db63bccb0 : ERROR Project github.com/stackhpc/ansible-collection-linux does not have the default branch master in 13s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/a0c665747c814068a1671be64532062b : ERROR Project github.com/stackhpc/ansible-collection-linux does not have the default branch master in 14s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2e40ef609b724b5e94d2326ae15c8cba : ERROR Project github.com/stackhpc/ansible-collection-linux does not have the default branch master in 15s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/d71ec963fd3944929e5afd66312741cb : ERROR Project github.com/stackhpc/ansible-collection-linux does not have the default branch master in 12s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/1c22723b19ba49ee8631dccc716bc65a : ERROR Project github.com/stackhpc/ansible-collection-linux does not have the default branch master in 16s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9138fe9cbd5c4febaf7b08cc39f3dd3f : ERROR Project github.com/stackhpc/ansible-collection-linux does not have the default branch master in 12s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/c0bd8cf7ec8e4c579e2a668c37550629 : ERROR Project github.com/stackhpc/ansible-collection-linux does not have the default branch master in 14s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/dd5a55fa8ba64d80a927e4cabdf6c40e : ERROR Project github.com/stackhpc/ansible-collection-linux does not have the default branch master in 12s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/aa331337f4e24d20856bd54a9729da8a : ERROR Project github.com/stackhpc/ansible-collection-linux does not have the default branch master in 21s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/7dc416a3fb5f4ec69d6f45849df08c54 : ERROR Project github.com/stackhpc/ansible-collection-linux does not have the default branch master in 17s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/adbd445f3f7645aa90576ca93e5f1a8e : ERROR Project github.com/stackhpc/ansible-collection-linux does not have the default branch master in 11s","accounts_in_message":[],"_revision_number":1},{"id":"fc1dd324462b281fe5715ecf225997e3ef292fd1","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-06-04 15:03:17.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"2885d8b8b39a1298011f93f68e11c974f018d8f9","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-06-04 15:16:07.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"49b4fb17416671db3dac0261ea653f2c8d6649f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-04 16:59:56.000000000","message":"Patch Set 3: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/86b3ec4af5fe42deb57ce2876a3ebcce\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b3502614af3402e9a202b6313ee0f20 : SUCCESS in 9m 47s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/37047a9bb2c84895b9bcdbe6f307ebba : FAILURE in 5m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/840053d3bfc0411fbf190569179c9bc0 : SUCCESS in 5m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/354da565330a4df587094edd5785c9c2 : SUCCESS in 4m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/10ed460b16d24e5fb4756185a6896aec : SUCCESS in 5m 37s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f332f4db26af4211b9208c8f5a656414 : SUCCESS in 8m 12s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b3af5786c5544aee90928bed67e8dae9 : SUCCESS in 4m 09s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4a03846847bc491488ccccf9e10d19dd : FAILURE in 3m 42s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ea8e485a245041f987900a1c81fd8be2 : SUCCESS in 13m 36s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/996743770d8a41228155b1ad6566ac49 : SUCCESS in 11m 50s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/04ddc73f268d47f3ba19d6a127d557c2 : SUCCESS in 14m 13s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/b1701a2dd1f047a6a48b1e41317d4905 : SUCCESS in 23m 48s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9a7c5becbe434187aca2b3e193ee835b : SUCCESS in 19m 30s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/16e62adb3bbd4932a4ff64ad6cb70269 : SUCCESS in 9m 07s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/e7ac942499144d46a087e0ca2d719477 : SUCCESS in 23m 52s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ea07e618330947d59d04bf07d40dbbc1 : SUCCESS in 12m 50s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/b3142f4bb1de4fa18c8d2ffdfec51e5c : FAILURE in 12m 48s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c39acb6a734c432eac931532cb617df2 : FAILURE in 8m 15s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/796d3034bff747279ee5933b6d36f504 : FAILURE in 8m 43s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/d662de12fae244c78579699fe855b88d : FAILURE in 7m 01s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/2138211ef39f427f93b6eab8920ce7bd : FAILURE in 7m 06s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f2da390c7dc14f5e95867f2c0af0c81b : FAILURE in 7m 29s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/c9773273699843fa864d82d65071f72d : FAILURE in 7m 44s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/68248fa07e3e4f288350372ec1e597b7 : SUCCESS in 1h 42m 33s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d264bd44d94f4b45902d1dfd12ad79f4 : SUCCESS in 1h 41m 32s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/187eb9c10abc4346bea6abf6c6e3467b : SUCCESS in 19m 02s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/b093ff1175a849f29aaeebc590e1fb12 : SUCCESS in 20m 32s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4c946c7d5d5842619f0de4dd18488481 : SUCCESS in 27m 09s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/40d9847e65f74848801206cd90ca7e0e : SUCCESS in 14m 07s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/4c229cb41ce54e5fa1a35bf709f15586 : SUCCESS in 22m 24s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a04116157dd0413b97d627963811ffc3 : SUCCESS in 28m 06s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/e5131b9e067a4e7e83c5e11e79ddce56 : SUCCESS in 14m 29s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e2bb5dd998ca46d0934114373c7060f1 : SUCCESS in 21m 04s","accounts_in_message":[],"_revision_number":3},{"id":"82a3b6fd31617ab4a2c154aca19d3aa5630bd362","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-06-08 08:22:25.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"90e9cd888b2b48b19f47588a7bb21ee6a0f2e49e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-06-08 08:27:31.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"7e5cf4f172c3cde6d85e3a91e41191fb69f96618","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-06-08 09:08:30.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"7a320d4c26ad96542c971e11010fce2bb7b95839","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-06-08 09:10:25.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"fe21b7748e18e0aa38c6899d086774c7c7dc3583","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-08 11:05:24.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c59a252f95474b34a3ab5e5c714cd904\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0dbec2244089484a9ada0d0c4a55e4be : SUCCESS in 10m 20s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/c5f5d731c81742849b5b495e57cc95c4 : SUCCESS in 8m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eb0a31ad88e5420aa71a0491f355a517 : SUCCESS in 5m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/46ed9d271c3d4198b4a5955c19d2b28b : SUCCESS in 4m 27s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/741e1d7ab0374f82b88d798df993ea2b : SUCCESS in 2m 56s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/24ba864fb2184071a4ca2784099d7a92 : SUCCESS in 7m 24s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d530c609000e42e79c6ac656776c4e0b : SUCCESS in 3m 54s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/421341b1af8846fa830c780fcac86b50 : SUCCESS in 4m 52s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5b081f9437a94ba3b42ea6a9cabdfb52 : SUCCESS in 12m 27s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2543b4650a5e40bb845f568a9386521e : SUCCESS in 11m 31s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/db8fad6e981f426c845cbddfde66eca0 : SUCCESS in 22m 15s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/54eebfec55704b779ea0788ebed61ddf : SUCCESS in 21m 29s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c48d22f2f66a4d20900cc93323ba81c8 : SUCCESS in 18m 43s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/b2045c0e5e3941a9a3812214ee31935b : SUCCESS in 18m 32s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/86280b3f19f84d42891f7eab37d0eab7 : SUCCESS in 26m 48s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8f9071ae328a4833a2b7bbd966ebdd20 : SUCCESS in 22m 08s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/311f00f65cb34b04b15e848c513f37f3 : SUCCESS in 18m 24s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/503de661cb4b4aa0b9fb8a3f68e9178e : SUCCESS in 19m 03s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/b6c21637bfb645cbb76ba65938fec698 : SUCCESS in 58m 56s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/f2a095bf410e49f58f4dd082beb48782 : SUCCESS in 59m 41s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/03fa3e4fd4504d9c8ad8c1a2eb130f13 : SUCCESS in 1h 02m 57s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f7bf34c1c4874628ba0598f2aa17d365 : FAILURE in 31m 17s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/0b963d4db1384dd5a17af5f6e337d784 : FAILURE in 16m 22s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/0ca9d68442a24b5e863e447398b5e157 : SUCCESS in 1h 41m 46s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7d24476c7f4d4b37b8ee47cdf9a87f78 : FAILURE in 29m 56s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/d31e5a219ad549c19351198810b98634 : SUCCESS in 23m 45s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/43412c9a4c45410bb25a5ddbd9fad598 : SUCCESS in 21m 28s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f05502754baf495d81827b45b2252954 : SUCCESS in 23m 19s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/c41cb5aa0cb04d9892947c8fe48776f7 : SUCCESS in 23m 54s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/f31aa6862fb94680b2c5df71da7b2776 : SUCCESS in 24m 04s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/397419f6b084456aa70db0164bcadbda : SUCCESS in 37m 03s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/c8cecfe8aa3a4117a20578e800a71fba : SUCCESS in 14m 30s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4479429c76e84662873a991686430584 : SUCCESS in 19m 16s","accounts_in_message":[],"_revision_number":6},{"id":"f3465b70eb006bd0de4b7efe628b3c5b76d42de5","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-06-08 12:15:34.000000000","message":"Patch Set 6:\n\n(1 comment)\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":6},{"id":"2bb28272e3f2bc7b5b21b806d628c66dedf83dbb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-08 14:05:02.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a203e5bb3bc947ed853060e28318b913\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ab486c522124a25a234f2aa0665c523 : SUCCESS in 4m 51s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/78218d4f9eef4855bf4b502b91c82f5f : SUCCESS in 6m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e2477c5a4c7c47018dc8beee4d174ad1 : SUCCESS in 3m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/75dcb9f1d7c646aa82a1ba7bdc108fb2 : SUCCESS in 4m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/88142c7aba9544039c758d9d4817b810 : SUCCESS in 4m 02s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e5e9b61d4ca247798f5da800f76f8d16 : SUCCESS in 6m 23s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/53faa292005a4e28a906a0118b96c636 : SUCCESS in 6m 14s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f2106564bab7451abb1fc71b5007bd6f : SUCCESS in 6m 32s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/cc1f692578ea4662b9e155eacf76558d : SUCCESS in 11m 22s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/cb8b414cd1cf49a5a34d7343aca77e65 : SUCCESS in 13m 52s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/6b5a5bfdee8845e8b95957a23407cd62 : SUCCESS in 12m 29s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/bf56ed21711f4e57b5b4e354420a49eb : SUCCESS in 17m 39s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c399febe069a451bbb6675c63ac6ba43 : SUCCESS in 16m 01s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/b9308c1719ba403196a4cb1940f538f3 : SUCCESS in 15m 09s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/e6109a3d322a45509c1d46fc68c764ea : SUCCESS in 15m 35s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5a3179ea5d5f44968594230505df42a5 : SUCCESS in 21m 11s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/c93d0a60287c446db1f486a2ab615ca0 : SUCCESS in 16m 20s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/87640e21e01b4ffeb12c4adce8ff492a : SUCCESS in 11m 10s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/6e6967640950495a8500867d170c1059 : SUCCESS in 1h 02m 33s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/2fccea650f6c4450888c44f6a6d0bfd4 : SUCCESS in 1h 02m 28s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/cd0bb65d58b74c209b30c85bc43dbd0a : SUCCESS in 58m 09s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a532b41514804c0b9fb85517f0f9a373 : FAILURE in 33m 17s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/22ea81668d864fffbf997c6aa9b448a7 : FAILURE in 28m 30s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/483193d69c704a48b9cd94d174959617 : SUCCESS in 1h 47m 59s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/42aeb37bfb2a443ba2044cbdc4e03393 : FAILURE in 32m 17s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/1f298aceb21d4826b203996c7f89b545 : SUCCESS in 20m 00s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/094c7066602e4abebcca6b86d39d1683 : SUCCESS in 15m 50s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/78821642099e492ea0f3022cc6d4c721 : SUCCESS in 13m 49s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/908fe36d7f714577bc1cc7fd0e42c0e3 : SUCCESS in 19m 01s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/1529f7c0e5494d0c9e5b2af917d07c97 : SUCCESS in 22m 02s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5bbb856c19594d37b85c0df4d8776911 : SUCCESS in 22m 41s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/b1db50a6570244a1ad6116fbb7f1a735 : SUCCESS in 24m 47s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/25f42b92b3be4e198dee6cce6ebad956 : SUCCESS in 21m 19s","accounts_in_message":[],"_revision_number":6},{"id":"9b8195fcd233c1261466e22b2b4599dedf7673b0","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-06-09 14:59:16.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"df5b88231ffceed85d86df2c01f9ca739c8afd9f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-09 16:14:33.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/91f0efb8836145afaa3542ecd3017f3b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f36af485dbab46c4b69ea9fa8a207d1f : SUCCESS in 9m 27s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/65cb754c4db7467ab42e1ff288527c51 : SUCCESS in 6m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f6ee107e6e5a44f59513d2cc0547e382 : SUCCESS in 6m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d673850309114820a8ccc24b9a07546d : SUCCESS in 4m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0060b06afa49410cbf2287f7b90ea103 : SUCCESS in 4m 20s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/fe4f5b4930ca47c289b726a37a5e48c7 : SUCCESS in 8m 14s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2ef80bc7375c4c3997a1a7acbab44222 : SUCCESS in 3m 29s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/84cf4f6d346f411cbbe79aa87d9c0608 : SUCCESS in 5m 38s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4cd7815d298142c8b08e753109584161 : SUCCESS in 12m 20s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2cf5948b4e464758b8d0f3c8d94a45a1 : SUCCESS in 10m 37s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/8ed8a4634aa64e379637d4f9bff029b0 : SUCCESS in 13m 03s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/3616385c305a42ad940ab15dddf4f594 : SUCCESS in 14m 14s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2f890ec127a3467c8726b831705bd66c : SUCCESS in 15m 35s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/940237f403d24816a0d6dcc2448fa980 : SUCCESS in 19m 00s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/f5936d55540741f69bf87da4fadc6da5 : SUCCESS in 13m 31s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/45271590b99f42fd9c8a6470e404674a : SUCCESS in 21m 17s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/ffd8916334614877b407b90125d46f49 : SUCCESS in 19m 15s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d144c16e04ae4667bbad4e51a5411115 : SUCCESS in 17m 20s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/d3666b98d48b441d87f4347d2d86d54b : FAILURE in 28m 44s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/0531daa1aa634ac8b15bd4cbbb81da6c : FAILURE in 22m 01s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/69a716f3a9f544bea8ece56dd4c1ac72 : FAILURE in 28m 13s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9337bfb462084e799ed3c18df0cea55a : FAILURE in 24m 29s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/161170ab4cab4e2095942608804e8192 : FAILURE in 17m 12s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/3a2a40fe8355425a98e1bfd9940dc554 : FAILURE in 1h 13m 09s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d9ec4ec1c81047f2ac4f1885a7219537 : FAILURE in 30m 16s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/b37d869887f2408385f112a6fd359ff9 : SUCCESS in 17m 42s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/0df618fda4d1469fab44f20d02b74917 : SUCCESS in 19m 34s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bf1b5338783142bc9948f5d9c6eedca2 : SUCCESS in 25m 44s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/0dbea4d2e781448ea12bc0eb594621cb : SUCCESS in 18m 19s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/c2b8f5a9b66f495881457bb1fa37de09 : SUCCESS in 26m 58s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e65037e8a0c648a5a604cb2d7a2c9db7 : SUCCESS in 34m 08s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/220463edbc1340bc860c86a7b701dc5c : SUCCESS in 20m 42s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f06cbe4434264df3885ec6563a728352 : SUCCESS in 22m 49s","accounts_in_message":[],"_revision_number":6},{"id":"04d839dddfd8008ad18619c6c9ad89535e3c35a1","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-06-22 12:35:12.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"3bf32c9dbf74b554fb556ccb09e0a95b8c194821","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-22 14:27:25.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/74300089fcff44dba12576bf4c59a622\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c12154952a3f44c89853f13229dc8410 : SUCCESS in 6m 19s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/e2aee0bd01174d359b461daa1cbbba6e : SUCCESS in 7m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/99a2b2955a34437c9fada020ad6f259f : SUCCESS in 4m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/acfe4dd613ba4c46a850f15c492ad4fe : SUCCESS in 3m 16s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/43d75e9b9ca44271b2ffd3deaa64e087 : SUCCESS in 4m 38s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/62b70a2b6fe840568301e57d67084b28 : SUCCESS in 6m 29s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ca342a1dd5124644b25c56791bd1946e : SUCCESS in 6m 22s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/836d2e7f8dff430082748a55eb2f39b3 : SUCCESS in 5m 03s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c2f2738bbe844f3b979f182248e8fb26 : SUCCESS in 13m 44s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f70aa9e146a6485fbe856a5fa603e847 : SUCCESS in 14m 22s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/ed1cac41cfd74b2bbc845831dc858b47 : SUCCESS in 21m 49s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/9659587cf5d946a48ec9c214bc4c9b68 : SUCCESS in 16m 48s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/60be7c2d577d45438c4af92be1e5d7c5 : SUCCESS in 10m 20s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/6a49570068cb423fb568881b7bd2d298 : SUCCESS in 16m 39s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/5b2e5448eebb40e691e08fa0d550673c : SUCCESS in 24m 01s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1243d4e7a9334955808e311ef5502390 : SUCCESS in 17m 35s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/96c66638b5ef41c7824297e475ecc26d : SUCCESS in 12m 00s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e00529b2f8434a0a9f670dd68d18b8e2 : SUCCESS in 21m 01s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/ce97a4c4d93c422ba4a63bfb4980b956 : SUCCESS in 1h 01m 53s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/85d550b52c2f42cdb03f16e17a20afc7 : SUCCESS in 59m 49s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/bd56564c90224409b1e19707b4547738 : SUCCESS in 1h 04m 13s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/325ff3e8c1354025a37705b3d54111b6 : SUCCESS in 58m 44s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/cf3380e0ab594da3b08fa611037f05db : SUCCESS in 1h 08m 25s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/e764a06939ad4697bec01528c6d4e016 : SUCCESS in 1h 44m 31s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ac49453ffa3e421da5c2354b43e89b53 : SUCCESS in 1h 51m 00s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/829cc705039b4a4580bee3e67fb91159 : SUCCESS in 15m 30s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/27d26ad6918b4707a0f5dc86f1bc931c : SUCCESS in 15m 19s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4849d4f424f1403aaea0e0d4aa28862d : SUCCESS in 14m 34s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/88ef0d3fb4574e2795696f99a8eac33f : SUCCESS in 19m 01s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/08f0db0aeab04732bfa06b4b38bbe0f8 : SUCCESS in 42m 34s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/603a858650cf46e0b6668193e0d01108 : SUCCESS in 28m 45s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/b29d40e7cfba453395dedf0e96cfff07 : SUCCESS in 14m 32s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c4f7704899c0496f8a6d90b03f788491 : SUCCESS in 14m 31s","accounts_in_message":[],"_revision_number":6},{"id":"9ecf7bc9c34d00c28bde22af4aac3f6ac5043722","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-07-06 10:22:16.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"95596a57e31f5b3bbf4cfd7fd4b7e653cb46e949","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-07-06 10:22:48.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"09683563530b9e3a759e9714accfab3454b0016e","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-07-09 15:51:35.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"7de3d2bd5a692f1d1c1bb4f3fec9341dec78d39b","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-07-10 10:29:21.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"7fc945813f7268111c59c84c7e9e9c3096a5991c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2026-07-10 15:02:22.000000000","message":"Patch Set 7: Published edit on patch set 6.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":7},{"id":"83316bf2e5430e43fee969d1f057fca0fd74bbb3","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2026-07-10 15:02:41.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"2807acb69d218be76f94908d52660050f4f6f708","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2026-07-10 15:12:57.000000000","message":"Patch Set 7: Code-Review+2\n\n(4 comments)","accounts_in_message":[],"_revision_number":7},{"id":"77707c4991f501e1de4bfc7ffb6b290c1d676c7b","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-07-10 16:01:53.000000000","message":"Patch Set 7: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"0325121e16cfcf9423965a3f6a772c008045292b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-10 16:52:55.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6fd86f20cd544cb987a5c801e5a37004\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c11a679844854afeafd2ab0c59a5a457 : SUCCESS in 9m 52s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/c3bc6a6a743b40f38b1b9154e11c4a11 : SUCCESS in 9m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c967ba9e29404cb3b9f5140b4e829193 : SUCCESS in 4m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb37028f1c4246298fee389158812d75 : SUCCESS in 4m 30s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/71815dfbaeb64a6ba1f1f75aa9c8cfa3 : SUCCESS in 4m 14s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/fb3901ed6f324fe5ab838f476265b5bc : SUCCESS in 6m 57s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/411b8b80a4384aa0842c34a172db2a96 : SUCCESS in 6m 28s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4979edc8afd24041b42df0e3af9de3b9 : SUCCESS in 5m 09s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/521e8f06ccc1497da1ce7419102dee24 : SUCCESS in 12m 54s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/635c95d7b9d34c6eaf795982dfe7a950 : SUCCESS in 6m 45s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/dc8ae2201dc94ce5ad4f78d7c8168c82 : SUCCESS in 12m 36s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/61cf0cf3119b4aa1aa2f70c9174aabfc : SUCCESS in 18m 31s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9ee6fe60bf424c9980940e7554cac740 : SUCCESS in 14m 23s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/88885df6046a42a2b0b82a8719daefd5 : SUCCESS in 18m 47s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/ca4b7febcff644589282a1cbd5ca8152 : SUCCESS in 20m 44s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cf32bbfd27fc4ddb83f2673d458ac388 : SUCCESS in 18m 47s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/b211f59bf1e64868b25ec61e23ef9a62 : SUCCESS in 11m 26s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ac3a786e686e413491cfb3da317f4cdc : SUCCESS in 17m 46s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/7ff3b087584645da99bf3ca491e1dccb : SUCCESS in 1h 10m 33s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/0fc6b962b53b4dd9a6faab858c2bf366 : SUCCESS in 1h 06m 00s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/0a2772bb371947fd93dff558a5e0b74d : SUCCESS in 1h 12m 39s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1b8d90beb5b44f4a9bddeb3babf19084 : SUCCESS in 1h 10m 49s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/dce9f3607004497b9784aef7ad11b09e : SUCCESS in 1h 12m 20s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/6c9cc1746f42434496c394f239d4f8b9 : SUCCESS in 1h 48m 36s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5e03f4e22d2741beb1fdccf8be14c633 : SUCCESS in 1h 48m 21s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/8c44498579814248905bfca70f1d4e2e : SUCCESS in 14m 48s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/767c8c35a2444fd7a369072965a38701 : SUCCESS in 29m 51s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e5601b8972bc4bae8fe8f9f9ae4cc471 : SUCCESS in 24m 22s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/84b7ac0cf23b4eb68084756fdacf312e : SUCCESS in 14m 18s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/54948a0592ca4af0904b9ae72bacc742 : SUCCESS in 22m 43s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/198547d70f4a4a14bc7eb17dcd355515 : SUCCESS in 28m 40s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/bb845c0030774b979a623e6e8d95adfa : SUCCESS in 23m 09s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/91ee098c8b3b4ae49399b644dee9ad4e : SUCCESS in 20m 29s","accounts_in_message":[],"_revision_number":7},{"id":"f495d6adc94858502cd46dee5f5d08f9b678bd10","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-07-10 18:12:43.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"814870cd9ed4af406634d5a73025a5cb25a5baef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-07-13 12:56:51.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":8},{"id":"61555a63badb6e1e9abd9b70d79989dbccd5a1dc","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-07-13 12:57:14.000000000","message":"Patch Set 8:\n\n(3 comments)","accounts_in_message":[],"_revision_number":8},{"id":"f2325159985124434d6ac339bc8e2686b046b6c5","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-07-13 14:47:06.000000000","message":"Patch Set 8: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":8},{"id":"82969127b669bd5981daa1e6c692a008de50f618","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-13 14:48:40.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f75bbe84b15a483f9d1850146c241e63\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5ff9ef5f4514728b08287eada2cae13 : SUCCESS in 5m 09s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/aa3f40a056254991bcf8635c9c57c2f5 : SUCCESS in 7m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f9dba248736b4411b2fc1c565f7d18fd : SUCCESS in 3m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9adf9daf377d42e284b81948a57a2f4e : SUCCESS in 4m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/288b6974131541bb9453f5e7c6477d18 : SUCCESS in 4m 04s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/44bedf0bbe274319867f1fa0cbc6317d : SUCCESS in 4m 21s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/320735a65a074e3790d17597d6afc1eb : SUCCESS in 5m 53s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/85b02605765d4f70b47a9c229a83ef4a : SUCCESS in 6m 45s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/2da057be90594038943fd813f375ec43 : SUCCESS in 13m 51s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/dc25a6e695e94765ac121417eeb4bcc4 : SUCCESS in 7m 55s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/9e07e1f645714d769a36e61b61212916 : SUCCESS in 17m 34s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/dd51584ae9e34fccbe1e2e42f0442a57 : FAILURE in 12m 35s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a422907d62fc41f7b163360c79c03547 : SUCCESS in 17m 07s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/39cb2a6f28594458a747695b433e155a : SUCCESS in 17m 52s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/51fcd550f43c4d789834776b6f25365e : SUCCESS in 19m 53s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b101760cfd9940dbb93e21b483ab3c3a : SUCCESS in 19m 20s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/7fb2871985d94b88af122ec0d7c4b486 : SUCCESS in 16m 39s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/aaca53d1c05847e99df4a7123c2252f9 : SUCCESS in 18m 48s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/7248cfbf609841e6afdf9b14fd04fc13 : FAILURE in 51m 32s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/e3c20d34483b488da3d24e260f2115a9 : FAILURE in 49m 57s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/d644147a39464608b507d690420956fd : FAILURE in 57m 14s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5adc1f80fdb54f1ea9f207e96e3c5652 : FAILURE in 52m 00s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/f73e960b091745f2bf8861a660e1a297 : FAILURE in 47m 55s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/f3e70cc1812a434aae49c810873681a8 : FAILURE in 1h 50m 53s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/865fc6f3cf614a60b9c73f40f30ed866 : FAILURE in 1h 43m 18s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/0a9cb4b3183749e2b320d65a2b15d03c : SUCCESS in 25m 01s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/edadf9ec3a924878bc973b1267c0b65e : SUCCESS in 21m 07s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b8ab2bd1133749f8bdaa2a834278eb2b : SUCCESS in 20m 56s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/fd38acd0e797471ebd70c1f062cfc570 : SUCCESS in 22m 54s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/5020342090354cd0a208bd9380222394 : SUCCESS in 27m 18s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/65d66651d7404f4b8536f734b30f2942 : SUCCESS in 43m 24s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/586c31ca853c4970910a826792032b66 : SUCCESS in 18m 13s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bfdcf57cb78a4b24bb1b2a86fa13c71d : SUCCESS in 25m 27s","accounts_in_message":[],"_revision_number":8},{"id":"b9737fd99234f9026bd0989273220cf045c61787","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2026-07-13 15:23:22.000000000","message":"Patch Set 8: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"074b75f80789cc720bb80624e252b36ca8e429ce","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-07-13 16:28:53.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"ed788976bd83fa5787cffea9414ea32a75627b10","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-13 17:12:05.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0dfcdd1b1af442e69dbeec90d0d1db47\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18d05b9bc1f94c51a8a9ef60d4de2388 : SUCCESS in 11m 04s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/bbaf496ed1634facb0ce3654d90d7d02 : SUCCESS in 7m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4ffe59b00e5f4cd398bd5a4e11ce4115 : SUCCESS in 2m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/18119624dbb542c29cc4144521ec99b7 : SUCCESS in 5m 21s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b693141d849548f0824574c7a144af0c : SUCCESS in 4m 09s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/be7631252dd249e7a1d387c82abc6d51 : SUCCESS in 7m 46s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/90257e274d954fcd89d879d11e915dfa : SUCCESS in 6m 09s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2028b5304d2d45798f6c07994ac4fd22 : SUCCESS in 5m 20s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/2983ff9cf97e49b192eda5e293f9b4e4 : SUCCESS in 12m 54s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/8d6cae30eec742188cdf9bb8bdbcd496 : SUCCESS in 11m 53s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/cd37c5695f1b446bb9448b82464bc4cb : SUCCESS in 14m 08s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/08ded8a6ba1d4d34b2fe9d31ff2a84f3 : SUCCESS in 12m 28s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a0ac6b3f6d3d4e34ab6e9d4dee3c1c7f : SUCCESS in 20m 52s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/96102fc77ca24d389790b615cecaa9c9 : SUCCESS in 11m 48s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/6d7b418f5ff54f349a8f1cf28d9e6877 : SUCCESS in 16m 15s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f288be47a4a84c3abfe3346f0d975688 : SUCCESS in 20m 02s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/78c5508ad6204643afc658a2a8d113ab : SUCCESS in 24m 32s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a0398155afb746718737b2520d4bfcfe : SUCCESS in 22m 43s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/17e7f07e3b874859909d8fca497389be : FAILURE in 50m 52s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/a98b73a2abb0401a825da1e354b529a2 : FAILURE in 52m 18s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/68f8d4b97d7945f4bd60bbbc03b73937 : FAILURE in 56m 07s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8e02cb7fe5e040f9aea33e517cdf732b : FAILURE in 49m 22s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/3c07789e6a164bc9abdd28e88317da25 : FAILURE in 44m 16s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/9d0842f4640241d9ba2abf3d46641fd4 : FAILURE in 1h 46m 57s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6265463f0bba4c69884d81c958312cc6 : FAILURE in 1h 38m 14s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/a75b07b1a7e448dd8bee0697b52c6a27 : SUCCESS in 17m 07s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/1cf4607d2a74487999eee8e24a8eb5cc : SUCCESS in 17m 16s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4861d0e247624ee7a964ebf285da132e : SUCCESS in 28m 07s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/de521ebf30454d43bf19e319d3a5c09d : SUCCESS in 22m 11s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/03f4ceccd8764ca4acc21fbf93d8cbeb : SUCCESS in 21m 47s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ba4a37cb6bb44d4ba09eccdf81bcff76 : SUCCESS in 38m 03s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/a1c397b9d43a4cc09a2cfdfa13f758f5 : SUCCESS in 16m 14s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dbc35c1bed824b04afc61b48b329aacf : SUCCESS in 28m 48s","accounts_in_message":[],"_revision_number":8},{"id":"011a5dd44a5ed341fd30e53b00acb856e4c27442","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-07-13 19:04:20.000000000","message":"Patch Set 8: Code-Review+1","accounts_in_message":[],"_revision_number":8},{"id":"671664ccca48e734869ca965e4c13f921752ba8c","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-07-17 15:35:36.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"c9c41ad2bc2c6a52fba856c7d4d70eb7e9c3876f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-17 17:25:00.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/538490591a5f4480add759f00fd313fe\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/020e44d520b24aec867d09c4d7602b38 : SUCCESS in 9m 26s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/d5dc703f0e7b49fe80f53eeccd9a1d0d : SUCCESS in 6m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0d5612c0d7d6442896b4eced8192b7c9 : SUCCESS in 6m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2606297493b84f35b72fb516d230da5f : SUCCESS in 6m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/313301c967984b63ae89c2bffa0f4797 : SUCCESS in 4m 34s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/4e8f5420f59b4400afbd39a4bdb5870e : SUCCESS in 7m 32s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/47caeaa14804423694b243706ab70f5f : SUCCESS in 7m 40s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f1f4c358c9e145da9a78a05db4831a39 : SUCCESS in 5m 46s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1125cde5a9fd4bb3aba2ce8c5903045f : SUCCESS in 12m 59s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c49fac9de96140b7851e221fcc2db460 : SUCCESS in 14m 30s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/93e142f7a4614b9b9f5b85b72c1db482 : RETRY_LIMIT in 6m 15s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/54099afb29c44406801a6ea164b34a6e : RETRY_LIMIT in 7m 49s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bc8e76175cff4506ae2e504eb665b49b : SUCCESS in 14m 10s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/d7f93c184b7e48408668d4c9dc16bb11 : SUCCESS in 19m 44s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/0c295017c45d4840bd52c16b0242393d : RETRY_LIMIT in 3m 45s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/db4316fc1b094bc9b6aadf012a3f34ca : SUCCESS in 20m 51s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/b829ec974cc4432baef8b2b401816b71 : RETRY_LIMIT in 5m 44s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ffceefd8c1014545b4b3bfd011eb4e70 : SUCCESS in 17m 18s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/5cb098ebbb8c44158b09df6fb5caed74 : RETRY_LIMIT in 5m 12s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/03bb9d4acd94479e8fb89ddc518c62a0 : RETRY_LIMIT in 5m 13s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/7394f176465e4b56ae2c3db171b63420 : RETRY_LIMIT in 4m 56s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a6992f29d6584082902248d03952be7b : SUCCESS in 1h 04m 15s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/e75b340f39924f50a127d948c1fe3161 : SUCCESS in 1h 04m 48s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/83a30cf1ae324ad49dc657ee9cfba4db : FAILURE in 5m 05s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9dbbf86e21b0485c8ba433c3a2c2c63a : SUCCESS in 1h 47m 59s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/0fb5332f60c54745b7f3986a37331e42 : SUCCESS in 19m 31s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/a3e189457e774d34a7615634656595b6 : SUCCESS in 22m 22s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/387a4e93f2244ef1a3f94dbe149926ee : SUCCESS in 23m 29s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/c5375df0cdce479f8807b0ef15f942a9 : SUCCESS in 25m 57s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/1d15904ddf754a4fa1d39652589d7af0 : SUCCESS in 36m 32s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b7bd3adb44984e0b9e0f1aa3c498ffd0 : SUCCESS in 36m 06s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/34eee936c8d44e4d95e861f5816614f1 : RETRY_LIMIT in 3m 10s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/53f1f01650e64c6c869f5547074f4268 : SUCCESS in 23m 25s","accounts_in_message":[],"_revision_number":8},{"id":"d575be5c4d019f91887bbbc7c0909a564e76cde0","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-08-25 12:55:21.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"8ebcc7361c13ea727cad4e116e00fad935c4aa3d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-25 15:05:46.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b9de3fa77d114931a895b162183d1f76\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84f314bab77c499db4f6d2a1ad083992 : SUCCESS in 8m 31s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/4ead728ed89d404e82fa9b7710daec38 : SUCCESS in 8m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3a7a607ae9464b9aa62441313e0e173b : SUCCESS in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/364caf6881b64aec959cd25a403354d5 : SUCCESS in 3m 09s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d03472a8c4ee480d8de416e9f1463cc0 : SUCCESS in 6m 07s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/9919f42798e44b3db3a1c78d9eebfe67 : SUCCESS in 4m 27s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f076fe48d7f544f9899cb5596610f8d3 : SUCCESS in 7m 07s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b2b1d4a36d534f3eab1450238dc653ee : SUCCESS in 3m 19s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/586450dafa49434bbd2948c5298a686a : SUCCESS in 14m 25s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6731e99656a841dfba0b1c3f4376b040 : SUCCESS in 11m 57s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/17f7a02c1a6c4f638ad9f4e9de617d22 : SUCCESS in 20m 13s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/5aaaebe5f9634de6992f5d5142fd3579 : SUCCESS in 16m 57s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/adc1bd0c440448268f59be1c08e6ed44 : SUCCESS in 18m 41s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/7536d7d8cf464e4ab06cf1daf0d77522 : SUCCESS in 17m 04s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/04518fbbf6d24eb388c5352d00d8ab5d : SUCCESS in 17m 12s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/afe34984c6de4883be23558ca7444d30 : SUCCESS in 22m 58s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/9e5c44ba10114523b933d58117bbab6f : SUCCESS in 20m 39s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2bf38dcce0b74a9990636ba09268bad7 : SUCCESS in 16m 21s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/2b96f50771ca440791209a712811d826 : SUCCESS in 1h 16m 38s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/2ce2e909e6bf4e3f9ac54cb8b443007c : SUCCESS in 1h 15m 09s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/7e9f852b5b74419c83bbf8ad3f9497a7 : SUCCESS in 1h 09m 02s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/50fd54300bd94afd9c83650067d94225 : FAILURE in 1h 17m 18s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/e686742692284d03a9b8530a673a2ea1 : SUCCESS in 1h 17m 50s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/c9ed49e014bb4873abd0a73d899e19e7 : SUCCESS in 2h 03m 06s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4eb49140c4c84364bafdc72ab2ca4796 : SUCCESS in 2h 07m 24s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/8ae9b859ac91470fa0b4ea12ce64ccdf : SUCCESS in 30m 59s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/989a25250f204f818bf35d6d64ca6ea9 : SUCCESS in 24m 39s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/12afb17a75da494189fd242fff2475c6 : SUCCESS in 25m 31s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/9e422f9d6a5a4d16b8a9db7771b2bdc5 : SUCCESS in 19m 44s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/4e76aa3438cb4191bb98fdfb39634bf3 : SUCCESS in 26m 52s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c151152bc8114e70b4ff289672559138 : SUCCESS in 41m 20s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/772681d8bee4486599363c651beb5d39 : SUCCESS in 24m 19s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3396062c590b499a85c228467b26b076 : SUCCESS in 18m 30s","accounts_in_message":[],"_revision_number":8},{"id":"806411b7203e083e6bdf571a2854321ce0af3597","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-08-26 10:36:52.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"8102fe1db73c31332f4e27bfb8d39762f17b5bb9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 12:38:02.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5a98709ba38f4cc7992292015d103375\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ba237ec207446598efe9da4a62dd063 : SUCCESS in 9m 30s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/4e96a190bc364a37b315a7dddcc263e6 : SUCCESS in 6m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bced70a251084d54baaff9cbd2aa09cc : SUCCESS in 4m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/095910487ae4485cb23ba10b29a5816e : SUCCESS in 5m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0c451e65698d46119eec0eb82e317953 : SUCCESS in 2m 37s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/5ff90e38b51f40c1a8131eda47bbedc9 : SUCCESS in 9m 45s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e980f03944d841c48be04a045d6b3f1d : SUCCESS in 5m 55s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d82fe3c086634463a2c41f6b15f4903a : SUCCESS in 5m 15s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/58da4da811b44c1f8340e552cc64d883 : SUCCESS in 11m 27s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/18db4506de2648e29ec255af442dbf9a : SUCCESS in 13m 37s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/23034092d85f463ca29f9dfa21d016f0 : SUCCESS in 16m 25s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/e8593feae76643f69dda1a51b71fd4aa : FAILURE in 16m 49s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/57097366faa24dd08eacca4443807d39 : SUCCESS in 19m 00s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/1397202f27834297bcd0799e00ac2440 : SUCCESS in 17m 35s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/6b11f2cfe7d84cb79a4927b59b4fc0bd : SUCCESS in 15m 02s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/519c0f0acda94c36aba7268fce32f509 : SUCCESS in 20m 05s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/cbcfdba0ded742d9a4ad313cdc557ba9 : SUCCESS in 19m 21s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8bf4c0f098a6414eb7fe281f2f13c52b : SUCCESS in 14m 54s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/1964a271370047f996a6d093b62c770e : SUCCESS in 1h 13m 57s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/8b3ead65d4d74ba0a9be5e4a2a4b8c10 : SUCCESS in 1h 10m 13s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/56d23b70032d4f5c96cd383a1c2326a5 : SUCCESS in 1h 06m 06s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0125127a2eff4e7d9e9e7677b4fc6bae : SUCCESS in 1h 16m 10s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/a055d01a83c9455084dcf962615cd675 : SUCCESS in 1h 17m 54s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/724450824fa94b2c84bc68c9284b28df : SUCCESS in 1h 57m 00s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fa3b9a4ddd114f7097969d3536d18402 : SUCCESS in 1h 59m 53s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/b0bc856d38db4b78a895b99591946b3d : SUCCESS in 18m 39s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/e5cbcd02304846ceaa329f65e9c60303 : SUCCESS in 15m 31s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b37bda43d9844a4186b15167ff6c7ff3 : SUCCESS in 24m 09s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/ea20bb92b78e4f04bfd7885ead77aac0 : SUCCESS in 13m 25s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/a6d23332466449a5b172b855945b7dc7 : SUCCESS in 23m 29s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0c9a9542ef4447f88d648b1c9b5165b6 : SUCCESS in 35m 35s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/249bdbdf5af843c68f1c589adc396af1 : SUCCESS in 14m 15s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5efc81d269da4215bfdb46e2328d3e4c : SUCCESS in 20m 37s","accounts_in_message":[],"_revision_number":8},{"id":"2c22154cb51a20ed5f7a36615eb8f884f569d4f9","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-08-28 10:23:06.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"fc6d4f3b515eb0d6733e23ce0494880ed5b126ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-28 13:11:46.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/33298a2080d542ffa364e26db9fface8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85ed891638694bd8ac9e388c691be79f : SUCCESS in 5m 13s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/c91b33905a9547ef86755703af2a8de2 : SUCCESS in 4m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a3af18702d93487ebb5fb92acd50cd2d : SUCCESS in 4m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ca0c8b82b854b55a71e563ab856eb42 : SUCCESS in 3m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b048cd1fdb9643d8b19fb4cb28524b6b : SUCCESS in 5m 04s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/35a18963beec4e009e167f49feba8093 : SUCCESS in 4m 09s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e188832cf5f043ccbc9e389ff1bfba7b : SUCCESS in 5m 47s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ab5b7d7755dd4e74aad915d284dc5d8a : SUCCESS in 3m 24s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/60a4da1ac8fa4bcdade5f730e3831f02 : SUCCESS in 12m 30s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6194d3d16c6546e384a9373bebff286d : SUCCESS in 11m 35s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/151376d69c854c37a7b720ed9c174dc3 : SUCCESS in 16m 25s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/7d2bfe9d8bef49228d08574e59f720f1 : SUCCESS in 14m 28s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/21dcfbe37b884411ab6bce41a2df33a3 : SUCCESS in 16m 26s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/16e6a0241eda40b089f4a90fa8f59ad3 : SUCCESS in 15m 38s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/dcff18c760cd4be287bc9f5df4fe9acb : SUCCESS in 16m 51s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f2ce2adc6586492792e397aa21382757 : SUCCESS in 18m 58s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/29b10786f8d14263b5298502f78521f9 : SUCCESS in 18m 34s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7c1c6a82954a4e0397420320da94af7c : SUCCESS in 18m 18s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/7d25955f42354aadb26b388798af553d : SUCCESS in 1h 10m 24s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/53530d287b6b4951ba145636349ade22 : FAILURE in 1h 17m 25s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/b6fb27b3f60843bb9ff08569d3637285 : SUCCESS in 1h 09m 54s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/13535b8ad4f64da78654f8928b392acc : SUCCESS in 1h 51m 40s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/00609385dc39440abfe7f266de3eb5ff : SUCCESS in 1h 43m 25s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/d384356fa5e54775b8bb8c44acf279ab : SUCCESS in 1h 49m 49s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e7ae09c69906484ea7aa02040f69aa78 : SUCCESS in 2h 47m 17s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/e134c022f81e49199e6b5aed9f21fcbb : SUCCESS in 19m 39s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/b44dd2492d4545ca8265efdd56c12cfa : SUCCESS in 26m 30s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2d124f84b87d480f8d8d05ef17d92734 : SUCCESS in 24m 19s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/d086481887fe46c3b888447692b1e1de : SUCCESS in 16m 23s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/c50c6485fe57439fbc3ce9f7a733a9b7 : SUCCESS in 22m 40s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9ea726829e79486693deb911fe6308fe : SUCCESS in 33m 44s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/2645b7eb08b64da891ee547bd26055ee : SUCCESS in 20m 26s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/254d2d2c3225427586294898dfff3832 : SUCCESS in 20m 17s","accounts_in_message":[],"_revision_number":8},{"id":"c38aa2ffead9820b58bd02efea98faa758f70938","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-09-01 09:01:54.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"2716dfa113c11619c39bee633217ca0144cef2a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 11:43:59.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/896f46fc935d441fa38861ba53a350e7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14fffadcff494bb38dfc035dbcd359c3 : SUCCESS in 4m 45s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/d9deb10f1bbf41bd84ecfbbc531ce622 : SUCCESS in 8m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/32b42e19d7d64bbaaeeb476c01420020 : SUCCESS in 2m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5908c053a5f4859b3723545abad4788 : SUCCESS in 4m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3e15c804d3034752b127332689bd3b84 : SUCCESS in 4m 24s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/1d38e0630d39468e934165fc37993d77 : SUCCESS in 7m 17s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7f425c9ece0248de9008f3b0955ba8d8 : SUCCESS in 3m 21s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/55701f31bc3e432fa1e32650502d3a42 : SUCCESS in 5m 44s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9cbad266b9f24181a6c059ed52dcafaf : SUCCESS in 13m 31s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ef8f1b3f26e24bc9bfaccfe4fb3e2c9e : SUCCESS in 14m 00s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/a3a0a2b20c3b43a8a1d6bb80e7279045 : SUCCESS in 19m 03s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/588a60df4a4746faa4f28ce1633912f2 : SUCCESS in 15m 24s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/38813875374d483790f2105662b89359 : SUCCESS in 17m 19s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/2276f4e3a11449c08660aa1836f8c73c : SUCCESS in 17m 41s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/4fb153650e534a03a268ab9ced474532 : SUCCESS in 17m 45s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/31bd4b45896e4d28979f5ebc227da7af : SUCCESS in 21m 04s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/68b2c61dc7f6430f889dc070d46ebf6e : SUCCESS in 13m 57s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8c0fc53fc7b04d218ff92cd527f80413 : SUCCESS in 19m 56s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/67280e2a849c4b129ef20e5b0a2fe22b : SUCCESS in 1h 07m 49s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/fc783b463a824ecaa2b6fcb48c19ee50 : SUCCESS in 1h 11m 06s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/7b6862c7f1f0462db4e70b8af6ae74c1 : FAILURE in 1h 30m 15s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f9466640a3de46adb70ef6ff7b673034 : TIMED_OUT in 2h 03m 23s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/77f85d0427244da7add1c4ae132abad9 : SUCCESS in 1h 46m 00s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/1e44c3d3cb824011ad4c6e64b8368697 : SUCCESS in 1h 54m 59s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6e1b0a87a12c45c4999bf3b349c9c297 : SUCCESS in 2h 41m 01s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/1579756c82c44703806075a3c04951f5 : SUCCESS in 17m 10s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/0ec924aa47954fb39b6b27e9ab014b21 : SUCCESS in 29m 09s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/296edc7148da41538ef74f71ee9d8ca2 : SUCCESS in 16m 56s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/a7c1c74e85f14cbabb1e3bfd1b44f860 : SUCCESS in 24m 39s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/d4ec92060ba44ec8b1be241ed09c995c : SUCCESS in 28m 01s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c68ed3f66d6540cc9469d0621f10d7f8 : SUCCESS in 32m 52s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/0a5521ba35f14a1086c997414310184b : SUCCESS in 21m 15s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/625cb106f1e747ec9ba2a3bc6ba64f65 : SUCCESS in 23m 34s","accounts_in_message":[],"_revision_number":8},{"id":"d1b2710fabe69d1731e14ea25613094ad371d212","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-09-02 09:16:38.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"450a859ece0417963b0a5b97800090b6651f9f16","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-02 12:11:17.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e23b6e4f315a4d27a0aa5e8d62c0046e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d13171d55417499cb25e94f8fc1442e1 : SUCCESS in 6m 17s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/b4dd231364da4e278bb1e9a473a3bcb9 : SUCCESS in 5m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/791ec6e269354c9395de47aadd1a036e : SUCCESS in 5m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/432caa07ff254d09bc02cc0c4b74439b : SUCCESS in 3m 59s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c66ea733ed9b4d5a9a61e3992db098d4 : SUCCESS in 4m 01s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/cc2e00790c314d608013a57ea6220eaf : SUCCESS in 5m 31s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3bdcd708e5dd406ab5b3dcb4d328dd2d : SUCCESS in 3m 39s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f46d1ca5619940baa5e35ba440e2e42c : SUCCESS in 6m 44s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/98a83f499d0b40cd9d5ed9f86f5c8c24 : SUCCESS in 14m 39s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/bc8ebf1d0da54e14870c9a0db04a5bc0 : SUCCESS in 8m 14s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/f9c6dc35c458445f84fd2bf43dfaf759 : SUCCESS in 21m 39s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/e7e0e89a8e024464b0d1663c1026a310 : SUCCESS in 16m 06s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e832401fd01249fd9e60c5f2ab0e4a08 : SUCCESS in 17m 24s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/c27439f630294b40a6f37f241cc56e5e : SUCCESS in 17m 38s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/b9f513b930ad437a8a1b900a9f19ab97 : SUCCESS in 17m 44s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a4d3ad8b81444329aa55e3256199c8e7 : SUCCESS in 21m 43s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/0196c728d91c4b4c8ff4262cc803f2ba : SUCCESS in 28m 09s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e6b6035548574774a3a7c709b1fbd0a2 : SUCCESS in 13m 51s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/0fad342ffb0c491f921621ceb521ae9b : SUCCESS in 1h 03m 42s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/09e9f5726eb5486a80b213f8f357b576 : SUCCESS in 1h 06m 39s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/9caba1ac0ac141d2b3ff1fbfa40c8161 : SUCCESS in 1h 00m 15s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6816e50d1fb5459b97ecaee3b3c20467 : SUCCESS in 1h 46m 00s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/7e5cfeef02c24dbaaafec65feee0daef : SUCCESS in 1h 49m 52s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/166a9f638b574cf39c25c5d54449a45c : SUCCESS in 1h 56m 21s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f6094662c1134c0bae62fc78349a6a9d : SUCCESS in 2h 53m 29s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/c7a72375afa94605bcb4c5326e178e16 : SUCCESS in 18m 02s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/7ecac07ba9df438395487bab439784c1 : SUCCESS in 25m 05s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0fa35c78ba2c446d859e1bb8f6a77b5b : SUCCESS in 24m 04s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/ad59cbb3dffe4f6ea1b207416a00c63f : SUCCESS in 22m 31s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/e87ad7470f794acc908ad6d2414c2c2f : SUCCESS in 25m 33s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b8716e5f4a29476986e12430955cab23 : SUCCESS in 29m 20s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/a8d103e5940c43a399b6d0618546b05f : SUCCESS in 15m 57s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/36ee8fdf7e8e42fd8ef656108f7da9a7 : SUCCESS in 16m 51s","accounts_in_message":[],"_revision_number":8},{"id":"834db898e8d5b0db293eed3e0ee03f8859a144a8","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-09-03 08:43:00.000000000","message":"Patch Set 8: Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"3252d77af8e570c9eb6c422c4ebf7b7f8497f17a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 08:43:32.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"31992d6555f868d458b94ef7b63ae2e2c0a7db25","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 11:22:25.000000000","message":"Patch Set 8: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f46785fe236e43f68226cd1816094cb5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4122b489df2a46568d6b8377779bcf1e : SUCCESS in 9m 46s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/3ef30395836843e7b09d4f527286cd80 : SUCCESS in 4m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/feb54506e5f0476fa644745d71927907 : SUCCESS in 4m 27s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d2d17ab3c1e44886a5efe4ceec007ee8 : SUCCESS in 6m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ef978fe132fa4938948b12d5564ba26f : SUCCESS in 7m 55s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d2ac8f9bd6584e19af53e1c2161f25f5 : SUCCESS in 7m 49s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/fa83cbb11ef54e5faea21de604bafa79 : SUCCESS in 14m 40s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/762d34e480f840a58939bd55980e0025 : SUCCESS in 13m 27s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/c1407fdf6c024641b5431d03b9086589 : SUCCESS in 19m 49s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6fe6660d76ea43e79424d9fff73f3bce : SUCCESS in 21m 17s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/d7b9cd7e6a0640309803832413b6ed7d : SUCCESS in 21m 03s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/958081a0c6e54b4083320ceaf6f8d496 : SUCCESS in 13m 49s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/6aa62043268e4eedac5971a97ac9db39 : SUCCESS in 24m 00s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5b251dfd1318496599d161c10dfdaa1d : SUCCESS in 18m 56s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/ad6a71af5026410d8863cffb6cea0acd : SUCCESS in 1h 11m 16s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/79b19c27caa44966ba166e5d66a231d5 : SUCCESS in 1h 12m 45s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/c03b5b238d474a83bd7a77dbc85af2ea : SUCCESS in 1h 07m 34s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/62b9865e21884d88af5d39f276526612 : SUCCESS in 1h 37m 44s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/6178b6c8aafe49c9915639629ecd7cb3 : FAILURE in 1h 55m 24s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/9450fac8990e44a9bc1b456df75c35cf : SUCCESS in 1h 49m 13s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a25b5bcbc3a24423b4a56dd170217c13 : SUCCESS in 2h 38m 05s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/f5c87e87100b47719366fcc6213ecff5 : SUCCESS in 29m 55s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/1365dc001d84430d8e342e010e33e035 : SUCCESS in 23m 57s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fbd5e51610eb47849ccc3982c84fc8e1 : SUCCESS in 22m 20s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/a5a1928a5875432084ec2540d3ceeb95 : SUCCESS in 18m 15s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/4e4f4baebd4740f8bfc653892d5feb6e : SUCCESS in 37m 50s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5d4bbbc9d87d49b198d618a0d712e67c : SUCCESS in 29m 59s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/4ce074d34ee74c66964be76f4408569a : SUCCESS in 22m 56s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/242ef2ae4bbf4fe3bb7b2305d32cc040 : SUCCESS in 21m 51s","accounts_in_message":[],"_revision_number":8},{"id":"e558d0fc18ae5db45ce274c2a82859f0057aaa0a","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-09-03 11:57:12.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"e3e17e905279750ce71c695a2ccc4869c02c6899","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 16:46:01.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1e18fa616a4d4c0a9427bbd4b5aad8d0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/915fd54857ae4b70996ebd49514d75e7 : SUCCESS in 8m 27s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/fe6c6413fc894e1ba24431cf0795552f : SUCCESS in 11m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/79851a824dd245f5b77d01e26f0133a0 : SUCCESS in 5m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b71c09891b3347e1a55c0a3aa4f77f31 : SUCCESS in 5m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/dd0eb20ad4e6420e8dae866f9adc3cf1 : SUCCESS in 4m 55s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/af9d8e7d67fd42ae841cfa9caed9a9b1 : SUCCESS in 9m 21s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4f0dc6e13fe84429b23478f41837fb8d : SUCCESS in 5m 33s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c6d09d1fcbc344c9991f88ce5bcb2065 : SUCCESS in 11m 13s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c7ef095ccace4f0f899891703f93b8c7 : SUCCESS in 19m 23s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a71f2f1cad8d440bb13608dac9677556 : SUCCESS in 13m 07s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/1cef9b7ed84d4c66b1c93fe104c2aed1 : SUCCESS in 15m 58s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/e3046cf4884c495cb99e2c77b22d99f4 : SUCCESS in 24m 00s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a648e657ef7a42e8875d8352aa8d72e5 : SUCCESS in 16m 47s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/4026c44d6c994cc1844c6dade5e3d50a : SUCCESS in 14m 32s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/3afa79f264374f528ab6a982943173ff : SUCCESS in 20m 02s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9926965d0462451baf8d9d81f17f6a6f : SUCCESS in 25m 39s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/4fa74f595d5e45b492793bfdb5c3e4d8 : SUCCESS in 21m 54s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ede3d5a83d4b440882a976f33c893b7a : SUCCESS in 23m 28s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/dc7538b1865140afa1386e58d43fb19f : SUCCESS in 1h 14m 34s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/50b2b6ccc3944d0ca9446058516266e4 : SUCCESS in 1h 14m 22s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/1d6291a8ae894f71b9874619bfb92241 : SUCCESS in 1h 12m 53s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7e8df2f61a874e569821d161b23ad2d7 : SUCCESS in 1h 43m 26s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/13d675ce3799453186d529f8f2e9ae47 : TIMED_OUT in 2h 03m 38s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/2731fb462390409bafc76b86f65b3b26 : SUCCESS in 1h 59m 44s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ce88288faf1c41bf8d4e5ae10e93ca20 : TIMED_OUT in 3h 05m 00s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/0a2dd726530d4f2fb4abb645b7f04452 : SUCCESS in 22m 35s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/2e8793e53fef453288b0785205e02a1b : SUCCESS in 18m 18s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/245eb831310a4886973bad75fb7c6bbb : SUCCESS in 31m 13s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/2a11d2c52e704c5596cfcff64e3b0e96 : SUCCESS in 26m 05s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/e98a78a650384b8c97fe0f1cfe1baf4f : SUCCESS in 27m 28s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7fa07742d972441b9070ac3447ff5648 : SUCCESS in 25m 06s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/2ca054a12dfb460aa064a6af9206c4a3 : SUCCESS in 18m 19s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ee0da7615c8744a281556bbfde9b597c : SUCCESS in 23m 19s","accounts_in_message":[],"_revision_number":8},{"id":"4cdae49a47453b05b54c2988717dfd1dc3429011","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-09-08 13:14:33.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"ca70d89bc6f05e8f6f949f42e8dfd5cbef739ccb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-08 15:16:38.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7bb36fba43d74ff89fac6c1adbcbeb35\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7ed3d42618842b0a8e733ac3fab81ce : SUCCESS in 9m 16s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/ad9bdae743b24c059cc496c5a8a33aae : SUCCESS in 8m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d70e6a4731fa4adaaf82a9baaf9f4cd6 : SUCCESS in 5m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/00660f609e9a4a9fb2537c9ae05c6bfb : SUCCESS in 4m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/759e0eda18f141efacbf92beef92337c : SUCCESS in 2m 37s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/61d20def69f449a39acd6d092c559e66 : SUCCESS in 6m 58s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/edf1386215594ca98794c5949497d53b : SUCCESS in 7m 36s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/1e4e4a404dac4aabbdc3cdbff47c284a : SUCCESS in 7m 47s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e75a298929b4432fbfd2301b7d505de5 : SUCCESS in 15m 56s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/957183f19f5043e6b4e1d1ffaf2854bb : SUCCESS in 12m 17s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/5ce984e29aba4e04965f58ecd7e6c510 : SUCCESS in 17m 58s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/4b3d62f566cc4234a39de7a43916cd49 : SUCCESS in 16m 44s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7a5b9d2b02864bd6bb55bf0b9910d2b5 : SUCCESS in 22m 13s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/444fb6126b8b42f58c4d007f2f6ce5e2 : SUCCESS in 14m 55s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/8b9f8c91d7da4bc1820b13b3918d0a11 : FAILURE in 19m 49s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3251d93c01ca4fd3ba27ef3a9d01978f : SUCCESS in 24m 09s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/83f856a584bf44aa87ab8e3a0fcbfb3a : SUCCESS in 18m 21s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e041980c26a14bf090c2e75ce666f63f : SUCCESS in 26m 53s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/9be29644a70a46c19e4f62d6b1547e97 : SUCCESS in 1h 05m 15s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/313b01da1b474d4d9da4b388ff97203f : SUCCESS in 1h 05m 48s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/deedba34dca146aca28b8c97e1e2250a : SUCCESS in 1h 14m 10s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1a01072cdf6d4022ba145222ad812fc9 : SUCCESS in 1h 43m 27s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/7d52f7b19c04402cb0d08079c6fcc431 : SUCCESS in 1h 35m 48s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/fc2db914a23d446da5aa0a219a9b2b7a : SUCCESS in 1h 59m 06s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/45f2df12208d443d8a286f7b4ad365bc : SUCCESS in 32m 09s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/d37590b3f05c4978b06425ed7c59e087 : SUCCESS in 29m 49s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d88a0453654f43b1b6bf3096ca398123 : SUCCESS in 25m 11s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/10c9c39e45144fe6bec2a7d0b8944b7e : SUCCESS in 24m 54s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/384aeeaa46cd43bbb05842f49d2ed738 : SUCCESS in 28m 14s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/11c41adaef024891bb52fdc61f61a37e : SUCCESS in 44m 08s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/f51f62c1ac5143e2a83c367ae1896b33 : FAILURE in 24m 22s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/75a2a4e1e8db401d981b5c031513312d : SUCCESS in 18m 50s","accounts_in_message":[],"_revision_number":8},{"id":"4afd66d0dd79b9082df0779fb0f6c819586012d1","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-09-10 09:56:13.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"0816bc2ca8e72243da09fdce92732375c9f6eefd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 12:07:19.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/53974e60e7fb4920896694a4855b40df\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df784946da3642619458f828b4727e34 : SUCCESS in 4m 51s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/64fb84e7306f406f94cc0dd8918ff541 : SUCCESS in 13m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/81529c827d264f249211b48a44fac74a : SUCCESS in 5m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/edd53f34b7244d9abec200edae2939b4 : SUCCESS in 3m 13s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2b776b07109940e2abc56ea0671fb4f0 : SUCCESS in 3m 00s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/2ea8aeaf50d84c9fbe7fee75de40bb2e : SUCCESS in 4m 38s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8dc1bbcadff742a9bca23078e7958e0a : SUCCESS in 6m 28s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6af3c0a2951248ccb4089793f6a218b2 : SUCCESS in 12m 51s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a1944ee11a8244a5ada53c15d22a356b : SUCCESS in 15m 28s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/9e12473e379343628cc6a506efd0836c : SUCCESS in 16m 27s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/f284f9716b124405a12e79b88cd9d763 : SUCCESS in 24m 25s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/0a8bc8b270af48f3b5d33cbcc2b49626 : SUCCESS in 26m 17s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/474df36f73904140a0d454ddd3bf459b : SUCCESS in 27m 35s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/3b0c19716a6c4ed993b3fb4827affa8f : SUCCESS in 27m 25s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/bf6647f77e9e40ce8315bc037749d772 : SUCCESS in 30m 20s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/22d6a3e7458f479d9d5948ed14d874ce : SUCCESS in 28m 48s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/0f161697114949f0a9d5b91c20164f54 : SUCCESS in 27m 03s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c6a5f01f20574e28b35d672c9992fa63 : SUCCESS in 28m 15s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/063b84ec749d4ae7bbac9cac80d17859 : SUCCESS in 1h 13m 36s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/8e44cf225c854fcb86905d05f91f0afb : SUCCESS in 1h 13m 01s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/460058301e11492093492c0b64e49302 : SUCCESS in 1h 19m 41s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c2bb5e5af8784384a0b0d06dbd7e7071 : SUCCESS in 1h 38m 18s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/100912ccb46c49bdb403c4dbca281bd3 : SUCCESS in 1h 50m 50s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/481604a3792b4843aa8cc8e0f08c8249 : SUCCESS in 2h 08m 54s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/5a10a44421ac4a249da6c4108513284f : SUCCESS in 33m 18s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/64ab9fbb1a0b4a33a26fa7fef3013cab : SUCCESS in 39m 10s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a6b80e8df0bc470880b4e8292c25b70c : SUCCESS in 28m 12s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/4a790168fd564104b2e65a1b76d2c813 : SUCCESS in 32m 09s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/fee295bdcf7c47788da9e53c82883307 : SUCCESS in 42m 45s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e31b24f885384746be30ca0d845cf124 : SUCCESS in 35m 10s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/369898af6ca54b81826a0119663bbd5d : SUCCESS in 27m 15s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/55daf60a70e84523a84c9f1008246a4e : SUCCESS in 27m 23s","accounts_in_message":[],"_revision_number":8},{"id":"b5c2198f6be5fb113334d23db933aee4e001edbd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 12:07:58.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"c2d338f42e01d1397c0ce1dad7efbf66876cfeeb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 15:20:59.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c358a94cecef45eaacd75c5b5e9c2dda\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23475dc8e5dc417ba4a9d7a6a564f08f : SUCCESS in 10m 54s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/511c076550cd468881229c3b6e76a5dd : SUCCESS in 11m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/90c6043564c74120bcacc5213dda392a : SUCCESS in 6m 20s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a14e91ca4f154e57be5d7b9ebc84fb21 : SUCCESS in 4m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/51c7f9e8f2384f9b86026084dc0547a8 : SUCCESS in 5m 38s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d3a5bfd4f7aa4e4d985afd70873bde8a : SUCCESS in 12m 58s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/682aa9fa91384432959e03a2d5131eb6 : SUCCESS in 23m 28s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2e47c7851d164917af238f9fab9c0bdd : SUCCESS in 15m 00s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/7560cd8a7840432da02ba909da7d77ca : SUCCESS in 39m 32s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c297ccbd315b458da45c46056d5c64a9 : SUCCESS in 28m 46s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/f4ce77d3f2d14372926a03b2c6a45327 : SUCCESS in 29m 24s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0637f9f05bf2488d84827bbb6d09566d : SUCCESS in 41m 03s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/bbbd67c9b9914070be105d8a48eab59c : SUCCESS in 34m 08s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/210b3a97dc4b409f8a8f8c9234110406 : SUCCESS in 23m 00s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/a3cb8cb5c6304cfa9fe9876f34cbf97a : SUCCESS in 1h 10m 31s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/1f8dfcf2a4524c128fac30052666e761 : SUCCESS in 1h 11m 47s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/2a1856b589da447aab5a7cef731d45c9 : SUCCESS in 1h 16m 17s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a0b14971b78140a189bd71b1eee3625c : SUCCESS in 1h 51m 52s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/e516fb1104b24b19890c1896d5ba910c : SUCCESS in 2h 01m 18s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/7691ad0a4a4a44fe98a25dc4c9cd8bf4 : SUCCESS in 2h 05m 54s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/4461f1131e09476981e4d7359d28a0fc : SUCCESS in 32m 09s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/51fa0e7f39e64b54833db8da5e12ec8f : SUCCESS in 32m 23s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5a497ab1aadc496bbdc59b0ea5236a88 : SUCCESS in 35m 39s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/8b91b142339d4138accd95c21acae5e2 : SUCCESS in 33m 24s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/88ebbcbe67534389afb0ccbf6afb0946 : SUCCESS in 45m 59s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9f5ac5d411d7447fb4036b19b862bff1 : SUCCESS in 52m 12s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/40c84800dd014cec95a5176dfd5fae36 : SUCCESS in 41m 24s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/86e1ccbebf78489dba9067b493703daf : SUCCESS in 35m 55s","accounts_in_message":[],"_revision_number":8},{"id":"0ce6268037855b83264b22e85c490964220d785f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 15:21:00.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":8},{"id":"243c60573ae32afbf6e36707ed492baca15e0e9c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 15:22:21.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bc50260cbef9448f89e4870e92a7a1f2\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99fe066b692444088852dc0fbd7ce3c8 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"b420a206cb135d06f722c79e0fcec206c8c1276d","revisions":{"62219ffbfd44b70f3672de8192960b4176c9b397":{"kind":"REWORK","_number":1,"created":"2026-06-04 11:04:44.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/75/991575/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/75/991575/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/75/991575/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/75/991575/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/75/991575/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/75/991575/1"}}},"commit":{"parents":[{"commit":"95e47148f1a2aa14429fe72c3879d7e6157789e9","subject":"Merge \"CI: Stop using image from Docker Hub\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/95e47148f1a2aa14429fe72c3879d7e6157789e9"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2026-06-04 11:03:18.000000000","tz":60},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2026-06-04 11:04:31.000000000","tz":60},"subject":"Never reboot without explicit permission","message":"Never reboot without explicit permission\n\nDepends-On: https://github.com/stackhpc/ansible-collection-linux/pull/50\nChange-Id: Icc517339b5ca1b66e750c0623f99451402da7cca\nSigned-off-by: Matt Crees \u003cmattc@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/62219ffbfd44b70f3672de8192960b4176c9b397"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/62219ffbfd44b70f3672de8192960b4176c9b397"}]},"branch":"refs/heads/master"},"ed4d8ba3e853e2ceb1c4dd3e87551ccf5d2d705e":{"kind":"REWORK","_number":2,"created":"2026-06-04 15:03:17.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/75/991575/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/75/991575/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/75/991575/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/75/991575/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/75/991575/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/75/991575/2"}}},"commit":{"parents":[{"commit":"95e47148f1a2aa14429fe72c3879d7e6157789e9","subject":"Merge \"CI: Stop using image from Docker Hub\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/95e47148f1a2aa14429fe72c3879d7e6157789e9"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2026-06-04 11:03:18.000000000","tz":60},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2026-06-04 15:03:05.000000000","tz":60},"subject":"Never reboot without explicit permission","message":"Never reboot without explicit permission\n\nChange-Id: Icc517339b5ca1b66e750c0623f99451402da7cca\nSigned-off-by: Matt Crees \u003cmattc@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ed4d8ba3e853e2ceb1c4dd3e87551ccf5d2d705e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ed4d8ba3e853e2ceb1c4dd3e87551ccf5d2d705e"}]},"branch":"refs/heads/master"},"3d934994a06b7574f366e9252bd7a180078cf136":{"kind":"REWORK","_number":3,"created":"2026-06-04 15:16:07.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/75/991575/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/75/991575/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/75/991575/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/75/991575/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/75/991575/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/75/991575/3"}}},"commit":{"parents":[{"commit":"95e47148f1a2aa14429fe72c3879d7e6157789e9","subject":"Merge \"CI: Stop using image from Docker Hub\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/95e47148f1a2aa14429fe72c3879d7e6157789e9"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2026-06-04 11:03:18.000000000","tz":60},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2026-06-04 15:15:31.000000000","tz":60},"subject":"Never reboot without explicit permission","message":"Never reboot without explicit permission\n\nBumps stackhpc.linux to 1.6.0 to support this.\n\nChange-Id: Icc517339b5ca1b66e750c0623f99451402da7cca\nSigned-off-by: Matt Crees \u003cmattc@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3d934994a06b7574f366e9252bd7a180078cf136"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3d934994a06b7574f366e9252bd7a180078cf136"}]},"branch":"refs/heads/master"},"05885dd2c8a7c07f028646f1c58ae34559480ff3":{"kind":"REWORK","_number":4,"created":"2026-06-08 08:22:25.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/75/991575/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/75/991575/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/75/991575/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/75/991575/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/75/991575/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/75/991575/4"}}},"commit":{"parents":[{"commit":"95e47148f1a2aa14429fe72c3879d7e6157789e9","subject":"Merge \"CI: Stop using image from Docker Hub\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/95e47148f1a2aa14429fe72c3879d7e6157789e9"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2026-06-04 11:03:18.000000000","tz":60},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2026-06-08 08:22:19.000000000","tz":60},"subject":"Never reboot without explicit permission","message":"Never reboot without explicit permission\n\nBumps stackhpc.linux to 1.6.0 to support this.\n\nChange-Id: Icc517339b5ca1b66e750c0623f99451402da7cca\nSigned-off-by: Matt Crees \u003cmattc@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/05885dd2c8a7c07f028646f1c58ae34559480ff3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/05885dd2c8a7c07f028646f1c58ae34559480ff3"}]},"branch":"refs/heads/master"},"9d8b0ccc335758ecfc62cc286ed2178a2587d6a8":{"kind":"REWORK","_number":5,"created":"2026-06-08 08:27:31.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/75/991575/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/75/991575/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/75/991575/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/75/991575/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/75/991575/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/75/991575/5"}}},"commit":{"parents":[{"commit":"95e47148f1a2aa14429fe72c3879d7e6157789e9","subject":"Merge \"CI: Stop using image from Docker Hub\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/95e47148f1a2aa14429fe72c3879d7e6157789e9"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2026-06-04 11:03:18.000000000","tz":60},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2026-06-08 08:27:24.000000000","tz":60},"subject":"Never reboot without explicit permission","message":"Never reboot without explicit permission\n\nBumps stackhpc.linux to 1.6.0 to support this.\n\nChange-Id: Icc517339b5ca1b66e750c0623f99451402da7cca\nSigned-off-by: Matt Crees \u003cmattc@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9d8b0ccc335758ecfc62cc286ed2178a2587d6a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9d8b0ccc335758ecfc62cc286ed2178a2587d6a8"}]},"branch":"refs/heads/master"},"e7de9bce6725cde32829729b1170e65d5429cb07":{"kind":"REWORK","_number":6,"created":"2026-06-08 09:10:25.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/75/991575/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/75/991575/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/75/991575/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/75/991575/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/75/991575/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/75/991575/6"}}},"commit":{"parents":[{"commit":"95e47148f1a2aa14429fe72c3879d7e6157789e9","subject":"Merge \"CI: Stop using image from Docker Hub\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/95e47148f1a2aa14429fe72c3879d7e6157789e9"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2026-06-04 11:03:18.000000000","tz":60},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2026-06-08 09:10:19.000000000","tz":60},"subject":"Never reboot without explicit permission","message":"Never reboot without explicit permission\n\nBumps stackhpc.linux to 1.6.0 to support this.\n\nChange-Id: Icc517339b5ca1b66e750c0623f99451402da7cca\nSigned-off-by: Matt Crees \u003cmattc@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e7de9bce6725cde32829729b1170e65d5429cb07"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e7de9bce6725cde32829729b1170e65d5429cb07"}]},"branch":"refs/heads/master"},"267e42920144550f28dd93f5bead1b866f1a56bd":{"kind":"REWORK","_number":7,"created":"2026-07-10 15:02:22.000000000","uploader":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"ref":"refs/changes/75/991575/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/75/991575/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/75/991575/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/75/991575/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/75/991575/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/75/991575/7"}}},"commit":{"parents":[{"commit":"95e47148f1a2aa14429fe72c3879d7e6157789e9","subject":"Merge \"CI: Stop using image from Docker Hub\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/95e47148f1a2aa14429fe72c3879d7e6157789e9"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2026-06-04 11:03:18.000000000","tz":60},"committer":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2026-07-10 15:02:22.000000000","tz":0},"subject":"Never reboot without explicit permission","message":"Never reboot without explicit permission\n\nBumps stackhpc.linux to 1.6.0 to support this.\n\nChange-Id: Icc517339b5ca1b66e750c0623f99451402da7cca\nSigned-off-by: Matt Crees \u003cmattc@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/267e42920144550f28dd93f5bead1b866f1a56bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/267e42920144550f28dd93f5bead1b866f1a56bd"}]},"branch":"refs/heads/master"},"b420a206cb135d06f722c79e0fcec206c8c1276d":{"kind":"REWORK","_number":8,"created":"2026-07-13 12:56:51.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/75/991575/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/75/991575/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/75/991575/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/75/991575/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/75/991575/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/75/991575/8"}}},"commit":{"parents":[{"commit":"95e47148f1a2aa14429fe72c3879d7e6157789e9","subject":"Merge \"CI: Stop using image from Docker Hub\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/95e47148f1a2aa14429fe72c3879d7e6157789e9"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2026-06-04 11:03:18.000000000","tz":60},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2026-07-13 12:56:13.000000000","tz":60},"subject":"Never reboot without explicit permission","message":"Never reboot without explicit permission\n\nBumps stackhpc.linux to 1.6.0 to support this.\n\nCloses-Bug: #2091713\nChange-Id: Icc517339b5ca1b66e750c0623f99451402da7cca\nSigned-off-by: Matt Crees \u003cmattc@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b420a206cb135d06f722c79e0fcec206c8c1276d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b420a206cb135d06f722c79e0fcec206c8c1276d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
