)]}'
{"id":"openstack%2Fkayobe~991622","triplet_id":"openstack%2Fkayobe~stable%2F2026.1~I122af26524197e3d67bbf46a75ff632b836beabe","project":"openstack/kayobe","branch":"stable/2026.1","topic":"customContainersEdits-stable/2026.1","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2026-06-04 19:10:32.000000000","reason":"\u003cGERRIT_ACCOUNT_15197\u003e replied on the change","reason_account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}}},"hashtags":[],"change_id":"I122af26524197e3d67bbf46a75ff632b836beabe","subject":"Fix naming of custom container variables","status":"MERGED","created":"2026-06-04 13:04:36.000000000","updated":"2026-06-04 22:23:43.000000000","submitted":"2026-06-04 22:22:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"991622-customContainersEdits-stable/2026.1","cherry_pick_of_change":990931,"cherry_pick_of_patch_set":2,"meta_rev_id":"7df668cd65653553f0947e23971d53b31d0ba86c","_number":991622,"virtual_id_number":991622,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-04 22:22:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"recommended":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"all":[{"value":1,"date":"2026-06-04 13:10:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},{"value":2,"date":"2026-06-04 16:49:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2026-06-04 13:16:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-06-04 14:18:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":0,"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},{"value":1,"date":"2026-06-04 16:49:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-04 13:04:36.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"state":"CC"},{"updated":"2026-06-04 13:10:09.000000000","updated_by":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"reviewer":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"state":"REVIEWER"},{"updated":"2026-06-04 13:16:12.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2026-06-04 14:18:51.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2026-06-04 14:42:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-04 16:49:31.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"}],"messages":[{"id":"e2a35657caab394b372565e26d35d1cd950a19c3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-06-04 13:04:36.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"c51cd0302101aaee1e64d57aa40342196ef806a3","author":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"date":"2026-06-04 13:10:09.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"ac7d13b38da0a54d0ae9dc08e921b1c4ed402546","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-06-04 13:16:12.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"a83dcd9dfc052c66de3b311479c7564740eb8e6a","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-06-04 14:18:51.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"38a5a972732af5186adffda2fb50f341171e99ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-04 14:42:55.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/944b797fc84e4347959de7008abe3099\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5c54af890754e9ba05f759697696e44 : SUCCESS in 10m 28s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/4ed1bd053dee4e378b28b54cb22faccb : SUCCESS in 7m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0d53b506397a45d8b80cb58039347d18 : SUCCESS in 4m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56e8cf36128a4be68eee074806c83281 : SUCCESS in 4m 10s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bd6bbe46a65744df8f7a7d92070e0c16 : SUCCESS in 3m 46s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e2fdee759fa0431781de43604a2f24d3 : SUCCESS in 7m 04s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0993d70637e6494ea054aacc099cd51f : SUCCESS in 7m 10s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ec06510d6865448288ee0aa3ae96ca92 : SUCCESS in 4m 52s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/27de2166842248c4957ab5fc3f2d1ad2 : SUCCESS in 7m 32s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/60960a96d5d54f0f913244a23a19ed34 : SUCCESS in 13m 06s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/f2d5a3d6b173446a9b6e595f0d15f320 : SUCCESS in 13m 07s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/66064834d6e949758f92dd28eab6a152 : SUCCESS in 22m 18s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/41fd80e8f47b46b79a92c41f91576ab3 : SUCCESS in 16m 42s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/4d98d0ca55754b6da0fbb8f4dea8ffb3 : SUCCESS in 20m 11s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/b3c0fd41e5cd458a9acabe390faa7110 : SUCCESS in 14m 54s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7cdb04a2c68f46d6a2c1e2efe7b053c4 : SUCCESS in 13m 56s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/fdef9c1ee7fa4a9b91268241700c80c1 : SUCCESS in 24m 32s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6cf44da7f766488f87319a9c6bda022a : SUCCESS in 18m 13s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/6a138ef920254bc4849bedcc0e93f4b9 : SUCCESS in 1h 02m 28s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/73783cf204a4457e9da2d174c1a44709 : SUCCESS in 1h 04m 21s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/5f7096efa87b4596b87cd65ceb2ca782 : SUCCESS in 59m 04s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bd0fcd6d2b1240cab2362c6917e6badc : SUCCESS in 1h 05m 15s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/97499d7e25a84e278627764273d096b4 : SUCCESS in 1h 06m 47s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/16549a44315c4a9ba42dd86a30b085dc : SUCCESS in 1h 33m 41s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/2c1518be9d8a47c19f6d24af41d474d4 : SUCCESS in 1h 27m 37s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a72d696b869d403f958374a698d299f8 : SUCCESS in 1h 33m 22s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5013caaf50a44fe4a9b66e12f74f4214 : SUCCESS in 1h 36m 59s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/5f249557d6bd4011b734412bc30905b3 : SUCCESS in 24m 25s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/2fc3c7f565524a8196f1c654305e8627 : SUCCESS in 17m 39s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/92b700b998ac41248a5ff6e149a89ec6 : SUCCESS in 22m 56s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/d405467b39774db7b106d3b497218b63 : SUCCESS in 23m 52s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/43a8eb6ed2474a4c8831ab9e377260e3 : SUCCESS in 36m 22s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/53879a762b21492aa5b80ad7fa4892d6 : SUCCESS in 38m 24s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/41590c999c4041ac8d4bf6e25c5979c6 : FAILURE in 16m 02s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/78b0ae8e78fb481996f1ec67e32952cb : SUCCESS in 17m 56s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/7f40c47a3c0b45ecb0f09f104c1ff4be : SUCCESS in 20m 48s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a248655dcaf140e8998f1203abb94b4e : SUCCESS in 23m 27s","accounts_in_message":[],"_revision_number":1},{"id":"cc456898d4f79bcd692b13c9d5e8134e538c7bb4","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-06-04 14:45:02.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b3a8f46d37b27d5bd80a634fb7e2be8af977cc44","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-04 16:23:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/642cef936d21419e9a5a4a1499a2d2a1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93ca89a4c1254e5bb66096438b9f2bd2 : SUCCESS in 5m 49s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/46b0dfbcf8c044319ded93b2cd2e6d89 : SUCCESS in 4m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/33b46e44a8574546b820331958a437ac : SUCCESS in 3m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8af6d0e08fba448daad53c3abc46ca75 : SUCCESS in 3m 30s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d5d2593cd16748f49f9ec30322d412ef : SUCCESS in 5m 34s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/6eb17cd23f134ce69bc7bb734520cf26 : SUCCESS in 4m 07s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dbe8d33c5d6348bfae82767f675b5bc6 : SUCCESS in 4m 17s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d9179b9d59284b50b41372c5e09f3702 : SUCCESS in 4m 04s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9779bd6b789f4096b79719791126f1b6 : SUCCESS in 6m 25s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/362e3da7af3447feacc9f1f9bf975d27 : SUCCESS in 6m 26s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/6dd105d8696c4bbabe8e65fc9c73a191 : SUCCESS in 19m 04s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/f3c513d7483347bab2af18741dd58ccc : SUCCESS in 12m 09s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f16197d03b1c432685e92537c1cb1f00 : SUCCESS in 18m 17s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/92db0e142dcd47b69a4ee2f19682b19b : SUCCESS in 15m 16s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/dfb3d7d97ff6422487bae1aa0414549f : SUCCESS in 20m 30s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/72c57e5b7b034494915807ab79167b60 : SUCCESS in 21m 26s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/ff20a981027a479395d67fe2681c5339 : SUCCESS in 12m 28s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/daf09974d345422d8589fd9b80cac6b5 : SUCCESS in 20m 05s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/b45aa719b5c542218d01a4d1d089ea00 : SUCCESS in 1h 00m 16s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/60315d5082a547bcbf1ee44b141c2db4 : SUCCESS in 1h 08m 06s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/3dff7418dfd147c5bd9ee1f3e11af68b : SUCCESS in 1h 05m 47s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/135b9cbe89ff41b280c553d66ca6d9ed : SUCCESS in 1h 02m 59s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/252a57d51a8a4a0a9d2436f724308f23 : SUCCESS in 1h 04m 03s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/50cdf1ab78f64c9eae147214059f6839 : SUCCESS in 1h 33m 00s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/eb8209d158624551bd152fa66e20ec77 : SUCCESS in 1h 24m 40s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6ceb5d8f7183426bbb3ae782d1121c94 : SUCCESS in 1h 31m 08s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c5f5616d318944f0835c197cfb666ce9 : SUCCESS in 1h 37m 00s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/a0532a9419a644fe9452a3f6628bfee1 : SUCCESS in 26m 22s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/26961f40650a4b0d97667d8efa7262bf : SUCCESS in 24m 46s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/825864dbb6284a298f417115ebe1089b : SUCCESS in 22m 13s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/674ad6e596c54d488f3a1aea2ad9847a : SUCCESS in 18m 53s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/82fd932977f14694a44f286f2a512831 : SUCCESS in 24m 54s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/bdb284335d9d412aac7d2c335f5b344b : SUCCESS in 39m 57s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d22f2331ee6148fe9049a569d6179b0a : SUCCESS in 37m 42s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/45a9bd40675f45a29da67e1793b4b800 : SUCCESS in 34m 49s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/c6d31ce3d1e94eb6b75afd7816e3bebb : SUCCESS in 15m 46s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a76dc6f468ab4e0a9cbf47d29088a191 : SUCCESS in 20m 34s","accounts_in_message":[],"_revision_number":1},{"id":"1e7ab9eb6ad4cc7c2a3d41d3405afa8a23abc169","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-06-04 16:49:31.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6a8387cc1f8c54c2946acbbaafb36a8418365f57","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-04 16:49:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"15d9f14d047afd839baeef96c03a4c5bc896f68f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-04 19:00:34.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/254ab5cfe0774ffd989a74772b6be86a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e38edbe9ee8b4c7c834b7ebc094d05da : SUCCESS in 10m 13s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/89ea75d922ae4ee8a389d416f07cd9f8 : SUCCESS in 8m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/caad1f1b0286437f85f7a20050e915ac : SUCCESS in 3m 39s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a9c1007c37804363afdbfd645fa18596 : SUCCESS in 3m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/90d767471ef64120b7f7a5bd6511b005 : SUCCESS in 6m 02s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/776b83b0a5564985ac96300100d4417c : SUCCESS in 5m 06s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/86849c16bc63424093797cedc37e25e1 : SUCCESS in 7m 37s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f920f1a7dad4432597d6a65cc1d56dc2 : SUCCESS in 13m 38s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/feb93abd71034f608345afc123b46531 : SUCCESS in 23m 21s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d98d965eb32e4f019faca7f2fc07bd53 : SUCCESS in 16m 09s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/a57fbbeaf6a7414cb238ba6b59f6be0f : SUCCESS in 16m 43s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8fcec7e428634a95a7fd6592f5f496f5 : SUCCESS in 22m 08s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/2e48126754504e56a7cc5a0b78a20dd0 : SUCCESS in 13m 26s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/edb094c26d4c48578f2604029ea447cd : SUCCESS in 17m 41s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/b5cefab4b68e45b682170ff15756114e : SUCCESS in 1h 07m 25s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/ba91c6aecd5a4703b959fd5eb34b58cf : SUCCESS in 1h 07m 49s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/91a55c7e167b49c79511e51aac7bae12 : SUCCESS in 1h 11m 10s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/433f1819d7af4e749d2bd8b0bd575ea5 : SUCCESS in 1h 14m 00s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/0130b5e1de1a4d37b591aed771a54fdb : SUCCESS in 1h 12m 14s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/fd7c8e98c31e4e41a7e664707265a3ef : FAILURE in 1h 35m 46s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/abaae055bee3463d8e7e6b9ab998a028 : SUCCESS in 1h 36m 32s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/94bbdeac691a49fdaf68a5a1b67db487 : SUCCESS in 2h 09m 46s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/627a00fb17944a0d872a0e858f2cfa23 : SUCCESS in 1h 44m 43s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/3547df41d4774b458e3c12280f16b54c : SUCCESS in 14m 57s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/9b465744c92740d1b590868c45f2c007 : SUCCESS in 16m 28s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ca2d3879ceb44958bdb473766c93860f : SUCCESS in 23m 50s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/273fcc54d8694261b9102f66db5fd662 : SUCCESS in 20m 31s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/308a8573adc54f2cb3364eac18cc2ad7 : SUCCESS in 27m 09s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/6df004f5bb1543d5908e11efee8c8530 : SUCCESS in 20m 43s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/93ec0d4d020c423497b77d034ab239a0 : SUCCESS in 35m 04s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/575f2ac4cc764e3992a4dffdc703ba37 : SUCCESS in 27m 43s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/1749d19c262a4c25b9dd04e6f20cc272 : SUCCESS in 16m 32s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f6f7a368731e4742ac6991a1b171d7ea : SUCCESS in 13m 12s","accounts_in_message":[],"_revision_number":1},{"id":"782ecd00192336e249d7f3b5eed18d5b23d5fe48","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-06-04 19:10:32.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"96956864de6f0bbd6a06144ea2f326afee76944d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-04 20:48:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7d0f0339e0194820a2bb7ee4e50c3686\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb22711ffb51442db6a5a47e9d261bc4 : SUCCESS in 9m 34s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/595d40fd8aa74c1095fe3e649c9332c6 : SUCCESS in 4m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ba5d3bdd0ffc48dcbbcda65de6bbf986 : SUCCESS in 4m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e8967344118477495cf24f2eca5858b : SUCCESS in 4m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2b36c5fdb7844189ab038f7f3478d85f : SUCCESS in 3m 22s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/d51af95edca249e48739e4a98cff3706 : SUCCESS in 8m 27s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6350589cdc214133b2c24539b8994911 : SUCCESS in 7m 21s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/01b02086fc194cc1954da0dad2c2e7ed : SUCCESS in 7m 28s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/68db96e83fbc4bc281a591ebebf96fdf : SUCCESS in 11m 10s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/4944b47e2236418badd4a1d361f07c21 : SUCCESS in 12m 17s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/ad3dd570ed394ef2a048baeaf6d1c95b : SUCCESS in 24m 06s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/7cc0d002207e447b842bd68e47e34e59 : SUCCESS in 23m 39s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a08f95c0b50144dfb7e768e451e9b137 : SUCCESS in 16m 01s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/287510d908a3430188d2c457ca5f3684 : SUCCESS in 16m 31s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/45ac725a6e324e68ab785ffbd531790a : SUCCESS in 22m 19s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7029936afd304d3a9bc318e2f013a730 : SUCCESS in 22m 40s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/3dd06eaecc0c4608ba82e62215e067ab : SUCCESS in 12m 49s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/93ae415b5efe4350bd16ed9f5d328cd3 : SUCCESS in 19m 00s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/09acdcfd93944ffc9e42427a806ed5ef : SUCCESS in 56m 23s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/12f7668827bd49a7938d49a792ed6d44 : SUCCESS in 1h 02m 24s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/f8427a6022e643819c85dd3bd1b626b7 : SUCCESS in 59m 42s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bc70aea6aece4b3fae147f308a3a86f9 : SUCCESS in 1h 02m 12s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/c5de0c0e770a4257a6588cad543df59f : SUCCESS in 1h 07m 55s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/b2bfe4db5a1649ac9483d7131a36414e : SUCCESS in 1h 32m 36s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/9eacc1bc21e94c8bb6c149e3764be717 : SUCCESS in 1h 30m 20s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/aa3c900723224fee847b4cf3f186225f : SUCCESS in 1h 33m 08s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a12ff378ed3348989e5dbad71f98904d : SUCCESS in 1h 35m 39s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/2a811249e42e49adbc16aae7adc6e0e7 : SUCCESS in 24m 30s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/c1a65489ea0d4e08b99b33dc9b680507 : SUCCESS in 17m 21s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5cce097d313d4e5cb560f6fd33e22d28 : SUCCESS in 22m 57s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/8c4d6a6fd7d2442d97b3f948bdefbd32 : SUCCESS in 22m 10s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/95511209b4e2462789515759eed88859 : SUCCESS in 19m 01s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/159fedf42d8947bb967c8ef9594f7679 : SUCCESS in 33m 02s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8c7d157bff5e462fb70bd4d02be4636e : SUCCESS in 29m 48s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/269d4d04c5164e07a814db75bb5864ee : SUCCESS in 29m 05s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/74fdfb7e5a0244758d1a83e5abd5a8f0 : SUCCESS in 21m 49s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6b6c88bd61db49b58072473a1671f04e : SUCCESS in 21m 35s","accounts_in_message":[],"_revision_number":1},{"id":"6c375e405174e63553c8ac4a09e928041d04f860","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-04 20:49:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2b6e3c18f5d4ea2377ed6c68a3ef8ec454d3da78","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-04 22:22:38.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a70dc468baf54bc69360d144da9259fa\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f603ebb5701e49bd9f07c410567cdb5a : SUCCESS in 9m 57s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/7be2d152b831408f9aeba8d0821c76cd : SUCCESS in 4m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/46a3e953f1d94327afc3b31311f268c8 : SUCCESS in 4m 21s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b3561bb5689541da99c0a6b3f50d0936 : SUCCESS in 4m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6a02850f5c034fe2b2c4355f4b24c732 : SUCCESS in 5m 19s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/45fbecce1d484544aa8b36f3d8a4d6f1 : SUCCESS in 5m 32s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/207f9ed940b24b248016b741a910a733 : SUCCESS in 12m 05s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/50aad240d34a43d6bb9b38e30a138a3b : SUCCESS in 13m 27s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/c7cf9a4e743a4dc5b7b32643f53de1e5 : SUCCESS in 22m 16s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/84b2ab2459aa4058a7513ab46540658b : SUCCESS in 8m 40s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/bb5c3f6b54fa44a2904f07ce5872ac4d : SUCCESS in 23m 39s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f434ee1645294f9ca401d0a9d8a6e42b : SUCCESS in 21m 32s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/361031cb94014c8d895995106717aa89 : SUCCESS in 16m 28s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a4302cf801bf477ebf8825f2389e0ee8 : SUCCESS in 13m 07s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/6fc4eb020b2f4370bc274dee3ea1a020 : SUCCESS in 59m 04s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/cb62b8a371764fd09f06933f0501dc17 : SUCCESS in 1h 02m 50s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/b563a14ba6904e10a4eb212808824d35 : SUCCESS in 59m 38s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/959b47dbb77542fc94f74d723eaad71a : SUCCESS in 57m 03s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/16ddc379985a4cb892e7c24e3664bc90 : SUCCESS in 1h 07m 46s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/cdca14371f6c45518af83bbeb7bf1b7a : SUCCESS in 1h 25m 22s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/59e06d3e57bc48bd842132490c93cf91 : SUCCESS in 1h 21m 52s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f978772e9c724e8fba1af176bd9b0d89 : SUCCESS in 1h 27m 21s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c1af11eda2d24f8ea9bf441d45f91069 : SUCCESS in 1h 32m 36s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/64c055b9008148f59b2d66cd88246f86 : SUCCESS in 18m 26s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/81c78f905571488e8863c999e95ba40a : SUCCESS in 14m 36s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/24203423494a4cfea01d1526844b6a40 : SUCCESS in 26m 10s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/1fd4836baa694ceda8838cdf5daaf0ea : SUCCESS in 19m 48s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/3d305e53674041f0a53489cab7f63f56 : SUCCESS in 17m 01s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/593af27f59bb4af1ab5e5ef2dc7de2e2 : SUCCESS in 18m 46s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a2cbfd2a8a95453f87748a33836b67cd : SUCCESS in 28m 56s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1ff97cefb7ff453b833254af5b48d94f : SUCCESS in 14m 31s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/f2645fc1e8274deab5a821a0f9da1764 : SUCCESS in 16m 42s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/63b2982878ba477db6accae7906d4e82 : SUCCESS in 21m 16s","accounts_in_message":[],"_revision_number":1},{"id":"238859a78a9d193cdf8fc08cca321ff7949987e8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-04 22:22:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"7df668cd65653553f0947e23971d53b31d0ba86c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-04 22:23:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/098ee64d95df451a9cbae65f0d626633\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29ea8464d0984fd49130aa47bf891228 : SUCCESS in 53s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a7d1d06223f6afb2a8714d1e20ef076406664159","revisions":{"a7d1d06223f6afb2a8714d1e20ef076406664159":{"kind":"REWORK","_number":1,"created":"2026-06-04 13:04:36.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/22/991622/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/22/991622/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/22/991622/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/22/991622/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/22/991622/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/22/991622/1"}}},"commit":{"parents":[{"commit":"ec5c196c0e19295ccf432a43357ad7d14532e74a","subject":"Add docs for kolla-playbook usage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ec5c196c0e19295ccf432a43357ad7d14532e74a"}]}],"author":{"name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","date":"2026-06-01 13:38:13.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-06-04 13:04:36.000000000","tz":0},"subject":"Fix naming of custom container variables","message":"Fix naming of custom container variables\n\nUpdates the custom container example in ansible-control.yml\nto the correct variable, and updates a description in\ninfra-vms.yml to reference the correct container.\n\nCloses-Bug: #2154741\nChange-Id: I122af26524197e3d67bbf46a75ff632b836beabe\nSigned-off-by: Leonie Chamberlin-Medd \u003cleonie@stackhpc.com\u003e\n(cherry picked from commit 444db636625b3ffae1b9cf03045463787fa485a0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a7d1d06223f6afb2a8714d1e20ef076406664159"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a7d1d06223f6afb2a8714d1e20ef076406664159"}]},"branch":"refs/heads/stable/2026.1","conflicts":{"ours":"ec5c196c0e19295ccf432a43357ad7d14532e74a","theirs":"444db636625b3ffae1b9cf03045463787fa485a0","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
