)]}'
{"id":"openstack%2Fkayobe~992204","triplet_id":"openstack%2Fkayobe~master~I9552c722cdb0215a5b4a0d7b65cf9a2239573efc","project":"openstack/kayobe","branch":"master","topic":"bug/2154438","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2026-07-15 17:30:48.000000000","reason":"\u003cGERRIT_ACCOUNT_28048\u003e replied on the change","reason_account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},"35345":{"account":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"last_update":"2026-06-18 08:42:27.000000000","reason":"\u003cGERRIT_ACCOUNT_35345\u003e replied on the change","reason_account":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"}},"37122":{"account":{"_account_id":37122,"name":"Sylvère K","display_name":"sylvr","email":"sylvere.kanapa@univ-lyon1.fr","username":"sylvr"},"last_update":"2026-07-15 21:09:58.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2026-07-13 09:15:45.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"35263":{"account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"last_update":"2026-07-14 07:57:35.000000000","reason":"\u003cGERRIT_ACCOUNT_35263\u003e replied on the change","reason_account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}}},"hashtags":[],"change_id":"I9552c722cdb0215a5b4a0d7b65cf9a2239573efc","subject":"Adds seed IP to default no-proxy list","status":"MERGED","created":"2026-06-08 10:33:09.000000000","updated":"2026-07-15 21:11:01.000000000","submitted":"2026-07-15 21:09:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":4,"has_review_started":true,"submission_id":"992204-bug/2154438","meta_rev_id":"b8bb449dc23ceebb9c26b6c7226b654cd45e27d6","_number":992204,"virtual_id_number":992204,"owner":{"_account_id":37122,"name":"Sylvère K","display_name":"sylvr","email":"sylvere.kanapa@univ-lyon1.fr","username":"sylvr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-07-15 21:09:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2026-07-10 18:12:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},{"value":1,"date":"2026-07-13 09:15:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-07-14 07:57:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-07-14 07:57:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2026-07-10 18:12:16.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-08 11:09:46.000000000","updated_by":{"_account_id":37122,"name":"Sylvère K","display_name":"sylvr","email":"sylvere.kanapa@univ-lyon1.fr","username":"sylvr"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2026-06-08 12:15:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-08 13:39:43.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2026-06-11 08:08:10.000000000","updated_by":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"reviewer":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"state":"REVIEWER"},{"updated":"2026-06-22 12:27:47.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"CC"},{"updated":"2026-07-10 18:12:16.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"}],"messages":[{"id":"4fffe9d03e62b8a76ae0cd3f91ce5d5980e7db3f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37122,"name":"Sylvère K","display_name":"sylvr","email":"sylvere.kanapa@univ-lyon1.fr","username":"sylvr"},"date":"2026-06-08 10:33:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"308654999b4cfc7111ee8752f6c33df389c23591","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-08 12:15:34.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c27329f7d4a3481c97d8b7df1e4c36f0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9948d82b87c345408961476405b3eb7a : SUCCESS in 9m 13s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/d928afc9b370405399538d0e124cf216 : SUCCESS in 6m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/14ad9a8cf25543369dd34ced72c2a3f3 : SUCCESS in 3m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/71d7c1ccf86c46d2b3032398b3f76cb6 : SUCCESS in 3m 09s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c62b46d23ede439eb078579fff4fb9d3 : SUCCESS in 3m 12s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/50bd324039b54e919320d099676d2d44 : SUCCESS in 6m 55s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c8d4b09cab204eb68b6566efc10d0461 : SUCCESS in 6m 38s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/37473720e7394dc9b63dc42d9778116c : SUCCESS in 3m 41s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/dbe2d97deaf443acb1ee7067b66c18ac : SUCCESS in 12m 16s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a673edac80fd40049d2e1d401e181503 : SUCCESS in 11m 31s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/c526163e762243e7833dade27b507bcf : SUCCESS in 12m 54s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/8aef1b801ecb4f7c816a572c7d87b172 : SUCCESS in 12m 47s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7bd4c520ca1e4229a28eb585b2f3ce9d : SUCCESS in 17m 18s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/0f81631ea3aa4f1da0a3f7ae845af421 : SUCCESS in 16m 19s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/6ddd229d613d497597c60fc08352012d : SUCCESS in 21m 06s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/82b75e7444e346128ddc064eba8269d7 : SUCCESS in 21m 53s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/b4b2b5264c94431cb09ac56857b608d5 : SUCCESS in 23m 55s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9128c8717f7c4779b67018c29fde1618 : SUCCESS in 16m 28s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/15cbe72ca1624c00ac13002f468518e8 : SUCCESS in 1h 01m 13s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/61d59377bfd6479485d1978647dba750 : SUCCESS in 1h 00m 29s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/b1a5494545d8423b80cb9f7944021d25 : SUCCESS in 1h 03m 23s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7e7e734b414f430ca603e81abefe4c20 : FAILURE in 29m 43s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/eeef98e92f84414dbda98ea3b0d8bc31 : FAILURE in 25m 06s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/8046f3acb9e942289ef3ba307ecb7ba3 : SUCCESS in 1h 41m 07s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/917bdcb31dab4b93b26e164836b9f426 : FAILURE in 29m 55s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/bb330dc13de94ece92bdbf9e8c136813 : SUCCESS in 17m 19s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/a188cb7158d449b494bc0dde5e7d308a : SUCCESS in 21m 17s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/93cffee18e6c4f44a41693311d3db820 : SUCCESS in 21m 50s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/ab1d36cdf0394059a998515e4045afe7 : SUCCESS in 14m 02s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/92f68d08f46b44769a4b45f80bb6a56d : SUCCESS in 31m 27s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/35693c4f24ef457a81971411094713a1 : SUCCESS in 38m 01s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/b62766fe15a246e2b1af15ce5b20ba81 : SUCCESS in 20m 04s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/01521211fac9478a83d5a7681bff8498 : SUCCESS in 19m 37s","accounts_in_message":[],"_revision_number":1},{"id":"8028707a19b19670248838a149e05d516a9397d4","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-06-08 14:45:57.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2937716398cc02e41288981de55b8ab4c0b5d623","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-06-09 10:19:25.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"5f9bdc07ae688cc4f30d2c43c656e49357cfd6c9","author":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"date":"2026-06-11 08:08:10.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"d74c4c658e13e8f1bae2a0133926b2643b2102c5","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":37122,"name":"Sylvère K","display_name":"sylvr","email":"sylvere.kanapa@univ-lyon1.fr","username":"sylvr"},"date":"2026-06-15 08:51:21.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"9f7d8597d845aa1a65fa9378778a10ec42af6704","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37122,"name":"Sylvère K","display_name":"sylvr","email":"sylvere.kanapa@univ-lyon1.fr","username":"sylvr"},"date":"2026-06-15 08:51:28.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"fe13db7053fb4ea1c4740dc81f520318153af102","tag":"autogenerated:gerrit:abandon","author":{"_account_id":37122,"name":"Sylvère K","display_name":"sylvr","email":"sylvere.kanapa@univ-lyon1.fr","username":"sylvr"},"date":"2026-06-15 08:51:52.000000000","message":"Abandoned\n\nI don\u0027t know what I\u0027m doing","accounts_in_message":[],"_revision_number":2},{"id":"cd05bbad99211f3189b7484fb1c49bd5883f7f1e","tag":"autogenerated:gerrit:restore","author":{"_account_id":37122,"name":"Sylvère K","display_name":"sylvr","email":"sylvere.kanapa@univ-lyon1.fr","username":"sylvr"},"date":"2026-06-15 08:52:04.000000000","message":"Restored","accounts_in_message":[],"_revision_number":2},{"id":"4bf6dd1500f11d967e0543a6da808c7adcf3a435","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":37122,"name":"Sylvère K","display_name":"sylvr","email":"sylvere.kanapa@univ-lyon1.fr","username":"sylvr"},"date":"2026-06-15 08:52:16.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"f9b19a050d21cc689a5459e197da2d45cb16cdc8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-15 10:33:47.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/be358d35bde7455ea441787dbbffcf38\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9cfdbee4ed3404dab60e30af24ffce8 : SUCCESS in 9m 29s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/1877fc54a64b4bfda241c9fc3145ffb7 : SUCCESS in 8m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a9e44bd707b441c6b91e1f415f3b0606 : SUCCESS in 5m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/00690766f79640faa705e530ae9b7595 : SUCCESS in 4m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/66546404741242fcbd6767b325f18ab0 : SUCCESS in 5m 56s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/d68f26e049274c8dad28191aa6ae4345 : SUCCESS in 7m 01s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0f9ebff860f941e0900846683e815eef : SUCCESS in 3m 27s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/5b53604c0b4244bca8e8d1bb04f3e099 : SUCCESS in 4m 52s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/107b2d03a27140479371712a9669fde1 : SUCCESS in 15m 30s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f0a286f599aa49e79f3c741607bc1480 : SUCCESS in 6m 16s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/cc168d3665a0475b96911a78232e0f72 : SUCCESS in 12m 11s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/5bd28b99c0ca4f6482d0e856d07e141e : SUCCESS in 11m 08s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6c7dc449031c41c7b56f1738141ac253 : SUCCESS in 12m 22s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/45a2fb487c69483f916294f676573870 : SUCCESS in 17m 28s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/70bc490644344d8ebc7e798e6269a6ed : SUCCESS in 14m 23s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7454af38ba3e4b2983f7c141314adb50 : SUCCESS in 13m 04s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/82f3584b262b45c88db5379eba2425f7 : SUCCESS in 15m 21s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4968bf8498d44d9eabc56553145c9996 : SUCCESS in 22m 00s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/fa509d4af0c346c0a93d0a0379f1ffe6 : POST_FAILURE in 58m 39s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/07c0a7445d4f41d4bc067951db7b7d86 : POST_FAILURE in 1h 03m 15s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/57154eeb3d8043f9a58583a9dd80f2f7 : SUCCESS in 1h 01m 06s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/77aee94ac7b64bf6b9ebe45d1fc0bd91 : POST_FAILURE in 1h 03m 00s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/c769f52b5b14455aa4bc7a3ce8fe602d : SUCCESS in 1h 03m 25s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/d1f0ee915ed14c04a3851dd22a719690 : SUCCESS in 1h 40m 55s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/55c4963aac0b4652b11f600587f1a3ed : SUCCESS in 19m 45s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/3df6651f83de421894531b2091de36cd : SUCCESS in 15m 04s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3c7d3e32d9c9422185d4d65b9e8dce7f : SUCCESS in 20m 07s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/fedc5f7f23de4178a38643793a98737e : SUCCESS in 17m 53s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/a5083a0516a244ba84b4f58fe088dfe1 : SUCCESS in 22m 06s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c817f508e6444740ad18327c98045cd7 : SUCCESS in 21m 40s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/99bdb71183a24503a2c2861a84489000 : SUCCESS in 26m 17s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d82c65ad1512486aa6f390b0d54b6744 : SUCCESS in 13m 17s","accounts_in_message":[],"_revision_number":2},{"id":"409c460361622c9432a62f39d2b340dc876f87ba","author":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"date":"2026-06-18 08:42:27.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"677d50a794dd6c4d066c59c9b6e00ff68ba02236","author":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"date":"2026-06-18 08:44:33.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"080719653be1e211a38a7903d78d4a464d718bc9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-18 10:33:25.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a5d55297305341aa89bac5dbd10cd1e2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d31abb7e30149b2a0a9b83d03687c0d : SUCCESS in 10m 21s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/b697833a44e34fa796638ecc06a10ef2 : SUCCESS in 5m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/be29386c6a8f443197f45e1307fd2c68 : SUCCESS in 5m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e28410eda34944fc8f5a7b8a034dbf30 : SUCCESS in 2m 27s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c57475a09d45401ea6dac39f825c4e3e : SUCCESS in 5m 32s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/4e5c562a84834360969973365b973f4a : SUCCESS in 9m 40s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cd1a7bca93234de0bd42efb264658fff : SUCCESS in 8m 07s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3a49e88d7fb949b4857ca4ea5a4c9edf : SUCCESS in 5m 06s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/73b970f850134ee0af3a915ef9063d72 : SUCCESS in 13m 26s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/16ae47d3dd074480b26f07068153d1b4 : SUCCESS in 10m 06s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/8b693d515bdf462db619429bd71b0519 : SUCCESS in 23m 33s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/8c037f7953e94d4f8a530208a7f29882 : SUCCESS in 12m 21s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bc2493a3c2634c8582b1d5a728eb0138 : SUCCESS in 17m 37s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/be7f00403a6f40c19d3d5b4a050d75b2 : SUCCESS in 13m 47s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/119650867ebc41c1bb00685981a25d5b : SUCCESS in 22m 26s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/24ce7179a5894aa19c436ea5529ff731 : SUCCESS in 21m 00s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/5f6a710344a149e08185c79366eeab02 : SUCCESS in 26m 14s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d0ccc34e0ee041d096d248641f3592fd : SUCCESS in 19m 06s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/c6a82ec59b264c13b902cdae3bf52609 : SUCCESS in 1h 04m 05s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/d86f8fffbdc44dbab91b3ffc98175498 : SUCCESS in 1h 05m 31s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/3ed9cc1a636f4c7e81604f30b09bb459 : SUCCESS in 1h 08m 12s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6e30f40d5f664167a8a507ebbb79b97c : SUCCESS in 1h 10m 48s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/ca471dcd7c0546308a40f535b016193c : SUCCESS in 1h 14m 35s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/8ff8f55114b64bbeb69c73283e302eea : SUCCESS in 1h 46m 24s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/72e3216d1f1c4ad4b1c3e96998f19deb : SUCCESS in 1h 48m 55s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/7db18676bc3745068d01864fc2431b76 : SUCCESS in 25m 45s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/baa5bc1169dc4d51b6616bc5fc8316aa : SUCCESS in 29m 48s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1b62e6ca87e4422bbc09740ffb9c17b5 : SUCCESS in 25m 25s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/3e8563f31fb64501b7a5c3cece29f17f : SUCCESS in 17m 14s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/6f6a83b2057446f594b85c5149609088 : SUCCESS in 29m 05s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/83e29989035c4cb49e5fd325b4c54c3d : SUCCESS in 40m 27s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/3df3670dbb72483e84480ac3a02d4276 : SUCCESS in 16m 41s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0ec7201e4862454d9f2495354141472e : SUCCESS in 21m 03s","accounts_in_message":[],"_revision_number":2},{"id":"2a659283e53d0bd98b089085091d2cb0e7a384ea","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-06-22 12:27:47.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7e32df9910e6b001322fe7d2a1fb85ae7ba54662","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-07-10 10:15:01.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"18f6043df642673f5174b932177b8cef0e1ba2f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-10 12:11:22.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2dec0a7a28954ddc987514efe8e37071\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9b84aae0c3941c092c0de57d7e33ab8 : SUCCESS in 9m 03s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/cdd4156d4e9c4d5fba03fb2dc76c1315 : SUCCESS in 5m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/95b44fc787294a22a71be5d45f13f4ee : SUCCESS in 4m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c23f3bce1e074220a15073d74ee08232 : SUCCESS in 4m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0097a2a86f2f4482aa6d20dc4e3d282d : SUCCESS in 3m 52s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/66a61c637be1427da0db3c14d15a9a95 : SUCCESS in 7m 16s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/008762d113094e78a654367c407afbf2 : SUCCESS in 6m 16s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2b53b786156f49719a5f331610433381 : SUCCESS in 5m 30s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/07af0a675b614e0888d7afe282904cfc : SUCCESS in 13m 28s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/96b29bd4870c48cd951780501c6051c7 : SUCCESS in 11m 41s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/a2db89d4abf04c95894101ecc8740b8e : SUCCESS in 15m 01s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/8c82f39a36b643fbabb214459ee5cce1 : SUCCESS in 15m 31s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0abc581d612d4abb94336a213bed6b78 : SUCCESS in 15m 59s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/23e3f29d520541e382003ddc7dc5ce96 : SUCCESS in 17m 39s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/dabc5e89499a418090f677a108e07263 : SUCCESS in 14m 03s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/777878b70493467e842ea6136d6eddbb : SUCCESS in 14m 35s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/6eefc875ebe2477bb47b5e312f85db4c : SUCCESS in 10m 25s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/28c7c818655e4bffaaef98c6366fbaa0 : SUCCESS in 18m 38s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/edb35f24541545b393129b1d97dc3681 : SUCCESS in 1h 07m 23s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/17e53be4713e4a13b7f2cf4bcc16e486 : SUCCESS in 1h 07m 22s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/1c5369e3e9c744eca76a97732e6810e9 : SUCCESS in 1h 42m 44s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/adb5ee1b38fe4d29bf130fd9aec5be37 : SUCCESS in 1h 12m 58s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/a581fcfad1224f028bae9638bd9b3a1d : SUCCESS in 1h 13m 02s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/19a72d7d10394d078f1a22cbaef05bd4 : FAILURE in 1h 02m 12s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a44792270a2b49eb9909f9fefe3b00a7 : SUCCESS in 1h 55m 20s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/61f38950f1cb407482b8829d5f87a8c8 : SUCCESS in 28m 01s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/2dd52101459e46f6af86c0d28fed1ba6 : SUCCESS in 27m 28s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5c365ed2120f4d9eb97b07fe41160e1e : SUCCESS in 27m 09s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/a1e1e50bb2bb4ade950c11ce37e1fad3 : SUCCESS in 21m 08s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/305aff2acc034c84a86c68b050516263 : SUCCESS in 21m 54s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e1e6c1c281864df6a5df96082cd5b769 : SUCCESS in 37m 08s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/cf018c6d7ba24773a9bf37572d6fd7bc : SUCCESS in 15m 14s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4eba2729161f42c0ae3b70f62d846d44 : SUCCESS in 19m 54s","accounts_in_message":[],"_revision_number":3},{"id":"de52819de694f8e822e36f58687cc0988b5d34dd","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-07-10 15:43:06.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"9dc1bd7a530104a1bc9281582ec13a45ce38acbf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-10 17:34:10.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c45e5ac656af48258ebf6d291b3adfe6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b1e98bb18014142b717fb1a18e51e8e : SUCCESS in 9m 00s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/36af2692b36049aaa82a8b99fb8f87a8 : SUCCESS in 6m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/db41c9f482c641028f7bd78391a4cef2 : SUCCESS in 4m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a2ed4bea829c4bc39f8c6510de344af8 : SUCCESS in 4m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fc07a394ad2f46a4a01bd8229fcc16f3 : SUCCESS in 6m 03s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/d92dc0302cd047c6bdae81f743cdc535 : SUCCESS in 4m 27s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cb4f7bbd328946fe931d7c4635d8f8c3 : SUCCESS in 3m 49s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7652a547d30e4fdf8750911a4855bf7c : SUCCESS in 4m 59s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/46343fcdff0d41c8828793da7d90ecb5 : SUCCESS in 10m 32s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b745aac0e7aa4d2389bfa777db8f6ca4 : SUCCESS in 11m 37s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/33ee7fd594154ad19faa339d2cc9e9b6 : SUCCESS in 11m 23s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/e885b89cd6c345cbb66fe1a47da3d3e4 : SUCCESS in 13m 13s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/48728f92dc934bb3995b30268ff5e96d : SUCCESS in 16m 37s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/0da27987a83e4b8a8dd88a03eb5a9b52 : SUCCESS in 16m 03s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/80e9ce18afb24615a091e1634d7364cb : SUCCESS in 14m 08s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0f7b39fdac664615ba2cf3b3ae6cf2c2 : SUCCESS in 20m 04s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/773421a9e18a48cf9f98fd024b5e0e0b : SUCCESS in 19m 41s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a04951a2b2b348399f5cc82760c37914 : SUCCESS in 14m 37s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/3e1ff313a8f742d0a3294ba246840acd : SUCCESS in 1h 02m 34s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/9464251dafee4588ae486042f8aef3c1 : SUCCESS in 1h 04m 36s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/ba130062ef6f4fdc846637de810704c4 : SUCCESS in 1h 04m 21s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1769d90fc08744f0810b79179751e8e9 : SUCCESS in 1h 04m 34s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/04c06caa873d41bd9839f9b8397064b2 : SUCCESS in 1h 05m 09s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/fcc3004fe6c84fff96fa4511aeb4a4db : SUCCESS in 1h 45m 02s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ee7d8d337d7e473f9d79a02fe5eeafb4 : SUCCESS in 1h 47m 56s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/cea3ee1106264eaebd34c0a79206e941 : SUCCESS in 14m 26s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/2bf301ec6e704b7aa05dee202d1694d4 : SUCCESS in 24m 58s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/aebad6d182e143ce90402676e7187ccb : SUCCESS in 22m 31s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/4821bf9dbdcf4772b55749037955219c : SUCCESS in 21m 50s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/bfe22a746894490d830fda744774d245 : SUCCESS in 40m 42s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/433c00d189284f51adf00b4b4d1ee4a9 : SUCCESS in 41m 17s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/78869ecceb6e4b49acfa1043e6dd4ddd : SUCCESS in 14m 16s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bc7f88ab4ec7406a828b382dcda38c40 : SUCCESS in 21m 28s","accounts_in_message":[],"_revision_number":3},{"id":"e95486adf5c185ee7d1be0b38b6e79051f90e7f6","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-07-10 18:12:16.000000000","message":"Patch Set 3: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":3},{"id":"196a4e9afa74db789128ddd46dc089c20bf70586","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-07-13 09:15:45.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"6d2329e550dadfb8c22f1bf66e62b8d2aaf4aa59","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-07-14 07:57:35.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"5b49826f88b0acf11fc65528f9dca615d76e43c5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-14 07:57:53.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"1419a2d19e59935719c56537270134f444c182e2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-14 09:46:58.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/35d9bcd0a6a444a285e876c557d73d58\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c06ef47de957447ba073ba6025a96f8b : SUCCESS in 9m 45s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/c6df915758c945b6a96844c66ad1c34d : SUCCESS in 7m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/59dd046b0b4348be8dccf728bbe5c28f : SUCCESS in 2m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/240b1727257a4b2497f1c91c1ae796b0 : SUCCESS in 4m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f94a3a0d24cf49bbbd1b88b4b9a0da05 : SUCCESS in 6m 42s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4ad659fda06d40d192e201b2a799a5c9 : SUCCESS in 5m 47s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/48513a2954f14534b5395ad57f3fdb51 : SUCCESS in 13m 04s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/9c4cf2bf58154c47a9ce9b5e7302d546 : SUCCESS in 12m 07s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/71af1220577f43b8ac04f393c2524b10 : SUCCESS in 19m 43s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c7fb320f26104b28a6d725e7c470da2b : SUCCESS in 16m 36s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/5be0508a13ef46149fc8202432301cfd : SUCCESS in 17m 51s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2d3803f95be24733a743eeb331396c04 : SUCCESS in 18m 27s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/4b66c9d4443e4b9f86e84d5127ddcd1e : FAILURE in 12m 16s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/92834b32e95444fe87684208f38b5b33 : SUCCESS in 19m 22s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/6a3bd94ec1c94da9b248466cdec58128 : FAILURE in 49m 14s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/4ae95c24675a4e76a12448b3654ac9b2 : FAILURE in 50m 29s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/76733ea5e4fe42258a31a51c4961fb35 : FAILURE in 58m 23s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cb6cc250cd1e44d587abb342a544bb28 : FAILURE in 54m 22s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/2a9c1484621541698f4754c7b2fec0e9 : FAILURE in 48m 48s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/a6402f331392472f801e13de8abf99ab : FAILURE in 1h 46m 33s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bdbcbae5836043cbabbc1d58fb279a6e : FAILURE in 1h 48m 17s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/7cb374ea17f54e0b887ce5fcc616caa3 : SUCCESS in 24m 17s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/61193f5c3e4e43e093d15780faed7989 : SUCCESS in 28m 25s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d983e0624e7846deb74553c78f1ad28e : SUCCESS in 26m 46s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/5e42e1e360264f698f3f306f2a22bfe6 : SUCCESS in 22m 31s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/81b63b8013e24540a2ecd54f13b017d0 : SUCCESS in 36m 15s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e4cdf74d17694d66b37e82fd1005bb7c : SUCCESS in 31m 02s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/b5cc51f3c9b84ef98d4106a23dc36968 : SUCCESS in 18m 35s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ac066c880cb048338d75d24b64a2d05f : SUCCESS in 21m 04s","accounts_in_message":[],"_revision_number":3},{"id":"524c278748e590c0895501df879f147466f12c30","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-07-15 17:30:48.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"e853677b0de6beb103cf704084d01eee128920c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-15 19:21:59.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6cfd46ea619a493caa68f1890518f7d0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a21b44d214414871b304ea35a1c32a68 : SUCCESS in 10m 26s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/edcd0403c3304c0085a12e4a48213edb : SUCCESS in 6m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3874bc2d13824c82a8665a7fcd8a05d1 : SUCCESS in 4m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/17a490cf40ec4174b0c1557c51122087 : SUCCESS in 2m 59s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/281b1152fab9420fa710f1fd46874292 : SUCCESS in 4m 17s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/99b3f8bb649047758d79e99f122b7601 : SUCCESS in 4m 35s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/79a7c744c720449d8104404318184817 : SUCCESS in 5m 37s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/932b28ca4c46476fa77c48dc4d59bc5a : SUCCESS in 5m 12s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/45ae29aeb36c4570959e17a56b678adc : SUCCESS in 14m 50s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2484cc9eb4db478a9393e9b04f09fe79 : SUCCESS in 7m 07s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/1ac49f4a51cd4a8ea07261d9efca02de : SUCCESS in 11m 55s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/d96426f2bd37438787f0289bb68b446e : SUCCESS in 23m 16s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fd165597cf4046418294c41d347aa313 : SUCCESS in 14m 42s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/18bfd7789ac24ce1914f22b875b8cbd8 : SUCCESS in 10m 34s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/39c8180b18784df1bbdf32479ae29a41 : SUCCESS in 16m 04s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/aa188af7377e40f1921f1de80b3736ea : SUCCESS in 21m 19s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/6f38ea010383414594ce3b1b06df8cde : SUCCESS in 10m 54s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/558f635043ae48ce940900fbaa3aae30 : SUCCESS in 17m 48s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/c6a6d3d1a6d64348bc56fbd51ee12618 : SUCCESS in 1h 04m 34s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/06af3361119f4da0ad444b69da27f6b6 : SUCCESS in 1h 03m 18s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/cef2b3026a704a18bef39d661db8cb17 : SUCCESS in 1h 05m 32s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/45378349213d4843b22d5b3695025612 : SUCCESS in 1h 05m 22s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/613363326602419c852abbe02f16ba35 : SUCCESS in 1h 05m 35s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/1ada0bea0a874882a0bfe1ab6eaf4574 : SUCCESS in 1h 45m 57s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8d8137df87ce45c4ab56fddffd9abbe7 : SUCCESS in 1h 49m 57s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/f44f65a30d7449dc874fec57265dafb7 : SUCCESS in 14m 19s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/017f361acd074d378be1be4baeccc6a0 : SUCCESS in 15m 35s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/464b27f871534a7591eb7821b4c3d48a : SUCCESS in 22m 23s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/06c6541f4d0e4f9bb1b8e948e4454c0d : SUCCESS in 19m 57s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/e48edc11e01c4d1398c790ba8e6f875e : SUCCESS in 22m 47s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4607fe14b10b4d8184daae5700ddce04 : SUCCESS in 20m 38s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/071485d46492410dad2406a669c3382f : SUCCESS in 19m 52s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/16f0678dc82a47d7920b852ffd8753b1 : SUCCESS in 19m 26s","accounts_in_message":[],"_revision_number":3},{"id":"9f500cb9a1a33e2acc00eab1cdc9d60668a4ebc2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-15 19:22:21.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"6fdaeb86cff66c33556cf024326cae23f8d1c9e8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-15 21:09:58.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1acb48152f4844868f9a086f4005ad2b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8d75acbdddd4ca79ec2dc59d8ea97b4 : SUCCESS in 9m 37s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/ebc3a5f17ebf4c4794b51d63e59bdb49 : SUCCESS in 6m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff0d1381ffb9485d88e1424cfb8bc9a0 : SUCCESS in 4m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e986408c078d424fb323b35e93d8543b : SUCCESS in 5m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4ae92970d26d4c9d9a50292de511f144 : SUCCESS in 5m 40s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9737cbdbf18d4df6acd96726525139c7 : SUCCESS in 3m 44s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/94171986bbf5421da6066c75d7ac17e8 : SUCCESS in 11m 02s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/307e412b030e467b9535c85d086b5a8a : SUCCESS in 9m 51s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/ba75c0b322c14ecb8ad9465f9115318e : SUCCESS in 11m 32s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5de38635851940d3a443594ac3db4f72 : SUCCESS in 16m 33s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/77673e32a81643ec98a46803d6efdbae : SUCCESS in 14m 03s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/047b91406e814c82927bce7f0d2e1e0d : SUCCESS in 19m 53s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/03e9798c92fb49c9bcd086041784da0f : SUCCESS in 12m 55s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/72e1d77a6bf84c75aed924ec0e27d3e0 : SUCCESS in 10m 43s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/a6befd65d8e54c039252cd888cf32140 : SUCCESS in 59m 20s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/47ca9773dd8a4795846e045e57ab874c : SUCCESS in 1h 00m 29s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/52eb8efd60d74ab3aedde31250929bf3 : SUCCESS in 1h 05m 52s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1daf06572c9e4d1d85152408aeaec7b9 : SUCCESS in 1h 05m 43s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/396ee7d173b04b4e9f0ae9609f391d81 : SUCCESS in 1h 05m 49s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/a95eb79e01e34befbb761a762b900487 : SUCCESS in 1h 44m 38s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/68a258130d6b46cea48fe08b71938270 : SUCCESS in 1h 47m 05s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/cacee573cabc4ff48f70735a10c2b451 : SUCCESS in 29m 13s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/43bac918d06f40f29fa8d237e470965b : SUCCESS in 26m 28s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b6acf8ff9616438c9933888cd3af6ec1 : SUCCESS in 13m 47s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/ecf5b996a6ec4c8cb60bfcec76136b30 : SUCCESS in 21m 40s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/02fd0376af3944569b1701e6faf5f141 : SUCCESS in 31m 38s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5c26614913e544bc85a40f086a79d8d3 : SUCCESS in 24m 24s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/0bcd39dc52024a6082384ddb57d8f395 : SUCCESS in 15m 31s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0c055e02cccd4480830dfce4da2ed7d7 : SUCCESS in 21m 09s","accounts_in_message":[],"_revision_number":3},{"id":"667e838d65fc84e81e6b35bce105ac3c9059e842","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-15 21:09:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"b8bb449dc23ceebb9c26b6c7226b654cd45e27d6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-15 21:11:01.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b9f83f90c1e140e285196775f6e96b2f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3013f21cbf894c6482842a8c821adcc6 : SUCCESS in 50s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b78db41d740944a1898bffa0dd8986e8 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"d91f7e3b468becafb40868f45a59d5281722e461","revisions":{"55962275355a1e2aa62a6dcf156599e3134187c3":{"kind":"REWORK","_number":1,"created":"2026-06-08 10:33:09.000000000","uploader":{"_account_id":37122,"name":"Sylvère K","display_name":"sylvr","email":"sylvere.kanapa@univ-lyon1.fr","username":"sylvr"},"ref":"refs/changes/04/992204/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/04/992204/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/992204/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/992204/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/992204/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/04/992204/1"}}},"commit":{"parents":[{"commit":"95e47148f1a2aa14429fe72c3879d7e6157789e9","subject":"Merge \"CI: Stop using image from Docker Hub\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/95e47148f1a2aa14429fe72c3879d7e6157789e9"}]}],"author":{"name":"SYLVERE KANAPA","email":"sylvere.kanapa@univ-lyon1.fr","date":"2026-06-08 08:25:39.000000000","tz":120},"committer":{"name":"SYLVERE KANAPA","email":"sylvere.kanapa@univ-lyon1.fr","date":"2026-06-08 10:32:54.000000000","tz":120},"subject":"Adds seed IP to default no-proxy list","message":"Adds seed IP to default no-proxy list\n\nUsing self-signed certificates and an externally managed proxy causes\nconnectivity issues when deploying the seed services.\n\nCloses-bug: #2154438\nChange-Id: I9552c722cdb0215a5b4a0d7b65cf9a2239573efc\nSigned-off-by: SYLVERE KANAPA \u003csylvere.kanapa@univ-lyon1.fr\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/55962275355a1e2aa62a6dcf156599e3134187c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/55962275355a1e2aa62a6dcf156599e3134187c3"}]},"branch":"refs/heads/master"},"17a6ec5d9f63c7f09276c4749eae9cbb8db9247c":{"kind":"REWORK","_number":2,"created":"2026-06-15 08:51:28.000000000","uploader":{"_account_id":37122,"name":"Sylvère K","display_name":"sylvr","email":"sylvere.kanapa@univ-lyon1.fr","username":"sylvr"},"ref":"refs/changes/04/992204/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/04/992204/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/992204/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/992204/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/992204/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/04/992204/2"}}},"commit":{"parents":[{"commit":"95e47148f1a2aa14429fe72c3879d7e6157789e9","subject":"Merge \"CI: Stop using image from Docker Hub\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/95e47148f1a2aa14429fe72c3879d7e6157789e9"}]}],"author":{"name":"SYLVERE KANAPA","email":"sylvere.kanapa@univ-lyon1.fr","date":"2026-06-08 08:25:39.000000000","tz":120},"committer":{"name":"Sylvère K","email":"sylvere.kanapa@univ-lyon1.fr","date":"2026-06-11 09:42:42.000000000","tz":0},"subject":"Adds seed IP to default no-proxy list","message":"Adds seed IP to default no-proxy list\n\nUsing self-signed certificates and an externally managed proxy causes\nconnectivity issues when deploying the seed services.\n\nCloses-bug: #2154438\nChange-Id: I9552c722cdb0215a5b4a0d7b65cf9a2239573efc\nSigned-off-by: SYLVERE KANAPA \u003csylvere.kanapa@univ-lyon1.fr\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/17a6ec5d9f63c7f09276c4749eae9cbb8db9247c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/17a6ec5d9f63c7f09276c4749eae9cbb8db9247c"}]},"branch":"refs/heads/master"},"d91f7e3b468becafb40868f45a59d5281722e461":{"kind":"REWORK","_number":3,"created":"2026-07-10 10:15:01.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/04/992204/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/04/992204/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/992204/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/992204/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/992204/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/04/992204/3"}}},"commit":{"parents":[{"commit":"95e47148f1a2aa14429fe72c3879d7e6157789e9","subject":"Merge \"CI: Stop using image from Docker Hub\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/95e47148f1a2aa14429fe72c3879d7e6157789e9"}]}],"author":{"name":"SYLVERE KANAPA","email":"sylvere.kanapa@univ-lyon1.fr","date":"2026-06-08 08:25:39.000000000","tz":120},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-07-10 10:10:10.000000000","tz":60},"subject":"Adds seed IP to default no-proxy list","message":"Adds seed IP to default no-proxy list\n\nUsing self-signed certificates and an externally managed proxy causes\nconnectivity issues when deploying the seed services.\n\nCloses-bug: #2154438\nChange-Id: I9552c722cdb0215a5b4a0d7b65cf9a2239573efc\nSigned-off-by: SYLVERE KANAPA \u003csylvere.kanapa@univ-lyon1.fr\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d91f7e3b468becafb40868f45a59d5281722e461"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d91f7e3b468becafb40868f45a59d5281722e461"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
