)]}'
{"id":"openstack%2Fkayobe~992288","triplet_id":"openstack%2Fkayobe~master~I12a9c874283856181e397378a2336efdf68a28e7","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2026-07-15 09:20:14.000000000","reason":"\u003cGERRIT_ACCOUNT_23084\u003e replied on the change","reason_account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}}},"hashtags":[],"change_id":"I12a9c874283856181e397378a2336efdf68a28e7","subject":"CI: add Rocky 10 aarch64 jobs","status":"NEW","created":"2026-06-08 15:26:44.000000000","updated":"2026-07-20 15:35:10.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":4,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":true,"meta_rev_id":"2e1ab11abbb2ed90f8277d16208dc1d6058d8a1e","_number":992288,"virtual_id_number":992288,"owner":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-07-20 07:38:43.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}]},"pending_reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}]},"reviewer_updates":[{"updated":"2026-06-08 16:47:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-06-08 17:16:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-09 07:52:32.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"}],"messages":[{"id":"3cf41164da929a3f8d29af7b5f592578250a55e5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-06-08 15:26:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b736b3c6b5453d0a7864de98a5b1b79c0512ae73","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-06-08 15:26:58.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"ae38cb55c93ec42ddd54d5241b59b27b3813186a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-06-08 15:29:41.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1650492cee56378dcfaa8d5100279b3d7bf13d7b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-08 16:47:49.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d1ef28789af147e5a33ee702502912e4\n\n- kayobe-overcloud-host-configure-rocky10-aarch64 https://zuul.opendev.org/t/openstack/build/c0716e1837874f1c9d68e94803067505 : SUCCESS in 35m 36s (non-voting)\n- kayobe-overcloud-rocky10-aarch64 https://zuul.opendev.org/t/openstack/build/400e517d1dca4fd5ac8d568699461aff : FAILURE in 43m 42s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3fe9d137abc04c7991bdbfc159266767536fdd7e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-08 17:16:12.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a4985a71f16f47d9ae438a9ff644f04d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f25343ee4a89479eba44d0a33142d26f : SUCCESS in 9m 28s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/36ed75916df3488ba5f45515d739ef83 : SUCCESS in 8m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7fd5d6d74a1341d2971205d812b7eb1e : SUCCESS in 3m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0cc057f002ba4f3d899a698e98ff55bc : SUCCESS in 4m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/61295cc0f60b48b1aa744dd53823c01f : SUCCESS in 4m 45s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/26adefa9f86d4d3fb6ba4ad2aefde314 : SUCCESS in 8m 43s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/917f24def16d44e8918ca9fd5999b224 : SUCCESS in 7m 01s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4ed15b72ae034e5db00e41ff8c5ce5e7 : SUCCESS in 6m 17s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/97d93e9d80a54ece8590dfa0042b7083 : SUCCESS in 12m 46s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2be9729eaeed4ba79ec0b9ce2cabdee2 : SUCCESS in 11m 06s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/aca0ac2ec9604bd49d5145af1bb75e2b : SUCCESS in 17m 45s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/af2559834ca14cb6a5f20f863ba5ebc4 : SUCCESS in 14m 40s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/38225c18fc7141e9987a263fb21a0f60 : SUCCESS in 14m 45s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/5d862348e1dd4819aa7d52c674c88b01 : SUCCESS in 17m 58s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/001bac9afb084eaea95d593eb5c6c547 : SUCCESS in 13m 42s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/09a38c25c61a44b7a20df36c4e4a9af9 : SUCCESS in 15m 34s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/27ea11fbbccd4220b954b79f460348eb : SUCCESS in 18m 36s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2dd06d27b493469ab20dda9d1073593d : SUCCESS in 18m 15s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/2e206ceeb8df43e994535c6cc0c0c32b : SUCCESS in 58m 09s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/450d73ddab9d4fd2a78523fb8bfe254c : SUCCESS in 1h 02m 18s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/9cc35c2150314b319b17ae2faae56447 : SUCCESS in 1h 07m 29s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f054936bb5ab403a8d19626233c80379 : FAILURE in 29m 09s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/1f42cf255fd444c98892ba84cc074c02 : FAILURE in 29m 07s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/6d5b3bab6609479b8c75ed3257ffb8ee : SUCCESS in 1h 44m 19s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bd21c6f5ea22462dbd06b4d18cdc5dab : FAILURE in 32m 00s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/4a27a99f13984006a7669c19940ca9f6 : SUCCESS in 14m 38s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/2a3eec8ee71f4d569827b62b7e4649ff : SUCCESS in 27m 03s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e7bfd5cbecec43d08831b1440de510d7 : SUCCESS in 25m 21s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/82b7627e91c74aac8449fb3c02b5bab5 : SUCCESS in 23m 18s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/fe73675a7ed3484486257a9be0157dce : SUCCESS in 21m 14s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6a26b5206d5c4ae78077f617c34ab179 : SUCCESS in 41m 13s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/c3828f2583df496781a4dde0baf3a82a : SUCCESS in 26m 39s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b4c3e9485fa34687901320364152c757 : SUCCESS in 19m 45s","accounts_in_message":[],"_revision_number":2},{"id":"eb876f929c792e6625185a25217c294106e39809","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-06-09 07:52:32.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"596d29f84e839594bcfc9293246f9a0deeece7f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-09 09:38:12.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d2c033c458f64730b4ff55a1a86dd884\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c394c58adf7c4c77ab3714e24afb75ac : SUCCESS in 10m 41s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/ad4fdc800cc44833a0642fdd887ca1c1 : SUCCESS in 6m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7e0cf2abca2644749786e205c03d2552 : SUCCESS in 4m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d88a6a707c204173b62c4d262a0b18d6 : SUCCESS in 4m 40s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b4b21373a27248c0a5fe7d0358b12ebf : SUCCESS in 3m 53s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/d3bf01ced47546fd8bacc08f2977deb7 : SUCCESS in 7m 08s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/86dfd6de9f584685b5ad8e0adae97d50 : SUCCESS in 6m 25s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/570533cb0d7a41a9979363bba2c9d999 : SUCCESS in 5m 41s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/74f4fe6b7bf7409f97a7a20c8a5708d0 : SUCCESS in 12m 15s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/afd1673659b140a980c24f45c57663de : SUCCESS in 13m 38s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/91f95c4be3f340b7abab9158b13125f2 : SUCCESS in 20m 53s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/8f8a4e040102482ab5d6758ae15652c8 : SUCCESS in 17m 32s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/43e8d9b0fa6c4eb0887b442ff0495498 : SUCCESS in 17m 48s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/0846646dd7ad4a648c6cafa41be5d9f1 : SUCCESS in 16m 56s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/a6f5235b37c44944b9d5f60e9537992d : SUCCESS in 14m 04s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6c1ae2e7f28246d0b32c569935252e78 : SUCCESS in 19m 22s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/a35ba3613bf6498aa3673ef60884cf5b : SUCCESS in 23m 59s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a611e97540064be1878a2912803f0146 : SUCCESS in 18m 59s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/15e5b091adf5435693fdbaeaa5b0f40a : SUCCESS in 1h 07m 49s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/7284f36cbe0c418397e4c25253dbcf61 : SUCCESS in 1h 02m 19s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/db2cd9cd205d4ab0b2d08f4d076256bf : SUCCESS in 1h 04m 16s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c34d9185b013415690f2ca14e0c16ce0 : SUCCESS in 1h 03m 25s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/747b144cbce64b1abe1f9aadf8455e2a : SUCCESS in 1h 02m 49s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/86719faeddde4b83b0b7017d944aebe2 : SUCCESS in 1h 43m 59s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/517753390bce4e569c93f0d73029911a : FAILURE in 30m 51s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/6a8f69945e3441fd8ce9ba83ba4cb3e4 : SUCCESS in 23m 06s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/2e472f34a2c34ee2ba7e39a1df41cca1 : SUCCESS in 24m 10s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b56ab8bbcb82429bb488c2fc1fced598 : SUCCESS in 24m 25s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/fb6b7035317c4ec3899aaad00cfd5396 : SUCCESS in 24m 44s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/b4c2d22c0d8441e69115043df23b6b35 : SUCCESS in 37m 52s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7f078223dc4b4f6ebda2ab7a3d5a7c99 : SUCCESS in 35m 47s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/8eb24bbe10634fcdb2a05ec9fcf3f13d : SUCCESS in 21m 21s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/66e8dddde5c14ad4a06deffa5744e1e5 : SUCCESS in 19m 19s","accounts_in_message":[],"_revision_number":2},{"id":"a12368c435f7f34f038466bf69e54001de106cb5","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-06-09 10:22:45.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"cfd05132bb4d6657d40695f12a57e2e60420c93d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-09 12:08:59.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f068aaabab3a42b1bf61fc2e005b4b09\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9da73e46a1184fb1beea2e36c1351e8d : SUCCESS in 7m 36s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/c504ba8e834049349e8e0c8bbd958988 : SUCCESS in 6m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5d56bd26bf3b41b89d413de53fa69fb6 : SUCCESS in 3m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d05c64b5e8f44cd4a17c063273ec9a85 : SUCCESS in 3m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/29fc9efc37564144970736b692a3228d : SUCCESS in 3m 37s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/0d3fbb7293fe4d9e869148bf4846966d : SUCCESS in 4m 54s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d2e29ac8e6e549c69e70c4bec7cc7d4c : SUCCESS in 4m 50s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/845817f0da654e4eb69de9ffea3e01ff : SUCCESS in 4m 10s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0ccb7e827f9c4a0f8d8bb1dd6139daf1 : SUCCESS in 10m 15s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/968880d7e2de45fca5220cdd5ba0e539 : SUCCESS in 12m 20s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/0d052c14290e461a9456bbcd4f4275d1 : SUCCESS in 21m 25s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/e70f0f3e0537412db8c1de5be07e063e : SUCCESS in 14m 34s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1b7de4da1e914440bdfd36163923b29b : SUCCESS in 10m 27s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/6e23db67f9b44712b77cac1d21c852f4 : SUCCESS in 9m 51s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/f4f5b52470624fd398918ee61fed30f0 : SUCCESS in 21m 35s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dbba285be08e49d2b73456ac04a3807a : SUCCESS in 13m 54s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/923d4775032449d3ab357b7ee3cc6a63 : SUCCESS in 11m 06s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3c92224122174dc2a8d8cae682544822 : SUCCESS in 10m 10s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/47fa6e8e2c8e4f719a612c090d4eff51 : SUCCESS in 1h 09m 56s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/af24e2a8af1f49cf9bf0b60395882843 : SUCCESS in 1h 04m 18s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/40257bf051e64c218d3f58d2c3545780 : SUCCESS in 55m 21s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d85c76ad13e24c2f91f21eb40dcb56b1 : SUCCESS in 1h 04m 30s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/7fe463e8182740adbe6c2f5e32617821 : SUCCESS in 1h 01m 43s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/1e5ef53fb3914edcad919490f9fcedb5 : SUCCESS in 1h 44m 37s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/434b61d3e5944cfcbcfcb04804e87f79 : FAILURE in 31m 05s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/dc52eb2a4797473aa212cf844a03ff37 : SUCCESS in 21m 59s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/8212e6ada3fa4465a16bd7a0ba9edd79 : SUCCESS in 14m 23s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/54c494e944034c5885316da654f76383 : SUCCESS in 20m 43s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/52b71d1ff3084c2f87cf9393a02cde8a : SUCCESS in 21m 08s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/f9b6130633c14bbb8b9834169cb3fe4c : SUCCESS in 20m 10s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/29fab3e44b72481d9e96a635ccc958c6 : SUCCESS in 41m 43s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/993e03898bfc4069856912cff40e69fd : SUCCESS in 20m 44s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5ef0dd75e3b8492582153124815ae335 : SUCCESS in 20m 55s","accounts_in_message":[],"_revision_number":3},{"id":"fb4c061a21f512f26bb72d090e7964d5dcbea95a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-09 15:33:03.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/82331819c1fb44c793fa7504903e6458\n\n- kayobe-overcloud-host-configure-rocky10-aarch64 https://zuul.opendev.org/t/openstack/build/d05645cb1b564d0695dbab2003422c9e : SUCCESS in 31m 15s (non-voting)\n- kayobe-overcloud-rocky10-aarch64 https://zuul.opendev.org/t/openstack/build/4f6092cf95d149a8b84b11b72e14de3b : FAILURE in 39m 15s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"ab883f4b62536d23a02a7a43b03279a58cdeb79b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-06-10 09:07:50.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"2f9f8f65c97a4b50c3fb44394af1db7390ec4691","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-10 09:50:54.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/41e49df9720b4033b986ca546c039ba5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45af767e9282438e9415c2d9715cfe7d : SUCCESS in 9m 33s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/db3db7403f164db192770af736cbd703 : SUCCESS in 6m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9c5a317743ab4df08bffd035be221d9b : SUCCESS in 4m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/948d193db84c432a905e7f0af4e9f7dc : SUCCESS in 4m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/453a8eec0ac445dbaccd36829f475dfe : SUCCESS in 5m 21s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/3c6d706e880643c19c6f1859d9b8dff6 : SUCCESS in 4m 30s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/479013056dd949b7a27913af54ada2a5 : SUCCESS in 2m 56s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e7ce9abfccd047f5bd1360df5cb9ecae : SUCCESS in 4m 48s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0956ba178c18427b8c652b742025d26f : SUCCESS in 9m 56s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1814a22eafbd45bea7f0853616bf1bf8 : SUCCESS in 6m 16s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/164b334524e54301974e931ff9a27d38 : SUCCESS in 16m 44s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/f9ba56bd4d294ef59de0f07b27a55ad2 : SUCCESS in 15m 34s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/51008525c79f4da195a2b89551d38a97 : SUCCESS in 16m 12s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/2fed3c87b2ec479f8b2950fddcde70cf : SUCCESS in 15m 53s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/f77a7ca8eff74d0881d8661e5a3b028a : SUCCESS in 16m 44s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/65f000370aa54cfba34045422fef3711 : SUCCESS in 12m 22s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/39399e7d1b3140e4972b79274edb2f21 : SUCCESS in 11m 49s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c6ff4ab9ff53426cae375e864054873f : SUCCESS in 16m 52s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/417b6768867b43f3883ef5e70a9b8e48 : FAILURE in 28m 59s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/6cadfefdb0184a139837e1e9b58bb770 : FAILURE in 29m 12s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/1c33e61cf864496b8fae4dade2a94e4d : FAILURE in 33m 15s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a8196470f88645c0a77552232b72b882 : FAILURE in 27m 47s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/007a7d841a324b7d8ba4576995912caf : FAILURE in 27m 49s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/bef00f61d9744811bbc69d459ea8bb11 : FAILURE in 31m 53s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5eae59f63adf438989fe53a21f233460 : FAILURE in 30m 11s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/4b140e0b71184dbaa08456dfdb509127 : SUCCESS in 25m 18s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/624d9b51852e4a768d5cc2d70524cb18 : SUCCESS in 24m 18s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/805922564cd04fc28b924c2eb67ec459 : SUCCESS in 17m 44s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/067f37a2fd114796a35424fedaf06858 : SUCCESS in 23m 50s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/721d8686b2004895b93146435b1a6620 : SUCCESS in 30m 09s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3ff75f26c14a48eea64615537ad23c47 : SUCCESS in 36m 56s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/c82cc318589442fea846c96be6e6943f : SUCCESS in 24m 59s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/de00108ca3b24590a2b12dfed428c504 : FAILURE in 20m 05s","accounts_in_message":[],"_revision_number":4},{"id":"70d722bf75e8715737fe458995dbb635b1f441e3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-10 11:10:47.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1b47a9d95ce849309a933e753eaa840d\n\n- kayobe-overcloud-host-configure-rocky10-aarch64 https://zuul.opendev.org/t/openstack/build/cb74a3b086994617a7cdd1fd903a83d0 : SUCCESS in 21m 39s (non-voting)\n- kayobe-overcloud-rocky10-aarch64 https://zuul.opendev.org/t/openstack/build/8d1b0c7364974000970a5447420e51fc : FAILURE in 1h 08m 51s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"9f8a857f1dfef169569562b655ca07192a554b05","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-06-13 10:20:18.000000000","message":"Patch Set 5: Patch Set 4 was rebased\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"17747680d1707ae526a53f2345671eed49e0d1c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-13 12:02:53.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ea00aaafa3c44331bd84f4b896293699\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/871a2fd5c36e4a04995e3c3ba0c1db52 : SUCCESS in 4m 35s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/c003f2525d914a068fbcac8a30a59bdf : SUCCESS in 3m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3f38f04d540b48d7b9a385f2af466996 : SUCCESS in 3m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50268f864dca4ccaac9cc9c1e4025ae4 : SUCCESS in 2m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a345121aa1cc4318b5abd0fece9da0e7 : SUCCESS in 3m 15s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/12c84695db0f4d40a8c15ec07d84abcb : SUCCESS in 7m 26s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ae8389a4708245d6b4b3f225e52ce8f9 : SUCCESS in 2m 58s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/489bd36e79ba4857b34a9b4d8e1cb613 : SUCCESS in 4m 38s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/01800c3fffd6437daaf3d036160e0637 : SUCCESS in 6m 12s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/9c26d8304770414caa892640502ceb48 : SUCCESS in 10m 59s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/fc81b7d866754ff69efb3305a4f5f56c : FAILURE in 15m 01s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/f9208b54a2434d06b53ccabe1e3f3a61 : FAILURE in 14m 47s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/97c2ada687f24f93b7aed8be3b882a59 : SUCCESS in 9m 23s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/498115bc9ee94e369282bf6220e50829 : SUCCESS in 16m 14s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/53b7499b0c684c8489174d45a16b4178 : SUCCESS in 13m 36s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a0db5e4306f34760bd8d6dc3703362b7 : SUCCESS in 15m 46s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/294d0769b4ab46738d3a20f5149a6c27 : SUCCESS in 11m 29s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8f353fd7e6da438a9786bb087a9bbeee : SUCCESS in 16m 05s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/628fe4c5a4f04c69a28a837cb3e10e42 : SUCCESS in 58m 32s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/f10f7744be04479eb24c9fca75c69450 : SUCCESS in 59m 04s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/086dc83cadcd4660bb317ec27857546a : SUCCESS in 59m 06s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/69e22bea91ae4c06a9ed119d7c0020d7 : SUCCESS in 1h 02m 45s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/b0c1a53c8a1d4112bb74e8efaec4c328 : SUCCESS in 1h 03m 06s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/56308be1be2e47198e1ba1d8cf13c959 : SUCCESS in 1h 41m 26s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/570d74e26ba04c8691786c90a0156409 : SUCCESS in 27m 22s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/1ae36bd4aaa6495f9408461f7df746f0 : SUCCESS in 14m 29s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/127c540107534626ba9f5b2783d148da : SUCCESS in 23m 06s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/30a0a3aa6b8c4b0cb31eba00ee235673 : SUCCESS in 21m 04s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/b9b2fadcee124f65b8e8e2db9b892e7b : SUCCESS in 21m 35s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dcdf12d5921d478ca41f999a82daebd7 : SUCCESS in 34m 26s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/4bd1ab04319e45509a01d36f830d79e3 : SUCCESS in 13m 50s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d35a5045188b464baf8c56f030c6b587 : SUCCESS in 19m 19s","accounts_in_message":[],"_revision_number":5},{"id":"b82bb82f75b8b14b74309b2c382f4b2d5cb4a21c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-16 06:33:41.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/423faca87a9b46918a3a8ed1addefd27\n\n- kayobe-overcloud-host-configure-rocky10-aarch64 https://zuul.opendev.org/t/openstack/build/91f5e8427a2348b9a70de78267cba225 : SUCCESS in 38m 01s (non-voting)\n- kayobe-overcloud-rocky10-aarch64 https://zuul.opendev.org/t/openstack/build/5035dd2a455f40dcbafd410ad454d4fd : FAILURE in 2h 00m 54s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"a29b398421d9d3b3042126a260c32bf4b62dbdea","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-06-18 08:04:11.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":6},{"id":"0a55597fe765eab56cab03399ab769eb6475a6e8","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-06-18 08:04:23.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"fa53798d18cfd79e80821c337911d91bb6117d8d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-18 10:03:00.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c6584d280475427ea82ab3c80fdd4711\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/111da1f5b5194e7d9460d4e513d1bbaf : SUCCESS in 5m 39s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/92c5041be15d45638db4a10dec3adbc2 : SUCCESS in 4m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/272ca811371645a29c82d24f54bee007 : SUCCESS in 4m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ab15dc065804020a3d98ff767f3db9e : SUCCESS in 4m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a9226659c4174b3d9008cac2e1a27ffa : SUCCESS in 3m 00s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/340900b35416445b988b48ed57664a13 : SUCCESS in 4m 29s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ebe61de62807405a9dc48a5ed4dee028 : SUCCESS in 6m 00s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/56a7342326ba41d1ae1ea8ae2c94377e : SUCCESS in 7m 05s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/bff05d9c37104cc3ae97d48e2c3eeabe : SUCCESS in 6m 26s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1f337fc6c09f4f9383777b1ccacdac95 : SUCCESS in 12m 24s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/6602af22e02146e8ad13e51ad5faee44 : SUCCESS in 12m 48s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/7b5cd0815c7a41dea84ef180fe2b7875 : SUCCESS in 24m 26s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b715c75a84a249f18ef83194f4618261 : SUCCESS in 18m 11s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/772b5635992a449bb751052d2908fc71 : SUCCESS in 15m 17s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/187d98a41d79412d8d7273ad64d8f226 : SUCCESS in 20m 28s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c7ea769426394a909e027de87026bb76 : SUCCESS in 19m 04s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/96ad2faf6b864e548d9aea05b73684db : SUCCESS in 26m 59s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4b18c2ef54ba42f78a37fb61894ccf83 : SUCCESS in 17m 16s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/7930630717684c37813da70f187ae3cf : SUCCESS in 1h 13m 42s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/f7afad677b174b0fb19965217be4973b : SUCCESS in 1h 07m 37s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/3592a7636f2e4c4daa3d23c4654bb972 : SUCCESS in 1h 06m 38s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3f19ad8f526d4d6d9685e8b1a6daa56c : SUCCESS in 1h 03m 59s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/6bf597c76a974407a863487e71b421a9 : SUCCESS in 1h 10m 06s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/9507d30234ed4e4e811c44dfd1192c36 : SUCCESS in 1h 55m 49s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e5660db3cb6b464fb465ac2e2f198157 : SUCCESS in 1h 55m 57s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/0f01ef880de343d88808d4e04465bf59 : SUCCESS in 25m 52s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/5a0389e9545543bea37357640c70a748 : SUCCESS in 16m 18s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ddb707a251464fcca8176dd8c1697f30 : SUCCESS in 19m 37s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/05390feda8014d988f113a9a961fce9e : SUCCESS in 14m 34s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/391e11eb25f64b67a421e0282afa7ea5 : SUCCESS in 22m 13s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e1ec8c04a16342d398805548a99d3ebe : SUCCESS in 35m 30s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/6f3a7ae15edb41839f416d15768a1c57 : SUCCESS in 16m 03s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/68157c0ddc424fd797ee78e0f5525f5c : SUCCESS in 14m 03s","accounts_in_message":[],"_revision_number":7},{"id":"bdadb9016259aeeaafa10d763e64e008fa4fe40a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-18 10:12:27.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0e17f33214054b8e890f2cc8ba0a2ce5\n\n- kayobe-overcloud-host-configure-rocky10-aarch64 https://zuul.opendev.org/t/openstack/build/444118af602540b8bd84df8d0d1c5ba5 : SUCCESS in 33m 18s (non-voting)\n- kayobe-overcloud-rocky10-aarch64 https://zuul.opendev.org/t/openstack/build/8418c2b3bee84de69c32b2f9d8a63234 : FAILURE in 1h 45m 03s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"bd3cb059c96962fd3899d34904cc13c190bf247e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-06-19 08:38:50.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":8},{"id":"7f6896b17bf19d05dcca936a73ae49a819ccde7b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-19 10:31:37.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b1ff128f80fa41478ea6235ee4328a97\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53956ca5370b4e0198f4778068c1b232 : SUCCESS in 9m 29s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/0eca5f4b8b6c4feaaa66938361689a62 : SUCCESS in 6m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/457d7e1ddb424f4c9b0bcb06e2e968de : SUCCESS in 4m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/368f2b9e9a7640b29bfdca875065092e : SUCCESS in 4m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/003835f22c084c40a5a41dd11d6d09ad : SUCCESS in 3m 28s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b85d8f0a8dba44d29bfabd1c8fa34013 : SUCCESS in 7m 19s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7a23b879abc343658f1a0fe71e05a05a : SUCCESS in 6m 10s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ff8124b80749425997abe997d5f72b6f : SUCCESS in 5m 20s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d5142f4e457446f7951ddadcbe6b46bd : SUCCESS in 7m 17s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a9176a9f256c4b2680293926a25b4a0a : SUCCESS in 11m 46s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/fde55c4e565b48768851d0d357749893 : SUCCESS in 15m 48s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/5bbf0eb245fe431f993156925b27ba76 : SUCCESS in 12m 35s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/43e17041b6c94cf3959ebc1d832889e3 : SUCCESS in 15m 15s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/b26e2995873e4955b623b2040431b839 : SUCCESS in 9m 57s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/16afabfa3df547589eda472207c220a9 : SUCCESS in 20m 58s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/58bf6b4e934348af9e25fe466756b8f9 : SUCCESS in 12m 27s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/d9400bcf237743f88d6ef0f1486bbdba : SUCCESS in 24m 55s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0ac368a74fc847089cbea08d0a718fc9 : SUCCESS in 15m 11s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/7db77924ccd449888f5144363658b5fd : SUCCESS in 1h 05m 11s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/bb5b363b049c4da6af7c8adb6c485dd9 : SUCCESS in 1h 04m 20s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/fe283980d85940838fc66ee19e63d135 : SUCCESS in 1h 04m 54s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b75b18caf271419fbde4f036b41507e0 : SUCCESS in 1h 05m 43s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/83961dd235b346cfbd66251c73ae1e8a : SUCCESS in 1h 05m 25s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/1fc15eaaa8b84b20a09ed502eaa5e038 : SUCCESS in 1h 43m 56s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7f91ce36a3514ee6b532b113bf2916f2 : SUCCESS in 1h 51m 25s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/fdb16d1651ad40c29671c2fb1bc94945 : SUCCESS in 24m 08s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/f709a811264148cb91d0f0b791c6fe8d : SUCCESS in 14m 47s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/537bd3e5cf524d85994555324507374c : SUCCESS in 21m 26s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/e5fb5d2e0a4c43a2a6489dcb6ebb8bcd : SUCCESS in 21m 28s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/e78044fe40bd46ca86af520848971763 : SUCCESS in 44m 09s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fd1d1ac03a6a4362a1305d26fa74f359 : SUCCESS in 35m 23s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/92b49dda82494adc928088b5257d9a0d : SUCCESS in 22m 10s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/74eed9088b234e9fb4ca444189b43142 : SUCCESS in 20m 43s","accounts_in_message":[],"_revision_number":8},{"id":"5285f6c85e508bda4fa9b100330a25d46711443a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-19 16:17:23.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9fed67bba0934c83ae13e9daeb4d659d\n\n- kayobe-overcloud-host-configure-rocky10-aarch64 https://zuul.opendev.org/t/openstack/build/5a2445aaa11b4888991ce216dd0bdfa7 : SUCCESS in 38m 35s (non-voting)\n- kayobe-overcloud-rocky10-aarch64 https://zuul.opendev.org/t/openstack/build/bd0fa55a11b846e4a3785bbdd91a2a7d : FAILURE in 1h 37m 50s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"a5406092ee72a186ecaa69aaf6b3d7f8e77a913b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-06-22 07:40:22.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":9},{"id":"44e6128709d97eb7092af07bb13a8536173a2578","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-22 09:31:05.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8b7ab17a97ae47b0be87593f3a203d9f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41f0fcbc18944c6baf902ce2a0c643c2 : SUCCESS in 9m 15s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/5e51b5e4033642d0b8502b7cdd557137 : SUCCESS in 7m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ac4ee8725d584a149626203ebb570225 : SUCCESS in 4m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/888fef4090e7498c80c93e5dead62e4e : SUCCESS in 5m 00s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/50a95c03c9674ac5a3af49a283f258b9 : SUCCESS in 4m 45s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/3510baef353a415a8adf63fb330e0b7c : SUCCESS in 7m 08s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7a34fb7e109140f3add2c3bab71d1bd6 : SUCCESS in 5m 45s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/af75bff4b5074894ab1a39be4ab4b82f : SUCCESS in 4m 59s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/cd336c962d044f67b28869a19eb143e4 : SUCCESS in 11m 49s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/22d45685048f45c8984e16afc577efed : SUCCESS in 11m 31s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/0409280d6ffd46728156c22d5801326d : SUCCESS in 14m 34s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/66430a9f4ab44f4fa0cce1fdec66213a : SUCCESS in 22m 13s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e62f2efc2a2049e4ab0b69b7b26fa069 : SUCCESS in 16m 39s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/1af561b30cfc46f892a6606079e894e5 : SUCCESS in 14m 05s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/48236f18400b4e3fbb5135f979600c61 : SUCCESS in 13m 44s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/55d7a4ac10784cb3a8506cac03f7cc82 : SUCCESS in 19m 09s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/e8770da0590f4815889afdd0113955d0 : SUCCESS in 22m 50s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/93ea851897394acab28a09e2a365dd11 : SUCCESS in 18m 12s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/4d9400f06c69492283cf3ae34c2cb007 : SUCCESS in 58m 39s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/8fc43c0773874fdd8e211e2707992bb9 : SUCCESS in 58m 17s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/bedb09b8177748b2a9b7f461487510d3 : SUCCESS in 1h 02m 12s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6d7b12edce1a4827ab9a046fd0dbf5a7 : SUCCESS in 1h 06m 34s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/7cc0b1325a294759a002f9ec4ca75632 : SUCCESS in 1h 01m 28s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/1223cc5a93444ad9b921ee837baa485d : SUCCESS in 1h 41m 14s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/36045c2ca7d04d98b25ffff9ebe19bb2 : SUCCESS in 1h 49m 19s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/44cc0256b95544d99014ddd722c38b11 : SUCCESS in 16m 37s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/e8138ab4c6f54132add2b1f628e0d504 : SUCCESS in 22m 33s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1e112718d30b4807b814f92c8a41a9a9 : SUCCESS in 24m 09s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/49e3a004cf73496a8c2e1538cbdc8cd9 : SUCCESS in 24m 49s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/3cd7cabac27243478882c1de8f3688e8 : SUCCESS in 27m 43s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/75212af33aef4414996a4291c6b05483 : SUCCESS in 20m 48s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/5df9a7b882984304a7abbc2b80c9da41 : SUCCESS in 20m 20s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4f69f0e7a25e4e3a860d7afe226f119e : SUCCESS in 14m 06s","accounts_in_message":[],"_revision_number":9},{"id":"b9e0badffc8c6bd0edae5e1998c5893ceb264a4c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-22 09:43:44.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6374b667e4ba431f8edb7c291f2a209e\n\n- kayobe-overcloud-host-configure-rocky10-aarch64 https://zuul.opendev.org/t/openstack/build/f8526bc2820040d19e70ac1567369608 : SUCCESS in 39m 29s (non-voting)\n- kayobe-overcloud-rocky10-aarch64 https://zuul.opendev.org/t/openstack/build/2a049c5e734740cfb21c0d5932266edf : FAILURE in 1h 38m 42s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"7b129947f1d21594becc2487d5e88fd37475542a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-06-22 12:56:33.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":10},{"id":"a4fa1c99313e1839099a6868cd86865f157f5080","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-22 14:55:07.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ac7fa7789c174fedbe8946b61133f57f\n\n- kayobe-overcloud-host-configure-rocky10-aarch64 https://zuul.opendev.org/t/openstack/build/9b1f6298d52b4bcba978a2ac0fe826d2 : SUCCESS in 40m 04s (non-voting)\n- kayobe-overcloud-rocky10-aarch64 https://zuul.opendev.org/t/openstack/build/70616f1b2edc40b49f7d7d22359c92c8 : FAILURE in 1h 43m 25s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"e0040e647a8829da45cc26b66fa63f5a264ec81d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-22 14:59:29.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e17ba1367e39460aad22816934a26e63\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e91792c53a0742b8b1d2efc4059862ba : SUCCESS in 10m 18s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/0b1674c516e74e1eb113ba6245fc79a5 : SUCCESS in 8m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9b7c9ce51ebb4d9fb12578665475e519 : SUCCESS in 6m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a02effad817644c28d757501a41c2a6b : SUCCESS in 4m 36s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/abec2c50a54b401dba5c5e7008a3203d : SUCCESS in 5m 51s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/11ce9912c25c406c8c26c9c19a06e112 : SUCCESS in 6m 43s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8b23eabd7a924a2b9968d91180bac8be : SUCCESS in 3m 30s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ed2b9dca067742708a3d984f3bd8a532 : SUCCESS in 6m 54s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/94915d8d77544c33a6831d12487f1a7e : SUCCESS in 10m 32s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/176cded781f94311bca20a5cddab7d94 : SUCCESS in 15m 01s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/6a6909c0f8144dcca6ac597fca5a1915 : SUCCESS in 19m 35s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/aa593c368cc34157b6c5e1560b405d0a : SUCCESS in 19m 39s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ba040ef2496f47929facd83a05bb0262 : SUCCESS in 20m 09s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/6d10f197cc67478a9995636af62fe636 : SUCCESS in 13m 56s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/fcad960f7f634f2ab088da9c27460870 : SUCCESS in 22m 02s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b6a52a4d714c43b3a2d6eaca9c86ef11 : SUCCESS in 16m 17s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/d6dd92a61e774b1ba5659628030e0f1f : SUCCESS in 27m 26s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/34584451edb5416197a2b590eb5a866e : SUCCESS in 20m 56s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/50d32b42c1164ff6a3f0b9e8316357e1 : SUCCESS in 1h 05m 56s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/ffc28b7068ac401eb9f9c796ae94d846 : SUCCESS in 1h 09m 20s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/e16d4502dd004f6caafcecde372600f6 : SUCCESS in 1h 05m 27s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1a7e7b78a7e24ec6b8aa37f897907206 : SUCCESS in 1h 07m 15s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/c0f13e1e14394737a12c38933ae106a4 : SUCCESS in 1h 12m 30s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/486c7d01c7764a2fa965400d60c767b9 : SUCCESS in 1h 58m 15s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6a758c872e6847f9b3da4c3b92c1cf37 : SUCCESS in 1h 50m 03s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/578fe9019c1c41cfb1bc5a2672ca2828 : SUCCESS in 19m 38s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/b4a55353bcfa4cb485876f500236bdd0 : SUCCESS in 18m 58s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5d22e23fad1a474a86aee42c7aa11590 : SUCCESS in 22m 02s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/426345ee75d448fcb8631575be9802ae : SUCCESS in 20m 31s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/a5a64fdee12744e69865af494c21a5b0 : SUCCESS in 25m 53s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e6b622c40c154ca8a5239c234d887e41 : SUCCESS in 25m 46s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/6453b5b5988d4440b3037cd3448153d7 : SUCCESS in 24m 53s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f49004c955f748dcab564ca82bdde690 : SUCCESS in 24m 36s","accounts_in_message":[],"_revision_number":10},{"id":"43215f0749ff8e30bc8ccfd825d6494d2934dfb0","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-07-15 08:49:44.000000000","message":"Patch Set 11: Patch Set 10 was rebased\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":11},{"id":"61f2fa9491abb8a8658932098926c080fa3f213a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-15 09:19:29.000000000","message":"Patch Set 11: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/kayobe-config-dev/+/992285 is needed.","accounts_in_message":[],"_revision_number":11},{"id":"a893926ee0afd1d44f1b63bf986fdb2d33072a73","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-07-15 09:20:14.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"e6b79702e39a2e8313200bbbbedb1950980638c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-15 11:02:47.000000000","message":"Patch Set 11:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5a13657d0fad405aaa485cc006552d6a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3f6368da50f4436841810589bffd24a : SUCCESS in 10m 30s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/8b3f1d46d11d4d9b85dc7d7f8b84e2db : SUCCESS in 8m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3367b96a3d3c4329aabf8bf9e0312d13 : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d36a88b1e5a64c96b883bf5ae3729299 : SUCCESS in 5m 12s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/15858326d4074fb9b8250d726ef166c3 : SUCCESS in 3m 55s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/4a8989d13436401888f83b3f069b487f : SUCCESS in 7m 31s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/10ac5efc7de74b2f89278983bf155f63 : SUCCESS in 6m 37s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7139e06b14894bc69d1bd49d252f0b79 : SUCCESS in 7m 17s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/23ac28e26c6f4efea3b3240ba92fba2f : SUCCESS in 15m 55s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/00b9f00a27134966b048fc591be4822b : SUCCESS in 14m 11s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/f6e3a09ee93741508e730742919d1b6f : SUCCESS in 24m 06s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/53641bcc845f43c2b903036c59407813 : SUCCESS in 17m 23s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9b9ffc803c784f8db8a4b719022ce8c2 : SUCCESS in 19m 39s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/efc3dd693a5e4da48fbde5b008ab67ee : SUCCESS in 17m 22s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/994f96e6c1a34a3da57d765533a8bb1f : SUCCESS in 18m 11s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9759a59dee4b4df3acf50019fae9f01a : SUCCESS in 22m 39s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/e4743978a7e7483f9e75facde6ceb1cc : SUCCESS in 18m 20s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e4b87cd03192490388e2c7ffdcc0e1f1 : SUCCESS in 20m 15s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/aaaee0733eb04967892ed290753e294a : FAILURE in 47m 25s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/da07ea83d54d40f0b069d6fbd935fcfb : FAILURE in 48m 39s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/f2bc5222cdd14bda8dd7814238d7a5a8 : FAILURE in 55m 00s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/395aabacbd2342d9a923c931285d70f4 : FAILURE in 49m 56s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/8a15b9fa1f5b47459705b34c67924942 : FAILURE in 49m 48s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/297a41d72a5a48f08c641c4fa1df1b89 : FAILURE in 1h 39m 31s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ab45eb067e7b4f00b81b1ab0ba1296f2 : FAILURE in 1h 41m 12s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/61efd78672d44af2a600d01747d8a78c : SUCCESS in 16m 52s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/def84ea112114f029b99d5cf8c00d1ab : SUCCESS in 25m 33s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ba5b784b610144c0a0cba0ac25f70a19 : SUCCESS in 25m 45s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/4615781d11474ef4a16e074c33437fc9 : SUCCESS in 25m 48s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/a6fef3fe1da445dc985789e5ddf32f2e : SUCCESS in 22m 58s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/141dc227002b4c77b22202b4a7ca6178 : SUCCESS in 32m 57s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/144eb22b83eb4440b957087073fa34f1 : SUCCESS in 23m 09s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d18df6b8a75d415383c9ff79962c7518 : SUCCESS in 22m 52s","accounts_in_message":[],"_revision_number":11},{"id":"20b7e07c137c465439c5e82b5b9f418fe1eb3bd0","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-07-15 12:52:57.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"5b7ddb9aea62cc591aaf653cb0d4969b0578dc50","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-15 15:04:13.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/094aad0322c04d1ab7c7028ba89aaab3\n\n- kayobe-overcloud-host-configure-rocky10-aarch64 https://zuul.opendev.org/t/openstack/build/58f84752ed26478f868dd6c026a246dc : SUCCESS in 42m 56s (non-voting)\n- kayobe-overcloud-rocky10-aarch64 https://zuul.opendev.org/t/openstack/build/9890a16f8a1745ccab74a78bf8bceeca : TIMED_OUT in 2h 13m 23s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"44a5ca4c90ca8c1deff35fc580445e8e8bcf9f03","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-15 15:24:24.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/50395f7236ad48fe9da131feee56ec72\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa603f833eec41b899e5eae9751f99d6 : SUCCESS in 9m 48s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/97b28473ab3b4dd8b374d9a640a21eb3 : SUCCESS in 4m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6d01c4c74e4143368675ff10996acd7d : SUCCESS in 3m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/889e90140dfb45139cdf605aa1a165b2 : SUCCESS in 4m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5df8ae64510147ea9684c1655e562b71 : SUCCESS in 4m 44s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/48f89fc6308e4aa6be66bed8ae59147d : SUCCESS in 6m 23s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1b3ccb44cfa249b8a7d07102529cd510 : SUCCESS in 3m 26s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7a16a4429aeb49d78e9501093f5f5209 : SUCCESS in 3m 27s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b10e72a1aa0240c39f72e389feba40b3 : SUCCESS in 16m 03s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/8359b87dd16e4dbd97aa5a6db1fcf224 : SUCCESS in 7m 01s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/8bb3e96ace214c94b717720976ff329f : SUCCESS in 13m 52s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/908ef238657c4d9bad3f2910c5422dfc : SUCCESS in 11m 57s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/07fd65235480400e86fe6d108ab1578f : SUCCESS in 21m 13s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/647abe34433b42e1a819fe11b8c56f56 : SUCCESS in 19m 51s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/32efb0b60a684a9fbcb60ab5e5be57b1 : SUCCESS in 16m 58s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bf9dc1cbadd64596a4b550ec0b296ab0 : SUCCESS in 23m 05s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/6c59e6263187483dba9d8baa62cd93d9 : SUCCESS in 20m 16s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e073fa273a2645b1a93dd31aacb5d57b : SUCCESS in 23m 41s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/bef935b394034a01ad5c7ed87c050f3a : SUCCESS in 1h 24m 57s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/f62c9931f85f4b88a0a6e69b04753840 : SUCCESS in 1h 01m 25s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/d82ee71399a04b34bd38b2dfe9485f34 : SUCCESS in 1h 07m 34s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b76d569675e749aeab89cb1849b2c015 : SUCCESS in 1h 11m 12s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/7e56a27ce8d840349cdd4173f2da0931 : SUCCESS in 1h 16m 33s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/f23286f0c9d84f74a6a2643f8bde2b01 : SUCCESS in 2h 28m 54s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bb22b885ffd1481ea43082537f8703b6 : SUCCESS in 1h 58m 53s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/c75b730b901041e980b4f1588c2bb200 : SUCCESS in 30m 08s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/733e453c52014cad9cbac4b4a2641815 : SUCCESS in 17m 24s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3f7816c495f94b9da5abc1c3fa8af8d1 : SUCCESS in 28m 54s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/3eb34816309f46b5874263b0e00c76af : SUCCESS in 26m 38s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/026c77cc7ddb4af7b8694b6f0e8d1a83 : SUCCESS in 35m 58s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/83911dc457314e98b58bb74f9e9118af : SUCCESS in 36m 54s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/a76e4be128bb419dac397b01cfa24039 : SUCCESS in 16m 09s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c0260b8787db4d579a2490f2f612046f : SUCCESS in 27m 47s","accounts_in_message":[],"_revision_number":11},{"id":"71b56251d4369ef6a0859d4a04214e3085cab843","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-07-20 07:35:34.000000000","message":"Patch Set 12: Patch Set 11 was rebased\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":12},{"id":"a1552bd77b14869cc83b4429eeb3066c40e6307e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-20 07:38:43.000000000","message":"Patch Set 12: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/kayobe-config-dev/+/992285 is needed.","accounts_in_message":[],"_revision_number":12},{"id":"e52ea2557b9f839a6222c775cfb9ea6589d504a1","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-07-20 07:39:24.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"9658ecb9e0aff0dfd9561e6a635586d3b7e92706","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-20 09:34:18.000000000","message":"Patch Set 12:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d14e070d46a5446b870590383d1ecbe7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d267335029c4e17acfa721a223991f7 : SUCCESS in 9m 37s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/a76939aae39e4fc698f0c735d2d0fb77 : SUCCESS in 7m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ff73d0ce352c4a1483610a99097c3ee7 : SUCCESS in 4m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/19917cd66d9a4f608b2f36fb1d469fa7 : SUCCESS in 4m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0ee078704231463ba084889d085e32f1 : SUCCESS in 4m 22s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/282ddac287b64ac190637f7068819ad3 : SUCCESS in 7m 09s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0d37a4cdae324bf9869c1a7cc296d3a2 : SUCCESS in 6m 06s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/deb8b00cec9244d3b65c1e3625df998d : SUCCESS in 5m 03s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f3e5e52534de4d31af0acbaa7dc0c0d8 : SUCCESS in 11m 34s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f52a7800b1044f26ad41b685942bf12f : SUCCESS in 14m 15s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/c73ea046212d45a8956b4b88d9c2fd34 : RETRY_LIMIT in 3m 00s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/2aca3abda202479189446ae1ce37f847 : RETRY_LIMIT in 6m 33s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e0961389d5b5468b8605677d48371a07 : SUCCESS in 16m 16s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/f8927aacb8544f87948e537481df42dc : SUCCESS in 15m 43s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/92f3fe70ac6248fa8f637e2e1139b317 : RETRY_LIMIT in 2m 36s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/77f01133c45a4b0d9ba4792da07fd5bd : SUCCESS in 22m 26s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/40a8d72dae714969b7d28da00eb0ea4f : RETRY_LIMIT in 4m 54s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6a06c06532b64fa3b03a787f52c0ef75 : SUCCESS in 19m 50s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/923ddb27638f42ccb70d81680b9fb6a0 : RETRY_LIMIT in 5m 35s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/9b5ae991c8794c4eb89f7220998836e4 : RETRY_LIMIT in 4m 23s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/d82ed9762c65449e98c9a2a1a7de34a9 : RETRY_LIMIT in 4m 26s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d556f637ecee4be0924004e4c52ec145 : SUCCESS in 58m 20s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/9bb5cce1abd54aedb9ec862d1dfaf62f : SUCCESS in 1h 05m 01s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/b698fe63935745bdb80e45d86586499b : FAILURE in 4m 50s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dfd7283cd0bd441b8569f089a06d6a99 : SUCCESS in 1h 52m 30s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/af38e58dc6cd48a6bf486aeb09dd4a71 : SUCCESS in 18m 58s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/163615446c334b208ae2322d8a38f2f5 : SUCCESS in 23m 07s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f31c644f7790458281dda6d6f915fd91 : SUCCESS in 28m 44s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/7cddf2ab5bbe412c93cf3b71e517f983 : SUCCESS in 16m 38s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/a11115a838eb447996c7662ebe4ee1c9 : SUCCESS in 35m 36s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8b69e3b94ce4479bbef8050770f143ab : SUCCESS in 31m 54s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/bf218dff95bc4f2a95b76a19abf0fbcc : RETRY_LIMIT in 2m 16s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4877bd9ce9ac4afe893f3a61f9a93698 : SUCCESS in 14m 10s","accounts_in_message":[],"_revision_number":12},{"id":"2e1ab11abbb2ed90f8277d16208dc1d6058d8a1e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-20 15:35:10.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bb989af57710497298397a6c2447b2fb\n\n- kayobe-overcloud-host-configure-rocky10-aarch64 https://zuul.opendev.org/t/openstack/build/b60aa55edbc342bab06c1621e344f905 : RETRY_LIMIT in 11m 12s (non-voting)\n- kayobe-overcloud-rocky10-aarch64 https://zuul.opendev.org/t/openstack/build/1c0bf975f424464a85f8e2a4ed50ad78 : RETRY_LIMIT in 9m 34s (non-voting)","accounts_in_message":[],"_revision_number":12}],"current_revision_number":12,"current_revision":"fa6c1e27891cb389ae30e3ab6d60dda3cd01b38a","revisions":{"0f6d02125f10f7705c4a7ba2798e2220ba596bdb":{"kind":"REWORK","_number":1,"created":"2026-06-08 15:26:44.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/88/992288/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/88/992288/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/88/992288/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/88/992288/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/88/992288/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/88/992288/1"}}},"commit":{"parents":[{"commit":"95e47148f1a2aa14429fe72c3879d7e6157789e9","subject":"Merge \"CI: Stop using image from Docker Hub\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/95e47148f1a2aa14429fe72c3879d7e6157789e9"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2026-06-08 15:26:30.000000000","tz":60},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2026-06-08 15:26:30.000000000","tz":60},"subject":"CI: add Rocky 10 aarch64 jobs","message":"CI: add Rocky 10 aarch64 jobs\n\nTest host configure and overcloud deployment on Rocky 10 aarch64.\n\nDepends-On: https://review.opendev.org/c/openstack/kayobe-config-dev/+/992285\n\nChange-Id: I12a9c874283856181e397378a2336efdf68a28e7\nSigned-off-by: Bartosz Bezak \u003cbartosz@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0f6d02125f10f7705c4a7ba2798e2220ba596bdb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0f6d02125f10f7705c4a7ba2798e2220ba596bdb"}]},"branch":"refs/heads/master"},"ea4d05e27d75204a0c2ec5df26c4bd32444493b4":{"kind":"REWORK","_number":2,"created":"2026-06-08 15:29:41.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/88/992288/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/88/992288/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/88/992288/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/88/992288/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/88/992288/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/88/992288/2"}}},"commit":{"parents":[{"commit":"95e47148f1a2aa14429fe72c3879d7e6157789e9","subject":"Merge \"CI: Stop using image from Docker Hub\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/95e47148f1a2aa14429fe72c3879d7e6157789e9"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2026-06-08 15:26:30.000000000","tz":60},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2026-06-08 15:29:37.000000000","tz":60},"subject":"CI: add Rocky 10 aarch64 jobs","message":"CI: add Rocky 10 aarch64 jobs\n\nTest host configure and overcloud deployment on Rocky 10 aarch64.\n\nDepends-On: https://review.opendev.org/c/openstack/kayobe-config-dev/+/992285\n\nChange-Id: I12a9c874283856181e397378a2336efdf68a28e7\nSigned-off-by: Bartosz Bezak \u003cbartosz@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ea4d05e27d75204a0c2ec5df26c4bd32444493b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ea4d05e27d75204a0c2ec5df26c4bd32444493b4"}]},"branch":"refs/heads/master"},"410d0715187e66e8e5dd7795cea1837b4bd084b0":{"kind":"REWORK","_number":3,"created":"2026-06-09 10:22:45.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/88/992288/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/88/992288/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/88/992288/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/88/992288/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/88/992288/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/88/992288/3"}}},"commit":{"parents":[{"commit":"95e47148f1a2aa14429fe72c3879d7e6157789e9","subject":"Merge \"CI: Stop using image from Docker Hub\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/95e47148f1a2aa14429fe72c3879d7e6157789e9"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2026-06-08 15:26:30.000000000","tz":60},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2026-06-09 10:22:39.000000000","tz":60},"subject":"CI: add Rocky 10 aarch64 jobs","message":"CI: add Rocky 10 aarch64 jobs\n\nTest host configure and overcloud deployment on Rocky 10 aarch64.\n\nDepends-On: https://review.opendev.org/c/openstack/kayobe-config-dev/+/992285\n\nChange-Id: I12a9c874283856181e397378a2336efdf68a28e7\nSigned-off-by: Bartosz Bezak \u003cbartosz@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/410d0715187e66e8e5dd7795cea1837b4bd084b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/410d0715187e66e8e5dd7795cea1837b4bd084b0"}]},"branch":"refs/heads/master"},"538889698c984c889574301a71d3777a8468190d":{"kind":"REWORK","_number":4,"created":"2026-06-10 09:07:50.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/88/992288/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/88/992288/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/88/992288/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/88/992288/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/88/992288/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/88/992288/4"}}},"commit":{"parents":[{"commit":"95e47148f1a2aa14429fe72c3879d7e6157789e9","subject":"Merge \"CI: Stop using image from Docker Hub\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/95e47148f1a2aa14429fe72c3879d7e6157789e9"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2026-06-08 15:26:30.000000000","tz":60},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2026-06-10 09:07:43.000000000","tz":60},"subject":"CI: add Rocky 10 aarch64 jobs","message":"CI: add Rocky 10 aarch64 jobs\n\nTest host configure and overcloud deployment on Rocky 10 aarch64.\n\nDepends-On: https://review.opendev.org/c/openstack/kayobe-config-dev/+/992285\n\nChange-Id: I12a9c874283856181e397378a2336efdf68a28e7\nSigned-off-by: Bartosz Bezak \u003cbartosz@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/538889698c984c889574301a71d3777a8468190d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/538889698c984c889574301a71d3777a8468190d"}]},"branch":"refs/heads/master"},"6f3e5c9eb724da104cc6db832117df1796581017":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2026-06-13 10:20:18.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/88/992288/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/88/992288/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/88/992288/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/88/992288/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/88/992288/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/88/992288/5"}}},"commit":{"parents":[{"commit":"63b35e747bc0579b095db4032df489bd617dbfc7","subject":"Merge \"Linking AGENTS.md to CLAUDE.md\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/63b35e747bc0579b095db4032df489bd617dbfc7"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2026-06-08 15:26:30.000000000","tz":60},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2026-06-13 10:20:18.000000000","tz":0},"subject":"CI: add Rocky 10 aarch64 jobs","message":"CI: add Rocky 10 aarch64 jobs\n\nTest host configure and overcloud deployment on Rocky 10 aarch64.\n\nDepends-On: https://review.opendev.org/c/openstack/kayobe-config-dev/+/992285\n\nChange-Id: I12a9c874283856181e397378a2336efdf68a28e7\nSigned-off-by: Bartosz Bezak \u003cbartosz@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6f3e5c9eb724da104cc6db832117df1796581017"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6f3e5c9eb724da104cc6db832117df1796581017"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"95e47148f1a2aa14429fe72c3879d7e6157789e9","ours":"538889698c984c889574301a71d3777a8468190d","theirs":"63b35e747bc0579b095db4032df489bd617dbfc7","merge_strategy":"recursive","contains_conflicts":false}},"3e4c9d17a8c7295bdbc8821a6254301ec47ec11c":{"kind":"REWORK","_number":6,"created":"2026-06-18 08:04:11.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/88/992288/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/88/992288/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/88/992288/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/88/992288/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/88/992288/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/88/992288/6"}}},"commit":{"parents":[{"commit":"95e47148f1a2aa14429fe72c3879d7e6157789e9","subject":"Merge \"CI: Stop using image from Docker Hub\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/95e47148f1a2aa14429fe72c3879d7e6157789e9"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2026-06-08 15:26:30.000000000","tz":60},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2026-06-18 08:04:05.000000000","tz":120},"subject":"CI: add Rocky 10 aarch64 jobs","message":"CI: add Rocky 10 aarch64 jobs\n\nTest host configure and overcloud deployment on Rocky 10 aarch64.\n\nDepends-On: https://review.opendev.org/c/openstack/kayobe-config-dev/+/992285\n\nChange-Id: I12a9c874283856181e397378a2336efdf68a28e7\nSigned-off-by: Bartosz Bezak \u003cbartosz@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3e4c9d17a8c7295bdbc8821a6254301ec47ec11c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3e4c9d17a8c7295bdbc8821a6254301ec47ec11c"}]},"branch":"refs/heads/master"},"2690144c07c039173284667a008e20b9e005fbff":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2026-06-18 08:04:23.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/88/992288/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/88/992288/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/88/992288/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/88/992288/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/88/992288/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/88/992288/7"}}},"commit":{"parents":[{"commit":"1f2df4b92302c485593abd3096de77131046224d","subject":"Merge \"Remove external_net_name variable\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1f2df4b92302c485593abd3096de77131046224d"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2026-06-08 15:26:30.000000000","tz":60},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2026-06-18 08:04:23.000000000","tz":0},"subject":"CI: add Rocky 10 aarch64 jobs","message":"CI: add Rocky 10 aarch64 jobs\n\nTest host configure and overcloud deployment on Rocky 10 aarch64.\n\nDepends-On: https://review.opendev.org/c/openstack/kayobe-config-dev/+/992285\n\nChange-Id: I12a9c874283856181e397378a2336efdf68a28e7\nSigned-off-by: Bartosz Bezak \u003cbartosz@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2690144c07c039173284667a008e20b9e005fbff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2690144c07c039173284667a008e20b9e005fbff"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"95e47148f1a2aa14429fe72c3879d7e6157789e9","ours":"3e4c9d17a8c7295bdbc8821a6254301ec47ec11c","theirs":"1f2df4b92302c485593abd3096de77131046224d","merge_strategy":"recursive","contains_conflicts":false}},"7da04e6acdf797dc2690f22d54653fd1df01dab4":{"kind":"REWORK","_number":8,"created":"2026-06-19 08:38:50.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/88/992288/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/88/992288/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/88/992288/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/88/992288/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/88/992288/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/88/992288/8"}}},"commit":{"parents":[{"commit":"1f2df4b92302c485593abd3096de77131046224d","subject":"Merge \"Remove external_net_name variable\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1f2df4b92302c485593abd3096de77131046224d"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2026-06-08 15:26:30.000000000","tz":60},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2026-06-19 08:38:43.000000000","tz":120},"subject":"CI: add Rocky 10 aarch64 jobs","message":"CI: add Rocky 10 aarch64 jobs\n\nTest host configure and overcloud deployment on Rocky 10 aarch64.\n\nDepends-On: https://review.opendev.org/c/openstack/kayobe-config-dev/+/992285\n\nChange-Id: I12a9c874283856181e397378a2336efdf68a28e7\nSigned-off-by: Bartosz Bezak \u003cbartosz@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7da04e6acdf797dc2690f22d54653fd1df01dab4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7da04e6acdf797dc2690f22d54653fd1df01dab4"}]},"branch":"refs/heads/master"},"8e0f2f39de91e92bcdcc6ea588ddf12ab6939667":{"kind":"REWORK","_number":9,"created":"2026-06-22 07:40:22.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/88/992288/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/88/992288/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/88/992288/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/88/992288/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/88/992288/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/88/992288/9"}}},"commit":{"parents":[{"commit":"1f2df4b92302c485593abd3096de77131046224d","subject":"Merge \"Remove external_net_name variable\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1f2df4b92302c485593abd3096de77131046224d"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2026-06-08 15:26:30.000000000","tz":60},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2026-06-22 07:40:15.000000000","tz":120},"subject":"CI: add Rocky 10 aarch64 jobs","message":"CI: add Rocky 10 aarch64 jobs\n\nTest host configure and overcloud deployment on Rocky 10 aarch64.\n\nDepends-On: https://review.opendev.org/c/openstack/kayobe-config-dev/+/992285\n\nChange-Id: I12a9c874283856181e397378a2336efdf68a28e7\nSigned-off-by: Bartosz Bezak \u003cbartosz@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8e0f2f39de91e92bcdcc6ea588ddf12ab6939667"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8e0f2f39de91e92bcdcc6ea588ddf12ab6939667"}]},"branch":"refs/heads/master"},"621906579bdc480d1b171031bea64507a009142d":{"kind":"REWORK","_number":10,"created":"2026-06-22 12:56:33.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/88/992288/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/88/992288/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/88/992288/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/88/992288/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/88/992288/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/88/992288/10"}}},"commit":{"parents":[{"commit":"1f2df4b92302c485593abd3096de77131046224d","subject":"Merge \"Remove external_net_name variable\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1f2df4b92302c485593abd3096de77131046224d"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2026-06-08 15:26:30.000000000","tz":60},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2026-06-22 12:56:29.000000000","tz":120},"subject":"CI: add Rocky 10 aarch64 jobs","message":"CI: add Rocky 10 aarch64 jobs\n\nTest host configure and overcloud deployment on Rocky 10 aarch64.\n\nDepends-On: https://review.opendev.org/c/openstack/kayobe-config-dev/+/992285\n\nChange-Id: I12a9c874283856181e397378a2336efdf68a28e7\nSigned-off-by: Bartosz Bezak \u003cbartosz@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/621906579bdc480d1b171031bea64507a009142d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/621906579bdc480d1b171031bea64507a009142d"}]},"branch":"refs/heads/master"},"45b1d74337f19344ee1b2161c790dbe841891a12":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2026-07-15 08:49:44.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/88/992288/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/88/992288/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/88/992288/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/88/992288/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/88/992288/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/88/992288/11"}}},"commit":{"parents":[{"commit":"e7dc5a78e5f00420fcad9e99a47a3f67b4669ebd","subject":"Update architecture page to reflect reality","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e7dc5a78e5f00420fcad9e99a47a3f67b4669ebd"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2026-06-08 15:26:30.000000000","tz":60},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2026-07-15 08:49:44.000000000","tz":0},"subject":"CI: add Rocky 10 aarch64 jobs","message":"CI: add Rocky 10 aarch64 jobs\n\nTest host configure and overcloud deployment on Rocky 10 aarch64.\n\nDepends-On: https://review.opendev.org/c/openstack/kayobe-config-dev/+/992285\n\nChange-Id: I12a9c874283856181e397378a2336efdf68a28e7\nSigned-off-by: Bartosz Bezak \u003cbartosz@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/45b1d74337f19344ee1b2161c790dbe841891a12"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/45b1d74337f19344ee1b2161c790dbe841891a12"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"1f2df4b92302c485593abd3096de77131046224d","ours":"621906579bdc480d1b171031bea64507a009142d","theirs":"e7dc5a78e5f00420fcad9e99a47a3f67b4669ebd","merge_strategy":"recursive","contains_conflicts":false}},"fa6c1e27891cb389ae30e3ab6d60dda3cd01b38a":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2026-07-20 07:35:34.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/88/992288/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/88/992288/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/88/992288/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/88/992288/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/88/992288/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/88/992288/12"}}},"commit":{"parents":[{"commit":"9a6e4333b2aa2fbe8dc80a93cd5aa3e2f1872970","subject":"Merge \"Adds seed IP to default no-proxy list\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9a6e4333b2aa2fbe8dc80a93cd5aa3e2f1872970"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2026-06-08 15:26:30.000000000","tz":60},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2026-07-20 07:35:34.000000000","tz":0},"subject":"CI: add Rocky 10 aarch64 jobs","message":"CI: add Rocky 10 aarch64 jobs\n\nTest host configure and overcloud deployment on Rocky 10 aarch64.\n\nDepends-On: https://review.opendev.org/c/openstack/kayobe-config-dev/+/992285\n\nChange-Id: I12a9c874283856181e397378a2336efdf68a28e7\nSigned-off-by: Bartosz Bezak \u003cbartosz@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fa6c1e27891cb389ae30e3ab6d60dda3cd01b38a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fa6c1e27891cb389ae30e3ab6d60dda3cd01b38a"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"e7dc5a78e5f00420fcad9e99a47a3f67b4669ebd","ours":"45b1d74337f19344ee1b2161c790dbe841891a12","theirs":"9a6e4333b2aa2fbe8dc80a93cd5aa3e2f1872970","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
