)]}'
{"id":"openstack%2Fkayobe~993103","triplet_id":"openstack%2Fkayobe~master~Ied80937b4f25ef36783807b064cd72c2182153de","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2026-06-22 12:31:01.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ied80937b4f25ef36783807b064cd72c2182153de","subject":"WIP:Try older constraints with tenks","status":"ABANDONED","created":"2026-06-12 12:16:12.000000000","updated":"2026-06-22 12:31:01.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a7067ac90e74e2e3947311aed6e4fa7abde82250","_number":993103,"virtual_id_number":993103,"owner":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-06-12 13:13:01.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-12 13:13:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2481a8d28b8b933db480fb6ca3aa4afd61fa3d40","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-06-12 12:16:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"969513d108351ac488ac1d38c39606930b4b6d6f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-12 13:13:01.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/eb8c2cea28204d7aa9dd866527cbd856\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0060f059c4e74d749fc189ba8c549209 : SUCCESS in 5m 32s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/678426b120694107ac67fa9b54f6786f : SUCCESS in 8m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3fb96f6d8cb04b56aa446b2265c1937b : SUCCESS in 4m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e2dd048e6830455da782242caebfaa02 : SUCCESS in 3m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2a3f47a6858c4f848577b41cd6222ab2 : SUCCESS in 3m 18s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/009c0d81e3c54fe59d9dd2ae9f503256 : SUCCESS in 6m 31s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2ecd09e4f5194e8fae2e740a06f1874f : SUCCESS in 7m 25s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d2877e2e86ae4a62937ac9d05e3c458f : SUCCESS in 6m 20s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0be30c98c4fa4d0a8099623de6e54658 : SUCCESS in 15m 36s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/bcbdae6962564e9da0957716c695bceb : SUCCESS in 14m 52s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/e9cd4d54ae8642ecbef53dcae0c002d7 : SUCCESS in 19m 45s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/6bf454ee55ed42329fac065bfedf0f17 : SUCCESS in 17m 54s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/557c774a436745a6840aa6678634fa75 : SUCCESS in 16m 05s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/ce071f0f5feb43a98f6b2bf02fef92d9 : SUCCESS in 14m 14s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/803aa68074584fb5a07724cda95c44e8 : SUCCESS in 29m 30s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c2fcc82f67b642db8da17d168efdc0cb : SUCCESS in 21m 13s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/d0c95f13debe401d9cabee45b9cc94b7 : FAILURE in 20m 02s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c7110ea6422549f38296b8cfb69ebcf4 : SUCCESS in 15m 08s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/483e47abf95b4797988dc3afd06e0642 : FAILURE in 47m 23s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/7d8d3b05bde34aa08e656a2cdee61123 : FAILURE in 45m 57s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/837ed49c71cf448a8e561a8532200b1a : FAILURE in 53m 03s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fc89f3dff49d43119cb30689b153dbc0 : FAILURE in 42m 04s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/c149d4a9096047febc56ea3abe804a8e : FAILURE in 46m 00s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/08a41271082c4054a02d4ab4ab885da5 : FAILURE in 41m 07s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/c3d2744467b74317bee8f7149e53962c : SUCCESS in 27m 38s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/46835fcd58854205b537c05d65197eff : SUCCESS in 18m 26s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/490897819bdd4c9a8c5b3327791c67d0 : SUCCESS in 31m 16s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/d468a9d4d2f74e8692cc8c75d6a80697 : SUCCESS in 24m 03s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/57e73ee2d55b44498a018d5a451fb74c : SUCCESS in 35m 25s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d7df13959cf2466289e1720967b66678 : SUCCESS in 36m 51s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/4c2f77fe36d2472b921e02fc10f8865c : SUCCESS in 19m 39s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f8ac96bbcf574b4dbca717d114e120b9 : SUCCESS in 19m 17s","accounts_in_message":[],"_revision_number":1},{"id":"a7067ac90e74e2e3947311aed6e4fa7abde82250","tag":"autogenerated:gerrit:abandon","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-06-22 12:31:01.000000000","message":"Abandoned\n\nFixed via https://review.opendev.org/c/openstack/tenks/+/993082 on tenks side","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b39cee3def81a7e71b80e15232286890cc73db35","revisions":{"b39cee3def81a7e71b80e15232286890cc73db35":{"kind":"REWORK","_number":1,"created":"2026-06-12 12:16:12.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/03/993103/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/03/993103/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/03/993103/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/03/993103/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/03/993103/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/03/993103/1"}}},"commit":{"parents":[{"commit":"69588969bd982a859961b0daade6b60c4e871081","subject":"[CI] Disable NTP checks in CI","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/69588969bd982a859961b0daade6b60c4e871081"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-06-12 10:52:30.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-06-12 12:16:04.000000000","tz":60},"subject":"WIP:Try older constraints with tenks","message":"WIP:Try older constraints with tenks\n\nThis is to workaround:\n\nansible_openstack.cloud.baremetal_node_payload.zip/ansible_collections/openstack/cloud/plugins/module_utils/openstack.py\\\",\nline 317, in openstack_cloud_from_module\\nAttributeError: module\n\u0027openstack\u0027 has no attribute \u0027version\u0027\\n\",\n\nChange-Id: Ied80937b4f25ef36783807b064cd72c2182153de\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b39cee3def81a7e71b80e15232286890cc73db35"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b39cee3def81a7e71b80e15232286890cc73db35"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
