)]}'
{"id":"openstack%2Fkayobe~993104","triplet_id":"openstack%2Fkayobe~master~Ia52fa057c4d938f0f3abeb1ae709d975a5b45b6a","project":"openstack/kayobe","branch":"master","topic":"tenks-collections","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2026-07-10 15:50:11.000000000","reason":"Change was marked work in progress"}},"hashtags":[],"change_id":"Ia52fa057c4d938f0f3abeb1ae709d975a5b45b6a","subject":"CI: Add openstack collection to tenks venv","status":"NEW","created":"2026-06-12 12:37:05.000000000","updated":"2026-07-10 17:41:13.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":5,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":true,"meta_rev_id":"6a46598ec098c03cb29c6d9a234ecca6cb32fb0b","_number":993104,"virtual_id_number":993104,"owner":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-07-10 17:41:13.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"reviewer_updates":[{"updated":"2026-06-12 12:53:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c6e9b0e11970fdb5dfae738cc2262ca0a1e28cd6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-06-12 12:37:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"09a89e44ebb34a37b65e8d74ad375a4c8464a0f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-12 12:53:29.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/be187e6aaa82472a96f39bc7a8e52b10\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe9f0092f24744b99a78372ced420d5d : SUCCESS in 5m 29s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/56d565ee190044fba0d818a544e47a17 : SUCCESS in 4m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/68c5b599e58d4afe8baa40e609b3f077 : SUCCESS in 4m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7681e15070204ff3b77fa327cab483d7 : SUCCESS in 3m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c52045a4c0ee43e78c9858d1cc71e254 : FAILURE in 2m 09s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b7bd5af3ab494308b0f55676b3097ee1 : SUCCESS in 4m 17s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f1ec8545b9af471ab724d4754b0ed98a : SUCCESS in 5m 04s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/580d8ec7d43b43aca89717da33b7c738 : SUCCESS in 6m 31s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/56f8ea527c994757b6f70e551df4301c : SUCCESS in 15m 02s","accounts_in_message":[],"_revision_number":1},{"id":"a6fb868678b005605a0b8a2d12bcac1f64657dea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-06-12 12:55:28.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"349d677e03bb20b84f39b13ea988f86ea6624cf4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-12 13:46:57.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e29b36fb30f64c7f84572dc43c24c09b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bdb801f4e1cd469eac4277f434c5d0b9 : SUCCESS in 10m 29s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/5927f784cbbe43209de74081501ef881 : SUCCESS in 10m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/60cdd55012104bbfab785cbc697f6c37 : SUCCESS in 4m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/16fba34a845a4c0894430a36c4ef73da : SUCCESS in 4m 05s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/686ad5e31c9f4c5fb643873687f109eb : SUCCESS in 5m 22s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/cf7a40cff77b47d7a0293b24439bbe71 : SUCCESS in 8m 34s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d76a854bbe774c9c8b3d902a357ab40a : SUCCESS in 6m 51s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/bf0590ec31d741a99e2179a1c4156ed1 : FAILURE in 50m 18s","accounts_in_message":[],"_revision_number":2},{"id":"df853defdd5570c8189f8b1cd25e43ec25c56d10","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-06-12 14:03:35.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"58039256048566db7de58b1578be142b021df96d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-12 14:50:03.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/737547a51b6045d1b46996ffa4e99746\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8fb6be9252d14d4fb72a4d3a418114b9 : SUCCESS in 6m 20s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/672beed80e7d4d119f7ba4754a533e82 : SUCCESS in 5m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e8c09ad58cb04db2b624ba4e577a2f02 : SUCCESS in 4m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/34754b3922204a9fa3f5da55cbaabc47 : SUCCESS in 3m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a04f27e4d1a2472bbe9acbfd7834de26 : SUCCESS in 4m 37s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e341e1d7adfd4e68ab7c304bdb79395c : SUCCESS in 4m 35s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/86b74adc22134f80964e860d3f3b3284 : SUCCESS in 3m 32s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/0bf1012784d4474ca34bff784e12e519 : FAILURE in 43m 17s","accounts_in_message":[],"_revision_number":3},{"id":"14ee6bec0040a1e5693425bed70cbb4ca959a549","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-06-22 12:44:31.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"507ebb1dbad2fe865e1e273b662062c023b8188c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-22 13:16:08.000000000","message":"Patch Set 4: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/tenks/+/993060 is needed.","accounts_in_message":[],"_revision_number":4},{"id":"f9515434c358b4ebda070b3e0a1c2ea535daae11","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-06-22 13:33:55.000000000","message":"Patch Set 5: Commit message was updated.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"c68275b55c4579efa74a45448cf70756572bbc1b","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-06-22 13:51:27.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"817acc806d408b944c44403b44e66bd41d549ae2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-22 14:16:23.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/05da1148ba11445484d644c34da81590\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1cc80cd012584d11a0005c170afa00c1 : SUCCESS in 9m 16s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/fc7fe72f0e1e4a97981333718693f827 : SUCCESS in 7m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/11e7e75d898e426183ecd06d895aca29 : SUCCESS in 5m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a8cfb167de5493591ac6fac9818d919 : SUCCESS in 3m 55s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/500c05a335f040579f27e0b3770ff48c : SUCCESS in 4m 16s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/1da4017ea9854a13a19a015fdfa3e6bc : SUCCESS in 5m 38s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5af843860e3e4ab4812a9ccd5e78f79d : SUCCESS in 5m 58s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d73652e4351b4e06ab72c35805b7a285 : SUCCESS in 3m 11s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c16d18d9ae284347ace0a1da3b391081 : SUCCESS in 15m 32s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1f8954c3bd6e43a8a0ed7b5f6d754472 : SUCCESS in 6m 50s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/7c6cf0d13d0046e4ac5eb379ecb292d3 : FAILURE in 40m 50s","accounts_in_message":[],"_revision_number":5},{"id":"51cb1597810056ba657eb71aa884ce684be6618d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-06-23 10:18:30.000000000","message":"Patch Set 6: Published edit on patch set 5.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":6},{"id":"45ba0e91b6de0436a007afbe1a715498c8f506ba","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-06-23 10:20:11.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"b8db8f521ac69abe8882a6bba3531ce5663c90ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-23 11:06:48.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8e2d6101bae7481ab7a38e87a04e83fc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0ed9d9b1d2949ae9ffd64383423228b : SUCCESS in 9m 02s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/854d0324ca8e4f1cbc50441887fe3da8 : SUCCESS in 6m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ae2af99749e3424fa58ffe91265bc8f6 : SUCCESS in 4m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/96db07b944944dcaa30469f661dde3e0 : SUCCESS in 2m 36s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d059017cf9ac42e38c7541fd11f45373 : SUCCESS in 4m 20s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f2e32a9b055449da929e0ddf25a96ae1 : SUCCESS in 6m 45s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fd567092905e479b8c0261dad745cea5 : SUCCESS in 6m 47s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c2a8cbf35a6b4ca9b347046917e152a3 : SUCCESS in 6m 24s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f4a181b3afd94a669aa74dcdefd17938 : SUCCESS in 11m 22s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/bafa0ab57e424fb895becb87f5580d87 : SUCCESS in 13m 07s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/224010c4a1ea4255942a45166643e64c : FAILURE in 41m 35s","accounts_in_message":[],"_revision_number":6},{"id":"f19829fd639a64bbf58d8a36559b44db63a90e23","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-06-23 15:42:02.000000000","message":"Patch Set 7: Published edit on patch set 6.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":7},{"id":"9d839e1bdd642c353d75eab3ebb30e14a9bbdab0","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-06-23 15:43:59.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"8de3c649b03b68468ef3da7ef19f9b556ebaaab2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-23 16:29:24.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d466c015e76e49a38a8ac0ac7ef22df2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78673a830ae04df58fb9595674a920e7 : SUCCESS in 8m 04s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/a3dd2740aa624c7aaeecee2eeec7dd5f : SUCCESS in 4m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/62669fa8379646a89f2c791d64502a90 : SUCCESS in 4m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/21c00291db7a401bad43adda8fbc9bd5 : SUCCESS in 3m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/00c5607767b64928bef9ed86cd40417f : SUCCESS in 2m 29s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/db38a1ff204a4a3e9eb6e4486dc78a1b : SUCCESS in 4m 45s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7be0de74ac444bc39ff922571dccad30 : SUCCESS in 6m 01s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7d4392dd47b143a5bd697da0b2f70c69 : SUCCESS in 4m 52s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a6ed1fd0f06348bfb303dd5643fc69f9 : SUCCESS in 7m 42s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/456eb90e4443429dacfe3506af59d33d : SUCCESS in 7m 05s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/7ee67c211cf54c0d881c372667442241 : FAILURE in 45m 09s","accounts_in_message":[],"_revision_number":7},{"id":"1d2492e437e5bb192e576ecc2f1d7e5b419608f4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-07-10 15:49:59.000000000","message":"Uploaded patch set 8: Commit message was updated.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":8},{"id":"83d73a6e3c494f1a7099629e3ab658e5c4b8b6c9","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-07-10 15:50:11.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":8},{"id":"289b2bc3c8489468883aa732c98400946dd01172","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-07-10 15:50:15.000000000","message":"Patch Set 8:\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"bed09dbd736079e05c98c8917a64a5dde5ec2742","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-10 16:36:32.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4afe21de31944ed6a92c3e608c406c51\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58096bc465434477a478135fbedd0249 : SUCCESS in 9m 03s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/2af9748f9e8443719408ad229b9344f1 : SUCCESS in 7m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/28d65696e91847be9a8704fbd5b49f3f : SUCCESS in 3m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f7172650efe4bbfb3263e7b1ab073db : SUCCESS in 3m 57s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d69ed464f90b4a55905756a579be716a : SUCCESS in 4m 13s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/53ea8b602e50435ba222701e1c7c4b6f : SUCCESS in 6m 53s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3e1a3634a64448c3bfc05e4b98ea4265 : SUCCESS in 6m 33s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/72aaa3d02823488f80eaa2ba20b7241e : SUCCESS in 3m 28s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/543778b1575a487ea870e1c2ac19e59d : SUCCESS in 6m 26s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3d32d55a3f4c4052aa509719717e329c : SUCCESS in 9m 00s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/b54a589dd1b246c380aa158978a23ef8 : FAILURE in 45m 06s","accounts_in_message":[],"_revision_number":8},{"id":"5204d824acf70843131922acd9c82c513e865675","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-07-10 16:51:50.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":9},{"id":"e48f5e768ed3347c59bb8b1ac521735c227490bb","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-07-10 16:52:07.000000000","message":"Topic set to tenks-collections","accounts_in_message":[],"_revision_number":9},{"id":"6a46598ec098c03cb29c6d9a234ecca6cb32fb0b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-10 17:41:13.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ec119ef37cae4d879647bdeee95ef9a7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a8319f7c5d240b2b432243e3d413e3c : SUCCESS in 4m 54s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/98b27488d9e9455cae5fb2f356127436 : SUCCESS in 7m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f7fee5356f844dc39c10b74c8ee63fab : SUCCESS in 3m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7972ff43c81d42fe866b3af9bf04e2e4 : SUCCESS in 2m 55s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d59243cfd741483dab71e953f8d6a58b : SUCCESS in 4m 00s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/42ed9280f025401d98d5515fc7b2d574 : SUCCESS in 4m 02s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f5ac1ac72f7e407d904dddae77c45bf4 : SUCCESS in 5m 32s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/27370bcafe214476b3ba5f919bc3961b : SUCCESS in 3m 19s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/19b1ba05c3a34692aedd08ae38a001d8 : SUCCESS in 13m 23s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/47168c8cb3e541e7b78d444dc7158df9 : SUCCESS in 12m 06s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/b4c439c9057744be8ec358177bdd8db6 : SUCCESS in 14m 46s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/7283ebeb713748edbcf3f9e2566c9e86 : SUCCESS in 14m 06s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4081c22e83514d2ea6fe25627a23034b : SUCCESS in 17m 19s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/121224267b744523a88fedb134f4d8b0 : SUCCESS in 14m 30s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/e2b4f75c9d5f4d7a99106a89863c8791 : SUCCESS in 14m 19s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b74f6d6aa0504aa6953dc7bdb0227675 : SUCCESS in 21m 05s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/40c2e694143245c4b4c20f57eac405ff : SUCCESS in 10m 56s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fbd9937f12414b62a7cc0f5992aa53ef : SUCCESS in 19m 26s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/c3d9fb4ae6734103bab098bf2aae7256 : FAILURE in 42m 13s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/c207fef16c7647139eb3b945a66c2615 : FAILURE in 41m 40s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/c12cebbb868f4926b0e8262047861084 : FAILURE in 48m 21s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/769dd5da0d2c45df874ee7dfac9ebcca : FAILURE in 41m 03s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/6d2cff927a5744a2a7c5d492bd759283 : FAILURE in 41m 46s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/c4081fbe377a43d2b49ec5758a904b0d : FAILURE in 42m 23s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/36178193ecdb4420bd6eb3cfab7936aa : FAILURE in 41m 51s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/78c0dcc05f5544eba844b7bc65e8f6f1 : SUCCESS in 14m 53s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/f35da47248f94e3190380b348e546087 : SUCCESS in 14m 23s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/739a50ad22aa4020b4d53256a014a9a3 : SUCCESS in 24m 31s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/12ae4509a80245dfab5ea02a117d4f60 : SUCCESS in 18m 55s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/0062ee103d324d6bb0156374ab8c9f53 : SUCCESS in 20m 48s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/68e80945cdcc48a0a82a5c8d525c5dbf : SUCCESS in 39m 30s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/692cf80a1ee34eee884cf051df128366 : SUCCESS in 14m 53s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c9fb3ede80d64b488ecd24a4b9b222d0 : SUCCESS in 21m 31s","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"c64ac06f36da640e99089ef3c7cf86ff5161fc49","revisions":{"4276325692f82952e69eb861dc65a18fc62e17d0":{"kind":"REWORK","_number":1,"created":"2026-06-12 12:37:05.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/04/993104/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/04/993104/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/993104/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/993104/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/993104/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/04/993104/1"}}},"commit":{"parents":[{"commit":"f95eb48fafea556b00bbeb11a47a1f903c838a72","subject":"Merge \"Drop kolla-tags and kolla-limit\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f95eb48fafea556b00bbeb11a47a1f903c838a72"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-06-12 12:30:11.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-06-12 12:36:59.000000000","tz":60},"subject":"WIP: Try install ansible collection openstack","message":"WIP: Try install ansible collection openstack\n\nChange-Id: Ia52fa057c4d938f0f3abeb1ae709d975a5b45b6a\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4276325692f82952e69eb861dc65a18fc62e17d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4276325692f82952e69eb861dc65a18fc62e17d0"}]},"branch":"refs/heads/master"},"09a4d04a558e03af75f426af265ec0acbf546713":{"kind":"REWORK","_number":2,"created":"2026-06-12 12:55:28.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/04/993104/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/04/993104/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/993104/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/993104/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/993104/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/04/993104/2"}}},"commit":{"parents":[{"commit":"f95eb48fafea556b00bbeb11a47a1f903c838a72","subject":"Merge \"Drop kolla-tags and kolla-limit\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f95eb48fafea556b00bbeb11a47a1f903c838a72"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-06-12 12:30:11.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-06-12 12:55:19.000000000","tz":60},"subject":"WIP: Try install ansible collection openstack","message":"WIP: Try install ansible collection openstack\n\nChange-Id: Ia52fa057c4d938f0f3abeb1ae709d975a5b45b6a\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/09a4d04a558e03af75f426af265ec0acbf546713"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/09a4d04a558e03af75f426af265ec0acbf546713"}]},"branch":"refs/heads/master"},"eaa22e42e7376aacecf8b48c6ee983f19efd11cc":{"kind":"REWORK","_number":3,"created":"2026-06-12 14:03:35.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/04/993104/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/04/993104/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/993104/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/993104/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/993104/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/04/993104/3"}}},"commit":{"parents":[{"commit":"f95eb48fafea556b00bbeb11a47a1f903c838a72","subject":"Merge \"Drop kolla-tags and kolla-limit\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f95eb48fafea556b00bbeb11a47a1f903c838a72"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-06-12 12:30:11.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-06-12 14:03:35.000000000","tz":0},"subject":"WIP: Try install ansible collection openstack","message":"WIP: Try install ansible collection openstack\n\nChange-Id: Ia52fa057c4d938f0f3abeb1ae709d975a5b45b6a\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/eaa22e42e7376aacecf8b48c6ee983f19efd11cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/eaa22e42e7376aacecf8b48c6ee983f19efd11cc"}]},"branch":"refs/heads/master"},"9fbfb8f824e21adc089dd6a0844d69968cd62c5c":{"kind":"REWORK","_number":4,"created":"2026-06-22 12:44:31.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/04/993104/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/04/993104/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/993104/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/993104/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/993104/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/04/993104/4"}}},"commit":{"parents":[{"commit":"1f2df4b92302c485593abd3096de77131046224d","subject":"Merge \"Remove external_net_name variable\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1f2df4b92302c485593abd3096de77131046224d"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-06-12 12:30:11.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-06-22 12:44:25.000000000","tz":60},"subject":"WIP: Try install ansible collection openstack","message":"WIP: Try install ansible collection openstack\n\nDepends-On: https://review.opendev.org/c/openstack/tenks/+/993060\nChange-Id: Ia52fa057c4d938f0f3abeb1ae709d975a5b45b6a\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9fbfb8f824e21adc089dd6a0844d69968cd62c5c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9fbfb8f824e21adc089dd6a0844d69968cd62c5c"}]},"branch":"refs/heads/master"},"3d82cb74176039fe3ff129c9e60ede457a35f72a":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2026-06-22 13:33:55.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/04/993104/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/04/993104/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/993104/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/993104/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/993104/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/04/993104/5"}}},"commit":{"parents":[{"commit":"1f2df4b92302c485593abd3096de77131046224d","subject":"Merge \"Remove external_net_name variable\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1f2df4b92302c485593abd3096de77131046224d"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-06-12 12:30:11.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-06-22 13:33:55.000000000","tz":0},"subject":"WIP: Try install ansible collection openstack","message":"WIP: Try install ansible collection openstack\n\nChange-Id: Ia52fa057c4d938f0f3abeb1ae709d975a5b45b6a\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3d82cb74176039fe3ff129c9e60ede457a35f72a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3d82cb74176039fe3ff129c9e60ede457a35f72a"}]},"branch":"refs/heads/master","description":"Edit commit message"},"90b89ad6ad9a830ae00447f252fa04c4ce05a95d":{"kind":"REWORK","_number":6,"created":"2026-06-23 10:18:30.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/04/993104/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/04/993104/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/993104/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/993104/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/993104/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/04/993104/6"}}},"commit":{"parents":[{"commit":"1f2df4b92302c485593abd3096de77131046224d","subject":"Merge \"Remove external_net_name variable\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1f2df4b92302c485593abd3096de77131046224d"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-06-12 12:30:11.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-06-23 10:18:29.000000000","tz":0},"subject":"WIP: Try install ansible collection openstack","message":"WIP: Try install ansible collection openstack\n\nChange-Id: Ia52fa057c4d938f0f3abeb1ae709d975a5b45b6a\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/90b89ad6ad9a830ae00447f252fa04c4ce05a95d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/90b89ad6ad9a830ae00447f252fa04c4ce05a95d"}]},"branch":"refs/heads/master"},"3dd58021652b58c6808dc12cb0395a1b88c22fd8":{"kind":"REWORK","_number":7,"created":"2026-06-23 15:42:02.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/04/993104/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/04/993104/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/993104/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/993104/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/993104/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/04/993104/7"}}},"commit":{"parents":[{"commit":"1f2df4b92302c485593abd3096de77131046224d","subject":"Merge \"Remove external_net_name variable\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1f2df4b92302c485593abd3096de77131046224d"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-06-12 12:30:11.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-06-23 15:42:02.000000000","tz":0},"subject":"WIP: Try install ansible collection openstack","message":"WIP: Try install ansible collection openstack\n\nChange-Id: Ia52fa057c4d938f0f3abeb1ae709d975a5b45b6a\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3dd58021652b58c6808dc12cb0395a1b88c22fd8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3dd58021652b58c6808dc12cb0395a1b88c22fd8"}]},"branch":"refs/heads/master"},"b518388e34da9d0584cec50c1cd6ddb3aec9cca1":{"kind":"NO_CODE_CHANGE","_number":8,"created":"2026-07-10 15:49:59.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/04/993104/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/04/993104/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/993104/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/993104/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/993104/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/04/993104/8"}}},"commit":{"parents":[{"commit":"1f2df4b92302c485593abd3096de77131046224d","subject":"Merge \"Remove external_net_name variable\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1f2df4b92302c485593abd3096de77131046224d"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-06-12 12:30:11.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-07-10 15:46:42.000000000","tz":0},"subject":"CI: Install ansible collection openstack in tenks venv","message":"CI: Install ansible collection openstack in tenks venv\n\nThis is in preparation for bumping the openstacksdk constraints in\ntenks[1] which is problematic with the version of the opentack\ncollection in community collection.\n\nChange-Id: Ia52fa057c4d938f0f3abeb1ae709d975a5b45b6a\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b518388e34da9d0584cec50c1cd6ddb3aec9cca1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b518388e34da9d0584cec50c1cd6ddb3aec9cca1"}]},"branch":"refs/heads/master"},"c64ac06f36da640e99089ef3c7cf86ff5161fc49":{"kind":"REWORK","_number":9,"created":"2026-07-10 16:51:50.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/04/993104/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/04/993104/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/993104/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/993104/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/993104/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/04/993104/9"}}},"commit":{"parents":[{"commit":"1f2df4b92302c485593abd3096de77131046224d","subject":"Merge \"Remove external_net_name variable\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1f2df4b92302c485593abd3096de77131046224d"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-06-12 12:30:11.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-07-10 16:51:35.000000000","tz":60},"subject":"CI: Add openstack collection to tenks venv","message":"CI: Add openstack collection to tenks venv\n\nThis prepares for bumping the upper constraints in tenks[1], which is\nproblematic with the version of the openstack collection in the Ansible\ncommunity collection. Installing the version from tenks requirements\nallows us to use a known working combination.\n\n[1] https://review.opendev.org/c/openstack/tenks/+/994295\n\nChange-Id: Ia52fa057c4d938f0f3abeb1ae709d975a5b45b6a\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c64ac06f36da640e99089ef3c7cf86ff5161fc49"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c64ac06f36da640e99089ef3c7cf86ff5161fc49"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
