)]}'
{"id":"openstack%2Fkayobe~993744","triplet_id":"openstack%2Fkayobe~stable%2F2026.1~I62c16b9d09607ab728068d64196b11ed83e03195","project":"openstack/kayobe","branch":"stable/2026.1","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2026-06-22 13:56:30.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I62c16b9d09607ab728068d64196b11ed83e03195","subject":"Fix paths to EPEL repositories","status":"MERGED","created":"2026-06-17 12:22:10.000000000","updated":"2026-06-22 13:58:32.000000000","submitted":"2026-06-22 13:56:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"993744","cherry_pick_of_change":992770,"cherry_pick_of_patch_set":4,"meta_rev_id":"1bd21bcff29c52173d79534ec9c301b89cba03ca","_number":993744,"virtual_id_number":993744,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-22 13:56:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2026-06-22 12:13:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":2,"date":"2026-06-17 12:54:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":1,"date":"2026-06-17 19:58:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2026-06-22 12:13:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-17 12:54:19.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2026-06-17 14:18:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-17 19:58:21.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2026-06-22 12:13:30.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"}],"messages":[{"id":"22f5d84bd86970c1dda39453340e81e739c25738","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-06-17 12:22:10.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"140e43d6daf9a3e8b0ecba43bf3f1d96cf10be41","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2026-06-17 12:54:19.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"57a7e1402355dad35128d748d2a3f8d3a6749883","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-06-17 13:22:33.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"152da29f3bcb0f327d6fa637879e4d67fbf7b884","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-17 14:18:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9dcaef98e1c54b548e65dbaa393969f1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d78baba5696544cfb7221cbdbd337c00 : SUCCESS in 10m 08s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/3057c4ecba874f658cac34902d076d3a : SUCCESS in 6m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5c5c4c74a0e84cd5b960e40cf1c53789 : SUCCESS in 4m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/110a4677bab74e02aebc9b225fa7dad3 : SUCCESS in 4m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/554af84c6496427c923e14823460c837 : SUCCESS in 5m 40s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/79b945dca7b44b429af726366df70552 : SUCCESS in 7m 18s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bebd87dde98d4a85ad4c4955dde1efa0 : SUCCESS in 6m 23s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ef3b7ce490af4d68bb7ae37cd14870a7 : SUCCESS in 6m 34s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/574510a04f414ef5a5f48b32e50d0a2e : SUCCESS in 12m 47s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/59704581d42d426c8445fe4d26e7a746 : SUCCESS in 10m 47s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/69954685a7f84c80b07592812e1ba103 : SUCCESS in 11m 07s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/b78a7be99eba4c048368f4ebd536e416 : SUCCESS in 19m 57s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/13d75a8f2c3f455590dce5a0f7424a56 : SUCCESS in 10m 03s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/6fe8f6eb7c6b494fbe3b13e167afb48e : SUCCESS in 14m 06s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/487ec85a6fb142e48ce957abe501a3ec : SUCCESS in 14m 15s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/006de5fdf043414dba2845e99b91ae67 : SUCCESS in 21m 43s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/d39da1568de046ba998574bf41b9433a : SUCCESS in 16m 03s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2375d894abcf48f491d541b62258085c : SUCCESS in 19m 43s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/43996239574e4009a9916fd6cff0788b : SUCCESS in 1h 10m 33s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/f63b0b83219b488da6e06ae80159fe18 : SUCCESS in 59m 53s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/02435b70c5614a3eadfeb381ab440d92 : SUCCESS in 1h 03m 24s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8311e4734a384b3a9bf28a4f04c5e110 : SUCCESS in 1h 25m 13s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/bf7df6c82cb34bb6a37c1447a2fc03a8 : SUCCESS in 1h 47m 56s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/54293552272a4941a813c15b43a18be3 : SUCCESS in 1h 15m 04s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/8890a6b19ec749989ed31f1b197fba73 : SUCCESS in 1h 29m 57s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/215a8c2017394b3995d409e50124d3a3 : SUCCESS in 1h 50m 00s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cb1e6b42ab5a4183992a39a011b04962 : SUCCESS in 1h 38m 02s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/7a3b9219057b4c8abd2e92fc5f3b0d20 : SUCCESS in 18m 40s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/e9571ce53cf14e44825699d9b6c02cca : SUCCESS in 28m 31s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e7a3c0565270425185932c6e567d358c : SUCCESS in 23m 48s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/767f648c372949a29b583847496bc7d0 : SUCCESS in 21m 34s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/fa16d4b53ede425e9339ce85cad1c340 : SUCCESS in 36m 00s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/91286184851443b2a18768cca63e5f37 : SUCCESS in 17m 14s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1ed53be47b424acfb905f233e5906512 : SUCCESS in 32m 53s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/16264e7c06c74b74b98ff250a60658b8 : SUCCESS in 30m 03s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/e46701ba6bc94d93856caae11dd19373 : SUCCESS in 25m 15s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5fc6a574f8ed4e48ae97f6903df27a50 : SUCCESS in 20m 05s","accounts_in_message":[],"_revision_number":1},{"id":"f4d75c5e2f57f4f85e35e1a3265d83bdfcbf5802","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-17 14:28:30.000000000","message":"Patch Set 1:\n\nBuild failed (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/19f059f46d814f969d51a81119582aa1\n\n- kayobe-ansible-control-host-configure-centos10s https://zuul.opendev.org/t/openstack/build/5035fb009f094fdc9d9a4da78b8d1914 : SUCCESS in 17m 24s (non-voting)\n- kayobe-infra-vm-centos10s https://zuul.opendev.org/t/openstack/build/98c21ae05d60455b91de5e5bf80e00f3 : SUCCESS in 15m 10s (non-voting)\n- kayobe-infra-vm-centos10s-cloud-image https://zuul.opendev.org/t/openstack/build/5a38a3160b2e40fcad610144e70c925e : SUCCESS in 18m 27s (non-voting)\n- kayobe-infra-vm-rocky10-cloud-image https://zuul.opendev.org/t/openstack/build/b2e86361a5534bc7a857b611ddd7bfd2 : SUCCESS in 30m 14s\n- kayobe-infra-vm-ubuntu-noble-cloud-image https://zuul.opendev.org/t/openstack/build/bf9975b5d10440228fd299a22385649a : SUCCESS in 27m 47s (non-voting)\n- kayobe-overcloud-centos10s https://zuul.opendev.org/t/openstack/build/92121eaabe04454495c487c0665f11fd : SUCCESS in 1h 02m 33s (non-voting)\n- kayobe-overcloud-host-configure-centos10s https://zuul.opendev.org/t/openstack/build/2683f5feab004394a372eccf23e19f2e : SUCCESS in 17m 42s (non-voting)\n- kayobe-overcloud-tls-centos10s https://zuul.opendev.org/t/openstack/build/52bf663a17034055b60f0ac6785518b7 : SUCCESS in 1h 04m 53s (non-voting)\n- kayobe-seed-centos10s https://zuul.opendev.org/t/openstack/build/f071395f216a45ab84538980e45cca47 : SUCCESS in 15m 42s (non-voting)\n- kayobe-seed-images-centos10s https://zuul.opendev.org/t/openstack/build/d187389c41f84334a4cc7cd2f2e16106 : FAILURE in 21m 23s (non-voting)\n- kayobe-seed-images-rocky10 https://zuul.opendev.org/t/openstack/build/7dd032079c1f4f44973db2d59ba2b812 : FAILURE in 17m 12s\n- kayobe-seed-images-rocky10-podman https://zuul.opendev.org/t/openstack/build/19bd679f68d24970973928136b50586e : FAILURE in 22m 10s\n- kayobe-seed-images-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7aafca30e4744d5ab44ddbdcce61a16c : FAILURE in 15m 10s\n- kayobe-seed-images-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/fa8dd5a6dfa74f7aac46b562b0a0a66a : FAILURE in 23m 39s\n- kayobe-seed-vm-centos10s https://zuul.opendev.org/t/openstack/build/7588a83da51d4f6abf262c8970816295 : SUCCESS in 21m 31s (non-voting)\n- kayobe-seed-vm-centos10s-cloud-image https://zuul.opendev.org/t/openstack/build/13201cf6e7744bb589767baafaff3b35 : SUCCESS in 32m 31s (non-voting)\n- kayobe-seed-vm-centos10s-q35 https://zuul.opendev.org/t/openstack/build/bc947657aa844dfba1eb101b53babbcf : SUCCESS in 21m 19s (non-voting)\n- kayobe-seed-vm-rocky10-cloud-image https://zuul.opendev.org/t/openstack/build/cf56f92af80e4c7abd859c3fee4475df : SUCCESS in 22m 25s\n- kayobe-seed-vm-rocky10-q35 https://zuul.opendev.org/t/openstack/build/b86c7a10c95845ee84a8016a911207e5 : SUCCESS in 17m 34s\n- kayobe-seed-vm-ubuntu-noble-cloud-image https://zuul.opendev.org/t/openstack/build/e75cc566673040d7bd9093039813d3db : SUCCESS in 27m 09s (non-voting)\n- kayobe-seed-vm-ubuntu-noble-q35 https://zuul.opendev.org/t/openstack/build/a045e107f62446c0b7c96cfdaf291940 : SUCCESS in 21m 11s","accounts_in_message":[],"_revision_number":1},{"id":"10711e9db727fe721a2770e0916d12f0a4636e5d","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-06-17 19:58:21.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"6bccd91334f867e9d6d7a8dfb003d91374877a3d","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-06-22 12:13:30.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1b404abd83fe17eb1ef9f8e78cdb1ab66614541b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-22 12:13:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"80aac05d181c7b42263b64865e1be5a860bcb3d0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-22 13:56:30.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/13d5c59c0c164cf1a8287df72abaac34\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77451e7d41ea4ab89ee1f006bec251e1 : SUCCESS in 4m 52s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/045ca552df9846b597834e6778ff63a8 : SUCCESS in 8m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2fe522c2c60d4e0490f828606d58f626 : SUCCESS in 5m 05s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/79ce13ae6ed242878322dd1e8c1c18f8 : SUCCESS in 4m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/29c9c5eccf0a4d75b901191ca36d78ad : SUCCESS in 3m 08s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/bfd752e8417d40eb94bb08444bcaadc2 : SUCCESS in 5m 21s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/984275313e17486ea7da3d89b7a35df9 : SUCCESS in 12m 21s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d851f81a057c4dde9f83fc690951dbe1 : SUCCESS in 13m 54s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/9f054d561bf442c88ff26b2d9a0c11fe : SUCCESS in 13m 24s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/53e79b5fea28496d9100a0173a372266 : SUCCESS in 19m 47s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/e0bd9bd5c86345d797b59615fb2935f0 : SUCCESS in 19m 36s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/716d48432e0243fd874a075e94ed70c5 : SUCCESS in 19m 31s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/6b06eef4e28044e7915cca314602344b : SUCCESS in 24m 16s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c040056dee2b44b4bb4d8c41ff0e42ac : SUCCESS in 18m 42s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/9c145cc980e24d8d84e1af37a94391c8 : SUCCESS in 1h 02m 06s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/0c8e122b60ba4583819d74d46239459c : SUCCESS in 1h 06m 51s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/68aac610b8fd4e49bde7ad03212b85df : SUCCESS in 1h 03m 32s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dc18ceba5d0c4c849ff0d68795caff0a : SUCCESS in 1h 01m 20s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/86a6f820a26c4e6783ac5f86a5b68bff : SUCCESS in 1h 05m 55s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/5a53f5238805442e849e658663d8669e : SUCCESS in 1h 36m 04s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/4454665dd4994561aa25b9d288b533ab : SUCCESS in 1h 34m 27s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/32a8f456ac2a4184b9098883883e34be : SUCCESS in 1h 35m 40s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e55c6f5c68374d288034062be81bdbc3 : SUCCESS in 1h 41m 45s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/d8af28c5d8414adab18ac1ba9f92aa4f : SUCCESS in 21m 13s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/51ecdc05f93a48f4b1a824c5e70dfc46 : SUCCESS in 16m 23s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d05787b492f040a6a37841472775193e : SUCCESS in 22m 37s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/f20dcafac78a4ac6af1f5bcd37746b41 : SUCCESS in 20m 45s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/b8c5b62c3ba04f8db9915174cfcabaf3 : SUCCESS in 29m 21s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/48d8290e8ab849bba95f84482dbbf75d : SUCCESS in 18m 46s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4a674118ca454fc6a3d16eb6efe66ac3 : SUCCESS in 31m 50s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fec91db3d70c4ca79c002302c7c4720c : SUCCESS in 29m 06s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/39469a8d1d8e4d45acdfeb6df65a0e14 : SUCCESS in 22m 14s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8b2acd009b784c4691a088c77dd25121 : SUCCESS in 14m 36s","accounts_in_message":[],"_revision_number":1},{"id":"2e7bf6e2fa01a0e6206cc6c1e6de07b243055101","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-22 13:56:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"1bd21bcff29c52173d79534ec9c301b89cba03ca","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-22 13:58:32.000000000","message":"Patch Set 1:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1977d69603064e67ab0dc0ac590a4d95\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af793b28d7624e1aadca1cb602d23a5f : SUCCESS in 52s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2dc1c801f320425c851803bb111f0e2a : FAILURE in 1m 06s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0487610a28befb172a8168d8aa4c858ec02e014c","revisions":{"0487610a28befb172a8168d8aa4c858ec02e014c":{"kind":"REWORK","_number":1,"created":"2026-06-17 12:22:10.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/44/993744/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/44/993744/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/44/993744/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/44/993744/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/44/993744/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/44/993744/1"}}},"commit":{"parents":[{"commit":"c15bcf6bee746e2f7df196d2298bda5cfc2e84ac","subject":"Merge \"CI: Stop using image from Docker Hub\" into stable/2026.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c15bcf6bee746e2f7df196d2298bda5cfc2e84ac"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-06-13 16:53:17.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-06-17 12:22:10.000000000","tz":0},"subject":"Fix paths to EPEL repositories","message":"Fix paths to EPEL repositories\n\nOur EPEL repository file was building paths using variable $releasever,\ni.e. `10`. This now maps to `/pub/epel/10.3` which contains packages\nthat may not be compatible with Rocky Linux 10.2.\n\nChange paths to use the minor version when defined (it is undefined on\nCentOS Stream). This will produce paths such as `/pub/epel/10.2` on\nRocky Linux 10.2.\n\nWe could also use `/pub/epel/10z` to point to the latest stable release,\nbut it returns a 404 error on download.fedoraproject.org.\n\nCloses-Bug: #2156407\nChange-Id: I62c16b9d09607ab728068d64196b11ed83e03195\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n(cherry picked from commit b00b6cc804ab18b5b76d3463727e660d0badc856)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0487610a28befb172a8168d8aa4c858ec02e014c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0487610a28befb172a8168d8aa4c858ec02e014c"}]},"branch":"refs/heads/stable/2026.1","conflicts":{"base":"63b35e747bc0579b095db4032df489bd617dbfc7","ours":"c15bcf6bee746e2f7df196d2298bda5cfc2e84ac","theirs":"b00b6cc804ab18b5b76d3463727e660d0badc856","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
