)]}'
{"id":"openstack%2Fkayobe~993747","triplet_id":"openstack%2Fkayobe~stable%2F2025.1~I62c16b9d09607ab728068d64196b11ed83e03195","project":"openstack/kayobe","branch":"stable/2025.1","attention_set":{},"removed_from_attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2026-06-22 14:48:48.000000000","reason":"Change was submitted"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2026-06-22 14:48:48.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I62c16b9d09607ab728068d64196b11ed83e03195","subject":"Fix paths to EPEL repositories","status":"MERGED","created":"2026-06-17 12:27:37.000000000","updated":"2026-06-22 14:50:12.000000000","submitted":"2026-06-22 14:48:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"993747","cherry_pick_of_change":992770,"cherry_pick_of_patch_set":4,"meta_rev_id":"adb9c39433284710e544e189ebf4de7ce1574cae","_number":993747,"virtual_id_number":993747,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-22 14:48:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2026-06-22 12:13:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2026-06-22 12:13:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-17 16:06:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-17 19:59:15.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2026-06-22 12:13:45.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"}],"messages":[{"id":"b43ebd13d53cf0fe6b8b5e0a87fece8f14bc16a6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-06-17 12:27:37.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"5094d99287a723416ea83e9bc34a896fe0316fb6","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-06-17 12:48:02.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f27dac6686c2be8190fc66d14c59b91cd9f1b8e3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-06-17 13:19:04.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Workflow-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"4bd680fb0514dd4da1488245f313a7d41f9b9127","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-06-17 13:22:16.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1e134c2ce7e6604ce5511f0c832f499c83009471","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-06-17 13:40:16.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"d05fdddcd2d78db257d80c354ebd4a30da741159","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-17 16:06:30.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1161c58661754fc8a2f6d05a8fb368bc\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3627a70086574d409ea96ef7eb0181dd : SUCCESS in 3m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e75c91f2968143f8abfb4b7c57dff29e : SUCCESS in 4m 17s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1a5bcf25c4b3474e8079aed4d4fa7e4e : SUCCESS in 6m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a381927dd9e4969b9d04a8cce841ae3 : SUCCESS in 6m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f930181122bd4d93a04e0d52a2b65a50 : SUCCESS in 5m 32s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/09ffa36036e54c8c864de89a8c26f4f2 : SUCCESS in 4m 54s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/bc7080c140c44796b6952f0ef6560406 : SUCCESS in 16m 07s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a9bca0cb68ca44cca07ed8a63b63bd10 : SUCCESS in 10m 52s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/0908300b67654fd38bd77d76fefb4319 : SUCCESS in 16m 32s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/94724f1740cb42e981a5bb714218bc57 : SUCCESS in 15m 29s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4c76947183cc48ec8901a12de3f0218e : SUCCESS in 9m 20s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/5dcfbdcf46bb49ce8de231ec0b152200 : SUCCESS in 16m 54s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/14603b5d05a2477eb1fe16d01940083d : SUCCESS in 21m 38s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0e2e37b4106848f89d9f0617c75353b8 : SUCCESS in 11m 46s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/c6f94715a4724ea680761a3a9fd95cac : SUCCESS in 37m 57s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/064f9498830e44e2a48bd8cd422de637 : SUCCESS in 1h 11m 33s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/a1ed0564ae0243a9b69476f07f346048 : SUCCESS in 48m 14s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/9fd5fbb9fe494531811dba8b1dc11b7f : SUCCESS in 1h 01m 23s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/44f38173532b4c7ca5b10e3609daef12 : SUCCESS in 1h 03m 20s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/476f3f58f776484f82a5e36b4e4f2d48 : SUCCESS in 41m 20s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bca8fed61d6841a7a7f9203ee6a45b10 : SUCCESS in 31m 44s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/7f079b0087824e8d8169337c251741a2 : SUCCESS in 34m 00s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/1fa5eb12c2b040eaadd3a895dabd222f : SUCCESS in 1h 53m 47s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/815915524c344d3b951cdc65786f7bf1 : SUCCESS in 2h 19m 28s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/d458155787324e61be880488e9081692 : SUCCESS in 11m 31s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/a076a100996c4666a0b3a14c0f3e6e08 : SUCCESS in 10m 46s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/901e92f217d54631984de1c2513efecb : SUCCESS in 11m 21s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/8a25d5c1c13946578e9324040e13ae92 : SUCCESS in 19m 01s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/65f505d2c99f4727a45652651a4c30d0 : SUCCESS in 18m 14s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/7c75e8eb51924cd5a7f76f3bbd9ee4b2 : SUCCESS in 18m 41s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/d228e4dc2644409db699c0815440fc97 : SUCCESS in 39m 37s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/bb4c8206e8bf4a728e5e2c86ef8c6975 : SUCCESS in 13m 31s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/8b62a5d9df1e4bf48d6be79957de1cd9 : SUCCESS in 10m 45s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/24e7049bc22f4ab0b47abb1145407d60 : SUCCESS in 11m 40s","accounts_in_message":[],"_revision_number":3},{"id":"1b59e3479f9468abb0b888fa2a9f0df49f97ed93","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-06-17 16:58:15.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"5abb2249ea54f60d62ea472f51e4224f558dbda6","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-17 18:29:39.000000000","message":"Patch Set 3:\n\nBuild succeeded (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d1fd376e6fd64ff1857ef6b87be7f706\n\n- kayobe-infra-vm-centos10s https://zuul.opendev.org/t/openstack/build/997d819fe326490bb18d17892ddc169e : SUCCESS in 8m 34s (non-voting)\n- kayobe-infra-vm-centos10s-cloud-image https://zuul.opendev.org/t/openstack/build/6e1fe6003f404f039504d98de374a5a0 : SUCCESS in 19m 08s (non-voting)\n- kayobe-infra-vm-centos9s https://zuul.opendev.org/t/openstack/build/377168031ca54f5896f130c50ce536eb : SUCCESS in 9m 52s (non-voting)\n- kayobe-infra-vm-centos9s-cloud-image https://zuul.opendev.org/t/openstack/build/0ae243762a824a6d86fa70e5b25117db : SUCCESS in 19m 35s (non-voting)\n- kayobe-infra-vm-rocky10-cloud-image https://zuul.opendev.org/t/openstack/build/a0bf90e0bd6b43f59a05e77f757fb074 : SUCCESS in 12m 21s\n- kayobe-infra-vm-rocky9-cloud-image https://zuul.opendev.org/t/openstack/build/382c9a26243945c6b89171fdecbb618d : SUCCESS in 27m 19s\n- kayobe-infra-vm-ubuntu-noble-cloud-image https://zuul.opendev.org/t/openstack/build/d16eac6b0ecc4a3bbdb5fc0c626a7265 : SUCCESS in 16m 50s (non-voting)\n- kayobe-overcloud-centos10s https://zuul.opendev.org/t/openstack/build/4eb20a5455844dd6873a246668a6a41a : SUCCESS in 1h 08m 05s (non-voting)\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/2a04bd773d3149d8b47c170687834b8c : SUCCESS in 33m 22s (non-voting)\n- kayobe-overcloud-host-configure-centos10s https://zuul.opendev.org/t/openstack/build/53c25e0bb8e5456e8fabf46b32da3359 : SUCCESS in 22m 41s (non-voting)\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/762895903f7d4128beb7c7a60bfe71a5 : SUCCESS in 25m 48s (non-voting)\n- kayobe-overcloud-tls-centos10s https://zuul.opendev.org/t/openstack/build/41e32e5eb5734e66ae7a49cd42a8a088 : SUCCESS in 1h 04m 44s (non-voting)\n- kayobe-seed-centos10s https://zuul.opendev.org/t/openstack/build/9e5e57072a1b4343b1b32054a0ddd2be : SUCCESS in 20m 38s (non-voting)\n- kayobe-seed-images-centos10s https://zuul.opendev.org/t/openstack/build/38dbb6b056874059b7e1adc26a4a5158 : SUCCESS in 43m 04s (non-voting)\n- kayobe-seed-images-centos9s https://zuul.opendev.org/t/openstack/build/d41cc5fb55494ceebabb71d7a12182f1 : SUCCESS in 59m 59s (non-voting)\n- kayobe-seed-images-rocky10 https://zuul.opendev.org/t/openstack/build/3db1276b908f4f7a9b158c7874979c08 : SUCCESS in 1h 03m 16s\n- kayobe-seed-images-rocky10-podman https://zuul.opendev.org/t/openstack/build/71127e88a3f845f98b02e8643d88c8a6 : SUCCESS in 44m 30s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/bd10f70759de468a947d9a2cab3fe7ea : SUCCESS in 21m 24s\n- kayobe-seed-images-rocky9-podman https://zuul.opendev.org/t/openstack/build/733cd785a7b7437c9888c0779362e5ed : SUCCESS in 1h 25m 12s\n- kayobe-seed-images-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0b2a1bbce5794fce9f3d95b2f4261ff3 : SUCCESS in 44m 39s\n- kayobe-seed-images-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/f7f36309ff54457f9c9c459850352265 : SUCCESS in 25m 12s\n- kayobe-seed-vm-centos10s https://zuul.opendev.org/t/openstack/build/e817ed9f5be34003ad840c9747f8ec5b : SUCCESS in 10m 34s (non-voting)\n- kayobe-seed-vm-centos10s-cloud-image https://zuul.opendev.org/t/openstack/build/884372923dce4974a47a099d2b7d26f7 : SUCCESS in 20m 53s (non-voting)\n- kayobe-seed-vm-centos10s-q35 https://zuul.opendev.org/t/openstack/build/ca6846ee134b4db69b23c2858a6dbca0 : SUCCESS in 8m 42s (non-voting)\n- kayobe-seed-vm-centos9s https://zuul.opendev.org/t/openstack/build/af05ad860133434cad7e02205eec1a74 : SUCCESS in 20m 05s (non-voting)\n- kayobe-seed-vm-centos9s-cloud-image https://zuul.opendev.org/t/openstack/build/f48fd073a2ed4f67a87329649a3eec57 : FAILURE in 28m 13s (non-voting)\n- kayobe-seed-vm-centos9s-q35 https://zuul.opendev.org/t/openstack/build/035d3e773ae64f61921fe30b27fe4cfd : SUCCESS in 17m 11s (non-voting)\n- kayobe-seed-vm-rocky10-cloud-image https://zuul.opendev.org/t/openstack/build/8aa06faeb4d14eea880f66c5f0dd861d : SUCCESS in 12m 19s\n- kayobe-seed-vm-rocky10-q35 https://zuul.opendev.org/t/openstack/build/2f811668b5e84d62a42ab88fdd1a8f25 : SUCCESS in 10m 15s\n- kayobe-seed-vm-rocky9-cloud-image https://zuul.opendev.org/t/openstack/build/96848d9f98fe4a9ca3dcb24f4de4fe22 : SUCCESS in 25m 48s\n- kayobe-seed-vm-rocky9-q35 https://zuul.opendev.org/t/openstack/build/b9d284fc0b0d4b26ad657c7b03b75d04 : SUCCESS in 17m 21s\n- kayobe-seed-vm-ubuntu-noble-cloud-image https://zuul.opendev.org/t/openstack/build/a10fc014eaea4436bf64b39b625f549b : SUCCESS in 17m 46s (non-voting)\n- kayobe-seed-vm-ubuntu-noble-q35 https://zuul.opendev.org/t/openstack/build/dcb89493c8284341a04dadcb2dfea3a9 : SUCCESS in 9m 01s","accounts_in_message":[],"_revision_number":3},{"id":"ed60a3826c8e0d3d7c791497302c8a5c39f41081","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-06-22 12:13:45.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"80cbf3f2a76f1aefaa81e85e0bc9a17ecf7daf62","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-22 12:14:12.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"4ec424ec5942a371ef4f4d4c7dc775a7f27a8b9a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-22 14:48:48.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d454233b3d134120afb3aa024bfc20ab\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3acd82fe1caf453ead2c4e5af47b050c : SUCCESS in 4m 29s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e03b4fddc45945ce85a1a99fe3134c25 : SUCCESS in 6m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e3b7cba1d504e6f9d64d0f2a3aadffc : SUCCESS in 9m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e557cfd7d3974898a9f2966e3ec1fa5e : SUCCESS in 6m 25s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3c9571aa9e9a49678a80ef7daff7fc00 : SUCCESS in 5m 43s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/78f385a1ea3c44f795d1dc46de00ba06 : SUCCESS in 12m 17s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/27ed4b8782eb4d2f80faca610e23efea : SUCCESS in 10m 57s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/abdc0a1b174245168df644d9515eebff : SUCCESS in 19m 02s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/ff0f852a1df04b10a11d283ddcacc444 : SUCCESS in 16m 42s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/01dc0901549d4140922cafad642e3f1f : SUCCESS in 12m 30s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/022f47004d854a288a3a52e649846c7c : SUCCESS in 16m 32s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/033deed9e7dc45cc84feaea70996b56a : SUCCESS in 22m 43s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9f42ac47c20e4ffe82bd5ab20e37c9d3 : SUCCESS in 16m 33s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/785a62938d404327804b39e37ea09bde : SUCCESS in 57m 04s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/30c4899f4ef346f6b76bbd5951ca73a2 : SUCCESS in 56m 40s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/125cf6154936463ea62865c06bd64242 : SUCCESS in 1h 09m 06s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/5d1e8bded8ea4a82bd2074d548f4b322 : SUCCESS in 1h 04m 55s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/b40f1835fbf5455f9c7d7c4d51dd89d1 : SUCCESS in 1h 21m 17s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/55e53cc54877499b8a2eb3fc733e29c5 : SUCCESS in 1h 09m 51s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f96af7dd1704450fbed908ebaa597fd2 : SUCCESS in 31m 36s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/9ee747b9a5324a44b41eeeefe2b99e73 : SUCCESS in 58m 34s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/5664eb498e5b485795f495609c651453 : SUCCESS in 1h 21m 12s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e10427b39a1047b491d5d7999c430913 : SUCCESS in 2h 29m 04s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/37cd2beb5cd24c4ca7141a6dff6f4ccf : SUCCESS in 17m 53s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/5b61bf3f947b46a78ab9f33bedf7f4b7 : SUCCESS in 23m 29s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/4d362ce3e4cb4be3a7a6cf26ddd6c911 : SUCCESS in 19m 32s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/51730cff63824290a40bd845545205df : SUCCESS in 18m 40s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e83f743a5f53473a95bf357b4b23cce6 : SUCCESS in 16m 29s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/6287fee0498c4871af1dcc98a3b33917 : SUCCESS in 18m 29s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/b14759a90f104ea280bab5d6b4534fee : SUCCESS in 23m 38s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/ccc907bcbdcd4279aa82afe62fdff6cc : SUCCESS in 10m 11s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/71d783a045804a9187a1af40c8f3aa41 : SUCCESS in 17m 54s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d3ba3ec2d9774858a14c9a1127575eae : SUCCESS in 8m 30s","accounts_in_message":[],"_revision_number":3},{"id":"1faeebba525ab03010a31815c5caab326cbb994c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-22 14:48:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"adb9c39433284710e544e189ebf4de7ce1574cae","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-22 14:50:12.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7a2dbc726c1640559facec98f4dcc4ae\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c078088368684abaa7144be151986301 : SUCCESS in 44s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/594e5a71f3f4439881a0e946b1e3b041 : SUCCESS in 1m 09s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"8d046017d52970daac3b1a585cf577c95bd86b5d","revisions":{"05486f3879f543af9b2be99e4c2824899fe81c11":{"kind":"REWORK","_number":1,"created":"2026-06-17 12:27:37.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/47/993747/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/47/993747/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/47/993747/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/47/993747/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/47/993747/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/47/993747/1"}}},"commit":{"parents":[{"commit":"383628fc236da3821f09786a2f51761cdaf54339","subject":"Adds support for custom watcher configuration files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/383628fc236da3821f09786a2f51761cdaf54339"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-06-13 16:53:17.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-06-17 12:27:37.000000000","tz":0},"subject":"Fix paths to EPEL repositories","message":"Fix paths to EPEL repositories\n\nOur EPEL repository file was building paths using variable $releasever,\ni.e. `10`. This now maps to `/pub/epel/10.3` which contains packages\nthat may not be compatible with Rocky Linux 10.2.\n\nChange paths to use the minor version when defined (it is undefined on\nCentOS Stream). This will produce paths such as `/pub/epel/10.2` on\nRocky Linux 10.2.\n\nWe could also use `/pub/epel/10z` to point to the latest stable release,\nbut it returns a 404 error on download.fedoraproject.org.\n\nCloses-Bug: #2156407\nChange-Id: I62c16b9d09607ab728068d64196b11ed83e03195\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n(cherry picked from commit b00b6cc804ab18b5b76d3463727e660d0badc856)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/05486f3879f543af9b2be99e4c2824899fe81c11"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/05486f3879f543af9b2be99e4c2824899fe81c11"}]},"branch":"refs/heads/stable/2025.1","conflicts":{"base":"63b35e747bc0579b095db4032df489bd617dbfc7","ours":"383628fc236da3821f09786a2f51761cdaf54339","theirs":"b00b6cc804ab18b5b76d3463727e660d0badc856","merge_strategy":"recursive","contains_conflicts":false}},"da8012b840cf09ab8dac6091fd32ecc614f85836":{"kind":"REWORK","_number":2,"created":"2026-06-17 13:19:04.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/47/993747/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/47/993747/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/47/993747/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/47/993747/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/47/993747/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/47/993747/2"}}},"commit":{"parents":[{"commit":"383628fc236da3821f09786a2f51761cdaf54339","subject":"Adds support for custom watcher configuration files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/383628fc236da3821f09786a2f51761cdaf54339"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-06-17 13:18:15.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-06-17 13:18:15.000000000","tz":120},"subject":"Fix paths to EPEL repositories","message":"Fix paths to EPEL repositories\n\nOur EPEL repository file was building paths using variable $releasever,\ni.e. `10`. This now maps to `/pub/epel/10.3` which contains packages\nthat may not be compatible with Rocky Linux 10.2.\n\nChange paths to use the minor version when defined (it is undefined on\nCentOS Stream). This will produce paths such as `/pub/epel/10.2` on\nRocky Linux 10.2.\n\nWe could also use `/pub/epel/10z` to point to the latest stable release,\nbut it returns a 404 error on download.fedoraproject.org.\n\nCompared to the original commit on master, this change is adapted to be\ncompatible with Rocky Linux 9.\n\nCloses-Bug: #2156407\nChange-Id: I62c16b9d09607ab728068d64196b11ed83e03195\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n(cherry picked from commit b00b6cc804ab18b5b76d3463727e660d0badc856)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/da8012b840cf09ab8dac6091fd32ecc614f85836"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/da8012b840cf09ab8dac6091fd32ecc614f85836"}]},"branch":"refs/heads/stable/2025.1"},"8d046017d52970daac3b1a585cf577c95bd86b5d":{"kind":"REWORK","_number":3,"created":"2026-06-17 13:40:16.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/47/993747/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/47/993747/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/47/993747/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/47/993747/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/47/993747/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/47/993747/3"}}},"commit":{"parents":[{"commit":"383628fc236da3821f09786a2f51761cdaf54339","subject":"Adds support for custom watcher configuration files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/383628fc236da3821f09786a2f51761cdaf54339"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-06-17 13:40:09.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-06-17 13:40:09.000000000","tz":120},"subject":"Fix paths to EPEL repositories","message":"Fix paths to EPEL repositories\n\nOur EPEL repository file was building paths using variable $releasever,\ni.e. `10`. This now maps to `/pub/epel/10.3` which contains packages\nthat may not be compatible with Rocky Linux 10.2.\n\nChange paths to use the minor version when defined (it is undefined on\nCentOS Stream). This will produce paths such as `/pub/epel/10.2` on\nRocky Linux 10.2.\n\nWe could also use `/pub/epel/10z` to point to the latest stable release,\nbut it returns a 404 error on download.fedoraproject.org.\n\nCompared to the original commit on master, this change is adapted to be\ncompatible with Rocky Linux 9.\n\nCloses-Bug: #2156407\nChange-Id: I62c16b9d09607ab728068d64196b11ed83e03195\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n(cherry picked from commit b00b6cc804ab18b5b76d3463727e660d0badc856)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8d046017d52970daac3b1a585cf577c95bd86b5d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8d046017d52970daac3b1a585cf577c95bd86b5d"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
