)]}'
{"id":"openstack%2Fkayobe~993763","triplet_id":"openstack%2Fkayobe~master~I659e89efcad628187caccd83018d5068eb4a3bf7","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2026-06-22 14:48:51.000000000","reason":"Change was submitted"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2026-06-22 14:48:51.000000000","reason":"Change was submitted"},"35263":{"account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"last_update":"2026-06-22 12:37:07.000000000","reason":"\u003cGERRIT_ACCOUNT_35263\u003e replied on the change","reason_account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}}},"hashtags":[],"change_id":"I659e89efcad628187caccd83018d5068eb4a3bf7","subject":"Bump stackhpc.openstack to 0.10.2","status":"MERGED","created":"2026-06-17 14:35:31.000000000","updated":"2026-06-22 14:49:56.000000000","submitted":"2026-06-22 14:48:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"993763","meta_rev_id":"22c8daf97e4cc27827af8c89e604490a91b53378","_number":993763,"virtual_id_number":993763,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-22 14:48:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2026-06-22 12:14:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-06-22 12:37:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-06-22 12:37:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-17 16:02:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-06-17 18:26:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-17 20:27:30.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2026-06-22 12:14:11.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2026-06-22 12:24:48.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"}],"messages":[{"id":"91116091f1d6093557cb2195a8378a1f3794b3e9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-06-17 14:35:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e93ff9f6c2f40df555d63ae7c0842635e9e07fc3","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-06-17 14:36:02.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d17c79ff69e864a13dc0d2ae406bb392701030f8","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-17 16:02:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d37f4f7c25ef48dfb9bc5805a57b4b30\n\n- kayobe-ansible-control-host-configure-centos10s https://zuul.opendev.org/t/openstack/build/7f45f82df36e47a39b839706893ce71f : SUCCESS in 18m 11s (non-voting)\n- kayobe-infra-vm-centos10s https://zuul.opendev.org/t/openstack/build/a6a9e4f994694493abdd96db0ccc3080 : SUCCESS in 24m 51s (non-voting)\n- kayobe-infra-vm-centos10s-cloud-image https://zuul.opendev.org/t/openstack/build/2173ecb3b7b947fbbdf0d1bc4bd73f84 : SUCCESS in 16m 53s (non-voting)\n- kayobe-infra-vm-rocky10-cloud-image https://zuul.opendev.org/t/openstack/build/37095f4d31e947f487025c3c88752f1f : SUCCESS in 18m 21s\n- kayobe-infra-vm-ubuntu-noble-cloud-image https://zuul.opendev.org/t/openstack/build/58e8385e003142679a47cf10d3ed2be4 : SUCCESS in 27m 11s (non-voting)\n- kayobe-overcloud-centos10s https://zuul.opendev.org/t/openstack/build/d1b13b5225f04b7086ca128a223d7fa7 : SUCCESS in 56m 41s (non-voting)\n- kayobe-overcloud-host-configure-centos10s https://zuul.opendev.org/t/openstack/build/4cbd786cb353478dbc45f806f0ebd07f : SUCCESS in 19m 10s (non-voting)\n- kayobe-overcloud-tls-centos10s https://zuul.opendev.org/t/openstack/build/3b325009e7c84449bba660028502794d : SUCCESS in 1h 05m 55s (non-voting)\n- kayobe-seed-centos10s https://zuul.opendev.org/t/openstack/build/202360d1f81d420bb5893b0fffdfc580 : SUCCESS in 20m 17s (non-voting)\n- kayobe-seed-images-centos10s https://zuul.opendev.org/t/openstack/build/d24c05d5720c42f4873e70ea07e068c9 : SUCCESS in 52m 30s (non-voting)\n- kayobe-seed-images-rocky10 https://zuul.opendev.org/t/openstack/build/7d1eedb037314589bb1bfb39bfb1e45b : SUCCESS in 37m 16s\n- kayobe-seed-images-rocky10-podman https://zuul.opendev.org/t/openstack/build/52ba6a21986347e7a391460e6febbf74 : SUCCESS in 1h 01m 02s\n- kayobe-seed-images-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bc673097117f4b5aaacb74512f655130 : SUCCESS in 34m 03s\n- kayobe-seed-images-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/a4922162199543c2a790673fc485fae3 : SUCCESS in 1h 21m 16s\n- kayobe-seed-vm-centos10s https://zuul.opendev.org/t/openstack/build/62e67232bf764b01ae22e71aecedeec4 : SUCCESS in 16m 04s (non-voting)\n- kayobe-seed-vm-centos10s-cloud-image https://zuul.opendev.org/t/openstack/build/027c799438f54e07bebab526a3c10606 : SUCCESS in 28m 13s (non-voting)\n- kayobe-seed-vm-centos10s-q35 https://zuul.opendev.org/t/openstack/build/64a4e6fbbf9f411f8f799f2986518c0f : SUCCESS in 22m 01s (non-voting)\n- kayobe-seed-vm-rocky10-cloud-image https://zuul.opendev.org/t/openstack/build/a4ae55d5b53f402b9ae9bbe17219cc4b : SUCCESS in 19m 17s\n- kayobe-seed-vm-rocky10-q35 https://zuul.opendev.org/t/openstack/build/f78ad44641574b6a8d4f7332d07a4186 : SUCCESS in 14m 25s\n- kayobe-seed-vm-ubuntu-noble-cloud-image https://zuul.opendev.org/t/openstack/build/efe80bef582e448797047e2318f1efe3 : SUCCESS in 29m 38s (non-voting)\n- kayobe-seed-vm-ubuntu-noble-q35 https://zuul.opendev.org/t/openstack/build/3ba8a4d37b7c4854a7d0016f1e15f24e : SUCCESS in 14m 46s","accounts_in_message":[],"_revision_number":1},{"id":"92e536efe14e20cf2acd463852d20cc3aa47834c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-06-17 16:27:58.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e15c96a0a32e58d14ab0a8f40a2b961365d2c4c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-17 18:26:50.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0aabed1b7f674f76be41c628c997d1c0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1fab6f998cd24511b4ea60a8bd9bcba7 : SUCCESS in 9m 48s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/b1fc0aad46514edda7780fa5cbd48b4b : SUCCESS in 6m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/46e8e03144fd41f4bcc57e1744d105b2 : SUCCESS in 6m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/34bf3e6d71e94ebd9f7deed6fef5aa05 : SUCCESS in 5m 20s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5567fab1e1d747a09bc1aa28c666c0e2 : SUCCESS in 4m 07s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/4c4265cc518847d48b5fafdc6f13a8f7 : SUCCESS in 5m 11s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/29d9d75abdeb426ea96ebc355db7486b : SUCCESS in 6m 29s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a483137792af484b8e8b20c4d3d15e66 : SUCCESS in 5m 41s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/399c4f9d08f14b4eb3c9a1f36e238df7 : SUCCESS in 12m 01s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d01980dc03574d04b8de338610f60a00 : SUCCESS in 11m 22s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/597303d6aa214dab9258dc148b85b1a9 : SUCCESS in 22m 37s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/e7bb3a19261e4989a7b131cdd5af96d9 : SUCCESS in 14m 26s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8b7f0b5a66a84aa4b271b399c385e7dd : SUCCESS in 17m 48s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/1131d5733fc247028e27d8884013cf90 : SUCCESS in 15m 15s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/4c402abdc4e0407e9c9862d28e7c3bb3 : SUCCESS in 21m 48s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7e8add5e7c53463a96c5cb52481521ad : SUCCESS in 18m 25s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/b3bb8940650b45cd89075452ad6f74c6 : SUCCESS in 17m 00s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/edba33e68fde4ac5b8a286dea7ded025 : SUCCESS in 9m 35s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/21c7f47b3ac140cb8183cb057932f414 : SUCCESS in 1h 00m 04s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/9d633c69f2264f8bbe163db2daf2422e : SUCCESS in 59m 43s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/91d6da6716ac494f8607fa6617979345 : SUCCESS in 1h 06m 02s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0f4d29509ac7486eade9471d832d3132 : SUCCESS in 58m 40s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/e559ec0b480745788f40ce266a9e0c9e : SUCCESS in 1h 02m 04s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/bbdde3a375e24ef7a13c7633e9c2b10d : FAILURE in 1h 31m 02s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fd446b1f59d24c69b6c1b5ced6527be5 : SUCCESS in 1h 54m 15s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/b45468670e944119bc91ee57a8e66bb6 : SUCCESS in 16m 25s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/6e530346c360424cb56b2bb6c4575b42 : SUCCESS in 23m 04s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/45dd4e53064243089c5ec408e77c6664 : SUCCESS in 13m 03s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/31d2ec4274f34da0804b66f7688cee03 : SUCCESS in 14m 19s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/9eb6362481fc4bf7aca42babd96a560f : SUCCESS in 41m 51s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b22a29da20804c3ba572313189dc6e5f : SUCCESS in 20m 11s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/2d03c9a0934d4ef7920c1dabf3eff5e9 : SUCCESS in 21m 23s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4cf74d9194294cd89e8018c3a9e3a7a4 : SUCCESS in 16m 25s","accounts_in_message":[],"_revision_number":2},{"id":"ea26a451bb254d6e1c095c704ab2fb573d1ae730","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-06-18 15:27:31.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"33f1768782236727a423e27dd5af3d33346c04ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-18 17:24:01.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eca6fd49887c40819502d611edef1e44\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f12e614f6764907a01f4a3a3ad47b2f : SUCCESS in 10m 25s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/900ce17d95324a6a9d0466e4c51c8928 : SUCCESS in 8m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f4f42b673dbf437b910f92d207cc1ffd : SUCCESS in 6m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42d08510703b40e1a5a8b072a52ef5d0 : SUCCESS in 6m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/dbe8e21b6b3d4555b2a450a47bee0f51 : SUCCESS in 4m 13s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/73e06297a14843f78846c3b62122b7a1 : SUCCESS in 9m 23s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fd889ea2ecde4b55a8b160977a402ce7 : SUCCESS in 6m 30s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/0f1a643e80764fe9bec9c6d1ed4b929b : SUCCESS in 4m 37s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e651445f6db54c7f94d014904e911ba3 : SUCCESS in 13m 36s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a4ea23cafc574b198a2e4e20e0067005 : SUCCESS in 11m 35s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/0e7209c781a24713a6e2189d071907ab : SUCCESS in 15m 21s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/eb8e832a9aef4a27b928818c720aa089 : SUCCESS in 14m 19s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7a50fc325858455e99e46575658d6db9 : SUCCESS in 18m 33s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/a0a4469612e34dde852fae8f6dfb887d : SUCCESS in 17m 28s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/74d9e15b852f42a1b2922ea4cb6854de : SUCCESS in 20m 20s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dc115b35234f4cc68a701828b162d60b : SUCCESS in 19m 49s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/e4799ee3297d4953b37d37f5baf97dee : SUCCESS in 21m 48s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a550a3da30bd44bab9b38d5edd69036e : SUCCESS in 18m 20s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/dbf414cdf862483b97ffc0fed508e703 : SUCCESS in 1h 01m 16s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/51270c6fb54e46f3b713e5450e475efc : SUCCESS in 1h 10m 32s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/37b7ca157b06458c8a7424f6208dca95 : SUCCESS in 1h 14m 35s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/26f28851305b49b391b9dea6c4eec4f4 : SUCCESS in 1h 16m 25s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/f2285482e29c4808a31321af9f99e192 : SUCCESS in 1h 04m 35s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/6f143b8cb87b426592ead2de08bf308f : SUCCESS in 1h 54m 26s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e57a10eef52842fe800a066bcbeb84c4 : SUCCESS in 1h 49m 00s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/6bf8e6af36a9495196050453dd09261d : SUCCESS in 28m 56s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/9aded9c8442b401cadd9269df95ce142 : SUCCESS in 21m 46s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a223104217b4441ab0ca514aec71faed : SUCCESS in 23m 54s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/477fcc0c7d164e909894d96f0b239d21 : SUCCESS in 22m 14s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/be3bc766b5864d8cbe2c77baa7e1d460 : SUCCESS in 23m 26s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bdd9f2495c014d55ab1f5709b1869042 : SUCCESS in 34m 16s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/1175f6b5ca634e029641dbe1764ab18c : SUCCESS in 27m 04s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/85237a73abf846abafea40bba0ec2f2a : SUCCESS in 22m 05s","accounts_in_message":[],"_revision_number":2},{"id":"6bd1143b7b18af95c4cce12ef4c674aacc4b74f5","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-06-22 12:14:11.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"b6ca09834d4d3bda0bdd41002175109adb849b03","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-06-22 12:37:07.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"664b79020675f21ac2b46c1e9f8cd2b343c90668","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-22 12:37:35.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"3166a8275ac9005907441ffd643e076d98e68562","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-22 14:48:51.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c113468b6f7a48a4a1f4ec740b4dda43\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/249f3ded6cf848799c6ea8975540ef0b : SUCCESS in 5m 11s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/7f654edcf8414fdc9e08f2eb796b27ab : SUCCESS in 6m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/11f8d50f39404a44bf05e1cfe1a2fb4c : SUCCESS in 4m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f45b22b58c8d4adebd5e67cd6d4e66fe : SUCCESS in 3m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6d1a14c4a96c47f2a0f98ed550311d6d : SUCCESS in 5m 33s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ae1d0719e4394bd09917cdc1e5ab27a6 : SUCCESS in 5m 30s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8186aa3bec2649ef8d40d373caf489ca : SUCCESS in 14m 14s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/4faa3b2cbe2c43f0a13309befcbc0005 : SUCCESS in 11m 10s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/b690b65b99e94fa6b02bf9a0b0944d3c : SUCCESS in 15m 37s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/840805c6883d4d569b2b1c433c8f3031 : SUCCESS in 15m 52s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/2302bf076def4d21a57ff99f86327488 : SUCCESS in 13m 42s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bb04eb66c3d74501bccb0806497bee16 : SUCCESS in 19m 36s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/f46244e4a8c342ccb1a7fa56303724af : SUCCESS in 12m 22s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/374e17fc7b6a40a39548110b3bdafb04 : SUCCESS in 19m 08s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/4abd102f3e96491ca250bafb2ff1bf5d : SUCCESS in 1h 01m 37s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/dbebfda9d79e44258f4025d891805e6c : SUCCESS in 1h 03m 47s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/9aee317047364bf48bdd47efb3239098 : SUCCESS in 56m 20s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7e6506863aed49698fd064240266fc9b : SUCCESS in 1h 06m 43s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/c786306d0da142c198b033421e66542b : SUCCESS in 1h 12m 18s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/52434b63cc8c4db6a1a1b2f98e4a4f9c : SUCCESS in 1h 56m 28s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f50773d057a84fd6afd81228e0de1305 : SUCCESS in 1h 40m 33s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/183864ca4c434fb484afc92dc87b9a36 : SUCCESS in 14m 34s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/340b0622b66f441bb16561b51dd7d16d : SUCCESS in 23m 09s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5b413ec1d6e54344843e76b945164597 : SUCCESS in 14m 09s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/f36b58688a154a2eac08bf8db19656ce : SUCCESS in 20m 20s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/8e6df7ed08fa49ceb6831aee82fb09ba : SUCCESS in 31m 24s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fe98206aa8d2419cbc8fb50b7eabd3fe : SUCCESS in 38m 44s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/fc41a5feccf346ff8c70addd587bbd8f : SUCCESS in 18m 59s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/39b3a0c3a56242b38d5802e6dae0aa02 : SUCCESS in 13m 20s","accounts_in_message":[],"_revision_number":2},{"id":"536b3c7a1cafc7045d9b8f3c89465c68ee196050","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-22 14:48:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"22c8daf97e4cc27827af8c89e604490a91b53378","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-22 14:49:56.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6cecedf3b32d4ea3b7820cc7d584eee5\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e094bb2672f4a32ac1d448bf54a01a4 : SUCCESS in 48s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/095362c2456a4d92a1a3c5e5436e12d6 : SUCCESS in 51s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"dd5eb04f59c832de4e583719d2611dc53217b251","revisions":{"bf356bf1352336e4da47041eba30cc9f5c04d4a1":{"kind":"REWORK","_number":1,"created":"2026-06-17 14:35:31.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/63/993763/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/63/993763/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/63/993763/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/63/993763/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/63/993763/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/63/993763/1"}}},"commit":{"parents":[{"commit":"38f47c4c5bb665098fac2007c3b58cfe0b5b4f1c","subject":"Merge \"Fix paths to EPEL repositories\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/38f47c4c5bb665098fac2007c3b58cfe0b5b4f1c"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-06-17 14:34:52.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-06-17 14:34:52.000000000","tz":120},"subject":"Bump stackhpc.openstack to 0.10.2","message":"Bump stackhpc.openstack to 0.10.2\n\nThis fixes a broken conditional causing IPA/overcloud image build\nfailures. No release note because this is only affecting upstream CI.\n\nChange-Id: I659e89efcad628187caccd83018d5068eb4a3bf7\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bf356bf1352336e4da47041eba30cc9f5c04d4a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bf356bf1352336e4da47041eba30cc9f5c04d4a1"}]},"branch":"refs/heads/master"},"dd5eb04f59c832de4e583719d2611dc53217b251":{"kind":"REWORK","_number":2,"created":"2026-06-17 16:27:58.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/63/993763/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/63/993763/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/63/993763/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/63/993763/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/63/993763/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/63/993763/2"}}},"commit":{"parents":[{"commit":"1f2df4b92302c485593abd3096de77131046224d","subject":"Merge \"Remove external_net_name variable\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1f2df4b92302c485593abd3096de77131046224d"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-06-17 16:27:11.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-06-17 16:27:11.000000000","tz":120},"subject":"Bump stackhpc.openstack to 0.10.2","message":"Bump stackhpc.openstack to 0.10.2\n\nThis fixes a broken conditional causing IPA/overcloud image build\nfailures.\n\nCloses-Bug: #2157021\nChange-Id: I659e89efcad628187caccd83018d5068eb4a3bf7\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/dd5eb04f59c832de4e583719d2611dc53217b251"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/dd5eb04f59c832de4e583719d2611dc53217b251"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
