)]}'
{"id":"openstack%2Fkayobe~995016","triplet_id":"openstack%2Fkayobe~master~I81c2b2a8e1ccde44854257ce8e34e3517c945a4d","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2026-08-07 18:30:01.000000000","reason":"Change was submitted"},"17669":{"account":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"last_update":"2026-08-07 14:51:27.000000000","reason":"\u003cGERRIT_ACCOUNT_17669\u003e replied on the change","reason_account":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"}},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2026-08-07 18:30:01.000000000","reason":"Change was submitted"},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2026-08-07 18:30:01.000000000","reason":"Change was submitted"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2026-08-07 18:30:01.000000000","reason":"Change was submitted"},"35263":{"account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"last_update":"2026-08-07 18:30:01.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I81c2b2a8e1ccde44854257ce8e34e3517c945a4d","subject":"Cleaner split of kolla-ansible install and configure","status":"MERGED","created":"2026-06-26 09:11:55.000000000","updated":"2026-08-07 18:31:03.000000000","submitted":"2026-08-07 18:30:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":34,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"995016","meta_rev_id":"1e17dae9a1db44a0d23841e9a72b4cbb718d423e","_number":995016,"virtual_id_number":995016,"owner":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-08-07 18:30:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":2,"date":"2026-08-07 14:51:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":2,"date":"2026-08-07 15:03:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2026-08-07 14:51:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":1,"date":"2026-08-07 15:03:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2026-08-07 14:26:20.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"}],"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-26 11:17:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-07-02 14:10:57.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"},{"updated":"2026-07-02 14:19:22.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"CC"},{"updated":"2026-07-07 12:32:21.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"CC"},{"updated":"2026-07-07 12:32:52.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2026-07-10 13:17:25.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2026-07-13 09:21:56.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2026-07-15 12:51:59.000000000","updated_by":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"reviewer":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"state":"CC"},{"updated":"2026-07-27 14:21:55.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2026-08-07 07:29:04.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2026-08-07 15:03:42.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"a785cda60f3db211269464f7a844770b6f50b407","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-06-26 09:11:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e3001f322e8849cee207ebafd0294e681cbc5b89","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-06-26 09:12:38.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"f4c270d88fa0567019b4c4d1033bf4b75021e9a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-26 11:17:14.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/92a32a33d8074ecfa5d728e48a981fbd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/766ab19b8bce41d595ccf7a3e809fd7a : SUCCESS in 8m 59s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/6d19e0d821cc4cc293a020998bd90db8 : SUCCESS in 7m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/edb2ecc2e5c948dfb66bc889cfa60f73 : SUCCESS in 4m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/49df2c6b8ad74a35818a69e3c2eb0163 : SUCCESS in 4m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6dd0731ca63e477f9ff429183d44e537 : SUCCESS in 4m 22s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/8aa9d7626cf7478392ca795f2a1a28e4 : SUCCESS in 3m 47s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0f1391dd224e47cb9bd1ed9664628b26 : SUCCESS in 7m 22s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/fcf140ba0ea74da7a0c65154a6527795 : SUCCESS in 5m 12s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/213caf0953ed42e8914f95fe27b19272 : FAILURE in 6m 06s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/25b3c35cefb4432e9655b1f195506acd : SUCCESS in 14m 03s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/9b46331530b247b6bee88804f5264509 : SUCCESS in 19m 19s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/02d361570b7148dca948258185899b1a : SUCCESS in 11m 25s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/49a050a88c7c44329e1b40a895d92b71 : SUCCESS in 17m 33s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/f4cc793b3837429a914b466c57a489ee : SUCCESS in 17m 15s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/2e9b73e554fd4285a5b8c9436b4e9d71 : SUCCESS in 14m 49s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f2354e1b5abe404da4c542d5dfe4b65e : SUCCESS in 19m 21s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/79422bd3ed1b4692a7c031988f926c1d : SUCCESS in 16m 11s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5014206bc13547e0bf1e1a0322d9cd65 : SUCCESS in 22m 55s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/57d95bb43efe4757a354b5b08e92a2a7 : SUCCESS in 1h 01m 07s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/028e6a4934af49d5897b07891fd288ae : SUCCESS in 1h 00m 01s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/63dedbbec6214df68cdfe15053f6d746 : SUCCESS in 1h 06m 27s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fdf835b5dfc74ba9b343787a91eb7007 : SUCCESS in 1h 06m 12s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/c50e4f149946424a9f90c5d920aaf345 : SUCCESS in 1h 10m 50s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/94e3174780e844fe9dd77e4a5b4565d6 : SUCCESS in 1h 50m 18s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/977811fb898a4e7492764d349911d900 : SUCCESS in 2h 04m 27s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/2e5f04ae895e49518ed822e97383bee4 : SUCCESS in 15m 36s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/11bd37803c094809bf79473c8ad7d090 : SUCCESS in 15m 55s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8d4baded339a43059d8105746a34a540 : SUCCESS in 22m 53s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/0ada9982d4ec436cbd60b0dfec669aa6 : SUCCESS in 23m 32s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/6b9f127e64c34966acf380431d58159a : SUCCESS in 23m 26s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cbfbe934b7c84019aa4c65410ced6b51 : SUCCESS in 42m 02s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/2fb26de198c74320967743f100c94140 : SUCCESS in 14m 47s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b6860ce385ee43698b868374379da0d4 : SUCCESS in 13m 35s","accounts_in_message":[],"_revision_number":1},{"id":"ad9d6e040a14619c52e782c5cd7c0c45a73648a0","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-06-26 17:08:04.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"589737235c44d52a202e60d6931da8d3b9738af7","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-06-26 17:36:19.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"5ef6fabe9b2493f8c30d74b8b5f76c82ae5214a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-26 18:50:07.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0d90f8f5d0384e4caff6093bc1dbe257\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/904363c7944647c28d8cb42298f83428 : SUCCESS in 9m 48s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/3c290b1b65c54bf29a0ce34afa87047b : FAILURE in 4m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/198af4f19c9a4f6a95dc61399dcebf86 : SUCCESS in 6m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b8fe0b69abc94dd09368c14bffa045e3 : SUCCESS in 5m 57s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/50f57e9e73c746278e2886ce4c5d8d82 : SUCCESS in 4m 50s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/50a07a48fee44c24a612ff0a83d53782 : SUCCESS in 4m 55s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b4f17e734f9f4ab3822aa90da0425d13 : SUCCESS in 4m 44s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3b9ee7e392fd44ababc1de846ebcebe6 : SUCCESS in 5m 14s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ee9413e2abdb4fdca6f443be5bc066ec : FAILURE in 5m 50s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/9cbb4c6517dd4a7e9e4390f4e5b255a0 : SUCCESS in 11m 25s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/bc6b0837648240bcb55290912b731892 : SUCCESS in 13m 10s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/9e352acb960c480e9eed8aff89f9da1b : SUCCESS in 12m 43s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4d1245444e2a41cd9af75feb6119463f : SUCCESS in 17m 05s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/4ded146172214b4e9af9f05fa2f2cd27 : SUCCESS in 16m 51s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/96b3771af3204f529250dc58771f062f : SUCCESS in 15m 14s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cc26fd71bc0c43739102aa4fb176a125 : SUCCESS in 19m 54s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/73880d4ffbef4ef1ab1e10d03792eb72 : SUCCESS in 21m 18s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7dba23249ffc4e54b3c8aec98387911d : SUCCESS in 10m 55s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/5a87d0977edd4610af24e40ecde22439 : SUCCESS in 59m 47s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/40e700ac2a994301b29d1f4404999902 : SUCCESS in 1h 00m 34s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/1462653c8e254f7c9d92828ded433cf3 : SUCCESS in 1h 01m 42s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/96aba7dfef9542419e9270fefe80af4c : SUCCESS in 1h 03m 52s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/0aedd59046f64d289ec2df41c8124051 : SUCCESS in 1h 02m 32s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/679e9522d10b4643bbf10f3c0a8d258e : FAILURE in 1h 06m 24s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4f5a4b0678be410fbdbfebd811660bc6 : FAILURE in 1h 08m 47s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/b3d19d2a24c94b68aae83852911a12fc : SUCCESS in 17m 38s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/7a17be4985d54dbdb35c29231780086e : SUCCESS in 29m 56s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/71139dc39a1e48d492fa7f7517d4b5ff : SUCCESS in 20m 30s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/80c699d9d99541be8262cff68bf215bf : SUCCESS in 21m 08s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/6fc3e8412e8749f08b2b3d142f9e50bd : SUCCESS in 22m 20s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ef6c35aaa45744ad880e4459a125edf2 : SUCCESS in 33m 19s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/179177d797ea4bffb06f8ff66cd78196 : SUCCESS in 25m 36s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/857f14effbde4aa2bb5770eb5800991f : SUCCESS in 19m 08s","accounts_in_message":[],"_revision_number":3},{"id":"da155ed3a495b48a34177b4cc6c02a655bd6465f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-06-29 10:42:09.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"bb9564e01268fb34ee59713a5479befa4d9b62e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-29 11:49:47.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/358e8353f9564e788efa841c1bf3cf1c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6cf7275f5fb146baa24bd2d4ef5051bd : SUCCESS in 9m 46s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/46b582a553bb40e3b83fb56d08b90340 : SUCCESS in 6m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/40bafaa9b2ea432088bab09ff4b3406b : SUCCESS in 4m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d74c303eb92c4ed1b0eef9a1a305aa7a : SUCCESS in 4m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9a504f499d4c4ab8876b4ad582a8d66c : SUCCESS in 3m 58s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/2897c4da68554e6aa8d192854fc9aa2f : SUCCESS in 6m 38s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b14ad5310ef549f383f4531c664241bc : SUCCESS in 6m 06s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7e8c551e13994f23abf944149ceb5c6a : SUCCESS in 5m 06s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8fcb2c52c66d4f9da37f6f7f01023525 : FAILURE in 5m 20s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/8c96f811d548452eb42635b70fc94637 : SUCCESS in 11m 36s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/81e35ea0582c439aa6c7796cce8c42ab : SUCCESS in 15m 45s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/42f56af5068945918aa8dd61a3d8a822 : SUCCESS in 10m 13s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8f276fe82ec1486ebe92c342934a551f : SUCCESS in 16m 17s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/fcbec018af0b446ab23031f7e66034f3 : SUCCESS in 9m 56s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/84c4fef5ca8b470f8b1845c8fee067ab : SUCCESS in 18m 29s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a613e94b618043fd98d58bbd9517710b : SUCCESS in 19m 17s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/8657112418554140b6a82a6aa120aae8 : SUCCESS in 17m 41s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bc73d33b36824543b7b3a285a770d9c4 : SUCCESS in 20m 13s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/f030e3a27ee64f288bfd4c783c114721 : SUCCESS in 57m 03s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/5123d7ef4b7b494fbbbcf4f50ca7dd12 : SUCCESS in 57m 44s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/7c51fba68ce344259f5f2191934ec13b : SUCCESS in 1h 00m 24s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/887bdb9d55cc4c49af195f9386a5e465 : SUCCESS in 1h 01m 50s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/42bb4d6f626c4847b5e021eef20dda04 : SUCCESS in 1h 02m 02s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/13a3c2f52ae5472b83d7559200f5e017 : FAILURE in 1h 06m 07s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/617f667c79034abb8b5b4897b52a73e0 : FAILURE in 1h 06m 02s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/d9f82eb54711448296a237e821dfd81d : SUCCESS in 18m 49s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/1b4b64a345fc4419bbab2c610e8dc342 : SUCCESS in 14m 53s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c211d76077c4444798608693222a7fa2 : SUCCESS in 26m 35s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/3ad2b727deb64c0cbaf0d9480d6b3fd0 : SUCCESS in 13m 30s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/055318b34b104ffd8f8125e1c52fae82 : SUCCESS in 20m 38s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6d812e19f3804e0bbe56aa9bfb8bea67 : SUCCESS in 30m 03s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/17e8769873804812b7d47694a126c3ea : SUCCESS in 18m 24s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0503e46a07fb40ce8f748b905289056b : SUCCESS in 18m 48s","accounts_in_message":[],"_revision_number":4},{"id":"f78f631c62ee74c070cf7638242489a67db70ddb","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-06-29 11:56:17.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"6da962caa714f2a521f0a8bcdcdd5c8397c8bb29","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-29 13:07:55.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5ae3be9148624e0f923853f793cfc89d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49f8fcab845d4f5096cc02fac7ecf33e : SUCCESS in 5m 33s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/23bb3b4aebd740c2ab58f750b6766a17 : SUCCESS in 6m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3c3f8c094b1f4f4e8b11fe6e3c9de5a6 : SUCCESS in 5m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/832832b49448479abcaa502c3cef9dc5 : SUCCESS in 5m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a30046a3532d48cbaeb10a10569a68ce : SUCCESS in 4m 09s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/fdf020cc18cf4a5cb5cb5221ee6ea6f5 : SUCCESS in 7m 00s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8f62dd4ecbd64b398e9650a511467b83 : SUCCESS in 6m 48s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/69b3537f169745cd87e5d4d39ba584e8 : SUCCESS in 4m 58s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1b43b383eb464d009319c03074e951f1 : FAILURE in 5m 50s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e7d2631103894144a354b3111cdd5056 : SUCCESS in 12m 15s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/89f5e813c61a4601bc95274c50b7d216 : SUCCESS in 12m 01s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/1e7697274abe452c850975ee5b6f0bd2 : SUCCESS in 15m 37s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a6dbf9f891a443a296b5c7ee202c5885 : SUCCESS in 15m 03s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/8263dd767e3d4296982a661da3e7ba87 : SUCCESS in 17m 29s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/dca2e6e5be644b7ab01af22b3d4beb9c : SUCCESS in 17m 58s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/af9ac61a772f4d9fa87306d6f590c943 : SUCCESS in 18m 47s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/a38987f773aa4992889261a19f123e7b : SUCCESS in 21m 54s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e422425dbc4c4ee7bcec2525f0f80a2c : SUCCESS in 10m 40s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/9e344abbf4704218a3d1f233b3e02c7e : SUCCESS in 58m 43s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/da7d1ce5379342ac9d4352ee63bdaaff : SUCCESS in 59m 44s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/af8473cde3724b288c005f890d0cc113 : SUCCESS in 1h 00m 38s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/36e46405eb53420cb50fbb5c7f43ecfc : SUCCESS in 1h 03m 17s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/e746fd9a96784f15b3f21acc7b88364d : SUCCESS in 1h 02m 29s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/6e5e34a032464620a9c983688b2a7cfc : FAILURE in 1h 05m 28s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/57d21634a265496fbceae68a7f057518 : FAILURE in 1h 06m 34s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/63868d72f9444ba49c062036efa9c4a8 : SUCCESS in 20m 10s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/e8abd3c22ab3455a914e9b528b895d2f : SUCCESS in 19m 27s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ea50b21547da4432b3d3dbf293a451ff : SUCCESS in 13m 38s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/b9d3024370834fe49b939e70b2ae9975 : SUCCESS in 18m 22s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/1746640202b64116ac500b4bfe9422c9 : SUCCESS in 36m 38s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7bc2fa5de2e44177926f1fce3c3fe63a : SUCCESS in 20m 34s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/e59d2964586b4213a3f38f016f76e415 : SUCCESS in 14m 30s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/34969df79fb54294aeaeaf5db23c664f : SUCCESS in 19m 36s","accounts_in_message":[],"_revision_number":5},{"id":"040b8a7ec939b8a9f5df1405266bce80772250dc","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-06-29 15:20:28.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":6},{"id":"898b981a02c965fdb74c231752ffc5388b0106f4","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-06-29 15:36:33.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"cd5f0829e7ba3db60452856516a454e494e2abd7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-29 16:41:39.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/afb8a9b1cb384a1eb35e9d192b174551\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d329fde880c6454892dff7a314b4f45e : SUCCESS in 5m 07s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/46c14a708adb40b4be388d3a8203d449 : SUCCESS in 3m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/327ca0b92a994fc787487a5dcc1a4381 : SUCCESS in 4m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/92d589d7edec480595aaefc17724e7a3 : SUCCESS in 3m 16s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/479f9ba36a7f4421a4781bee92a14e29 : SUCCESS in 3m 42s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/15d70ec997bb4272b95209dab57c54c2 : SUCCESS in 8m 16s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5f98a695768849299a21cfd59a205894 : SUCCESS in 3m 04s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/562f3f009e534d858476a604612f28c4 : SUCCESS in 5m 17s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/289c8bf6da6e4dff9eb3573995d31434 : SUCCESS in 12m 10s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b784bda503484789ae7705539567edbb : SUCCESS in 11m 21s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/e2bc08fa4471400ba61b28577535959f : SUCCESS in 11m 14s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/b2ca1c614374414483ac5173d0488eb0 : SUCCESS in 11m 32s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e139f8e7463746aabf615db53176aadf : SUCCESS in 15m 11s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/2bb93f126b254a529bbaf02e1ef72696 : SUCCESS in 16m 31s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/ab3ee072c4b84db08e4e94f12a1779d1 : SUCCESS in 24m 50s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/37307a23e4124541bebe2bf6481b16b2 : SUCCESS in 11m 58s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/a14d59a2b5d347319a7331a7730f71ee : SUCCESS in 18m 04s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/09140cfbfbfc452fad0d15f882ed366a : SUCCESS in 10m 25s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/016df0d066a44d4bbfc5e74947bda804 : SUCCESS in 59m 40s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/ae3063ea7e8f40c8b3f45c8512da8815 : SUCCESS in 58m 38s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/75b5ab4723e749b18178bbc7d5f068b3 : SUCCESS in 59m 14s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2bd68b718fdc4895863cc7a01bc99236 : SUCCESS in 56m 04s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/9c4cc0ccdb9241aca6557d2a674edecf : SUCCESS in 58m 13s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/3461ee60b40a4dd6ad6b4b519034495a : FAILURE in 1h 02m 43s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/78e6efa07cb74cb99f0e77309bafe0a7 : FAILURE in 1h 04m 01s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/cb1013b173804354919cf00d8635a2f4 : SUCCESS in 14m 16s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/7c268e2e7bda4420a04609d42989f936 : SUCCESS in 18m 13s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7a9d886a46914496a3bada2f5d006105 : SUCCESS in 22m 56s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/3984bfd14ea94931a3eaae4a22f53312 : SUCCESS in 21m 03s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/899e002bb83b4fea993aca07b01776b9 : SUCCESS in 20m 10s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b708e377e64345fe908c8742c3e051ad : SUCCESS in 35m 26s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/e0f8246c10184d8ca7b09f586df3eb78 : SUCCESS in 15m 51s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a063b4bb521a4049b73d2e71bf450185 : SUCCESS in 19m 42s","accounts_in_message":[],"_revision_number":7},{"id":"feb428c76fc198bcd2c8408fd4ac22e4b019f1bc","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-06-29 18:17:10.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":8},{"id":"0b403b08d859575a813a133a164d66de864c6e9d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-06-29 18:29:37.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"f00487ef84a9b7dfa58a81a56e0bc355fc4716e1","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-06-29 18:38:07.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"36c45ac295f1a7cfe8b5345365456a2335f79f9a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-29 20:25:02.000000000","message":"Patch Set 10: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d49c062a92ae432595d8b35655ac4db2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f62d2e93830342f8a08b89bb6aa75355 : SUCCESS in 5m 48s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/1974d796783c4887a545514fc00b4b06 : SUCCESS in 6m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/207ad3190a7049e0896bef0001e30db2 : SUCCESS in 4m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe24c86c1d5a4841b947e898ec3a5942 : FAILURE in 3m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/527d47833fdd4f4cbd11830e8378a837 : SUCCESS in 2m 53s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ccafc98463f84d70a7d2722f6eee16eb : SUCCESS in 4m 22s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/91427d0ca1294c0a8d9b55eae1d35bff : SUCCESS in 4m 04s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6241505c63664743ac23e9a6dfcb58e5 : SUCCESS in 3m 39s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/863e0110c28040e48f2fec27e537057b : SUCCESS in 6m 33s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a59571dbc7f245acaffe9493ae6fa192 : SUCCESS in 6m 46s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/e6ff4e5dd0994a8cbef92ea8e10b093d : SUCCESS in 24m 31s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/f4eadb7ff4084f3e98087b68b47d19d8 : SUCCESS in 12m 06s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/acb4831ee9e2402382275870f0c57367 : SUCCESS in 11m 01s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/ee46a205c46241e59a5119c02e9edbf3 : SUCCESS in 13m 38s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/d0f87553bc4a4ca6b6b0be940e3d21fc : SUCCESS in 24m 47s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ed637e5b8d6e4aaeb17e68d5f613cbb8 : SUCCESS in 15m 06s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/5087dd915623480e8ba6458cb86fc832 : SUCCESS in 17m 22s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fa5608fdc27b4576b4a84024e13416cf : SUCCESS in 18m 26s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/ad744f9593aa46b0bee36357a24ce57d : SUCCESS in 1h 02m 27s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/d5af9acceb11444dae2e797d5c2b68b1 : SUCCESS in 58m 05s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/7b0c736ed1a74df79f88fceb3034073e : SUCCESS in 59m 54s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/92495a5fb5b347a08c46cb82f0149559 : SUCCESS in 1h 03m 04s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/f3c813f7ee1a41e2a938b6c8f1e5200c : SUCCESS in 1h 02m 30s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/95d3c30be9a84991b6b99393c7393ead : SUCCESS in 1h 42m 04s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0f68b5a760594892900961a124978d22 : SUCCESS in 1h 45m 27s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/1b8ca1a61df449d5888fbcf68eb12498 : SUCCESS in 22m 33s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/8fd2bfa19f4a45379c90cff18017aa4d : SUCCESS in 14m 36s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ee97aa86241642cd9cbb7ec4a212fa63 : SUCCESS in 23m 23s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/c2f6ae1645924b50a68d269ae8e4b96f : SUCCESS in 13m 34s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/d53d7d2215734a9db6b796b90bb991bd : SUCCESS in 21m 51s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/091fe6f254684f18b0f7a1471cbb3dd3 : SUCCESS in 34m 37s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/42dcd040b1724e0698883fb845c04ec9 : SUCCESS in 17m 13s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/61739050ee0b4a55ad990284ea24da8e : SUCCESS in 17m 26s","accounts_in_message":[],"_revision_number":10},{"id":"f685e89aaea333fcead5beec6ec215f7c21a1b7d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-06-30 08:27:30.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":11},{"id":"66eb249f1b744ddef41d5191d462823ff66c456d","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-06-30 08:31:10.000000000","message":"Patch Set 11:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":11},{"id":"e5e6f646681937200d14e9f27c2bba68602ab72a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-30 10:28:42.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1d005454f15144d1b7f288c239f2aeac\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9bdcf52e982243b89ce22e2b41c0ee10 : SUCCESS in 9m 09s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/db694dcd4e9c430c85ce59e58b57e3ec : SUCCESS in 6m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5923169bc0e34b5aba4a424b42c5b370 : SUCCESS in 4m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/26a7893b013a4764b6768336f6fdc571 : SUCCESS in 5m 09s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c596d55800cb4112ab2c96b7fed97ac4 : SUCCESS in 5m 31s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/507357937597403290d92c4126b614f7 : SUCCESS in 8m 02s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e5853487dd614626971bfe2ef1a54b98 : SUCCESS in 6m 53s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/01fa6f0f8ea6405b899c4837204a7fab : SUCCESS in 4m 56s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9fcb96cc79fa456bacd5ee8fb2726b12 : SUCCESS in 11m 19s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/9eab206422e248c394395fcd03694c4b : SUCCESS in 12m 39s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/3de78c935d5c4d84b5be3fa4850012c4 : SUCCESS in 11m 32s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/83a1ba0441184db3a6fd341f2cb7ca86 : SUCCESS in 11m 14s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3642f0f7bff94e22b62d43e6b804258b : SUCCESS in 15m 35s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/0fd06074647c4bc2b79dd109eb885a50 : SUCCESS in 17m 13s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/83b7e179d069415882d3f6bafcee8eca : SUCCESS in 18m 54s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/06e89eda14af428aa57cbe66c0384dbe : SUCCESS in 20m 47s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/f2ea3b54ab0a47119e09960604c3cf93 : SUCCESS in 12m 04s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/12c1c6bdbf2c4b088b7d4108479c75d4 : SUCCESS in 20m 35s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/d2bc3fb32f5b4930ad8503b7d1f57540 : SUCCESS in 57m 16s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/bf43d98b767e46d4bf6dd3625b3d6145 : SUCCESS in 1h 02m 14s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/37ba122e78b94809b54493e024e22853 : SUCCESS in 1h 02m 56s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/94562392ef7a4160929910935d154a05 : SUCCESS in 1h 04m 36s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/7d1dde2d7e04471aad81befcaa29aa3f : SUCCESS in 1h 04m 11s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/d918b5d120c346f4a524f01e28c680e9 : SUCCESS in 1h 46m 14s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/aa978e8340ad4fec999ddb37b1f39e5c : SUCCESS in 1h 58m 50s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/45df56a2afce4b81a38802ab85ba1d17 : SUCCESS in 14m 37s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/c42408b05e114aefbb78fbdad6ff84b4 : SUCCESS in 15m 51s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/70ef90243b3c4a7f85a8cce4742b2cf2 : SUCCESS in 21m 06s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/6b1bcd439c444525a55b28f49cf56b83 : SUCCESS in 15m 36s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/af6e116d2a9e478e80ede2b498c9165f : SUCCESS in 20m 52s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/799b90a7d74647a7bf15e56703c8f01b : SUCCESS in 33m 38s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/6f83c8ddcea8491fbc1c0cba8f6e7999 : SUCCESS in 14m 25s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9af5ee425ce440a3b505c2ee2e1f8250 : SUCCESS in 13m 53s","accounts_in_message":[],"_revision_number":11},{"id":"9800f46a70a9648d02ca49a0493e21995a3592b2","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-06-30 10:48:17.000000000","message":"Topic stackhpc/2026.1 removed","accounts_in_message":[],"_revision_number":11},{"id":"213feab2f6f15b30eaebe16b8bd2f04b7971bdd1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-06-30 10:56:51.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":12},{"id":"3727ab3c84a98bc7f541e0f24b9ed3a23b74a8eb","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-06-30 12:31:47.000000000","message":"Patch Set 12: Review-Priority+1","accounts_in_message":[],"_revision_number":12},{"id":"eb4741da31b539607216b1ff53f8c63bb138c388","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-30 13:00:18.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b943aed1d6864ebcbe375e79d8b120a1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1bee704bbdf34d758e35cee0372b2ede : SUCCESS in 9m 00s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/951b852d0ecd44b6b6f32e8f0c4ce116 : SUCCESS in 6m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/518188c3c10841a085eb2362d4733023 : SUCCESS in 4m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3847f1e7ccdb48d584c26db729c997e9 : SUCCESS in 3m 54s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9c6260ba4f384c89b6788e93eccc2302 : SUCCESS in 4m 07s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/87f8bb3fe3144356947a1d88042a6787 : SUCCESS in 4m 56s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c71ce951c4c443829c6f1f2973c71521 : SUCCESS in 6m 04s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/606a22e738c948279467a3fe92ffd022 : SUCCESS in 6m 02s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c981d6ec20af474b871f41906bf05f6b : SUCCESS in 14m 28s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/0cdab16407c44eb1b5e6ca28083520aa : SUCCESS in 12m 00s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/eb0ca13886d74f04a3e180c4d5fc6801 : SUCCESS in 14m 43s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/81ed615ef39b46bf87e12ff5b244d0bf : SUCCESS in 14m 46s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7f650cf75eaa4e19af2d3405a62258c1 : SUCCESS in 17m 56s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/5fa8fb76a2e44a0b8c597dc5a76d470c : SUCCESS in 14m 25s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/b3e04236e5bc413e9afe22656fba973a : SUCCESS in 21m 13s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ad6a43143a9e43b8aa8ff52a91e5f372 : SUCCESS in 20m 59s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/28463e6ad3bb4d629319cfc716fe37ac : SUCCESS in 19m 31s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ec5a4e4323c64099ae0f8f81d760b700 : SUCCESS in 13m 25s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/42decbdeba5648849b6847ef6306dc1d : SUCCESS in 1h 09m 15s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/f6b7480ae92146c4a47f318bafa62fe8 : SUCCESS in 1h 08m 53s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/7ba5afabfa6b46c1a55cdb9e2cb8271c : SUCCESS in 1h 10m 58s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/48ec5f25d2974bdf9d2d2d669922e1de : SUCCESS in 1h 06m 38s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/23b4335738c14f49a5be379b2af0ea73 : SUCCESS in 1h 08m 50s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/78ae44dcf1884cc482de2e100e8c4d53 : SUCCESS in 2h 02m 15s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3782a1bef35847ecbea8a8992ff2b989 : FAILURE in 2h 02m 14s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/36a542e88e1b4eab80122397dd407e24 : SUCCESS in 23m 34s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/47ca40733cdd47a892ecd8df58102225 : SUCCESS in 27m 07s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/421d7b3d216d4687b6683c126ffb5bf4 : SUCCESS in 16m 22s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/283e7dee75a04993ad64ad113cc5e46b : SUCCESS in 18m 29s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/a24c8cc6e91e42e0a34c6c1496f9adbf : SUCCESS in 41m 58s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8fe8fba2019a4d65aa473dd7902db003 : SUCCESS in 34m 58s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/434674c3339141f29815e66d27cc7c55 : SUCCESS in 27m 14s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0503a137a6544cfba685ec17c029c303 : SUCCESS in 23m 26s","accounts_in_message":[],"_revision_number":12},{"id":"f3ac2a3152130b0aa7983bb73c5baea600f607e7","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-06-30 13:36:05.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"207cc8de0b2a5ba2723a6736605a721d3569ba5b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-30 15:29:07.000000000","message":"Patch Set 12:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/df6f5684f71d44089b5a9d0d46c76584\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c2e53f83c96a44bb8d397208a8c35789 : SUCCESS in 9m 57s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/0aaf45770cff484991854185fcaefd1a : SUCCESS in 4m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/75527a7473134a91b87df0980989421c : SUCCESS in 4m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c339558094841e7864c16ae4908034e : SUCCESS in 3m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/debf768ff3bc47ffa8eeb081f3645f71 : SUCCESS in 2m 47s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/652e3ac7fd9645ccbd7f70c594531b6c : SUCCESS in 4m 33s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e75716efd9cf4f779cb4e8c4ee8ee814 : SUCCESS in 3m 41s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/1a782a4749504a5ca48286cfecfc1419 : SUCCESS in 4m 25s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/205d7828dcbf4e32a6841428db00c5c0 : SUCCESS in 12m 12s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/cd8aa89f06634383aa8cbee3870405c7 : SUCCESS in 13m 59s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/7b9b47550ded47fe94c7ed6a276c2d8e : SUCCESS in 18m 16s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/b18ac096fc1b47099ed9687f365174e8 : SUCCESS in 22m 57s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a47806fe485f4f25948b65517d06ebb3 : SUCCESS in 18m 24s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/1cd2df0935fb4794b57ecc87d2078e6c : SUCCESS in 16m 09s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/37ea15f653d347009a60e26dd30e880b : SUCCESS in 14m 06s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bdf41c458bc7427da9b262f0c6a5f548 : SUCCESS in 22m 26s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/7cea5a9cb9ae44f58953fb6e5047b1b7 : SUCCESS in 24m 21s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9c1bc7e3a5564fe98b9fbb5ca8648289 : SUCCESS in 10m 59s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/7f23ee3f755944299a965a90db41867b : SUCCESS in 1h 04m 06s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/70a1cb47ce09434b89d6c82cb13c3c3d : SUCCESS in 1h 01m 51s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/5254831fd9824149a9d9a68dd63ab367 : SUCCESS in 1h 02m 07s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/97e78549cd724eb2a13044de72bdfc99 : SUCCESS in 1h 04m 00s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/ba364a2d67624857887721910d3770cc : FAILURE in 1h 31m 42s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/816dec07207d430ebf34f01db4b832b6 : SUCCESS in 1h 46m 16s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/181c4991b1a6472b889534bd1227c01f : SUCCESS in 1h 48m 53s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/fa77e949c1ec442c8ec7b932795a7946 : SUCCESS in 22m 43s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/a34b45c3d27341feb85802e53be0ed82 : SUCCESS in 16m 54s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c79e44a2595a44e0b018cc33200f6948 : SUCCESS in 13m 13s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/eb4c638f232645f88186a02dd1132135 : SUCCESS in 13m 18s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/3abf812a26ac4737aa2c2001eabbdde3 : SUCCESS in 41m 10s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/92a23b7c718548e694fbac8ccae9697f : SUCCESS in 20m 42s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/7eb0994bb9164385909af34cb273911b : SUCCESS in 21m 24s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ce4270cb9c734db78ca3fb2f4ec47b13 : SUCCESS in 20m 43s","accounts_in_message":[],"_revision_number":12},{"id":"aaf5d64635166c525735c8b1bae2ad7aad9a3b9e","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-07-02 14:10:57.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"1c07418bcdc66cc7c0848ab5207d1c5334ae29f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-02 16:13:25.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/50a6deb8e5174eaba7ef11d2975f9cf4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d435105046d8406db657869f0b63907c : SUCCESS in 8m 30s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/bc57c772989b45b6abed0e52cd446920 : SUCCESS in 5m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0e3f350bd7ef4db695a7c5d7b2036fa3 : SUCCESS in 5m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/916f6d649e5e4670b371f41a1a74f04b : SUCCESS in 2m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ebc6367398664eb098409b97f8608572 : SUCCESS in 4m 25s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/d47cfed5d42a4e2ca42aeb09c13efca2 : SUCCESS in 7m 25s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9f57b40f76844d1796491a0acc909546 : SUCCESS in 6m 15s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/52c2bb3760db4f36bdc467aeef7a8bc5 : SUCCESS in 4m 58s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/77cc2927e9f949be873d57826a87d85f : SUCCESS in 11m 40s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b661418c64a2498c8aa2dffc4c0ab322 : SUCCESS in 9m 24s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/30d79169adf642c296a97ccf6d0561af : SUCCESS in 11m 51s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/78b0dddbafad46969893be7283c97b74 : SUCCESS in 13m 51s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9367c317f94b4f7d9721b90a2938724e : SUCCESS in 11m 01s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/d3e18ecc2550467e91e5f97c29de5c7f : SUCCESS in 16m 03s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/1bb45102df044fa7bc2ca910c6ed8ac8 : SUCCESS in 20m 01s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4290cfe21ace4eee9871ae51b4110b48 : SUCCESS in 13m 03s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/4ebe7ee716be4db5b58a122cbc30b6b8 : SUCCESS in 12m 41s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5bdadc5889f54b319d955bfc03e64cc5 : SUCCESS in 10m 47s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/821421e981c94000920e4881b55e58c6 : SUCCESS in 59m 29s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/3401cdf4bf0e45e783ecffb77ba6bfc4 : SUCCESS in 1h 02m 16s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/81b2f94b244345fc8b867b6d2fe9f893 : SUCCESS in 1h 00m 59s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/52d27a802df74d8db60d76dc153ba5e8 : SUCCESS in 1h 05m 40s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/48f5f39f0b6e4b1e847a76632ce7c5ef : SUCCESS in 1h 07m 00s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/6e44bdd06e4f44ca9d64b061c3f07331 : SUCCESS in 1h 47m 22s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4576238126bf47869810b11ac441459d : SUCCESS in 1h 49m 58s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/04da8bab20b6498cb54146ae09f80ea0 : SUCCESS in 15m 01s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/d0cee2200f704b62992de6e26e0d22f5 : SUCCESS in 14m 43s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/41bd5b56d80b47d3af886002e7c48741 : SUCCESS in 15m 45s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/6e37dc226b3e424d96cdd2b3c568c7bb : SUCCESS in 23m 00s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/1b971ea2e9aa4cd4bba49ad795c97ce2 : SUCCESS in 21m 28s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1e8bc80e763c416787fe8847ce05625b : SUCCESS in 33m 22s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/0333314eb1834dc59344c5a17a51691d : SUCCESS in 15m 16s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c1c7415f0f0d4b23bb7cdd97f7cc5a5b : SUCCESS in 17m 44s","accounts_in_message":[],"_revision_number":12},{"id":"07d8694c14875af54951fb66773da493a22bb589","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-07-07 12:32:21.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"13038c7e5385be8a01fee3d64d908dfc18099848","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-07-07 12:32:52.000000000","message":"Patch Set 12: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"704508f10d07a5c909ca75b3fdc55a0ce3540b05","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-07-09 15:48:16.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"0c19859c0e83918249f7c1b42201a0f40fa3e8d3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-07-09 15:48:25.000000000","message":"Patch Set 13: Published edit on patch set 12.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":13},{"id":"540a895f04120dbb00a639d72a9d241e0bfabcbd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-09 17:46:59.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8a094656913044309c5f9ec17222d097\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/394ff1defe9e4145a69cc589d6b638e8 : SUCCESS in 9m 52s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/9434d773fefe4d66ae781ce6895399a8 : SUCCESS in 7m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2aa47825277640d3aca44c19e315995a : SUCCESS in 5m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/63ffe6eaef2a4b6ea5199863135c6a7f : SUCCESS in 4m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/001c3729a3c2411aa5e2320c639085be : SUCCESS in 4m 33s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/325f1b9cc5dd49bdade00b47d1bc9606 : SUCCESS in 6m 53s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/60a048925b814797a7147a0c847d70e9 : SUCCESS in 6m 41s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/303d2e984a2b4f7b90cf78a3455a67ba : SUCCESS in 5m 03s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/bbd55dd9f8ac412ab4b5677fcffd52d9 : SUCCESS in 15m 02s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c48c25279fe3477489c822b5c82e12cb : SUCCESS in 12m 26s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/9aa266e8c1a843beaca88a9b6cfebee7 : SUCCESS in 17m 20s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/2854389228d14fcc834f38240b439422 : SUCCESS in 25m 29s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9f78b03cfcb9466b8cbae50fd23b8ebe : SUCCESS in 14m 12s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/c095f9ea33714f0eaba449370046ccea : SUCCESS in 16m 13s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/dcac79c631eb41f6a1e55e737e7c096c : SUCCESS in 22m 10s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/50b8231e9fd44a49b288c604eaf72ea7 : SUCCESS in 19m 08s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/955aea14785f446f9c5445845c894ed5 : SUCCESS in 18m 55s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e95eab842faa4371b4743e937d28129a : SUCCESS in 19m 06s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/cb86b3c96ff042c6a8fefbe23266ccb3 : SUCCESS in 1h 00m 26s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/1980b62a69be45f399c98f8db8fdf4fa : SUCCESS in 1h 01m 21s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/0aa3e4c30d684941ab19b1751f96ffb4 : SUCCESS in 1h 12m 21s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/517856c7a23141ff8c3db9cd99f9dbba : FAILURE in 1h 29m 27s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/9460d1c484f84177a71cbbaef0f7a993 : SUCCESS in 1h 07m 21s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/76498ceebda04971947c6c5aaf8e1c33 : SUCCESS in 1h 57m 17s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0aacc12d03184d0e9afd66c2ec0d3b88 : FAILURE in 16m 31s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/b3c775c9e06349b18170f71438402dff : SUCCESS in 28m 15s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/aeaef676486a4385ad4fe53483efc9a4 : SUCCESS in 21m 32s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fc616436c8cc45f58f9122d9bb561ad2 : SUCCESS in 22m 20s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/531990f20505490994675f692148d929 : SUCCESS in 29m 33s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/cd363d23d74c474686d2fc4ea9cb63a2 : SUCCESS in 35m 13s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ac83ce1bc54d4106b0817df7b7487947 : FAILURE in 26m 13s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/880eeae80ddd4af0a79fbf375ee9b20e : SUCCESS in 27m 31s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ff714ae3000f4cf3a96497f324b86904 : SUCCESS in 20m 22s","accounts_in_message":[],"_revision_number":13},{"id":"c99bdba35004fd67027f671fdd22a0ae3ae78f8e","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-07-10 09:13:00.000000000","message":"Patch Set 13: Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"873b6e4fd63228b922b3f527e6284fb2aadcfaa7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-10 11:14:42.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/af217fc5da47415d8fc65428aee69e6e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b6745f1a6b847c59f10733aa90eb5cc : SUCCESS in 5m 10s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/4d5a488564a44e0e881ce14609ea5370 : SUCCESS in 9m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/23745988578149df8adc0e2e71dff4bb : SUCCESS in 4m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e84911cd82fc440c90742391af406e20 : SUCCESS in 4m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/67a963ae0749497eaae1eef4c1ad8131 : SUCCESS in 6m 02s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a98bfcc447c74cc7bb2a981a24d28c06 : SUCCESS in 7m 12s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6635dae2394b4abe9c36ec2b5fcea430 : SUCCESS in 6m 44s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/addcb8d2f7f74086bb7b0896c621a481 : SUCCESS in 10m 18s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c449f127196b4cae8a63bba186f160e7 : SUCCESS in 7m 21s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c429810d3faa44fbadc15f30460c3edf : SUCCESS in 12m 48s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/d53f4d8d8af8402896938c1650264076 : SUCCESS in 12m 03s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/015879c55b5e4dbeb22f17d03cd77ab3 : SUCCESS in 13m 44s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bc3f761c94d44703893e701c85ec4521 : SUCCESS in 18m 01s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/16c032b08348483e8371dbfd6ed378e3 : SUCCESS in 16m 48s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/538dd75752f24658927b059d0c97bee4 : SUCCESS in 22m 19s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9be2d9176a244250ad8955359e4217cb : SUCCESS in 20m 42s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/662b71d5da4345fd9e3a721473365333 : SUCCESS in 25m 56s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a325e7964fb14c44a9d6a8d575d2e8b0 : SUCCESS in 18m 38s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/2bffcac4ab6a48ffae545870564f988f : SUCCESS in 1h 03m 50s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/abd47ae2454e46409f1698cd0b1eb733 : SUCCESS in 1h 04m 03s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/385e778ce58341c488be68e24276f10d : SUCCESS in 1h 06m 43s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bb8d7a41a817452fbe38970a84498c8a : SUCCESS in 1h 13m 10s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/4db9159f090043d8bd300e5f7bd7cd84 : SUCCESS in 1h 09m 27s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/f24edf9d44634c249c60b5f5abcab5ed : SUCCESS in 1h 44m 52s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/93565bb9edaf4da88704f42c31b2b641 : SUCCESS in 1h 57m 43s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/2014623586ed4419bb66292c958f2747 : SUCCESS in 17m 11s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/66494f28504a4739aa2981d039e5943f : SUCCESS in 15m 17s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cab51a7efd434b4fa9ce4163cc97c417 : SUCCESS in 25m 23s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/9d1fe8280f6848a9ad828414af03405d : SUCCESS in 22m 42s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/87cc70cf879342cda536f6250c2665dc : SUCCESS in 27m 32s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2a413830ea56441695867e1556dc0549 : SUCCESS in 37m 27s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/ee88ecd23d914f5e8538c7093f6b4831 : SUCCESS in 15m 20s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4aeccbd4f4ab49ca98ea6ee24b267738 : SUCCESS in 21m 47s","accounts_in_message":[],"_revision_number":13},{"id":"60bbce2ffb7dec7ae6e7ddb5200e6b4e6290eb20","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2026-07-10 13:17:25.000000000","message":"Patch Set 13: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":13},{"id":"b0ff0a50d71ed6359cfd87bca88ea4bc1708d269","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-07-13 09:21:56.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"eb815cf3632f4e202e10f8383340edb3bed9cddd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-07-13 14:48:47.000000000","message":"Patch Set 14: Published edit on patch set 13.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":14},{"id":"cfebe0d79b52e70b8a5ff04112d1836b2e53f275","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-13 16:29:29.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a614a848f3204840b4730395c3dec52c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9112db6a3fe4522909a08b00ae05b51 : SUCCESS in 8m 31s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/f5cf7db93f314f4298b166883efc749f : SUCCESS in 6m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/78ac138e999548dc9574d5f65185dabf : SUCCESS in 3m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/81551efc08a344468c34fa4871dfce22 : SUCCESS in 4m 21s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/22ae50931cd143dfbb69c65e28a442e1 : SUCCESS in 4m 23s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/95c7064ebf944b27bf36f5a979446ef3 : SUCCESS in 4m 19s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/704b4fc8b41d45ef808db822dc426493 : SUCCESS in 3m 39s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c30159af7f2941af819305ca10c03564 : SUCCESS in 8m 28s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/80177629ae724d618d0e4d72c04b9200 : SUCCESS in 13m 00s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/0f20572e38ae4160b481825bf681fe6c : SUCCESS in 13m 32s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/910b69634e3d49b0a52308f5cbb0ea47 : SUCCESS in 17m 21s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/c7dcb34f882a4066a05fd80cf40ece8b : SUCCESS in 20m 38s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b268ae2258a04cb3b7c38d8d9ddbf973 : SUCCESS in 13m 32s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/2d2f2ecad32c491188fc638ae0e37e6b : SUCCESS in 19m 56s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/6ddb205edbe54bbdba6b4dc3c417dbec : SUCCESS in 18m 03s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/958769c6c3294a03b5b5ddea19fe86ca : SUCCESS in 14m 53s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/4dd19a5c75ab41fd832390875ac104e6 : SUCCESS in 11m 47s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/466e804db8ca42609c32633a1703c78e : SUCCESS in 18m 27s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/b082c9dbf502495cb8abcd024a230c75 : FAILURE in 40m 57s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/b9a320f6f4bd40c1b78143b2da29c484 : FAILURE in 45m 06s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/2e1ee6da9e0b4dad896223d7f64f9709 : FAILURE in 47m 29s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f5ba5bc934e74351a0edd4285b556d2d : FAILURE in 40m 37s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/8d92b82fa4574215b1e9773602728699 : FAILURE in 46m 28s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/56d77fc9b73147ba9ede73b307811fff : FAILURE in 1h 39m 34s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b5e621bc45f54e50b8e5932783c493e2 : FAILURE in 1h 34m 27s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/995d93fb7160450db84e1d545900dc32 : SUCCESS in 15m 43s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/a04716c1aed140dd837f0036e8444304 : SUCCESS in 17m 38s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/04c74e062d61425abab0231349fd8a7f : SUCCESS in 16m 17s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/8d1c71d38e5a4341ab7f13d316e26395 : SUCCESS in 20m 27s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/01c35d01de2d4162bf900d63b4ab678e : SUCCESS in 22m 42s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5f7d160442fc4de6a63e094b6b44fc81 : SUCCESS in 28m 09s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/62af29642e2d488f949573bbdfc3cfe6 : SUCCESS in 17m 18s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e323d751eba84116926a9c4d150cac6e : SUCCESS in 15m 22s","accounts_in_message":[],"_revision_number":14},{"id":"1a47130cc5f3ab82bf32d52f5ed8219f97443791","author":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"date":"2026-07-15 12:51:59.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"4d07b034753c0ddb22035f1be27065bd758075d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-15 14:44:44.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/24124e415dfe460c9857b41c85938f1d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/865679ca0a7d413d8cde560fefb0baff : SUCCESS in 9m 20s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/af58f589f1a543bda71b6c927bf2be65 : SUCCESS in 10m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a43c9fba02994b18bc05ab80084be2da : SUCCESS in 4m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/725becb0d527493c8f8d6abd92a696e3 : SUCCESS in 4m 36s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/196a615ffa7b4af9b947a8b4432a11e8 : SUCCESS in 4m 15s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/4ab3f7a0e62e4113b1691ee70b28e0f8 : SUCCESS in 6m 55s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/59d17edbab734713be8de7c9626fadcb : SUCCESS in 3m 04s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f59bd31aaefd4d20a20c373456c6cfd6 : SUCCESS in 8m 39s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b5337153f36b4668bfa2f567d6bc2524 : SUCCESS in 7m 01s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/fa0bf02baee34791bc3677b96302cb16 : SUCCESS in 13m 14s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/7f6e58b3b8c24147acc2b400ad19d5eb : SUCCESS in 11m 38s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/9e1a8e9cc23d4868984db196a9db9012 : SUCCESS in 13m 49s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c098abad5b264726bd1b08d002e9e406 : SUCCESS in 14m 40s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/3071b97e5e9f4434aa3d9b3e2331dcf2 : SUCCESS in 21m 49s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/9e2e1a56e4294ff0b53ada5a6d2ba74e : SUCCESS in 20m 46s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/35f67bfabffc4634806a34f6f5cb52fc : SUCCESS in 14m 39s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/2e82cf062dd349ea9b560d03c33d835a : SUCCESS in 12m 57s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/29a3431b65574ce59ecc09ae9743b51c : SUCCESS in 27m 29s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/03b8393ca3c4400a85c5fc17e8ac9ed9 : SUCCESS in 1h 03m 35s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/bcc6bbbfabda4078b435adeebcaf3fa6 : SUCCESS in 59m 42s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/68c52e7b4f084ce29b1e1d2c1d8b7465 : SUCCESS in 1h 05m 03s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0266fdfcc0e14d4f81a4d1f3ceb45e59 : SUCCESS in 1h 07m 38s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/48684f39e8bc4912bd7ab000aeafc7e5 : SUCCESS in 1h 14m 26s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/8e04999b59414a95aa91a905f3fd2791 : SUCCESS in 1h 42m 21s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c1e0b1fac76946d99435593f1cbc4b72 : SUCCESS in 1h 51m 25s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/4427aa9aa8f2408e93e3afe34954fc84 : SUCCESS in 24m 21s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/90064f5478b141a3b2c9ef33875dceb0 : SUCCESS in 15m 15s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/266594826c5646668faf3ff4bdf76b92 : SUCCESS in 26m 52s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/ea4a7b7c8482402b8457f66bfaba9ff6 : SUCCESS in 17m 23s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/7c1c566751ef4a2f9f4a2b5a882996c6 : SUCCESS in 29m 22s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/671c5603ba0941699e6161aa30876ac7 : SUCCESS in 21m 47s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/9672bc1a737742c2826ae291a4a98ef0 : SUCCESS in 23m 57s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1340e672e3fe4518bd8e853b81f266af : SUCCESS in 26m 49s","accounts_in_message":[],"_revision_number":14},{"id":"4da3576f76f49dcb15ce2a1e2c2adf8c2dda6e59","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-07-16 05:57:53.000000000","message":"Patch Set 14: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":14},{"id":"aecacb6c31339dd6053254159df48f4493b47b42","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2026-07-16 08:09:04.000000000","message":"Patch Set 14: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"82c04e485a446d5fedf7f10198aa3fc1ff5bb028","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-07-16 16:18:45.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"0c0c045d250bd29a2e383947efbf24d290b373ec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-07-17 15:24:28.000000000","message":"Patch Set 15: Commit message was updated.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"**changekind:NO_CODE_CHANGE** OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":15},{"id":"ff7f1015fe6abbfa51c51dc4c0e58163aba45d94","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-07-17 15:25:12.000000000","message":"Patch Set 15: Review-Priority+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":15},{"id":"8981690cb94dc92930bd21179bdb39f08daa54a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-17 17:53:18.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/829c1c4957f04f36bbb1b6655860a836\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b1eefd0a46d44b5a81b2a94939bb744 : SUCCESS in 9m 45s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/bee0b907f184400f8e5ffebc43fe9ee4 : SUCCESS in 7m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f7ea0e8033874a1dbda3af096f1bee09 : SUCCESS in 5m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6711e72ec695412cacba921de992b69e : SUCCESS in 4m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fd920f2f03e04da9b41ae4cb8052cd5e : SUCCESS in 4m 59s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f886b7087fa04a798802a984c642ca5a : SUCCESS in 7m 23s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/476d21583b6746b8abd0a287f6840667 : SUCCESS in 6m 20s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c9bb24689a4d4d6991816f91bbc02bbd : SUCCESS in 5m 42s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c285aab18b304c778b8d1ddba99a0d48 : SUCCESS in 12m 37s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ba5751c6569a475e88b6209c2675e4d4 : SUCCESS in 14m 50s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/925a1555832b4ba685c8b4235aa92f47 : RETRY_LIMIT in 7m 35s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/a8bc2d1882194c0593ae6607eafbdc08 : RETRY_LIMIT in 6m 19s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3708e4eeef5747e8b34fab6607c85da2 : SUCCESS in 19m 01s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/e6dff637a45646b392ed70621dcca6c7 : SUCCESS in 17m 27s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/9e2c717f3ac5428ea19ab1b983f458c1 : RETRY_LIMIT in 4m 49s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/17d1a011ab60428b8efa546b013f0dd0 : SUCCESS in 21m 55s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/81a3a18ebf5649eea63375b514e9fc4e : RETRY_LIMIT in 5m 40s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/501409d790e7411f82a253c53310931c : SUCCESS in 20m 58s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/4b8afa923f36443eb2bfa65a81614f13 : RETRY_LIMIT in 4m 51s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/cfce86c679ce4f1fad374d6f049093ab : RETRY_LIMIT in 4m 35s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/198f43363c984846902b9cfbbcb74cbc : SUCCESS in 55m 28s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e5d9ccdc90844d6bb9e62d5bfaa36924 : SUCCESS in 59m 51s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/bbab87ab147f4ed4bf1734a8412c8ea6 : SUCCESS in 1h 03m 00s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/eea0cdc5efa94fec984c1069e21a4917 : FAILURE in 4m 40s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/97c7dbd902c54cbd80594453bead8c29 : SUCCESS in 1h 16m 00s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/6aedd52e9fda41f696363e3303f1ea9f : SUCCESS in 30m 11s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/27be9de3a37541069b23bd0ec26ea4a2 : SUCCESS in 19m 58s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/07733bb108de44838d27f306a721eb01 : SUCCESS in 23m 20s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/47711dc2717c4fd5888c520262e6b8dc : SUCCESS in 23m 57s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/9c10aba965b64e25928fc693f70e03f7 : SUCCESS in 35m 50s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a943fd02a259442f93f509e127405bae : SUCCESS in 35m 36s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/bfadafd5f8b541c6b009a86d75100c48 : RETRY_LIMIT in 5m 21s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5fd04fe769c94c0494148221ff283c04 : SUCCESS in 21m 40s","accounts_in_message":[],"_revision_number":15},{"id":"a645d5d1abd8c4940f75f0dca50f47491c24d671","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-07-23 13:45:35.000000000","message":"Patch Set 15:\n\n(2 comments)","accounts_in_message":[],"_revision_number":15},{"id":"9bfc145152178513d4132d9104dbc9f0abf03cb3","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2026-07-23 14:28:23.000000000","message":"Patch Set 15: Code-Review+2","accounts_in_message":[],"_revision_number":15},{"id":"b96c387106d2cff329b536b3ac8e56585522de26","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-23 15:44:48.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a89ab41abe5e43af82ffde3e0d2c7c38\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/092081e5a0184cebb78719d02fb5ee22 : SUCCESS in 9m 51s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/2459dcfc85a94036a95bc3d838eff4e8 : SUCCESS in 5m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9399abc2ab914992b4f05527ae0cca7b : SUCCESS in 4m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/80cae65565f64dcc9ee6d4f26a79336e : SUCCESS in 4m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b2533bf4ece84bc0b0d48a527b4d688d : SUCCESS in 4m 01s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/8433528e84824e48ae18f6d5a4d040d7 : SUCCESS in 6m 43s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/236a3fd35c844935998f3d75a6e1e828 : SUCCESS in 6m 06s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/cb3c140444d84052b6b87a91abb0be98 : SUCCESS in 5m 16s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/541df958e4794e85b86842a695df41f7 : SUCCESS in 8m 55s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ef73e2253433439a957481ec133c633f : SUCCESS in 11m 43s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/373606281184483c919148e93e882cb1 : SUCCESS in 16m 07s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/9ac026b2ace94d118b30eb45b3007d46 : SUCCESS in 21m 36s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/09817fc9632c4a728ef162ca92e034bd : SUCCESS in 20m 52s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/4670c2afece343cca26a108539bef8c4 : SUCCESS in 17m 11s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/652e486e036b497e9cdd9cd5d1bc9170 : SUCCESS in 22m 21s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6860ffb8b2c44b9cb171bed7245da2b6 : SUCCESS in 19m 22s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/fb43e9f319a14896af6024eacc9d4c41 : SUCCESS in 21m 00s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dacc6a47f29c4ae7b27cc020e2206ac3 : SUCCESS in 22m 51s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/759f5fd0f6f14ab496e88c0e0de0ff0f : SUCCESS in 1h 11m 00s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/662a23dedd7146b89cfcb627b67cb6f3 : SUCCESS in 1h 17m 28s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/de9b7a62b87244fcba9f1f02f1bc901b : SUCCESS in 1h 19m 15s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a9d903312f6f44208cec173d7fbf30dc : SUCCESS in 1h 16m 53s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/827afe1b17954b0db4e8f704524b7498 : SUCCESS in 1h 17m 04s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/07381c628321464c94f1e932375e8952 : SUCCESS in 1h 49m 44s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1cada4e2671f470f89eeda67dddc0bd6 : SUCCESS in 1h 58m 17s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/25c0093d8b374a249060554d3ae82b77 : SUCCESS in 19m 34s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/9010b653017c481e9814300273ce2a6d : SUCCESS in 25m 05s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2b29a1077a204e52a02b6ee602bc7af7 : SUCCESS in 26m 41s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/904836d07b244bfe80ddff27cc8afd9f : SUCCESS in 22m 50s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/5a25213a3d9a4d9f8e677f59b2a85b28 : SUCCESS in 41m 37s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/af8f0ef338af49b3969ed6897ecbe8fa : SUCCESS in 41m 08s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/a49bdcf0c3fc46f8b59457bcc2c1a18c : SUCCESS in 25m 48s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/50a1c743a66d4db58183470319029866 : SUCCESS in 21m 30s","accounts_in_message":[],"_revision_number":15},{"id":"747c9b89488ed6549bca3bd9d42b0dae3f81ce52","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-07-24 18:36:30.000000000","message":"Patch Set 15: Code-Review+1","accounts_in_message":[],"_revision_number":15},{"id":"ded5e34e4bda649e9c111c8d0557c59ba80c7ef8","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-07-27 14:21:55.000000000","message":"Patch Set 15: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":15},{"id":"ec65db130d24e76cb18dd977b8db2996be8d128d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-07-27 15:59:55.000000000","message":"Patch Set 16: Published edit on patch set 15.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":16},{"id":"2a7b700b328e169179a9316e0c1679fc16345679","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-07-27 16:15:38.000000000","message":"Uploaded patch set 17.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":17},{"id":"b820bd3ba16130b567ee903dae578b6ab2770cc2","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-07-27 16:16:08.000000000","message":"Patch Set 17: Review-Priority+1\n\n(4 comments)","accounts_in_message":[],"_revision_number":17},{"id":"704fdacd103b32acd6e3716b8e4d78d22d7d692d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-27 18:25:33.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/172917d9f843464cbba3438d0a1c1446\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a97bfefc767748d69bef33445a6c038d : SUCCESS in 8m 10s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/27f2cb2b650f463abf4f57ef33c4d1f7 : SUCCESS in 6m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e682791162a240508be6ec9c81139460 : SUCCESS in 4m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a1cf7f845cf74b9e9a4bb26ce6e737cb : SUCCESS in 4m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5a0cb6a5abb34b9c8e9efb8aecfffbaf : SUCCESS in 4m 22s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/39fda1fc9b3d4f308144565c9086205e : SUCCESS in 6m 42s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bf0c1dde903143a78caff762ace8a5c3 : SUCCESS in 6m 02s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e071c77622d2419ea2291532221bdc2f : SUCCESS in 4m 54s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/55dd648b38cf4b17ab30da23b3b01c90 : SUCCESS in 11m 42s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/239b91015dce465590f9febbe3db6f09 : SUCCESS in 11m 22s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/9fd67d6044094aac8f4d7321e9ef1650 : SUCCESS in 21m 23s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/4979841a9b6744d9af19795beb23be04 : SUCCESS in 21m 06s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/54255a4f4dd1422ab163b11a168648b6 : SUCCESS in 19m 22s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/9c657d01cb8a4a9fb2d04524a94b4e54 : SUCCESS in 11m 56s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/8685f3991d5244e888bde7ef6e25b873 : SUCCESS in 25m 10s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/846031ba3028415fa18ffe5dc36e6ef1 : SUCCESS in 18m 55s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/c825fc86276545bab2efa545efcc290d : SUCCESS in 17m 16s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5c06f77fed69415e87dd8740da07620d : SUCCESS in 19m 01s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/065e8fd07f1c48d282ed04e42dbf09b6 : SUCCESS in 1h 23m 21s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/36be251983574608bcc7aa8bac5fcb66 : SUCCESS in 1h 15m 12s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/8689df00221b4fda963fd8ac0bb77005 : SUCCESS in 1h 15m 51s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/41f3aaf8b63e4229ae0a2994e52c2c97 : SUCCESS in 1h 30m 48s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/87f659645270492f86dc16f12be5e82f : SUCCESS in 1h 30m 43s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/b1dea3f9b10e433b813d39104db26cbe : SUCCESS in 2h 03m 56s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6aa6d5475f7243eebb981ae4a7b66c3c : SUCCESS in 2h 08m 43s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/df970273fc77475e9058b6772959aa89 : SUCCESS in 27m 22s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/5706238641ff40c6b502edf87cbc2f12 : SUCCESS in 20m 08s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ce6e6b3332ce4561811d6e6df3fb54e0 : SUCCESS in 22m 54s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/274df2d09ad748599b7e945718ad3233 : SUCCESS in 13m 57s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/e51e5fd2de184aa49fe9453422c7f221 : SUCCESS in 39m 31s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e781451dcaae4aec8188f900b722e921 : SUCCESS in 33m 31s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/19a306c47d9944769a729c3d1a11a311 : SUCCESS in 20m 54s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a44db24096af4cc5a09c33757540b089 : SUCCESS in 21m 06s","accounts_in_message":[],"_revision_number":17},{"id":"f17280f0277c1afd70175521a7b1fd4ad337aacd","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-08-05 14:51:35.000000000","message":"Patch Set 17: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"d7adf04f8c23b95d2f17c2f135de16b1ddb5affd","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-08-07 07:29:04.000000000","message":"Patch Set 17: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"056a48d8c2445aeeca972323cff2db1f56fcccf2","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-08-07 07:29:15.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"8ad797cfd138c07dfd844904ccebacbe137ab153","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-08-07 07:30:35.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"db31770f328726a6a1cba52fd280f268ea44a0c1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-08-07 14:23:43.000000000","message":"Uploaded patch set 18.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":18},{"id":"1f90b01e626644d27c76f083049fc1ec2c16962a","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-08-07 14:25:48.000000000","message":"Patch Set 18: Code-Review+2","accounts_in_message":[],"_revision_number":18},{"id":"dbc0a4b1991731953590959b8ff67b776fffe049","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-08-07 14:26:20.000000000","message":"Uploaded patch set 19.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":19},{"id":"55716d49ef03f44e139691446defb493a4f8be2f","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2026-08-07 14:51:27.000000000","message":"Patch Set 19: Code-Review+2 Workflow+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":19},{"id":"8a77b119d5aad4f39e98759af81b6355d4713d08","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-08-07 15:03:42.000000000","message":"Patch Set 19: Backport-Candidate+1 Code-Review+2","accounts_in_message":[],"_revision_number":19},{"id":"b65b2427a9aca78c6528305af405af938e339a02","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-07 16:27:25.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a93eb97a858a43dd91317560c80515b1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e0d33920de4422c8eca8ab2407a2c50 : SUCCESS in 9m 59s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/204ed07a2c984d5181ccdc028c3d0862 : SUCCESS in 6m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ba932c260eda4e79862a256aa5115068 : SUCCESS in 5m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/879137c540df40598998515e62881cf4 : SUCCESS in 4m 30s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/809258d4ed6548d7a3249e2bd974e0b3 : SUCCESS in 4m 07s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/6ec3e7c73df5407f97cb2acf296cba4d : SUCCESS in 6m 43s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c12f86b5cadd4ad991268b73df1fd14f : SUCCESS in 5m 52s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b7806da643d245ef9df6d91667e56c00 : SUCCESS in 5m 23s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/00541abf51a24dd3aeccb4d00de145ce : SUCCESS in 10m 34s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/0c9ed70ebc824b30884e09e1498a819f : SUCCESS in 12m 38s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/506cf3cfc8bc43a58ff4c462811c0e56 : SUCCESS in 16m 05s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/412e2c0254464853b556bff05b0b97cf : SUCCESS in 16m 11s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3e3fea39c7b34988ab4b5ba264b1a391 : SUCCESS in 20m 14s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/110fbaa498d24cc5a54ab72f8e4da20c : SUCCESS in 15m 42s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/16ae9173c636466981a70a3756f40d42 : SUCCESS in 21m 03s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4dd50b8068bc4b2aacc558b3b3c2a987 : SUCCESS in 18m 32s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/954468693bc848b0a10f702e6a98b2c2 : SUCCESS in 27m 46s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a9752609b29d401a98fd61c3a66cd295 : SUCCESS in 16m 17s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/c5aa23d3463b48b3b0ab5a76138ba9fd : SUCCESS in 1h 08m 12s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/9d44730ad9ea4482aa4d043666a689f4 : SUCCESS in 1h 07m 59s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/fc6b48071f5b49a19cf18275ae967ec7 : SUCCESS in 1h 07m 54s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/27c82c999e7740daade48c36a302f4c2 : SUCCESS in 1h 13m 11s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/b04dfd6f9d1943aaa5e31ede20ae02df : SUCCESS in 1h 12m 38s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/438cad151e684b238fa0b5f7ed638ea3 : SUCCESS in 1h 54m 16s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0015616b3578478a93af6cb842230277 : SUCCESS in 1h 59m 16s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/b79767fd283646e2856aaa5bf63b783d : SUCCESS in 26m 09s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/e0958f98854b4ae78105d10e1359bbc3 : SUCCESS in 26m 50s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0e41df12a2644777bb9998ff5bc5d7b0 : SUCCESS in 15m 43s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/00028d06a89c41e484572b53a8c72fad : SUCCESS in 15m 41s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/968b867314664605aaad2581190bffe9 : SUCCESS in 33m 31s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/44a0ffcd663c40e9be4d309ae4f8a45e : SUCCESS in 21m 32s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/f256ae1f2ee94ea59a2f6e88423a3663 : SUCCESS in 25m 30s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/692317c850574775b0f1e694e140949b : SUCCESS in 13m 20s","accounts_in_message":[],"_revision_number":19},{"id":"6d064034159a52fb6f2b6106796b3c40283110bc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-07 16:27:45.000000000","message":"Patch Set 19: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":19},{"id":"90a12d3a56e6dd48f8e65466c23f3c9a10ce51f5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-07 18:30:00.000000000","message":"Patch Set 19: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fc4a2d0fc8b94c4d9243fd40c73c7d59\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b914ae88cee5416c948c8d188209043e : SUCCESS in 5m 27s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/3feaf4f785f94bd3a1104ccf343fdb27 : SUCCESS in 5m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae5ed46584764adcb9110c52f5aba55e : SUCCESS in 3m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/02a5659dd4a543b5890b80ed862d2a4a : SUCCESS in 4m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/88736af700c2479296499d07884b8931 : SUCCESS in 3m 42s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/40c83b2709bb4a4c9c8ddbd92808b769 : SUCCESS in 3m 54s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ca0bf5d5a2e143c1b5a1f9fe64916144 : SUCCESS in 7m 09s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2548226bb64845f686154b79fe697845 : SUCCESS in 6m 26s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/8e5291275e91474da0976bfe2da5281f : SUCCESS in 11m 56s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5ee9958b7ca34603976004644dcdd2b4 : SUCCESS in 11m 35s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/fc22b7cb116440ce947184dac11be968 : SUCCESS in 16m 15s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/401a8b7f23c044c1b31e532de9ddd94b : SUCCESS in 18m 13s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/67d588a16c0b4b20b99f2e5d4bfe764f : SUCCESS in 15m 30s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/459090ad6bfc4cb5978676edd75fdc0e : SUCCESS in 11m 15s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/4f7a54b95b664fe08eea57ccde9dfb21 : SUCCESS in 1h 04m 36s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/41cd31a0494042cd9188d7289ae64cac : SUCCESS in 1h 07m 26s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/5678827f61ed4bebb16109288abd8faf : SUCCESS in 1h 06m 44s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b279ec830f8f480aa1bde4c9f69b970e : SUCCESS in 1h 09m 07s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/1fc74d1d95f5406f9fcda177ea7ae38a : SUCCESS in 1h 12m 58s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/4dce83f2b5484ace96cf2939942b46e3 : SUCCESS in 1h 59m 47s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/93997ffd64074cff8546d276f6c24ff6 : SUCCESS in 2h 01m 24s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/59036879cdb540edaa78eef27049b020 : SUCCESS in 15m 58s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/06bfbbfc4bc14231902917db0b6adb2e : SUCCESS in 24m 31s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a433e112fc124612a42814d0ee099ec0 : SUCCESS in 25m 04s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/27c6b2dce553420bb852d596406b76a0 : SUCCESS in 20m 59s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/f521cec7697f405bb8ece35e237aeed3 : SUCCESS in 21m 57s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7ea9712f2ecc4e6ebc9f41892a0d6d52 : SUCCESS in 22m 43s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/5ac266c3361847cea8c744d841746e40 : SUCCESS in 14m 37s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/24a757f8ad20423698d9bdcdf3022b10 : SUCCESS in 19m 39s","accounts_in_message":[],"_revision_number":19},{"id":"3013727339720eaeee6ee17eec4f6eeab2f41aff","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-07 18:30:01.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":19},{"id":"1e17dae9a1db44a0d23841e9a72b4cbb718d423e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-07 18:31:03.000000000","message":"Patch Set 19:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/16c60f7dd8ce4f91bd8fb001e5801075\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2364e5be4a64167881685c41fc37b17 : SUCCESS in 44s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/185c61cab5a14c20b2771cb4d4386d68 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":19}],"current_revision_number":19,"current_revision":"7b191d9165ccd710dfc3bbdebc907b27b718c3d3","revisions":{"d8f9fcfc822b07c33355ab7a3fdcf813bc87dcf8":{"kind":"REWORK","_number":1,"created":"2026-06-26 09:11:55.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/16/995016/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/16/995016/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/995016/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/995016/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/995016/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/16/995016/1"}}},"commit":{"parents":[{"commit":"127c3ae67d8c2e1f0c61b7e45b522f5291defe43","subject":"Merge \"Simplify inspection rules\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/127c3ae67d8c2e1f0c61b7e45b522f5291defe43"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-06-26 09:07:15.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-06-26 09:11:49.000000000","tz":60},"subject":"Cleaner split of kolla-ansible install and configure","message":"Cleaner split of kolla-ansible install and configure\n\nThis allows us to install kayobe without generating the config.  This\nwas previous possible before the change to merge kolla-tags with\ntags[1].\n\n[1] https://github.com/openstack/kayobe/commit/fb1767ad7f42235f576ec5ae3fd6f623f207e512\n\nChange-Id: I81c2b2a8e1ccde44854257ce8e34e3517c945a4d\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d8f9fcfc822b07c33355ab7a3fdcf813bc87dcf8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d8f9fcfc822b07c33355ab7a3fdcf813bc87dcf8"}]},"branch":"refs/heads/master"},"078e3e1482bdf6370a6349c3e156dd9db55bbc03":{"kind":"REWORK","_number":2,"created":"2026-06-26 17:08:04.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/16/995016/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/16/995016/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/995016/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/995016/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/995016/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/16/995016/2"}}},"commit":{"parents":[{"commit":"127c3ae67d8c2e1f0c61b7e45b522f5291defe43","subject":"Merge \"Simplify inspection rules\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/127c3ae67d8c2e1f0c61b7e45b522f5291defe43"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-06-26 09:07:15.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-06-26 17:06:39.000000000","tz":0},"subject":"Cleaner split of kolla-ansible install and configure","message":"Cleaner split of kolla-ansible install and configure\n\nThis allows us to install kayobe without generating the config.  This\nwas previous possible before the change to merge kolla-tags with\ntags[1].\n\n[1] https://github.com/openstack/kayobe/commit/fb1767ad7f42235f576ec5ae3fd6f623f207e512\n\nChange-Id: I81c2b2a8e1ccde44854257ce8e34e3517c945a4d\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/078e3e1482bdf6370a6349c3e156dd9db55bbc03"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/078e3e1482bdf6370a6349c3e156dd9db55bbc03"}]},"branch":"refs/heads/master"},"42ffe9d62b018194d1109c0993241aeebad81c59":{"kind":"REWORK","_number":3,"created":"2026-06-26 17:36:19.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/16/995016/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/16/995016/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/995016/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/995016/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/995016/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/16/995016/3"}}},"commit":{"parents":[{"commit":"127c3ae67d8c2e1f0c61b7e45b522f5291defe43","subject":"Merge \"Simplify inspection rules\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/127c3ae67d8c2e1f0c61b7e45b522f5291defe43"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-06-26 09:07:15.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-06-26 17:36:11.000000000","tz":0},"subject":"Cleaner split of kolla-ansible install and configure","message":"Cleaner split of kolla-ansible install and configure\n\nThis allows us to install kayobe without generating the config.  This\nwas previous possible before the change to merge kolla-tags with\ntags[1].\n\n[1] https://github.com/openstack/kayobe/commit/fb1767ad7f42235f576ec5ae3fd6f623f207e512\n\nChange-Id: I81c2b2a8e1ccde44854257ce8e34e3517c945a4d\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/42ffe9d62b018194d1109c0993241aeebad81c59"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/42ffe9d62b018194d1109c0993241aeebad81c59"}]},"branch":"refs/heads/master"},"e4f48e54d65001e26fdcbdc523b43e479f458e14":{"kind":"REWORK","_number":4,"created":"2026-06-29 10:42:09.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/16/995016/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/16/995016/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/995016/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/995016/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/995016/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/16/995016/4"}}},"commit":{"parents":[{"commit":"127c3ae67d8c2e1f0c61b7e45b522f5291defe43","subject":"Merge \"Simplify inspection rules\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/127c3ae67d8c2e1f0c61b7e45b522f5291defe43"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-06-26 09:07:15.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-06-29 10:40:52.000000000","tz":60},"subject":"Cleaner split of kolla-ansible install and configure","message":"Cleaner split of kolla-ansible install and configure\n\nThis allows us to install kayobe without generating the config.  This\nwas previous possible before the change to merge kolla-tags with\ntags[1].\n\n[1] https://github.com/openstack/kayobe/commit/fb1767ad7f42235f576ec5ae3fd6f623f207e512\n\nChange-Id: I81c2b2a8e1ccde44854257ce8e34e3517c945a4d\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e4f48e54d65001e26fdcbdc523b43e479f458e14"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e4f48e54d65001e26fdcbdc523b43e479f458e14"}]},"branch":"refs/heads/master"},"ce71d00a752596a6f619336c520e435bd224b736":{"kind":"REWORK","_number":5,"created":"2026-06-29 11:56:17.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/16/995016/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/16/995016/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/995016/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/995016/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/995016/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/16/995016/5"}}},"commit":{"parents":[{"commit":"127c3ae67d8c2e1f0c61b7e45b522f5291defe43","subject":"Merge \"Simplify inspection rules\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/127c3ae67d8c2e1f0c61b7e45b522f5291defe43"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-06-26 09:07:15.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-06-29 11:56:03.000000000","tz":60},"subject":"Cleaner split of kolla-ansible install and configure","message":"Cleaner split of kolla-ansible install and configure\n\nThis allows us to install kayobe without generating the config.  This\nwas previous possible before the change to merge kolla-tags with\ntags[1].\n\n[1] https://github.com/openstack/kayobe/commit/fb1767ad7f42235f576ec5ae3fd6f623f207e512\n\nChange-Id: I81c2b2a8e1ccde44854257ce8e34e3517c945a4d\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ce71d00a752596a6f619336c520e435bd224b736"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ce71d00a752596a6f619336c520e435bd224b736"}]},"branch":"refs/heads/master"},"21c262cabb46de89dbd724ee6b575355c89197f6":{"kind":"REWORK","_number":6,"created":"2026-06-29 15:20:28.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/16/995016/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/16/995016/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/995016/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/995016/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/995016/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/16/995016/6"}}},"commit":{"parents":[{"commit":"127c3ae67d8c2e1f0c61b7e45b522f5291defe43","subject":"Merge \"Simplify inspection rules\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/127c3ae67d8c2e1f0c61b7e45b522f5291defe43"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-06-26 09:07:15.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-06-29 15:20:21.000000000","tz":60},"subject":"Cleaner split of kolla-ansible install and configure","message":"Cleaner split of kolla-ansible install and configure\n\nThis allows us to install kayobe without generating the config.  This\nwas previous possible before the change to merge kolla-tags with\ntags[1].\n\n[1] https://github.com/openstack/kayobe/commit/fb1767ad7f42235f576ec5ae3fd6f623f207e512\n\nChange-Id: I81c2b2a8e1ccde44854257ce8e34e3517c945a4d\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/21c262cabb46de89dbd724ee6b575355c89197f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/21c262cabb46de89dbd724ee6b575355c89197f6"}]},"branch":"refs/heads/master"},"b76eb69be2228007684a2063582709b0615bf2b9":{"kind":"REWORK","_number":7,"created":"2026-06-29 15:36:33.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/16/995016/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/16/995016/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/995016/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/995016/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/995016/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/16/995016/7"}}},"commit":{"parents":[{"commit":"127c3ae67d8c2e1f0c61b7e45b522f5291defe43","subject":"Merge \"Simplify inspection rules\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/127c3ae67d8c2e1f0c61b7e45b522f5291defe43"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-06-26 09:07:15.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-06-29 15:36:25.000000000","tz":60},"subject":"Cleaner split of kolla-ansible install and configure","message":"Cleaner split of kolla-ansible install and configure\n\nThis allows us to install kayobe without generating the config.  This\nwas previous possible before the change to merge kolla-tags with\ntags[1].\n\n[1] https://github.com/openstack/kayobe/commit/fb1767ad7f42235f576ec5ae3fd6f623f207e512\n\nChange-Id: I81c2b2a8e1ccde44854257ce8e34e3517c945a4d\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b76eb69be2228007684a2063582709b0615bf2b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b76eb69be2228007684a2063582709b0615bf2b9"}]},"branch":"refs/heads/master"},"366e90aded0c7eb53bd18d0ef12e65bb2cbe15ab":{"kind":"REWORK","_number":8,"created":"2026-06-29 18:17:10.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/16/995016/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/16/995016/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/995016/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/995016/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/995016/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/16/995016/8"}}},"commit":{"parents":[{"commit":"127c3ae67d8c2e1f0c61b7e45b522f5291defe43","subject":"Merge \"Simplify inspection rules\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/127c3ae67d8c2e1f0c61b7e45b522f5291defe43"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-06-26 09:07:15.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-06-29 18:16:48.000000000","tz":60},"subject":"Cleaner split of kolla-ansible install and configure","message":"Cleaner split of kolla-ansible install and configure\n\nThis allows us to install kayobe without generating the config. This is\nuseful in a docker image build where will later mount the config in.\n\nChange-Id: I81c2b2a8e1ccde44854257ce8e34e3517c945a4d\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/366e90aded0c7eb53bd18d0ef12e65bb2cbe15ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/366e90aded0c7eb53bd18d0ef12e65bb2cbe15ab"}]},"branch":"refs/heads/master"},"47415c90ef3973f0ee618aea109b4c91598d776b":{"kind":"REWORK","_number":9,"created":"2026-06-29 18:29:37.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/16/995016/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/16/995016/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/995016/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/995016/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/995016/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/16/995016/9"}}},"commit":{"parents":[{"commit":"127c3ae67d8c2e1f0c61b7e45b522f5291defe43","subject":"Merge \"Simplify inspection rules\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/127c3ae67d8c2e1f0c61b7e45b522f5291defe43"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-06-26 09:07:15.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-06-29 18:28:48.000000000","tz":60},"subject":"Cleaner split of kolla-ansible install and configure","message":"Cleaner split of kolla-ansible install and configure\n\nThis allows us to install kayobe without generating the config. This is\nuseful in a docker image build where will later mount the config in.\n\nChange-Id: I81c2b2a8e1ccde44854257ce8e34e3517c945a4d\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/47415c90ef3973f0ee618aea109b4c91598d776b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/47415c90ef3973f0ee618aea109b4c91598d776b"}]},"branch":"refs/heads/master"},"9d538669da5d7dbe9968d779796b67f037062369":{"kind":"REWORK","_number":10,"created":"2026-06-29 18:38:07.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/16/995016/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/16/995016/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/995016/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/995016/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/995016/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/16/995016/10"}}},"commit":{"parents":[{"commit":"127c3ae67d8c2e1f0c61b7e45b522f5291defe43","subject":"Merge \"Simplify inspection rules\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/127c3ae67d8c2e1f0c61b7e45b522f5291defe43"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-06-26 09:07:15.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-06-29 18:38:01.000000000","tz":60},"subject":"Cleaner split of kolla-ansible install and configure","message":"Cleaner split of kolla-ansible install and configure\n\nThis allows us to install kayobe without generating the config. This is\nuseful in a docker image build where will later mount the config in.\n\nChange-Id: I81c2b2a8e1ccde44854257ce8e34e3517c945a4d\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9d538669da5d7dbe9968d779796b67f037062369"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9d538669da5d7dbe9968d779796b67f037062369"}]},"branch":"refs/heads/master"},"08a0d4afbc92226552c22ef7061b3b3cdf9bf749":{"kind":"REWORK","_number":11,"created":"2026-06-30 08:27:30.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/16/995016/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/16/995016/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/995016/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/995016/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/995016/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/16/995016/11"}}},"commit":{"parents":[{"commit":"127c3ae67d8c2e1f0c61b7e45b522f5291defe43","subject":"Merge \"Simplify inspection rules\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/127c3ae67d8c2e1f0c61b7e45b522f5291defe43"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-06-26 09:07:15.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-06-30 08:27:23.000000000","tz":60},"subject":"Cleaner split of kolla-ansible install and configure","message":"Cleaner split of kolla-ansible install and configure\n\nThis allows us to install kayobe without generating the config. This is\nuseful in a docker image build where will later mount the config in.\n\nChange-Id: I81c2b2a8e1ccde44854257ce8e34e3517c945a4d\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/08a0d4afbc92226552c22ef7061b3b3cdf9bf749"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/08a0d4afbc92226552c22ef7061b3b3cdf9bf749"}]},"branch":"refs/heads/master"},"9e21f84f43f16e152ca1729d986b0ec9e371950c":{"kind":"REWORK","_number":12,"created":"2026-06-30 10:56:51.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/16/995016/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/16/995016/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/995016/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/995016/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/995016/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/16/995016/12"}}},"commit":{"parents":[{"commit":"127c3ae67d8c2e1f0c61b7e45b522f5291defe43","subject":"Merge \"Simplify inspection rules\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/127c3ae67d8c2e1f0c61b7e45b522f5291defe43"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-06-26 09:07:15.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-06-30 10:56:44.000000000","tz":60},"subject":"Cleaner split of kolla-ansible install and configure","message":"Cleaner split of kolla-ansible install and configure\n\nThis allows us to install kayobe without generating the config. This is\nuseful in a docker image build where will later mount the config in.\n\nChange-Id: I81c2b2a8e1ccde44854257ce8e34e3517c945a4d\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9e21f84f43f16e152ca1729d986b0ec9e371950c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9e21f84f43f16e152ca1729d986b0ec9e371950c"}]},"branch":"refs/heads/master"},"44be81597317789799286b4e1a7f1fdcd0809cd2":{"kind":"REWORK","_number":13,"created":"2026-07-09 15:48:25.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/16/995016/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/16/995016/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/995016/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/995016/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/995016/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/16/995016/13"}}},"commit":{"parents":[{"commit":"127c3ae67d8c2e1f0c61b7e45b522f5291defe43","subject":"Merge \"Simplify inspection rules\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/127c3ae67d8c2e1f0c61b7e45b522f5291defe43"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-06-26 09:07:15.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-07-09 15:47:54.000000000","tz":0},"subject":"Cleaner split of kolla-ansible install and configure","message":"Cleaner split of kolla-ansible install and configure\n\nThis allows us to install kayobe without generating the config. This is\nuseful in a docker image build where will later mount the config in.\n\nChange-Id: I81c2b2a8e1ccde44854257ce8e34e3517c945a4d\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/44be81597317789799286b4e1a7f1fdcd0809cd2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/44be81597317789799286b4e1a7f1fdcd0809cd2"}]},"branch":"refs/heads/master"},"b0541a4d4dea381266da5fdd952f9e1949bd80c2":{"kind":"REWORK","_number":14,"created":"2026-07-13 14:48:47.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/16/995016/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/16/995016/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/995016/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/995016/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/995016/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/16/995016/14"}}},"commit":{"parents":[{"commit":"127c3ae67d8c2e1f0c61b7e45b522f5291defe43","subject":"Merge \"Simplify inspection rules\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/127c3ae67d8c2e1f0c61b7e45b522f5291defe43"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-06-26 09:07:15.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-07-13 14:48:36.000000000","tz":0},"subject":"Cleaner split of kolla-ansible install and configure","message":"Cleaner split of kolla-ansible install and configure\n\nThis allows us to install kayobe without generating the config. This is\nuseful in a docker image build where will later mount the config in.\n\nChange-Id: I81c2b2a8e1ccde44854257ce8e34e3517c945a4d\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b0541a4d4dea381266da5fdd952f9e1949bd80c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b0541a4d4dea381266da5fdd952f9e1949bd80c2"}]},"branch":"refs/heads/master"},"607092995740c4ed073ceee834eddbdcbf72bc58":{"kind":"NO_CODE_CHANGE","_number":15,"created":"2026-07-17 15:24:28.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/16/995016/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/16/995016/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/995016/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/995016/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/995016/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/16/995016/15"}}},"commit":{"parents":[{"commit":"127c3ae67d8c2e1f0c61b7e45b522f5291defe43","subject":"Merge \"Simplify inspection rules\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/127c3ae67d8c2e1f0c61b7e45b522f5291defe43"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-06-26 09:07:15.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-07-17 15:24:28.000000000","tz":0},"subject":"Cleaner split of kolla-ansible install and configure","message":"Cleaner split of kolla-ansible install and configure\n\nThis allows us to install Kayobe without generating the configuration.\nThis is useful when building a Docker image, where you want to install\nall dependencies at build time but do not require the runtime\nconfiguration. The runtime configuration can then be generated when the\ncontainer is executed.\n\nChange-Id: I81c2b2a8e1ccde44854257ce8e34e3517c945a4d\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/607092995740c4ed073ceee834eddbdcbf72bc58"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/607092995740c4ed073ceee834eddbdcbf72bc58"}]},"branch":"refs/heads/master","description":"Edit commit message"},"a5e5ec39e076f9ef81ab38ba86eeba0d2a501b83":{"kind":"REWORK","_number":16,"created":"2026-07-27 15:59:55.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/16/995016/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/16/995016/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/995016/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/995016/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/995016/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/16/995016/16"}}},"commit":{"parents":[{"commit":"127c3ae67d8c2e1f0c61b7e45b522f5291defe43","subject":"Merge \"Simplify inspection rules\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/127c3ae67d8c2e1f0c61b7e45b522f5291defe43"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-06-26 09:07:15.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-07-27 15:59:47.000000000","tz":0},"subject":"Cleaner split of kolla-ansible install and configure","message":"Cleaner split of kolla-ansible install and configure\n\nThis allows us to install Kayobe without generating the configuration.\nThis is useful when building a Docker image, where you want to install\nall dependencies at build time but do not require the runtime\nconfiguration. The runtime configuration can then be generated when the\ncontainer is executed.\n\nChange-Id: I81c2b2a8e1ccde44854257ce8e34e3517c945a4d\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a5e5ec39e076f9ef81ab38ba86eeba0d2a501b83"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a5e5ec39e076f9ef81ab38ba86eeba0d2a501b83"}]},"branch":"refs/heads/master"},"e1fb1096ec1cfdb5dda8acbde7b887f52e37c281":{"kind":"REWORK","_number":17,"created":"2026-07-27 16:15:38.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/16/995016/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/16/995016/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/995016/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/995016/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/995016/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/16/995016/17"}}},"commit":{"parents":[{"commit":"127c3ae67d8c2e1f0c61b7e45b522f5291defe43","subject":"Merge \"Simplify inspection rules\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/127c3ae67d8c2e1f0c61b7e45b522f5291defe43"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-06-26 09:07:15.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-07-27 16:15:31.000000000","tz":60},"subject":"Cleaner split of kolla-ansible install and configure","message":"Cleaner split of kolla-ansible install and configure\n\nThis allows us to install Kayobe without generating the configuration.\nThis is useful when building a Docker image, where you want to install\nall dependencies at build time but do not require the runtime\nconfiguration. The runtime configuration can then be generated when the\ncontainer is executed.\n\nChange-Id: I81c2b2a8e1ccde44854257ce8e34e3517c945a4d\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e1fb1096ec1cfdb5dda8acbde7b887f52e37c281"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e1fb1096ec1cfdb5dda8acbde7b887f52e37c281"}]},"branch":"refs/heads/master"},"9a2949ddf568c5316df345252ada94b6725e5687":{"kind":"REWORK","_number":18,"created":"2026-08-07 14:23:43.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/16/995016/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/16/995016/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/995016/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/995016/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/995016/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/16/995016/18"}}},"commit":{"parents":[{"commit":"085e76af67a6bc4a6180781c7994ea59d0c8ea4c","subject":"Merge \"Add support for QoS egress/ingress settings in systemd-networkd\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/085e76af67a6bc4a6180781c7994ea59d0c8ea4c"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-06-26 09:07:15.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-08-07 14:22:28.000000000","tz":120},"subject":"Cleaner split of kolla-ansible install and configure","message":"Cleaner split of kolla-ansible install and configure\n\nThis allows us to install Kayobe without generating the configuration.\nThis is useful when building a Docker image, where you want to install\nall dependencies at build time but do not require the runtime\nconfiguration. The runtime configuration can then be generated when the\ncontainer is executed.\n\nChange-Id: I81c2b2a8e1ccde44854257ce8e34e3517c945a4d\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9a2949ddf568c5316df345252ada94b6725e5687"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9a2949ddf568c5316df345252ada94b6725e5687"}]},"branch":"refs/heads/master"},"7b191d9165ccd710dfc3bbdebc907b27b718c3d3":{"kind":"REWORK","_number":19,"created":"2026-08-07 14:26:20.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/16/995016/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/16/995016/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/995016/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/995016/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/995016/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/16/995016/19"}}},"commit":{"parents":[{"commit":"085e76af67a6bc4a6180781c7994ea59d0c8ea4c","subject":"Merge \"Add support for QoS egress/ingress settings in systemd-networkd\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/085e76af67a6bc4a6180781c7994ea59d0c8ea4c"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-06-26 09:07:15.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-08-07 14:26:15.000000000","tz":120},"subject":"Cleaner split of kolla-ansible install and configure","message":"Cleaner split of kolla-ansible install and configure\n\nThis allows us to install Kayobe without generating the configuration.\nThis is useful when building a Docker image, where you want to install\nall dependencies at build time but do not require the runtime\nconfiguration. The runtime configuration can then be generated when the\ncontainer is executed.\n\nChange-Id: I81c2b2a8e1ccde44854257ce8e34e3517c945a4d\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7b191d9165ccd710dfc3bbdebc907b27b718c3d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7b191d9165ccd710dfc3bbdebc907b27b718c3d3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
