)]}'
{"id":"openstack%2Fkayobe~995021","triplet_id":"openstack%2Fkayobe~master~I77d5727b3692833dbb498a743320a90cbf575217","project":"openstack/kayobe","branch":"master","topic":"ansi_bump","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2026-09-07 11:58:54.000000000","reason":"\u003cGERRIT_ACCOUNT_28048\u003e replied on the change","reason_account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},"34940":{"account":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"last_update":"2026-09-07 14:03:00.000000000","reason":"Change was submitted"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2026-09-07 14:03:00.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I77d5727b3692833dbb498a743320a90cbf575217","subject":"Bump up Ansible supported versions to 13.x/14.x","status":"MERGED","created":"2026-06-26 09:57:22.000000000","updated":"2026-09-07 14:04:10.000000000","submitted":"2026-09-07 14:03:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":26,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"995021-ansi_bump","meta_rev_id":"8df4782a38692bf075951092e9ce35f6a435fac7","_number":995021,"virtual_id_number":995021,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-07 14:03:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2026-09-07 11:58:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2026-09-07 13:25:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-09-07 10:02:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2026-09-07 11:58:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"reviewers":{"CC":[{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"}],"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-26 11:51:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-07-29 15:25:20.000000000","updated_by":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"reviewer":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"state":"CC"},{"updated":"2026-08-20 16:37:25.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"CC"},{"updated":"2026-09-07 11:03:08.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2026-09-07 11:03:08.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2026-09-07 11:03:08.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2026-09-07 11:03:08.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2026-09-07 11:03:08.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2026-09-07 11:03:08.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2026-09-07 13:25:07.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"bf13a2260ec24231fce8a84e897cd5c56b0ff5fe","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-06-26 09:57:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8ebff7ee93bbada4c586396500798c5b0b36ab3b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-26 11:51:37.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/22d0a02fe5a3487aafddbe421013e4b4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5f5658c3549430e8ba34182fa974c7f : SUCCESS in 9m 07s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/ae815a2da2e14c838b0b65deb2e4cbc2 : SUCCESS in 6m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d6dfaf668f72491aa0084204416e7a16 : SUCCESS in 4m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/61a7aaecb9b246d7977c7255984af76e : SUCCESS in 3m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/34f735e86dea48be8d338d836f93ca55 : SUCCESS in 2m 59s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f8d4ccdebeb14f7ea977a640b1323b52 : SUCCESS in 4m 47s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c47ceff01eee4f948d4d72edf2da39ec : SUCCESS in 4m 32s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/17ae5814587c40af9633d8e8c63cab59 : FAILURE in 5m 27s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/cbdef2b3f1c34309a0bb3d140d7ce276 : FAILURE in 9m 44s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3e308256e9334d5e94a211117891bb1b : FAILURE in 5m 43s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/309221696a1248d3ac549f04a1f05f61 : FAILURE in 14m 21s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/00bdec44663d4e94890141f1b4f5aa8d : FAILURE in 14m 31s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/67dc85ccd4d64948b40947ac2c8f7440 : FAILURE in 19m 37s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/5d225e810d68452e827f236f51f82686 : FAILURE in 19m 41s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/7f6853381cac4519a9e74fc1d31a50f2 : FAILURE in 15m 09s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9e4ba423916d4ff2966b77aff76bfeaf : FAILURE in 18m 43s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/d3cd234f1250472290477254e0db0ca9 : FAILURE in 14m 20s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fa6105b61f8b49c78905765cd0739770 : FAILURE in 14m 41s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/128103f0ba144a8e90d6224cd872152e : FAILURE in 17m 04s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/7e13dc7cd49b422e9334f83cbf6bbe44 : FAILURE in 17m 50s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/3e152f8e92fe48d59f21d9ae3e4a45da : FAILURE in 18m 28s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8250d1b75d1a413683617d94d3bc3367 : FAILURE in 18m 31s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/5ff56e9ca2ac43398f48613a89035696 : FAILURE in 19m 03s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/7a1343d2ca184ca6928003b84e8001db : SUCCESS in 1h 47m 01s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9732a6457d104939ab64c7f9ca9359f9 : SUCCESS in 1h 53m 23s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/6f02a088d51d49d4bb02c564b4204965 : FAILURE in 22m 38s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/6783e84611804394947c53c3a743994b : FAILURE in 17m 32s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0dc68ce8071743c8be4b72436167e0ba : FAILURE in 19m 23s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/ca5ff0af9e974203af58700795b4faa4 : FAILURE in 16m 09s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/85ef8167a3844f28a08f50859ec30dff : SUCCESS in 31m 14s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/976d43ae091f4a9f903060f340475f0c : SUCCESS in 26m 18s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/c35734e03daf4908b2506d15ab8f0b4f : FAILURE in 20m 43s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7ad83a4f7820400cb969bb16c8c54292 : FAILURE in 14m 11s","accounts_in_message":[],"_revision_number":1},{"id":"11ef56571a66f56ee265857d3601e1d1f89be930","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"real_author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2026-07-22 09:55:30.000000000","message":"Patch Set 2: Patch Set 1 was rebased on behalf of \u003cGERRIT_ACCOUNT_15197\u003e\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"_revision_number":2},{"id":"4cf3e01b30f5b7a17d0482c0ff6f704064d2e2e5","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-07-22 10:13:03.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"7124b611f507ba70a9425271caa33738bf3e79e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-22 12:09:06.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9e1af6852e0c4e2c9e953966a263e8e7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/431a0686d53d438497d2a86762621542 : SUCCESS in 10m 44s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/3e42f61c54c84354a476895e9cf329a6 : SUCCESS in 6m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a82bd8992be04a6db087bdfdc009baf1 : SUCCESS in 4m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6adb4bb3ce5a4b1ea17b66bd97cd66ab : SUCCESS in 6m 50s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/95a9f21963a64c49b991176306cf97c9 : SUCCESS in 4m 01s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/5002340660804d549c7ca81f11249cd7 : SUCCESS in 7m 22s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eb31a1f3dade4c50b77aa55e1943b96e : SUCCESS in 5m 48s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/09bbc34e0c7048859c12367c33f60ed8 : SUCCESS in 6m 39s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ca1e2361814d47859f6e22fbeae33b12 : FAILURE in 7m 12s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3a1d605c822e444fa45f107ba18ae25e : FAILURE in 3m 41s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/2238ee0d59334f8ca850a88fcbe0981a : FAILURE in 24m 50s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/db110e51f03d43f292c2a6d9f9f4735a : FAILURE in 13m 12s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4717eb66a5b3474b97b39767037093cf : FAILURE in 19m 36s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/07dd3de0b7604f298951ed97cf7b4cb8 : FAILURE in 21m 17s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/06225781b062484498d807f4f4e7c5ba : FAILURE in 12m 25s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c6f73ed0af304e8dbcec0157f912123d : FAILURE in 17m 22s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/5e102bc1981f4b679990054d93ebd7b4 : FAILURE in 19m 16s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b9210da2f92547339caf5821318d1066 : FAILURE in 19m 53s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/0174c91fd23045c4942bef72dc451b55 : FAILURE in 15m 35s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/67fb706118334f42a3e2c35ced589759 : FAILURE in 15m 35s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/5e281fee1045484d8f982af3051b4968 : FAILURE in 17m 10s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c63d78452a10498b9106891c7f2ad9eb : FAILURE in 16m 37s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/1c07eba16d7845f5bf6ba2386f046eca : FAILURE in 16m 35s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/e24591ac1e554cf98b6f5b74da70d40d : SUCCESS in 1h 47m 49s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cdae1e06b6184a11aef67129abd93b66 : SUCCESS in 1h 55m 00s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/3cda4eceecf84a5ea981b4eb23c30c34 : FAILURE in 12m 30s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/94562a046f754078a5a9ca6b6ae9c6bc : FAILURE in 12m 35s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/229d8cbcb2fe44459612cf2d8c2dec10 : FAILURE in 20m 07s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/d67fa7b6bd0745de8d5ed251a6b7f268 : FAILURE in 20m 11s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/9043d2308daf4dc0ae07215ee052453e : SUCCESS in 23m 49s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1acf7c9e66bf4f43a15e77627b4c5951 : SUCCESS in 39m 42s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/220c7ccef2da4fa6b68d489533136823 : FAILURE in 15m 00s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4ae96ed0580843a98050ef0e4a292828 : FAILURE in 19m 58s","accounts_in_message":[],"_revision_number":3},{"id":"34ade223801045ef3944e4736632ce2edf77012b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2026-07-24 13:33:10.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"fba137212f2560ea16005d2d263966545b82dc02","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-24 14:57:58.000000000","message":"Patch Set 4: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b57e614682794c1699b0a95b138b515d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92e1afd0d7304104a072333a6f9acd07 : SUCCESS in 9m 29s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/f50c0ce493734ecebe5216fa56669e57 : SUCCESS in 7m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/67110a17b7174267b85d30d055af2e6c : SUCCESS in 4m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8752b15bba584ab1af78e1bf46ddd216 : FAILURE in 4m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a3d53aee91794991a01013696ccf8ff1 : SUCCESS in 4m 09s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f96adfa02476446d9a2dd6a5c53e867f : SUCCESS in 6m 50s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/145f48be8d32470e8259f2d0f3bf8cf0 : SUCCESS in 6m 57s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/89ea9b54212e4caf83e96176a4379ecf : SUCCESS in 5m 53s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9d766df0ebd4451e92dee4db81794d4d : SUCCESS in 13m 00s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/8fa8348e6f3d430883baa5e2a0e56328 : FAILURE in 7m 54s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/db2ac4fdb08f4646859abaad0708b366 : SUCCESS in 23m 23s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/3d0c07c0828949369f6a2d570c307701 : SUCCESS in 20m 09s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/92ceb341518444b09f31fbb0c4195dc2 : SUCCESS in 17m 41s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/946d39af7c5c4cdf80fa79f690a31aa9 : SUCCESS in 17m 35s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/878de7f3d44449f19ab600da14d16b15 : SUCCESS in 19m 14s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7533283f5ce44960a53e26641b185fa4 : SUCCESS in 19m 07s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/17d8048a9547474ea27704bb4cb0dbab : SUCCESS in 19m 27s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c796acb8f4a14384a171674de80f08c5 : SUCCESS in 25m 20s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/3870b3c2cdc3469e8ba8febb1de9495c : SUCCESS in 1h 14m 27s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/390fe12ee5c648fcb281eb27e274d07b : SUCCESS in 1h 15m 11s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/b5e8e5a3c4fa4641b7c2b607e057f677 : SUCCESS in 1h 13m 17s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c69546d9397142aaa112735da8d1cedb : SUCCESS in 1h 22m 16s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/19ba466dfb5349be8c49bbcd51320119 : SUCCESS in 1h 18m 35s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/02814450799e433684a84da97f240a9b : FAILURE in 1h 11m 18s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7b1d3dcefc4b4ca7a40e8316cebeef45 : FAILURE in 1h 12m 54s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/d0a5fe83949b4076b056c10e20a1b1b8 : SUCCESS in 23m 23s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/04bdd3020aca4bb6840f28a1bc489a99 : SUCCESS in 27m 04s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d7d6fc6dbe3c48e585299006083f2a61 : SUCCESS in 25m 28s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/54bd5f65d1dd4026acf0b2aeca973899 : SUCCESS in 24m 31s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/94ed77291588430fb8b37809089ebdf7 : FAILURE in 40m 26s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b7a5a9ddf95f465091f9205a2100875a : FAILURE in 45m 18s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/7f10ed1f346c4528a4d70eb7de3e0b0f : SUCCESS in 26m 17s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/657fad96a358414a85ecda4d19472970 : SUCCESS in 22m 01s","accounts_in_message":[],"_revision_number":4},{"id":"53717eb6cd15a8b6e3dc77b138ebd883d0ea2769","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2026-07-26 12:48:09.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"5a458c1fa440b97eb96e542d1674d1a1e6a904d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-26 14:42:57.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/da007a6f895f4810bede5f71e4c1f9aa\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/560ba4088e684889974322945269f710 : SUCCESS in 9m 09s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/d4742d852c1942f1a30b92f631db22c5 : SUCCESS in 3m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8ec47872bf3b442bae015339e7b020b5 : SUCCESS in 4m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/facb2c0012214c52b581761910e898fa : SUCCESS in 2m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7566ebc920144e5eb72ad229da6d6436 : SUCCESS in 4m 04s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/7d14773b100944b2981efc4c53f65acc : SUCCESS in 6m 48s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/54824c2eaf164e00b572de6e8d61cbbc : SUCCESS in 6m 03s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6fedb263d6df446abd15f5ee60ced1d0 : SUCCESS in 3m 22s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8c87ea2b17c546a8a72713f4b11ff633 : SUCCESS in 11m 38s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/206b1a713ad340eda87a0ed4528c4e14 : FAILURE in 5m 19s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/9dc1e31af4fb436a852216340812479c : SUCCESS in 14m 10s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/545bbf0278734174be131f42ea9721f3 : SUCCESS in 10m 27s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a30e082981ce4dbcaeff43db8d51b7dc : SUCCESS in 15m 29s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/7495871fb3474cf3a341431feb1d8601 : SUCCESS in 14m 05s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/0808275e09b94b16a106d384fe0d922b : SUCCESS in 20m 34s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bbf977b878b246bf9076f9a0583eba50 : SUCCESS in 11m 39s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/1bd1c17360f340d79f3c18b56b41c872 : SUCCESS in 22m 36s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/39fe1cc28e884a60aa6d05f3085db00f : SUCCESS in 18m 38s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/b564137df62d4685a2dda50c56046650 : SUCCESS in 1h 11m 04s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/8f8bf26152b04efeaaa9dc7816efbba5 : SUCCESS in 1h 09m 35s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/e99c8290efdf4c68a436618f2067c67f : SUCCESS in 1h 06m 31s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b6aab36ea32b48e995f2bb641e99d213 : SUCCESS in 1h 16m 42s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/4ba8d945571a4de2817c307ddc46cfa0 : FAILURE in 1h 18m 11s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/74f9447607ff430687d89d4c14bc78bf : SUCCESS in 1h 47m 14s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9fee36caf37b41f7b1a634c11e22ede5 : SUCCESS in 1h 54m 05s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/61eff520ac424880b813c1acd3310173 : SUCCESS in 27m 48s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/657a9b7aa5fb46fc841ca46a8fa13883 : SUCCESS in 20m 32s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/354d3169aaab43259e89d0b278ef6aa5 : SUCCESS in 22m 34s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/56080947863643d3b1be3beff1a0104d : SUCCESS in 25m 35s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/1886e8c971a249a4b967f75ee76f8e71 : SUCCESS in 27m 06s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8d46f3a19b74410bad65579a944b9d2d : SUCCESS in 24m 16s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/1ce754931095479481ae7857c33597f7 : SUCCESS in 14m 53s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/19d5051991b944aab9618079841319a0 : SUCCESS in 12m 01s","accounts_in_message":[],"_revision_number":5},{"id":"7bb2aed6463359463041c4951cd0715366bd2de1","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2026-07-26 19:12:44.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":6},{"id":"97aa77443682ad747e6aec0511cfb9a5a32bb296","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-26 21:08:54.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a786d26fb2dc4741ad76cb042f80f897\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0b94d2fef3f449ba7b80f468687b971 : SUCCESS in 8m 40s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/444954277b3f48fe9cc72e8b25c1011e : SUCCESS in 3m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2b6b968365b3492b9cbaf18770a1b309 : SUCCESS in 6m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/48fa3cd49bf346058792c99090c2b7c4 : SUCCESS in 4m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0d5dd00772e64903b7d7c1a890d82c04 : SUCCESS in 4m 11s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/4032117ce48640a39d32a76322e2aaed : SUCCESS in 6m 52s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7e26b4c21f704953a2f8cff61ef4b75f : SUCCESS in 5m 36s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ee81395e9186417fa3a99791d938be62 : SUCCESS in 4m 46s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/3571797d7cb449319ed6e2522316fe92 : SUCCESS in 11m 20s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e01949811c7848e28c33a0a5de1d0ea8 : FAILURE in 5m 17s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/41408a0a4b58462f93c5a174dab00950 : SUCCESS in 11m 26s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/3fd3449f5173472e9df41f577d36e791 : SUCCESS in 11m 31s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8386c4153157467e9ecb6af8dea8735a : SUCCESS in 9m 50s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/2b4abbb3f7d84212b77378c09a7cc2ca : SUCCESS in 11m 35s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/b892ed837bb74e0bbf350cb0af65470f : SUCCESS in 19m 03s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d446d1da3afc4c8280eb15de6e853e3c : SUCCESS in 18m 02s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/d11eccd56103495cb092546223b247d1 : SUCCESS in 15m 50s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d5954a65e3e347faa71d95660aad5a5d : SUCCESS in 16m 06s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/e449ea076d5e426abf893beb18b81299 : SUCCESS in 1h 11m 49s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/ac8909f433ff4ce882df4db013b74ab7 : SUCCESS in 1h 10m 58s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/05aafc47b37a41b6ad3bd39d273a97cf : SUCCESS in 1h 07m 33s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0d9d31457e7040929d1cbb1f70dfd111 : SUCCESS in 1h 11m 23s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/c4edc582fd894557b01cdf00625cd346 : SUCCESS in 1h 14m 50s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/d53fae191d5642cb9f965c08064c2d9c : SUCCESS in 1h 41m 30s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/eb8f9a28f42a45e6ab21b9501a796a74 : SUCCESS in 1h 55m 10s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/65b0cc0f6a2b4f98b1faefbc4ac54df8 : SUCCESS in 21m 21s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/c190968a0f9146129be813c97ddf1356 : SUCCESS in 21m 34s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0116deb80bbd421e9a1092528a096c64 : SUCCESS in 22m 23s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/4785ac68f38f46a494ec209e10315b5d : SUCCESS in 18m 10s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/8f173c0358f14b9ca19385b3d5bf1329 : SUCCESS in 31m 23s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b5892b6538bd45758c64cbfc34d3084e : SUCCESS in 33m 18s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/dae9f12e75db4c6cb9e6c060277b0eb6 : SUCCESS in 20m 57s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/82e988b3d069414d8cb4f2c271c1d1c7 : SUCCESS in 19m 58s","accounts_in_message":[],"_revision_number":6},{"id":"e935faa27cccdbdabf579930f65cc91d9eb58dee","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2026-07-27 08:14:38.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":7},{"id":"97ac22fbdfee8c9e46a55bd1b90cad4abddb55a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-27 10:10:04.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5d9052c50b4b4ad796d0834597dac80b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b35adda405c84acf8d46154d2ad72b5b : SUCCESS in 8m 08s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/8d754e94c6864158a5751433590d2cda : SUCCESS in 4m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/92f40a88e7c4417b8aecc0a44dd6f470 : SUCCESS in 3m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/77f91e9b40b54e4ebd3ae1ad59246203 : SUCCESS in 3m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2739d8a1e5ba4b65af7a7eabccb0f56c : SUCCESS in 3m 09s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f3e8f76e52d14a55a4d549cae2b5d2ef : SUCCESS in 7m 34s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/484dece2ca2e4adfbd7866cda7497c65 : SUCCESS in 5m 50s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c68cd208da0f4750be397b2c5fb94919 : SUCCESS in 5m 00s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ebee008ebe7a4ba2b3d53cce6559e066 : SUCCESS in 6m 22s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/9c83083e9f8e44cdab59f10c25d7d85d : FAILURE in 3m 53s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/cc4c21d0f86841ad8423cfe9253b1925 : SUCCESS in 16m 17s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/c85aa21783664fa8a281318238c3ff38 : SUCCESS in 16m 51s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/83c093b78ad44a7b9f687c495de38e4d : SUCCESS in 15m 57s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/08df446b60bb4c7e9e15a0f07b1c1cd9 : SUCCESS in 16m 59s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/11605cf59cb946faadd3a6d5661566b0 : SUCCESS in 14m 02s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ab53ac023fd54c049e4195bbd87d2244 : SUCCESS in 19m 11s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/cb6c99264ce9445bbacbc2bbc30ed9f7 : SUCCESS in 24m 04s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/33d2c1806d02411683138f57545d351e : SUCCESS in 18m 14s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/4164b99c05184b3fa4b19498fcfcfbf9 : FAILURE in 1h 19m 01s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/88039d5855ed450da4ba42a6a697480a : SUCCESS in 1h 09m 30s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/0ad8d016588844b4a2cd23cfdc7afe79 : SUCCESS in 1h 10m 56s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/adbbdc4a2ef44c79895f5ceb4953862f : SUCCESS in 1h 17m 06s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/a2fcb3325a7840198754da5d31128801 : SUCCESS in 1h 15m 33s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/9738e46b827b40fb9977835759d509f6 : SUCCESS in 1h 49m 30s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0f1148f27baa49d89d1376ecdc1c3727 : SUCCESS in 1h 54m 21s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/8858341670334fe2abc64101bf12d4c1 : SUCCESS in 26m 48s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/201a0b38b7ff446c9ef1c6c86cb721bd : SUCCESS in 22m 36s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cd392e13550f435eb237c6ca92e0b61c : SUCCESS in 26m 31s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/89a2e089de0d4f2196ae488799c315b8 : SUCCESS in 22m 01s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/364c04eaa7fc4cc095830e211118e8a4 : SUCCESS in 42m 33s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0740e19e90e94e5fb37996465383b876 : SUCCESS in 41m 09s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/2d8abd0971ba4430b3627e0b342f237e : SUCCESS in 27m 35s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/591155febc974f34a805ceb6eb2161ea : SUCCESS in 13m 34s","accounts_in_message":[],"_revision_number":7},{"id":"109cf73adf76152acef5e93ad969fd68429bc58e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2026-07-29 08:31:51.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":8},{"id":"c96236285c1d9162509fe0e1bc50a44c2c501296","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2026-07-29 10:12:02.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"7f4ee56dbdc4925d85415bad085138ed2f912eaf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-29 12:54:20.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/33c2664defd84dc59a26407e750985b4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fce0c51ec6614a74a5d81273bc06f84d : SUCCESS in 9m 02s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/0e2607c86c8b4ab7b0a1c9a67a305205 : SUCCESS in 8m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e1be4d4af6da4aeb81cc7719e294d6c8 : SUCCESS in 4m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e76f998f8a24ee5b924bd44ca04f739 : SUCCESS in 4m 10s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/954af5f1117946f8bd5b6e62ad2ff5bd : SUCCESS in 4m 07s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/1888375ed9cb49668753dd5978842723 : SUCCESS in 7m 31s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7980451304c84c7f9aa9f79eb12edb6e : SUCCESS in 6m 26s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/321d91f20d7a4b02b95c53af18fa2ad4 : SUCCESS in 7m 25s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6dee98a68e75489891a4264fb5ad7480 : SUCCESS in 11m 41s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c8cf61fd2787437094167c0ea9efd557 : SUCCESS in 12m 23s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/da006588aa4743a5bec0acf72f7a76a8 : SUCCESS in 23m 29s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/6f004ab9f4fe4b398ff04ba32713180c : SUCCESS in 21m 24s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d8b74af300ef4efab690c08ef57d8220 : SUCCESS in 16m 25s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/2f0d0673eb494d8092d3937a0461f780 : SUCCESS in 17m 04s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/d244d06c9a1f4e2fb7b2380aadd866e2 : SUCCESS in 25m 11s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f812762cce23438d9d9c91ceac94ad4e : SUCCESS in 18m 56s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/f9588add43da43ce9d3b8273b8988a2a : SUCCESS in 23m 41s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5439e1107ac1475cb8f31d4708faa497 : SUCCESS in 17m 50s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/e37e421c4a18486fb5e18f68c2bd3390 : RETRY_LIMIT in 9m 25s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/fb1c4674525d4983b714452b1b6e2364 : RETRY_LIMIT in 9m 42s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/246dd146909e46abb6bf9407a1ad22fc : RETRY_LIMIT in 6m 31s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d5a0a5a0d01441108cc708b025bb3887 : SUCCESS in 50m 21s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/49367db473da4be6b8defb24f359e195 : SUCCESS in 1h 24m 16s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/620a7212836f4d17846de18f4611ba6e : FAILURE in 5m 44s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8d691638c43b45b2882979c8e454f941 : SUCCESS in 1h 27m 35s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/62055446430e4d0ebb82266353265d16 : SUCCESS in 28m 28s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/30c5742e11c2442a94d9d42cf912ce88 : SUCCESS in 23m 44s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/da6639f147714626b6a3fef214d55157 : SUCCESS in 22m 25s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/2c4525ed33e646a0b935b77356a7de1a : SUCCESS in 21m 42s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/cfd29d58c4f14a41acc4f8815790cce6 : SUCCESS in 39m 16s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2afd950f034045fcaacc029e1b1f5206 : SUCCESS in 37m 16s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/e200019035b94a5ab70e83f2cb0379a2 : SUCCESS in 26m 46s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5bcc5b39a3b844b49a99894ec10a2c9f : SUCCESS in 14m 06s","accounts_in_message":[],"_revision_number":9},{"id":"34c17cd0ab31f326f04099803fbbf642981c5f2b","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2026-07-29 15:25:20.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"0bb47bca1fde969e64fa45777dc3bd9980e15952","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-29 17:51:39.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/00453235a3404dcd8faf030f1bb2c852\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9970c962bbac435d945b77b0dad1c6e7 : SUCCESS in 6m 35s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/b098dcdae9ad48b798c19c05e1210ace : SUCCESS in 9m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/411a25dd0b594668b2a34801b932c7ae : SUCCESS in 3m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/16fcc486e0044bb79742630baa816227 : SUCCESS in 5m 09s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6b99146692ab4fedafac677f0f542ff7 : SUCCESS in 4m 15s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e6887e6cbb76426cb13b407696e01b16 : SUCCESS in 6m 02s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/148d676a3ac44ec4adaa3e5d5d36cd8a : SUCCESS in 6m 07s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9e554a53f0ae4013ae9af571dfa540b3 : SUCCESS in 6m 20s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d263fefd584445588f6ac8e567d06158 : SUCCESS in 15m 17s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1298add5b626442d9362bcbe2df49cea : SUCCESS in 8m 57s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/214f972d6ffb416bb0e2dc0c6d8e85d2 : SUCCESS in 14m 59s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/8cc5483e8f074cef893c0231a50f00c7 : SUCCESS in 16m 42s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0400c7e2842749059ff871d49729ee27 : SUCCESS in 11m 59s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/7414f40bc4a04cdb852aef05c0cf87e8 : SUCCESS in 19m 13s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/e4933c77fa2743c4ab7a978d8b85d4cd : SUCCESS in 17m 56s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a93953f0b9ec44778217d3387063c1a9 : SUCCESS in 20m 36s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/4482bcce554649058061628945fb292e : SUCCESS in 14m 55s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3e15400ef2d1483d83704ab9894cb8bf : SUCCESS in 14m 16s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/6a607bfe1fa547d9929e54ed44b51e84 : SUCCESS in 1h 26m 33s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/67fe267e62a645318a7406284fc19979 : SUCCESS in 1h 19m 46s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/67d1f3b350f24b0ca6d69eb7f65b7bba : SUCCESS in 1h 10m 38s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a0a2ddf9ba24449f81fd80d5e0a53cf4 : SUCCESS in 1h 31m 16s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/ce5711f386b04a7b85781f7642071726 : SUCCESS in 1h 26m 43s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/800a91118f5642648969ba7ac6160195 : SUCCESS in 2h 03m 03s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f2c0f5d9122143fb916ac6e5a7052af8 : FAILURE in 2h 24m 09s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/b0a4f992ae1b48f3898e93476eaa675b : SUCCESS in 17m 54s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/fd1a853fc87244fb826e47543f1381be : SUCCESS in 17m 46s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/71bd38da9ff84aa39b4360d880562e2b : SUCCESS in 21m 14s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/1516e160a6404030a16fff23de377fa8 : SUCCESS in 24m 13s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/c72683c524ea42ec8ecf9380e015c546 : SUCCESS in 43m 06s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/28368ce09a1749b59ce711393abdb563 : SUCCESS in 34m 38s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/9ca96ebae13145b1b8e6e41815094c65 : SUCCESS in 17m 38s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1ade3f7be5324ba494ba37cf7ab3556f : SUCCESS in 15m 43s","accounts_in_message":[],"_revision_number":9},{"id":"94493fa38747a35db0329154e3e8e07216785dc1","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2026-07-30 08:52:25.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"f4da1a584a46fd5d14a08db332d53eb84c71b3ff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-30 11:11:20.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/00bc26f0728d47d2833996d49d6b89f2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e8751c88dbb4d42ad36072324962fcc : SUCCESS in 9m 58s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/49ee4a97f8714150be1599cfee7c04eb : SUCCESS in 10m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f2c148ceeee24062bdcb87be4476b147 : SUCCESS in 4m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a96dfc7d763a4aa3be399ddfd2580c9b : SUCCESS in 4m 05s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f766d9a2936647ae803f560ca754f758 : SUCCESS in 4m 01s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/fc1652666cdb4645a7e4b4c016e1af8c : SUCCESS in 7m 56s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/97aaf836aaae47d788b65a5ee5f2b1b6 : SUCCESS in 6m 52s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/20bf917f8a3c46259b0b38f032692dcb : SUCCESS in 8m 09s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e0957e94f13443ae89b47ac80c9ccd8b : SUCCESS in 14m 47s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/88e8407a2ae547d6aeac9da118a7b9ff : SUCCESS in 13m 37s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/0b3316c674794185ac10f143973c4e86 : SUCCESS in 19m 51s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/24b9335c178f43879034bc3dea00ed1a : SUCCESS in 18m 21s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2e04b10cbd664fefa0af4179315b078f : SUCCESS in 20m 42s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/ee014017631c4705af7bc2c68c91102d : SUCCESS in 22m 13s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/7ec265ec9c4f44b886c103cadfb7a37f : SUCCESS in 25m 15s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7b1684b9f65a47b5a74992a1e96a4f0e : SUCCESS in 23m 59s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/0f334e85d96947f0b42026863b067e53 : SUCCESS in 19m 32s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0f1774fe7f80416b952aebaee6874ccf : SUCCESS in 24m 35s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/cef47d55ae274e679bdf45a8014c7047 : SUCCESS in 1h 20m 45s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/c0c3860d5e694a019bfb99c74e36d681 : SUCCESS in 1h 14m 54s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/2a4b18f84a4445d9b4376bcb5f81a411 : SUCCESS in 1h 12m 20s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/acf597df9f594f3a98410988fe1274c2 : SUCCESS in 1h 23m 23s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/dd9ea2b970cd4dd984b45f87168fde18 : SUCCESS in 1h 18m 51s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/640162c84e4e4bf3af269797a1fcc3c7 : SUCCESS in 2h 17m 47s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/daff7b9cf4534c7eb2252785c9c2ca6b : SUCCESS in 2h 03m 48s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/a5f0a3cf6adc4b36b3eb4ba0bea20412 : SUCCESS in 33m 22s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/32142f7298544e58aae0bd65c903bb37 : SUCCESS in 30m 03s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/963f77aebd4d42b0846bf29ee387b607 : SUCCESS in 28m 50s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/8711952d560845ecb3929dd210f91c9b : SUCCESS in 27m 11s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/de8609f14e154bc48604c6c923eba5a8 : SUCCESS in 34m 52s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/179089d42b7e4baabb30fbbc308b21a7 : SUCCESS in 39m 50s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/2e7efff074274b888011fd880731c3da : SUCCESS in 25m 03s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2655b33266a54ff783172b74c2a9b306 : SUCCESS in 25m 50s","accounts_in_message":[],"_revision_number":9},{"id":"4c856dab4642201616d66d3212f59c3cc78cd636","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2026-07-30 11:11:54.000000000","message":"Patch Set 9:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":9},{"id":"e43be8a862aecfae16c896efc9318cfddb1c83ea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2026-07-30 11:12:16.000000000","message":"Patch Set 10: Commit message was updated.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":10},{"id":"8e90200127eae3d4defd2b45dd873a449305ff18","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-07-30 11:29:11.000000000","message":"Patch Set 10:\n\n(3 comments)","accounts_in_message":[],"_revision_number":10},{"id":"a92f66080eed5467c5e85d80bd748f4786a9ca32","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-07-30 11:29:23.000000000","message":"Patch Set 10: Code-Review-1","accounts_in_message":[],"_revision_number":10},{"id":"7c2a9696c4aa61a24ffc712c985d9be8060df714","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2026-07-30 11:30:48.000000000","message":"Patch Set 11: Commit message was updated.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":11},{"id":"632ad1dafc6dbf5c5649912c41f22a36b82b7e0d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2026-07-30 12:46:53.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"dd1eb9fb7168876b784b8df28946d37f8770e151","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2026-07-30 12:47:49.000000000","message":"Patch Set 12:\n\n(3 comments)","accounts_in_message":[],"_revision_number":12},{"id":"aa02ad8247313b9120d98b1da73c9cbd28df6cb1","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-07-30 12:52:41.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"1386b35b7c35f228fd5a565065a57f4432b292e7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2026-07-30 14:07:54.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"eea9b8d7b5051fbe9d5aea70a0ed5a54319e3055","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-30 16:09:30.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/27747513ee6b441180b8232e19a5520e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b88e79451f5244aa9c3f28bd1a1ac180 : SUCCESS in 6m 07s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/77b76613ac7e449d80f0a6a92176e89f : SUCCESS in 6m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1a21c7534b1c4947948047a959669c2a : SUCCESS in 3m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ae45b42e6174b8f9a23b3e059940c12 : SUCCESS in 4m 27s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9390b21bde294c7a84760eedecfc4611 : SUCCESS in 4m 36s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/48901969e7f74f8092782e459f28ccc7 : SUCCESS in 6m 39s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0307e72b77344d3aa3f9e22a1f7ee914 : SUCCESS in 3m 44s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/325a8748308f4cccb99100ed87ecc2dd : SUCCESS in 5m 31s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ade1d55a410c445bbc4bbd82e8b17214 : SUCCESS in 11m 59s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/839f746de773437d833576926af104bd : SUCCESS in 15m 25s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/b58458a2c245481e8984c38555ab1d21 : SUCCESS in 19m 24s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/92c3d8d4cb6a4cc49d32a9aa9bd30f2c : SUCCESS in 12m 13s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3a34b808726142adbf63a330b24a0735 : SUCCESS in 14m 03s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/ba216f303ebe4943aeaa891d349d3867 : SUCCESS in 16m 28s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/da3514028370424884dd685866623789 : SUCCESS in 15m 02s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0029ff9fc7d04db2861b8430e194e145 : SUCCESS in 19m 16s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/87edd2289d3841c5a0a4dd22930f6241 : SUCCESS in 13m 49s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/822b5849b1b448e99d436c059d228113 : SUCCESS in 18m 14s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/c2b58a5b80df424ca2988004df10b07f : SUCCESS in 1h 15m 49s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/fae170eff2d441d1a85eaa1ff29bb928 : SUCCESS in 1h 14m 58s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/5c9af1d80c2c4f558c1864bab02b81c7 : SUCCESS in 1h 14m 34s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3883404662344f3ab3695cc05ce523a2 : FAILURE in 1h 37m 52s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/2101f9a923244ac5b51b8ad7b329ef1a : SUCCESS in 1h 14m 53s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/864fa4c44cf341dab07e0341a9188ea1 : SUCCESS in 1h 44m 59s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/03a060fa1ccc4df0907a0909e7625276 : SUCCESS in 1h 59m 28s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/1d0901187ef24d299f9ea66ffca6db38 : SUCCESS in 16m 12s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/adc08d1419fd47c29e3e33147caf0ab5 : SUCCESS in 27m 25s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/534c723095e648e08873c0e94cd2ea89 : SUCCESS in 22m 41s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/f73e395db1ba4382948fe832f72f5f3d : SUCCESS in 19m 55s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/4d993d26fb3d482289ca984a6b368dcb : SUCCESS in 23m 15s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5a359d9b01e7479fa7afa3ca6067b1ba : SUCCESS in 39m 38s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/62740fe57e714546934032a8d21c5a3e : SUCCESS in 24m 47s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/335db2e8c8874685bbcdb359e3878796 : SUCCESS in 21m 10s","accounts_in_message":[],"_revision_number":13},{"id":"421ea54881e9bb716f0c8f8ae9ae82cca9418058","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2026-07-31 07:18:32.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"5a334e6e4c70fecf43b64691c7fe5a7a368b4421","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2026-07-31 08:08:53.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"32def098448cd245431b690f17f69864e2fd4081","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-31 10:15:42.000000000","message":"Patch Set 13:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/698383a9dd6447c889f0dc0c4e969991\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31435425b5d24b5d8076761dd0949df0 : SUCCESS in 5m 46s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/b61b5c82ba7145159a05b5cd607e1f9b : SUCCESS in 8m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6d977a9d4b32424a85098b8a2bd03416 : SUCCESS in 2m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/030eb0b5d244464088728bc93d6584ee : SUCCESS in 4m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/43b3af1e2a0549b1987d4f0f4f3e926e : SUCCESS in 4m 01s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b7337b3018124249b2fe755bb8682e75 : SUCCESS in 6m 51s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7995640550b84e49ba99c1a645c57331 : SUCCESS in 6m 41s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/618f8bd1e1324b4f84703904167eccb2 : SUCCESS in 4m 26s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/71ed251f9bfd457cb3096a4f139bc089 : SUCCESS in 11m 01s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b683a107f1bf47f98320b3899dfd4821 : SUCCESS in 7m 02s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/b1097fc9b4a745e69cab8bdfd71ae6a4 : SUCCESS in 12m 45s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/f855e7c4b2df47fa8866b27fff1abec4 : SUCCESS in 18m 48s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/434d95f568664d1b8534d9a64021b437 : SUCCESS in 17m 39s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/9bedd5334cb54272b87fbcbb852c214d : SUCCESS in 12m 59s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/8fc0ff054b5f423399c3a2bb271e3a64 : SUCCESS in 25m 04s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a5144789c8b24afeb27d813f2905c7f0 : SUCCESS in 20m 45s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/99db0a361c504c3ca44485183afc6cce : SUCCESS in 13m 06s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8a64147883a043b2a15687ff7174bece : SUCCESS in 16m 57s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/a495d1f61b1945c18d6543349517e349 : SUCCESS in 1h 14m 16s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/a4ba3755ea6642c5a2683287384d07c5 : SUCCESS in 1h 14m 26s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/5ad76fa2d6f14d038a78de6c3ad41398 : SUCCESS in 1h 10m 13s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e08da10febe442aaadbfcfd4962ca02e : SUCCESS in 1h 16m 06s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/f31ae1960f4942f78be5b898e6de08bc : SUCCESS in 1h 14m 25s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/fdc622facf3a40608dd0e8428e3c5949 : SUCCESS in 1h 59m 12s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/932239863dd346ec814e3a35bff008a8 : SUCCESS in 2h 01m 00s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/426cfd87c3ec4c56bd494c7b34f37b02 : FAILURE in 21m 16s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/98616322b37c443a832fdaf978819a76 : FAILURE in 16m 02s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a7b2cdf73b8a4d6dbf15edfbf75bbff1 : FAILURE in 20m 59s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/dd79162b4ec243d6a97d36a9f1936fa2 : FAILURE in 15m 15s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/d401946285164f6693f94db40fe2d307 : FAILURE in 24m 19s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/056ca941a474483594f8f2438a1dba76 : FAILURE in 35m 28s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/5bb558c4fc59432d836bf7d3985819d5 : SUCCESS in 19m 39s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6de8ab495f5e430da99cae41de61dec3 : SUCCESS in 21m 24s","accounts_in_message":[],"_revision_number":13},{"id":"242a8d66bc73f03309c026e9cdf129f8391e16d3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2026-07-31 11:33:09.000000000","message":"Patch Set 14: Patch Set 13 was rebased\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":14},{"id":"8da4e2afad5a0cb543f5064b8f5ce815f5bc4ff3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-31 13:34:49.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cd75223b1d99459399e3a3c9c9eb10ef\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04c6f65d75724d46affc705c511defe0 : SUCCESS in 8m 02s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/37c2f720791548e8ac26432d416c1d75 : SUCCESS in 10m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f2c081f23b764ad680532ccf16645d91 : SUCCESS in 5m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/be305823aed44ca1a4be98cdaeaf7329 : SUCCESS in 4m 39s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7d9184c5d1b54717943e9bc81759210d : SUCCESS in 3m 03s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/9b2e9f212d83484e8bb1b9336d9e26cf : SUCCESS in 6m 39s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a2b9655b67a4487a9bbca9653c061f3e : SUCCESS in 6m 31s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e59d3b0e28ae4e69860a7adce832fc25 : SUCCESS in 5m 43s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c3b9fc10cdf54057b8f872fa0b09bc3f : SUCCESS in 7m 19s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3f8763209d454bcdbbffc1439c5733c6 : SUCCESS in 11m 47s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/98ff13b7a09841a5a9e63eeeb541cd0e : SUCCESS in 12m 25s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/e1ef3c51b20446269384cd895dab70db : SUCCESS in 13m 03s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9b9e6f069ee74f38adf0ddfc840a8845 : SUCCESS in 16m 54s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/a093eba0e76d4fa489fd554cb9d77017 : SUCCESS in 20m 24s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/da33ff7d033b4d939ae7ef1c5d426406 : SUCCESS in 13m 59s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7e646ea9672d4d3ca94f9faa48eae747 : SUCCESS in 21m 23s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/042b1ab42b3849289d274235da195a23 : SUCCESS in 12m 31s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5ceb4202d4864e3ea9ec065d6324523e : SUCCESS in 16m 48s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/850416ec8ee442eb9edc355aaa8e281c : SUCCESS in 1h 13m 37s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/f5c853e360534c25aba4ad509b744b0d : SUCCESS in 1h 09m 50s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/fbb4151d350f44379e91db98782db1ea : SUCCESS in 1h 10m 41s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dcb638111d92457081f7438803985c4f : FAILURE in 1h 18m 52s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/ae69e02eef994927aa7688a49e3d17ca : SUCCESS in 1h 13m 16s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/84d837cd5b1f4b68b4400a73f4849407 : SUCCESS in 1h 54m 54s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6d3fe0854cbd436484034dc2720852d2 : SUCCESS in 2h 00m 01s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/4fcb8baf88334e0b86bc080dfb56e605 : FAILURE in 15m 41s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/0fc4aeee4efa48ac8380748e8c04ab1d : FAILURE in 15m 36s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d8254b0df8934ceab6fe2521749e185f : FAILURE in 21m 57s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/c713fbb5380c4c9ca3e9caa8190e75bd : FAILURE in 19m 27s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/082590c9dd454aeaaf595a838712372a : FAILURE in 20m 58s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2d2aeb3e01734834902425ba1b6b6e46 : FAILURE in 43m 04s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/acd4f135ba774d22bef6376011fcf2a1 : SUCCESS in 14m 25s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1cfe5aec478f43d6a1f40fe0a107e5d7 : SUCCESS in 24m 07s","accounts_in_message":[],"_revision_number":14},{"id":"af1dae5a14dbdf1b33326ad7cdf6da1b13cb9cb7","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2026-08-03 08:54:53.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"4c3b129f6faf58b0f49077a5ffd7c5624d00237c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-03 11:08:56.000000000","message":"Patch Set 14:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2cf04796d244455596abfff46fe6db1a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6edb8a8754f440db8e01b39776e70ac5 : SUCCESS in 5m 07s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/60ca5f579b104f22b3834bc63d7beecf : SUCCESS in 8m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/edbc24bbf64c49fcb28dad5a024788df : SUCCESS in 4m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af85bbb08c8442c69be307f4a7958a19 : SUCCESS in 2m 55s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b185c0abc9f14f39b4d3ec890990ae81 : SUCCESS in 4m 21s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/932cd55c0154463a95188adae6faa3d1 : SUCCESS in 8m 37s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ad97b599689b44bd884b01c2fa16e184 : SUCCESS in 3m 58s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6c514a03c1e643919c6c926c8a37107c : SUCCESS in 3m 31s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1b0d4b7525e14cf2bc4884c9fe909a57 : SUCCESS in 13m 45s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1bb3590f88554aa890544f90963ede59 : SUCCESS in 13m 02s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/a37262b3dc2f4bc4bc0e5227a0117dc1 : SUCCESS in 20m 11s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/162e9eb3657b48f19dbe5964967a40ba : SUCCESS in 13m 15s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1920137fd782408b858f91f1f6584578 : SUCCESS in 10m 43s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/70531803b4c64a84bc5ccef0b732213b : SUCCESS in 16m 33s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/fce25ef4d3bc4e0f89261d67707b9da0 : SUCCESS in 19m 11s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3bd6c118c846490984a34c854c7d6141 : SUCCESS in 14m 12s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/ee475717bd6a4574b499cc2b657f7ad7 : SUCCESS in 12m 46s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bdd8bbfed2904f37815f921c40e846fc : SUCCESS in 19m 50s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/44e27f7f7aa243058672c1dc7e36ab98 : SUCCESS in 1h 16m 45s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/9626b49865ae41f7bbec1ad2dd552943 : SUCCESS in 1h 10m 32s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/c39c78e040f54012b031ae8b21682c40 : SUCCESS in 1h 12m 39s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8f4723f3bce74e5790e6cddc46d5bfbc : SUCCESS in 1h 21m 53s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/5bf79aeecde0477793164486b32a4041 : FAILURE in 1h 32m 44s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/c29438623b044cf4afd9ba42fc324c63 : SUCCESS in 2h 09m 59s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b3b717ed6f094ea8b53bf1abc19abeff : SUCCESS in 2h 12m 40s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/82df4b4226d144428dc5e1496da3ec59 : FAILURE in 14m 45s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/b235f1c35adc43ca835601a7498db3d8 : FAILURE in 21m 58s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/848f3c936e1e4875ade58816625c95bf : FAILURE in 13m 09s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/28d7fcaf33d8492a802c1c1cd7237a09 : FAILURE in 18m 50s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/c5d8fde73b1b4557b752f024d3e8577b : FAILURE in 27m 55s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/12bb05b9b0f84d748e40ae05c32c08e9 : FAILURE in 22m 20s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/034a32a84af5420782f9e886cf9745a8 : SUCCESS in 11m 35s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9d560d8e6ec24ef18d1288aa4df1ddf2 : SUCCESS in 22m 27s","accounts_in_message":[],"_revision_number":14},{"id":"3bf332456bcbc976dc621d3001a92fd07955d64b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2026-08-03 13:04:01.000000000","message":"Uploaded patch set 15.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":15},{"id":"22ab53d79f8c4c5818401990d7a886f78ecc7382","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-03 15:06:35.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5a4b2afcfb464b2cb7345dae1ae5780d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18b56288b388440bab74e9cd958f8f4b : SUCCESS in 5m 06s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/6459106c164c413dac9ad559d0698b8f : SUCCESS in 6m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/93e25eb2820147a78bae3604faab18b7 : SUCCESS in 5m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d689665a6d5c4f5c998dfd596a864a71 : SUCCESS in 3m 50s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2cf52f867bb144889667a6f4b50ce25b : SUCCESS in 3m 53s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e6efd974603c46d58d41c88839cd207d : SUCCESS in 4m 07s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4f00bb1749ab40488715edc209eca716 : SUCCESS in 7m 27s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a4ab6d2ab4be433c9eb95eade2ff7728 : SUCCESS in 5m 02s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e3e4a8af9e6d4ed88778fa6d1bf671f7 : SUCCESS in 7m 34s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/89201bdddfb143e08c9138b5a239af8d : SUCCESS in 11m 52s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/00d0f831d1ac4ae38bfb27361a32390d : SUCCESS in 11m 38s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/906929a352a74cc2850976e02f7cd3fe : SUCCESS in 11m 40s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d0a6c4b18949489a98a339b6318631be : SUCCESS in 16m 04s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/f236d62f54fc415c8b9bfc317c245fdf : SUCCESS in 15m 11s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/8992c7ed2ccf4ee6b338512a48b10152 : SUCCESS in 14m 49s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/44d87c2ee16649f987549bae2a0c6aa9 : SUCCESS in 13m 07s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/ab100f15e3c648788c6cc26ddd16980a : SUCCESS in 16m 39s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a6b275ecb6db4e4aa85a316bdf7ee81e : SUCCESS in 16m 31s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/0afde2b7adf74a3cbc27094982e37826 : SUCCESS in 1h 07m 03s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/81c5e6698fe4423a8f646d83ea12f5cd : SUCCESS in 1h 09m 43s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/d72f031945374eeeb55af887eec61acf : SUCCESS in 1h 06m 59s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7200449b096c41cb969b34ce269f57ec : SUCCESS in 1h 16m 03s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/a21324bc669745f2b7b10b718b6b2044 : SUCCESS in 1h 16m 15s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/6e9d0209f2674a0297985d7c327a7dab : FAILURE in 1h 44m 32s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a61aab4ca9c64488a3f3dbcbd92fc3c2 : SUCCESS in 2h 01m 45s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/0cc2cdb308114b3e93eaae64e96718a3 : SUCCESS in 18m 26s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/ab8d0c7b682c4928b8bd1b90b01a96c3 : SUCCESS in 24m 13s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/05901c3cafef483788d0a591eefa1ae2 : SUCCESS in 24m 09s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/645e377f56e74b50ba7faef26e1526dc : SUCCESS in 15m 00s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/424bce8b31c040549a61b7c80a20b2ac : SUCCESS in 37m 54s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/09af22a4f50945a488bcc26f208836c6 : SUCCESS in 39m 17s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/0c3a85ac805648c69af8e4ba08b294a3 : SUCCESS in 16m 17s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0efa6b952aa748f0baa898d449345e2c : SUCCESS in 21m 55s","accounts_in_message":[],"_revision_number":15},{"id":"315b826e221a1e36e869bae26c4546b77c37ba3c","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2026-08-04 16:23:44.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"b887028c9e907dc6d9c7ab912fc07df376b9b3bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-04 18:21:57.000000000","message":"Patch Set 15:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2496ee768d41444ba7c60df6820a3ddd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c9a23538b8f41688d9a5db8d9d3586e : SUCCESS in 9m 26s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/a9b268e23525412686c680c7c35ea9e5 : SUCCESS in 6m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fec54a07c4f54a15aefb060092b875cf : SUCCESS in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/077230697b9947788baa66b60f90cff7 : SUCCESS in 4m 21s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/696c9e545b64423a8f0a278ec6364ee8 : SUCCESS in 4m 11s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f33f419a727b44818629130de6cd7c0f : SUCCESS in 6m 32s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4805237fb2454fb3af125f7c7871410a : SUCCESS in 6m 43s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/832d4383bb3c4de69b3bf0ad0ab37aa4 : SUCCESS in 4m 55s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9d6b092d7b86487e8bc2eefd550a095f : SUCCESS in 11m 32s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/cb5f93ae657d417a951eadc7cc9a346d : SUCCESS in 12m 59s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/ec192cccfce348ca89b76c657ef5852c : SUCCESS in 15m 48s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/32b2c7d3c5924eeeb66b74670dcc7eef : SUCCESS in 19m 10s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/65796808bcef450dbd7c206ca8820be8 : SUCCESS in 15m 06s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/3754337611b64192af85b45f08fb925d : SUCCESS in 14m 41s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/82f5b94f81084bc2b5679082569dae99 : SUCCESS in 14m 28s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ad1c6e10b3004116a000241af6a4bc84 : SUCCESS in 18m 36s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/5eff3dc4016f4f5fbc0d04a981ba4a31 : SUCCESS in 16m 38s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dbd7e723fe9648b3b40248ee293f52ad : SUCCESS in 18m 29s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/052a8be2db664e359682289178a39fae : FAILURE in 47m 27s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/6600d0f4f5464bf78cf418514b680bb0 : FAILURE in 51m 47s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/84511f189e2b451eb9e4fe6b659c3c09 : FAILURE in 59m 43s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1824ed80045d42d6b28f9447c4eb262a : FAILURE in 49m 50s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/e3c119a3e1424728ac662b26c30bb797 : FAILURE in 52m 07s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/5ff33a2a2f6c4540a87a55c7593c0423 : FAILURE in 1h 55m 15s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b79b04c2534a4260a31fd0791cc9e8cd : FAILURE in 1h 55m 48s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/2dc74f5608be4430aa95462480ec654a : SUCCESS in 15m 27s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/188f1b359a964a969dc2bdabb888c925 : SUCCESS in 28m 20s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/86298dfe291f45daa55a11feb7cb509c : SUCCESS in 24m 03s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/5db01323e8a24584b37c45d2de4c392d : SUCCESS in 21m 34s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/61e1a7d1b68443f28475e6f39ed6df9c : SUCCESS in 39m 42s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1c9c6dcab42143cb8d6deae1bcfcd99f : SUCCESS in 36m 17s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/df95c20ddb1747f5935b70f4d4ba7a03 : SUCCESS in 14m 34s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6c0fc0bfd25040a689884cf6a5585fb3 : SUCCESS in 20m 01s","accounts_in_message":[],"_revision_number":15},{"id":"316702130b344889d812a02a9e24ad7b4b9aa8b6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2026-08-05 06:22:11.000000000","message":"Patch Set 16: Patch Set 15 was rebased\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":16},{"id":"0a7bc6d6581e3a0d377a728da2564a05bfab00c6","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-08-05 07:07:28.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"6f2487f143d0c3f5e79f3017b710e143863e9d1b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-05 08:24:02.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6c08976d5e4840acb0f18e2b3bf2a273\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27a927010807419dadd52a430c8f25c0 : SUCCESS in 9m 36s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/ef05747e81ff42588bedb2ee0ce734c7 : SUCCESS in 7m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d2c9fff9c800454fa5e9be5b9391ccb2 : SUCCESS in 4m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d91f5b61673242d294c0b35abe846f55 : SUCCESS in 4m 20s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3eb0a54ed87940e688daba0df5f33377 : SUCCESS in 4m 01s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/abcddd450c1740888423fc689136d4d1 : SUCCESS in 6m 21s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/84c4ab7e64be436b9858c841c5f3616d : SUCCESS in 6m 46s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/578b698282d441879e5835d2d48b3e35 : SUCCESS in 5m 10s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f53bd5affb614b7c88d04896ad3bf7bf : SUCCESS in 11m 30s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2200d77c8cb94bb58ea87b59f9dd42f7 : SUCCESS in 13m 10s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/d1031db8630f471a8da47db73760f136 : SUCCESS in 11m 35s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/8390c09019e74afc8eff5c7c0ca80ba7 : SUCCESS in 10m 32s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b0795db9092a4c1ba8e1dab06c15ebeb : SUCCESS in 16m 25s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/31927215f62c4cd8afec64ecefd352ea : SUCCESS in 17m 56s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/16176560800947378162d1582dd0b362 : SUCCESS in 26m 34s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5e571b9b59c542e48e8fb41093977a7f : SUCCESS in 19m 55s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/e9d7581d67cd4a6b870dc1383d22ad53 : SUCCESS in 23m 42s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bb5c5a4a4d21427983de7592d445e60c : SUCCESS in 19m 46s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/18eb6237e6c24798bee1a5483cc9c3c1 : SUCCESS in 1h 09m 31s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/eea4477619c04463bc40b0af4d3064ee : SUCCESS in 1h 07m 57s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/eacf680e15784cc7bb390c44422e42ef : SUCCESS in 1h 09m 01s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2e85355966384e42af39f38bce0aecbe : SUCCESS in 1h 18m 17s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/c2f9bc87be9c492e8aea9f8aebf5bcf8 : SUCCESS in 1h 19m 14s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/78dc74e104c548f888cdfec4ab81033c : SUCCESS in 1h 55m 13s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/432f4377985043e8914fc78676637fc2 : SUCCESS in 2h 00m 09s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/0b313ad1a2bb4740ba26d6f32c63fae6 : SUCCESS in 24m 50s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/602a5076763143f488b5ef62dc183913 : SUCCESS in 29m 35s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d643e562f5b349049d753aecadfcbd32 : SUCCESS in 24m 29s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/3d6002cf368e4f77832bc44b82d6d820 : SUCCESS in 22m 28s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/c4ba0770300d4458989bd9be2651284b : SUCCESS in 43m 08s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2ca8845f3129483a916afc0acf842ff3 : SUCCESS in 36m 48s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/95388d014fc940b4aa24e8bc0eb34f9f : SUCCESS in 18m 47s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d79749e1133d4df8928e6b2bec8dc84d : SUCCESS in 21m 51s","accounts_in_message":[],"_revision_number":16},{"id":"88d2cbb445fdac2f368919122db98590e9050423","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2026-08-05 10:13:59.000000000","message":"Uploaded patch set 17.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":17},{"id":"1a6a0fee72102784d12e476a4b3069466fff0fc4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-05 12:16:09.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d92f81f513c9483c8c2b7fe81ab49d71\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98ee3bed007a45d588530e5159c3905e : SUCCESS in 8m 57s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/55349aba05834ef3aedb7106afa58499 : SUCCESS in 6m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/928daaec28db46cb9a1095ebf56b4a4c : SUCCESS in 5m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f37860bdee5545aa9cc5918e970dba4b : SUCCESS in 2m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/34ade70baf504ca99e14b52d2c4a1be8 : SUCCESS in 4m 15s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/130ead2c02e64ee78f869b2b463de678 : SUCCESS in 7m 18s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d0020480e2f24a9ebacc8ea8f36eed73 : SUCCESS in 3m 34s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/603e1763a20c45eb856edf26c9ac07c1 : SUCCESS in 5m 20s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/da056476b1304c3190b79fbfd78f02b5 : SUCCESS in 9m 11s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/00974af40bb34609a067a8686e243801 : SUCCESS in 12m 20s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/aa69df5b504f43bc94e664bea3ae682b : SUCCESS in 14m 00s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/e123177282be4812a2643232b9ff7aa2 : FAILURE in 8m 02s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fd2540b888de4e2e98ed6087dc97c271 : SUCCESS in 10m 53s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/b89a1cd48eb948ffbc852f20d7dc9189 : SUCCESS in 19m 50s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/0a604fe217a44beaa26ef8b5bb464118 : SUCCESS in 16m 13s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1d4dc1f4ff264893a96ae8e6f68a6bd8 : SUCCESS in 18m 58s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/4299fe109dc1442d8df494cf17aed95b : SUCCESS in 23m 30s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cef14ba1ab79424ea5ebcf67ce9ec30a : SUCCESS in 19m 01s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/d594fc63947140a78001ae7f7eebb75e : SUCCESS in 1h 10m 44s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/41da555473cf41f8bd9083c2ca024f7a : SUCCESS in 1h 11m 06s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/1da582000dd54666a6b05ecb91716d51 : SUCCESS in 1h 03m 06s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a2f703fd6f864e6a995288e5c388078a : SUCCESS in 1h 18m 56s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/777f2f62293e416da469b21747c2c261 : SUCCESS in 1h 19m 03s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/d2705a7c18c34cc6aa2943cb11093dad : SUCCESS in 1h 49m 42s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a8ccf14cd0e44aed8a57af57aa003e9d : SUCCESS in 2h 00m 38s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/2f815b04c8c74c92b7716b08ccc87bc6 : FAILURE in 17m 13s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/8df34e5919d941679462364ff1d95a6e : FAILURE in 16m 05s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b93bfba14d4c40a6bf445f95ffe010c3 : FAILURE in 20m 58s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/d431bfe2c38041ccb1d66d6039dbc695 : FAILURE in 18m 14s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/060d521de6f24e9f8d51d271df8ab34f : FAILURE in 24m 23s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fd21a7a9a04d43189c83a52da2034dd5 : FAILURE in 34m 40s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/c004c14646f848eabd0aa345fcb1b5f8 : SUCCESS in 26m 12s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/aca58a195cfb4105bddb1b703ce093a9 : SUCCESS in 15m 22s","accounts_in_message":[],"_revision_number":17},{"id":"6825c72d269ee516434accecd8424c648b63da03","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2026-08-06 09:45:32.000000000","message":"Patch Set 18: Patch Set 17 was rebased\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":18},{"id":"c1c23af67a1d970d88158344ebaa4439e127a074","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-08-06 09:54:27.000000000","message":"Patch Set 18:\n\n(2 comments)","accounts_in_message":[],"_revision_number":18},{"id":"60af71d5bb761d61e50cec3fef08b684e88358a7","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-08-06 09:56:07.000000000","message":"Patch Set 18:\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"a254ccafa1b5b1ca298553e1b659013e63bd6641","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-06 12:11:24.000000000","message":"Patch Set 18: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/642b384602d0434f89631803c29a95d8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/935421756133430b82dfd1e9f2a7f1cd : SUCCESS in 9m 39s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/0cc5a680f1864fe19a68c7a278cb6e4e : FAILURE in 4m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e52c462b9351495a9995a8fc2bb0ca3b : SUCCESS in 5m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3460724bc44a44af90276cb2486e58c1 : SUCCESS in 4m 55s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/15e8d77523b341b49edb6715b02b2339 : SUCCESS in 4m 10s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/59c5416bc6d447159202a7d003b335ab : SUCCESS in 6m 56s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/50a94331c7d34ae49f8f35343d137173 : SUCCESS in 5m 56s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/415101a6fc214ddf97ea1b5e3b30d4f3 : SUCCESS in 6m 34s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/11f7a03d93bf45f0807d8a920eff107f : SUCCESS in 17m 05s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a47047cbe8354b1c83365b8aa8bc0f25 : SUCCESS in 15m 48s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/24eeb94bbf0543f7b388ebe38f4efba5 : SUCCESS in 24m 08s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/4a08ddec9a6b498fa822821201c6e6ec : SUCCESS in 19m 30s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5d44decc71744c988addb170d613a218 : SUCCESS in 15m 47s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/38303e3055d541f5b2b380e3d3de1856 : SUCCESS in 17m 11s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/1582d385ecfb42a8a219b8154cfa02ea : SUCCESS in 25m 19s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/715dad6f615e4b519250e8f076771c2b : SUCCESS in 20m 42s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/9740fac5b7024cf2bc27856dca4b491a : SUCCESS in 17m 43s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5050fd534cc04883ba10c4edb927fa00 : SUCCESS in 22m 33s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/35e871406ecd427a9215b365cd99a642 : SUCCESS in 1h 16m 51s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/4c9387bb8c364e3688b391ffb4a524b8 : SUCCESS in 1h 12m 44s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/3c61b3cae44240bbbe9983e51d72e35d : SUCCESS in 1h 12m 32s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/910b788cb5f446fd8d39729b4931c119 : SUCCESS in 1h 24m 11s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/b182cd29fc794d06ad0f6111b77fec94 : FAILURE in 44m 32s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/3615c7069176429dbc7c002d60989cf0 : FAILURE in 9m 40s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/70c1e959069f426882fe515c0b0ecf7a : SUCCESS in 2h 24m 22s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/5149de37f6f7415daeca867e0902843f : FAILURE in 29m 50s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/488e02503cc345d2811fb330a9faa0ba : SUCCESS in 21m 02s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/373fe667ec2c405f8220436bdf3b8d92 : SUCCESS in 25m 02s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/5edc674d2d7a4c5faf9ee981ca0707e3 : SUCCESS in 23m 39s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/03e89b3fa61f4939b11174155f7a3646 : FAILURE in 12m 32s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4cf593074ee4459eaa9f8a87dbd57a3e : SUCCESS in 39m 33s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/f5c919dff9f5405c9ad8384742cf5ea5 : SUCCESS in 26m 09s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/50a20c3d9e8443aa8e92818575233812 : SUCCESS in 20m 36s","accounts_in_message":[],"_revision_number":18},{"id":"450c70ff9ced6cb10c93fdbb2cb9e6c40bfbc556","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2026-08-06 19:34:09.000000000","message":"Uploaded patch set 19: Patch Set 18 was rebased. Commit message was updated.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":19},{"id":"dab458fb1cfbb94e3595901f27489b5148b1a78e","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2026-08-06 19:34:27.000000000","message":"Patch Set 19:\n\n(3 comments)","accounts_in_message":[],"_revision_number":19},{"id":"07bb8f57e83f49911f1d911160b2c25747bf1e35","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-08-06 20:51:21.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"ee1beadff3f371d26072f6f15dbc99734ea641ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-06 21:36:47.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6ac5559d7c874b44b16d2431f5f0fb73\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be13302b71d44c92a79f1aeb19a498e7 : SUCCESS in 9m 46s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/484b2732a69a4ea9a95725e2b32bfb32 : SUCCESS in 6m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/28e21ae85c5e4c6085d0dd898731db7b : SUCCESS in 6m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4174d7ae0d2a4e11845698cc6a01cf2b : SUCCESS in 3m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0a2c0ae8c003485b84c2b8378e85e7d8 : SUCCESS in 3m 53s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/d91ab66ec8b341bda42909e15949bcf6 : SUCCESS in 6m 57s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e6225b2360f141b4b418717e5f37fda7 : SUCCESS in 5m 42s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/bfaff582b1ea43f4a6c9b6b9152444b5 : SUCCESS in 6m 02s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/73c9359fb27f4de092c813e2c7e602d4 : SUCCESS in 11m 02s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/26c4b079582948aba2eace7f1038614a : SUCCESS in 5m 52s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/29bcb63c949f4a67995043163f73cbca : SUCCESS in 12m 32s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/a2f6b05cc7b344cdb51588a6f589e818 : SUCCESS in 22m 15s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/362056e1f7e3448388009ed2597683d2 : SUCCESS in 17m 09s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/4f0ede215cd94263bd54f381cae62e15 : SUCCESS in 19m 27s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/92328fb8ea0941a3990b1289e78516fb : SUCCESS in 15m 53s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7b1b037737724a78a59edc1dfa3fc4c9 : SUCCESS in 18m 58s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/9f682f38c95e4ca6ae55ba6e5057f5f2 : SUCCESS in 10m 56s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/952bca9fba754b54830cf8cd6be4092f : SUCCESS in 10m 45s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/bb0321b3f03c434d86b580573491e804 : SUCCESS in 1h 10m 41s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/c197c7173a9943ddb937cb1e8d7c0669 : SUCCESS in 1h 07m 54s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/d4f375e8922743749c0fdb7cfb36d24a : SUCCESS in 1h 08m 32s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4120b55d59dc4f86bd6b97c223a42293 : SUCCESS in 1h 16m 58s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/5b032526864746b0876c23f94f457119 : SUCCESS in 1h 14m 31s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/22c4b8b5f4724def85b178cd89ef9c14 : SUCCESS in 1h 55m 20s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/aa8f6e712b714249a5cde855ff0f6e08 : SUCCESS in 2h 01m 25s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/54868843fea04045a98a769447a72e7c : SUCCESS in 26m 12s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/c3397da147444845b1c90159d43cda1b : SUCCESS in 24m 21s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8dcdd009181946eca1a0c842579851de : SUCCESS in 16m 14s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/ab891ec44ed44d50a9b34b13307bb304 : SUCCESS in 15m 24s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/24651913b80f4c52ba60f5fc33cf74df : SUCCESS in 40m 11s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bac7d3915ee64907a0edef871e9eb264 : SUCCESS in 35m 55s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/c62fd6264c754dcf843bf781475b51aa : SUCCESS in 22m 38s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/affc5c6d01ce4cfd98ce7d71c6f05adb : SUCCESS in 19m 05s","accounts_in_message":[],"_revision_number":19},{"id":"43559349dbc476c8846d0fdcf44289e48ca5569f","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-08-20 16:37:25.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"130aac2e1c8b79d57381c9eac66373d3368a6216","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-08-24 06:52:41.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"81cc46c847648fae5ee62404d2e02fe30a6e73c8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2026-09-04 09:15:03.000000000","message":"Patch Set 20: Patch Set 19 was rebased\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":20},{"id":"9f4ac03bdab8afd63d18851475a3bb1375cb5d95","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2026-09-04 09:15:59.000000000","message":"Uploaded patch set 21.","accounts_in_message":[],"_revision_number":21},{"id":"fdd8039e7ad10779f9433d5e39fb786cd80ee229","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2026-09-04 09:16:15.000000000","message":"Patch Set 22: Patch Set 21 was rebased","accounts_in_message":[],"_revision_number":22},{"id":"35c5e21ca0ca59d17db186ef3053ce1e6d545b0a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-04 11:20:15.000000000","message":"Patch Set 22: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8330fc1faa014ee3abc8ac45ba200cf0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f53f1e86127d4c009e144fcd135ca876 : SUCCESS in 10m 30s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/65b27161e783493f9958480aa8db5bc3 : SUCCESS in 6m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7afe0d703a3f4525924c4bbc8dba4793 : SUCCESS in 4m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f8e86405b2ee46a195733d7154f3e8f9 : SUCCESS in 2m 30s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e6b09ccf7aa1489abb7c40f9a654aae0 : SUCCESS in 3m 59s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/1bc6a9ffd8504063becb7dc231a73c0e : SUCCESS in 4m 42s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/07be5d5fccb446bcab765dbf4fdd9b66 : SUCCESS in 6m 08s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/15e7a6c68c884d3387eb6a0f7981602b : SUCCESS in 4m 24s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e199010900e44a2181f8ae47c5311357 : FAILURE in 16m 14s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/039de729130b43889983182f5319f9c5 : SUCCESS in 13m 07s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/f08f92de09a4490988e7f9b9c966b66f : SUCCESS in 21m 04s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/73b6496c1f5747ebab2973a539984034 : SUCCESS in 13m 23s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0a8148677714494faec230e19487e947 : SUCCESS in 18m 24s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/443431b7b30c40f7b45f2f1cb9875f97 : SUCCESS in 21m 00s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/8a95baeb36714c538a5ca5d7479c8a7f : SUCCESS in 17m 06s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/90a4c8e7ef974c1195dfacbd74d1baef : SUCCESS in 22m 24s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/ce8d900078db490090c797bb5e0527b6 : SUCCESS in 15m 08s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0e7869ea0e8e40e59579734a61f589e4 : SUCCESS in 21m 58s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/b6c61bf9ab2244b38b32f8ec07932e21 : FAILURE in 1h 21m 41s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/dcfa4e6450754f41a0eb84a4d7f90eb2 : FAILURE in 1h 18m 50s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/8a2bf4e38e0348ef9b6d468af9c16c23 : TIMED_OUT in 2h 02m 21s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a3c12fe006d743c2bf0b50c19ef87a22 : FAILURE in 40m 22s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/43233901c89a4108aa494139a29a3e75 : TIMED_OUT in 2h 02m 36s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/a6f8d001a97a44e3b2b92eac651a0c69 : FAILURE in 1h 18m 11s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2ef0d06b0ee54eb1b9614aa8baa1c220 : FAILURE in 41m 18s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/444ac7e31676495eafc56ba230be5d41 : SUCCESS in 30m 19s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/147732c526684a46a8ea0f3b6104df07 : SUCCESS in 17m 16s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3a28441661384cc69bf6e10623478697 : SUCCESS in 15m 05s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/e29f25e3ed4b4fcfbe188ca93f861f8d : SUCCESS in 15m 35s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/e9454be9fedf42b7972fcfee03e956a9 : SUCCESS in 26m 40s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3a00520ba3564f99ae4d4e484f3c6bd5 : SUCCESS in 41m 13s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/16bd2f23d7a346c7893d478fc6542c4d : SUCCESS in 21m 35s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6f31540b69d3452fab438584decdcad4 : SUCCESS in 15m 50s","accounts_in_message":[],"_revision_number":22},{"id":"ab8e537829a4d74c0fb9d23ac85d348a29fc76dc","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2026-09-07 08:56:01.000000000","message":"Patch Set 22:\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"2a5021a72c7448eba508b30ef37450a292d4b0d4","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-09-07 10:02:13.000000000","message":"Patch Set 22: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"fe59a3fe0ce465e2da70ea90932a604ceeb3f9d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 11:00:06.000000000","message":"Patch Set 22: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fd358b2015c4453ba39505d508967b5c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26550eb8df5c473fa3712ffa115e1681 : SUCCESS in 8m 58s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/343a5c680e8b483d8e33e4fc9ae6c727 : SUCCESS in 9m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5606944638394285a2d44822799aab07 : SUCCESS in 4m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0664ba5ebecf4ce89e68b1dbe9575b1c : SUCCESS in 4m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1fbc243a7f6b4a3193982f304bb13b59 : SUCCESS in 4m 20s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b233241802714f5298a04b9fb6ea4e67 : SUCCESS in 7m 08s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/21f6001081c547f3a8ab31625067e511 : SUCCESS in 6m 26s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c350cfc6d55249b1b2f99078681c0e88 : SUCCESS in 7m 12s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0e255a7ea0b3430685eb23aafb99dd90 : SUCCESS in 13m 48s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1cc1f27012154603af6210db17c0a616 : SUCCESS in 13m 06s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/89ce65373dd34d9781b00e01ab76b818 : SUCCESS in 23m 35s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/d73dba7a89c54de3aee98d32368bce46 : SUCCESS in 25m 34s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/47606e34a6724522bf496fe96524f57e : SUCCESS in 21m 22s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/e2f92e45881a4c3cb78e905191097cef : SUCCESS in 20m 27s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/0d82c48030364bedbfb6351705b7985a : SUCCESS in 24m 54s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e5fdd0d374cf46aba745c5c8516733bd : SUCCESS in 28m 11s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/cb0b9be6da1d4129996a1208d288b8e5 : SUCCESS in 22m 22s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9191786a9f44494bb990893e02e69821 : SUCCESS in 23m 35s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/fb13ea773465432c905668e34528b26a : SUCCESS in 1h 04m 14s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/60d6f6cb256543968b4aff918a87cb49 : SUCCESS in 1h 05m 58s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/626f902bc1804d3c9f2cd45b6c439b69 : SUCCESS in 1h 14m 30s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dd8c650026bc499b941b11d10e5bddd5 : SUCCESS in 1h 40m 17s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/fc5d65a3e32f4e418b4f3804b6b053be : SUCCESS in 1h 48m 12s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/724e3af79e804ae9be921431277da862 : SUCCESS in 2h 01m 40s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/1ca6433f7fbc41fcbd865dc5a0ba0d7f : SUCCESS in 29m 13s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/b28bf7260ac7450f92621d12f6454f2e : SUCCESS in 27m 33s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/049c21d4528b48bebb507893956d5698 : SUCCESS in 27m 00s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/dcb6aceabbb84fe295e605b36ca63e6e : SUCCESS in 25m 30s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/83ccda0e8d8f421a99ecd3c0ddeedbe6 : SUCCESS in 34m 54s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cc729ff7575c48b2abdb1f3fa62a0b61 : SUCCESS in 47m 02s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/030ef62f433d4b4d9c84aaf121598069 : SUCCESS in 20m 31s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/33d8363d0ced4c86bfc27a8f7861cb09 : SUCCESS in 22m 33s","accounts_in_message":[],"_revision_number":22},{"id":"bebabc5177e8d49a3f177dd31f2f7e61baf7325e","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-09-07 11:58:54.000000000","message":"Patch Set 22: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":22},{"id":"cbc7fd4f96672567afe1a9d2c6a92201479cf564","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 11:59:19.000000000","message":"Patch Set 22: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":22},{"id":"3dccbd6cb487aa83c27f2542949c9d7bef71a481","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-09-07 13:25:07.000000000","message":"Patch Set 22: Code-Review+1","accounts_in_message":[],"_revision_number":22},{"id":"d5375e8ac004108b764bd5c980935d850540b609","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 14:03:00.000000000","message":"Patch Set 22: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/894fe5a765da4030a09e91c5717acf43\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ca6f2169aaa4324ab1d174daa60210d : SUCCESS in 5m 54s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/61c6d13f9ead40e185bbd1d1e71cce81 : SUCCESS in 9m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a76d6a6bf7849b29241e405c614cd80 : SUCCESS in 4m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d680cc833e7342d18c2715de9c63e5c0 : SUCCESS in 4m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/939ab03fdde44ebea41429f9353f261c : SUCCESS in 3m 59s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a43ae7e0625f42e8ab0b52e98bc6c71e : SUCCESS in 6m 05s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e3d75f519b4e42a59f0c01908c4faa40 : SUCCESS in 14m 45s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/530272354c1a478f9ae1def962dcde72 : SUCCESS in 7m 59s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/3ba90d6673934065b4369fe7aaded11d : SUCCESS in 17m 19s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/22555c5cc1e54b3c987404fbaebd37df : SUCCESS in 13m 42s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/40f9e037f50542d3ab6b5471a1ef9d9c : SUCCESS in 22m 25s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1b7288ba2db74ba9b38561a4b445a4f7 : SUCCESS in 21m 05s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/438ac5da22524a13a9ed738a502324c5 : SUCCESS in 19m 35s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/466681fe125f4df3a25c6d2be020e849 : SUCCESS in 21m 02s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/6def662a7300487387f06738d06efebd : SUCCESS in 1h 01m 32s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/0df0f8ee4eba42d1b2478f98df80ea88 : SUCCESS in 1h 02m 20s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/545f89bd651349c5bd510977c5fbcc3d : SUCCESS in 1h 05m 38s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/165815b3d24146ecb26e3c31bde80331 : SUCCESS in 1h 23m 33s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/bb5338cf63984f40a6782f33f6856e14 : SUCCESS in 1h 39m 37s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/24c0059158984a1c95b63db30929615d : SUCCESS in 2h 03m 06s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/c9063c6a4fb84812affba09ab8538af1 : SUCCESS in 19m 50s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/263168c4c56c446f95e7f7a0029fc28b : SUCCESS in 22m 23s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/33406dc31c2c42b589bdb18d0dafe211 : SUCCESS in 24m 41s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/c601b9719e094f71b926ff41619f1c29 : SUCCESS in 23m 12s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/247fc83742cb4ccba69ed1478fbe81c8 : SUCCESS in 25m 04s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d4afb3e219fb41099f39824b2213dab8 : SUCCESS in 36m 17s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/aceb257c664d4aa4bc898140612d57a9 : SUCCESS in 23m 14s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/79d7dd92ce774778b3d3324a31dbb174 : SUCCESS in 22m 47s","accounts_in_message":[],"_revision_number":22},{"id":"89402415147ad69f8a0d43e01e589f2bc35ab33b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 14:03:00.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":22},{"id":"8df4782a38692bf075951092e9ce35f6a435fac7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 14:04:10.000000000","message":"Patch Set 22:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7ad0706940d142ac8f3a959b36def6f1\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06e5642a122141348c37580df230e275 : SUCCESS in 52s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3cea8fd5f8554179b1bdac425a278d68 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":22}],"current_revision_number":22,"current_revision":"4250507f17400eacf73cb218247e28b348c39976","revisions":{"05371f7cca1430d83f67792435eb40c1ba30570e":{"kind":"REWORK","_number":1,"created":"2026-06-26 09:57:22.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/21/995021/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/21/995021/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/21/995021/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/21/995021/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/21/995021/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/21/995021/1"}}},"commit":{"parents":[{"commit":"127c3ae67d8c2e1f0c61b7e45b522f5291defe43","subject":"Merge \"Simplify inspection rules\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/127c3ae67d8c2e1f0c61b7e45b522f5291defe43"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-06-26 09:57:16.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-06-26 09:57:16.000000000","tz":120},"subject":"[WIP] Bump ansible version","message":"[WIP] Bump ansible version\n\nChange-Id: I77d5727b3692833dbb498a743320a90cbf575217\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/05371f7cca1430d83f67792435eb40c1ba30570e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/05371f7cca1430d83f67792435eb40c1ba30570e"}]},"branch":"refs/heads/master"},"26cc0f3a77b0baa2c1c4568a05ddfe2afab1207f":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-07-22 09:55:30.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"real_uploader":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"ref":"refs/changes/21/995021/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/21/995021/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/21/995021/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/21/995021/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/21/995021/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/21/995021/2"}}},"commit":{"parents":[{"commit":"45d22bf0e45ab76945700c7c6da15bf66193e0bf","subject":"Remove leftover device mapper references from the code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/45d22bf0e45ab76945700c7c6da15bf66193e0bf"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-06-26 09:57:16.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-07-22 09:55:30.000000000","tz":0},"subject":"[WIP] Bump ansible version","message":"[WIP] Bump ansible version\n\nChange-Id: I77d5727b3692833dbb498a743320a90cbf575217\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/26cc0f3a77b0baa2c1c4568a05ddfe2afab1207f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/26cc0f3a77b0baa2c1c4568a05ddfe2afab1207f"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"127c3ae67d8c2e1f0c61b7e45b522f5291defe43","ours":"05371f7cca1430d83f67792435eb40c1ba30570e","theirs":"45d22bf0e45ab76945700c7c6da15bf66193e0bf","merge_strategy":"recursive","contains_conflicts":false}},"3ec4e484f4b5d1d353eaf81eb2b3cb6e48fbcb39":{"kind":"REWORK","_number":3,"created":"2026-07-22 10:13:03.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/21/995021/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/21/995021/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/21/995021/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/21/995021/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/21/995021/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/21/995021/3"}}},"commit":{"parents":[{"commit":"45d22bf0e45ab76945700c7c6da15bf66193e0bf","subject":"Remove leftover device mapper references from the code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/45d22bf0e45ab76945700c7c6da15bf66193e0bf"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-07-22 10:10:26.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-07-22 10:10:26.000000000","tz":120},"subject":"[WIP] Bump ansible version","message":"[WIP] Bump ansible version\n\nThe junipernetworks.junos collection was removed from Ansible 14 [1].\nWe now Install it via ansible-galaxy instead.\n\n[1] https://docs.ansible.com/projects/ansible/latest/collections/junipernetworks/junos/index.html\n\nChange-Id: I77d5727b3692833dbb498a743320a90cbf575217\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3ec4e484f4b5d1d353eaf81eb2b3cb6e48fbcb39"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3ec4e484f4b5d1d353eaf81eb2b3cb6e48fbcb39"}]},"branch":"refs/heads/master"},"c31ae473aff73850b8535bc344bfd083187aa4c0":{"kind":"REWORK","_number":4,"created":"2026-07-24 13:33:10.000000000","uploader":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"ref":"refs/changes/21/995021/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/21/995021/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/21/995021/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/21/995021/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/21/995021/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/21/995021/4"}}},"commit":{"parents":[{"commit":"45d22bf0e45ab76945700c7c6da15bf66193e0bf","subject":"Remove leftover device mapper references from the code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/45d22bf0e45ab76945700c7c6da15bf66193e0bf"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-07-22 10:10:26.000000000","tz":120},"committer":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2026-07-24 13:33:09.000000000","tz":120},"subject":"[WIP] Bump ansible version","message":"[WIP] Bump ansible version\n\nThe junipernetworks.junos collection was removed from Ansible 14 [1].\nWe now Install it via ansible-galaxy instead.\n\n[1] https://docs.ansible.com/projects/ansible/latest/collections/junipernetworks/junos/index.html\n\nChange-Id: I77d5727b3692833dbb498a743320a90cbf575217\nCo-Authored-By: Rafal Lewandowski \u003crafal@stackhpc.com\u003e\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c31ae473aff73850b8535bc344bfd083187aa4c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c31ae473aff73850b8535bc344bfd083187aa4c0"}]},"branch":"refs/heads/master"},"49f3d7a555d43c258ce5044a8c05fd2b02b8061e":{"kind":"REWORK","_number":5,"created":"2026-07-26 12:48:09.000000000","uploader":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"ref":"refs/changes/21/995021/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/21/995021/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/21/995021/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/21/995021/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/21/995021/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/21/995021/5"}}},"commit":{"parents":[{"commit":"45d22bf0e45ab76945700c7c6da15bf66193e0bf","subject":"Remove leftover device mapper references from the code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/45d22bf0e45ab76945700c7c6da15bf66193e0bf"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-07-22 10:10:26.000000000","tz":120},"committer":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2026-07-26 12:47:57.000000000","tz":120},"subject":"[WIP] Bump ansible version","message":"[WIP] Bump ansible version\n\nThe junipernetworks.junos collection was removed from Ansible 14 [1].\nWe now Install it via ansible-galaxy instead.\n\n[1] https://docs.ansible.com/projects/ansible/latest/collections/junipernetworks/junos/index.html\n\nChange-Id: I77d5727b3692833dbb498a743320a90cbf575217\nCo-Authored-By: Rafal Lewandowski \u003crafal@stackhpc.com\u003e\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/49f3d7a555d43c258ce5044a8c05fd2b02b8061e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/49f3d7a555d43c258ce5044a8c05fd2b02b8061e"}]},"branch":"refs/heads/master"},"42a560ac72a74cce6ea9683c582286a09dc4efce":{"kind":"REWORK","_number":6,"created":"2026-07-26 19:12:44.000000000","uploader":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"ref":"refs/changes/21/995021/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/21/995021/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/21/995021/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/21/995021/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/21/995021/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/21/995021/6"}}},"commit":{"parents":[{"commit":"45d22bf0e45ab76945700c7c6da15bf66193e0bf","subject":"Remove leftover device mapper references from the code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/45d22bf0e45ab76945700c7c6da15bf66193e0bf"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-07-22 10:10:26.000000000","tz":120},"committer":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2026-07-26 19:12:37.000000000","tz":120},"subject":"[WIP] Bump ansible version","message":"[WIP] Bump ansible version\n\nThe junipernetworks.junos collection was removed from Ansible 14 [1].\nWe now Install it via ansible-galaxy instead.\n\n[1] https://docs.ansible.com/projects/ansible/latest/collections/junipernetworks/junos/index.html\n\nChange-Id: I77d5727b3692833dbb498a743320a90cbf575217\nCo-Authored-By: Rafal Lewandowski \u003crafal@stackhpc.com\u003e\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/42a560ac72a74cce6ea9683c582286a09dc4efce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/42a560ac72a74cce6ea9683c582286a09dc4efce"}]},"branch":"refs/heads/master"},"0345bd5db95bfbd957f7a016a088b223272358a5":{"kind":"REWORK","_number":7,"created":"2026-07-27 08:14:38.000000000","uploader":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"ref":"refs/changes/21/995021/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/21/995021/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/21/995021/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/21/995021/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/21/995021/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/21/995021/7"}}},"commit":{"parents":[{"commit":"45d22bf0e45ab76945700c7c6da15bf66193e0bf","subject":"Remove leftover device mapper references from the code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/45d22bf0e45ab76945700c7c6da15bf66193e0bf"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-07-22 10:10:26.000000000","tz":120},"committer":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2026-07-27 08:14:31.000000000","tz":120},"subject":"[WIP] Bump ansible version","message":"[WIP] Bump ansible version\n\nThe junipernetworks.junos collection was removed from Ansible 14 [1].\nWe now Install it via ansible-galaxy instead.\n\n[1] https://docs.ansible.com/projects/ansible/latest/collections/junipernetworks/junos/index.html\n\nChange-Id: I77d5727b3692833dbb498a743320a90cbf575217\nCo-Authored-By: Rafal Lewandowski \u003crafal@stackhpc.com\u003e\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0345bd5db95bfbd957f7a016a088b223272358a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0345bd5db95bfbd957f7a016a088b223272358a5"}]},"branch":"refs/heads/master"},"be77424563bf96ee2b479d4997e58711da76bff1":{"kind":"REWORK","_number":8,"created":"2026-07-29 08:31:51.000000000","uploader":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"ref":"refs/changes/21/995021/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/21/995021/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/21/995021/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/21/995021/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/21/995021/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/21/995021/8"}}},"commit":{"parents":[{"commit":"45d22bf0e45ab76945700c7c6da15bf66193e0bf","subject":"Remove leftover device mapper references from the code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/45d22bf0e45ab76945700c7c6da15bf66193e0bf"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-07-22 10:10:26.000000000","tz":120},"committer":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2026-07-29 08:31:40.000000000","tz":120},"subject":"[WIP] Bump ansible version","message":"[WIP] Bump ansible version\n\nThe junipernetworks.junos collection was removed from Ansible 14 [1].\nWe now Install it via ansible-galaxy instead.\n\n[1] https://docs.ansible.com/projects/ansible/latest/collections/junipernetworks/junos/index.html\n\nChange-Id: I77d5727b3692833dbb498a743320a90cbf575217\nCo-Authored-By: Rafal Lewandowski \u003crafal@stackhpc.com\u003e\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/be77424563bf96ee2b479d4997e58711da76bff1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/be77424563bf96ee2b479d4997e58711da76bff1"}]},"branch":"refs/heads/master"},"22d6433e9e4e1d69742f884431dee520def845e2":{"kind":"REWORK","_number":9,"created":"2026-07-29 10:12:02.000000000","uploader":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"ref":"refs/changes/21/995021/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/21/995021/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/21/995021/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/21/995021/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/21/995021/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/21/995021/9"}}},"commit":{"parents":[{"commit":"45d22bf0e45ab76945700c7c6da15bf66193e0bf","subject":"Remove leftover device mapper references from the code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/45d22bf0e45ab76945700c7c6da15bf66193e0bf"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-07-22 10:10:26.000000000","tz":120},"committer":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2026-07-29 10:05:14.000000000","tz":120},"subject":"[WIP] Bump ansible version","message":"[WIP] Bump ansible version\n\nThe junipernetworks.junos collection was removed from Ansible 14 [1].\nWe now Install it via ansible-galaxy instead.\n\n[1] https://docs.ansible.com/projects/ansible/latest/collections/junipernetworks/junos/index.html\n\nChange-Id: I77d5727b3692833dbb498a743320a90cbf575217\nCo-Authored-By: Rafal Lewandowski \u003crafal@stackhpc.com\u003e\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/22d6433e9e4e1d69742f884431dee520def845e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/22d6433e9e4e1d69742f884431dee520def845e2"}]},"branch":"refs/heads/master"},"1a716e12006b6089f007f04326c3f60c170a0033":{"kind":"NO_CODE_CHANGE","_number":10,"created":"2026-07-30 11:12:16.000000000","uploader":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"ref":"refs/changes/21/995021/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/21/995021/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/21/995021/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/21/995021/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/21/995021/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/21/995021/10"}}},"commit":{"parents":[{"commit":"45d22bf0e45ab76945700c7c6da15bf66193e0bf","subject":"Remove leftover device mapper references from the code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/45d22bf0e45ab76945700c7c6da15bf66193e0bf"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-07-22 10:10:26.000000000","tz":120},"committer":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2026-07-30 11:12:16.000000000","tz":0},"subject":"Bump ansible version","message":"Bump ansible version\n\nThe junipernetworks.junos collection was removed from Ansible 14 [1].\nWe now Install it via ansible-galaxy instead.\n\n[1] https://docs.ansible.com/projects/ansible/latest/collections/junipernetworks/junos/index.html\n\nChange-Id: I77d5727b3692833dbb498a743320a90cbf575217\nCo-Authored-By: Rafal Lewandowski \u003crafal@stackhpc.com\u003e\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1a716e12006b6089f007f04326c3f60c170a0033"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1a716e12006b6089f007f04326c3f60c170a0033"}]},"branch":"refs/heads/master","description":"Edit commit message"},"d45d0c38b876fb9937f8066589fa24acc1d06e7c":{"kind":"NO_CODE_CHANGE","_number":11,"created":"2026-07-30 11:30:48.000000000","uploader":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"ref":"refs/changes/21/995021/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/21/995021/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/21/995021/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/21/995021/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/21/995021/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/21/995021/11"}}},"commit":{"parents":[{"commit":"45d22bf0e45ab76945700c7c6da15bf66193e0bf","subject":"Remove leftover device mapper references from the code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/45d22bf0e45ab76945700c7c6da15bf66193e0bf"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-07-22 10:10:26.000000000","tz":120},"committer":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2026-07-30 11:30:48.000000000","tz":0},"subject":"Bump ansible version","message":"Bump ansible version\n\nThe junipernetworks.junos collection was removed from Ansible 14 [1].\nWe now install it via ansible-galaxy instead.\n\n[1] https://docs.ansible.com/projects/ansible/latest/collections/junipernetworks/junos/index.html\n\nChange-Id: I77d5727b3692833dbb498a743320a90cbf575217\nCo-Authored-By: Rafal Lewandowski \u003crafal@stackhpc.com\u003e\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d45d0c38b876fb9937f8066589fa24acc1d06e7c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d45d0c38b876fb9937f8066589fa24acc1d06e7c"}]},"branch":"refs/heads/master","description":"Edit commit message"},"507e907e9b9e9e3fe1a5b8af8761ec39fd465547":{"kind":"REWORK","_number":12,"created":"2026-07-30 12:46:53.000000000","uploader":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"ref":"refs/changes/21/995021/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/21/995021/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/21/995021/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/21/995021/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/21/995021/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/21/995021/12"}}},"commit":{"parents":[{"commit":"45d22bf0e45ab76945700c7c6da15bf66193e0bf","subject":"Remove leftover device mapper references from the code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/45d22bf0e45ab76945700c7c6da15bf66193e0bf"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-07-22 10:10:26.000000000","tz":120},"committer":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2026-07-30 12:41:43.000000000","tz":120},"subject":"Bump ansible version","message":"Bump ansible version\n\nThe junipernetworks.junos collection was removed from Ansible 14 [1].\nWe now install it via ansible-galaxy instead.\n\n[1] https://docs.ansible.com/projects/ansible/latest/collections/junipernetworks/junos/index.html\n\nAdded check to ``copy_result[\u0027invocation\u0027]`` as the copy_result shape\nchanged in new Ansible/ansible-core.\n\nReplaced docker_image module with community.docker.docker_image_build.\n\nChange-Id: I77d5727b3692833dbb498a743320a90cbf575217\nCo-Authored-By: Rafal Lewandowski \u003crafal@stackhpc.com\u003e\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/507e907e9b9e9e3fe1a5b8af8761ec39fd465547"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/507e907e9b9e9e3fe1a5b8af8761ec39fd465547"}]},"branch":"refs/heads/master"},"3c76e932a276e70e20144d24b0e93e00de41df2f":{"kind":"REWORK","_number":13,"created":"2026-07-30 14:07:54.000000000","uploader":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"ref":"refs/changes/21/995021/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/21/995021/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/21/995021/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/21/995021/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/21/995021/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/21/995021/13"}}},"commit":{"parents":[{"commit":"45d22bf0e45ab76945700c7c6da15bf66193e0bf","subject":"Remove leftover device mapper references from the code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/45d22bf0e45ab76945700c7c6da15bf66193e0bf"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-07-22 10:10:26.000000000","tz":120},"committer":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2026-07-30 14:07:46.000000000","tz":120},"subject":"Bump ansible version","message":"Bump ansible version\n\nThe junipernetworks.junos collection was removed from Ansible 14 [1].\nWe now install it via ansible-galaxy instead.\n\n[1] https://docs.ansible.com/projects/ansible/latest/collections/junipernetworks/junos/index.html\n\nAdded check to ``copy_result[\u0027invocation\u0027]`` as the copy_result shape\nchanged in new Ansible/ansible-core.\n\nReplaced docker_image module with community.docker.docker_image_build.\n\nChange-Id: I77d5727b3692833dbb498a743320a90cbf575217\nCo-Authored-By: Rafal Lewandowski \u003crafal@stackhpc.com\u003e\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3c76e932a276e70e20144d24b0e93e00de41df2f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3c76e932a276e70e20144d24b0e93e00de41df2f"}]},"branch":"refs/heads/master"},"4f0ef6817f0c145eab00979434d7cb304e7e786b":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2026-07-31 11:33:09.000000000","uploader":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"ref":"refs/changes/21/995021/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/21/995021/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/21/995021/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/21/995021/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/21/995021/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/21/995021/14"}}},"commit":{"parents":[{"commit":"6f67922715c0e430f39b84a19fb4427611ac574b","subject":"Update guidance about install method","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6f67922715c0e430f39b84a19fb4427611ac574b"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-07-22 10:10:26.000000000","tz":120},"committer":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2026-07-31 11:33:09.000000000","tz":0},"subject":"Bump ansible version","message":"Bump ansible version\n\nThe junipernetworks.junos collection was removed from Ansible 14 [1].\nWe now install it via ansible-galaxy instead.\n\n[1] https://docs.ansible.com/projects/ansible/latest/collections/junipernetworks/junos/index.html\n\nAdded check to ``copy_result[\u0027invocation\u0027]`` as the copy_result shape\nchanged in new Ansible/ansible-core.\n\nReplaced docker_image module with community.docker.docker_image_build.\n\nChange-Id: I77d5727b3692833dbb498a743320a90cbf575217\nCo-Authored-By: Rafal Lewandowski \u003crafal@stackhpc.com\u003e\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4f0ef6817f0c145eab00979434d7cb304e7e786b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4f0ef6817f0c145eab00979434d7cb304e7e786b"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"45d22bf0e45ab76945700c7c6da15bf66193e0bf","ours":"3c76e932a276e70e20144d24b0e93e00de41df2f","theirs":"6f67922715c0e430f39b84a19fb4427611ac574b","merge_strategy":"recursive","contains_conflicts":false}},"43b6157e5c6fd878a2d4c9c026a70f00c81926a0":{"kind":"REWORK","_number":15,"created":"2026-08-03 13:04:01.000000000","uploader":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"ref":"refs/changes/21/995021/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/21/995021/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/21/995021/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/21/995021/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/21/995021/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/21/995021/15"}}},"commit":{"parents":[{"commit":"45d22bf0e45ab76945700c7c6da15bf66193e0bf","subject":"Remove leftover device mapper references from the code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/45d22bf0e45ab76945700c7c6da15bf66193e0bf"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-07-22 10:10:26.000000000","tz":120},"committer":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2026-08-03 13:03:52.000000000","tz":120},"subject":"Bump ansible version","message":"Bump ansible version\n\nThe junipernetworks.junos collection was removed from Ansible 14 [1].\nWe now install it via ansible-galaxy instead.\n\n[1] https://docs.ansible.com/projects/ansible/latest/collections/junipernetworks/junos/index.html\n\nAdded check to ``copy_result[\u0027invocation\u0027]`` as the copy_result shape\nchanged in new Ansible/ansible-core.\n\nReplaced docker_image module with community.docker.docker_image_build.\n\nChange-Id: I77d5727b3692833dbb498a743320a90cbf575217\nCo-Authored-By: Rafal Lewandowski \u003crafal@stackhpc.com\u003e\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/43b6157e5c6fd878a2d4c9c026a70f00c81926a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/43b6157e5c6fd878a2d4c9c026a70f00c81926a0"}]},"branch":"refs/heads/master"},"622832f8c1287aa7e458f90139565af4365fd8aa":{"kind":"TRIVIAL_REBASE","_number":16,"created":"2026-08-05 06:22:11.000000000","uploader":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"ref":"refs/changes/21/995021/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/21/995021/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/21/995021/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/21/995021/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/21/995021/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/21/995021/16"}}},"commit":{"parents":[{"commit":"6f67922715c0e430f39b84a19fb4427611ac574b","subject":"Update guidance about install method","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6f67922715c0e430f39b84a19fb4427611ac574b"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-07-22 10:10:26.000000000","tz":120},"committer":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2026-08-05 06:22:11.000000000","tz":0},"subject":"Bump ansible version","message":"Bump ansible version\n\nThe junipernetworks.junos collection was removed from Ansible 14 [1].\nWe now install it via ansible-galaxy instead.\n\n[1] https://docs.ansible.com/projects/ansible/latest/collections/junipernetworks/junos/index.html\n\nAdded check to ``copy_result[\u0027invocation\u0027]`` as the copy_result shape\nchanged in new Ansible/ansible-core.\n\nReplaced docker_image module with community.docker.docker_image_build.\n\nChange-Id: I77d5727b3692833dbb498a743320a90cbf575217\nCo-Authored-By: Rafal Lewandowski \u003crafal@stackhpc.com\u003e\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/622832f8c1287aa7e458f90139565af4365fd8aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/622832f8c1287aa7e458f90139565af4365fd8aa"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"45d22bf0e45ab76945700c7c6da15bf66193e0bf","ours":"43b6157e5c6fd878a2d4c9c026a70f00c81926a0","theirs":"6f67922715c0e430f39b84a19fb4427611ac574b","merge_strategy":"recursive","contains_conflicts":false}},"d3588af0fe0054652a7734774a9a9c36d063c696":{"kind":"REWORK","_number":17,"created":"2026-08-05 10:13:59.000000000","uploader":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"ref":"refs/changes/21/995021/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/21/995021/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/21/995021/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/21/995021/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/21/995021/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/21/995021/17"}}},"commit":{"parents":[{"commit":"45d22bf0e45ab76945700c7c6da15bf66193e0bf","subject":"Remove leftover device mapper references from the code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/45d22bf0e45ab76945700c7c6da15bf66193e0bf"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-07-22 10:10:26.000000000","tz":120},"committer":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2026-08-05 10:13:48.000000000","tz":120},"subject":"Bump ansible version","message":"Bump ansible version\n\nThe junipernetworks.junos collection was removed from Ansible 14 [1].\nWe now install it via ansible-galaxy instead.\n\n[1] https://docs.ansible.com/projects/ansible/latest/collections/junipernetworks/junos/index.html\n\nAdded check to ``copy_result[\u0027invocation\u0027]`` as the copy_result shape\nchanged in new Ansible/ansible-core.\n\nReplaced docker_image module with community.docker.docker_image_build.\n\nChange-Id: I77d5727b3692833dbb498a743320a90cbf575217\nCo-Authored-By: Rafal Lewandowski \u003crafal@stackhpc.com\u003e\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d3588af0fe0054652a7734774a9a9c36d063c696"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d3588af0fe0054652a7734774a9a9c36d063c696"}]},"branch":"refs/heads/master"},"08b94b256f0a723d3967f11e777651d5509482c0":{"kind":"TRIVIAL_REBASE","_number":18,"created":"2026-08-06 09:45:32.000000000","uploader":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"ref":"refs/changes/21/995021/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/21/995021/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/21/995021/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/21/995021/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/21/995021/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/21/995021/18"}}},"commit":{"parents":[{"commit":"8a0d1d9026d7ba9a02d3316ffab7aad5eb994a5d","subject":"Merge \"arista-switch: Fix switch interface configuration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8a0d1d9026d7ba9a02d3316ffab7aad5eb994a5d"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-07-22 10:10:26.000000000","tz":120},"committer":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2026-08-06 09:45:32.000000000","tz":0},"subject":"Bump ansible version","message":"Bump ansible version\n\nThe junipernetworks.junos collection was removed from Ansible 14 [1].\nWe now install it via ansible-galaxy instead.\n\n[1] https://docs.ansible.com/projects/ansible/latest/collections/junipernetworks/junos/index.html\n\nAdded check to ``copy_result[\u0027invocation\u0027]`` as the copy_result shape\nchanged in new Ansible/ansible-core.\n\nReplaced docker_image module with community.docker.docker_image_build.\n\nChange-Id: I77d5727b3692833dbb498a743320a90cbf575217\nCo-Authored-By: Rafal Lewandowski \u003crafal@stackhpc.com\u003e\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/08b94b256f0a723d3967f11e777651d5509482c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/08b94b256f0a723d3967f11e777651d5509482c0"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"45d22bf0e45ab76945700c7c6da15bf66193e0bf","ours":"d3588af0fe0054652a7734774a9a9c36d063c696","theirs":"8a0d1d9026d7ba9a02d3316ffab7aad5eb994a5d","merge_strategy":"recursive","contains_conflicts":false}},"8d24a4dcb4207eb48ebbfd332896e7ef62368860":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":19,"created":"2026-08-06 19:34:09.000000000","uploader":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"ref":"refs/changes/21/995021/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/21/995021/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/21/995021/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/21/995021/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/21/995021/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/21/995021/19"}}},"commit":{"parents":[{"commit":"45d22bf0e45ab76945700c7c6da15bf66193e0bf","subject":"Remove leftover device mapper references from the code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/45d22bf0e45ab76945700c7c6da15bf66193e0bf"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-07-22 10:10:26.000000000","tz":120},"committer":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2026-08-06 19:33:02.000000000","tz":120},"subject":"Bump up Ansible supported versions to 13.x/14.x","message":"Bump up Ansible supported versions to 13.x/14.x\n\nThe junipernetworks.junos collection was removed from Ansible 14 [1].\nWe now install it via ansible-galaxy instead.\n\nAdded check to ``copy_result[\u0027invocation\u0027]`` as the copy_result shape\nchanged in new Ansible/ansible-core.\n\nReplaced docker_image module with community.docker.docker_image_build.\n\n[1] https://docs.ansible.com/projects/ansible/latest/collections/junipernetworks/junos/index.html\n\nChange-Id: I77d5727b3692833dbb498a743320a90cbf575217\nCo-Authored-By: Rafal Lewandowski \u003crafal@stackhpc.com\u003e\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8d24a4dcb4207eb48ebbfd332896e7ef62368860"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8d24a4dcb4207eb48ebbfd332896e7ef62368860"}]},"branch":"refs/heads/master"},"981aa66158cdd11e5eddfd58eee16618892a25fa":{"kind":"TRIVIAL_REBASE","_number":20,"created":"2026-09-04 09:15:03.000000000","uploader":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"ref":"refs/changes/21/995021/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/21/995021/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/21/995021/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/21/995021/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/21/995021/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/21/995021/20"}}},"commit":{"parents":[{"commit":"39661017d8fb8a2e1d8c85bac37ba2e0e4fe7942","subject":"Add NGS support for Cisco NX-OS switches","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/39661017d8fb8a2e1d8c85bac37ba2e0e4fe7942"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-07-22 10:10:26.000000000","tz":120},"committer":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2026-09-04 09:15:03.000000000","tz":0},"subject":"Bump up Ansible supported versions to 13.x/14.x","message":"Bump up Ansible supported versions to 13.x/14.x\n\nThe junipernetworks.junos collection was removed from Ansible 14 [1].\nWe now install it via ansible-galaxy instead.\n\nAdded check to ``copy_result[\u0027invocation\u0027]`` as the copy_result shape\nchanged in new Ansible/ansible-core.\n\nReplaced docker_image module with community.docker.docker_image_build.\n\n[1] https://docs.ansible.com/projects/ansible/latest/collections/junipernetworks/junos/index.html\n\nChange-Id: I77d5727b3692833dbb498a743320a90cbf575217\nCo-Authored-By: Rafal Lewandowski \u003crafal@stackhpc.com\u003e\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/981aa66158cdd11e5eddfd58eee16618892a25fa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/981aa66158cdd11e5eddfd58eee16618892a25fa"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"45d22bf0e45ab76945700c7c6da15bf66193e0bf","ours":"8d24a4dcb4207eb48ebbfd332896e7ef62368860","theirs":"39661017d8fb8a2e1d8c85bac37ba2e0e4fe7942","merge_strategy":"recursive","contains_conflicts":false}},"24e42dd44a8929a0bb039a771860e4faccf02554":{"kind":"REWORK","_number":21,"created":"2026-09-04 09:15:59.000000000","uploader":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"ref":"refs/changes/21/995021/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/21/995021/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/21/995021/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/21/995021/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/21/995021/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/21/995021/21"}}},"commit":{"parents":[{"commit":"45d22bf0e45ab76945700c7c6da15bf66193e0bf","subject":"Remove leftover device mapper references from the code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/45d22bf0e45ab76945700c7c6da15bf66193e0bf"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-07-22 10:10:26.000000000","tz":120},"committer":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2026-09-04 09:14:49.000000000","tz":120},"subject":"Bump up Ansible supported versions to 13.x/14.x","message":"Bump up Ansible supported versions to 13.x/14.x\n\nThe junipernetworks.junos collection was removed from Ansible 14 [1].\nWe now install it via ansible-galaxy instead.\n\nAdded check to ``copy_result[\u0027invocation\u0027]`` as the copy_result shape\nchanged in new Ansible/ansible-core.\n\nReplaced docker_image module with community.docker.docker_image_build.\n\n[1] https://docs.ansible.com/projects/ansible/latest/collections/junipernetworks/junos/index.html\n\nDepends-On: I1da2d8a91af173c7dbe3bb00cc6b791f5a9a620c\n\nChange-Id: I77d5727b3692833dbb498a743320a90cbf575217\nCo-Authored-By: Rafal Lewandowski \u003crafal@stackhpc.com\u003e\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/24e42dd44a8929a0bb039a771860e4faccf02554"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/24e42dd44a8929a0bb039a771860e4faccf02554"}]},"branch":"refs/heads/master"},"4250507f17400eacf73cb218247e28b348c39976":{"kind":"TRIVIAL_REBASE","_number":22,"created":"2026-09-04 09:16:15.000000000","uploader":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"ref":"refs/changes/21/995021/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/21/995021/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/21/995021/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/21/995021/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/21/995021/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/21/995021/22"}}},"commit":{"parents":[{"commit":"39661017d8fb8a2e1d8c85bac37ba2e0e4fe7942","subject":"Add NGS support for Cisco NX-OS switches","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/39661017d8fb8a2e1d8c85bac37ba2e0e4fe7942"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-07-22 10:10:26.000000000","tz":120},"committer":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2026-09-04 09:16:15.000000000","tz":0},"subject":"Bump up Ansible supported versions to 13.x/14.x","message":"Bump up Ansible supported versions to 13.x/14.x\n\nThe junipernetworks.junos collection was removed from Ansible 14 [1].\nWe now install it via ansible-galaxy instead.\n\nAdded check to ``copy_result[\u0027invocation\u0027]`` as the copy_result shape\nchanged in new Ansible/ansible-core.\n\nReplaced docker_image module with community.docker.docker_image_build.\n\n[1] https://docs.ansible.com/projects/ansible/latest/collections/junipernetworks/junos/index.html\n\nDepends-On: I1da2d8a91af173c7dbe3bb00cc6b791f5a9a620c\n\nChange-Id: I77d5727b3692833dbb498a743320a90cbf575217\nCo-Authored-By: Rafal Lewandowski \u003crafal@stackhpc.com\u003e\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4250507f17400eacf73cb218247e28b348c39976"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4250507f17400eacf73cb218247e28b348c39976"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"45d22bf0e45ab76945700c7c6da15bf66193e0bf","ours":"24e42dd44a8929a0bb039a771860e4faccf02554","theirs":"39661017d8fb8a2e1d8c85bac37ba2e0e4fe7942","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
