)]}'
{"id":"openstack%2Fkayobe~995679","triplet_id":"openstack%2Fkayobe~stable%2F2025.1~I0a5100347d8755a452d2744a458b3177e41196ca","project":"openstack/kayobe","branch":"stable/2025.1","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2026-07-10 09:19:12.000000000","reason":"\u003cGERRIT_ACCOUNT_28048\u003e replied on the change","reason_account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2026-07-10 11:53:37.000000000","reason":"Change was submitted"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2026-07-10 11:53:37.000000000","reason":"Change was submitted"},"35263":{"account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"last_update":"2026-07-10 11:53:37.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I0a5100347d8755a452d2744a458b3177e41196ca","subject":"Add support for url sources in kolla_sources","status":"MERGED","created":"2026-07-01 13:20:41.000000000","updated":"2026-07-10 11:54:50.000000000","submitted":"2026-07-10 11:53:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"995679","cherry_pick_of_change":970268,"cherry_pick_of_patch_set":5,"meta_rev_id":"8b229d9ae48d0b0bd40e4d6817acb761f44bfbb4","_number":995679,"virtual_id_number":995679,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-07-10 11:53:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2026-07-10 09:19:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2026-07-01 13:23:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2026-07-10 09:19:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"CC":[{"_account_id":38626,"name":"Eric Le Lay","display_name":"Eric Le Lay","email":"eric@stackhpc.com","username":"elelaysh","status":"Proud to work for StackHPC"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-01 13:20:41.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":38626,"name":"Eric Le Lay","display_name":"Eric Le Lay","email":"eric@stackhpc.com","username":"elelaysh","status":"Proud to work for StackHPC"},"state":"CC"},{"updated":"2026-07-01 13:23:21.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2026-07-01 13:23:21.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2026-07-01 13:23:21.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2026-07-01 15:45:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"32e7063824b38f83c4cb7837b931a3d19dbc6b16","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-07-01 13:20:41.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"7ef83ce9a21c8f04fbec9ad81091f2fd529ffee8","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-07-01 13:23:21.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"e1bf0bce9b2830c26fbb8010e9a8ef5498a53300","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-01 15:45:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b520c85a8881490e9277d7b78e51f88e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c691b1ddf7704c3d9967029dc70a081e : SUCCESS in 4m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ae7ac18ca2047e99e7653226ebd5844 : SUCCESS in 4m 17s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/08dca02a44c54b81a89c41abbe778cc0 : SUCCESS in 5m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac79140f5efd428da9f9b77f18b2fed2 : SUCCESS in 10m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6a156018acab453f87a9fdd3d7931cb3 : SUCCESS in 6m 00s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/854184c0c50d46c591bd079cc2700811 : SUCCESS in 5m 58s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/79055f3432bf4ed59ab3e29616588100 : SUCCESS in 12m 32s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/9936729d07444b099bb18cb365bda2ef : SUCCESS in 11m 43s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/b9b6d7f9e5c84c909b2a25b3729069a5 : SUCCESS in 19m 49s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/f5728faf21334f1abdef9e2c0a5826fb : SUCCESS in 17m 40s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/860faa0361c347c2a7b33a45d0c470ef : SUCCESS in 14m 01s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/ed2bb0fcc952414f8ab8da7ac5264215 : SUCCESS in 8m 33s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/c4c5a267716846cd96f192a0c9c23d84 : SUCCESS in 21m 37s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a3b612f307cf477e929c3ec499856695 : SUCCESS in 17m 40s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/1c4d749ab4cc4f65ba35e8f0a62ad728 : SUCCESS in 1h 00m 01s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/dc0130ff87904679bbcdb11215b66160 : SUCCESS in 34m 18s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/585d16f6bbbf4aae8361e470bfcfaae3 : SUCCESS in 1h 06m 41s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/6b2c915200dc4d9a836c2a18092322fe : SUCCESS in 1h 07m 11s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/b822512df74a4313a24c17a63c5a1028 : SUCCESS in 1h 17m 33s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/ff2fe3a64d914800b217069968b868b9 : SUCCESS in 1h 17m 32s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8d48e0bcfd9542df94aa933905de1747 : SUCCESS in 1h 00m 23s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/6da2d8fe784c48148f038fd4be929548 : SUCCESS in 1h 01m 10s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/cca708a1fa74418894bccac9d8840d9b : SUCCESS in 2h 18m 36s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8e29737370f14b75be1d4364bb2e500e : SUCCESS in 2h 08m 26s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/9886b2d58b6d489387e5c264805cb3be : SUCCESS in 15m 02s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/b500bb14f27e4ffa8c3feef1136bfc3e : SUCCESS in 10m 07s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/ee08482a8fef45ac95ffa6ecbc8bdb23 : SUCCESS in 19m 47s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/388e89838acb4122821b573756b8ee6e : SUCCESS in 19m 32s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/902bafb163d84d0996d81bfa0770bc1a : SUCCESS in 18m 00s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/54a6e24c4200467891220c901ff08ce8 : SUCCESS in 16m 44s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/bf74eb945e654d81a5a2d409e5ef5145 : SUCCESS in 36m 57s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/567ab3c692d2402c9cb276f1e33a6146 : SUCCESS in 12m 26s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/dc36d27fcd7c4ea5920333cdb008dca0 : SUCCESS in 18m 02s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0343027e818d4526a6d47d374a347edb : SUCCESS in 13m 32s","accounts_in_message":[],"_revision_number":1},{"id":"7fbad0bb019ea6d4fd6f16ccb8ae42babfe4c2f5","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-07-10 09:19:12.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"53ba06da05f1b51dd8fbda11b1671966858f7760","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-10 09:19:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3011620ad5eab882056dd9877c0f5037c7a3b6b1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-10 11:53:37.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5d4205c3cf8d424b9c54c03b2ed2faf6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f07206bcd7841c3a8c6a4167947f742 : SUCCESS in 4m 58s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3ab6d016080b47468d5bb96f063ac78c : SUCCESS in 4m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0895e04229640bf9b16c18b12974855 : SUCCESS in 10m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/42a03650ace14b6b8fcb85f5e70bb8d8 : SUCCESS in 6m 26s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/98bfac89af7443c5846401825401d51a : SUCCESS in 5m 34s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/815510b410234572afffb015a43f9a6f : SUCCESS in 13m 52s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1adc0fbb6a0e4fd99c840f73332b298f : SUCCESS in 8m 09s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/3c4649e2f9a44bb8b1a5090b1ab4b900 : SUCCESS in 20m 06s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/f1fc8b170688443b9dd7678d5b026166 : SUCCESS in 17m 26s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1379af4713484563bdd74edb907c0fb6 : SUCCESS in 15m 17s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/4c12a8ededd64042b605b1eac6d0d3c2 : SUCCESS in 22m 07s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/7617ff5079a44e7d9ccdf6acf9564e3e : SUCCESS in 26m 08s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a63f7180e4cb4d96bb72924f93e56e44 : SUCCESS in 16m 24s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/75090cbcaaa84d6a9d9330b4c47dc97f : SUCCESS in 1h 12m 48s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/1e408d9d45b04d53945438d8ca2ba666 : SUCCESS in 1h 17m 26s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/8acb964afc9049eb8197c1626e81039c : SUCCESS in 1h 14m 05s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/52dcde81b9d1494798238daf2bed032e : SUCCESS in 1h 10m 46s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/260dd3d539224cd09fb9de782ee823a9 : SUCCESS in 1h 19m 37s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/cbf1e7824b54473b9f4246429647ae3f : SUCCESS in 1h 15m 26s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/802530b1355e4c25b0153d9dac1b562c : SUCCESS in 57m 57s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/874e1c7ecf6b47b2bd5cb9f4ff3158e2 : SUCCESS in 57m 19s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/31879705eb1645e4aa6f83be493b1f2c : SUCCESS in 2h 28m 11s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/22f5987978864ee8b597e6a0e5a57e0d : SUCCESS in 2h 14m 26s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/c1380c035c664380b428c649376a8d29 : SUCCESS in 22m 09s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/875ff40aec384c6d820936d3b61a7f7b : SUCCESS in 11m 22s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/bee893d396d046ce8599731eb2360372 : SUCCESS in 20m 43s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/d85306f4015a42a48febee6f7294ac18 : SUCCESS in 21m 23s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/21e89a2391b740eb99454a1fbca6c0ce : SUCCESS in 21m 37s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/879c6744fcfa45e3a61b9c68a066eeb0 : SUCCESS in 16m 14s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/418dbb7a1e99473f834c8b9020bae60a : SUCCESS in 36m 44s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/008adc34f1674c139faecddd717daef5 : SUCCESS in 11m 54s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/cfaa7c1aeab941709459aa5c88977969 : SUCCESS in 18m 59s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/981290cee40040d79006aa3c88e5a3e4 : SUCCESS in 14m 49s","accounts_in_message":[],"_revision_number":1},{"id":"3a17661f4e80b3f9598b05a8449e58d353e640ca","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-10 11:53:37.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"8b229d9ae48d0b0bd40e4d6817acb761f44bfbb4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-10 11:54:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fff7932308e149fa8e6c6e12cc9cb598\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/634906a508504739acf0181cd695ec52 : SUCCESS in 48s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1e10c0f0882841768a17ff9996194dd9 : SUCCESS in 49s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7a2d2ca94f680f8b07ec504e02419bfc33e31c91","revisions":{"7a2d2ca94f680f8b07ec504e02419bfc33e31c91":{"kind":"REWORK","_number":1,"created":"2026-07-01 13:20:41.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/79/995679/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/79/995679/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/79/995679/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/79/995679/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/79/995679/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/79/995679/1"}}},"commit":{"parents":[{"commit":"8d046017d52970daac3b1a585cf577c95bd86b5d","subject":"Fix paths to EPEL repositories","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8d046017d52970daac3b1a585cf577c95bd86b5d"}]}],"author":{"name":"Eric Le Lay","email":"eric@stackhpc.com","date":"2025-12-09 14:43:40.000000000","tz":60},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2026-07-01 13:20:41.000000000","tz":0},"subject":"Add support for url sources in kolla_sources","message":"Add support for url sources in kolla_sources\n\nPreviously the version and sha256 attributes would not be generated in\nkolla-build.conf, preventing the override for sources with url type.\nOnly git and local types worked.\n\nChange-Id: I0a5100347d8755a452d2744a458b3177e41196ca\nSigned-off-by: Eric Le Lay \u003ceric@stackhpc.com\u003e\n(cherry picked from commit ca230c71c9642586c2aea27f70111e6888095aa4)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7a2d2ca94f680f8b07ec504e02419bfc33e31c91"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7a2d2ca94f680f8b07ec504e02419bfc33e31c91"}]},"branch":"refs/heads/stable/2025.1","conflicts":{"base":"62721ab89a794b22495e6a7a861d1550123667bd","ours":"8d046017d52970daac3b1a585cf577c95bd86b5d","theirs":"ca230c71c9642586c2aea27f70111e6888095aa4","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
