)]}'
{"id":"openstack%2Fkayobe~995699","triplet_id":"openstack%2Fkayobe~stable%2F2026.1~Icd3730c50acc9f8f72cb3651b7e099e5eaae6973","project":"openstack/kayobe","branch":"stable/2026.1","topic":"bug/2158773-stable/2026.1","attention_set":{},"removed_from_attention_set":{"39213":{"account":{"_account_id":39213,"name":"Jed Preist","display_name":"Jed Preist","email":"jed@stackhpc.com","username":"jed-preist"},"last_update":"2026-07-02 14:30:33.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Icd3730c50acc9f8f72cb3651b7e099e5eaae6973","subject":"Fix kolla inventory hostname formatting","status":"MERGED","created":"2026-07-01 14:57:35.000000000","updated":"2026-07-02 14:31:47.000000000","submitted":"2026-07-02 14:30:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"995699-bug/2158773-stable/2026.1","cherry_pick_of_change":995656,"cherry_pick_of_patch_set":3,"meta_rev_id":"668772da7fb3ba4929df740ba422d9501eab4aaa","_number":995699,"virtual_id_number":995699,"owner":{"_account_id":39213,"name":"Jed Preist","display_name":"Jed Preist","email":"jed@stackhpc.com","username":"jed-preist"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-07-02 14:30:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"recommended":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":1,"date":"2026-07-01 15:20:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-07-02 12:53:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":2,"date":"2026-07-02 07:54:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-07-02 12:53:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-01 15:20:16.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2026-07-01 16:38:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-07-02 07:54:46.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2026-07-02 12:53:50.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"}],"messages":[{"id":"6097e5269b9950310215cde863309c79266089a0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39213,"name":"Jed Preist","display_name":"Jed Preist","email":"jed@stackhpc.com","username":"jed-preist"},"date":"2026-07-01 14:57:35.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"d81c0289dcbb79e841358b577983c2a8dfd16449","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-07-01 15:20:16.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"0603376ab7b9350604943c1b541e5d0094ffce7d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-01 16:38:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d146df96e7064d77a41fee807ae65279\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/539f44a1798b4805bf4740742ba87755 : SUCCESS in 10m 36s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/47e5af4c1a854705bd27f71486423293 : SUCCESS in 7m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/84a556e465e0497ab0a643731db803e9 : SUCCESS in 3m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e833fe5b82a645fbb7a89d9222d993e2 : SUCCESS in 5m 36s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bed7f331d8784db1a9f59ef3790deee7 : SUCCESS in 5m 25s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/36753e0ef1834d40b6c8e97a8a15a350 : SUCCESS in 7m 34s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9e3cffafee3a4874a453480ddd849c6e : SUCCESS in 6m 08s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e9883e6fe10e467ab54208ff79eaef53 : SUCCESS in 7m 33s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5c1553bca8b6482aa0f10f56128dbca6 : SUCCESS in 8m 44s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/bc40e84aa5be4a78aab18056778b9c0a : SUCCESS in 7m 46s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/df271a2647d84a599827276a560df26a : SUCCESS in 18m 16s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/409a80ad1a9a4e3ca28ca1f9ca809673 : SUCCESS in 12m 30s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bfebe8d4d617473b842b4d86e0209dbf : SUCCESS in 19m 25s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/8649319f972e445aab6fc96ca54f9d35 : SUCCESS in 16m 46s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/b6ba0e0f06f64db8bebd39b16cf5cd1c : SUCCESS in 15m 59s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a0fe6738b07245369f79c820862c985d : SUCCESS in 15m 04s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/09e23a5f02b54aa58b53e75a8c024f63 : SUCCESS in 14m 40s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d63a3bb31ba6471f89b5f560284576dc : SUCCESS in 16m 11s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/9a5e5c77147c4594a2e02567586d1a80 : SUCCESS in 1h 02m 44s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/da821ecc17e54bf4a67a613e74bfd20b : SUCCESS in 1h 05m 31s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/ddf18bfede3c4a2e9d92ed6badb41a12 : SUCCESS in 1h 04m 01s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/42294468e30f41c49e87ea37aaadf67c : SUCCESS in 1h 06m 32s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/6e4942a83479400db557d3d652c70c72 : SUCCESS in 1h 05m 58s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/bdae51fec563439ca6adf2c0fa74fbf3 : SUCCESS in 1h 34m 38s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/2acad2cf45ee4f4b8e6fcf2cf314124b : SUCCESS in 1h 31m 15s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ba5eebb23e314bcdbeea550e17e5952f : SUCCESS in 1h 31m 38s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3aa88b7cfef54bfda8ef9ae801253172 : SUCCESS in 1h 39m 43s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/ef9afa5a6727454a879141799c3b23a4 : SUCCESS in 23m 01s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/c2705b2e30e04c719d23af54897b2f2d : SUCCESS in 17m 51s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9abb714300cb4153bd6c89abb7d7f62d : SUCCESS in 26m 53s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/d0fb27f4bb3b409fbc0d40c4f0b140ee : SUCCESS in 22m 45s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/c90b5808760544ac9a47e43987ff90c9 : SUCCESS in 36m 53s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/2318eadb3a564317bf4c0bf5bae05cb0 : SUCCESS in 19m 43s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1ca44fe4b15f4fdda05217df7d19e3c0 : SUCCESS in 38m 22s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/720516c8447542edb9272d8d0facb4eb : SUCCESS in 29m 23s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/df4e0563a2a94345bb631a838622c45e : SUCCESS in 26m 15s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0c16e80388714e49ac65eb0df91ec8fa : SUCCESS in 15m 48s","accounts_in_message":[],"_revision_number":1},{"id":"a16136c1783f5bfccd8dbffab5c15277b764ae03","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-07-02 07:54:46.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a650dd8a068fe5a2ce630e4b8b03c62495b47a77","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-07-02 12:53:50.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4e89414fa88489da03bfb3e09aa801f600f3cc05","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-02 12:54:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7eef837b6842d8acc2ac2614e63d6c0979e12887","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-02 14:30:33.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dea00a377e9e45038ce022f1369afc6b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e99766168c143c4bf3a528f95b01239 : SUCCESS in 4m 35s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/a11e1cb7ede945678f0ead079a656c28 : SUCCESS in 9m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b25e2a906b4444caa8bbba5eb0b1dc12 : SUCCESS in 3m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c6194695fad643eea12872f50d45463f : SUCCESS in 4m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d082ba60245e44bfa9eea6dcc65ce2f2 : SUCCESS in 6m 41s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/60e8527983064ed78f2f86d080a9d44f : SUCCESS in 7m 37s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6a042e92bd774dccb282e5ead731db8e : SUCCESS in 14m 49s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f4426c7512934fb69661a19d7d358404 : SUCCESS in 13m 40s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/9dd85a531d134b3d9642552b3f21a228 : SUCCESS in 11m 14s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/98c2c2ab673e4f36b28cc478b328bb83 : SUCCESS in 9m 47s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/4a17620925dd4f8796180aee8a842123 : SUCCESS in 14m 49s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b657338337454641b4baec766ec5a278 : SUCCESS in 26m 16s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/7757285fa74b4ede8641b0f2b35920c0 : SUCCESS in 19m 07s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/95f40aba9649423a9a242c62d22b1528 : SUCCESS in 21m 29s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/c2102be4aac9495d8e05bbb659065577 : SUCCESS in 1h 03m 19s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/fe0691d45220491daf151a6e89aee708 : SUCCESS in 59m 36s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/f0625bbf55df4372bd54574c97d14df7 : SUCCESS in 1h 02m 20s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2240b855494146d2b1d7af6a663aa127 : SUCCESS in 1h 06m 19s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/faa8ec6735fc41468ba55b414afaee4f : SUCCESS in 1h 04m 25s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/d9cd66d8e79841fbaec3daac51ccd9c4 : SUCCESS in 1h 33m 43s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/7d56c13ff75b4524a95beb9b6116bfac : SUCCESS in 1h 35m 43s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2969424eaf584bfc85fb810bb78377ba : SUCCESS in 1h 31m 20s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6d33e526e1cc4db0bb5e7a271b923393 : SUCCESS in 1h 35m 41s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/9d8e6d65002743e285bd3b36edfd5a60 : SUCCESS in 15m 42s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/91ae0541135149379e00956cd3bcaaae : SUCCESS in 15m 54s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/670b7f517d4e49f19d3876d74f1ef64e : SUCCESS in 26m 43s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/ad62bfb5a4fb418dbfd96351a839b737 : SUCCESS in 25m 19s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/b2970bfecd2044e5a795b3cfb2606078 : SUCCESS in 41m 22s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/074ecfa2fa67439d8c8bda2a5834dcc7 : SUCCESS in 36m 35s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/031de163c6744493b4f70f651d4f925f : SUCCESS in 29m 25s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/aaada3ac488044259748c44a0406ebe0 : SUCCESS in 33m 38s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/e54760d4ef7a4c43b7ad60100bbceeb1 : SUCCESS in 22m 20s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b0ff697cf98f409796ec87155c07698c : SUCCESS in 24m 21s","accounts_in_message":[],"_revision_number":1},{"id":"fd9eb53ddffda5e3bfcb56dc734e88c7abe2be4a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-02 14:30:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"668772da7fb3ba4929df740ba422d9501eab4aaa","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-02 14:31:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5c094176bdcd4342a3eed500258b4849\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/803909c065134957afb7459c867ad2e6 : SUCCESS in 1m 00s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/83d12c94bfee4f76a0d37de69b9a4ef1 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2f4d55f3d42caec8d5d863a659415556ba8e48af","revisions":{"2f4d55f3d42caec8d5d863a659415556ba8e48af":{"kind":"REWORK","_number":1,"created":"2026-07-01 14:57:35.000000000","uploader":{"_account_id":39213,"name":"Jed Preist","display_name":"Jed Preist","email":"jed@stackhpc.com","username":"jed-preist"},"ref":"refs/changes/99/995699/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/99/995699/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/99/995699/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/99/995699/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/99/995699/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/99/995699/1"}}},"commit":{"parents":[{"commit":"6edd10477daa1c71cd2094acc12ff9f996ecb22d","subject":"Merge \"CI: Remove override-checkout for stable/2026.1\" into stable/2026.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6edd10477daa1c71cd2094acc12ff9f996ecb22d"}]}],"author":{"name":"Jed Preist","email":"jed@stackhpc.com","date":"2026-07-01 09:33:31.000000000","tz":60},"committer":{"name":"Jed Preist","email":"jed@stackhpc.com","date":"2026-07-01 14:57:35.000000000","tz":0},"subject":"Fix kolla inventory hostname formatting","message":"Fix kolla inventory hostname formatting\n\nChange the overcloud-top-level jinja template to be less aggressive\nremoving whitespace, so that it doesn\u0027t remove the newlines between\nhosts in a group.\n\nCloses-Bug: #2158773\nChange-Id: Icd3730c50acc9f8f72cb3651b7e099e5eaae6973\nSigned-off-by: Jed Preist \u003cjed@stackhpc.com\u003e\n(cherry picked from commit 61baaf7183b99739a2d5abb886df0a26b83f7716)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2f4d55f3d42caec8d5d863a659415556ba8e48af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2f4d55f3d42caec8d5d863a659415556ba8e48af"}]},"branch":"refs/heads/stable/2026.1","conflicts":{"base":"4b7ff8fa8cd93eda79dc19775bbbd05007c1f248","ours":"6edd10477daa1c71cd2094acc12ff9f996ecb22d","theirs":"61baaf7183b99739a2d5abb886df0a26b83f7716","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
