)]}'
{"id":"openstack%2Fkayobe~996842","triplet_id":"openstack%2Fkayobe~master~I6ab964be87bcd18a71f8e5d4abe29c7cdb26b190","project":"openstack/kayobe","branch":"master","attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2026-07-17 18:17:24.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{"35263":{"account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"last_update":"2026-08-05 14:14:59.000000000","reason":"\u003cGERRIT_ACCOUNT_35263\u003e replied on the change","reason_account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}}},"hashtags":[],"change_id":"I6ab964be87bcd18a71f8e5d4abe29c7cdb26b190","subject":"Fix vault password env var validation","status":"NEW","created":"2026-07-10 18:41:02.000000000","updated":"2026-08-05 16:29:12.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"ed521161f77a5338e700e61ee10a5542ccbc1260","_number":996842,"virtual_id_number":996842,"owner":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-07-17 18:17:24.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-10 20:57:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-07-13 16:28:28.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"}],"messages":[{"id":"57b333c5ae627f8104e3e50e9a41fec3edd2bb09","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-07-10 18:41:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bf7890e96d62003b7a2fcc1e26118568b7086ad1","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-07-10 18:43:57.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"2189b329265a6d56b2fd80c7ee6bd7c528afff57","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-07-10 19:00:45.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"bcf31cb0ef260187c461a6aa4c520440b0b87bc7","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-07-10 19:10:08.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"968dd1b55c7cd8e1c53d6da06e9287805013353b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-10 20:57:59.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/70fde5c4b1214656b723781fcfab9f13\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b9baadf71f14bebb1f6db6fa95d11fb : SUCCESS in 5m 36s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/c9b63c0a1c954c5f9422fbaec62c7e3b : SUCCESS in 4m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/83b2089c2e8b44c7a67362913d026767 : SUCCESS in 2m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f59a89665404c4a84d4e64bba379c75 : SUCCESS in 2m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2999cd82e484471db74a214c630f7425 : SUCCESS in 3m 43s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/cf48b2d18c084a2db8838ba4fea4bcd8 : SUCCESS in 3m 56s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f96a827b62ab4759b545364c8260409e : SUCCESS in 5m 50s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e92cc4a583274bc9974c211beae26268 : SUCCESS in 3m 32s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/496da8c0f6e34d1882a28e63362d7a9f : SUCCESS in 12m 27s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/33abda719cb142aabd9372a7d7829a77 : SUCCESS in 11m 34s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/8e203a0dacb54728970ddb26ff2df200 : SUCCESS in 11m 58s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/fec758b5b9d74e84820f56a6777d9cf6 : SUCCESS in 11m 23s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fa3399afc69b422085838d015e986b95 : SUCCESS in 17m 19s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/36683098851e47c28cf9caa9ebd818ab : SUCCESS in 16m 53s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/e8ccee42453743a4b709be99eea1d82e : SUCCESS in 14m 06s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e47c298d5f194f97816836046be29d64 : SUCCESS in 21m 11s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/498e0af6cf894efe8f1316bd25b6d0b8 : SUCCESS in 24m 07s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1cca19a109e74a568ec30b4544064098 : SUCCESS in 10m 29s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/7605e98244734743aeb9df914ade03a4 : SUCCESS in 1h 05m 09s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/cd4303d9a7044dd7a685591c639cf8e5 : SUCCESS in 1h 01m 01s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/804130d6dd504860b9b26eaf547023da : SUCCESS in 1h 05m 52s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0d553491ae1f4f21800c440a807bb5b4 : SUCCESS in 1h 05m 22s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/7ca95bdb0cca4178b0176722ed3be709 : SUCCESS in 1h 04m 38s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/aa00a73b810e43c3a743010e948b9933 : SUCCESS in 1h 45m 00s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/71428beec2cc471bb07075c8d793e53d : SUCCESS in 1h 42m 07s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/3042dd0f04564438bcfae6f7be57a705 : SUCCESS in 22m 22s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/a36894d397d64df6839a9cf1d6e5f18f : SUCCESS in 27m 51s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1fc8d0a32fb440fa93de58af3d48b006 : SUCCESS in 27m 59s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/d823f6a426cf4d768818eb43aa59c7d6 : SUCCESS in 21m 41s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/bedf30cb0abe494d852c313755fd51f6 : SUCCESS in 41m 45s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d65d0b399f254e738fbddd659ca893db : SUCCESS in 20m 59s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/b89f8fe7d9bc46699d8e154f2f7b2d7d : SUCCESS in 14m 42s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/84a80f46bcbf4667b2ee7396b8e0b608 : SUCCESS in 17m 33s","accounts_in_message":[],"_revision_number":3},{"id":"dfb744372d2b84a7b5bd5714d709216f0aea7756","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-07-13 14:50:20.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":3},{"id":"7e793fc33b415011b41492b0f88d7f58a4835b21","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-07-13 16:28:28.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f5027a39876b3b5e54a9d306bf30302dd6e87075","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-07-17 16:25:46.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"120ffce7e0b685be86b3f7ce6a06183d93a5300c","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-07-17 16:26:08.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"84a00e9b30ebaa418a572a21fc9b5e5bda10b305","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-17 18:17:24.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ccaee6d51ea244fbb62a24981831307b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ce8024bffcb4916ba0a5cf1dea22e24 : SUCCESS in 9m 38s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/32d80151da9544bea466d019f49de1de : SUCCESS in 7m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a35e466b60154c4d93d0f69e133ce5ff : SUCCESS in 4m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc651392db84480aaca9fa2ca63c0ab4 : SUCCESS in 4m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0451b98c42634ead93c62542dde391f6 : SUCCESS in 4m 35s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e715a00588ab4e0582589f6aec1de27d : SUCCESS in 7m 11s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4b1de6b5de6d477ba49d77cf4d1dff84 : SUCCESS in 6m 00s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/0f698b9486134e7f991e309d11d47289 : SUCCESS in 7m 41s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a5eb6a67fef542b7ad67bf4f6553c375 : SUCCESS in 10m 51s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d968f518da6c41a6a39bfa937306e809 : SUCCESS in 9m 32s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/295895fe28c644be8cf5f1f1a676dc2a : RETRY_LIMIT in 4m 45s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/d53b8e86f3454d95ab2925d9778ba141 : RETRY_LIMIT in 6m 27s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b26584e1e31f4d189f99d6a5145656b4 : SUCCESS in 17m 12s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/6724d145b8704a04b21d8aca95e808c8 : SUCCESS in 19m 31s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/f1415f27f23e4c4d862bf23c19c809df : RETRY_LIMIT in 5m 04s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e92c2ad2c457484f9cd381827890ccd1 : SUCCESS in 19m 42s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/7a23bacc14e947ea8c551a2505c0675c : RETRY_LIMIT in 5m 20s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cda07f3792d24617966d09803f5c9882 : SUCCESS in 19m 38s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/6912d63d57414cf79e51696a3f4afc7f : RETRY_LIMIT in 4m 32s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/644833bb01e74079ad239b92839ec9a4 : RETRY_LIMIT in 4m 26s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/9920dd9e632c4d678f4a8c8190099fcf : RETRY_LIMIT in 5m 21s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e65206ccb44c4d1cb213f499d8614a93 : SUCCESS in 1h 07m 00s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/b7b6851b46734ff3b3ef77ecfaa73d33 : SUCCESS in 1h 01m 07s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/7058356e1bf5438a9e83e0c3f308f1b2 : FAILURE in 4m 53s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7bdd5e1339964049b14a6f8f31d89270 : SUCCESS in 1h 49m 12s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/70c951ed54454eb792be14b85710a179 : SUCCESS in 30m 10s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/fe3d98cdca774ba4b00722c60c7076f4 : SUCCESS in 29m 13s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9fc97b61c4e64db4b22dbbc67fbf082a : SUCCESS in 24m 06s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/cb2cae2281dd4f819dd873b6423e0d5d : SUCCESS in 21m 33s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/39de5599a7a04e45a30ad89484f06199 : SUCCESS in 41m 46s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d28ba2ec1d434b9fb9134814e9e49e9e : SUCCESS in 34m 57s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/b2b7f831066c4fb4b5215a54796641a2 : RETRY_LIMIT in 3m 47s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/25649f63c26d42e0917e5c0fc7841900 : SUCCESS in 20m 28s","accounts_in_message":[],"_revision_number":4},{"id":"5093d7ae5ac422267bcde880963d115381494855","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-08-05 14:14:59.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"ed521161f77a5338e700e61ee10a5542ccbc1260","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-05 16:29:12.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4711316898504005a3151808d40dc4fa\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc00305e7de14a84ad71868833ead5b4 : SUCCESS in 10m 03s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/c4eef00c855c45f09ee88c15b6452a90 : SUCCESS in 4m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5df7095d88af4ec59b37c8e2229cf157 : SUCCESS in 4m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/26b7eb2abe1d4618bfb0cc882defdd33 : SUCCESS in 2m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b123cdb73b3c45bb911c9076caf1a279 : SUCCESS in 5m 51s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b51fe3005d1e429bad90d5a2cfa5bd2d : SUCCESS in 5m 45s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/13c66ebd059b423f9abb514c56e29a64 : SUCCESS in 6m 31s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/eb9d984535d341788b709f6afa1beb79 : SUCCESS in 5m 44s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/88bffe27e91e41caab9347c86fe6eed8 : TIMED_OUT in 30m 58s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/9380a23b0f8542ec939d72142efa6750 : SUCCESS in 11m 26s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/74c1fb8399ec44748974e126b5679405 : SUCCESS in 12m 03s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/3cef6641139140c18bd150e417d0eccd : SUCCESS in 23m 08s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/631b1082c2d548a6b326c0423759cf5f : SUCCESS in 13m 32s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/7a4ef2df4cbb43659e397133bc00b481 : SUCCESS in 12m 12s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/d4d397623d954e66aa6164384254ef0f : SUCCESS in 14m 04s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2b3e674c497543ea942e5b689f054090 : SUCCESS in 12m 32s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/cf36b6f713c3456692523a1cfe8626ce : SUCCESS in 12m 01s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/957ee39a024d47ba8241510b710e52a7 : SUCCESS in 21m 44s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/b147c83badee4f12a5a3c1902fbc5d3f : SUCCESS in 1h 15m 18s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/dade30a8cfb747529de4d50b4193166d : SUCCESS in 1h 10m 12s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/148b0b2a31f546529a27279f79f63abe : SUCCESS in 1h 13m 09s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4bceaf6e78f54b4380b32011cc5f43f0 : SUCCESS in 1h 21m 27s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/faa5662bd351463694ad509c3e5ce3fa : SUCCESS in 1h 24m 18s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/dc5c77eef53f4288a97ddc39750c66a8 : SUCCESS in 2h 05m 04s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/15eb615f8a204cdaabeccb1f7a5eb5ee : SUCCESS in 2h 13m 06s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/4986ba55da6d4bc78c42f3cce74f3c84 : SUCCESS in 17m 14s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/edd3d29f281446c7bec2254f9e919f2d : SUCCESS in 17m 34s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8aff4d868ff1449aa12df04dd8b9df19 : SUCCESS in 14m 54s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/932bdf82790949a0a87df65bdf656030 : SUCCESS in 15m 33s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/92da630559ad4131a93dea21617d4bce : SUCCESS in 24m 27s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b1343f92475f4e2d90d35ca26592837b : SUCCESS in 38m 29s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/026ebb2b35a04dacb44a6e3e67d003cb : SUCCESS in 26m 18s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c0f6235b1d3344459d418298f76c8338 : SUCCESS in 22m 11s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"f5d81b0f7fbecc40f8df716d3f6a12067191a5ad","revisions":{"da5065123a075ffb49f2aa06931188c17748ef20":{"kind":"REWORK","_number":1,"created":"2026-07-10 18:41:02.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/42/996842/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/42/996842/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/42/996842/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/42/996842/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/42/996842/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/42/996842/1"}}},"commit":{"parents":[{"commit":"e7dc5a78e5f00420fcad9e99a47a3f67b4669ebd","subject":"Update architecture page to reflect reality","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e7dc5a78e5f00420fcad9e99a47a3f67b4669ebd"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-07-10 18:36:13.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-07-10 18:40:56.000000000","tz":60},"subject":"Fix vault password env var validation","message":"Fix vault password env var validation\n\nOnly check KAYOBE_VAULT_PASSWORD when determining the default vault\npassword file. Previously ANSIBLE_VAULT_PASSWORD_FILE could incorrectly\ntrigger vault password helper usage.\n\nCloses-Bug: #2160369\nChange-Id: I6ab964be87bcd18a71f8e5d4abe29c7cdb26b190\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/da5065123a075ffb49f2aa06931188c17748ef20"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/da5065123a075ffb49f2aa06931188c17748ef20"}]},"branch":"refs/heads/master"},"0fe92466b0c09f38beec2f17e958b732dc11ece8":{"kind":"REWORK","_number":2,"created":"2026-07-10 19:00:45.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/42/996842/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/42/996842/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/42/996842/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/42/996842/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/42/996842/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/42/996842/2"}}},"commit":{"parents":[{"commit":"e7dc5a78e5f00420fcad9e99a47a3f67b4669ebd","subject":"Update architecture page to reflect reality","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e7dc5a78e5f00420fcad9e99a47a3f67b4669ebd"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-07-10 18:58:00.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-07-10 19:00:40.000000000","tz":60},"subject":"Fix vault password env var validation","message":"Fix vault password env var validation\n\nOnly check KAYOBE_VAULT_PASSWORD when determining the default vault\npassword file. Previously ANSIBLE_VAULT_PASSWORD_FILE could incorrectly\ntrigger vault password helper usage.\n\nWhen using --vault-password-file with ansible-vault, remove\nANSIBLE_VAULT_PASSWORD_FILE from environment to avoid conflicting\nvault password sources.\n\nCloses-Bug: #2160369\nChange-Id: I6ab964be87bcd18a71f8e5d4abe29c7cdb26b190\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0fe92466b0c09f38beec2f17e958b732dc11ece8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0fe92466b0c09f38beec2f17e958b732dc11ece8"}]},"branch":"refs/heads/master"},"fafa54a7aab791e78e5496836e14166badafdbf2":{"kind":"REWORK","_number":3,"created":"2026-07-10 19:10:08.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/42/996842/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/42/996842/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/42/996842/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/42/996842/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/42/996842/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/42/996842/3"}}},"commit":{"parents":[{"commit":"e7dc5a78e5f00420fcad9e99a47a3f67b4669ebd","subject":"Update architecture page to reflect reality","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e7dc5a78e5f00420fcad9e99a47a3f67b4669ebd"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-07-10 19:06:34.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-07-10 19:09:52.000000000","tz":60},"subject":"Fix vault password env var validation","message":"Fix vault password env var validation\n\nRemove ANSIBLE_VAULT_PASSWORD_FILE from environment in\nupdate_environment() when its password is being used. This prevents\nconflicts when ansible-vault is called with --vault-password-file, which\ncan occur in kolla_passwords.py when kolla-ansible runs.\n\nCloses-Bug: #2160369\nChange-Id: I6ab964be87bcd18a71f8e5d4abe29c7cdb26b190\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fafa54a7aab791e78e5496836e14166badafdbf2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fafa54a7aab791e78e5496836e14166badafdbf2"}]},"branch":"refs/heads/master"},"f5d81b0f7fbecc40f8df716d3f6a12067191a5ad":{"kind":"REWORK","_number":4,"created":"2026-07-17 16:25:46.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/42/996842/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/42/996842/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/42/996842/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/42/996842/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/42/996842/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/42/996842/4"}}},"commit":{"parents":[{"commit":"e7dc5a78e5f00420fcad9e99a47a3f67b4669ebd","subject":"Update architecture page to reflect reality","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e7dc5a78e5f00420fcad9e99a47a3f67b4669ebd"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-07-10 19:06:34.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-07-17 16:25:44.000000000","tz":60},"subject":"Fix vault password env var validation","message":"Fix vault password env var validation\n\nRemove ANSIBLE_VAULT_PASSWORD_FILE from environment in\nupdate_environment() when its password is being used. This prevents\nconflicts when ansible-vault is called with --vault-password-file, which\ncan occur in kolla_passwords.py when kolla-ansible runs.\n\nCloses-Bug: #2160369\nChange-Id: I6ab964be87bcd18a71f8e5d4abe29c7cdb26b190\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f5d81b0f7fbecc40f8df716d3f6a12067191a5ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f5d81b0f7fbecc40f8df716d3f6a12067191a5ad"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
