)]}'
{"id":"openstack%2Fkayobe~997502","triplet_id":"openstack%2Fkayobe~master~I5c99da4c4ecc4995a31e96ffc9de4fe2620cdecb","project":"openstack/kayobe","branch":"master","topic":"addReconfigureDC","attention_set":{},"removed_from_attention_set":{"38288":{"account":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"last_update":"2026-07-21 18:18:43.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2026-07-17 20:43:27.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2026-07-21 14:26:06.000000000","reason":"\u003cGERRIT_ACCOUNT_15197\u003e replied on the change","reason_account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},"35263":{"account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"last_update":"2026-07-21 18:18:43.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5c99da4c4ecc4995a31e96ffc9de4fe2620cdecb","subject":"Add missing reconfigure.yml for manage-containers","status":"MERGED","created":"2026-07-16 09:45:29.000000000","updated":"2026-07-21 18:20:18.000000000","submitted":"2026-07-21 18:18:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"997502-addReconfigureDC","meta_rev_id":"99b8b00e539df091f23d7de4b79010e2e68b469b","_number":997502,"virtual_id_number":997502,"owner":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-07-21 18:18:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2026-07-17 14:59:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2026-07-17 20:43:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-07-21 14:26:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-07-21 14:26:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":1,"date":"2026-07-17 14:18:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":1,"date":"2026-07-17 14:18:28.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-16 10:13:11.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2026-07-16 12:13:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-07-16 12:14:25.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2026-07-16 16:15:56.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2026-07-17 14:59:55.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"}],"messages":[{"id":"f3c966a890d5057132558a6a7e31d8a463939f43","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"date":"2026-07-16 09:45:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"db953f555169395bf1227f6f16b9415d2be4a8de","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"date":"2026-07-16 10:07:32.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4daae0907b8f16d7f7dd8b28057edb657be309e1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"date":"2026-07-16 10:12:40.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"1c4c3037ff9df6c13edc8df89cbe83d11d879525","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-07-16 10:13:11.000000000","message":"Patch Set 3: Backport-Candidate+1 Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":3},{"id":"89de606a9ad99f89fd690864b58a92dfc40275ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-16 12:13:22.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/01b733d8f36a4457b50b3d5127c55465\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b4152bf59f249a9b7a220d5ed9979d3 : SUCCESS in 9m 08s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/bb969f6069714a00bccb22326ffbdff9 : SUCCESS in 7m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/61d665d6a9574f6d96cd8ce14065ded1 : SUCCESS in 4m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce2d48e861ce4e9d8f727050d8ad630b : SUCCESS in 4m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fcaa7e6f46e046f5889e7a79562296fb : SUCCESS in 4m 32s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/cbb6cde808fc4026bd086b600083088a : SUCCESS in 7m 53s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/afcb092c764b40ab8f9b9950d693c6e2 : SUCCESS in 6m 03s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/56bdde86d4c4436a962f95e2c3a22e1f : SUCCESS in 4m 32s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d53a5cf99cdd4bacb6805061d3e9bf99 : SUCCESS in 8m 45s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/54a4544fa8c444c2a46d57c3214c85e9 : SUCCESS in 10m 12s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/44fbebd839544f01a60f5400153e6590 : SUCCESS in 17m 06s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/2d90e6c9676a4e90b9bf7c2cd35e925e : SUCCESS in 11m 16s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8d2d77ca507b47fb9714d9141199715c : SUCCESS in 14m 56s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/79d39f8a7a2e47478ef73877942365c1 : SUCCESS in 14m 59s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/7c016f21485d4f2ab849aef6cc800a68 : SUCCESS in 16m 57s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/261f3a357289407b925816c0d4b5aaef : SUCCESS in 18m 46s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/8fe5fca1d82a43ffa9efb88ccb663c35 : SUCCESS in 17m 39s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c4b5898d194c415b959f0f0cdb5e967c : SUCCESS in 19m 00s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/0758de61bb2c4af786376f5481e49501 : SUCCESS in 1h 08m 22s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/6a6852fccba64192b241459af4790f9f : SUCCESS in 57m 45s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/1c86143473fb4217b161dbf8b64bea74 : SUCCESS in 1h 09m 00s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1bbf10a2c8ee479fbca456f16ade4e7f : SUCCESS in 1h 12m 23s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/a0c1f0f6f93f4d5288df56b7ba29ef9f : SUCCESS in 1h 11m 29s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/d8e5857e043641309ff470dd3ff553a3 : SUCCESS in 1h 55m 35s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2a91d2fa0ff6420fb3a732f27a78caa6 : SUCCESS in 1h 59m 45s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/feb8356ea54d4eb4a7299a888a3fd8fc : SUCCESS in 23m 23s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/cd27bdc488aa4acfb4f3637e4d985a71 : SUCCESS in 14m 32s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/201f13c39dfd451d831681bf8d558874 : SUCCESS in 23m 33s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/479b58178c454f7aacb492d46209fdf0 : SUCCESS in 24m 17s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/0ddfef5e002e451fbb400028214ac92a : SUCCESS in 20m 56s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3f23c377010e4db38f3b2472639e2f0c : SUCCESS in 33m 51s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/288b3d67a2e74b2fa991fe95b678f1da : SUCCESS in 19m 00s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5b1943e0d080463ab0f5e464b851e2e3 : SUCCESS in 20m 33s","accounts_in_message":[],"_revision_number":3},{"id":"4c9042f602cfd04321d6c13eb4dd1ded5688dee5","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-07-16 12:14:25.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"420dc420e8dae97f4a2923922eaf3ce3f4e2b44e","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-07-16 16:15:56.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"7bf208fd619f8a72579f7fb4ad7fc6123c366e82","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"date":"2026-07-17 09:21:31.000000000","message":"Uploaded patch set 4.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"6cd2cea81c169ab240c0620669eac87b3c778ad7","author":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"date":"2026-07-17 09:22:03.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"49c89b1d67559dbf112bce29bdb2d9fdc6de45d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-17 11:15:12.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ca6a04e69e04476caa62528e11d12d41\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d81a45b221342c9bd663d3e545ebd42 : SUCCESS in 12m 06s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/a5950ee9313e4491af5177679eeff3ef : SUCCESS in 8m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/496440ba30f94c4186eb2fe01c5e55a5 : SUCCESS in 4m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b8cd6bb50e08408ba07da3a8b8bfa364 : SUCCESS in 2m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/704fb9d2d17d409a823c0c7b8f91bee8 : SUCCESS in 2m 35s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/efb158f13a33404ab17699dc44eb03b9 : SUCCESS in 6m 36s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/27a9654d28be47c9806207b9205f7a50 : SUCCESS in 8m 01s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/bc0d987980dc40c0a7cb75e7890375ff : SUCCESS in 5m 46s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/50ad277dd8214366a99e99b4c7a75e13 : SUCCESS in 13m 39s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2db6c26840ae42aa9434db2273873890 : SUCCESS in 13m 39s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/b2922d4c12984b54ba3842819347edc4 : SUCCESS in 23m 47s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/c8faae62f6594b7bb35dcae7fe21d748 : SUCCESS in 17m 16s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cea19870baf647f88623d652d251bafb : SUCCESS in 17m 04s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/d432d05f4e854dafa491b0ccf79c3357 : SUCCESS in 17m 03s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/1cfd7fc20d5643d7bb00c379712561c5 : SUCCESS in 20m 24s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/54f358a9c3554c8e8627093e7b9456be : SUCCESS in 13m 05s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/f7d7b47f836447619b7f5b82bce6fe8f : SUCCESS in 14m 18s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8caf51802f5042dcaed1b28c3759e7fa : SUCCESS in 25m 32s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/528f07578e1a468b9f486aca9c4d6613 : SUCCESS in 1h 07m 04s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/a745dacddb2544d58fa75b52bd42340b : SUCCESS in 1h 04m 20s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/c327ce77615f42d4849f21d1969b857f : SUCCESS in 1h 08m 21s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9d7805d741244aee910b149b14fbf4b1 : SUCCESS in 1h 07m 56s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/ba80498dcedb438a9560ef9ff3fcdbb1 : SUCCESS in 1h 08m 32s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/dfde57edcba846d8bd90e504b188362a : SUCCESS in 1h 51m 56s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f8349e90d4e74d2fb56fb303bda138c1 : SUCCESS in 1h 52m 37s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/63d49d7a188f4412875d615788f06b15 : SUCCESS in 24m 30s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/b3e64077048344a4869b807f3cc92ef6 : SUCCESS in 30m 18s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4edd59af81634882a568b95f47a52ca8 : SUCCESS in 27m 30s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/f66898b35d7141b9aa44ae77a8016a35 : SUCCESS in 20m 13s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/e28777f697c642ccacae98d465b1932d : SUCCESS in 35m 50s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/276afa7abc374450990cb65067f32eee : SUCCESS in 35m 55s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/53ef428a5cfd4d78afdd8d258fdeab5b : SUCCESS in 16m 11s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6ead7cf7c5aa42978b07600b067d9ba7 : SUCCESS in 13m 50s","accounts_in_message":[],"_revision_number":4},{"id":"9471cc3562498c5a6c9f9c567b34464bb6514c61","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-07-17 13:53:35.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"6d9d0ecddadfdb6d35634354e6a12d3f7f26ca1b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"date":"2026-07-17 14:18:28.000000000","message":"Uploaded patch set 5.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"ea969c4906b912b35ac8f5890ed23027c28fe281","author":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"date":"2026-07-17 14:18:58.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"5763fd8850fe862cf5f6fa1506097a99cf977a76","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-07-17 14:59:55.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"346f90b47b25c42b5da5bcb4793d49ce3b80b99a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-17 16:16:18.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/93bc8ae5cb324b8d8122c00ccbb9a495\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e05ac9bc80db459c91ac0d78aa0a296b : SUCCESS in 10m 02s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/21f02acc68be4aae8dec3d34fa9b3ea3 : SUCCESS in 7m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/330bd541ed2d4ce083799d4c995765cd : SUCCESS in 5m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b674edd07077403f9443bfac8d5f70bb : SUCCESS in 4m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1827738c437b4ab0829f85deab9ffff9 : SUCCESS in 4m 46s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/850f3439529e465e8982c25f46284217 : SUCCESS in 9m 48s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8c555b74ab0d407c8ca5b393ba8c5252 : SUCCESS in 6m 15s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/403d5776002f4fbf852393a31609a9c4 : SUCCESS in 5m 06s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/89148720feb44d4b95abc6e4a34b601e : SUCCESS in 13m 27s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e8d14ebe761d42e88b06aa0fe59830ce : SUCCESS in 12m 33s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/5c7835fcd105439194476d767ae92481 : RETRY_LIMIT in 5m 14s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/f016f26b70c5491e856f39eaa9e61fee : RETRY_LIMIT in 5m 30s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8958b9f8fb4042c6bc0402d08ec01b83 : SUCCESS in 18m 30s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/6318c3776d4d40729735e9514845a4d2 : SUCCESS in 18m 28s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/b9ab17a7f61c455da9d18c3ff7a759d8 : RETRY_LIMIT in 3m 58s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8dba574c24844feb9f6a64f57ccc8795 : SUCCESS in 20m 43s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/018a1f894c024bdb8bdee4e23afe3e34 : RETRY_LIMIT in 7m 09s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/246fa27ab69c4dc684e06e84b7ff73cc : SUCCESS in 20m 16s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/a054c368025f44b491b13b21f75e2d36 : RETRY_LIMIT in 5m 29s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/857dc804efb44b82bb9ec0f974095c55 : RETRY_LIMIT in 4m 22s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/790274c98b654ab297868a5ae86005d6 : RETRY_LIMIT in 5m 10s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6bfa3bb3b7fa4540bea2f9d2d1952332 : FAILURE in 1h 35m 56s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/995ffc1431964b02a35ada4dbc4428db : SUCCESS in 1h 14m 31s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/df1f37213b2c43ba880309d4fdab0543 : FAILURE in 5m 48s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/25cec457a70741f38e660e1799bc6170 : SUCCESS in 1h 52m 34s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/392f56f0cf8a4c3980df8448f99617ad : SUCCESS in 20m 26s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/cab41ba6281f4dfea3325e2e21b0ca85 : SUCCESS in 30m 03s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d9911f927a0146149b2938ffa5b05dcf : SUCCESS in 19m 16s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/63f9e3dd04bb408cb2205050567071d6 : SUCCESS in 23m 14s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/be0c24b50c254eac9a64fe6d6023b019 : SUCCESS in 37m 59s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9e2a49b834f448f59982b607ceec7380 : SUCCESS in 35m 15s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/736cfbe9c60141aeb0c5a504eea09be5 : RETRY_LIMIT in 3m 22s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1ebf33c25a94443ba9e8fb5981683f1c : SUCCESS in 19m 58s","accounts_in_message":[],"_revision_number":5},{"id":"d2f41bca7ad7c423d8fd9fd1b93349848b8620a7","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-07-17 20:43:27.000000000","message":"Patch Set 5: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"59cfa2e09550865e1019b0389fa95c8c968bd2fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-17 22:34:05.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e0dc89644c734f72b6b806b41a0c1d4a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/afaf1be2a7d44e939d2deca2a666bd50 : SUCCESS in 9m 08s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/e9dd8640494a4840946a69388d491ce6 : SUCCESS in 6m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5d70aa49c5834c50adb294a4f22439ce : SUCCESS in 5m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9176e7c63fe141a2b31734d5273b9d9c : SUCCESS in 4m 20s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fcfcce51b9e8418cadc0e67bda71a2af : SUCCESS in 3m 09s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f3dc8a1077c74403a58e8ed5d1eff4a1 : SUCCESS in 6m 19s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f96695e69c9d4e46ad624bce628cda4d : SUCCESS in 6m 50s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/61949591089149b7848e439aec40bcd2 : SUCCESS in 5m 07s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b87506c5f51a46d7a1a3ddfd99ed981a : SUCCESS in 8m 20s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b8fddb9e8f324e559e1d6e46ec845f86 : SUCCESS in 9m 22s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/ce49068f315143769703d1284be293dc : SUCCESS in 16m 46s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/585333ce7516474b9bfec1270008dea3 : SUCCESS in 13m 08s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c4ffd76099004115bac1dfb6d7823456 : SUCCESS in 19m 17s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/8996e036edc343f7bc3d6315d3c680a0 : SUCCESS in 15m 54s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/5807e07d6b8c4a42be3be0b4d3521adc : SUCCESS in 16m 00s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1b579352ebaf41a4b1d6a34b374a6b7b : SUCCESS in 19m 41s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/6a2781f137284d309bde32780ec02097 : SUCCESS in 11m 30s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4aa5ccbd819f45f9baa9130f68b9be56 : SUCCESS in 15m 49s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/a904fd0a5b6d4efd856a1876c104e2e7 : RETRY_LIMIT in 4m 29s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/1100b94aa4ec4a1e97fd9ab8e6e8774f : RETRY_LIMIT in 5m 22s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/23db5e15c19d4c119cb64a9d03a1accb : RETRY_LIMIT in 4m 28s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/44ff4b20ec8d4652816fbed39b2f1ffc : SUCCESS in 1h 05m 01s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/466c3c863ed14a55b4f87f72f06c445c : SUCCESS in 1h 00m 44s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/fe38440698b04ac8b3fdc468a1afbff4 : FAILURE in 5m 25s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/83f468d22e9644f2ac73fae416e162d7 : SUCCESS in 1h 49m 37s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/36082bb0e21b4185aed5bb5cdc9caab2 : SUCCESS in 21m 49s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/c1333c9bd42445e29922030e39fa85e2 : SUCCESS in 17m 14s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1b3d949861a14f28acc83669f8b0c06a : SUCCESS in 15m 44s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/781a0856c1724f328aa7aa85de99ce87 : SUCCESS in 23m 08s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/7c8b464155d248edb10b2348215e5e73 : SUCCESS in 44m 35s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5de5f69c6def43f1920d2c4058e1ec30 : SUCCESS in 35m 30s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/062f8f1b1b7640fa922c881c27af7d5d : SUCCESS in 18m 05s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0a056eb0ca944bfca3ba9a8f0998da10 : SUCCESS in 19m 48s","accounts_in_message":[],"_revision_number":5},{"id":"c096f2442e4a018c2fdf55e67df833a52d6e9030","author":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"date":"2026-07-21 14:23:02.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"76471a6c3e47cdb9cd3a15ed0bcb7ddee66fa0b8","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-07-21 14:26:06.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"a1816a430cb1bd1e573655c2228e9d45aa608ff0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-21 16:30:01.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/967f780c1d1c4a9f9bebc0ee2db788a4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de5372d6a11749b78eb11791b9a886ed : SUCCESS in 8m 13s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/bb97d05771534e6c9d9923abf642c4ab : SUCCESS in 9m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fc292f535d68416a921a402f68f9dc40 : SUCCESS in 3m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e2660ffe5ae84d19a7e1b7dfc9408318 : SUCCESS in 4m 34s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/15466cc98a8d44e1adc642921aa258d3 : SUCCESS in 4m 47s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c44103c05f89484b87978fd1e0b73be2 : SUCCESS in 9m 57s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2f0e656f6223489a85799eb269f92d7b : SUCCESS in 6m 19s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/fe53c0f47a7843e4ad32bbfb2d7554ed : SUCCESS in 5m 30s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5cd929c00a6f45f787196555d21aad5a : SUCCESS in 6m 55s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c22cc35a8c5e4e118e56660b8623acba : SUCCESS in 15m 12s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/8729ba3a5c1940cfa5ac7d18750933e3 : SUCCESS in 17m 33s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/85d7811a9ed241d6bffd09320a870941 : SUCCESS in 16m 47s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7a2c8462c6b640019e2d912f48d88295 : SUCCESS in 10m 35s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/c5797fdebca24a4caf65a4599b2703c7 : SUCCESS in 9m 34s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/2ca0d0d4afbe4d21adec823c58a63974 : SUCCESS in 25m 36s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/63308a347e93439e80deeb522fb58ac0 : SUCCESS in 12m 56s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/01a55a527d234889831cb20eed5235f4 : SUCCESS in 22m 57s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/534f825162344e6e97b3cfa5d621af81 : SUCCESS in 22m 53s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/4442d3b2611b412494bb5f86ffb68f81 : SUCCESS in 1h 02m 11s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/93c0c65d914144d58aac3bafef2d04eb : SUCCESS in 1h 02m 14s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/effae1aa2dc746f3b2b04cda6b5166c8 : SUCCESS in 1h 13m 58s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ecb1fec9a17e4e76be40ac6865314095 : SUCCESS in 1h 05m 59s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/d8ac0922352d4805a37cd6d69e484715 : SUCCESS in 1h 07m 49s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/2eee60cee09941e696b8fbea1a67f957 : SUCCESS in 1h 46m 36s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/23233aa49fe94a0b88f31ae4d767bef5 : SUCCESS in 2h 05m 56s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/6b23f43e2e6c4f27befb160916116783 : SUCCESS in 28m 28s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/26e9963614ed4dbc989af01c4ad1895c : SUCCESS in 27m 56s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/508fb7ae78ba47d7865b3495e467f8c2 : SUCCESS in 18m 37s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/21b5bac3e72e45c8a2d7d847de09d62e : SUCCESS in 21m 57s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/3d4d26a7f39948c5a7b22b3573642dea : SUCCESS in 42m 27s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a18b2f4bcebb483886191d90b1aefaa2 : SUCCESS in 41m 28s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/a2e891ebf14540aebcf99ef9fad804ab : SUCCESS in 15m 36s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e0e7d23cd68540fe8f928f54f9e49913 : SUCCESS in 20m 25s","accounts_in_message":[],"_revision_number":5},{"id":"ce4d396a2aff4e6af6cc9d8a7eccdf2167eb1316","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-21 16:30:20.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"de23269a899baba36a7ae0c92c2a8e0aea1ed343","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-21 18:18:42.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ab7eaa1e08b64eb68136acfdde9f05f5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65e18a58e4624a599e8101b7cf825c7d : SUCCESS in 5m 49s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/5d3842bff88f4cd0a3a1ffa0641e2983 : SUCCESS in 5m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ccad02b999dc47d4aca55101c3d71e89 : SUCCESS in 6m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/94e569a883914b66b90afdd194cedbd6 : SUCCESS in 5m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0810592073b64ca69b2e8b072c2307fb : SUCCESS in 4m 30s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2bc16d539e154cc7b63774bdc08edaa1 : SUCCESS in 7m 09s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/43aecb00f2b94927a20a0959ffc6d170 : SUCCESS in 11m 50s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3cf0ce93122143ba9a4095468c41a905 : SUCCESS in 13m 16s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/0e112f240e17424ba6332f78591cd81c : SUCCESS in 11m 34s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c1b54c3a0ac446d6af1713f01349e40f : SUCCESS in 11m 53s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/5bc1c97d3c7c45448702eb034024af8a : SUCCESS in 15m 33s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/341a3e0827bd4a0abc0c9c5d6290552e : SUCCESS in 18m 22s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/771dd0af2f814f109ea54398ac40ea3a : SUCCESS in 11m 34s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d184c05e8ce446b6bf14e2f106998cd2 : SUCCESS in 18m 29s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/1eb9de87f2da412ea801921acf7f53a7 : SUCCESS in 1h 00m 42s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/2b80f22b041842c8be24a419f9af95f4 : SUCCESS in 1h 05m 32s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/5e09df5e7149462490cb2dc007559a45 : SUCCESS in 1h 04m 30s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/62e3d6d7a67b41d0b77991550bdf1a9b : SUCCESS in 1h 09m 27s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/1697983a7f264305867e2677a935dd0d : SUCCESS in 1h 02m 02s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/d802323cc0114640a22ea07b4c5d0021 : SUCCESS in 1h 41m 58s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6acb37cbf2a149feb0c79d7080fee116 : SUCCESS in 1h 47m 36s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/40424b0456844cc492eb217062ad4ca9 : SUCCESS in 15m 17s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/dc77119bfbe64f24bd79db40fcf173d6 : SUCCESS in 14m 45s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/53e78477331843ccb2d9105dc101de21 : SUCCESS in 16m 10s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/a60bb6c91e324df0a22969e7e36e7dbc : SUCCESS in 15m 38s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/568fcbf467e64d319d5c17282b939aa8 : SUCCESS in 28m 46s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d32cdf11d08548a58220e3cbcd1ec680 : SUCCESS in 38m 28s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/0bdb7e5b0a7048708324b18d561c55ea : SUCCESS in 16m 09s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/28d3b6ebdc5a42d6a6f23d1cd1e9c876 : SUCCESS in 19m 43s","accounts_in_message":[],"_revision_number":5},{"id":"c293fcdf993bc5eb36683df00a18a969546c521f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-21 18:18:43.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"99b8b00e539df091f23d7de4b79010e2e68b469b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-21 18:20:18.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5e398d14f48446788cdb9daadafec4fa\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/155fec2e8bb9491292b21d7272a71a20 : SUCCESS in 1m 02s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a76197cc74134254acbba0ea33a5871c : SUCCESS in 52s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"65cc58144a50e86a861bd1e7eafe5d055eebb755","revisions":{"05bbb17fe90c2541037f6230656e6270e2c97233":{"kind":"REWORK","_number":1,"created":"2026-07-16 09:45:29.000000000","uploader":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"ref":"refs/changes/02/997502/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/02/997502/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/997502/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/997502/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/997502/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/02/997502/1"}}},"commit":{"parents":[{"commit":"9a6e4333b2aa2fbe8dc80a93cd5aa3e2f1872970","subject":"Merge \"Adds seed IP to default no-proxy list\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9a6e4333b2aa2fbe8dc80a93cd5aa3e2f1872970"}]}],"author":{"name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","date":"2026-07-16 09:39:43.000000000","tz":60},"committer":{"name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","date":"2026-07-16 09:45:22.000000000","tz":60},"subject":"Adds missing reconfigure.yml for manage-containers","message":"Adds missing reconfigure.yml for manage-containers\n\nCloses-Bug: #2160941\nChange-Id: I5c99da4c4ecc4995a31e96ffc9de4fe2620cdecb\nSigned-off-by: Leonie Chamberlin-Medd \u003cleonie@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/05bbb17fe90c2541037f6230656e6270e2c97233"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/05bbb17fe90c2541037f6230656e6270e2c97233"}]},"branch":"refs/heads/master"},"583d410ba07420fce6d8bdba2e8f316c9319c7e2":{"kind":"REWORK","_number":2,"created":"2026-07-16 10:07:32.000000000","uploader":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"ref":"refs/changes/02/997502/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/02/997502/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/997502/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/997502/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/997502/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/02/997502/2"}}},"commit":{"parents":[{"commit":"9a6e4333b2aa2fbe8dc80a93cd5aa3e2f1872970","subject":"Merge \"Adds seed IP to default no-proxy list\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9a6e4333b2aa2fbe8dc80a93cd5aa3e2f1872970"}]}],"author":{"name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","date":"2026-07-16 09:39:43.000000000","tz":60},"committer":{"name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","date":"2026-07-16 10:07:24.000000000","tz":60},"subject":"Adds missing reconfigure.yml for manage-containers","message":"Adds missing reconfigure.yml for manage-containers\n\nCloses-Bug: #2160941\nChange-Id: I5c99da4c4ecc4995a31e96ffc9de4fe2620cdecb\nSigned-off-by: Leonie Chamberlin-Medd \u003cleonie@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/583d410ba07420fce6d8bdba2e8f316c9319c7e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/583d410ba07420fce6d8bdba2e8f316c9319c7e2"}]},"branch":"refs/heads/master"},"f2eb48a58684482e78b705dc257be89dafee45a2":{"kind":"REWORK","_number":3,"created":"2026-07-16 10:12:40.000000000","uploader":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"ref":"refs/changes/02/997502/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/02/997502/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/997502/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/997502/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/997502/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/02/997502/3"}}},"commit":{"parents":[{"commit":"9a6e4333b2aa2fbe8dc80a93cd5aa3e2f1872970","subject":"Merge \"Adds seed IP to default no-proxy list\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9a6e4333b2aa2fbe8dc80a93cd5aa3e2f1872970"}]}],"author":{"name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","date":"2026-07-16 09:39:43.000000000","tz":60},"committer":{"name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","date":"2026-07-16 10:12:32.000000000","tz":60},"subject":"Adds missing reconfigure.yml for manage-containers","message":"Adds missing reconfigure.yml for manage-containers\n\nCloses-Bug: #2160941\nChange-Id: I5c99da4c4ecc4995a31e96ffc9de4fe2620cdecb\nSigned-off-by: Leonie Chamberlin-Medd \u003cleonie@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f2eb48a58684482e78b705dc257be89dafee45a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f2eb48a58684482e78b705dc257be89dafee45a2"}]},"branch":"refs/heads/master"},"7f7605e0761e05454ced8e262b0f482fd1bc430b":{"kind":"REWORK","_number":4,"created":"2026-07-17 09:21:31.000000000","uploader":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"ref":"refs/changes/02/997502/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/02/997502/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/997502/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/997502/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/997502/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/02/997502/4"}}},"commit":{"parents":[{"commit":"9a6e4333b2aa2fbe8dc80a93cd5aa3e2f1872970","subject":"Merge \"Adds seed IP to default no-proxy list\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9a6e4333b2aa2fbe8dc80a93cd5aa3e2f1872970"}]}],"author":{"name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","date":"2026-07-16 09:39:43.000000000","tz":60},"committer":{"name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","date":"2026-07-17 09:21:22.000000000","tz":60},"subject":"Adds missing reconfigure.yml for manage-containers","message":"Adds missing reconfigure.yml for manage-containers\n\nCloses-Bug: #2160941\nChange-Id: I5c99da4c4ecc4995a31e96ffc9de4fe2620cdecb\nSigned-off-by: Leonie Chamberlin-Medd \u003cleonie@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7f7605e0761e05454ced8e262b0f482fd1bc430b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7f7605e0761e05454ced8e262b0f482fd1bc430b"}]},"branch":"refs/heads/master"},"65cc58144a50e86a861bd1e7eafe5d055eebb755":{"kind":"REWORK","_number":5,"created":"2026-07-17 14:18:28.000000000","uploader":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"ref":"refs/changes/02/997502/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/02/997502/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/997502/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/997502/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/997502/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/02/997502/5"}}},"commit":{"parents":[{"commit":"9a6e4333b2aa2fbe8dc80a93cd5aa3e2f1872970","subject":"Merge \"Adds seed IP to default no-proxy list\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9a6e4333b2aa2fbe8dc80a93cd5aa3e2f1872970"}]}],"author":{"name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","date":"2026-07-16 09:39:43.000000000","tz":60},"committer":{"name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","date":"2026-07-17 14:18:10.000000000","tz":60},"subject":"Add missing reconfigure.yml for manage-containers","message":"Add missing reconfigure.yml for manage-containers\n\nCloses-Bug: #2160941\nChange-Id: I5c99da4c4ecc4995a31e96ffc9de4fe2620cdecb\nSigned-off-by: Leonie Chamberlin-Medd \u003cleonie@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/65cc58144a50e86a861bd1e7eafe5d055eebb755"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/65cc58144a50e86a861bd1e7eafe5d055eebb755"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
