)]}'
{"id":"openstack%2Fkayobe~997692","triplet_id":"openstack%2Fkayobe~master~Ie3c0d8f1b326705b84ac49f8bc94da8f07827f8c","project":"openstack/kayobe","branch":"master","topic":"dns-resolvd","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2026-09-14 08:16:59.000000000","reason":"\u003cGERRIT_ACCOUNT_28048\u003e replied on the change","reason_account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},"35345":{"account":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"last_update":"2026-09-14 11:00:28.000000000","reason":"Change was submitted"},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2026-09-14 11:00:28.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2026-09-11 21:11:57.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"34940":{"account":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"last_update":"2026-09-14 11:00:28.000000000","reason":"Change was submitted"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2026-09-14 11:00:28.000000000","reason":"Change was submitted"},"35263":{"account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"last_update":"2026-09-14 11:00:28.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie3c0d8f1b326705b84ac49f8bc94da8f07827f8c","subject":"Enable systemd-resolved configuration for Debian based distros","status":"MERGED","created":"2026-07-17 08:45:05.000000000","updated":"2026-09-14 11:01:34.000000000","submitted":"2026-09-14 11:00:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":24,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"997692-dns-resolvd","meta_rev_id":"7668eed94b0b72609dfe8760d52d88ca0810fc05","_number":997692,"virtual_id_number":997692,"owner":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-14 11:00:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2026-09-14 08:16:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},{"value":1,"date":"2026-09-11 21:11:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2026-09-14 08:16:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"}],"CC":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-17 08:59:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-07-17 15:14:11.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"CC"},{"updated":"2026-07-22 15:59:44.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2026-07-29 13:23:50.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"},{"updated":"2026-08-03 17:38:19.000000000","updated_by":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"reviewer":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"state":"REVIEWER"},{"updated":"2026-08-03 20:32:18.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2026-08-05 14:08:56.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2026-08-05 14:48:04.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"}],"messages":[{"id":"eb18927c0ae0fcebec1526c26b5dbd1c094df5a9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2026-07-17 08:45:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1941296449411039a54c02fca630fb738cecd498","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2026-07-17 08:46:31.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6a195319adfcc76329d448abdcb8cf9c9f65afca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-17 08:59:25.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/986eed19c16048d989b66f483b3d6cc3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5be3ac301b944378f8eaee1a3090dca : ERROR Project github.com/stackhpc/ansible-collection-linux does not have the default branch master in 9s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/d222072f3a534d8da56d0692852e7421 : ERROR Project github.com/stackhpc/ansible-collection-linux does not have the default branch master in 5s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9d973914dd0642d8af8e36bc71fdd9e5 : ERROR Project github.com/stackhpc/ansible-collection-linux does not have the default branch master in 7s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f04c9755684245359d918200f2ebb8ad : ERROR Project github.com/stackhpc/ansible-collection-linux does not have the default branch master in 5s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/30c95c15975d4f859d5efe0a1c39656f : ERROR Project github.com/stackhpc/ansible-collection-linux does not have the default branch master in 5s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e82b201328de48259a1b561dad2f98fa : ERROR Project github.com/stackhpc/ansible-collection-linux does not have the default branch master in 5s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3c19c581ca32482294fed380474b5f19 : ERROR Project github.com/stackhpc/ansible-collection-linux does not have the default branch master in 5s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/80e2259017584f2ea7658dfeea594ff1 : ERROR Project github.com/stackhpc/ansible-collection-linux does not have the default branch master in 8s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/3937b6a59d024eb6904c9d95f86487ba : ERROR Project github.com/stackhpc/ansible-collection-linux does not have the default branch master in 10s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6cadf078b04f4ce881e02b1a41ca4dee : ERROR Project github.com/stackhpc/ansible-collection-linux does not have the default branch master in 6s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/df95ed7f6e5c4badac55407eac8aaa37 : ERROR Project github.com/stackhpc/ansible-collection-linux does not have the default branch master in 21s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/c6d8b4468e4d4a26842bfe1ab2a0e98b : ERROR Project github.com/stackhpc/ansible-collection-linux does not have the default branch master in 19s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ddb2da09f080438d984f2907d9c6975c : ERROR Project github.com/stackhpc/ansible-collection-linux does not have the default branch master in 12s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/97a5fa310faa489494eb38c6cf634852 : ERROR Project github.com/stackhpc/ansible-collection-linux does not have the default branch master in 27s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/fc17e9bfd755417c911548f7618c4260 : ERROR Project github.com/stackhpc/ansible-collection-linux does not have the default branch master in 15s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3a64c83cb92e4b1ebf4a0f1161db09b4 : ERROR Project github.com/stackhpc/ansible-collection-linux does not have the default branch master in 33s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/c23d177301d1470a95652d8250038f6e : ERROR Project github.com/stackhpc/ansible-collection-linux does not have the default branch master in 19s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7bd593287b854681b350fd356aa7b33f : ERROR Project github.com/stackhpc/ansible-collection-linux does not have the default branch master in 18s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/5122a4016f3042de8f3758b1561a51d6 : ERROR Project github.com/stackhpc/ansible-collection-linux does not have the default branch master in 17s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/704605b3a4cc4eaca47f32c8042b1ae7 : ERROR Project github.com/stackhpc/ansible-collection-linux does not have the default branch master in 12s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/8192c3b6d7614c2e9898d8902524c478 : ERROR Project github.com/stackhpc/ansible-collection-linux does not have the default branch master in 14s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/37cc0f103001483fabc00f0fe2b0a0bc : ERROR Project github.com/stackhpc/ansible-collection-linux does not have the default branch master in 13s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/47495a9c9e2e4d4c821a984fa04e9d53 : ERROR Project github.com/stackhpc/ansible-collection-linux does not have the default branch master in 23s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/609912d5a220455fb7e423662d776dcd : ERROR Project github.com/stackhpc/ansible-collection-linux does not have the default branch master in 22s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/97d1a677b7b245bea58a6ad522914280 : ERROR Project github.com/stackhpc/ansible-collection-linux does not have the default branch master in 26s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/a3f8d73148f04926a79c1db048311f52 : ERROR Project github.com/stackhpc/ansible-collection-linux does not have the default branch master in 13s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/449c7960a0af4a9d847c0ace0f603b5d : ERROR Project github.com/stackhpc/ansible-collection-linux does not have the default branch master in 22s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0a89154a9f6b4903aed160329bef8ec9 : ERROR Project github.com/stackhpc/ansible-collection-linux does not have the default branch master in 32s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/15d906f0df8c4330b8a0a1eef7dfa57c : ERROR Project github.com/stackhpc/ansible-collection-linux does not have the default branch master in 11s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/3e922d4abf56408db16439fb22730025 : ERROR Project github.com/stackhpc/ansible-collection-linux does not have the default branch master in 16s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8773dc97717a48b1afe20868776f6cf6 : ERROR Project github.com/stackhpc/ansible-collection-linux does not have the default branch master in 20s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/a4e2afc0e1c84ba39ed01f35760430a5 : ERROR Project github.com/stackhpc/ansible-collection-linux does not have the default branch master in 22s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0641aebeb6944ef8bb4b19bbef2c69e8 : ERROR Project github.com/stackhpc/ansible-collection-linux does not have the default branch master in 18s","accounts_in_message":[],"_revision_number":2},{"id":"b0fae7d2c48ab131692d4c8083e7ddb16e2e6d79","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2026-07-17 10:25:44.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a797c984246ca766ced01f8ddd8abed10eac896f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-17 12:17:24.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e20b4a8332744b95920d99f06a585639\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3f01d064aa34b0cbb470e34edc34a47 : SUCCESS in 8m 47s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/46f227208b184ba7aba0507d546dd6d7 : SUCCESS in 6m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/da7afe41c1f1489b8ef5feaa9b2fd244 : SUCCESS in 4m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8bc29de33d79470ba423c06c9b664925 : SUCCESS in 4m 16s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1fb365af66d247a3a20f438be0cd16c5 : SUCCESS in 6m 30s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/db97dadfc36645ea80ec153f1416047b : SUCCESS in 6m 57s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/34fcf10413c04482b911a2b829bb2f9a : SUCCESS in 5m 47s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/5c408e9e754a475b8d612e885b543e98 : SUCCESS in 4m 50s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/97765c51a4954a6cb330bb107d73e47e : SUCCESS in 11m 12s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/36015bec65d64170b7142e6435e1f89d : SUCCESS in 14m 40s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/a1a4c9f5ba554d9181695a44a13b8dab : FAILURE in 17m 48s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/7ec5b3d951a24813a3b0e3a81816605f : SUCCESS in 16m 39s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/54ed64594a054f7ea1f290846de738ce : SUCCESS in 20m 56s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/e59b8b802ff047ab84a7d63199217db4 : SUCCESS in 15m 22s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/df626e0fa4af4617a327273e383b2429 : SUCCESS in 16m 30s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dcc80aa8cd52404ca637203dda14623b : SUCCESS in 18m 36s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/7ef011227ac34f4a847ed9e92413a36b : SUCCESS in 26m 20s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3b19bd24e3444f059a29c119a23c4234 : SUCCESS in 16m 33s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/f626e753c60e4228be3bd5b0926fed3d : SUCCESS in 1h 02m 32s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/06d0897ef74d46338fe9e87c498eec1a : SUCCESS in 55m 23s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/f85ba1a1e51b466aa63bd803a3b30bcf : SUCCESS in 53m 17s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d9309dc39c5248b0a13a13007d8c4280 : SUCCESS in 1h 07m 14s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/186d75bb9dda4faba3ff9890f6d39e11 : SUCCESS in 1h 07m 02s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/72e47f9d979240c9bdee045808fab5e5 : SUCCESS in 1h 49m 32s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/39797ffdf20345658b5d2dfd789ee324 : SUCCESS in 1h 20m 23s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/2aca3b2509584b17a1a2d94c9aa2e47b : SUCCESS in 23m 47s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/de3ea88b90ad4e17835aaaa5c0e5710c : SUCCESS in 23m 10s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c306cfb42c7b43e48e541a839686fff4 : SUCCESS in 23m 54s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/90c4cd8acf494d618741cf7436794021 : SUCCESS in 23m 06s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/3cf5f915cc7748259e07c84b5feab234 : SUCCESS in 39m 53s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f0a28d6925904c2fa363e73936e6fdf7 : SUCCESS in 34m 12s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/a5477200b9584980ae73557b18cd7a27 : SUCCESS in 16m 49s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a98009b8ac6b4d6ba27b02ebcce537dd : SUCCESS in 20m 09s","accounts_in_message":[],"_revision_number":2},{"id":"8c854cf0375ecd3ce24e3fd9be4367d490284042","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2026-07-17 12:30:05.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"583854906bd2944fe1f8c68807276690d0056010","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-17 13:34:55.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/618563a1daa04b0c97c8d49912399eea\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea1732fe5e004c05be69eb843a0429a2 : SUCCESS in 10m 39s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/0204afd5b7c944819dff59ac75206e5a : FAILURE in 4m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a1b8555dbf994060a5839b54862ae1cb : SUCCESS in 4m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb1d2ebc329e4d08ab0bbfbdd1e7b8e5 : SUCCESS in 4m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/44a3d6f0b7174a2cb8d5c1f40d99dc29 : SUCCESS in 4m 47s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c7335e9cbe65428ea6f495c108ee07c8 : SUCCESS in 7m 14s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/473fbd1eaa8d4bf7ba573f7f89dc0db9 : SUCCESS in 6m 36s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ba9b7cf790f941e1ab78d4f8746d77fd : FAILURE in 7m 34s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/fdaf147aeff0471d9a2329d5ef91dacf : SUCCESS in 13m 30s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/4906d315d7544bbbabe69f03d1d79c1e : SUCCESS in 11m 50s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/a50b957c92c640f5a231f349af504aa6 : FAILURE in 7m 37s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/52c63df202d24dde8bb6f331fad06832 : FAILURE in 8m 17s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/84a9388d9ec6425a9a510ca43f1e2b68 : FAILURE in 8m 26s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/69cb3f5abd2c4794a88a0d2f13623ca2 : FAILURE in 7m 57s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/f824890194134d8bbba77eddade6702f : FAILURE in 8m 26s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/800328405bd441b7b772f5ed373b0195 : FAILURE in 8m 11s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/c3d3a99cec1f49248970e84640cd918b : FAILURE in 8m 27s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/25e20d8ac7c3421da56cf0d399277dab : FAILURE in 7m 58s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/c2edaed31861465da4351460c0cff162 : FAILURE in 9m 34s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/5803811be7a5426ca57db08b8d674ed0 : FAILURE in 8m 41s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/3c11e94dcae5431c8d74ac76d2581496 : FAILURE in 8m 06s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/37ffb160ed354a92bc5434b189e627c3 : FAILURE in 8m 18s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/c3b8ef5b1aee4404958988e445959c99 : FAILURE in 8m 09s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/4bdb64fb110d454d9fa30d239abe9efd : FAILURE in 1h 03m 24s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c4b38a85bf4d4070a04cfdd37c57a3fc : FAILURE in 1h 02m 54s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/7f136c6f276448cdb05e7d465b9f1ff6 : FAILURE in 7m 15s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/10cccaf7d2c74dd0a26ab9bfcbf5e897 : FAILURE in 7m 29s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/17a7b0852f3d42e7904585c055cc3ad4 : FAILURE in 9m 08s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/7f6d7f5de7af4bcc9f30df0bf1a6bb87 : FAILURE in 7m 44s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/6296afb410dd4892b8e3e7f172a1181d : FAILURE in 41m 31s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e3c4da8d2e284ce591562917c9341e24 : FAILURE in 42m 38s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/a5c88d7d4d59446baa555970ee30ab9e : FAILURE in 9m 15s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4d3b41f2acff4e57b4831d7ffb6a2579 : FAILURE in 10m 04s","accounts_in_message":[],"_revision_number":3},{"id":"914d2d3e5c52a22f9b44a5ac66f956474ba2e49e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-07-17 15:11:40.000000000","message":"Patch Set 4: Published edit on patch set 3.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"64d75f79c9a06128ef181d4b0b6f805be2dbe532","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-07-17 15:14:11.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"00cc62a8e5046d4f245775b89e279c9c9c0b94c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-17 17:42:33.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c55f8118b0204b2a97bf12ffac9a63e5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3706fd21eaf478f8247acc8f34e9c2e : SUCCESS in 9m 19s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/d06febcd4e99491b844133676b3ea242 : SUCCESS in 8m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/214bcd82d52b4d6a86d0bf2398354d46 : SUCCESS in 6m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f320fa8bb80e44f19c6110c58b3642a9 : SUCCESS in 4m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/47dc2f79020044f6a6132aef3a2fa78c : SUCCESS in 4m 55s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/79b2939f5c12476b8b75d150063b5dd3 : SUCCESS in 7m 03s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/afd3a75f9e6644c5a332427c1e667439 : SUCCESS in 7m 22s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/62da61bb2ee34845afb381579acf2782 : SUCCESS in 7m 32s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1dbeb6fd360246c59ca9ccc234077e06 : SUCCESS in 14m 17s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/02b1a9089ad443cf8589884183d9391d : SUCCESS in 12m 06s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/337878f00f6948c49be1ad2b366018b9 : RETRY_LIMIT in 6m 15s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/927a429b6b80402fbc060036e9defc4b : RETRY_LIMIT in 6m 06s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c61399fd8d6949de997f0d83c30152fc : SUCCESS in 21m 14s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/2b6ff1be2e044e62ba549754b1ed03b1 : SUCCESS in 22m 01s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/8114818a1fe44f28952668f8612c5bad : RETRY_LIMIT in 4m 28s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a35db81d2e4e47ef959f0ff09cb2acfd : SUCCESS in 21m 59s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/e783479857fe4f1ca9b358e761d0ccdf : RETRY_LIMIT in 5m 34s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ab3f53372187449990b831463d0744cc : SUCCESS in 19m 05s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/ae61b15e9611482a8304596ac147ba00 : RETRY_LIMIT in 4m 47s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/c973ddd8728b4e5ba379b3f269bffa3e : RETRY_LIMIT in 5m 35s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/7c5981f392b44325b416824125edead2 : RETRY_LIMIT in 4m 30s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1c0b29a2d62942ae92b86db5bab6fabc : SUCCESS in 1h 05m 44s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/e8426c545c144ae2bd8fe48a7209224c : SUCCESS in 1h 10m 46s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/2ff35c4b00594e1aaaf2fe30aa22cc67 : FAILURE in 5m 34s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e0a252a7d2c146359bab9107a4a8bc5c : SUCCESS in 2h 23m 52s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/990f80aef9474d2fb10b037227de34fa : SUCCESS in 26m 47s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/c7a21e4863cb41b28e80ae3b599cf779 : SUCCESS in 23m 05s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5f87540533384b48b8b8f18f57a11e9f : SUCCESS in 26m 13s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/901e5af3bd7443918d1f5a6a6e8096dd : SUCCESS in 23m 25s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/5ceed961a4924f1897a03a943510e0f1 : SUCCESS in 45m 40s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/af053a5cbb724a398879ee040ff9f2e0 : SUCCESS in 41m 47s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/dba09a13ae484faca66395e98893210c : RETRY_LIMIT in 3m 42s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/938447edb7b34741bedf3ef2b972b2ce : SUCCESS in 21m 36s","accounts_in_message":[],"_revision_number":4},{"id":"49efbb67b8b1d76a6820fec20e843e5beef73f7d","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2026-07-20 07:57:59.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"e214610bbe7bf6435fd06387d100eaf28c0163dd","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2026-07-20 07:58:33.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"ff56f7c82346fdff510fbfd2a159c0d5f91da177","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-20 09:49:26.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2ea88f53071c417f97816cf6531eb980\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a69920ef7f5b42c4bcb7bb94143a0e71 : SUCCESS in 10m 27s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/5f2aac07c5594d36a0935573851d7008 : SUCCESS in 6m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/edbd8373bc9e444496b87dccf90a5597 : SUCCESS in 4m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/36bab2c1a0854b72a292b409b51431d1 : SUCCESS in 4m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7f5ffef714df43ae91bd56744ef96df1 : SUCCESS in 4m 20s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a7619b56b4674704831ab0d83c474fed : SUCCESS in 6m 53s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/23f5622521d944afb91f3d0d2a6531da : SUCCESS in 6m 05s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/5755d8f634e04ad1aff3e65e466cf63e : SUCCESS in 5m 29s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e5a9855be0c145f68fae8118efd506a6 : SUCCESS in 15m 42s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7641ccb8e45245df95c77f309b26c22e : SUCCESS in 11m 57s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/83fff9eb7f2e45859ac8d5de65d95f3d : RETRY_LIMIT in 4m 53s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/9fdbf456af3b4436b530145a0cfa2269 : RETRY_LIMIT in 3m 29s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/83261c5e0d41432594b13483b8ee981c : SUCCESS in 17m 57s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/16693837e5354c8eaec0631ef011b7d3 : SUCCESS in 16m 38s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/ce92e01da85f4c048bbcf3b377d5aabb : RETRY_LIMIT in 2m 15s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1c684d4334f2461aa106aa4c74cce6bb : SUCCESS in 19m 03s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/671dbfad1f72430b90385788fee16840 : RETRY_LIMIT in 4m 41s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4e10615033a240cf8942adb8178f8926 : SUCCESS in 17m 38s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/4a0120158ac9416d8065f6507f2b85a7 : RETRY_LIMIT in 4m 34s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/d6ab576fb4ab4d108e4078849d557f2b : RETRY_LIMIT in 4m 02s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/d3ce3f988ad3445abf6a85f344dca1e1 : RETRY_LIMIT in 4m 03s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dda1302f3536466085e10a5ae86f9f63 : SUCCESS in 1h 01m 16s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/c285fd3e2d0741f09ec10bce6382541a : SUCCESS in 1h 07m 11s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/2af2ef953b2c43c2a16a38c83666d85d : FAILURE in 5m 00s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d6c834cb9a8b4f27a9ace13ebe3f2ce8 : SUCCESS in 1h 49m 53s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/c379f86e2a1a438ba673ff9d6d78683e : SUCCESS in 14m 59s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/eedce3f87379472fa0d69ce302364226 : SUCCESS in 24m 24s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ecd1c3e34f024b0892216bd269c4d700 : SUCCESS in 22m 42s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/2416efb6aa6f4932bb00d724b0c254c0 : SUCCESS in 23m 02s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/bdc722e30e0f4c1fb7cdb09d8d7fa35f : SUCCESS in 38m 23s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/416945ceec5b4e79acbb0f9fb533c024 : SUCCESS in 40m 06s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/4007c96599d043febe73172db23efa06 : RETRY_LIMIT in 2m 44s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5a6aaa1e13f34ddd9ad65966897e80a7 : SUCCESS in 21m 25s","accounts_in_message":[],"_revision_number":4},{"id":"3648f085571069d3ec6b4ced2e390d5a5f31e543","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2026-07-21 07:23:45.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"6998f3dc5c4a132234fbf64da7fb27e9a0223d2b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-21 09:14:34.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9578a6b8ee7c40a28405158838973825\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9fdb9c6f8c2d4718bd3bb75b918582f4 : SUCCESS in 9m 35s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/38c02c8f02284f7390b02baf98c5b989 : SUCCESS in 6m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0f747fb1f003451b882740eef8a7a529 : SUCCESS in 4m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b6dbea22dbac4f11b80be8d1d7ac4da2 : SUCCESS in 5m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/48b37aed0d76409f811fbb5826c942e1 : SUCCESS in 4m 47s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a10dbeeb63f94b55ac5b2e5c3599ae7f : SUCCESS in 7m 01s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6c9255099151479f9d7d69bd956e78df : SUCCESS in 6m 04s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/1e0702372d2342e0b7c5a48b68eb376c : SUCCESS in 5m 22s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/fc6331ff1c194e9694614809341c07e1 : SUCCESS in 11m 53s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/bf41f05cd8bb4f0493c3bbb6237f6ae5 : SUCCESS in 12m 23s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/ab73a5b512a743eba6561aad892d4b5b : RETRY_LIMIT in 3m 58s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/0d7691106ecf4a8f855a2cff9f5fa472 : RETRY_LIMIT in 3m 39s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b43edb1736254e1e922700aab4498f83 : SUCCESS in 14m 52s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/de975c5b4e994005a0c0e49384d73b89 : SUCCESS in 17m 22s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/3fa277d05fd8410e8d2721f162516b11 : RETRY_LIMIT in 2m 56s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1154cc7002ff40caa928cf9f83145d8c : SUCCESS in 20m 36s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/b7a4a4cc55de43c992385c7faac23b90 : RETRY_LIMIT in 3m 33s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2167b98a191a4000b66c69658920866c : SUCCESS in 17m 39s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/33eb739173c54e369a789a1ec3ba1d3c : RETRY_LIMIT in 4m 18s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/0e84ffde68564487825128c2da5189d6 : RETRY_LIMIT in 4m 17s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/55ec4951860a494d9de60b29ff49c19f : RETRY_LIMIT in 4m 36s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b23929e015244fb089d8d7b95707bf7e : SUCCESS in 1h 01m 36s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/86a05ddf5a744b1a9a79ceb2bd9c7dfa : SUCCESS in 1h 06m 36s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/4706912712b642289b6e9e6cca20b0aa : FAILURE in 5m 10s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/177771a1592b40d281c05aef1b5b4fa8 : SUCCESS in 1h 49m 33s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/72db29db25934387b26f76bd784d95c4 : SUCCESS in 16m 30s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/462dc0b5df08429e82ba750b1d578d3f : SUCCESS in 27m 09s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/03cf2ccf7da943c08d3d93c80bcf0a18 : SUCCESS in 23m 59s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/c55f5f00e2114eb8b7d51dd440b5d5cc : SUCCESS in 22m 08s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/6effdd088584453fa2c5804ec45785b1 : SUCCESS in 21m 16s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9c6dc1db2f8b432d8725b4c22761ea33 : SUCCESS in 34m 51s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/f6424f6bf4084cfeaef8fa9c23adb641 : RETRY_LIMIT in 2m 49s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c33f03485d654add95a5026e6e375a3d : SUCCESS in 21m 06s","accounts_in_message":[],"_revision_number":4},{"id":"0a134dc4d6329f5227064f17ed890fae8ca68932","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2026-07-21 19:01:42.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"9a90fc17a20ddecea7cb7924b6815c4864f476a9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"real_author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2026-07-21 19:02:15.000000000","message":"Patch Set 5: Patch Set 4 was rebased on behalf of \u003cGERRIT_ACCOUNT_28048\u003e\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}],"_revision_number":5},{"id":"82dd91e22d9af8c79253244e34128922004d234f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-21 20:59:52.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/59593d194b504d3783933f3945ba62e3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/15a9c3cf90ab445ca2dee381c9685eb4 : SUCCESS in 10m 23s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/b91a0ef44e82464ba3a4a0bef37a4882 : SUCCESS in 6m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/805d06df9183407087b1d27dfc81d62d : SUCCESS in 4m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/65c4cde0662e4896ac806d12deb6b84c : SUCCESS in 2m 34s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e0651a1f60a64048a6fad27ee8ae6cd2 : SUCCESS in 4m 30s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c5dfc470d73642b4a5bfce7c929e4b8a : SUCCESS in 8m 43s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a3d322bd20434b7782a16ef15cba92d4 : SUCCESS in 5m 52s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/22474eb7e3514964b2380dff6a2a19d4 : SUCCESS in 5m 08s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/41b5b03e12204f6f9ce32b3669d6359e : SUCCESS in 11m 46s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/653f5dd7819f43fbad209f6adfcf0bae : SUCCESS in 14m 04s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/6dc31b54de5d413ca68e9cd51528b9dc : SUCCESS in 17m 06s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/1c0f3f31c581455681a23ecd8927f9aa : SUCCESS in 21m 32s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6630d763ce734ac09dde64b87f30a609 : SUCCESS in 15m 12s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/c489345ed7db425f94cda38c31cd1708 : SUCCESS in 15m 06s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/b17ec1e5d0ed4b9395e3843e3b941d7d : SUCCESS in 20m 12s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e66100c500b04cb592799abd8c745fed : SUCCESS in 18m 48s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/3481d1cc280f49bc888e23854ab340b6 : SUCCESS in 23m 42s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/911e5a3f8c114272810341e378a5f962 : SUCCESS in 22m 15s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/05aa3f78e75a444d82b82cb1dc62713e : SUCCESS in 1h 01m 35s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/53355f3b2ad34980bd85f64c0d0df9ee : SUCCESS in 1h 03m 24s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/2fab79e9ab4a495aaea813be8863816e : SUCCESS in 1h 14m 53s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/72f308a79baf40a586442d2917ea1f9b : SUCCESS in 1h 06m 55s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/e1463897e71f481fa99953a80670458d : SUCCESS in 1h 02m 03s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/c4a086b8e9994ca595887709dd01fe84 : SUCCESS in 1h 49m 48s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0ec30fbe45ab4282b4fc3f4cd832cd33 : SUCCESS in 1h 55m 02s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/14f157b62e114571aa9f1779fef9b075 : SUCCESS in 21m 25s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/41021a092cd5401b81eb17325353bdb5 : SUCCESS in 21m 05s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a2815e186b8f4ab38bb17b301b1d0446 : SUCCESS in 22m 32s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/168aeb9bbe8b4dc59772057daed7684b : SUCCESS in 22m 21s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/430cb8c9be70433cbb03b8c4872ba0fa : SUCCESS in 23m 53s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d9c713da911a4fb1b572e3644a0d1529 : SUCCESS in 34m 55s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/c9f86217413e472594415b13ef0107b9 : SUCCESS in 20m 23s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e6c1645244fd4b528c210a0fe0b88b79 : SUCCESS in 20m 02s","accounts_in_message":[],"_revision_number":5},{"id":"786ac5dddf40806f8cc24c1df84c3d81c696c797","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-07-22 15:58:46.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"6d78186f1c89d502de880fb28a18724c449f8160","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-07-22 15:59:44.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"829fe6fe46b8becbbb3a862cc3145f32a0a7cb5b","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2026-07-24 08:45:17.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"9ed284c6e72284109be32bd4605bf34141b16289","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"real_author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2026-07-27 11:10:58.000000000","message":"Patch Set 6: Patch Set 5 was rebased on behalf of \u003cGERRIT_ACCOUNT_28048\u003e\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}],"_revision_number":6},{"id":"726d291f463d640d7c0df6df431de82be7e737d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-27 13:07:11.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b1f48e3ba9c04959b026a58c1723a3d5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e69c922ed811455c84f39bbb93afd207 : SUCCESS in 10m 10s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/8fcadbb809554fc2b6d126fc1e3190b0 : SUCCESS in 7m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/80718c55af9d4f649b7c2dfbd88778c9 : SUCCESS in 4m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a61c145598544e78ee50778dc074662 : SUCCESS in 4m 20s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b4f2860bce234316848f3f811922fbfe : SUCCESS in 4m 01s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f5952e6beac94d26a87e249ab5995f79 : SUCCESS in 6m 59s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b020a6ad425b4b25ae2462d0b547f750 : SUCCESS in 6m 08s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/fe6e5c5b3fc44333b34c77b851ef3ded : SUCCESS in 5m 28s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8afcdb6338c1495b97ebb68018a5304e : SUCCESS in 11m 54s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1b7a62c0cd264dd1a8f64b8a12410615 : SUCCESS in 9m 08s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/85322d6635d943b180f5a53fb982e496 : SUCCESS in 21m 49s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/474593315a3f48898ce00ce820093eef : SUCCESS in 16m 27s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/aca96a1135c94873b5223660743e6c20 : SUCCESS in 14m 45s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/8ae31d43134d47068cc73c411032aa19 : SUCCESS in 15m 04s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/6d53f8c62047415485c50b0810477f9f : SUCCESS in 25m 23s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ae16120d2cdb4006892f3d33b2475e89 : SUCCESS in 19m 40s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/6d23fa2139624779b51cebb4398955f9 : SUCCESS in 21m 36s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6c35572f37ad4e7ea45f43269b3070ab : SUCCESS in 17m 57s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/ebecb68d69c84ad282c0905232dde926 : SUCCESS in 1h 14m 06s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/b4293b94c6fd4400b4ba813d660518fe : SUCCESS in 1h 14m 24s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/8ced792d63b64b0f97743b53bb2b1cbe : SUCCESS in 1h 14m 02s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/71a09ba771be46788d32323eddf4543c : SUCCESS in 1h 18m 19s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/3e13eaddb4584b8e811855eb806a6ce5 : SUCCESS in 1h 17m 42s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/600cdc99bbf84133ba0f2218da9d9597 : SUCCESS in 1h 52m 56s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e8daacc3c99e4126b0defaa7b30964a7 : SUCCESS in 1h 54m 38s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/872d73d858034abb87686f3b0ef11bcf : SUCCESS in 24m 56s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/63069a5f434e4881b4030b40674ce230 : SUCCESS in 24m 11s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4831b466e75c40feb2df9931467089b3 : SUCCESS in 23m 59s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/94f60b9f35df4880922909ffc95186e4 : SUCCESS in 21m 20s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/70ddbd46fe574d058ec1463bb4bfb977 : SUCCESS in 37m 42s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/547dfb7a0a7c457db7d99f1ba81e7a70 : SUCCESS in 33m 54s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/f34f899248f14208af1467f170dc4999 : SUCCESS in 21m 57s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b21f04750da34423bf8a74590eb75abd : SUCCESS in 21m 32s","accounts_in_message":[],"_revision_number":6},{"id":"db19de3730f881ca0ab64b325ed66227e5321970","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-07-29 13:23:50.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"d529ea02f460e83e7be5d381453acc93781a9d6c","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-07-31 09:15:24.000000000","message":"Patch Set 6: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"570748f783b0cafca420fc32e449a0dd0d955c19","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2026-07-31 10:51:23.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"66fcec3cd8829b4969e5335a11ae723411b20ae3","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-07-31 16:22:18.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"91158b304aeec122324f23789fd207f11ad4e3d6","author":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"date":"2026-08-03 17:38:19.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"dfff11eba10872510676f28e3a13251b8cb3b013","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-08-03 20:32:18.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"2fbf2477b44e181bf11ed18b30a94e7123dfd0a4","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-08-05 14:08:56.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"a1585020146607d13949318330b54cf606df4181","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-05 14:09:34.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"1812f41625e6272959719e8d7aac2ab5b1aa0f54","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-08-05 14:48:04.000000000","message":"Patch Set 6: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"d6f48859e2353fe3029ef791e529ef2fe349e840","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-05 14:48:22.000000000","message":"Patch Set 6: Verified-2\n\nThis change is unable to merge due to a missing merge requirement.","accounts_in_message":[],"_revision_number":6},{"id":"06d30fb6472ce445fc0893f5fa1b19239c6c7c37","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-08-05 14:54:44.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"af1e4b030bbe3d95e90d259bbb36c51e0694d0ce","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2026-08-25 10:49:05.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-2 (copy condition: \"NEVER\")\n* Workflow+1, Workflow-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":7},{"id":"6a0b4df621c85a56eb64714be7d4265bd268e5ca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2026-08-25 10:49:18.000000000","message":"Patch Set 8: Patch Set 7 was rebased","accounts_in_message":[],"_revision_number":8},{"id":"1f106eb5e8f03a3eb5cc2c5d05bbd9d20075b15b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-25 12:08:24.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e8119428cedc4a7496ce71489effa5b3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13348f6468c744ca81bb77c5aec5d2c9 : SUCCESS in 10m 20s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/24c28b8a1c574e75b8b916b7131adb35 : FAILURE in 5m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e5ecc1c28aa14c8e9014e470825e0400 : SUCCESS in 5m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd45c0a6403e4597a8b18e007afb24df : SUCCESS in 4m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c94a065aaaf64d3783e3ac6188551a98 : SUCCESS in 5m 45s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/fd6e05d1c8f94c17b906bb8837c184e9 : SUCCESS in 6m 56s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/94cf253d9af545ebbe68b479e2748749 : SUCCESS in 8m 13s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b350f50b59334e569b2c5545110e0389 : FAILURE in 10m 15s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e4766236a221448eb3265a06db3a5f2a : SUCCESS in 16m 49s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/06cd70fd6d174181a441268279140405 : SUCCESS in 15m 10s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/0f784d4cdeb945e9a4591e6f59247d29 : FAILURE in 13m 05s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/13e2ada18f874d87a7574e4e05afb9c2 : FAILURE in 9m 43s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3c0bf7272a114f70820ccfc505d9014c : FAILURE in 13m 51s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/f37c98c9421d405aa06682a41ab9b4e6 : FAILURE in 14m 38s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/760032d6ae6b450a986fb367383a909d : FAILURE in 7m 15s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/494caf07f866435ebe263ac7b36d5317 : FAILURE in 8m 43s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/ce75a39eecf14f868ae432f247711150 : FAILURE in 10m 26s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0a5621b5a3e7493eb1cdbb4e35be17f0 : FAILURE in 9m 25s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/60e606b9ae544c15aef8a03fbfac4591 : FAILURE in 8m 08s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/441b38f8b0ee4961abfa779770f6957a : FAILURE in 6m 37s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/c3a9ad00f15f4df08c6cad54484aeb8c : FAILURE in 7m 39s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/82464475211f4059a9548027888f807a : FAILURE in 6m 35s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/a7095c802146434488d3d71c04aea986 : FAILURE in 7m 22s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/f939ef99323a4e82a8022dc4d23b840f : FAILURE in 1h 06m 18s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/82f7b98ab1e94359b9cbf72fb2379e90 : FAILURE in 1h 17m 05s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/b0b65ccdc1104209b62dd480ab890f6b : FAILURE in 8m 10s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/2a9915d0e9da45d5bcfaf57f75786900 : FAILURE in 10m 51s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c3cd43d87e4c423894d2730d91fe9a85 : FAILURE in 13m 04s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/8f6996142250486b91eb2f7f84e95b04 : FAILURE in 9m 13s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/34c3c42bfe0742528ce083bd44d18229 : FAILURE in 29m 41s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/28b2e301fd4c44d3a2a698f6934216b8 : FAILURE in 29m 39s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/9f4b6b302f614386875c3e134aabd051 : FAILURE in 6m 22s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/90c5157eaa1a42acafac8a1e4ef9a7eb : FAILURE in 12m 31s","accounts_in_message":[],"_revision_number":8},{"id":"c57b4d05ea14cdacb9272e899aafb6720976b0cc","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2026-08-25 13:24:45.000000000","message":"Patch Set 8:\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"84c6cde61d47349e11c21de0eb3817745775f309","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2026-08-26 09:24:59.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"b63bde4eda06a2009f5367f07ab36e4bbd78e395","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 10:35:59.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/772c9f4329424a9787035546c2f9396f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e3d1cf72694a4dfdb16aa69ba6d31ef3 : SUCCESS in 4m 59s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/344d65f51dc84b839cf069eb6d958ae9 : FAILURE in 6m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e36c7e1292274022b49338a93f91af83 : SUCCESS in 6m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07bd1523062247a49bc2d642f43ccbbf : SUCCESS in 5m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7ac5e44e144b4d84880a44ffe254c53f : SUCCESS in 2m 39s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/2a2f4319a13e499c93544276df38dbca : SUCCESS in 5m 10s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/91e856ac9f804159a7d1e650224198a6 : SUCCESS in 3m 40s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6d2b4c3a69914ab985aa60665a9dc596 : FAILURE in 7m 17s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/889a232af3374a7aabd1255af1017f65 : SUCCESS in 13m 35s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a91a335f2908453fb9203c231a307551 : SUCCESS in 10m 45s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/9ef88c902b514c09a33518d2b325e882 : FAILURE in 9m 03s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/633263d33b8745f3bfc5d6759822edf3 : FAILURE in 7m 42s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c50df8b34a4a4a428ca00703311fb941 : FAILURE in 8m 50s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/20279f35e5c74ac1b45436c0462848ba : FAILURE in 9m 25s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/bc7f52cfcfe04daaabaadb65ae518c29 : FAILURE in 7m 32s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7d7cd780b1d749f7a753664906330fcb : FAILURE in 6m 40s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/0dce25c7b20149589e5025d65c351dae : FAILURE in 11m 39s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0acb055fad7c4d1a9fc1fae8e9aef265 : FAILURE in 8m 32s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/08b0f4efc93e4c9cb5bd21e09cf32d6d : FAILURE in 8m 52s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/d5f508db807341e9bde843c89fdb26b0 : FAILURE in 9m 36s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/e1737bc337f546d1b5614a9795ab1beb : FAILURE in 8m 29s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f4ea95fea62e462ab1c044907ee8e59a : FAILURE in 8m 45s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/db5bdbd8bbf942848d6192704ef5a74b : FAILURE in 9m 04s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/0626195bfd674c5588df63dde347170b : FAILURE in 1h 00m 47s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/91e88982999645c58b90569d4dcd531b : FAILURE in 1h 09m 55s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/cb36413413a64c0aa1c3c6733f420961 : FAILURE in 7m 30s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/ac283f4e08334755827740e5136a1da1 : FAILURE in 7m 29s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0e6cecb1f95b4314a03226b9be7ebceb : FAILURE in 8m 20s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/745f7ebb033740c7a866798d39928a3b : FAILURE in 8m 33s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/f38485e024b443c79ad55877e5a3cfd9 : FAILURE in 34m 24s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4e35a10b40914e6c8cc197ae28ac4456 : FAILURE in 30m 02s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/80368e8cc7b345a6bc420a218197b5c3 : FAILURE in 7m 36s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7e3f4353e12e4ba6a91379f1b238e13f : FAILURE in 8m 31s","accounts_in_message":[],"_revision_number":8},{"id":"23e4826c0932cf31d9381c77a9588f4b9a570403","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2026-08-26 11:52:47.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":9},{"id":"122a8e23702f4022f7f113f7a459ba7258121a4e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 14:06:45.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2bb74f1207bf4e368fb376fe85c74ae9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/733ebfba3c794dbda4ae466bcc6b3564 : SUCCESS in 5m 16s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/4ff64acbb51e4b1890ac73d818058606 : SUCCESS in 5m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f80c57b581f946f8b32e1592792cd848 : SUCCESS in 4m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/11de30358e3047d2b439a94e46f48ac6 : SUCCESS in 2m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d252c7fe6a654bafa10143dfcaf9084d : SUCCESS in 2m 58s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/66a89b51019546deab5f5fcbb8913e80 : SUCCESS in 7m 20s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a63f96f6cf4a471894ef3ad13e43948d : SUCCESS in 6m 26s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/cc709bde8f9f4ba4be3ef9901ccb72ba : SUCCESS in 5m 41s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0b7a31cc52a94f8c944f4165e271a525 : SUCCESS in 13m 01s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/fb13338b08dc45ea9f221a7feef3fca6 : SUCCESS in 12m 46s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/028bf3ba180e415a897576a7114d9871 : SUCCESS in 10m 59s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/34452ead636e41199d430d55018bbe27 : SUCCESS in 24m 51s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ec3261925940415783b9083b345bdddc : SUCCESS in 21m 12s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/b0a6c0bc09794004baee4f0df0ea7a76 : SUCCESS in 20m 00s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/8abb8c5a06324ed39ed8a4406dfde40a : SUCCESS in 28m 50s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/03fbd72531564894a6e10e7a1a45357f : SUCCESS in 22m 28s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/803593ab35db4ee995897e17df4dfc73 : SUCCESS in 16m 51s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0b93a3de7b27462db29919dfb67cf59a : SUCCESS in 20m 36s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/481349ad250243879b5d94fec4db769d : SUCCESS in 1h 12m 26s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/787c722ef803418385e11d6e458d2ab0 : SUCCESS in 1h 15m 35s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/6edca4d5df5f4d67a8b6b820173a8f0b : SUCCESS in 1h 05m 46s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/af76eeeffd2e4ab1a99e27aec7bd8c95 : SUCCESS in 1h 22m 43s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/bc8e85d33e314601bfd52311ad8d3b45 : SUCCESS in 1h 21m 31s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/f1216db79af4460b975bf798ba48a944 : SUCCESS in 1h 59m 53s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/456c8cde46284fcdb6160ceed1474cb3 : SUCCESS in 2h 11m 03s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/fc3e80e41d0b45f88b8de69dcc401ea4 : SUCCESS in 15m 05s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/0b66128f1a594cbcbd49746cc6017d59 : SUCCESS in 19m 47s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/50509efaf3194025b072132a6b1b2e68 : SUCCESS in 25m 26s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/e0d27462c4374b1285361f4b63a9e14e : SUCCESS in 24m 10s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/986e1f46652647b88d7dd4741e71f33b : SUCCESS in 21m 37s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7588e87e32a4452faa9211d5d7d3e48c : SUCCESS in 24m 32s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/c73301a51e414753a5b03a24b8059d2f : SUCCESS in 24m 41s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/19dfd664a81b446986860c3f5559431a : SUCCESS in 19m 05s","accounts_in_message":[],"_revision_number":9},{"id":"0e5881ca72e6feabcdcd3e3499e8cb60ec811230","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-09-01 09:48:04.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":9},{"id":"1ce3aa636aca5b1504d3300a5dfb06135fa268dc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 09:48:30.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"5f964f83f6508f2c112b5c175d5bed1df99b72e3","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-09-01 11:08:01.000000000","message":"Patch Set 9: Code-Review+1","accounts_in_message":[],"_revision_number":9},{"id":"e61e2bfbcc9c640e90a88b20ae290908d9965842","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 12:52:39.000000000","message":"Patch Set 9: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e6fe60b435fb44738864a749b5cd680d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09a372752bed4fbda27e6ad5e9b2ca51 : SUCCESS in 10m 55s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/cac0f29f53e44e0ea0ac12d19de6bbd7 : SUCCESS in 7m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e671d48b8f44255a67ea8e61d703fbc : SUCCESS in 3m 21s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6547f19150e14676ad085d06a6e72438 : SUCCESS in 2m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/11371c9bb13f4eca978651d2d42d7bef : SUCCESS in 6m 17s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7a7d5bb428c34959b5472813d681e596 : SUCCESS in 7m 06s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/84ccea7da44545988709e9a50f2544f1 : SUCCESS in 16m 40s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/9e7fa0e416f54a1d825e739e996a8e59 : SUCCESS in 11m 04s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/228f89b31caf489dabce3dfafeac7cd1 : SUCCESS in 11m 34s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/339925035d6c4225b58e75c951407c80 : SUCCESS in 12m 56s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/eb3cc7f572ad413787c8a3d6da975f95 : SUCCESS in 20m 16s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/82e7141535a64cd988566325f2399807 : SUCCESS in 14m 04s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/481a4084830e4f4082ab4ad2750b5ad0 : SUCCESS in 11m 56s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/02031cf436a84272b85389de75261a33 : SUCCESS in 23m 33s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/38c7da9b03d145b3aeb2090604d337e8 : SUCCESS in 1h 07m 24s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/bf7884ad874e4cb690c87ae06b308832 : SUCCESS in 1h 06m 24s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/0a9690faee38402488e78f81688595d0 : SUCCESS in 1h 05m 19s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e48b12bcf8094ac28046770fc0c67b93 : SUCCESS in 1h 52m 14s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/17288d5cd8a64c3bb64623a1119a4629 : SUCCESS in 1h 25m 44s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/c1347c653ddf46e3bc98b59bb9d4f302 : SUCCESS in 1h 48m 59s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3921c0dc74b843a1a5fa8eb6e07ddd91 : TIMED_OUT in 3h 03m 30s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/b64fc1c727444c07b3b0bbc411e40e38 : SUCCESS in 25m 48s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/2ea9cf999c284faa9fab782a186bbe67 : SUCCESS in 19m 58s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2fdff64562bd48bfa6a6680cb5017555 : SUCCESS in 17m 05s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/0bef5402895d4bdd8db3d2630fb92f4d : SUCCESS in 16m 40s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/5182e5a8abe8484f93e8b262bb0740c5 : SUCCESS in 38m 51s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2114ddd019e14b03ad967788cb8b99d9 : SUCCESS in 36m 12s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/06d67b67ba5d4e2491503ca9c6b535b0 : SUCCESS in 16m 02s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/34bdbcb4b6f64b1fbe3e38769ae6c2f6 : SUCCESS in 18m 29s","accounts_in_message":[],"_revision_number":9},{"id":"0ca8f260d47995fc97cf32ebed8b540f228624ce","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-09-10 11:18:45.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"d0628f407291a43565b94999eaf451453ab3eea2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 13:18:39.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/64d0718592f243758db0c963e9e82e89\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2239f9afd6994e76976040545ebed4c5 : SUCCESS in 9m 20s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/3d50456e3eec468d9f9df748d6cc6a3b : SUCCESS in 10m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bd73f2068cfa4236abe1cbb530f84483 : SUCCESS in 4m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b585305af7848b1968d57058462ca09 : SUCCESS in 4m 27s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7b9624d887514413aebca38d94248a79 : SUCCESS in 2m 25s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/5fa77b08c8ed426082aa0a2485472eb3 : SUCCESS in 7m 16s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5811c2f51cd8478b8bf465701fb72d34 : SUCCESS in 7m 19s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d350ee36614e4364bf0341604d15980e : SUCCESS in 9m 19s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ac91fcaeb25e48418f38757f145f9446 : SUCCESS in 14m 43s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6fb20daa2612415f925d35002ceb3d4a : SUCCESS in 13m 49s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/a8cddf6991b64c7e9ff3eebaacce3a6e : SUCCESS in 27m 21s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/89b8bbb2bd12454bbea3f09dd70343cc : SUCCESS in 22m 47s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b138c572f53546a1b91b84fa1d97d347 : SUCCESS in 23m 04s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/8c745e45258f4f6cab207854905eab5f : SUCCESS in 24m 17s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/5c3aaf7c85264c488ac55140839ce4fa : SUCCESS in 34m 47s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dcfa55aea3a041f4b8b1e73db72574c0 : SUCCESS in 29m 17s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/a4d7b016e25840b89e0b8cb00a8b8c5b : SUCCESS in 28m 10s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c931bda5696b4f328b3a4feb74ae7552 : SUCCESS in 27m 02s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/13c462699e1f403aa643253621b8e29b : SUCCESS in 1h 02m 08s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/63c03efaf8a44e929f3cf9223fb1f28c : SUCCESS in 1h 05m 54s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/8b345c8bd6124ff89a53aeee4003b601 : SUCCESS in 1h 19m 30s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/54aa12ac562a4b88a87f617b110de6b5 : SUCCESS in 1h 49m 35s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/a7e7ec69b7274912b88512cdc0997da0 : SUCCESS in 1h 53m 27s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/a3acc1f93abc40f1a254d915a873eeec : SUCCESS in 1h 57m 47s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/335c39c5f1ac49eb8610500803ebbfd2 : SUCCESS in 32m 12s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/9eb062dc585f43c4bb135dc70ca3aa00 : SUCCESS in 31m 44s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3d47b2d4bc9048a3846e21c8ba1a607f : SUCCESS in 23m 23s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/ebae7a71969147b4ae0de7a062db8c4f : SUCCESS in 29m 39s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/115309d5f8b54efb8daa2b2b91e1e39d : SUCCESS in 45m 09s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/de7cd99628b14ce398a364ec9e89cb28 : SUCCESS in 44m 14s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/e99a667ed1de4b679184b9307a20fadd : SUCCESS in 22m 06s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/67d5c733683c480987904aff26122b3f : SUCCESS in 28m 36s","accounts_in_message":[],"_revision_number":9},{"id":"e5351e8aafdad43fcbf07dcb3cf4b0b5856af6d4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 15:21:04.000000000","message":"Patch Set 9: Verified-2\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/kayobe for 997692,9","accounts_in_message":[],"_revision_number":9},{"id":"f7ca9511228a35736256c980692597344aa4f0f2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2026-09-11 08:07:49.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-2 (copy condition: \"NEVER\")\n* Workflow+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":10},{"id":"5f3783b40294f4f72d879a6d2e3148f0341355a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-11 10:13:18.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e783bf50407f4922b57499db99abcc0a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bef66de97fb94941bb2fb12196b90b8d : SUCCESS in 9m 10s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/45c43a7e93b7461882014a8aeb14a7a7 : SUCCESS in 11m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/765e4c3cd90741369fb94a99bd26fde7 : SUCCESS in 4m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2458e4622b0d4272873e21eaf0474e33 : SUCCESS in 4m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7822a93c981141f296c1cbb18adf71f6 : SUCCESS in 3m 54s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e61f31d87a244dacab255fda4acc0961 : SUCCESS in 6m 46s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a6f27ecb5e2d4984b10cf1604476b288 : SUCCESS in 6m 27s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/84da6a186b8045b8bbd4e06c8bbddd9d : SUCCESS in 9m 11s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c4114de2d69145d48268ff056ea0a79d : SUCCESS in 19m 14s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/41847734748a4a7aaa44826cc74e0cb3 : SUCCESS in 13m 03s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/5eca37470d5e48af865685456cb6d824 : SUCCESS in 29m 16s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/26a674e9d54747debd77ef2d860af3ce : SUCCESS in 30m 09s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bd13c4876d474e22b2812cc20e29051a : SUCCESS in 22m 15s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/c8fd3490945d4d81a3a22fbb312987c4 : SUCCESS in 24m 38s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/343e0833472541b386904c5ff490bae3 : SUCCESS in 29m 48s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c690a2345b1c4102aff2ed6a75cdca62 : SUCCESS in 26m 12s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/efad921a3f5942f680c08bd6d2c816b7 : SUCCESS in 27m 10s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/43f2d0858b6c4e9aa676879d572aa893 : SUCCESS in 24m 27s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/0a4c32e964734b1e812d9be723b1e4fe : SUCCESS in 1h 07m 17s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/11fa886bde474b6bbfbfd650996b02da : SUCCESS in 1h 08m 24s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/4d1de3ff654a43fcb085384acc2c5a24 : SUCCESS in 1h 18m 48s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f28b120ac6634db680b6f78ad611987f : SUCCESS in 1h 45m 46s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/42794a25c3d74ddd8017358e6a8dbbaa : SUCCESS in 1h 43m 42s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/91dd2c4b3e614b03a4e4e451756ba170 : SUCCESS in 2h 04m 12s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/f7e47012a7794d80897fc091ee4c666d : SUCCESS in 36m 06s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/bef7a216b1c54cf2998d8eb42cf2e8c4 : SUCCESS in 28m 33s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/41ecb73ff61145fda2abf11f2a78aaaf : SUCCESS in 32m 02s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/9f767c0d70fa4dc7924d69016fd88ce3 : SUCCESS in 29m 09s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/dc288fd3f6234b92aeffd0cbdce20904 : SUCCESS in 45m 53s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ff470a7e4ab6483585fe63d070d534ab : SUCCESS in 42m 50s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/153ad2f7c5234fa8b1553e5e1a0f5011 : SUCCESS in 25m 05s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bcf7c1f49a424d1cade466dfb6bbce8c : SUCCESS in 25m 27s","accounts_in_message":[],"_revision_number":10},{"id":"659ecdd41fd15b6983c53f89d9a2964bb944114e","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2026-09-11 10:21:29.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"a17cde85f26f2a3f4060bccc1a267866155c68c0","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-09-11 10:54:35.000000000","message":"Patch Set 10: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":10},{"id":"eb5230d28a782500cb889240d5baffa04ea072b4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-11 10:55:28.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"7a68467fbb0573b99f83ef84db7e2ad4ad0c5971","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-09-11 11:46:46.000000000","message":"Patch Set 10: Code-Review-1 Workflow-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":10},{"id":"f5e32c504989a375bd43467a8d1811485b3f035f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-11 11:46:54.000000000","message":"Patch Set 10: Verified-2\n\nThis change is unable to merge due to a missing merge requirement.","accounts_in_message":[],"_revision_number":10},{"id":"e9081781df82348113b11c70d066daba91972f4b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-11 12:43:30.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/93b3be19585d47dc8e64e01857d2d109\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/845cb4ffd40a42cd9a419eeab83261f7 : SUCCESS in 7m 45s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/b8b67579eb954236a06736d44841276d : SUCCESS in 9m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a88f2507f7e94a17a32c281c661825ef : SUCCESS in 4m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/634cd39dde8249af9a9be73085d4b060 : SUCCESS in 3m 59s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/947061381fac44cfb1204bcfaaeeb8a3 : SUCCESS in 3m 55s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/2f6778d3cc8747cc8bb08619c42cf47f : SUCCESS in 7m 16s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/deacfa58f0974d57902c0edab7a98ee4 : SUCCESS in 6m 13s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/dc20d1d5bee94617b0b1e6e2b7fef863 : SUCCESS in 10m 57s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ed55ca628d534d518db03635aa917456 : SUCCESS in 16m 54s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/85a2953e38e641a2886852f4e688afa8 : SUCCESS in 12m 12s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/6b3ae5c683f84210a36443b947fe8ef1 : SUCCESS in 24m 22s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/73a3272e856148fa881e964780a13ba3 : SUCCESS in 21m 22s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f4bb56a3c6484b63bec8a9c0e84c7533 : SUCCESS in 25m 28s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/aad9c3d995344f7b820de10a00891d30 : SUCCESS in 23m 52s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/e111dc7efade4c489d050b6a736bbba0 : SUCCESS in 32m 34s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0aa5e0f8850f4158853e0a47f1f3b3f3 : SUCCESS in 26m 14s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/7976c0cdb3c946d58ea3b8641e7f8adb : SUCCESS in 28m 13s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e0b033bdcae946418dec0d24cdbd5a56 : SUCCESS in 19m 57s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/8d894cb728f34bff86dc56accfb5b157 : SUCCESS in 1h 08m 04s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/04eb496f31ec4d38a56c0383770d7305 : SUCCESS in 1h 03m 13s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/61d008526679401c9a403e9dcbe71a01 : SUCCESS in 1h 10m 17s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f59835dd236d42b7a84a0373bdd854a7 : SUCCESS in 1h 35m 12s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/8e32b920168043f38d827aa798ffa473 : TIMED_OUT in 2h 21m 17s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/d45b0db93fce4cefbd72b569f76eaba2 : SUCCESS in 2h 02m 58s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/ef24cc70382a4dc696b1525767f1f2c3 : SUCCESS in 31m 15s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/9dab48316d2843fc8e5e977e3adcee2e : FAILURE in 31m 18s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3e7fc6e325b64472bdc835962abd1120 : SUCCESS in 25m 53s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/7479b472266f4d0bab0cfbee37e01758 : SUCCESS in 34m 26s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/9d88472df08849c7b6e5880b2a6c5245 : SUCCESS in 32m 52s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ac2a6ab7f6894937ad16c22c045931ec : SUCCESS in 39m 15s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/eff14fa1d9064308a7e080daec080589 : SUCCESS in 26m 12s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/614ec2b1e17d49b8bf05aacaa713d90b : SUCCESS in 26m 43s","accounts_in_message":[],"_revision_number":10},{"id":"6b55a96a592f47020e064244913ca4993ce5485f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2026-09-11 13:00:43.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n* Workflow+1, Workflow-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":11},{"id":"20810448553daba1c0aad01e91479e70f03fd18a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-11 15:06:14.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/70afd272e2214cd0ab4933db6939e33b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/874fae5356124683adb3a42d2ae3d3c5 : SUCCESS in 8m 52s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/54b7530975c042a6a2f80f6a5e24e295 : SUCCESS in 13m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/196297e2368b4af5b0ed410119e8a2f6 : SUCCESS in 2m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb4a5a0e6d824a228e8eedfae69c91da : SUCCESS in 5m 29s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8e6ea30f5c73478c8f4e17de85c05c68 : SUCCESS in 3m 12s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/d3f4a7fb8b8b42abb9af8547f922522e : SUCCESS in 4m 35s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/14c1c7b511914e918c6301c5fa248a2b : SUCCESS in 3m 02s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/fcca9be20be84951be425dace6e7aeda : SUCCESS in 12m 33s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/2f3737f6963a47838483ed4258e02013 : SUCCESS in 16m 47s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a815e31498644d0d9926637231539a4e : SUCCESS in 14m 22s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/5ba63d6eb15c457e97480c068e817cd9 : SUCCESS in 21m 30s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/a8c456e54b1642d3b2e04491c760da1b : SUCCESS in 34m 33s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b8f05c39276b4cc983ae2689ddec7b75 : SUCCESS in 28m 49s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/79eba1c514864bd6be26d6c5834635ba : SUCCESS in 32m 38s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/0294941adbf34c3dbae8ed05d59366ab : SUCCESS in 29m 00s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1e6f79ebe45c4d29a8b205a62c819f10 : SUCCESS in 29m 37s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/df08220a9f1a4932b1ff0c6e21a401b1 : SUCCESS in 28m 01s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c1932d4cd21b4c40828907e3394634a3 : SUCCESS in 31m 14s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/369c0ad70ce94e6f9cee435f157005c3 : SUCCESS in 1h 04m 25s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/bb4d1bb254d64ffcbf7380bbdd797767 : SUCCESS in 1h 12m 54s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/47c1e25262e248bf977d4c482aa34932 : SUCCESS in 1h 16m 37s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/94796a4585a4414c93d5e5c13ca42221 : SUCCESS in 1h 49m 59s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/481f1f58e3b14be5a9c9d4badf1ddd98 : SUCCESS in 1h 43m 03s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/de70ba7e025d406e80e1f394f55a887b : SUCCESS in 2h 04m 26s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/c833f1a9be7b4e1d979cfc985366e24e : SUCCESS in 26m 20s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/99901365a14942d1af53e63246b301a6 : SUCCESS in 24m 03s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/564dae7719724d6d8b9c470d4de08874 : SUCCESS in 23m 00s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/2538e8c9e99844e8a1c90484b09d25b1 : SUCCESS in 30m 35s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/d32cfac42d87494d945542240158062a : SUCCESS in 38m 00s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c7b611b7301647ff9bf7150f7e1000db : SUCCESS in 35m 57s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/2e41069719ca471180469c6956a6cd25 : SUCCESS in 26m 29s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a3b83c157222452c8a9f51d5ec25f213 : SUCCESS in 26m 00s","accounts_in_message":[],"_revision_number":11},{"id":"474d0bc8bfeb3ba9812c11228b78107675207152","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-09-11 21:11:57.000000000","message":"Patch Set 11: Code-Review+1","accounts_in_message":[],"_revision_number":11},{"id":"7e5a0f212bdfc661d10c4f0e7b02fc14a6045099","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2026-09-14 07:50:26.000000000","message":"Patch Set 11:\n\n(2 comments)","accounts_in_message":[],"_revision_number":11},{"id":"858d31da284fa39bb58632b4ff1e36a1d9d06980","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-09-14 08:16:59.000000000","message":"Patch Set 11: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":11},{"id":"e475ded40c22c89d7ae81bdb0f3b2195f90c6bf4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-14 08:17:27.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"b478574fce44fcedd7f3fd8831f35c5ca5967f2e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-14 11:00:28.000000000","message":"Patch Set 11: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a0a5d3fec36a4f85b0c18cd3e89a41aa\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/faf88cc5ca744bfa885ce2a70fb2afc6 : SUCCESS in 9m 05s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/7404a2837a4e4187aa735470b23f8b8c : SUCCESS in 16m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bce29c8d078048d194e8275b24a950bd : SUCCESS in 4m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fdbcbce2b56146e395345d90df99f32b : SUCCESS in 5m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2190c1a5c5c6463fb909365d1ea3c4aa : SUCCESS in 7m 01s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/054d7bbad5924b2292ed857cefb7aee4 : SUCCESS in 12m 28s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d7d682df5ae747bd8fa898ddeb1ec49c : SUCCESS in 23m 20s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/630adfa8bf554661b92252bb8382c0fb : SUCCESS in 15m 48s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/7549c9aec98e48259d960f481e3acadf : SUCCESS in 28m 57s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/669be78066b24bd78ca65fa85f24be4d : SUCCESS in 28m 03s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/e74d555d493e40619a915cb00290f455 : SUCCESS in 30m 05s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a032592dee084e20b5324ca7caf53b87 : SUCCESS in 33m 39s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/807ae1047f2748a8b964aea003adafee : SUCCESS in 32m 55s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/66d3ce8ba18347809b2b04018603b36d : SUCCESS in 34m 18s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/135c349c690540349aadb24f9a18ca74 : SUCCESS in 1h 18m 29s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/8fca5088cb5445078082865fd69ab091 : SUCCESS in 1h 18m 43s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/06e554e459104242a30525f8177b7282 : SUCCESS in 1h 20m 26s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/49acaa737c2a494cae56052776a9fd99 : SUCCESS in 2h 41m 29s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/fb76b10f8a77410d8b1175b1fe6f49d3 : SUCCESS in 1h 39m 26s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/bad767ed54b04e8ab56b6e88bcac3d8a : SUCCESS in 2h 18m 51s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/e38fd98a1b85448797f9d72db2b2956a : SUCCESS in 39m 48s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/310cf4baee2a4e03ba754e79bb8a11a6 : SUCCESS in 34m 31s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/028b8934c1af4f929a6a495c1fe93de1 : SUCCESS in 36m 39s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/5eae10b433464f11a76c62da56cfb426 : SUCCESS in 33m 55s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/efb621418b1041469d9b74e17fe989d9 : SUCCESS in 39m 36s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/796feb8e3e164bec99da127cb3e35392 : SUCCESS in 52m 48s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/71bf249a3af74b00a86dbf54b575cb74 : SUCCESS in 32m 47s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a3d11a64cc784f2fb3f7e2711e054fe4 : SUCCESS in 36m 16s","accounts_in_message":[],"_revision_number":11},{"id":"8714089e843e303da7f3c5078ff2e1c0ca553299","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-14 11:00:28.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":11},{"id":"7668eed94b0b72609dfe8760d52d88ca0810fc05","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-14 11:01:34.000000000","message":"Patch Set 11:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7e7230eecd4549e59af5794e489a801c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47aa1b6bc0ae4c609f0a8e985721331d : SUCCESS in 52s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c64a7c46d93a4246b603b47f92b85cf1 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"0f80cd9e5b395b0211c06d7de48dfb61350ea82f","revisions":{"7530db4340014c1b06d4952fd6d1de3eb4628782":{"kind":"REWORK","_number":1,"created":"2026-07-17 08:45:05.000000000","uploader":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"ref":"refs/changes/92/997692/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/92/997692/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/92/997692/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/92/997692/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/92/997692/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/92/997692/1"}}},"commit":{"parents":[{"commit":"e7dc5a78e5f00420fcad9e99a47a3f67b4669ebd","subject":"Update architecture page to reflect reality","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e7dc5a78e5f00420fcad9e99a47a3f67b4669ebd"}]}],"author":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2026-07-17 08:42:21.000000000","tz":120},"committer":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2026-07-17 08:45:04.000000000","tz":120},"subject":"Enable systemd-resolved configuration for Debian based distros","message":"Enable systemd-resolved configuration for Debian based distros\n\nDepends-On: https://github.com/stackhpc/ansible-collection-linux/pull/51\nChange-Id: Ie3c0d8f1b326705b84ac49f8bc94da8f07827f8c\nSigned-off-by: Rafal Lewandowski \u003crafal@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7530db4340014c1b06d4952fd6d1de3eb4628782"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7530db4340014c1b06d4952fd6d1de3eb4628782"}]},"branch":"refs/heads/master"},"b54efeba7874c832a5904e1045a4a2a01755972e":{"kind":"REWORK","_number":2,"created":"2026-07-17 08:46:31.000000000","uploader":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"ref":"refs/changes/92/997692/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/92/997692/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/92/997692/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/92/997692/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/92/997692/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/92/997692/2"}}},"commit":{"parents":[{"commit":"e7dc5a78e5f00420fcad9e99a47a3f67b4669ebd","subject":"Update architecture page to reflect reality","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e7dc5a78e5f00420fcad9e99a47a3f67b4669ebd"}]}],"author":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2026-07-17 08:42:21.000000000","tz":120},"committer":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2026-07-17 08:45:30.000000000","tz":120},"subject":"Enable systemd-resolved configuration for Debian based distros","message":"Enable systemd-resolved configuration for Debian based distros\n\nDepends-On: https://github.com/stackhpc/ansible-collection-linux/pull/51\nChange-Id: Ie3c0d8f1b326705b84ac49f8bc94da8f07827f8c\nSigned-off-by: Rafal Lewandowski \u003crafal@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b54efeba7874c832a5904e1045a4a2a01755972e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b54efeba7874c832a5904e1045a4a2a01755972e"}]},"branch":"refs/heads/master"},"d754d5ca0ebd67f0cb7490abfc65b8fdda62ed02":{"kind":"REWORK","_number":3,"created":"2026-07-17 12:30:05.000000000","uploader":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"ref":"refs/changes/92/997692/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/92/997692/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/92/997692/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/92/997692/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/92/997692/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/92/997692/3"}}},"commit":{"parents":[{"commit":"e7dc5a78e5f00420fcad9e99a47a3f67b4669ebd","subject":"Update architecture page to reflect reality","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e7dc5a78e5f00420fcad9e99a47a3f67b4669ebd"}]}],"author":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2026-07-17 08:42:21.000000000","tz":120},"committer":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2026-07-17 12:29:56.000000000","tz":120},"subject":"Enable systemd-resolved configuration for Debian based distros","message":"Enable systemd-resolved configuration for Debian based distros\n\nDepends-On: https://github.com/stackhpc/ansible-collection-linux/pull/51\nChange-Id: Ie3c0d8f1b326705b84ac49f8bc94da8f07827f8c\nSigned-off-by: Rafal Lewandowski \u003crafal@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d754d5ca0ebd67f0cb7490abfc65b8fdda62ed02"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d754d5ca0ebd67f0cb7490abfc65b8fdda62ed02"}]},"branch":"refs/heads/master"},"64a177f1d3055e8a64d8b99d982064816ca3c40b":{"kind":"REWORK","_number":4,"created":"2026-07-17 15:11:40.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/92/997692/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/92/997692/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/92/997692/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/92/997692/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/92/997692/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/92/997692/4"}}},"commit":{"parents":[{"commit":"e7dc5a78e5f00420fcad9e99a47a3f67b4669ebd","subject":"Update architecture page to reflect reality","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e7dc5a78e5f00420fcad9e99a47a3f67b4669ebd"}]}],"author":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2026-07-17 08:42:21.000000000","tz":120},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-07-17 15:11:40.000000000","tz":0},"subject":"Enable systemd-resolved configuration for Debian based distros","message":"Enable systemd-resolved configuration for Debian based distros\n\nDepends-On: https://github.com/stackhpc/ansible-collection-linux/pull/51\nChange-Id: Ie3c0d8f1b326705b84ac49f8bc94da8f07827f8c\nSigned-off-by: Rafal Lewandowski \u003crafal@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/64a177f1d3055e8a64d8b99d982064816ca3c40b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/64a177f1d3055e8a64d8b99d982064816ca3c40b"}]},"branch":"refs/heads/master"},"64c83ab5772fdc8c8d87bcc9b5a736d8ee4b0542":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2026-07-21 19:02:15.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"real_uploader":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"ref":"refs/changes/92/997692/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/92/997692/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/92/997692/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/92/997692/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/92/997692/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/92/997692/5"}}},"commit":{"parents":[{"commit":"7688d5c4b99ebf6979236b53e52964afd680915f","subject":"Merge \"Add missing reconfigure.yml for manage-containers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7688d5c4b99ebf6979236b53e52964afd680915f"}]}],"author":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2026-07-17 08:42:21.000000000","tz":120},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-07-21 19:02:15.000000000","tz":0},"subject":"Enable systemd-resolved configuration for Debian based distros","message":"Enable systemd-resolved configuration for Debian based distros\n\nDepends-On: https://github.com/stackhpc/ansible-collection-linux/pull/51\nChange-Id: Ie3c0d8f1b326705b84ac49f8bc94da8f07827f8c\nSigned-off-by: Rafal Lewandowski \u003crafal@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/64c83ab5772fdc8c8d87bcc9b5a736d8ee4b0542"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/64c83ab5772fdc8c8d87bcc9b5a736d8ee4b0542"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"e7dc5a78e5f00420fcad9e99a47a3f67b4669ebd","ours":"64a177f1d3055e8a64d8b99d982064816ca3c40b","theirs":"7688d5c4b99ebf6979236b53e52964afd680915f","merge_strategy":"recursive","contains_conflicts":false}},"b79c829c4f24e399dd697af38a77bc2a7fb0fabf":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2026-07-27 11:10:58.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"real_uploader":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"ref":"refs/changes/92/997692/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/92/997692/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/92/997692/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/92/997692/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/92/997692/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/92/997692/6"}}},"commit":{"parents":[{"commit":"eef7ec67df076d7da58f43937a088b0fb1b5d8fb","subject":"Make httpd image buildable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/eef7ec67df076d7da58f43937a088b0fb1b5d8fb"}]}],"author":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2026-07-17 08:42:21.000000000","tz":120},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-07-27 11:10:58.000000000","tz":0},"subject":"Enable systemd-resolved configuration for Debian based distros","message":"Enable systemd-resolved configuration for Debian based distros\n\nDepends-On: https://github.com/stackhpc/ansible-collection-linux/pull/51\nChange-Id: Ie3c0d8f1b326705b84ac49f8bc94da8f07827f8c\nSigned-off-by: Rafal Lewandowski \u003crafal@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b79c829c4f24e399dd697af38a77bc2a7fb0fabf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b79c829c4f24e399dd697af38a77bc2a7fb0fabf"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"7688d5c4b99ebf6979236b53e52964afd680915f","ours":"64c83ab5772fdc8c8d87bcc9b5a736d8ee4b0542","theirs":"eef7ec67df076d7da58f43937a088b0fb1b5d8fb","merge_strategy":"recursive","contains_conflicts":false}},"e8b39d5794576d39918e4789f322349ad0b046ea":{"kind":"REWORK","_number":7,"created":"2026-08-25 10:49:05.000000000","uploader":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"ref":"refs/changes/92/997692/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/92/997692/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/92/997692/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/92/997692/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/92/997692/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/92/997692/7"}}},"commit":{"parents":[{"commit":"e7dc5a78e5f00420fcad9e99a47a3f67b4669ebd","subject":"Update architecture page to reflect reality","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e7dc5a78e5f00420fcad9e99a47a3f67b4669ebd"}]}],"author":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2026-07-17 08:42:21.000000000","tz":120},"committer":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2026-08-25 10:48:49.000000000","tz":120},"subject":"Enable systemd-resolved configuration for Debian based distros","message":"Enable systemd-resolved configuration for Debian based distros\n\nDepends-On: https://github.com/stackhpc/ansible-collection-linux/pull/51\nChange-Id: Ie3c0d8f1b326705b84ac49f8bc94da8f07827f8c\nSigned-off-by: Rafal Lewandowski \u003crafal@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e8b39d5794576d39918e4789f322349ad0b046ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e8b39d5794576d39918e4789f322349ad0b046ea"}]},"branch":"refs/heads/master"},"ed5b49c438c4f0e64a97954d1f6dc58910c476ff":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2026-08-25 10:49:18.000000000","uploader":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"ref":"refs/changes/92/997692/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/92/997692/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/92/997692/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/92/997692/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/92/997692/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/92/997692/8"}}},"commit":{"parents":[{"commit":"720683bc01f182056e8c94ee90a7aefa49daf117","subject":"Merge \"dell-switch: Fix invocation of dellemc.os* modules\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/720683bc01f182056e8c94ee90a7aefa49daf117"}]}],"author":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2026-07-17 08:42:21.000000000","tz":120},"committer":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2026-08-25 10:49:18.000000000","tz":0},"subject":"Enable systemd-resolved configuration for Debian based distros","message":"Enable systemd-resolved configuration for Debian based distros\n\nDepends-On: https://github.com/stackhpc/ansible-collection-linux/pull/51\nChange-Id: Ie3c0d8f1b326705b84ac49f8bc94da8f07827f8c\nSigned-off-by: Rafal Lewandowski \u003crafal@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ed5b49c438c4f0e64a97954d1f6dc58910c476ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ed5b49c438c4f0e64a97954d1f6dc58910c476ff"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"e7dc5a78e5f00420fcad9e99a47a3f67b4669ebd","ours":"e8b39d5794576d39918e4789f322349ad0b046ea","theirs":"720683bc01f182056e8c94ee90a7aefa49daf117","merge_strategy":"recursive","contains_conflicts":false}},"e29f39efa8f82090d59cbdac6ad3770a91d4cd4c":{"kind":"REWORK","_number":9,"created":"2026-08-26 11:52:47.000000000","uploader":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"ref":"refs/changes/92/997692/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/92/997692/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/92/997692/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/92/997692/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/92/997692/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/92/997692/9"}}},"commit":{"parents":[{"commit":"e7dc5a78e5f00420fcad9e99a47a3f67b4669ebd","subject":"Update architecture page to reflect reality","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e7dc5a78e5f00420fcad9e99a47a3f67b4669ebd"}]}],"author":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2026-07-17 08:42:21.000000000","tz":120},"committer":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2026-08-26 11:52:39.000000000","tz":120},"subject":"Enable systemd-resolved configuration for Debian based distros","message":"Enable systemd-resolved configuration for Debian based distros\n\nDepends-On: https://github.com/stackhpc/ansible-collection-linux/pull/51\nChange-Id: Ie3c0d8f1b326705b84ac49f8bc94da8f07827f8c\nSigned-off-by: Rafal Lewandowski \u003crafal@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e29f39efa8f82090d59cbdac6ad3770a91d4cd4c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e29f39efa8f82090d59cbdac6ad3770a91d4cd4c"}]},"branch":"refs/heads/master"},"e1433293809ddea6137a886e5d52b2a00a68da14":{"kind":"REWORK","_number":10,"created":"2026-09-11 08:07:49.000000000","uploader":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"ref":"refs/changes/92/997692/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/92/997692/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/92/997692/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/92/997692/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/92/997692/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/92/997692/10"}}},"commit":{"parents":[{"commit":"3ec0c1258b0be34131dd12aa08a31695b855a815","subject":"Merge \"Never reboot without explicit permission\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3ec0c1258b0be34131dd12aa08a31695b855a815"}]}],"author":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2026-07-17 08:42:21.000000000","tz":120},"committer":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2026-09-11 08:07:14.000000000","tz":120},"subject":"Enable systemd-resolved configuration for Debian based distros","message":"Enable systemd-resolved configuration for Debian based distros\n\nDepends-On: https://github.com/stackhpc/ansible-collection-linux/pull/51\nChange-Id: Ie3c0d8f1b326705b84ac49f8bc94da8f07827f8c\nSigned-off-by: Rafal Lewandowski \u003crafal@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e1433293809ddea6137a886e5d52b2a00a68da14"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e1433293809ddea6137a886e5d52b2a00a68da14"}]},"branch":"refs/heads/master"},"0f80cd9e5b395b0211c06d7de48dfb61350ea82f":{"kind":"REWORK","_number":11,"created":"2026-09-11 13:00:43.000000000","uploader":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"ref":"refs/changes/92/997692/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/92/997692/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/92/997692/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/92/997692/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/92/997692/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/92/997692/11"}}},"commit":{"parents":[{"commit":"3ec0c1258b0be34131dd12aa08a31695b855a815","subject":"Merge \"Never reboot without explicit permission\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3ec0c1258b0be34131dd12aa08a31695b855a815"}]}],"author":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2026-07-17 08:42:21.000000000","tz":120},"committer":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2026-09-11 13:00:29.000000000","tz":120},"subject":"Enable systemd-resolved configuration for Debian based distros","message":"Enable systemd-resolved configuration for Debian based distros\n\nDepends-On: https://github.com/stackhpc/ansible-collection-linux/pull/51\nChange-Id: Ie3c0d8f1b326705b84ac49f8bc94da8f07827f8c\nSigned-off-by: Rafal Lewandowski \u003crafal@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0f80cd9e5b395b0211c06d7de48dfb61350ea82f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0f80cd9e5b395b0211c06d7de48dfb61350ea82f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
