)]}'
{"id":"openstack%2Fkayobe~997984","triplet_id":"openstack%2Fkayobe~master~I94d3a989932f5b70b05d512df1a4d319b01b3bd9","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2026-07-21 14:20:33.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I94d3a989932f5b70b05d512df1a4d319b01b3bd9","subject":"CI: Install kernel-modules-extra","status":"MERGED","created":"2026-07-20 12:57:57.000000000","updated":"2026-07-21 14:21:59.000000000","submitted":"2026-07-21 14:20:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"997984","meta_rev_id":"3b9e4163f0009b2f131449cc3292f13e3480bb65","_number":997984,"virtual_id_number":997984,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-07-21 14:20:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"recommended":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"all":[{"value":1,"date":"2026-07-21 09:18:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},{"value":2,"date":"2026-07-21 12:11:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2026-07-21 08:53:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-07-21 08:51:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":0,"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},{"value":1,"date":"2026-07-21 12:11:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},{"value":2,"date":"2026-07-21 09:03:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-20 13:49:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-07-21 08:51:15.000000000","updated_by":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"reviewer":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"state":"REVIEWER"},{"updated":"2026-07-21 08:53:01.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2026-07-21 09:18:11.000000000","updated_by":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"reviewer":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"state":"REVIEWER"},{"updated":"2026-07-21 12:11:34.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"}],"messages":[{"id":"87de5eb0079a18a5f59024ee67825716a9f0b6a6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-07-20 12:57:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7c70f46522eac74181f5337a7ab790c07801067f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-20 13:49:41.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/749035b77ece40d9a7d0781af66e38fd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e30a2d6241674e0dac475b7dc4c9bd24 : SUCCESS in 10m 24s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/3fed548f78ba495f9427b804f1b6e50c : SUCCESS in 7m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/90ea61fc075e4ca992b33cf17638135f : SUCCESS in 4m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ee9dc49eb42471eb715b68e89e30975 : SUCCESS in 5m 10s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3487a9f51dd54d2eabe1948fc735f11f : SUCCESS in 4m 03s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/7a9fd067ab924f8581bf34822793cb3c : SUCCESS in 8m 31s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ea22c0064f2f4637867198d41cc9c4c8 : SUCCESS in 6m 04s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9694a5f3269d4af3b307b360ed1bcc47 : SUCCESS in 7m 19s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a42f2749b72a40f184d06742d339c6bc : SUCCESS in 12m 57s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a58e0eb4c4994ed58489d5fc19cdce01 : SUCCESS in 12m 16s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/7aa433e2f0084054ab2752f5afd04e77 : RETRY_LIMIT in 2m 29s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/364c163208454582841e897496b8f406 : RETRY_LIMIT in 3m 57s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/527067268dfc4647b8a333bf65220527 : RETRY_LIMIT in 2m 01s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/e55b94bb897b4ec7938b61ce042f2cbc : RETRY_LIMIT in 2m 36s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/de58930962b4419d820eb700112dc36f : RETRY_LIMIT in 1m 59s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d1be36c667c74c3cad5c9812bed647a8 : RETRY_LIMIT in 2m 30s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/0f62d2ba907348a095332adf61460ee1 : RETRY_LIMIT in 2m 38s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b2562c4ca3ea4dc5b6a194297ea97fb3 : RETRY_LIMIT in 1m 46s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/f6f890cf4fa847189ed7a3db0d5ad9ab : RETRY_LIMIT in 2m 32s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/aaf4aeec1e934659a83b1e15d5ae3d9f : RETRY_LIMIT in 2m 47s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/d132d91e854d4217bab2491031fdea8b : RETRY_LIMIT in 3m 11s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/97cdc2d018fa49cba54eccf1c02ca817 : RETRY_LIMIT in 1m 43s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/10dc352a79994e72b10d88a79209662b : RETRY_LIMIT in 2m 46s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/1adf7985cb11431d9a248b82822bc429 : RETRY_LIMIT in 1m 50s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/14a351d239d54ad6a7bfed7f6447c84e : RETRY_LIMIT in 2m 21s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/e52359487d6d4dc68a8c5aed924162f9 : RETRY_LIMIT in 2m 26s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/c61c9155db4d42bea56fcf0de60e62ab : RETRY_LIMIT in 2m 51s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/de6f02b632254c94a13e9c82e11ade25 : RETRY_LIMIT in 2m 51s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/e224482d00a74b1c9ae57b40aa9810af : RETRY_LIMIT in 2m 36s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/9327b64f4f184c0ca1ca75d90a381440 : RETRY_LIMIT in 2m 41s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8d856c04b36049d4acaafbe6746918f4 : RETRY_LIMIT in 1m 49s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/f6b6d61e64a7462db537abea7321e13e : RETRY_LIMIT in 1m 51s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8d9d32cdc63e41d28b0f4219891f2787 : RETRY_LIMIT in 1m 48s","accounts_in_message":[],"_revision_number":1},{"id":"7c15292ededa34664270e207675fd8d7c8c4ab93","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-07-21 07:58:29.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"ee1292ff794d6a3adfca1bf7add5723b761f8ef4","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2026-07-21 08:51:15.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"e8639101cf4bf79659ddf4944ea2c5fbd2ecd12d","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-07-21 08:53:01.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"a8404b9ec59d18f4f7d7b49d309e561ff98240fe","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-07-21 09:03:33.000000000","message":"Patch Set 2: Review-Priority+2","accounts_in_message":[],"_revision_number":2},{"id":"383e2138295a738d4bff134c953738115f0360a3","author":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"date":"2026-07-21 09:18:11.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"b2a313bfc03b1ab385cfe8f3fc53ff14d0f0a831","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-21 10:00:52.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9f92b041ec8b41b8912b4877086b84af\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc5ceb0a4ba94ef9a1c594e27ae3a8df : SUCCESS in 5m 30s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/c8b30948459c42f6aeff276d0c540300 : SUCCESS in 6m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5ccc610a419b4140b705809e7e078ec9 : SUCCESS in 4m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c26ca142da4466888f1a4231150a9bc : SUCCESS in 4m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2be563d8f4524b22b1898b2a090adfe6 : SUCCESS in 4m 30s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/dcf2091ea1f84c3e8bb1e628b12d3aaa : SUCCESS in 7m 20s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c9acaa4b030d4b63bf1e763b5716aaf1 : SUCCESS in 4m 51s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/039fec6b948d42db93c3e2e5d6d5fcf6 : SUCCESS in 6m 05s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5ed389e60d2c4ec5a3a7a5541bbc4783 : SUCCESS in 13m 35s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2a7d9fe7f8054731963c1397321162ab : SUCCESS in 15m 23s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/ff978e2301ab4888b2757933d6661a6c : SUCCESS in 12m 38s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/64ff82817c384537ac04c36b4850f693 : SUCCESS in 16m 38s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c0d29c0f60d940288e5b172a0ef9cee7 : SUCCESS in 17m 33s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/3f7d237f7c014f70ac160b58335caecc : SUCCESS in 18m 33s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/e364b189e3404af2b4d3dcfe4fc897ff : SUCCESS in 18m 27s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/420f7c8f490b4b00a47da0efe7fbf75f : SUCCESS in 19m 54s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/6737658acac44e67875492549d397f58 : SUCCESS in 15m 57s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5ef978bbc7264514985642fc74889c28 : SUCCESS in 14m 07s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/6581db2277444b23a695fdf6381c26a6 : SUCCESS in 1h 09m 25s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/0bef7b4a351d4e318233e04d6c33e208 : SUCCESS in 1h 10m 36s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/60017cbcb0b24ba1bc385fe48ef8f92d : SUCCESS in 1h 08m 52s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7c2f79fc20fa40a0abc4d5f801451843 : SUCCESS in 1h 14m 10s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/890a4a94010a4849a6a1f532a4b36d2e : SUCCESS in 1h 14m 27s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/15b0ce037c1c4a92be265e95cad695f6 : SUCCESS in 1h 58m 06s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0dc6de304cca4969b7bd97edbb61d032 : SUCCESS in 2h 01m 00s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/369481195f6149b1b177140ec0369f5a : SUCCESS in 18m 27s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/076bdf7b67bd4ba79e498284d0939457 : SUCCESS in 18m 37s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/37ea203a423545f4a93ce036651c2aa5 : SUCCESS in 24m 37s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/d0533b93bac14da4bd2ef94f81a0e842 : SUCCESS in 17m 18s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/ed6b3f50200d480592432b04fe0ab0a4 : SUCCESS in 26m 31s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7669eb4c64644155a84826e6c2ba7cc7 : SUCCESS in 37m 03s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/35f4370285b3431e81102a180e0b0823 : SUCCESS in 19m 20s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c3f0e6a0ae7047cb8c09eddc32ec279f : SUCCESS in 16m 44s","accounts_in_message":[],"_revision_number":2},{"id":"d97390bf5d99fd0c2c3653e8d022f216ad96721c","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-07-21 12:11:34.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"2c9d000fa7f225ac038557da9d8408e499863046","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-21 12:11:52.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"34e93560cd60f5ed6dbb724a72c819a78db2f40a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-21 14:20:32.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c295711c809e48089c67e43572d868a4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3f27b64f4954c619413283d4aa1fbc2 : SUCCESS in 8m 12s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/64994af6f2a9492eb30ffcc4cd792ab0 : SUCCESS in 5m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b573b33884ec44c88dd040659cd16302 : SUCCESS in 4m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f95decefa8354dec9729dd7d8ee54e32 : SUCCESS in 3m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5c5fc23ac3074a6684143e330d9b67d5 : SUCCESS in 6m 20s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/5754302b8d2e4882992552c24d7cde8a : SUCCESS in 5m 25s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8e09aa53d18249209fe7aa1777d0c101 : SUCCESS in 12m 07s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/8bfc630733b44af7add52a38f69f0b14 : SUCCESS in 10m 32s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/b50cfbe78bab48a49dc6ee7380ac755e : SUCCESS in 14m 15s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/94a41e4375494fd390324020b2183433 : SUCCESS in 14m 18s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/026ed9a7b6ad47c5ba6e5b8ea88064fb : SUCCESS in 20m 30s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4f9427f1887e49d287084b09d3aba866 : SUCCESS in 19m 27s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/ee8e068740154ae5818fb948aec83cd9 : SUCCESS in 24m 42s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c37d4f564f5b4423830c94bce6c4d795 : SUCCESS in 21m 26s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/ffb82bae829543d39ed50ff3e312d2dc : SUCCESS in 1h 13m 35s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/3d7d72335e184057bb0150500c3fe1d4 : SUCCESS in 1h 14m 15s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/822f3b11373a4ebba39d3cf746dfa7e4 : SUCCESS in 1h 10m 06s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/37f2f7881d5b482d94eaf412079424a5 : SUCCESS in 1h 07m 02s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/4f3980c484e84cb3a61d99b076933204 : SUCCESS in 1h 07m 50s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/c07d4be133a542989b1affb299862ecb : SUCCESS in 1h 51m 32s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/704770baabe64385abc275cf22192a67 : SUCCESS in 2h 06m 39s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/d524001494214f91878d7c748a04c53e : SUCCESS in 28m 10s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/5304193bb0ba462c9d7ec757a98cd0a3 : SUCCESS in 17m 59s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/240c16ae13584ab58f478a4ecce0787a : SUCCESS in 24m 03s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/33f098433dae48fa8f8cac08ab80c710 : SUCCESS in 22m 29s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/4f6855fb411b4561b4cf4bfeff16284c : SUCCESS in 42m 20s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9bd8f3c1ff594d04ae5d01d65882d19d : SUCCESS in 38m 53s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/af5655e6d5bb4bedad5cdb147fe21ebb : SUCCESS in 17m 50s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b35340aa195d428eb4dd90d81fae18bb : SUCCESS in 23m 46s","accounts_in_message":[],"_revision_number":2},{"id":"235b31cc0f88bb69736a6c9e9546098206e38185","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-21 14:20:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"3b9e4163f0009b2f131449cc3292f13e3480bb65","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-21 14:21:59.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b4f03c866d9e48a9a0e33a4b5b35a357\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/daa70634e0ec43c5969014726c8fa0a0 : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"2c714a5a43864e6407e6c44d12f7da54e0cf9141","revisions":{"232c79847b8dd6aaffbcbbd2a29e09a0b7ad10dc":{"kind":"REWORK","_number":1,"created":"2026-07-20 12:57:57.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/84/997984/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/84/997984/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/997984/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/997984/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/997984/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/84/997984/1"}}},"commit":{"parents":[{"commit":"9a6e4333b2aa2fbe8dc80a93cd5aa3e2f1872970","subject":"Merge \"Adds seed IP to default no-proxy list\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9a6e4333b2aa2fbe8dc80a93cd5aa3e2f1872970"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-07-20 12:57:39.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-07-20 12:57:39.000000000","tz":120},"subject":"CI: Install kernel-modules-extra","message":"CI: Install kernel-modules-extra\n\nThis is required to fix various firewall configuration issues, such as:\n\n    sysctl: cannot stat /proc/sys/net/bridge/bridge-nf-call-iptables: No such file or directory\n\nChange-Id: I94d3a989932f5b70b05d512df1a4d319b01b3bd9\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/232c79847b8dd6aaffbcbbd2a29e09a0b7ad10dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/232c79847b8dd6aaffbcbbd2a29e09a0b7ad10dc"}]},"branch":"refs/heads/master"},"2c714a5a43864e6407e6c44d12f7da54e0cf9141":{"kind":"REWORK","_number":2,"created":"2026-07-21 07:58:29.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/84/997984/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/84/997984/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/997984/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/997984/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/997984/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/84/997984/2"}}},"commit":{"parents":[{"commit":"9a6e4333b2aa2fbe8dc80a93cd5aa3e2f1872970","subject":"Merge \"Adds seed IP to default no-proxy list\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9a6e4333b2aa2fbe8dc80a93cd5aa3e2f1872970"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-07-21 07:58:20.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-07-21 07:58:20.000000000","tz":120},"subject":"CI: Install kernel-modules-extra","message":"CI: Install kernel-modules-extra\n\nThis is required to fix various firewall configuration issues, such as:\n\n    sysctl: cannot stat /proc/sys/net/bridge/bridge-nf-call-iptables: No such file or directory\n\nChange-Id: I94d3a989932f5b70b05d512df1a4d319b01b3bd9\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2c714a5a43864e6407e6c44d12f7da54e0cf9141"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2c714a5a43864e6407e6c44d12f7da54e0cf9141"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
