)]}'
{"id":"openstack%2Fkayobe~998271","triplet_id":"openstack%2Fkayobe~stable%2F2026.1~I5c99da4c4ecc4995a31e96ffc9de4fe2620cdecb","project":"openstack/kayobe","branch":"stable/2026.1","topic":"addReconfigureDC-stable/2026.1","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2026-07-22 13:11:10.000000000","reason":"\u003cGERRIT_ACCOUNT_15197\u003e replied on the change","reason_account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}}},"hashtags":[],"change_id":"I5c99da4c4ecc4995a31e96ffc9de4fe2620cdecb","subject":"Add missing reconfigure.yml for manage-containers","status":"MERGED","created":"2026-07-22 07:34:03.000000000","updated":"2026-07-22 13:13:01.000000000","submitted":"2026-07-22 13:11:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"998271-addReconfigureDC-stable/2026.1","cherry_pick_of_change":997502,"cherry_pick_of_patch_set":5,"meta_rev_id":"c18d5115107c0e703c945f8ef6bfe2b1ef7284a1","_number":998271,"virtual_id_number":998271,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-07-22 13:11:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2026-07-22 11:03:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2026-07-22 10:03:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-07-22 13:11:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2026-07-22 11:03:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-07-22 13:11:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"}],"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-22 07:34:03.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"state":"CC"},{"updated":"2026-07-22 09:18:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-07-22 10:03:05.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2026-07-22 11:03:51.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"}],"messages":[{"id":"644e690cc8e1284db67b465f7b7d84b163df711c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-07-22 07:34:03.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"1e1b786b479cbbd50227b5614102c3a3c4af93dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-22 09:18:55.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6d7f8b87c6f54cfd926a16aff0eb94b5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc9e26e285364bc5841061f131063ccc : SUCCESS in 5m 33s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/f0a45764d55e4375b112ab621282fb45 : SUCCESS in 6m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/da9c28b941264210a5c34e055d93bf15 : SUCCESS in 3m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9947ea61d074e8796788fa2cec3ee79 : SUCCESS in 4m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0ca55761eefa4ff2843f2f89e12c212a : SUCCESS in 4m 33s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/31fe4a32a5b54faab262f97c1b1cd88b : SUCCESS in 7m 22s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e38047bf2abd4e1393732b3119be90d1 : SUCCESS in 4m 51s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a1f84b6c9afc46bb9d0c910bd629e67d : SUCCESS in 4m 15s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/643a1c9da9f4488a89d315d765d8a2e7 : SUCCESS in 12m 22s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/37ffd287411742039eb8a2a764af2ff5 : SUCCESS in 11m 32s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/74049b9a514f4ec3931122240ced1465 : RETRY_LIMIT in 3m 50s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/741fe230dfb646cab589646369c86185 : RETRY_LIMIT in 5m 30s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a6f97450ad8640a39b5d5743f1e80cb6 : SUCCESS in 15m 38s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/605f4889ba6b4818a2ab0ee18764ced8 : SUCCESS in 10m 52s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/2dfcd9822a46432e97fe3b65f48561f3 : RETRY_LIMIT in 3m 01s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/17b3cdefaea3490a8ce92a3cfd5b2eb6 : SUCCESS in 18m 31s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/716f126818664d6f8ab33d4b3fbb13ed : RETRY_LIMIT in 5m 42s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/277b436d2646465b902dfc8285e968d4 : SUCCESS in 17m 27s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/b18f05d775da44648756e00ca3422ecf : RETRY_LIMIT in 4m 28s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/b9f3b34cb5cc463391fcf32b91bf8eca : RETRY_LIMIT in 4m 16s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/342338c71fee4e9ea943f026912c2015 : RETRY_LIMIT in 4m 55s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9d537924ed1b4306a64c2bc81937d713 : SUCCESS in 1h 07m 43s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/9402fc96dcfb46a5b9c1a2a6b70a3e29 : SUCCESS in 1h 04m 01s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/df96309429694dd09b99ccbc2d210545 : FAILURE in 5m 33s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/bccc61f4f96e4507b1d38fbc639d925e : FAILURE in 5m 39s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7647f8cc127e424a84e191190247bf48 : SUCCESS in 1h 37m 11s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/54365caa89c24dc3acf5d7cc3a68e7f8 : SUCCESS in 1h 43m 38s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/4b8f0a46902144f39356aca0e63e89ee : SUCCESS in 21m 57s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/30c2fbd724a444908ddad94b80e8ba71 : SUCCESS in 17m 59s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/93a870a25bea41d095ef609ac5203a0c : SUCCESS in 14m 37s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/9fc7d1682af040859d4f1037a98b0fda : SUCCESS in 15m 37s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/cb364fd51b12440da479fa021d931a99 : SUCCESS in 17m 47s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/da97cd4b7c6147be927fdc49d32be44c : SUCCESS in 30m 51s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e795cd2b7fc44b92a32be8535e8c75d3 : SUCCESS in 29m 51s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9ab1ad6973ca4aa7bfc63b1ec728a6d7 : SUCCESS in 17m 37s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/54772db3a9b549b3b3238af5b26e375c : RETRY_LIMIT in 3m 17s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7979a9fc2d344e8fb27960d21e7ef878 : SUCCESS in 19m 50s","accounts_in_message":[],"_revision_number":1},{"id":"6685dd66a0eed4d5f0f200512a548d1d38575a2c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-07-22 09:50:38.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"9f281942fbb5b5c9c35d370decb2bc3018c919c6","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-07-22 10:03:05.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"ff3debaf5729cb46c1eb4bad774576c84875f892","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-07-22 11:03:51.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"2a96798a9a9eb880438f8e7c5a55211729050174","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-22 11:32:06.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/04fa0ef14b1b4d73a77bd820d3fcb50d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f44b14dc15364444af0a78c9594bc589 : SUCCESS in 9m 34s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/bdca201c33f14142b057b5244cd9b1dd : SUCCESS in 6m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/35c712e5d34b460cb1951f941cc3097a : SUCCESS in 4m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6572eddf0be0422d85744046ff36552b : SUCCESS in 4m 29s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/cc89b879a2b0424c8df8cffc30b72e62 : SUCCESS in 4m 03s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/2ea4ab6404e04ca0b9d49ee090680370 : SUCCESS in 4m 07s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e6d4ae70adac4ada9986dddfbad52b7c : SUCCESS in 5m 57s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7e9e1b8a23fb45a6a0d71731346c8a71 : SUCCESS in 6m 18s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c8235571b2ab46d5a8f7747084178478 : SUCCESS in 16m 14s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/dde2d5c68685446b8afb24fbb989e489 : SUCCESS in 9m 42s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/17d26ad7351f4c7693e56a0782df4cb6 : SUCCESS in 15m 12s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/c16d300af3824a37969f073c7c69cb92 : SUCCESS in 28m 03s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ddd32988d6524646a7342da59797dd0b : SUCCESS in 18m 02s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/bccde4e171574127979a0c84ae983a95 : SUCCESS in 20m 33s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/24e24f7852ed4202a792ca3160efbe07 : SUCCESS in 30m 42s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f6cfd4ce88f14eb8a46a4f9ae630764c : SUCCESS in 19m 45s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/fcaf2124becd4b00a1fd5a74d2253089 : SUCCESS in 18m 53s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/48742fe7f1f44de5bea65b12413ad1eb : SUCCESS in 20m 34s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/6166b49817d6432099a721a0820e1654 : SUCCESS in 1h 06m 44s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/d305b337421246758e80812fd7ca3c9e : SUCCESS in 1h 01m 09s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/53879bcc813545a5bf4adff14e7096a0 : SUCCESS in 1h 08m 42s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e4d3201d1a7e4aec9dca0200bb62da81 : SUCCESS in 1h 07m 17s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/448278c24da1444b9c6a014f77b2ffbb : SUCCESS in 1h 06m 53s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/9e572809fb6048d6b6960aac1f52b5a3 : SUCCESS in 1h 38m 08s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/3c23b355f1b34182a3042f6509abf527 : SUCCESS in 1h 33m 18s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5b9840c970a64a3e8cb7b69e5200fde5 : SUCCESS in 1h 34m 09s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/76ac0aac57dc431aa0643ea889370ca1 : SUCCESS in 1h 40m 28s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/50dba0cc8ba1458f9f2adc4aa829260d : SUCCESS in 19m 19s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/19d4358e6dda432f9e781aa52c321288 : SUCCESS in 28m 36s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2473341432df40799c9cdc170206cfd8 : SUCCESS in 24m 19s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/74f076d2055246b19e6f7fc535e0c8f9 : SUCCESS in 28m 19s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/d60a8a49abfb456f9806f71429ace4f3 : SUCCESS in 26m 25s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/a81dfefdd00b461986e622f6306a0262 : SUCCESS in 30m 56s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/95d071c02bbc46298562efd948dc81cb : SUCCESS in 31m 57s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3bbd17e4f3ce433caacdb22593140882 : SUCCESS in 35m 28s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/13fed21b9d1d441db5475ac2537b4e4d : SUCCESS in 27m 06s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dcb4b69bb649417c844c3e7dbc4e29ad : SUCCESS in 17m 56s","accounts_in_message":[],"_revision_number":2},{"id":"5ad8e78c58b18e3d1c31de7bea28286dfb4389b2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-22 11:32:29.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9cc93ea8332bd9e4ef3f5b64ded299d541bb0f73","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-07-22 13:11:10.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"ec00dcce36b369bb731c85a56f450cbada5b8780","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-22 13:11:44.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/44746ce42f7a40e1a19d7313b0d41824\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78765b45870d463c8017a8d661394689 : SUCCESS in 8m 27s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/59a376e625154a79898f7288be74140b : SUCCESS in 6m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba9b434be2ee4af389c915ccca526d72 : SUCCESS in 3m 12s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d8aa1c427fbf48a19a30840299583961 : SUCCESS in 4m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0e541ab7e4464ef7b9bc3afe67c16686 : SUCCESS in 6m 41s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e44fea8b9e4a45ad9ecd0bc2db4be400 : SUCCESS in 5m 49s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f9b3e209196e406ea00b8c48365a3716 : SUCCESS in 12m 12s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/bdb8dd8735624dc7a9049171b30a7a11 : SUCCESS in 12m 16s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/3feb3ff1d9ee43cea86ecefa000facbf : SUCCESS in 12m 53s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/daa07b103eb24913a94d0eb938b361b9 : SUCCESS in 9m 53s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/2b98243a414a4a4dad0503859966552c : SUCCESS in 27m 08s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b15b01a2df9344fdafebf51aa1c0f250 : SUCCESS in 12m 57s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/f959940b4d2c409083a09df1bc8109f7 : SUCCESS in 12m 04s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/13a8cc64946b4bd99482aeb11f901769 : SUCCESS in 15m 53s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/5a24d79fa3fe4f61a56bc39ac28b2ad0 : SUCCESS in 1h 09m 19s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/75fa7867a21641b5b086a8a6f5433a7c : SUCCESS in 1h 02m 59s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/bc7916d162a24187bfc4ec557c348ff8 : SUCCESS in 1h 08m 05s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3c3842450b9d41da81d50593dd90114d : SUCCESS in 1h 08m 37s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/4c31aa8a14ba4aa99a3507954ef3fb2d : SUCCESS in 1h 10m 13s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/6dec86eae2ac4f848f30aca279932f9f : SUCCESS in 1h 36m 07s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/e8ef7770c53b4e71a9caa442376bdb4f : SUCCESS in 1h 28m 03s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e1d958ad42ee4cc3850b7d43bd091627 : SUCCESS in 1h 33m 02s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f8d27305d2324b949f506b41710c4a0c : SUCCESS in 1h 38m 29s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/085f0fd52534498484875e0e3294daa8 : SUCCESS in 17m 42s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/44baa10f501f4c7288c5bc7183125746 : SUCCESS in 21m 52s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/aa1097e5da5c4224b849ea00b3767123 : SUCCESS in 15m 54s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/d84313e4611c4c17be549e0df9020e84 : SUCCESS in 21m 24s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/8da6a8b86dca417ab1f1bf740744bc30 : SUCCESS in 35m 11s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/4fefcd0027ff4e8e8421e63a7130dcd6 : SUCCESS in 29m 50s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6d9ee85669d04641b961985e9178da4c : SUCCESS in 30m 29s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/72256c6bf7b54b6ab172ca0a58f7e153 : SUCCESS in 31m 04s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/e2fa6e83f0c249c49e20fb4395734471 : SUCCESS in 16m 42s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/668b5873bcf3427db580f8f08fd1f9ee : SUCCESS in 15m 20s","accounts_in_message":[],"_revision_number":2},{"id":"e381af0d395a95c8339f008f356b72fc4e118c35","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-22 13:11:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"c18d5115107c0e703c945f8ef6bfe2b1ef7284a1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-22 13:13:01.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/da43e77fe74f4e229277d48b9fb5a2e8\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a14c2aefe6e243c489f1ede8cf9f078c : SUCCESS in 57s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/35e657b4ffb64a93a0fc5d04a03ae858 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"622332fe837835b3e46e5f37fcec80ebc09fa725","revisions":{"afd88e0682c51e8e2ebd74be415133ca9dbbed4a":{"kind":"REWORK","_number":1,"created":"2026-07-22 07:34:03.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/71/998271/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/71/998271/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/998271/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/998271/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/998271/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/71/998271/1"}}},"commit":{"parents":[{"commit":"2f4d55f3d42caec8d5d863a659415556ba8e48af","subject":"Fix kolla inventory hostname formatting","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2f4d55f3d42caec8d5d863a659415556ba8e48af"}]}],"author":{"name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","date":"2026-07-16 09:39:43.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-07-22 07:34:03.000000000","tz":0},"subject":"Add missing reconfigure.yml for manage-containers","message":"Add missing reconfigure.yml for manage-containers\n\nCloses-Bug: #2160941\nChange-Id: I5c99da4c4ecc4995a31e96ffc9de4fe2620cdecb\nSigned-off-by: Leonie Chamberlin-Medd \u003cleonie@stackhpc.com\u003e\n(cherry picked from commit 65cc58144a50e86a861bd1e7eafe5d055eebb755)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/afd88e0682c51e8e2ebd74be415133ca9dbbed4a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/afd88e0682c51e8e2ebd74be415133ca9dbbed4a"}]},"branch":"refs/heads/stable/2026.1","conflicts":{"base":"9a6e4333b2aa2fbe8dc80a93cd5aa3e2f1872970","ours":"2f4d55f3d42caec8d5d863a659415556ba8e48af","theirs":"65cc58144a50e86a861bd1e7eafe5d055eebb755","merge_strategy":"recursive","contains_conflicts":false}},"622332fe837835b3e46e5f37fcec80ebc09fa725":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-07-22 09:50:38.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/71/998271/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/71/998271/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/998271/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/998271/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/998271/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/71/998271/2"}}},"commit":{"parents":[{"commit":"739ca7b9ea2cab8de7bc27d667ff48d8a51040fa","subject":"CI: Install kernel-modules-extra","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/739ca7b9ea2cab8de7bc27d667ff48d8a51040fa"}]}],"author":{"name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","date":"2026-07-16 09:39:43.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-07-22 09:50:38.000000000","tz":0},"subject":"Add missing reconfigure.yml for manage-containers","message":"Add missing reconfigure.yml for manage-containers\n\nCloses-Bug: #2160941\nChange-Id: I5c99da4c4ecc4995a31e96ffc9de4fe2620cdecb\nSigned-off-by: Leonie Chamberlin-Medd \u003cleonie@stackhpc.com\u003e\n(cherry picked from commit 65cc58144a50e86a861bd1e7eafe5d055eebb755)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/622332fe837835b3e46e5f37fcec80ebc09fa725"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/622332fe837835b3e46e5f37fcec80ebc09fa725"}]},"branch":"refs/heads/stable/2026.1","description":"Rebase","conflicts":{"base":"2f4d55f3d42caec8d5d863a659415556ba8e48af","ours":"afd88e0682c51e8e2ebd74be415133ca9dbbed4a","theirs":"739ca7b9ea2cab8de7bc27d667ff48d8a51040fa","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
