)]}'
{"id":"openstack%2Fkayobe~998273","triplet_id":"openstack%2Fkayobe~stable%2F2026.1~I94d3a989932f5b70b05d512df1a4d319b01b3bd9","project":"openstack/kayobe","branch":"stable/2026.1","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2026-07-22 12:04:22.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I94d3a989932f5b70b05d512df1a4d319b01b3bd9","subject":"CI: Install kernel-modules-extra","status":"MERGED","created":"2026-07-22 07:41:13.000000000","updated":"2026-07-22 12:05:19.000000000","submitted":"2026-07-22 12:04:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"998273","cherry_pick_of_change":997984,"cherry_pick_of_patch_set":2,"meta_rev_id":"a6e8d8f24401c953263bb38fc6b8ab1bdf0a7470","_number":998273,"virtual_id_number":998273,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-07-22 12:04:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2026-07-22 10:17:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2026-07-22 10:02:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2026-07-22 10:02:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2026-07-22 10:17:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-07-22 09:50:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"reviewers":{"REVIEWER":[{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-22 09:34:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-07-22 09:50:30.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2026-07-22 09:50:30.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2026-07-22 09:50:30.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2026-07-22 09:50:30.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"state":"REVIEWER"},{"updated":"2026-07-22 09:50:30.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2026-07-22 09:50:30.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2026-07-22 09:50:30.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2026-07-22 10:02:55.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"382dc0ccb1ac4d59bbf66487bd335eaf55ec78be","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-07-22 07:41:13.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"45e08a3908ad42fd4ca8b7c7464cbb8a5e49ed0b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-22 09:34:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2b68c68e8bf7456c8282f53f83ad6021\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2b5fd799d8841ff810ead83e92da2fa : SUCCESS in 9m 32s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/9d6274d9400f4655b2497294eb186346 : SUCCESS in 6m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ce31dea3f0bb47a29fda25e6c3ff7c55 : SUCCESS in 4m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/109a3924c6794ba1b38bd4137a4a2a7b : SUCCESS in 4m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b5a734aac012450881cc4dc1da0fbfcd : SUCCESS in 5m 08s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/6a88f402daad476fae995e5de649d6ff : SUCCESS in 6m 38s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2e4c1759d69f457da043b83048b9daa1 : SUCCESS in 6m 06s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/07a887c5a27d4d4bbb058ee7491f5188 : SUCCESS in 4m 52s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1140b62070e742b494ed4cd6a8779da6 : SUCCESS in 11m 59s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/59ce25f0987a4510bd1a9fa2f25878ec : SUCCESS in 11m 55s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/fbea5a643ed24370af60dd20ff00d53f : SUCCESS in 15m 47s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/e4ef969fe164439e98ceb6405b486f22 : SUCCESS in 14m 26s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/78e6dedb58904caca8b87b82d0a9f9ea : SUCCESS in 16m 28s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/b806f85f326e4fac94fc0fb600d0f585 : SUCCESS in 13m 34s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/b1764361ce544e6a81441189fcb363ef : SUCCESS in 15m 44s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1921e70bba674b98ae159bd1c2ac3088 : SUCCESS in 15m 44s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/7f1067b29edb4ba79e57202ea871ff2c : SUCCESS in 11m 33s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0301c5407b9544738d2933321ef524a1 : SUCCESS in 17m 38s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/4b480d2ffb1a41eaa2a85ac47c912e02 : SUCCESS in 1h 18m 03s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/5fbd696ccc624ed698208f44e96d380e : SUCCESS in 1h 18m 28s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/fcb3d35d1b114e62873581cee7077f6d : SUCCESS in 1h 05m 27s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/be4a05234e16442cb3b98cda5551fd2d : SUCCESS in 1h 05m 29s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/70238d78e89141459cd139d3fa0795cb : SUCCESS in 1h 23m 44s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/e43cd4378f124a74afc06dabc5ea3855 : SUCCESS in 1h 52m 09s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/500900b42a3c49e4ade94dfb869d094b : SUCCESS in 1h 32m 35s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a0a86213bab04396ae62c3705a4f1f24 : SUCCESS in 1h 36m 18s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0215e4d015b84a6fb822892a335d438e : SUCCESS in 1h 41m 14s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/0e8abcb2dd17487bb9bbd9d08af6d006 : SUCCESS in 23m 27s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/b83b0efa07e840f2a2ac3d3a563b0401 : SUCCESS in 18m 21s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4558361ba1544b7786843ff84d161689 : SUCCESS in 23m 47s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/aa4ef8b903654e76ab9b11a3911aa58e : SUCCESS in 22m 07s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/dfbaccb112a443de86f4c23add96fe0d : SUCCESS in 33m 19s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/32a3827e1f164c09b0d2c94637b5b16e : SUCCESS in 25m 08s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5211c3d85d20413ea4aa7d0a97cc4321 : SUCCESS in 32m 40s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/36140cf446dd40dead7c495b47454312 : SUCCESS in 31m 58s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/c67f022570d04d18a1e9d30146f8a699 : SUCCESS in 16m 34s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7a405507e5044588ae4b8bbc87e91344 : SUCCESS in 21m 49s","accounts_in_message":[],"_revision_number":1},{"id":"7156004a2e6b6f15f9918cda65e60ee5359b2091","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-07-22 09:50:30.000000000","message":"Patch Set 1: Review-Priority+2","accounts_in_message":[],"_revision_number":1},{"id":"012786f72aa1dbeeec1d0a76f4d0e28547bf0c8b","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-07-22 10:02:33.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"58f1ce9196dd9ecd290591893fffa0990a5fea2f","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-07-22 10:02:55.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"53a424c7d6d68729690de15fe918913c79ef18d5","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-07-22 10:17:58.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"cf2ed41142e37d243ee1ac02197c870dc6f3e16a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-22 10:18:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"638d4c6d5eefb98a78ff1af892811053af7541e7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-22 12:04:22.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/adc0760c0071488797917f2a0594cb60\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1d546a1b5e943f098ecc69dc6dfd6f7 : SUCCESS in 4m 55s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/df9c62750b95464da972c8e74714fb2d : SUCCESS in 4m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca0bbd214d554156b88025d994f343a8 : SUCCESS in 4m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d3fbc180151a47a08b76839b1e3a0f1e : SUCCESS in 4m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/74338110c31f4ddcbadfdb63e7c46e04 : SUCCESS in 4m 48s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/132f94e2635a4b3dadd13a4e0d1940cf : SUCCESS in 4m 46s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c394a90a2cbb4b56b559b92c9f2ddfd1 : SUCCESS in 15m 32s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/beb3b62f698b48adbc28ef4566ded135 : SUCCESS in 15m 51s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/4e207ffb234841a0a7ad5f74e1f744aa : SUCCESS in 21m 03s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3ae73dc9578544858dc1f7895d6c85c6 : SUCCESS in 11m 22s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/1dcd2cfec73c47ca94b9e303040c0dfb : SUCCESS in 19m 28s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2326cdcfe3da49b0a5b032a205f8d6ba : SUCCESS in 19m 33s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/04c39953ea364867af5d77a31876e431 : SUCCESS in 23m 08s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8b044d66cd0c472292676a129abffefb : SUCCESS in 20m 08s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/5d75bee8327143d0a1ee90ee7a755469 : SUCCESS in 58m 15s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/800fa8900fdd49c987bbee053297a4ee : SUCCESS in 1h 04m 19s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/d0e1aecd6ac84ce69dfc3551417f9ca4 : SUCCESS in 1h 04m 39s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/458f6f6bd20b49039063d3cafb6a2300 : SUCCESS in 1h 06m 47s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/7d3d16c6e59947e78b573c1d83c2e6c7 : SUCCESS in 1h 07m 54s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/ea55cdebf1c34e4abf5a6647ac3291f1 : SUCCESS in 1h 44m 47s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/3d6b6fd096dc465ea86db22efede68ee : SUCCESS in 1h 36m 53s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/11362adc11e04687856e098b17dd67a4 : SUCCESS in 1h 38m 51s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/54ed8f5a7a1849bca3e54e8e30edd82a : SUCCESS in 1h 39m 51s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/c17a5b1c055549f99d84aa2a12401d93 : SUCCESS in 15m 03s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/be833920988c4e3a9dfb9c3491c670d7 : SUCCESS in 15m 20s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2c3c0d0c69594ae09ae31ba8b598206a : SUCCESS in 22m 53s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/672743d5b2ac41ebadfd8c52a86673b3 : SUCCESS in 14m 27s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/b3d6c012daa64381955867e9c42140f9 : SUCCESS in 16m 55s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/127457be99904de78c6fc2457876dbb1 : SUCCESS in 21m 17s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dec652b4e8ee4fceb8503e0c2a2bd8bf : SUCCESS in 31m 01s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ade6eb346bcc46d69c50d72a0d3b3336 : SUCCESS in 30m 31s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/7f77b92865c344969db4124b6d3ecb7a : SUCCESS in 23m 32s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3df22c04707a407b933ad0e73f0f287e : SUCCESS in 20m 43s","accounts_in_message":[],"_revision_number":1},{"id":"37b288e896b29bc58f27f78f91b6e5054fcec886","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-22 12:04:22.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"a6e8d8f24401c953263bb38fc6b8ab1bdf0a7470","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-22 12:05:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/28b5b287be844d27bfc557cd3704e5f3\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91323f5ed96442dea2af715dc6a77632 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"739ca7b9ea2cab8de7bc27d667ff48d8a51040fa","revisions":{"739ca7b9ea2cab8de7bc27d667ff48d8a51040fa":{"kind":"REWORK","_number":1,"created":"2026-07-22 07:41:13.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/73/998273/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/73/998273/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/998273/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/998273/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/998273/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/73/998273/1"}}},"commit":{"parents":[{"commit":"2f4d55f3d42caec8d5d863a659415556ba8e48af","subject":"Fix kolla inventory hostname formatting","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2f4d55f3d42caec8d5d863a659415556ba8e48af"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-07-21 07:58:20.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-07-22 07:41:13.000000000","tz":0},"subject":"CI: Install kernel-modules-extra","message":"CI: Install kernel-modules-extra\n\nThis is required to fix various firewall configuration issues, such as:\n\n    sysctl: cannot stat /proc/sys/net/bridge/bridge-nf-call-iptables: No such file or directory\n\nChange-Id: I94d3a989932f5b70b05d512df1a4d319b01b3bd9\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n(cherry picked from commit 2c714a5a43864e6407e6c44d12f7da54e0cf9141)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/739ca7b9ea2cab8de7bc27d667ff48d8a51040fa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/739ca7b9ea2cab8de7bc27d667ff48d8a51040fa"}]},"branch":"refs/heads/stable/2026.1","conflicts":{"base":"9a6e4333b2aa2fbe8dc80a93cd5aa3e2f1872970","ours":"2f4d55f3d42caec8d5d863a659415556ba8e48af","theirs":"2c714a5a43864e6407e6c44d12f7da54e0cf9141","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
