)]}'
{"id":"openstack%2Fkayobe~998274","triplet_id":"openstack%2Fkayobe~stable%2F2025.2~I94d3a989932f5b70b05d512df1a4d319b01b3bd9","project":"openstack/kayobe","branch":"stable/2025.2","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2026-07-22 12:47:48.000000000","reason":"\u003cGERRIT_ACCOUNT_15197\u003e replied on the change","reason_account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}}},"hashtags":[],"change_id":"I94d3a989932f5b70b05d512df1a4d319b01b3bd9","subject":"CI: Install kernel-modules-extra","status":"MERGED","created":"2026-07-22 07:41:57.000000000","updated":"2026-07-22 19:38:04.000000000","submitted":"2026-07-22 19:37:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"998274","cherry_pick_of_change":997984,"cherry_pick_of_patch_set":2,"meta_rev_id":"3bdb757bf53b3ae5f58ad247e9a9005fe494cbcb","_number":998274,"virtual_id_number":998274,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-07-22 19:37:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2026-07-22 10:18:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2026-07-22 10:03:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2026-07-22 10:18:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-22 09:18:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-07-22 10:03:32.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2026-07-22 10:18:17.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"}],"messages":[{"id":"d40ba7f2cefc32b6607049566e0ada6f19f06fd8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-07-22 07:41:57.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"0b4c10d6ac5a2472aa21e812cbcddf233edfdda0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-22 09:18:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/11da4093a5404a1fabb831dd6d094240\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8f532db53774f718017235340e359db : SUCCESS in 9m 53s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/f980df00fc8441529308b25529d6fbc1 : SUCCESS in 6m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/35c1bcf275594095b60ff0883f54bd95 : SUCCESS in 5m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6de11b378f74464ac51d54e350cb65f : SUCCESS in 4m 34s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/76d1da2f99f3435094200dcdbaf787e1 : SUCCESS in 4m 38s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a5965c56f99e438590891a8d414c746e : SUCCESS in 5m 37s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/341ed98a51b14043907ebe541970e8b2 : SUCCESS in 6m 47s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/23b4683681cb4c5d91908924bc8b6b5c : SUCCESS in 4m 13s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8093a55bb21a4396920be45614ace783 : SUCCESS in 12m 01s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/0e393a558390481a92305b347de371a2 : SUCCESS in 11m 54s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/aae7ab8648f0445d8aa50b6b736b1abb : SUCCESS in 14m 54s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ac24113a1e214668ae2f6815a8a6fd87 : SUCCESS in 12m 15s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/f97b017a0e0041408306ef5e08c1a2f3 : SUCCESS in 16m 38s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0f8fb3cf917f4b0ea8bb96b6e2c30272 : SUCCESS in 16m 33s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/55e1865d71e843a99f4567f16c0f75cf : SUCCESS in 55m 41s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/bbe4fc9162b64f0b9ddf6c877754578d : SUCCESS in 56m 34s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/6b49243abf12431dbd3b553fcc453c23 : SUCCESS in 57m 40s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fc0b06dbc2bc4f6792dcf0ea5d2f326d : SUCCESS in 1h 00m 14s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/b587a01742c54dc48ce87a6c22a49a7b : SUCCESS in 59m 38s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/57a4e345ab0b45aa82fc3a550b5eabf9 : SUCCESS in 1h 29m 33s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/213b915d891e4f0caf1d3c36839a64c7 : SUCCESS in 1h 34m 44s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/1a8c17cf4c1247ffa977e52850a3cdf4 : SUCCESS in 11m 04s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/a9fece995671456c9456806c3be048a6 : SUCCESS in 16m 54s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/abcc4f50e52a41b49db12e3f6c51df99 : SUCCESS in 19m 00s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/ce52d431070243e9b1b86d6761f92332 : SUCCESS in 16m 48s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/c1f613b224eb469b90959878505f01f2 : SUCCESS in 31m 35s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b18266094ae146f0ab33045d9ec8c023 : SUCCESS in 31m 40s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/a28a18f3b6574e01ad16b3c6646d8039 : SUCCESS in 13m 35s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/14729de80c944d4292dca3d7b73a4c66 : SUCCESS in 13m 38s","accounts_in_message":[],"_revision_number":1},{"id":"aadb6b0b39e49e4b73edd80f91537770ce38e7ef","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-07-22 10:03:32.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"991356708a0b1d2ccfeadc7dd774374e61df92ad","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-07-22 10:18:17.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"28a1d82fd9edd0d193422ac5cf7c82cc1640cdac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-22 10:18:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"961ebb3451204e77a8a57a881cee6d1ece928a3b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-22 12:04:24.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/30eabce4c56d4944b835c15c8224d6c1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f347ce4259924d0a88aa302913b8a777 : SUCCESS in 8m 59s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/5e8b6bf665a3474bb9270644a819ba5f : SUCCESS in 7m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/477b47fb648645d9a553ef99480b5f06 : SUCCESS in 4m 23s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9582abfdd04c4fde89d1e715ed2acf7c : SUCCESS in 4m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ea71531f991c412591a0bc167758d3a8 : SUCCESS in 6m 12s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/244cee1544554453bbf4b8c56f38aeb2 : SUCCESS in 6m 30s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5739f3d28f1c4677b045a87731852435 : SUCCESS in 12m 30s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a113cdcf77514d97970022ad67f7682d : SUCCESS in 11m 52s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/07ea86edfcd44814ae6b261dbd0c4d20 : SUCCESS in 20m 05s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6902eb8008bf48f483e43eb51dd270d7 : SUCCESS in 14m 10s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/136af8a9ee144c2db681eabf0c08f430 : SUCCESS in 11m 25s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b9d1e9ad281d4238a35d3e6dbdafe5df : SUCCESS in 16m 30s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/4cbdde452ebc46bd88870d78521eb919 : SUCCESS in 1h 15m 50s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/83b6604a07794d4ea1b7b3e0edc2dffe : SUCCESS in 55m 18s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/7ba9388cac4d4ac3adb996fecc81c902 : FAILURE in 1h 17m 48s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8b411b140f68400f990ab69fe3f01554 : SUCCESS in 1h 01m 41s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/b73fa26f5a9246c09caad409c2c00780 : FAILURE in 31m 30s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/ddf440485e9f46269897c57b2d0ae3b4 : SUCCESS in 1h 28m 45s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/10eb6ccc6a3344038dc29531c6df1441 : SUCCESS in 1h 33m 27s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/5e3ca39ff1c64b439954eba7eff5b914 : SUCCESS in 12m 55s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/6b65ce30576747fbb4f8ae87dabc1896 : SUCCESS in 10m 15s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ab637c80472644f4848b7bcfd85a6ec1 : SUCCESS in 18m 24s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/69bc3485b5a4447587b91670a3165a78 : SUCCESS in 10m 33s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/2618063ebfad450fbf0d2fed67d32f6b : SUCCESS in 27m 49s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e2bfff8e829147a397556a8a6cf20673 : SUCCESS in 27m 54s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/625ff8fafac344269eaaf3f9934bedcc : SUCCESS in 16m 18s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/69be2424cf5b4fb7bbdf1097613910a6 : SUCCESS in 8m 36s","accounts_in_message":[],"_revision_number":1},{"id":"811290d543430ea7a2830ae48e582b932e90c911","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-07-22 12:47:48.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c9bc35e541003042259108ae5fa6f632ceef10e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-22 14:18:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/26ef2d2895e24ea6a260f9d1432e4d40\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da2a89a0589a412b9e27b865ac773449 : SUCCESS in 5m 17s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/2cf4c7c463c547f1814e02bdff50e3a8 : SUCCESS in 8m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f1f1f52c0f2e45aab7d277d83e4ea6ea : SUCCESS in 5m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8626adfbd9354b928653a3349affbe44 : SUCCESS in 4m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/306ebc6bfce7446abf5d9cc37d4a7fea : SUCCESS in 6m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8ac12f06a9e24002ac271cf1830e0d20 : SUCCESS in 9m 02s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/109c8324bca644d696012f6b645e6728 : SUCCESS in 2m 55s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a6750eba40ef4cb9b1023d8213f96ccf : SUCCESS in 3m 26s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/893ef40748594291a4edda6faf95d5f9 : SUCCESS in 15m 15s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/8745237ac5a2486c97e9ee962f6652ba : SUCCESS in 11m 16s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/0e9c7efa167c47fe80d5a767b7afb731 : SUCCESS in 10m 00s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cc6994ff5ed84d3890810ecf903820ec : SUCCESS in 17m 13s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/e2ece3aeb59741e1973054bd3b1b6054 : SUCCESS in 14m 35s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/74c88a418ca744d69b8d8f655458650d : SUCCESS in 21m 19s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/d6bf9a3a237e4b9594b390653ff31c97 : SUCCESS in 55m 19s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/ff50f65c88164e59a1bbed0c3711130e : SUCCESS in 58m 08s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/416a8a63259049afa7ffae40931b23e2 : SUCCESS in 1h 01m 11s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/583e29b131d64904969c0f6a17f192bf : SUCCESS in 1h 03m 19s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/7619101534774daf9ec155798d5c5f93 : SUCCESS in 1h 04m 47s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/08ac95e6ac394e3e870a6fcea5afc402 : SUCCESS in 1h 24m 40s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/abb51d42762649c0a3f13154896fe98c : SUCCESS in 1h 27m 54s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/d1c25647f60f4c81b1ff1f3ff8d8486d : SUCCESS in 28m 31s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/0d9d8cc3942e4b3aacb893fc39f99f55 : SUCCESS in 24m 47s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/74bccd20be8b4e4e9c72080aa25e5497 : SUCCESS in 17m 57s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/8f74c8dc3ebf497bab0ae0425cdac77f : SUCCESS in 16m 45s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/e728d9ff5ea34cfaab0dde83873ff07b : SUCCESS in 37m 23s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ea408b7828d74d4fae863b9b1a2c30e8 : SUCCESS in 38m 42s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/8e3be72bf8b9445ebd7f7670293667e5 : SUCCESS in 11m 23s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/63d6ddaf52384706850af0a0365bf4a7 : SUCCESS in 15m 15s","accounts_in_message":[],"_revision_number":1},{"id":"f2b966b6359ba9c98c8be634428663620ac667cb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-22 14:18:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4cb4b9da50a5042fbc384715abc42565a7aa1642","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-22 19:37:07.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/64a71548796945ee91ac2435f792943a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a388d8f6ece3456da7b9f735da25ce3a : SUCCESS in 9m 56s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/2546e26fc76345388a01366c90c8499d : SUCCESS in 9m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f6b775b5ccb4134951ba948bf700491 : SUCCESS in 5m 49s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3d68fdfb7ded4763ab761ba40313c81b : SUCCESS in 4m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ebde4737901343b1b2a2050f12253ed2 : SUCCESS in 7m 17s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/fb5b3d60c7284048bcea95112ac5cdda : SUCCESS in 5m 46s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ae4c0830c075419989ffec6c44498a5f : SUCCESS in 15m 55s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/007d13610b244ad5ae0b020dcec4d91d : SUCCESS in 14m 23s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/60a00ec747104edea69d8019421ee203 : SUCCESS in 15m 21s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5df8542bcb684e239040755e7cf56018 : SUCCESS in 13m 13s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/8b6bc1cd38b44979b95bc8fc01b3a317 : SUCCESS in 18m 13s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3d466e9e9371448f9b3efacae1c0bf56 : SUCCESS in 16m 27s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/721f78b2349b43b687b68e3804c74a73 : SUCCESS in 47m 12s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/9c408626b1c54ee2a3cd6c84874649ad : SUCCESS in 38m 37s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/14c8ca174979499c85f2866ff53d5fb9 : SUCCESS in 40m 10s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4fc8cb2850c745e1934812ed39da2652 : SUCCESS in 1h 08m 45s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/b80c125dd7df468bb187cfc7f0d5ffdc : SUCCESS in 37m 09s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/a7635f93be1c44469f18df3976c4c857 : SUCCESS in 1h 05m 24s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5d80a494e3ca4628a24687b38d4cca43 : SUCCESS in 1h 00m 38s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/58e191a96d0e4c6d808ec09e90fd8712 : SUCCESS in 24m 33s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/8019bffe9d014630a8c8d1369fd8f6b1 : SUCCESS in 18m 51s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3ba1f745c67b40f58a1bf4f664c86141 : SUCCESS in 18m 59s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/8c1a8812e13646858cb95f87608956cc : SUCCESS in 17m 04s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/cf52697680984c39b4632ecfbe72cb91 : SUCCESS in 29m 54s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fbf5b26b7ea347bdb13ca7ce5f386078 : SUCCESS in 30m 38s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/3af04d71677e4eea9ceb0a7d471c7d2a : SUCCESS in 17m 43s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/19e303dfd01a4752b1443c0ab1086a5b : SUCCESS in 15m 12s","accounts_in_message":[],"_revision_number":1},{"id":"60ef0a9f8c8b6c0ca36bc0aaaf1c0bb5bbbce5e0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-22 19:37:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"3bdb757bf53b3ae5f58ad247e9a9005fe494cbcb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-22 19:38:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/314d1fb2d5f74112801610876309cabd\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24904b2f751a459c9c99574117ebdebc : SUCCESS in 43s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1ecd602cb701c13f905082fe3bcb7ee1451875be","revisions":{"1ecd602cb701c13f905082fe3bcb7ee1451875be":{"kind":"REWORK","_number":1,"created":"2026-07-22 07:41:57.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/74/998274/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/74/998274/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/998274/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/998274/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/998274/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/74/998274/1"}}},"commit":{"parents":[{"commit":"2f57c0d1e9e29c46deb395b4d44cc66f7d091b7b","subject":"Add support for url sources in kolla_sources","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2f57c0d1e9e29c46deb395b4d44cc66f7d091b7b"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-07-21 07:58:20.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-07-22 07:41:57.000000000","tz":0},"subject":"CI: Install kernel-modules-extra","message":"CI: Install kernel-modules-extra\n\nThis is required to fix various firewall configuration issues, such as:\n\n    sysctl: cannot stat /proc/sys/net/bridge/bridge-nf-call-iptables: No such file or directory\n\nChange-Id: I94d3a989932f5b70b05d512df1a4d319b01b3bd9\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n(cherry picked from commit 2c714a5a43864e6407e6c44d12f7da54e0cf9141)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1ecd602cb701c13f905082fe3bcb7ee1451875be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1ecd602cb701c13f905082fe3bcb7ee1451875be"}]},"branch":"refs/heads/stable/2025.2","conflicts":{"base":"9a6e4333b2aa2fbe8dc80a93cd5aa3e2f1872970","ours":"2f57c0d1e9e29c46deb395b4d44cc66f7d091b7b","theirs":"2c714a5a43864e6407e6c44d12f7da54e0cf9141","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
