)]}'
{"id":"openstack%2Fkayobe~998287","triplet_id":"openstack%2Fkayobe~stable%2F2026.1~I3717723a9213617272e9ae97664d4ad8febc0aab","project":"openstack/kayobe","branch":"stable/2026.1","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2026-07-22 19:37:10.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3717723a9213617272e9ae97664d4ad8febc0aab","subject":"Remove leftover device mapper references from the code","status":"MERGED","created":"2026-07-22 09:52:32.000000000","updated":"2026-07-22 19:38:10.000000000","submitted":"2026-07-22 19:37:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"998287","cherry_pick_of_change":997714,"cherry_pick_of_patch_set":3,"meta_rev_id":"339b6269f85cf25c424a606ec70221959d39dab4","_number":998287,"virtual_id_number":998287,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-07-22 19:37:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2026-07-22 15:20:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2026-07-22 10:14:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2026-07-22 11:33:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2026-07-22 16:13:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-22 09:52:32.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"state":"CC"},{"updated":"2026-07-22 10:06:15.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2026-07-22 11:33:59.000000000","updated_by":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"reviewer":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"state":"REVIEWER"},{"updated":"2026-07-22 11:57:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-07-22 15:20:35.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"}],"messages":[{"id":"80ebbd8ebebcfd869dcc3f13bfa16115631f1216","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-07-22 09:52:32.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"a9c42f7168d6062bac146a29964ac81a8764b9a6","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-07-22 10:06:15.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"df66f129e81009bd27c64699535ff65375db79a9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-07-22 10:14:02.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n","accounts_in_message":[],"_revision_number":2},{"id":"671a2196b62e1c8ec3fdab5834456b1a40d14ec7","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2026-07-22 11:33:59.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"aedf0b0f8401771249b615aaa8933cad257f0cc1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-22 11:57:40.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b70e2b2466ba400e94ca7a61331a33b2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df6351544fa74fa69cf9b76c0f4ba6f2 : SUCCESS in 4m 41s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/7481c8748d2e4c48912742fbaba4305a : SUCCESS in 7m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/90bb8cb0e2c84f12becf18c6a0bec751 : SUCCESS in 5m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/81fc10b335294d099bb29f1f1d528ccd : SUCCESS in 5m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5773cdc082a8418a99950a464fea103c : SUCCESS in 4m 49s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/0277c04cad244628ba586bfc390c58c7 : SUCCESS in 7m 47s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/95bbec4211e845718263db37be15a5e1 : SUCCESS in 5m 54s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/5b630576b0b5423b9f11bef9dc58dfef : SUCCESS in 6m 01s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0b7c73d4f8324e248c9b4c740a2bf109 : SUCCESS in 15m 39s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/4c6ccc37aaaa46c2a702596fb6b12efe : SUCCESS in 13m 50s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/8e2315d4d9194f9da9a6dae3c78c9d81 : FAILURE in 8m 34s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/145f8a597d304680b4f4f05401ed3701 : SUCCESS in 16m 58s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8c982ce356ec4b17bcb864c2624b767d : SUCCESS in 16m 30s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/a08d39829a1842448b079f00c0b92a56 : SUCCESS in 15m 32s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/5b82197e1d154950b6bad69b21d3ed24 : SUCCESS in 14m 24s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2aa98a483d06486abd24f7eb938bd92a : SUCCESS in 20m 31s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/7fddcba3d5304424b65a6ffa113bad30 : SUCCESS in 13m 59s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/80a2e2cf44c54c528108b42ea3a89b4d : SUCCESS in 17m 42s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/9cfad6cebc74481d92e49b67267fd4bf : SUCCESS in 1h 04m 37s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/9a1e4ab40119467f9f35ea04144621a8 : SUCCESS in 1h 04m 55s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/8a33853cec5a41eb9a40d53f652b2762 : SUCCESS in 1h 02m 33s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/74e29bf38cf641958c700219a9014a6d : SUCCESS in 1h 07m 50s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/793ad2ff37b34772ba58494fc58df487 : SUCCESS in 1h 05m 36s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/6815395f0257440a95788f98dd05072e : SUCCESS in 1h 38m 14s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/d94ae0baadb546a18a75cf5594d85f8a : SUCCESS in 1h 34m 09s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/222ba30e37df4552b1201e3115c61127 : SUCCESS in 1h 38m 38s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/eaf69657ae974c6db9c4c3dc8b1cedc4 : SUCCESS in 1h 41m 54s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/0fb8fd3f344440a285d3b3a2dccf33cc : SUCCESS in 25m 20s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/9a314faa56e04cde94624265aabdc2e9 : SUCCESS in 28m 42s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4e20df986d054688878141c776deb0ad : SUCCESS in 24m 49s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/bc678f0ec8574087b49ea40766a8ca4a : SUCCESS in 25m 34s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/a4bb31a7a44f4cf0b5ade2f6b402815e : SUCCESS in 36m 24s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/6b99d59de0134f3bb9844f1c4a7347e0 : SUCCESS in 39m 38s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0047a79ec57d4146a79e46da5e631c23 : SUCCESS in 33m 10s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0472cdbf840b48e79d198bde1cc68e5b : SUCCESS in 31m 39s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/81197e8c3de14ed68c229c4b71ed3d30 : SUCCESS in 26m 45s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c0140d18273b49feb677ab3dfec21903 : SUCCESS in 23m 10s","accounts_in_message":[],"_revision_number":2},{"id":"60c6402d11330d25e63db86a67bbbe734965623f","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-07-22 13:10:48.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"47f62855043225295816ca0eaaf556e5e4bfb05a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-22 15:14:31.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/63af6e613f3743738823ff3065a6d125\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8dd019aa79447caa4306d2274bcc06f : SUCCESS in 9m 23s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/6b32c8fa54984680a89224e34f373f52 : SUCCESS in 8m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/658f87308c2a4f76a68f024ee3067fee : SUCCESS in 4m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ecfef8818b684fd58712de15aee16d3c : SUCCESS in 4m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/683a819f3f9f4f53949f04727b830a52 : SUCCESS in 3m 44s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/2e9a67cc4fb04ca79c84d2683d3d15e6 : SUCCESS in 7m 39s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a1dbf9c322df420997d8142e3ef0f9f9 : SUCCESS in 6m 13s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/cc23e22e325a48eb9dacd8ba3bb5da8c : SUCCESS in 7m 02s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e95ac2fe43f5453699b538a38b4026ec : SUCCESS in 14m 24s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/af9b51189ace4e01a27d6a370cd73e25 : SUCCESS in 13m 18s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/ab4e9e5a9fa24d3791cb65671978442d : SUCCESS in 17m 26s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/63b48da6ad8247e0a1568531b7fdd703 : SUCCESS in 24m 05s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/26d7ddbe592947fab3a7398b44d5a663 : SUCCESS in 19m 21s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/cc74bcf52982430d93f9df0fcacc206c : SUCCESS in 17m 16s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/2d4bb09c7cda44c4af3e0c9a221e862b : SUCCESS in 23m 21s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8818b69c2a2949fd88203ee77e586cfc : SUCCESS in 19m 27s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/7716831eeb5b4125be709ff373f87c84 : SUCCESS in 25m 52s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/389174ff7ae0420cb456391a8e347eaa : SUCCESS in 21m 14s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/d74d599cc38345c498dffa608c25ad2b : SUCCESS in 1h 05m 07s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/dc96b03727314dee9bb885c2bca52760 : SUCCESS in 1h 09m 28s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/e3e8dc0688ea498db3439d5e1d8a3b35 : SUCCESS in 56m 54s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8acb49616da94da4930faf9ce365b577 : SUCCESS in 1h 14m 00s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/497c089ddd43497088eb9fc4dd9579bd : SUCCESS in 1h 09m 29s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/0a35356b1bcb458ebfe86dabdbc5fefb : SUCCESS in 1h 42m 35s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/79f6ae7dcb4c47a98e3033658741f683 : SUCCESS in 1h 35m 23s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/03abae6166fc48f9afc7bce46c0ccbe0 : SUCCESS in 1h 43m 17s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4c3ae02b4e814daa8995369efecbcbcf : SUCCESS in 1h 45m 15s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/73903013429c43a9a14b8444b56c91f5 : SUCCESS in 35m 44s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/2452ea353d2f4a44b9d324ded9c4f43b : SUCCESS in 29m 51s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/03a1b2763c8a41a582cda61672444d48 : SUCCESS in 25m 06s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/df95cf87e87a473ca00807575f4d25bf : SUCCESS in 25m 17s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/5e5160fdc75c4a03930475aa7d499828 : SUCCESS in 35m 03s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/dfb6c5038b324be0b1453e140af04b03 : SUCCESS in 40m 18s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bd8426289e9440fb9ef85d8c43cde4a9 : SUCCESS in 33m 40s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1c4d51a258bc4133ac8220def7bb21ae : SUCCESS in 34m 54s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/48501fba1ac94a739c1ebb7536ac2957 : SUCCESS in 22m 39s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a935974f738444218e2f54608835aaa3 : SUCCESS in 23m 18s","accounts_in_message":[],"_revision_number":2},{"id":"dcef7cd1f0dfc6da1e8a9a2f97b94e907145b8ab","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-07-22 15:20:35.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"907edc066921d86d993fcba26250f729490f0640","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-07-22 16:13:45.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"4ca1fdad737a895cc0b69e07d860fa26a762d662","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-22 16:14:08.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"90aa579be5fcbf622301681293773b5bc964ec47","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-22 19:37:10.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/070beee9597d490bbd5c85c994781884\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e98e3c1567a243fba0e0519da60af40e : SUCCESS in 10m 24s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/8d4ecbb33b0d4a5197952e952f0e08ed : SUCCESS in 7m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b0f4e3fec4f484d998c4c563ff7b122 : SUCCESS in 4m 45s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/94e8bbbf10364c25b837b744c1803b9f : SUCCESS in 3m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/81a04917f11c439c9d6bf976d030f0cd : SUCCESS in 7m 11s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a4a22dc7ceec4263b99ca07a3c842727 : SUCCESS in 5m 57s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5a2a6549c2044c6eb71f6524116daf5c : SUCCESS in 12m 50s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/669f4b273ed14c17bed1318aa0aff360 : SUCCESS in 13m 55s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/0ec882c3fa574dfb96b64cc87b2787e3 : SUCCESS in 22m 16s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/83666c9041de47769d5f5f010d713148 : SUCCESS in 17m 33s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/48b2f28909124cfebd2dd7ddbd3e5288 : SUCCESS in 22m 17s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/69af421ce1584930836a32541cd5c87a : SUCCESS in 20m 36s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/3f0eae14e1794faaac971418272ac92a : SUCCESS in 25m 16s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/103e193fc148439fa69b13f2192226cd : SUCCESS in 17m 52s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/73647040f41e4f2ab049eedb4ec0b791 : SUCCESS in 1h 03m 28s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/58d6cb17d3e24668b9aee511a2ad63e8 : SUCCESS in 1h 06m 18s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/7d3d0ca61ef44b5eb3c32c1b8ccc96bc : SUCCESS in 1h 03m 12s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3988cf8ff18c44ad9b7868ccbd9e810c : SUCCESS in 59m 08s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/fe7197752f6440638c8e16cf8d8a982a : SUCCESS in 1h 07m 34s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/1f57edff3ec24361a8d4b99c839a5dee : SUCCESS in 1h 33m 43s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/f5bab696e8e34799bcc39697f6440d4a : SUCCESS in 1h 36m 35s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e9bb17b5401c4ebf94ff04bf8a47a826 : SUCCESS in 1h 32m 51s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d0fc44d20e574dac86ab108059311b03 : SUCCESS in 1h 35m 16s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/88d8b679b890423cb8fac2074ad12d3d : SUCCESS in 29m 22s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/c2e21d6b180b4147a143e1aacc654392 : SUCCESS in 22m 18s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f96c8bd97bb54dabafaf6a33a3567f63 : SUCCESS in 22m 28s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/96ce760a28c34da2923def29fa76106a : SUCCESS in 22m 06s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/ffa8b1dae85c4e4e91790ddeff509653 : SUCCESS in 29m 41s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/4cb0566cb6ec41c8bc327414cc7bc110 : SUCCESS in 37m 38s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d146f588deab4448b4f25dfbf6f40e29 : SUCCESS in 32m 55s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f3cbea690d91442cae084a10ac3c0b7a : SUCCESS in 31m 45s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/99864abb33474ced8af729271d7c027f : SUCCESS in 22m 00s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/009c83848d1e4d9d8941bf3badb1cad5 : SUCCESS in 20m 18s","accounts_in_message":[],"_revision_number":2},{"id":"e088c414081cd1ba860267fda13a4f79a559c83e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-22 19:37:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"339b6269f85cf25c424a606ec70221959d39dab4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-22 19:38:10.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/da8ab745b3b14fcdac0d12e69431a712\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd6d58ce786e4a24a867d26016ccbaca : SUCCESS in 49s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"762731bcc897976bc630f7c8bf42ced2db07514d","revisions":{"68515e355d67129566f6ba4b1887016375021e4b":{"kind":"REWORK","_number":1,"created":"2026-07-22 09:52:32.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/87/998287/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/87/998287/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/87/998287/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/87/998287/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/87/998287/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/87/998287/1"}}},"commit":{"parents":[{"commit":"2f4d55f3d42caec8d5d863a659415556ba8e48af","subject":"Fix kolla inventory hostname formatting","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2f4d55f3d42caec8d5d863a659415556ba8e48af"}]}],"author":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2026-07-17 12:15:53.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-07-22 09:52:32.000000000","tz":0},"subject":"Remove leftover device mapper references from the code","message":"Remove leftover device mapper references from the code\n\nCloses-Bug: #2096784\nChange-Id: I3717723a9213617272e9ae97664d4ad8febc0aab\nSigned-off-by: Rafal Lewandowski \u003crafal@stackhpc.com\u003e\n(cherry picked from commit 45d22bf0e45ab76945700c7c6da15bf66193e0bf)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/68515e355d67129566f6ba4b1887016375021e4b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/68515e355d67129566f6ba4b1887016375021e4b"}]},"branch":"refs/heads/stable/2026.1","conflicts":{"base":"7688d5c4b99ebf6979236b53e52964afd680915f","ours":"2f4d55f3d42caec8d5d863a659415556ba8e48af","theirs":"45d22bf0e45ab76945700c7c6da15bf66193e0bf","merge_strategy":"recursive","contains_conflicts":false}},"762731bcc897976bc630f7c8bf42ced2db07514d":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-07-22 10:14:02.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/87/998287/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/87/998287/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/87/998287/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/87/998287/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/87/998287/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/87/998287/2"}}},"commit":{"parents":[{"commit":"739ca7b9ea2cab8de7bc27d667ff48d8a51040fa","subject":"CI: Install kernel-modules-extra","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/739ca7b9ea2cab8de7bc27d667ff48d8a51040fa"}]}],"author":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2026-07-17 12:15:53.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-07-22 10:14:02.000000000","tz":0},"subject":"Remove leftover device mapper references from the code","message":"Remove leftover device mapper references from the code\n\nCloses-Bug: #2096784\nChange-Id: I3717723a9213617272e9ae97664d4ad8febc0aab\nSigned-off-by: Rafal Lewandowski \u003crafal@stackhpc.com\u003e\n(cherry picked from commit 45d22bf0e45ab76945700c7c6da15bf66193e0bf)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/762731bcc897976bc630f7c8bf42ced2db07514d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/762731bcc897976bc630f7c8bf42ced2db07514d"}]},"branch":"refs/heads/stable/2026.1","description":"Rebase","conflicts":{"base":"2f4d55f3d42caec8d5d863a659415556ba8e48af","ours":"68515e355d67129566f6ba4b1887016375021e4b","theirs":"739ca7b9ea2cab8de7bc27d667ff48d8a51040fa","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
