)]}'
{"id":"openstack%2Fkayobe~998886","triplet_id":"openstack%2Fkayobe~stable%2F2026.1~I8e75b5e9b56586ced9c0a9d88bb80a8535397f50","project":"openstack/kayobe","branch":"stable/2026.1","topic":"tenks-redfish-ci-stable/2026.1","attention_set":{},"removed_from_attention_set":{"39213":{"account":{"_account_id":39213,"name":"Jed Preist","display_name":"Jed Preist","email":"jed@stackhpc.com","username":"jed-preist"},"last_update":"2026-07-31 17:15:11.000000000","reason":"\u003cGERRIT_ACCOUNT_39213\u003e replied on the change","reason_account":{"_account_id":39213,"name":"Jed Preist","display_name":"Jed Preist","email":"jed@stackhpc.com","username":"jed-preist"}}},"hashtags":[],"change_id":"I8e75b5e9b56586ced9c0a9d88bb80a8535397f50","subject":"CI: Emulate Redfish baremetal with Tenks","status":"MERGED","created":"2026-07-28 09:05:13.000000000","updated":"2026-07-31 21:01:04.000000000","submitted":"2026-07-31 21:00:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"998886-tenks-redfish-ci-stable/2026.1","cherry_pick_of_change":996221,"cherry_pick_of_patch_set":13,"meta_rev_id":"1949e6582128f50b8182684441164eeb7469860d","_number":998886,"virtual_id_number":998886,"owner":{"_account_id":39213,"name":"Jed Preist","display_name":"Jed Preist","email":"jed@stackhpc.com","username":"jed-preist"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-07-31 21:00:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2026-07-31 13:47:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2026-07-28 10:01:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2026-07-31 13:47:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}],"CC":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-28 10:01:24.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2026-07-28 10:49:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-07-30 09:15:42.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"},{"updated":"2026-07-31 13:47:52.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"}],"messages":[{"id":"39313fa1371a282b34bb4b26be5c9b04624c0718","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39213,"name":"Jed Preist","display_name":"Jed Preist","email":"jed@stackhpc.com","username":"jed-preist"},"date":"2026-07-28 09:05:13.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"d88cc549a84eb84c73d4ba60120006b6c65a5fb4","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-07-28 10:01:24.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"6886378459703114cfb261bb5b9185ab93a2e68e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-28 10:49:48.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cc5510836d904a13a545ba442f80924c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d8d0f9683a8486faa7a79a11af19dd0 : SUCCESS in 9m 20s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/7d0e1ce863fc4be89a757eb9a832bcff : SUCCESS in 5m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b7526fadade443f7aa31e89e13596652 : SUCCESS in 2m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3cb3c13256f0436e8092ba2a13e7ca05 : SUCCESS in 4m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e34a473b0e034394bc26c927dd8f4b42 : SUCCESS in 3m 04s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/74bb70f2ba6541e9bd77f52d2a62464b : SUCCESS in 6m 57s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8b01bda1baa747a5b8aadf4690587ba2 : SUCCESS in 3m 06s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/46cebed3cf734594a03d42f91cc5fef3 : SUCCESS in 4m 26s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/017ec3b9bef241cbaeaad35b821d7dd6 : SUCCESS in 7m 21s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/54713580e7c24dac8cdae6443223994e : SUCCESS in 7m 12s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/0c8caf9dd7f5454c836e27d10783d428 : SUCCESS in 23m 54s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/9c4d352bf6724bc9b16e4a75cd7661a9 : SUCCESS in 22m 40s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/219a6824c6b14715abe34bfe8972e2b0 : SUCCESS in 18m 31s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/9273250768db40a1a5d424d6c6b58b39 : SUCCESS in 18m 26s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/ddcea1431f9041b0afcb37649231dd6f : SUCCESS in 25m 26s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d7154b57256a4ddca4bb96f224f527cc : SUCCESS in 13m 00s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/8156215462034b68a1ac166369e32a84 : RETRY_LIMIT in 4m 41s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c211b9a879cc401eabea3cde9372fb16 : SUCCESS in 12m 37s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/170fa2e4c23f4a4a83f975e39ea371f7 : RETRY_LIMIT in 6m 07s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/cf10dd5ff4be450bb67b8dfa6f5c3686 : RETRY_LIMIT in 7m 00s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/d010b86dc16f46bebb7f1ba19a1f43cc : RETRY_LIMIT in 5m 45s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/672eae8b6c51464b8b7dc394eb073ce1 : SUCCESS in 1h 17m 14s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/c81ba0c04cfd4217902b112d68f8b719 : SUCCESS in 1h 18m 27s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/c4ef8ce6d9da4c20bac2b0d7421e8b7d : FAILURE in 6m 56s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/94de8e369f0742eebabc08f888a14b76 : FAILURE in 7m 28s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1ee7e93d4aae4e329131f245060f19a2 : SUCCESS in 1h 38m 53s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/616b7f35c4da45958b017b0173c0abd6 : SUCCESS in 1h 42m 56s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/a6f9dc8b1473461f8df435129c97c744 : SUCCESS in 28m 11s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/380ef3cc520d454fa2d7b11d042d3f3b : SUCCESS in 28m 57s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e0be0904b1584d97b280f05057aff530 : SUCCESS in 14m 30s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/0eafccf98a424831942072ed2a43cb55 : SUCCESS in 14m 20s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/525f7613cbb94d238132cc73d7222056 : SUCCESS in 18m 56s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/23b89f30fa164fe687637e00b6512ca7 : SUCCESS in 19m 45s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/55f2628ba9b04b67826e094f1866556b : SUCCESS in 33m 46s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0f5a9683562e48989bf4ffcbafded5f3 : SUCCESS in 24m 31s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/cc802cc18c7143b5afeef7cc30c23d07 : SUCCESS in 23m 12s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bc880d4092ed44f6b9ec11b9fb79e8b4 : SUCCESS in 19m 56s","accounts_in_message":[],"_revision_number":1},{"id":"5d25b2a7f531bedf124583507cc882b8fd828e02","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-07-30 09:15:42.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f714bd53f4958ff98893add3040c283dbb3caef1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-30 11:18:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f5ef182f5b4a4f6b82dab9787b254fcc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e4588d5b04e407b9aaed93463d9767e : SUCCESS in 9m 26s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/54ebf89c66304bc69f6fe97e04cb7f20 : SUCCESS in 8m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ffd0d165b6144411a01659567430e744 : SUCCESS in 5m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a71cb1a6cae4985b74811e95b12fb07 : SUCCESS in 3m 34s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2b6689ed4976416dacdecf24d2b0b8d3 : SUCCESS in 4m 36s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a608822cb03b4898b6ea1cd2f2b76b0e : SUCCESS in 6m 48s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/118ab296dcfc4af4bd82494c4d7ba649 : SUCCESS in 6m 06s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2d7cec896140404e802d53e170998f54 : SUCCESS in 6m 29s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/7f94477ab7604e78a3b9f71915e5ca07 : SUCCESS in 14m 58s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2857d17d72b24cdb9701510f7d345bca : SUCCESS in 16m 32s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/7f1f508a4bad400cb77f9b3bcd65fa6e : SUCCESS in 14m 00s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/aecba86a49344bdab5b2118e1d1751a6 : SUCCESS in 24m 35s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ab50eca3ec9f4be5a62a824fe5ec86d3 : SUCCESS in 20m 46s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/104626070ecc4e25a09ac8bb069a0eaa : SUCCESS in 18m 26s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/6b6e77b4547c4b03a46786b962d2a6f5 : SUCCESS in 23m 22s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0cca10553c5049c5a186d0adb1d95a9a : SUCCESS in 23m 11s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/c9bf9c1b1b5e4206910beea274e61d7f : SUCCESS in 13m 33s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b346f9a73582419f975bf9417e8d5f99 : SUCCESS in 13m 45s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/0554cb7a90fa41c9b8bd7c9c81a306f9 : SUCCESS in 1h 13m 26s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/e2bb4946f83a47d9b0c36b8c4d53b469 : SUCCESS in 1h 16m 39s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/8113a2cc975547c68319ce85db0d6143 : SUCCESS in 1h 12m 08s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4cbc7e26a07b412fa8776fd2b3a5770c : SUCCESS in 1h 21m 34s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/18bcf534f3a941a1b1c885b1c5d56865 : SUCCESS in 1h 21m 28s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/0f174135743744bb8647f6e31fe3804d : SUCCESS in 1h 39m 53s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/d0811d43d0f3418488d71c8775886030 : SUCCESS in 1h 45m 33s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7236f5a58f7a47208d750094c1a0dc15 : SUCCESS in 1h 49m 54s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c62528f068d348afb647334fd3cbe7e6 : SUCCESS in 1h 53m 54s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/0263ed9eb51d489cab26ae5ba05fc39b : SUCCESS in 26m 17s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/e2759c51575d4fca92594c2abb8b69c3 : SUCCESS in 17m 26s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/219745708ac245c8bb722ac4a49e34a4 : SUCCESS in 25m 09s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/93b51adb39784bf19cb36b76ef17aaf5 : SUCCESS in 22m 33s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/d55cf581353c43ab822366e591378500 : SUCCESS in 28m 45s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/f6524b4c26c34b6ab94a20ac05093cb3 : SUCCESS in 20m 27s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1a1e073aa6ee454599f8ff4138ab2ffa : SUCCESS in 35m 48s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b72ebdd2e9114cab92cd871fe61455c9 : SUCCESS in 34m 08s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/67ce250735dc435a91cd38d57bf86fbc : SUCCESS in 23m 44s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7d34eddc18e245be98ccce28b32cd195 : SUCCESS in 17m 43s","accounts_in_message":[],"_revision_number":1},{"id":"e43a776e80c85684980d7ec3b828fd0edbedd3f6","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-07-31 13:47:52.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9e3426feefa0517553e14c55318d3d551d45e345","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-31 13:48:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0e0d6f8e310c3ab06d1a3bceac00b333f4cfe97b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-31 15:53:11.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e27f2a9ed51e4740833eeded1ae7ff2e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8b52e63a0514db192a64561fdab2527 : SUCCESS in 9m 57s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/800f3464af0444d28a1cdd1a30e6ba6b : SUCCESS in 8m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6a9a0b26175d488b919435d30ba4ec46 : SUCCESS in 4m 39s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fa8c69d1a5a54c1c981296722b1c8340 : SUCCESS in 4m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a9e723a9745e47a9a5b11d73d2f801a7 : SUCCESS in 6m 08s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b289835fdb364ec6b5610fd181c7dfd9 : SUCCESS in 6m 46s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1e4dfca500ff4d3c9265aeaeba84db45 : SUCCESS in 14m 19s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7d3034a7c69a43508f519321221e1867 : SUCCESS in 13m 36s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/e51e019df7624c7a8bf57d591a9fcfa8 : SUCCESS in 12m 10s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b361c487b783429eb2ecccfa6640bff0 : SUCCESS in 16m 23s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/9ed77c5e8ccf4df48a9cb826fcf08d09 : SUCCESS in 19m 41s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/43eecb74a0224e54a93c15b11ae8094d : SUCCESS in 18m 31s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/b9ba0b8aa8f545edb2b1eb4e8a3163ac : SUCCESS in 23m 03s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1030f1ce97b541f198d1413121b654bc : SUCCESS in 21m 54s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/5fbcc8d733ed4b58897eb9c37f7fef80 : SUCCESS in 1h 23m 08s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/6ac26ca672754fce87351dd85dc013c7 : SUCCESS in 1h 20m 38s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/c75bf0e0ef9b43a99af328cf18626eec : SUCCESS in 1h 26m 39s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/edd86c58f9a9470eb9f5f20d3fd2b6e5 : SUCCESS in 1h 31m 43s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/08fcacc405de46ab8f74ac9c4c9d3ffd : SUCCESS in 1h 19m 30s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/7c237fb7731b48bbb6c32e2ef68177ad : SUCCESS in 1h 54m 30s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/57a86a6044374b6fabf6e417cb5148ac : SUCCESS in 1h 34m 32s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9228d342382242e59b380fa086561703 : SUCCESS in 1h 40m 38s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9e1a5c181dfe4ef69959723f47adb6a0 : SUCCESS in 2h 03m 50s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/648e913eddcd4dd1aa19d4bab84dbb59 : SUCCESS in 16m 05s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/eb16396d2d5843e6afcc0ca8726c39fd : POST_FAILURE in 17m 33s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b250b5d4946c4fa389e7852e993663d3 : SUCCESS in 23m 59s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/3d53e005a5924af5ba078fede40a9aaa : SUCCESS in 21m 03s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/c5f446d1c29445a8b3217f4e857cda5d : SUCCESS in 38m 30s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/fe1c0f567b36406383658f87d9a544e5 : SUCCESS in 22m 30s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/14d9359db6f24269bbea05069381aac6 : SUCCESS in 31m 35s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/af224f3f55d94a4aa5b979233351ce5c : SUCCESS in 42m 05s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/15e7f6f4ea024133a5cfa83d0b4d5695 : SUCCESS in 19m 01s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8c18215ecc1741b1bd0b18a55c3ec591 : SUCCESS in 21m 23s","accounts_in_message":[],"_revision_number":1},{"id":"7bd4c284fe0e2a01f99420cd3185e1f3db436371","author":{"_account_id":39213,"name":"Jed Preist","display_name":"Jed Preist","email":"jed@stackhpc.com","username":"jed-preist"},"date":"2026-07-31 17:15:11.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"acd45b607eb3fa031e09c5d44858271b81e6ed2e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-31 19:05:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5a80bb5dbc264f44aec7728c2439c701\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/728c716f9fb943929a6cc541fc3c38e7 : SUCCESS in 11m 19s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/c1c2fc4949d54842be3fbbd98fbb87a8 : SUCCESS in 4m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b2e9fe4da5a145bcbdbf18431e1bb8cf : SUCCESS in 4m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/76f67630d1d34e6bb7362abb835fb8bb : SUCCESS in 2m 55s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e9ba510cdfa54693b58aad7db624d46f : SUCCESS in 6m 08s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/0c2a4120ba5d49b78f0055d15cb7e9e1 : SUCCESS in 4m 32s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5ac482c36b394e96b3eb827a39fd3b34 : SUCCESS in 6m 30s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c2eb9e4698864ef0bc9b5780efc83f1f : SUCCESS in 6m 07s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6658934e51614bf5bbbae87f1ab50ca2 : SUCCESS in 8m 36s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/8bfb7d73bcdb4bc9b929a4db7e35d6e0 : SUCCESS in 6m 54s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/df3ef4e5714a4b0b884f11c7b4f9c96c : SUCCESS in 12m 23s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/f1a09a46d357400fb57ae7765e31ecdb : SUCCESS in 11m 09s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/00fa3adec0a143aaadc16f19c2202031 : SUCCESS in 12m 31s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/c225e65471254ea9a201e9cb9c35c572 : SUCCESS in 12m 26s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/2ad12d7078b14bcfa0b6f40cb2e9ff75 : SUCCESS in 18m 27s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/178caa8e4b3a42289198cc9e355b5b97 : SUCCESS in 13m 22s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/7aa11958ec4341a0a945eac96af3faa6 : SUCCESS in 11m 25s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5d9f9ff8ca444607862a0f2cc2ef35c1 : SUCCESS in 17m 30s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/00d25c804b7a4add9888aa73b2e10e49 : SUCCESS in 1h 16m 40s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/f87834c3afe348bf807140bfa9f1cb17 : SUCCESS in 1h 16m 29s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/5c76fb4c21894d45956b9b337cf1ec87 : SUCCESS in 1h 15m 03s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bedd48be143746079ef827cc209704fa : SUCCESS in 1h 17m 59s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/9ba62dfd6ed448bb94a048b496328742 : SUCCESS in 1h 18m 19s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/7a64a4be414148d9a41f49feea081aa6 : SUCCESS in 1h 45m 12s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/9cf8a71c6d224cd9ae574a078dba8c1c : SUCCESS in 1h 42m 28s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7da0e9b740874b0391d5485f60c5b998 : SUCCESS in 1h 41m 03s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/60b36978cc3545dc9a0c3a541dad509e : SUCCESS in 1h 48m 59s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/448619282f1f4399990e8bcf0ab834a7 : SUCCESS in 23m 04s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/6acb0067d56a4020b86708b28592fd55 : SUCCESS in 15m 17s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/936618123c3347d0a076707905933016 : SUCCESS in 21m 54s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/38cb99fbe5a3404a8b9dff16ce711dd7 : SUCCESS in 28m 32s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/032610b0a9084d36a4c6a66ac8afbd8a : SUCCESS in 17m 52s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/abeaa036eafd4a01b3a39e76c24fc215 : SUCCESS in 33m 13s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/542b846131a04a43915d5f037f617199 : SUCCESS in 32m 17s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bff46208cf57467292277c70c4efea1c : SUCCESS in 29m 55s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/8b3bacae38344ccdb706cb170333bb89 : SUCCESS in 15m 12s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4b979a2812c34ef98a54a19d1281238d : SUCCESS in 21m 05s","accounts_in_message":[],"_revision_number":1},{"id":"8c6f0d4594dafbeb22d290da7f369981aa9e548c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-31 19:05:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"79f85def65f452fc7f97a7ef4b0800dca85d0798","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-31 21:00:08.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/01ac29cdb8794df4a23ef9cba13c31a1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ba9c1cc0f14485f910fb6c310e17883 : SUCCESS in 9m 11s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/1f38795765994367af13d83ab5b7445f : SUCCESS in 7m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4fc07a9924de4f059280e565140f40f4 : SUCCESS in 4m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4eca884be9824ae087a19d980a55a2af : SUCCESS in 2m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f61ca288b4de48a18095ed462bd9f25e : SUCCESS in 6m 03s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/5f1d1042f1554cfca8e4d849d5172ad2 : SUCCESS in 5m 25s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/7413122714ea405d9d78eee4e9e6474a : SUCCESS in 12m 11s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/362b02a761a14f0e97a1f1879b814731 : SUCCESS in 13m 41s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/0119f81d42524a15a35836950e0e87f1 : SUCCESS in 23m 42s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a2d561299e3e4b558255c7cf9bf69e79 : SUCCESS in 16m 34s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/1c55367a81b34851b3bbcf79dbf57125 : SUCCESS in 14m 02s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ff990c7b5ae74de5bbb87e22eb31e3b8 : SUCCESS in 18m 01s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/760a17cd1a6e4e288f1f66e95343e657 : SUCCESS in 26m 00s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0cac8fe0cb1f46089991af9097c42cc3 : SUCCESS in 11m 06s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/30fda7a5cb644bb4820dae55ec618573 : SUCCESS in 1h 20m 10s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/4b60c2a5b5a64a2f9647a29428b88f1d : SUCCESS in 1h 15m 11s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/20a3694f781a4881b654d13ba9d7a7a9 : SUCCESS in 1h 09m 57s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d0a22e57a3e348aaaac7789dbbb40dea : SUCCESS in 1h 25m 55s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/8312ab68543c4a90a5e769c898860c79 : SUCCESS in 1h 24m 31s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/8b3a609c025641ad8e3252ce6445222f : SUCCESS in 1h 40m 50s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/e2db7dec2bb147bfb627dac6e15ac358 : SUCCESS in 1h 38m 03s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/50602aae904d4d129c009b3489da2b2c : SUCCESS in 1h 43m 00s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3fe283fb0c36410aa6c4863536419bda : SUCCESS in 1h 53m 43s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/1125d201e48f4bfab075d0991a879e01 : SUCCESS in 15m 30s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/c079b7349ac644f5a4eafefd9f8b1035 : SUCCESS in 22m 13s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5e9636447cd64eb18a64fb647eb03fe6 : SUCCESS in 15m 01s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/06868d4c0a0a476cb24c8e01919fc30b : SUCCESS in 23m 28s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/a000aa46145c42a5b96f0b19d1bbdca9 : SUCCESS in 18m 37s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/7b395f72294b493b8482ae969eb6e952 : SUCCESS in 43m 20s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a592760d98b04ebd9a65dedcbf68b6fc : SUCCESS in 34m 04s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d3d70a2b95c94fc5b394b3ef587e66ce : SUCCESS in 33m 10s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/274525eac4b0406a83d736e0c6b0078e : SUCCESS in 15m 10s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/74c697d636a44e4f85e17ef749c856c3 : SUCCESS in 16m 05s","accounts_in_message":[],"_revision_number":1},{"id":"860c9e961149e68f9d213f5e5b5e37eb8228a1b8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-31 21:00:09.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"1949e6582128f50b8182684441164eeb7469860d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-31 21:01:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/99bc6205f52f4c3aba2f67c992099f02\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/820a00e5cf84442480a3b98bb5ff6ba5 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8b57ee4e916a47dbaf53394b5ba8c2494b9eb1c1","revisions":{"8b57ee4e916a47dbaf53394b5ba8c2494b9eb1c1":{"kind":"REWORK","_number":1,"created":"2026-07-28 09:05:13.000000000","uploader":{"_account_id":39213,"name":"Jed Preist","display_name":"Jed Preist","email":"jed@stackhpc.com","username":"jed-preist"},"ref":"refs/changes/86/998886/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/86/998886/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/998886/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/998886/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/998886/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/86/998886/1"}}},"commit":{"parents":[{"commit":"cbb0fb8958739b364ba9a8874aceedc637feeb1d","subject":"Adds seed IP to default no-proxy list","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cbb0fb8958739b364ba9a8874aceedc637feeb1d"}]}],"author":{"name":"Jed Preist","email":"jed@stackhpc.com","date":"2026-07-07 08:21:04.000000000","tz":60},"committer":{"name":"Jed Preist","email":"jed@stackhpc.com","date":"2026-07-28 09:05:13.000000000","tz":0},"subject":"CI: Emulate Redfish baremetal with Tenks","message":"CI: Emulate Redfish baremetal with Tenks\n\nEmulate baremetal using Tenks to test the deployment of Redfish\nbaremetal with Ironic.\n\nDepends-On: https://review.opendev.org/c/openstack/tenks/+/995697\nChange-Id: I8e75b5e9b56586ced9c0a9d88bb80a8535397f50\nSigned-off-by: Jed Preist \u003cjed@stackhpc.com\u003e\n(cherry picked from commit b8bb880f769025e60fa2a794b13ef58b38a940db)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8b57ee4e916a47dbaf53394b5ba8c2494b9eb1c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8b57ee4e916a47dbaf53394b5ba8c2494b9eb1c1"}]},"branch":"refs/heads/stable/2026.1","conflicts":{"base":"e26af03e1877fce209b6f2b3c0eac23ab671000e","ours":"cbb0fb8958739b364ba9a8874aceedc637feeb1d","theirs":"b8bb880f769025e60fa2a794b13ef58b38a940db","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
