)]}'
{"id":"openstack%2Fkayobe~999968","triplet_id":"openstack%2Fkayobe~stable%2F2025.2~I35827eeb132fc7cd75a85aa0b33367d0a56dd33f","project":"openstack/kayobe","branch":"stable/2025.2","attention_set":{},"removed_from_attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2026-08-26 14:10:14.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I35827eeb132fc7cd75a85aa0b33367d0a56dd33f","subject":"Fix Bifrost extra_kernel_params validation","status":"MERGED","created":"2026-08-06 12:43:48.000000000","updated":"2026-08-26 14:11:38.000000000","submitted":"2026-08-26 14:10:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"999968","cherry_pick_of_change":999605,"cherry_pick_of_patch_set":6,"meta_rev_id":"1781bfcff3455f2b93c5df1ad42478b41eb74fdb","_number":999968,"virtual_id_number":999968,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-08-26 14:10:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2026-08-26 12:14:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2026-08-12 12:30:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-08-14 07:46:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2026-08-26 12:14:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"reviewers":{"REVIEWER":[{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-08-06 12:43:48.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"},{"updated":"2026-08-06 14:23:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-08-12 11:02:58.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"CC"},{"updated":"2026-08-14 07:46:12.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2026-08-24 15:01:22.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2026-08-24 15:01:22.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2026-08-24 15:01:22.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2026-08-24 15:01:22.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2026-08-24 15:01:22.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"state":"REVIEWER"},{"updated":"2026-08-24 15:01:22.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2026-08-24 15:01:22.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"}],"messages":[{"id":"d355716b7eac6a6f88dc4b219e7200244bf97d3d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-08-06 12:43:48.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"53e7e2513fcdfb9fa69be362302abd0395c24f21","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-06 14:23:57.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2a199ad66a9441e8b4e8997851e320e6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7d16ce73ba34b879cf26631162cbd06 : SUCCESS in 9m 25s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/d4a74b34d7894755923ae9ded110addd : FAILURE in 4m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aef36939be16401883965d6695c5a0a3 : SUCCESS in 4m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/23eb986dc1f9476a9f5a3ec1361c2106 : SUCCESS in 5m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4da3d7efbb6841e2a8616334dcdb642c : SUCCESS in 5m 34s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a3d4673c2b514e029e41e6c0a83274d7 : SUCCESS in 6m 57s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b3673d0d9f234e018d041059d7cc46fe : SUCCESS in 6m 10s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a3ee174dad584b38a1bc71e7a32671e6 : SUCCESS in 5m 43s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/7566a61b9fef41aca14aed501dd5764f : SUCCESS in 15m 24s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a36b0eb0a4a9490baf1521a2def6b0c6 : SUCCESS in 11m 35s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/f45d4fb1c68f48479ddbcd8f6f37af68 : SUCCESS in 11m 02s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a9a8e7dcdc9444aba251f5c7b190fe48 : SUCCESS in 13m 53s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/646a1f6924a54e88af2d57ba8a528602 : SUCCESS in 22m 50s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f0fa6933f3ab459c98fa9ffcee8e0677 : SUCCESS in 18m 56s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/c37bcff444eb4593b83d9da24b1320bc : SUCCESS in 1h 11m 11s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/75f53f5eaf3349429a54a4ec30b24855 : SUCCESS in 1h 08m 40s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/28309e24f62a4ec59c20066918b4f72d : FAILURE in 45m 43s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/366ef147131f4fd09c99cfdecb9d23c1 : SUCCESS in 1h 18m 03s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/24e4caaaef5e4d9ab35b74e1f3c56484 : SUCCESS in 1h 16m 11s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/6397b7b5d8194855bffbaf89fdd1ef2f : SUCCESS in 1h 34m 32s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9ef425496be647af8c5d7725e050186f : SUCCESS in 1h 38m 30s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/9d769acc8c2642d6adfaf3b4fdc6e2e8 : SUCCESS in 14m 32s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/57589e80df524616b9355bcc7eb0ba54 : SUCCESS in 14m 42s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1c3d7204e2164cc5aa3714cb1209280b : SUCCESS in 19m 48s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/c98da3a9df584064bf7405f6bc66fff7 : FAILURE in 16m 01s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/828ca5fb1a534d89bad05eb3757fe280 : FAILURE in 8m 59s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3e820fa2a3fa4e65983f416c23fde2bd : SUCCESS in 40m 27s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/557e9506ea944a958f760ca6bbacc62c : SUCCESS in 27m 08s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4dc136e0422746d690fa3cf98ff12969 : SUCCESS in 17m 06s","accounts_in_message":[],"_revision_number":1},{"id":"0a970de75b0288efa2ab02a8a2ea47bb85741f7d","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-08-12 11:02:58.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"89257a960c400e05c2ac65ddc3cd0fc7246be18f","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-08-12 12:30:34.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"dd2d2efc0b18a1df75dca0c42d30addac420e988","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-12 12:51:54.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d62c92051d48435dbcab0ce748823524\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/054e8f4a3b2f45eb907748813f0ea1a8 : SUCCESS in 8m 55s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/730b5d93d7cb4894ad54ce59691aa90e : SUCCESS in 6m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0a6e3771e25f4dad8565cba9234dbbc9 : SUCCESS in 4m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d65c0f473714dd2b553b8290daed519 : SUCCESS in 4m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f49f3b53065140fcb1f60401633c7451 : SUCCESS in 3m 21s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0937e98d68314f04a182ee775b7f560f : SUCCESS in 6m 58s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/36c1427b30844c90a5f7dfba18c22f0d : SUCCESS in 5m 51s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6c77fce3054245a7ad4e24621760d3f9 : SUCCESS in 5m 21s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/21f6ee6320774c9b867ac5f75df9f856 : SUCCESS in 13m 31s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/0d8bae6632934ff6aa0e757162dcf6d2 : SUCCESS in 12m 42s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/31a20ad72d0a49519e220c8440b6a88f : SUCCESS in 20m 17s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b370a507fb8c44e3b33eaaf08ac718ba : SUCCESS in 9m 22s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/87f2c91d62354cbaab441bde7747ad59 : SUCCESS in 17m 58s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fb1c7ed985044a0f8886c1fa6b12f168 : SUCCESS in 17m 06s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/e5660085e93046f98430e5a9422441e1 : SUCCESS in 1h 05m 54s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/2d324ed5f6514c25b3052da7d0194fcf : SUCCESS in 1h 09m 58s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/aa9f4ffec94e4505a23f728e9d8cddd1 : SUCCESS in 1h 05m 59s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a6c0686b58f04a729e5d992873976359 : SUCCESS in 1h 15m 02s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/70eb444ed0914cbc8756115df49c5679 : SUCCESS in 1h 17m 13s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/7ed8cfecfb9445de848483356529d644 : SUCCESS in 1h 40m 00s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8a32ce5062444ed4bba8495e011cc4d2 : SUCCESS in 1h 47m 40s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/d5514af41c0a40ba9f7cadcf59efa68e : SUCCESS in 24m 17s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/683b77268bf04dd19179f820f86f64b5 : SUCCESS in 18m 38s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2a2d4062f7c04c05987bc61b6f2c86ea : SUCCESS in 19m 26s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/8182f871bb4648959920097e0c0c5977 : SUCCESS in 10m 28s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/31662ba3fbc34c2fae8ffc6865ed1699 : SUCCESS in 36m 45s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/49a8a15993164196b61fcbe6f250b8db : FAILURE in 16m 55s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/c470926c333a419ba714d580fba04860 : SUCCESS in 19m 56s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3a731f90eb684420be459e3ee4e42647 : SUCCESS in 16m 18s","accounts_in_message":[],"_revision_number":1},{"id":"41dbb82842d3add16dd7975bbe9ec3f0174a06f9","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-08-13 10:37:24.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bc4219d1da3bb79be75edfb89bf2eda9fe854b9d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-13 12:11:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/96adc7c1343d4ecda775cf99e040dbdf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2afb17280d704c5a973d977b8bd82342 : SUCCESS in 5m 18s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/bf40960293254d3183690d2c33ad4092 : SUCCESS in 6m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0b9092de3776400cbf5f06f401d801c8 : SUCCESS in 5m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/95e609e26b694b57925f76b6b29dc79c : SUCCESS in 6m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/17496adc1bab456fa2aceb9e4d2dcaab : SUCCESS in 3m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b4d02447d149401092064ffffc11da46 : SUCCESS in 7m 21s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aae13017f99941cd9981593be626dc9f : SUCCESS in 6m 23s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d1d40b32f5b94450986a11138c2c7079 : SUCCESS in 6m 39s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/96ea3847a6354e19a5776aee765ff882 : SUCCESS in 14m 09s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ac9a0d11f26f4f5197877ac908a6b98d : SUCCESS in 11m 04s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/1aae42017d434d7db93206853e8d1d26 : SUCCESS in 10m 15s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8ed818b2d4234eb58998bc04b37599c1 : SUCCESS in 15m 21s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/e357128c616740adbfa38c2011bf1501 : SUCCESS in 10m 39s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3f9c9e8892cb4c3f9b5343924751ff47 : SUCCESS in 16m 38s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/8398567355ac417cbb23bd6c89234865 : SUCCESS in 1h 07m 12s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/d986b67e2d974f0fa7fec501da722372 : SUCCESS in 1h 03m 26s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/4b235be5c5174a4f88d22d8be51e4004 : SUCCESS in 1h 05m 00s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c6b5003786c64aa6ba38762ee5f9ebdc : SUCCESS in 1h 10m 11s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/c4911d63b2de4c6fb60a754ab87ab8d6 : SUCCESS in 1h 11m 30s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/47cbde5f0a984f4a99f5dd17b2615508 : SUCCESS in 1h 28m 51s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f9718aae10014bab8ccf52a31fd7b826 : SUCCESS in 1h 32m 37s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/a86713403d2c410bbd9ad1b854d42148 : SUCCESS in 12m 09s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/69c39c87271f41daae0a514e5bb1b1c4 : SUCCESS in 23m 18s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f4bd00ad8f8a4837a6db98a666f9aec7 : SUCCESS in 18m 10s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/c5be9187d20742bcbd39f2c5ff467c9c : SUCCESS in 12m 18s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/79f019d10bf24a22b919f7dc9f6d500f : SUCCESS in 36m 41s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a30b344b9cf54bd18cafa88f45e3b8bd : SUCCESS in 28m 14s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/98accb84a31e45f7b2abcc3a2f039bb3 : SUCCESS in 10m 17s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6091a8facf3d48729b6e7d79cd7feecc : SUCCESS in 15m 04s","accounts_in_message":[],"_revision_number":1},{"id":"7579895e621717ed4cb677e121a815105b4bc75f","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-08-14 07:46:12.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"9c73f4e9055971c126a80463b12dd402dab95f34","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-08-26 12:14:36.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1b7ae2ba071c937951ca82eb89c125f97f164c9a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 12:15:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"67b8042ac61b51efe712f897762e634729ea7c93","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 14:10:14.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9a099043c22641f0bf7fcb49dce96940\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8de6f49e83a942538628cb819d306a31 : SUCCESS in 7m 11s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/054074f8a46c4542b60aec2311a19da3 : SUCCESS in 7m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/897c069d696e4a7885c29df33317f246 : SUCCESS in 2m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/23c084cc122f4e71b8c9cf877e6d3b16 : SUCCESS in 3m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8a56f4c0837d41129155230375d416ef : SUCCESS in 6m 31s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b120203fb0ef4ea39061a8d18b8326af : SUCCESS in 5m 37s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f4e4f457e0544ba28798e3b4fe217f23 : SUCCESS in 7m 29s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a478ceb8467449ee9b932af4cde89b35 : SUCCESS in 7m 24s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/1d08a4eb8b0649f8beb71d0af16fb75c : SUCCESS in 10m 44s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6afde5bb1546473da42078089f7b483a : SUCCESS in 8m 29s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/53d675938a824985bbabe73050b9128f : SUCCESS in 16m 07s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4e4319e31c6144df981961c3c9c591ce : SUCCESS in 19m 31s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/4b39a08c07804f29b56a49d8d94cf75e : SUCCESS in 1h 12m 58s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/8cf5f548099445dcba156326bf3c733b : SUCCESS in 1h 16m 39s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/db4e9b8fb358473f8149bf37a897cd2d : SUCCESS in 1h 15m 13s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a146f52513134c39ba496776c1d884c9 : SUCCESS in 1h 13m 58s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/24002b31b9124800bae6d3e6da2db1b1 : SUCCESS in 1h 17m 32s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/81e31f2c161c414a8d603a7e7dc83214 : SUCCESS in 1h 39m 22s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f2b3af1b0bb342e9bf100187a8a32385 : SUCCESS in 1h 45m 00s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/c2db7bdd806e44b89a1fdab8f9435d02 : SUCCESS in 16m 35s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/970cfe7453b44dd383b392aa0427d223 : SUCCESS in 11m 24s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/71a6781e8c334b24b37a973cc3d34159 : SUCCESS in 22m 14s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/4669248ee50540bba391dbd596154e2c : SUCCESS in 16m 19s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/a28cc404c3ee4ec9a49776a32c7f54a5 : SUCCESS in 20m 05s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/38d935cffbf04a348004acedc94003d0 : SUCCESS in 31m 26s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/074d26820f3b4907ae92dc0b50ea5fe2 : SUCCESS in 19m 58s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f0ddc52651b4485d8579db912215f6d2 : SUCCESS in 15m 28s","accounts_in_message":[],"_revision_number":1},{"id":"d5d41e1fb937c062554ad191ec0422cde63d1538","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 14:10:14.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"1781bfcff3455f2b93c5df1ad42478b41eb74fdb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 14:11:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/301d16232c604756ba5605f89c68fc17\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8b177fb01244d68a287ff081cae2378 : SUCCESS in 1m 00s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b67a42d213cf4ef59dfc99380c4f0114 : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b8443a896feccd2ab526569a31c7705ac9f3b1de","revisions":{"b8443a896feccd2ab526569a31c7705ac9f3b1de":{"kind":"REWORK","_number":1,"created":"2026-08-06 12:43:48.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/68/999968/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/68/999968/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/68/999968/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/68/999968/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/68/999968/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/68/999968/1"}}},"commit":{"parents":[{"commit":"74dbf8fcfc1d5e2651d5834ef8feb96ab7f23dce","subject":"Merge \"Fix error when failing to retrieve inspection rule\" into stable/2025.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/74dbf8fcfc1d5e2651d5834ef8feb96ab7f23dce"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-08-04 09:11:55.000000000","tz":120},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2026-08-06 12:43:48.000000000","tz":0},"subject":"Fix Bifrost extra_kernel_params validation","message":"Fix Bifrost extra_kernel_params validation\n\nIronic recently added parsing of extra_kernel_params options [1].\nKayobe was generating \"ipa-inspection-benchmarks\u003d\" when extra-hardware\nwas disabled, which failed to validate.\n\nUpdate ipa_kernel_options_default to avoid producing kernel parameters\nwith no value.\n\n[1] https://review.opendev.org/c/openstack/ironic/+/992306\n\nCloses-Bug: #2162755\nChange-Id: I35827eeb132fc7cd75a85aa0b33367d0a56dd33f\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n(cherry picked from commit 8aae85bb9016579f1240a056e6525ffdc5b19df1)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b8443a896feccd2ab526569a31c7705ac9f3b1de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b8443a896feccd2ab526569a31c7705ac9f3b1de"}]},"branch":"refs/heads/stable/2025.2","conflicts":{"base":"6f67922715c0e430f39b84a19fb4427611ac574b","ours":"74dbf8fcfc1d5e2651d5834ef8feb96ab7f23dce","theirs":"8aae85bb9016579f1240a056e6525ffdc5b19df1","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
