)]}'
{"id":"openstack%2Fkeystone-specs~373983","triplet_id":"openstack%2Fkeystone-specs~master~I97f7a34d398ba673d7733fb2aaa490ebc9298afd","project":"openstack/keystone-specs","branch":"master","topic":"bp/improved-oidc-support","hashtags":[],"change_id":"I97f7a34d398ba673d7733fb2aaa490ebc9298afd","subject":"OpenID Connect improved support","status":"MERGED","created":"2016-09-21 09:58:56.000000000","updated":"2020-01-02 15:09:50.000000000","submitted":"2020-01-02 15:08:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":42,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"373983-1577977707175-46a14a57","meta_rev_id":"a7d630e096868084b075f0997669520fc89f48f7","_number":373983,"virtual_id_number":373983,"owner":{"_account_id":91,"name":"Alvaro","email":"aloga@ifca.unican.es","username":"aloga"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":0,"_account_id":18338,"name":"Ron De Rose","email":"ronald.de.rose@intel.com","username":"derosenet"},{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":8978,"name":"Marek Denis","email":"marek.denis+openstack@gmail.com","username":"marek-denis"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":17860,"name":"Samuel de Medeiros Queiroz","email":"samueldmq@gmail.com","username":"samueldmq"},{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":7191,"name":"Jamie Lennox","email":"jamielennox@gmail.com","username":"jamielennox"},{"value":0,"_account_id":8119,"name":"Eric Brown","email":"eric_wade_brown@yahoo.com","username":"ericwb"},{"value":0,"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},{"value":0,"_account_id":91,"name":"Alvaro","email":"aloga@ifca.unican.es","username":"aloga"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-02 15:08:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"all":[{"value":2,"date":"2019-12-17 20:25:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":0,"_account_id":18338,"name":"Ron De Rose","email":"ronald.de.rose@intel.com","username":"derosenet"},{"value":2,"date":"2019-12-10 18:34:51.000000000","_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":8978,"name":"Marek Denis","email":"marek.denis+openstack@gmail.com","username":"marek-denis"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":17860,"name":"Samuel de Medeiros Queiroz","email":"samueldmq@gmail.com","username":"samueldmq"},{"value":2,"date":"2020-01-02 14:58:47.000000000","_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":7191,"name":"Jamie Lennox","email":"jamielennox@gmail.com","username":"jamielennox"},{"value":0,"_account_id":8119,"name":"Eric Brown","email":"eric_wade_brown@yahoo.com","username":"ericwb"},{"value":0,"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},{"value":0,"_account_id":91,"name":"Alvaro","email":"aloga@ifca.unican.es","username":"aloga"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"all":[{"value":0,"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":0,"_account_id":18338,"name":"Ron De Rose","email":"ronald.de.rose@intel.com","username":"derosenet"},{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":8978,"name":"Marek Denis","email":"marek.denis+openstack@gmail.com","username":"marek-denis"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":17860,"name":"Samuel de Medeiros Queiroz","email":"samueldmq@gmail.com","username":"samueldmq"},{"value":1,"date":"2020-01-02 14:58:47.000000000","_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":7191,"name":"Jamie Lennox","email":"jamielennox@gmail.com","username":"jamielennox"},{"value":0,"_account_id":8119,"name":"Eric Brown","email":"eric_wade_brown@yahoo.com","username":"ericwb"},{"value":0,"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},{"value":0,"_account_id":91,"name":"Alvaro","email":"aloga@ifca.unican.es","username":"aloga"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":91,"name":"Alvaro","email":"aloga@ifca.unican.es","username":"aloga"},{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":7191,"name":"Jamie Lennox","email":"jamielennox@gmail.com","username":"jamielennox"},{"_account_id":8119,"name":"Eric Brown","email":"eric_wade_brown@yahoo.com","username":"ericwb"},{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"_account_id":8978,"name":"Marek Denis","email":"marek.denis+openstack@gmail.com","username":"marek-denis"},{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"_account_id":17860,"name":"Samuel de Medeiros Queiroz","email":"samueldmq@gmail.com","username":"samueldmq"},{"_account_id":18338,"name":"Ron De Rose","email":"ronald.de.rose@intel.com","username":"derosenet"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2016-09-28 17:01:35.000000000","updated_by":{"_account_id":8978,"name":"Marek Denis","email":"marek.denis+openstack@gmail.com","username":"marek-denis"},"reviewer":{"_account_id":8978,"name":"Marek Denis","email":"marek.denis+openstack@gmail.com","username":"marek-denis"},"state":"REVIEWER"},{"updated":"2016-10-17 15:01:56.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"},{"updated":"2016-11-01 15:23:03.000000000","updated_by":{"_account_id":7191,"name":"Jamie Lennox","email":"jamielennox@gmail.com","username":"jamielennox"},"reviewer":{"_account_id":7191,"name":"Jamie Lennox","email":"jamielennox@gmail.com","username":"jamielennox"},"state":"REVIEWER"},{"updated":"2016-11-15 19:01:18.000000000","updated_by":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"reviewer":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"state":"REVIEWER"},{"updated":"2016-11-16 01:51:14.000000000","updated_by":{"_account_id":8119,"name":"Eric Brown","email":"eric_wade_brown@yahoo.com","username":"ericwb"},"reviewer":{"_account_id":8119,"name":"Eric Brown","email":"eric_wade_brown@yahoo.com","username":"ericwb"},"state":"REVIEWER"},{"updated":"2016-11-16 14:15:02.000000000","updated_by":{"_account_id":17860,"name":"Samuel de Medeiros Queiroz","email":"samueldmq@gmail.com","username":"samueldmq"},"reviewer":{"_account_id":17860,"name":"Samuel de Medeiros Queiroz","email":"samueldmq@gmail.com","username":"samueldmq"},"state":"REVIEWER"},{"updated":"2017-02-28 18:16:53.000000000","updated_by":{"_account_id":18338,"name":"Ron De Rose","email":"ronald.de.rose@intel.com","username":"derosenet"},"reviewer":{"_account_id":18338,"name":"Ron De Rose","email":"ronald.de.rose@intel.com","username":"derosenet"},"state":"REVIEWER"},{"updated":"2019-12-10 18:34:51.000000000","updated_by":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"reviewer":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"state":"REVIEWER"},{"updated":"2019-12-17 20:25:35.000000000","updated_by":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"reviewer":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"state":"REVIEWER"},{"updated":"2020-01-02 14:58:47.000000000","updated_by":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"reviewer":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"state":"REVIEWER"},{"updated":"2020-01-02 15:08:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6b507e429346b633295f389754fd8bce7b094227","author":{"_account_id":91,"name":"Alvaro","email":"aloga@ifca.unican.es","username":"aloga"},"date":"2016-09-21 09:58:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c2d1b665500fe050bab08724e3d6eadbec168230","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-09-21 10:05:02.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-keystone-specs-docs-ubuntu-xenial http://logs.openstack.org/83/373983/1/check/gate-keystone-specs-docs-ubuntu-xenial/1dbeeeb/ : FAILURE in 3m 04s\n- gate-keystone-specs-python27-ubuntu-xenial http://logs.openstack.org/83/373983/1/check/gate-keystone-specs-python27-ubuntu-xenial/aba2537/ : FAILURE in 3m 08s","accounts_in_message":[],"_revision_number":1},{"id":"9545ceb8247612c0938dc785ff2563bea25bb234","author":{"_account_id":91,"name":"Alvaro","email":"aloga@ifca.unican.es","username":"aloga"},"date":"2016-09-21 10:16:33.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d58dc503c85522f839e7116315a93dd918ec4ecb","author":{"_account_id":91,"name":"Alvaro","email":"aloga@ifca.unican.es","username":"aloga"},"date":"2016-09-21 10:17:22.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"09fb409fc66545860c6d9c3223b2f5c3de5d0b01","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-09-21 10:20:28.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-keystone-specs-docs-ubuntu-xenial http://docs-draft.openstack.org/83/373983/3/check/gate-keystone-specs-docs-ubuntu-xenial/88698c2//doc/build/html/ : SUCCESS in 2m 09s\n- gate-keystone-specs-python27-ubuntu-xenial http://logs.openstack.org/83/373983/3/check/gate-keystone-specs-python27-ubuntu-xenial/9530d80/ : SUCCESS in 2m 31s","accounts_in_message":[],"_revision_number":3},{"id":"7692f0d5ca0895a91783c7c8b6bbf83ad6d73f24","author":{"_account_id":6482,"name":"Steve Martinelli","email":"s.martinelli@gmail.com","username":"stevemar"},"date":"2016-10-08 07:32:31.000000000","message":"Patch Set 3:\n\n(10 comments)\n\nstill reviewing, leaving comments so they don\u0027t get lost","accounts_in_message":[],"_revision_number":3},{"id":"cefc5c20581a4ecc7cffbb6755a6b452c4822424","author":{"_account_id":91,"name":"Alvaro","email":"aloga@ifca.unican.es","username":"aloga"},"date":"2016-10-17 14:54:17.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"240f3346b7a548316b3896f5b12165b5bd24d125","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-10-17 15:01:56.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-keystone-specs-docs-ubuntu-xenial http://docs-draft.openstack.org/83/373983/4/check/gate-keystone-specs-docs-ubuntu-xenial/28f35a1//doc/build/html/ : SUCCESS in 2m 38s\n- gate-keystone-specs-python27-ubuntu-xenial http://logs.openstack.org/83/373983/4/check/gate-keystone-specs-python27-ubuntu-xenial/8de90df/ : SUCCESS in 4m 49s","accounts_in_message":[],"_revision_number":4},{"id":"e0a3dab4ea0d86f43519e4318d26a8c5a5276fef","author":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"date":"2016-11-15 19:01:18.000000000","message":"Patch Set 4: Code-Review-1\n\nLike the general idea, although I suspect that the right place to do this work is in the Apache module, not in the application specific code.  Won\u0027t hold it up if this is the only way to make it happen","accounts_in_message":[],"_revision_number":4},{"id":"0cbf7d8689c2e89d86942a550dab53acf9a714e8","author":{"_account_id":8119,"name":"Eric Brown","email":"eric_wade_brown@yahoo.com","username":"ericwb"},"date":"2016-11-16 01:51:14.000000000","message":"Patch Set 4: Code-Review-1\n\n(20 comments)","accounts_in_message":[],"_revision_number":4},{"id":"3695a09b46ebd1b90e30122ecde9b082497a0ea6","author":{"_account_id":17860,"name":"Samuel de Medeiros Queiroz","email":"samueldmq@gmail.com","username":"samueldmq"},"date":"2016-11-16 14:15:02.000000000","message":"Patch Set 4: Code-Review-1\n\n(2 comments)\n\nThis should be done in the same way as adding native support for SAML2 in keystone, please see the spec [1] and talk to David Stanek if you need more info.\n\n[1] https://review.openstack.org/#/c/397860/","accounts_in_message":[],"_revision_number":4},{"id":"4052fc05ccf8822ccaef61ef57b99fbf72e522e0","author":{"_account_id":17860,"name":"Samuel de Medeiros Queiroz","email":"samueldmq@gmail.com","username":"samueldmq"},"date":"2016-11-16 14:18:45.000000000","message":"Patch Set 4:\n\nThe idea looks good overall, just fix the typos found by Eric. It should be good to go (the spec -\u003e the concept) other than that.","accounts_in_message":[],"_revision_number":4},{"id":"7c32b0a360ee8c0e8fd491a4490550b2c03bfbd7","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2016-11-22 15:59:16.000000000","message":"Patch Set 4: Code-Review-1\n\n(7 comments)\n\nHi Alvaro,\n\nI have a couple suggestions inline. I\u0027m looking forward to discussing this more!","accounts_in_message":[],"_revision_number":4},{"id":"b9c65cad32429717cb5f2eb849d1c7e6fcd20bce","author":{"_account_id":91,"name":"Alvaro","email":"aloga@ifca.unican.es","username":"aloga"},"date":"2016-11-29 22:05:21.000000000","message":"Patch Set 4:\n\nHi all.\n\nThanks for the input, I\u0027ve been offline of this topic for several weeks, but I am catching this up again. I\u0027ll reply to the coments during this week.\n\nThanks again!","accounts_in_message":[],"_revision_number":4},{"id":"2b46c2344282e857e8985c8fe0c29da7919a07f7","author":{"_account_id":6482,"name":"Steve Martinelli","email":"s.martinelli@gmail.com","username":"stevemar"},"date":"2016-12-16 02:09:18.000000000","message":"Patch Set 4: Code-Review-2\n\nprocedural -2 since we are at spec freeze","accounts_in_message":[],"_revision_number":4},{"id":"e03d6bdc749cf7e4c17aee1a7a3d560a00ddf4c4","author":{"_account_id":6482,"name":"Steve Martinelli","email":"s.martinelli@gmail.com","username":"stevemar"},"date":"2017-02-23 21:23:48.000000000","message":"Patch Set 4: -Code-Review","accounts_in_message":[],"_revision_number":4},{"id":"2229dc1a0fc4234b2a8cfe13c69f43b824f4952e","author":{"_account_id":91,"name":"Alvaro","email":"aloga@ifca.unican.es","username":"aloga"},"date":"2018-03-15 16:39:18.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"0216a081ad6aa10b91f92aeba9ade3a55a713b0c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-03-15 16:51:20.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs http://logs.openstack.org/83/373983/5/check/build-openstack-sphinx-docs/113c286/ : FAILURE in 3m 33s","accounts_in_message":[],"_revision_number":5},{"id":"cc462f6b28f1450ea0a7473d5c8142738280403d","author":{"_account_id":91,"name":"Alvaro","email":"aloga@ifca.unican.es","username":"aloga"},"date":"2018-03-16 10:27:00.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"d220707b277219ede1b6a020ebe8aa13648d8ec0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-03-16 11:12:00.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/83/373983/6/check/build-openstack-sphinx-docs/01b69f5/html/ : SUCCESS in 3m 43s","accounts_in_message":[],"_revision_number":6},{"id":"20c92d67d95e60ee061cf91402b2976e6a1d9225","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-12-10 18:34:51.000000000","message":"Patch Set 6: Code-Review+2\n\n(2 comments)\n\nI think this is fine to merge to the backlog now. The development of the auth plugin is being done as an external plugin, I think when the issues with that (e.g. https://bugs.launchpad.net/keystone/+bug/1854041) have been worked out and we should re-propose this for the V cycle and integrate the plugin in upstream keystone.","accounts_in_message":[],"_revision_number":6},{"id":"4a5464946c72a25f8a2baa5b419c5e4560b0df28","author":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"date":"2019-12-17 20:25:35.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"43cb8272393ca7e284b53e20f451abd6cb01c4c3","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2020-01-02 14:58:47.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1\n\n(1 comment)\n\nNice work setting the context and describing the problem.","accounts_in_message":[],"_revision_number":6},{"id":"9b037829fd3eddda7e45375ca276f4d497cc71f5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-02 14:58:58.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"bf974f3c93ca41d24d28262dd0d8d369a190a3ce","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-02 15:08:26.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/773678debfc040d7ac88385893964167 : SUCCESS in 4m 36s","accounts_in_message":[],"_revision_number":6},{"id":"0f784a1fdb6c3d9e6827722a6d1a524aee2d3f08","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-02 15:08:27.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"1da0c2fe6c09beaa0e32ac7d8e77a064f6aff00e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-02 15:09:50.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-specs https://zuul.opendev.org/t/openstack/build/f10702d771aa4a7cbdf8025787819bb5 : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"1d246348112cc890db5d0d6fe8d091161e05d831","revisions":{"37aa0a70182dddd6bc07a1fda245847f5166bff7":{"kind":"REWORK","_number":1,"created":"2016-09-21 09:58:56.000000000","uploader":{"_account_id":91,"name":"Alvaro","email":"aloga@ifca.unican.es","username":"aloga"},"ref":"refs/changes/83/373983/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone-specs","ref":"refs/changes/83/373983/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/83/373983/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/83/373983/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/83/373983/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone-specs refs/changes/83/373983/1"}}},"commit":{"parents":[{"commit":"32efe5a298ebdc8b892e4c4dd56d003f9d277dd5","subject":"Merge \"Revert spec change for Microversions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/32efe5a298ebdc8b892e4c4dd56d003f9d277dd5"}]}],"author":{"name":"Alvaro Lopez Garcia","email":"aloga@ifca.unican.es","date":"2016-09-21 09:57:29.000000000","tz":120},"committer":{"name":"Alvaro Lopez Garcia","email":"aloga@ifca.unican.es","date":"2016-09-21 09:58:54.000000000","tz":120},"subject":"OpenID Connect improved support","message":"OpenID Connect improved support\n\nOpenID Connect is supported in Keystone by leveraging the Apache\nmod_auth_oidc module and the Keystone Federation plugin. OpenID Connect\nworks fine when accessing OpenStack thorugh the dashboard, but it\nrequires additional configuration steps to make it work when using the\nOpenStack CLI tools.  This blueprint aims at improving the support, so\nthat the same outcome is obtained, regardless of the way the user\naccesses the cloud.\n\nImplements blueprint: oidc-improved-support\n\nChange-Id: I97f7a34d398ba673d7733fb2aaa490ebc9298afd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/37aa0a70182dddd6bc07a1fda245847f5166bff7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/37aa0a70182dddd6bc07a1fda245847f5166bff7"}]},"branch":"refs/heads/master"},"56a30228199d4e8cf4f2f0508c51c4d8ead3dca2":{"kind":"REWORK","_number":2,"created":"2016-09-21 10:16:33.000000000","uploader":{"_account_id":91,"name":"Alvaro","email":"aloga@ifca.unican.es","username":"aloga"},"ref":"refs/changes/83/373983/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone-specs","ref":"refs/changes/83/373983/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/83/373983/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/83/373983/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/83/373983/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone-specs refs/changes/83/373983/2"}}},"commit":{"parents":[{"commit":"32efe5a298ebdc8b892e4c4dd56d003f9d277dd5","subject":"Merge \"Revert spec change for Microversions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/32efe5a298ebdc8b892e4c4dd56d003f9d277dd5"}]}],"author":{"name":"Alvaro Lopez Garcia","email":"aloga@ifca.unican.es","date":"2016-09-21 09:57:29.000000000","tz":120},"committer":{"name":"Alvaro Lopez Garcia","email":"aloga@ifca.unican.es","date":"2016-09-21 10:13:44.000000000","tz":120},"subject":"OpenID Connect improved support","message":"OpenID Connect improved support\n\nOpenID Connect is supported in Keystone by leveraging the Apache\nmod_auth_oidc module and the Keystone Federation plugin. OpenID Connect\nworks fine when accessing OpenStack thorugh the dashboard, but it\nrequires additional configuration steps to make it work when using the\nOpenStack CLI tools.  This blueprint aims at improving the support, so\nthat the same outcome is obtained, regardless of the way the user\naccesses the cloud.\n\nImplements blueprint: oidc-improved-support\n\nChange-Id: I97f7a34d398ba673d7733fb2aaa490ebc9298afd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/56a30228199d4e8cf4f2f0508c51c4d8ead3dca2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/56a30228199d4e8cf4f2f0508c51c4d8ead3dca2"}]},"branch":"refs/heads/master"},"74ba5fa35226955dd6c5b1d28e686a6fe9c2f320":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2016-09-21 10:17:22.000000000","uploader":{"_account_id":91,"name":"Alvaro","email":"aloga@ifca.unican.es","username":"aloga"},"ref":"refs/changes/83/373983/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone-specs","ref":"refs/changes/83/373983/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/83/373983/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/83/373983/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/83/373983/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone-specs refs/changes/83/373983/3"}}},"commit":{"parents":[{"commit":"32efe5a298ebdc8b892e4c4dd56d003f9d277dd5","subject":"Merge \"Revert spec change for Microversions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/32efe5a298ebdc8b892e4c4dd56d003f9d277dd5"}]}],"author":{"name":"Alvaro Lopez Garcia","email":"aloga@ifca.unican.es","date":"2016-09-21 09:57:29.000000000","tz":120},"committer":{"name":"Alvaro Lopez Garcia","email":"aloga@ifca.unican.es","date":"2016-09-21 10:17:04.000000000","tz":120},"subject":"OpenID Connect improved support","message":"OpenID Connect improved support\n\nOpenID Connect is supported in Keystone by leveraging the Apache\nmod_auth_oidc module and the Keystone Federation plugin. OpenID Connect\nworks fine when accessing OpenStack thorugh the dashboard, but it\nrequires additional configuration steps to make it work when using the\nOpenStack CLI tools.  This blueprint aims at improving the support, so\nthat the same outcome is obtained, regardless of the way the user\naccesses the cloud.\n\nImplements blueprint: improved-oidc-support\n\nChange-Id: I97f7a34d398ba673d7733fb2aaa490ebc9298afd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/74ba5fa35226955dd6c5b1d28e686a6fe9c2f320"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/74ba5fa35226955dd6c5b1d28e686a6fe9c2f320"}]},"branch":"refs/heads/master"},"8214762e59ddf226ce906fa6d8d09d0db8036435":{"kind":"REWORK","_number":4,"created":"2016-10-17 14:54:17.000000000","uploader":{"_account_id":91,"name":"Alvaro","email":"aloga@ifca.unican.es","username":"aloga"},"ref":"refs/changes/83/373983/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone-specs","ref":"refs/changes/83/373983/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/83/373983/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/83/373983/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/83/373983/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone-specs refs/changes/83/373983/4"}}},"commit":{"parents":[{"commit":"32efe5a298ebdc8b892e4c4dd56d003f9d277dd5","subject":"Merge \"Revert spec change for Microversions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/32efe5a298ebdc8b892e4c4dd56d003f9d277dd5"}]}],"author":{"name":"Alvaro Lopez Garcia","email":"aloga@ifca.unican.es","date":"2016-09-21 09:57:29.000000000","tz":120},"committer":{"name":"Alvaro Lopez Garcia","email":"aloga@ifca.unican.es","date":"2016-10-17 14:54:09.000000000","tz":120},"subject":"OpenID Connect improved support","message":"OpenID Connect improved support\n\nOpenID Connect is supported in Keystone by leveraging the Apache\nmod_auth_oidc module and the Keystone Federation plugin. OpenID Connect\nworks fine when accessing OpenStack thorugh the dashboard, but it\nrequires additional configuration steps to make it work when using the\nOpenStack CLI tools. This spec aims at improving the support, so\nthat the same outcome is obtained, regardless of the way the user\naccesses the cloud.\n\nImplements blueprint: improved-oidc-support\n\nChange-Id: I97f7a34d398ba673d7733fb2aaa490ebc9298afd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/8214762e59ddf226ce906fa6d8d09d0db8036435"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/8214762e59ddf226ce906fa6d8d09d0db8036435"}]},"branch":"refs/heads/master"},"e61b28821f79d46258a23c72f01addea3b45dfae":{"kind":"REWORK","_number":5,"created":"2018-03-15 16:39:18.000000000","uploader":{"_account_id":91,"name":"Alvaro","email":"aloga@ifca.unican.es","username":"aloga"},"ref":"refs/changes/83/373983/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone-specs","ref":"refs/changes/83/373983/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/83/373983/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/83/373983/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/83/373983/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone-specs refs/changes/83/373983/5"}}},"commit":{"parents":[{"commit":"7364e988541988a711edf1cad4af4046565c4874","subject":"Fix typos in keystone-specs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/7364e988541988a711edf1cad4af4046565c4874"}]}],"author":{"name":"Alvaro Lopez Garcia","email":"aloga@ifca.unican.es","date":"2016-09-21 09:57:29.000000000","tz":120},"committer":{"name":"Alvaro Lopez Garcia","email":"aloga@ifca.unican.es","date":"2018-03-15 16:39:05.000000000","tz":60},"subject":"OpenID Connect improved support","message":"OpenID Connect improved support\n\nOpenID Connect is supported in Keystone by leveraging the Apache\nmod_auth_oidc module and the Keystone Federation plugin. OpenID Connect\nworks fine when accessing OpenStack thorugh the dashboard, but it\nrequires additional configuration steps to make it work when using the\nOpenStack CLI tools. This spec aims at improving the support, so\nthat the same outcome is obtained, regardless of the way the user\naccesses the cloud.\n\nImplements blueprint: improved-oidc-support\n\nChange-Id: I97f7a34d398ba673d7733fb2aaa490ebc9298afd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/e61b28821f79d46258a23c72f01addea3b45dfae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/e61b28821f79d46258a23c72f01addea3b45dfae"}]},"branch":"refs/heads/master"},"1d246348112cc890db5d0d6fe8d091161e05d831":{"kind":"REWORK","_number":6,"created":"2018-03-16 10:27:00.000000000","uploader":{"_account_id":91,"name":"Alvaro","email":"aloga@ifca.unican.es","username":"aloga"},"ref":"refs/changes/83/373983/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone-specs","ref":"refs/changes/83/373983/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/83/373983/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/83/373983/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/83/373983/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone-specs refs/changes/83/373983/6"}}},"commit":{"parents":[{"commit":"7364e988541988a711edf1cad4af4046565c4874","subject":"Fix typos in keystone-specs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/7364e988541988a711edf1cad4af4046565c4874"}]}],"author":{"name":"Alvaro Lopez Garcia","email":"aloga@ifca.unican.es","date":"2016-09-21 09:57:29.000000000","tz":120},"committer":{"name":"Alvaro Lopez Garcia","email":"aloga@ifca.unican.es","date":"2018-03-16 10:20:04.000000000","tz":60},"subject":"OpenID Connect improved support","message":"OpenID Connect improved support\n\nOpenID Connect is supported in Keystone by leveraging the Apache\nmod_auth_oidc module and the Keystone Federation plugin. OpenID Connect\nworks fine when accessing OpenStack thorugh the dashboard, but it\nrequires additional configuration steps to make it work when using the\nOpenStack CLI tools. This spec aims at improving the support, so\nthat the same outcome is obtained, regardless of the way the user\naccesses the cloud.\n\nImplements blueprint: improved-oidc-support\n\nChange-Id: I97f7a34d398ba673d7733fb2aaa490ebc9298afd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/1d246348112cc890db5d0d6fe8d091161e05d831"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/1d246348112cc890db5d0d6fe8d091161e05d831"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
