)]}'
{"id":"openstack%2Fkeystone-specs~391624","triplet_id":"openstack%2Fkeystone-specs~master~I2487fe09091e9b054a91a9853b99f3dc14fbeebb","project":"openstack/keystone-specs","branch":"master","topic":"bp/token-verify-role-check","hashtags":[],"change_id":"I2487fe09091e9b054a91a9853b99f3dc14fbeebb","subject":"Role Check from Middleware","status":"MERGED","created":"2016-10-29 21:28:30.000000000","updated":"2021-09-24 20:29:41.000000000","submitted":"2016-12-13 23:23:19.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":431,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"e014f718e4d418ce21ffb31adf1934ee99d3114d","_number":391624,"virtual_id_number":391624,"owner":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":2,"date":"2016-12-13 23:23:19.000000000","post_submit":true,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":11716,"name":"Tim Kelsey","email":"timothy.kelsey@uk.ibm.com","username":"tkelsey"},{"value":0,"_account_id":964,"name":"Anne Gentle","email":"annegentle@justwriteclick.com","username":"annegentle"},{"value":0,"_account_id":11397,"name":"Doug Chivers","email":"dougchivers@gmail.com","username":"dg"},{"value":0,"_account_id":6534,"name":"David Chadwick","email":"d.w.chadwick@truetrust.co.uk","username":"d-w-chadwick"},{"value":0,"_account_id":9162,"name":"Lana Brindley","email":"openstack@lanabrindley.com","username":"loquacity"},{"value":0,"_account_id":9098,"name":"Nathan Kinder","email":"nkinder@redhat.com","username":"nkinder"},{"value":0,"_account_id":5707,"name":"Henry Nash","email":"henryn@linux.vnet.ibm.com","username":"henry-nash"},{"value":0,"_account_id":11022,"name":"Rodrigo Duarte Sousa","email":"rodrigodsousa@gmail.com","username":"rodrigods"},{"value":0,"date":"2017-01-05 12:30:11.000000000","post_submit":true,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":0,"_account_id":6482,"name":"Steve Martinelli","email":"s.martinelli@gmail.com","username":"stevemar"},{"value":0,"_account_id":9237,"name":"Kevin Fox","email":"kevin@efox.cc","username":"kfox1111"},{"value":0,"_account_id":11029,"name":"Jamie Finnigan","email":"jamiefinnigan@gmail.com","username":"jamiefinnigan"},{"value":0,"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},{"value":0,"date":"2016-12-26 12:07:21.000000000","post_submit":true,"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"value":0,"_account_id":7063,"name":"Robert Clark","email":"hyakuhei@gmail.com","username":"hyakuhei"},{"value":0,"_account_id":12000,"name":"Ian Cordasco","email":"sigmavirus24@gmail.com","username":"sigmavirus24"},{"value":0,"_account_id":17893,"name":"Luke Hinds","email":"lhinds@redhat.com","username":"lhinds"},{"value":0,"_account_id":12325,"name":"Nathaniel Dillon","email":"nathaniel.dillon@alumni.depaul.edu","username":"sicarie"},{"value":0,"_account_id":9382,"name":"Shilla Saebi","email":"shilla.saebi@gmail.com","username":"shillasaebi"},{"value":0,"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":0,"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},{"value":0,"_account_id":7725,"name":"David Stanek","email":"dstanek@dstanek.com","username":"dstanek"},{"value":0,"_account_id":17645,"name":"Shan Guo","email":"guoshan.jolie@gmail.com","username":"guoshan"},{"value":0,"_account_id":10670,"name":"Michael McCune","email":"elmiko@redhat.com","username":"mimccune"},{"value":0,"_account_id":10608,"name":"Matthew Edmonds","email":"edmondsw@us.ibm.com","username":"edmondsw"},{"value":0,"date":"2017-01-03 16:39:22.000000000","post_submit":true,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":12342,"name":"WuKong","email":"rebirthmonkey@gmail.com","username":"RHE"},{"value":0,"_account_id":8119,"name":"Eric Brown","email":"eric_wade_brown@yahoo.com","username":"ericwb"},{"value":0,"_account_id":1528,"name":"Stanislaw Pitucha","email":"viraptor@gmail.com","username":"stanislaw-pitucha"},{"value":0,"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},{"value":0,"_account_id":12542,"name":"Johannes Grassler","email":"jgr-launchpad@btw23.de","username":"jgrassler"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11022,"name":"Rodrigo Duarte Sousa","email":"rodrigodsousa@gmail.com","username":"rodrigods"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":11716,"name":"Tim Kelsey","email":"timothy.kelsey@uk.ibm.com","username":"tkelsey"},{"value":0,"_account_id":964,"name":"Anne Gentle","email":"annegentle@justwriteclick.com","username":"annegentle"},{"value":0,"_account_id":11397,"name":"Doug Chivers","email":"dougchivers@gmail.com","username":"dg"},{"value":0,"_account_id":6534,"name":"David Chadwick","email":"d.w.chadwick@truetrust.co.uk","username":"d-w-chadwick"},{"value":0,"_account_id":9162,"name":"Lana Brindley","email":"openstack@lanabrindley.com","username":"loquacity"},{"value":0,"_account_id":9098,"name":"Nathan Kinder","email":"nkinder@redhat.com","username":"nkinder"},{"value":0,"_account_id":5707,"name":"Henry Nash","email":"henryn@linux.vnet.ibm.com","username":"henry-nash"},{"value":2,"date":"2016-12-13 23:18:24.000000000","_account_id":11022,"name":"Rodrigo Duarte Sousa","email":"rodrigodsousa@gmail.com","username":"rodrigods"},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":2,"date":"2016-12-13 19:00:56.000000000","_account_id":6482,"name":"Steve Martinelli","email":"s.martinelli@gmail.com","username":"stevemar"},{"value":0,"_account_id":9237,"name":"Kevin Fox","email":"kevin@efox.cc","username":"kfox1111"},{"value":0,"_account_id":11029,"name":"Jamie Finnigan","email":"jamiefinnigan@gmail.com","username":"jamiefinnigan"},{"value":2,"date":"2016-12-13 18:20:03.000000000","_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},{"value":0,"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"value":0,"_account_id":7063,"name":"Robert Clark","email":"hyakuhei@gmail.com","username":"hyakuhei"},{"value":0,"_account_id":12000,"name":"Ian Cordasco","email":"sigmavirus24@gmail.com","username":"sigmavirus24"},{"value":0,"_account_id":17893,"name":"Luke Hinds","email":"lhinds@redhat.com","username":"lhinds"},{"value":0,"_account_id":12325,"name":"Nathaniel Dillon","email":"nathaniel.dillon@alumni.depaul.edu","username":"sicarie"},{"value":0,"_account_id":9382,"name":"Shilla Saebi","email":"shilla.saebi@gmail.com","username":"shillasaebi"},{"value":0,"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":0,"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},{"value":0,"_account_id":7725,"name":"David Stanek","email":"dstanek@dstanek.com","username":"dstanek"},{"value":0,"_account_id":17645,"name":"Shan Guo","email":"guoshan.jolie@gmail.com","username":"guoshan"},{"value":0,"_account_id":10670,"name":"Michael McCune","email":"elmiko@redhat.com","username":"mimccune"},{"value":0,"_account_id":10608,"name":"Matthew Edmonds","email":"edmondsw@us.ibm.com","username":"edmondsw"},{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":12342,"name":"WuKong","email":"rebirthmonkey@gmail.com","username":"RHE"},{"value":0,"_account_id":8119,"name":"Eric Brown","email":"eric_wade_brown@yahoo.com","username":"ericwb"},{"value":0,"_account_id":1528,"name":"Stanislaw Pitucha","email":"viraptor@gmail.com","username":"stanislaw-pitucha"},{"value":0,"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},{"value":0,"_account_id":12542,"name":"Johannes Grassler","email":"jgr-launchpad@btw23.de","username":"jgrassler"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11022,"name":"Rodrigo Duarte Sousa","email":"rodrigodsousa@gmail.com","username":"rodrigods"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":11716,"name":"Tim Kelsey","email":"timothy.kelsey@uk.ibm.com","username":"tkelsey"},{"value":0,"_account_id":964,"name":"Anne Gentle","email":"annegentle@justwriteclick.com","username":"annegentle"},{"value":0,"_account_id":11397,"name":"Doug Chivers","email":"dougchivers@gmail.com","username":"dg"},{"value":0,"_account_id":6534,"name":"David Chadwick","email":"d.w.chadwick@truetrust.co.uk","username":"d-w-chadwick"},{"value":0,"_account_id":9162,"name":"Lana Brindley","email":"openstack@lanabrindley.com","username":"loquacity"},{"value":0,"_account_id":9098,"name":"Nathan Kinder","email":"nkinder@redhat.com","username":"nkinder"},{"value":0,"_account_id":5707,"name":"Henry Nash","email":"henryn@linux.vnet.ibm.com","username":"henry-nash"},{"value":1,"date":"2016-12-13 23:18:24.000000000","_account_id":11022,"name":"Rodrigo Duarte Sousa","email":"rodrigodsousa@gmail.com","username":"rodrigods"},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":0,"_account_id":6482,"name":"Steve Martinelli","email":"s.martinelli@gmail.com","username":"stevemar"},{"value":0,"_account_id":9237,"name":"Kevin Fox","email":"kevin@efox.cc","username":"kfox1111"},{"value":0,"_account_id":11029,"name":"Jamie Finnigan","email":"jamiefinnigan@gmail.com","username":"jamiefinnigan"},{"value":0,"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},{"value":0,"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"value":0,"_account_id":7063,"name":"Robert Clark","email":"hyakuhei@gmail.com","username":"hyakuhei"},{"value":0,"_account_id":12000,"name":"Ian Cordasco","email":"sigmavirus24@gmail.com","username":"sigmavirus24"},{"value":0,"_account_id":17893,"name":"Luke Hinds","email":"lhinds@redhat.com","username":"lhinds"},{"value":0,"_account_id":12325,"name":"Nathaniel Dillon","email":"nathaniel.dillon@alumni.depaul.edu","username":"sicarie"},{"value":0,"_account_id":9382,"name":"Shilla Saebi","email":"shilla.saebi@gmail.com","username":"shillasaebi"},{"value":0,"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":0,"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},{"value":0,"_account_id":7725,"name":"David Stanek","email":"dstanek@dstanek.com","username":"dstanek"},{"value":0,"_account_id":17645,"name":"Shan Guo","email":"guoshan.jolie@gmail.com","username":"guoshan"},{"value":0,"_account_id":10670,"name":"Michael McCune","email":"elmiko@redhat.com","username":"mimccune"},{"value":0,"_account_id":10608,"name":"Matthew Edmonds","email":"edmondsw@us.ibm.com","username":"edmondsw"},{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":12342,"name":"WuKong","email":"rebirthmonkey@gmail.com","username":"RHE"},{"value":0,"_account_id":8119,"name":"Eric Brown","email":"eric_wade_brown@yahoo.com","username":"ericwb"},{"value":0,"_account_id":1528,"name":"Stanislaw Pitucha","email":"viraptor@gmail.com","username":"stanislaw-pitucha"},{"value":0,"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},{"value":0,"_account_id":12542,"name":"Johannes Grassler","email":"jgr-launchpad@btw23.de","username":"jgrassler"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"_account_id":964,"name":"Anne Gentle","email":"annegentle@justwriteclick.com","username":"annegentle"},{"_account_id":1528,"name":"Stanislaw Pitucha","email":"viraptor@gmail.com","username":"stanislaw-pitucha"},{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":5707,"name":"Henry Nash","email":"henryn@linux.vnet.ibm.com","username":"henry-nash"},{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"_account_id":6482,"name":"Steve Martinelli","email":"s.martinelli@gmail.com","username":"stevemar"},{"_account_id":6534,"name":"David Chadwick","email":"d.w.chadwick@truetrust.co.uk","username":"d-w-chadwick"},{"_account_id":7063,"name":"Robert Clark","email":"hyakuhei@gmail.com","username":"hyakuhei"},{"_account_id":7725,"name":"David Stanek","email":"dstanek@dstanek.com","username":"dstanek"},{"_account_id":8119,"name":"Eric Brown","email":"eric_wade_brown@yahoo.com","username":"ericwb"},{"_account_id":9098,"name":"Nathan Kinder","email":"nkinder@redhat.com","username":"nkinder"},{"_account_id":9162,"name":"Lana Brindley","email":"openstack@lanabrindley.com","username":"loquacity"},{"_account_id":9237,"name":"Kevin Fox","email":"kevin@efox.cc","username":"kfox1111"},{"_account_id":9382,"name":"Shilla Saebi","email":"shilla.saebi@gmail.com","username":"shillasaebi"},{"_account_id":10608,"name":"Matthew Edmonds","email":"edmondsw@us.ibm.com","username":"edmondsw"},{"_account_id":10670,"name":"Michael McCune","email":"elmiko@redhat.com","username":"mimccune"},{"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},{"_account_id":11022,"name":"Rodrigo Duarte Sousa","email":"rodrigodsousa@gmail.com","username":"rodrigods"},{"_account_id":11029,"name":"Jamie Finnigan","email":"jamiefinnigan@gmail.com","username":"jamiefinnigan"},{"_account_id":11397,"name":"Doug Chivers","email":"dougchivers@gmail.com","username":"dg"},{"_account_id":11716,"name":"Tim Kelsey","email":"timothy.kelsey@uk.ibm.com","username":"tkelsey"},{"_account_id":12000,"name":"Ian Cordasco","email":"sigmavirus24@gmail.com","username":"sigmavirus24"},{"_account_id":12325,"name":"Nathaniel Dillon","email":"nathaniel.dillon@alumni.depaul.edu","username":"sicarie"},{"_account_id":12342,"name":"WuKong","email":"rebirthmonkey@gmail.com","username":"RHE"},{"_account_id":12542,"name":"Johannes Grassler","email":"jgr-launchpad@btw23.de","username":"jgrassler"},{"_account_id":17645,"name":"Shan Guo","email":"guoshan.jolie@gmail.com","username":"guoshan"},{"_account_id":17893,"name":"Luke Hinds","email":"lhinds@redhat.com","username":"lhinds"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2016-11-02 13:54:06.000000000","updated_by":{"_account_id":12342,"name":"WuKong","email":"rebirthmonkey@gmail.com","username":"RHE"},"reviewer":{"_account_id":12342,"name":"WuKong","email":"rebirthmonkey@gmail.com","username":"RHE"},"state":"REVIEWER"},{"updated":"2016-11-03 20:27:24.000000000","updated_by":{"_account_id":9237,"name":"Kevin Fox","email":"kevin@efox.cc","username":"kfox1111"},"reviewer":{"_account_id":9237,"name":"Kevin Fox","email":"kevin@efox.cc","username":"kfox1111"},"state":"REVIEWER"},{"updated":"2016-11-04 02:39:32.000000000","updated_by":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"reviewer":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"state":"REVIEWER"},{"updated":"2016-11-04 02:39:32.000000000","updated_by":{"_account_id":964,"name":"Anne Gentle","email":"annegentle@justwriteclick.com","username":"annegentle"},"reviewer":{"_account_id":964,"name":"Anne Gentle","email":"annegentle@justwriteclick.com","username":"annegentle"},"state":"REVIEWER"},{"updated":"2016-11-04 02:39:32.000000000","updated_by":{"_account_id":1528,"name":"Stanislaw Pitucha","email":"viraptor@gmail.com","username":"stanislaw-pitucha"},"reviewer":{"_account_id":1528,"name":"Stanislaw Pitucha","email":"viraptor@gmail.com","username":"stanislaw-pitucha"},"state":"REVIEWER"},{"updated":"2016-11-04 02:39:32.000000000","updated_by":{"_account_id":7063,"name":"Robert Clark","email":"hyakuhei@gmail.com","username":"hyakuhei"},"reviewer":{"_account_id":7063,"name":"Robert Clark","email":"hyakuhei@gmail.com","username":"hyakuhei"},"state":"REVIEWER"},{"updated":"2016-11-04 02:39:32.000000000","updated_by":{"_account_id":8119,"name":"Eric Brown","email":"eric_wade_brown@yahoo.com","username":"ericwb"},"reviewer":{"_account_id":8119,"name":"Eric Brown","email":"eric_wade_brown@yahoo.com","username":"ericwb"},"state":"REVIEWER"},{"updated":"2016-11-04 02:39:32.000000000","updated_by":{"_account_id":9098,"name":"Nathan Kinder","email":"nkinder@redhat.com","username":"nkinder"},"reviewer":{"_account_id":9098,"name":"Nathan Kinder","email":"nkinder@redhat.com","username":"nkinder"},"state":"REVIEWER"},{"updated":"2016-11-04 02:39:32.000000000","updated_by":{"_account_id":9162,"name":"Lana Brindley","email":"openstack@lanabrindley.com","username":"loquacity"},"reviewer":{"_account_id":9162,"name":"Lana Brindley","email":"openstack@lanabrindley.com","username":"loquacity"},"state":"REVIEWER"},{"updated":"2016-11-04 02:39:32.000000000","updated_by":{"_account_id":9382,"name":"Shilla Saebi","email":"shilla.saebi@gmail.com","username":"shillasaebi"},"reviewer":{"_account_id":9382,"name":"Shilla Saebi","email":"shilla.saebi@gmail.com","username":"shillasaebi"},"state":"REVIEWER"},{"updated":"2016-11-04 02:39:32.000000000","updated_by":{"_account_id":10670,"name":"Michael McCune","email":"elmiko@redhat.com","username":"mimccune"},"reviewer":{"_account_id":10670,"name":"Michael McCune","email":"elmiko@redhat.com","username":"mimccune"},"state":"REVIEWER"},{"updated":"2016-11-04 02:39:32.000000000","updated_by":{"_account_id":11029,"name":"Jamie Finnigan","email":"jamiefinnigan@gmail.com","username":"jamiefinnigan"},"reviewer":{"_account_id":11029,"name":"Jamie Finnigan","email":"jamiefinnigan@gmail.com","username":"jamiefinnigan"},"state":"REVIEWER"},{"updated":"2016-11-04 02:39:32.000000000","updated_by":{"_account_id":11397,"name":"Doug Chivers","email":"dougchivers@gmail.com","username":"dg"},"reviewer":{"_account_id":11397,"name":"Doug Chivers","email":"dougchivers@gmail.com","username":"dg"},"state":"REVIEWER"},{"updated":"2016-11-04 02:39:32.000000000","updated_by":{"_account_id":11716,"name":"Tim Kelsey","email":"timothy.kelsey@uk.ibm.com","username":"tkelsey"},"reviewer":{"_account_id":11716,"name":"Tim Kelsey","email":"timothy.kelsey@uk.ibm.com","username":"tkelsey"},"state":"REVIEWER"},{"updated":"2016-11-04 02:39:32.000000000","updated_by":{"_account_id":12000,"name":"Ian Cordasco","email":"sigmavirus24@gmail.com","username":"sigmavirus24"},"reviewer":{"_account_id":12000,"name":"Ian Cordasco","email":"sigmavirus24@gmail.com","username":"sigmavirus24"},"state":"REVIEWER"},{"updated":"2016-11-04 02:39:32.000000000","updated_by":{"_account_id":12325,"name":"Nathaniel Dillon","email":"nathaniel.dillon@alumni.depaul.edu","username":"sicarie"},"reviewer":{"_account_id":12325,"name":"Nathaniel Dillon","email":"nathaniel.dillon@alumni.depaul.edu","username":"sicarie"},"state":"REVIEWER"},{"updated":"2016-11-04 02:39:32.000000000","updated_by":{"_account_id":17893,"name":"Luke Hinds","email":"lhinds@redhat.com","username":"lhinds"},"reviewer":{"_account_id":17893,"name":"Luke Hinds","email":"lhinds@redhat.com","username":"lhinds"},"state":"REVIEWER"},{"updated":"2016-11-07 13:40:06.000000000","updated_by":{"_account_id":12542,"name":"Johannes Grassler","email":"jgr-launchpad@btw23.de","username":"jgrassler"},"reviewer":{"_account_id":12542,"name":"Johannes Grassler","email":"jgr-launchpad@btw23.de","username":"jgrassler"},"state":"REVIEWER"},{"updated":"2016-11-07 14:20:52.000000000","updated_by":{"_account_id":5707,"name":"Henry Nash","email":"henryn@linux.vnet.ibm.com","username":"henry-nash"},"reviewer":{"_account_id":5707,"name":"Henry Nash","email":"henryn@linux.vnet.ibm.com","username":"henry-nash"},"state":"REVIEWER"},{"updated":"2016-11-09 08:30:09.000000000","updated_by":{"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},"reviewer":{"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},"state":"REVIEWER"},{"updated":"2016-11-10 03:30:53.000000000","updated_by":{"_account_id":17645,"name":"Shan Guo","email":"guoshan.jolie@gmail.com","username":"guoshan"},"reviewer":{"_account_id":17645,"name":"Shan Guo","email":"guoshan.jolie@gmail.com","username":"guoshan"},"state":"REVIEWER"},{"updated":"2016-11-23 15:55:48.000000000","updated_by":{"_account_id":6534,"name":"David Chadwick","email":"d.w.chadwick@truetrust.co.uk","username":"d-w-chadwick"},"reviewer":{"_account_id":6534,"name":"David Chadwick","email":"d.w.chadwick@truetrust.co.uk","username":"d-w-chadwick"},"state":"REVIEWER"},{"updated":"2016-12-08 22:44:30.000000000","updated_by":{"_account_id":10608,"name":"Matthew Edmonds","email":"edmondsw@us.ibm.com","username":"edmondsw"},"reviewer":{"_account_id":10608,"name":"Matthew Edmonds","email":"edmondsw@us.ibm.com","username":"edmondsw"},"state":"REVIEWER"},{"updated":"2016-12-13 00:28:53.000000000","updated_by":{"_account_id":7725,"name":"David Stanek","email":"dstanek@dstanek.com","username":"dstanek"},"reviewer":{"_account_id":7725,"name":"David Stanek","email":"dstanek@dstanek.com","username":"dstanek"},"state":"REVIEWER"},{"updated":"2016-12-13 18:20:03.000000000","updated_by":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"reviewer":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"state":"REVIEWER"},{"updated":"2016-12-13 19:00:56.000000000","updated_by":{"_account_id":6482,"name":"Steve Martinelli","email":"s.martinelli@gmail.com","username":"stevemar"},"reviewer":{"_account_id":6482,"name":"Steve Martinelli","email":"s.martinelli@gmail.com","username":"stevemar"},"state":"REVIEWER"},{"updated":"2016-12-13 23:18:24.000000000","updated_by":{"_account_id":11022,"name":"Rodrigo Duarte Sousa","email":"rodrigodsousa@gmail.com","username":"rodrigods"},"reviewer":{"_account_id":11022,"name":"Rodrigo Duarte Sousa","email":"rodrigodsousa@gmail.com","username":"rodrigods"},"state":"REVIEWER"},{"updated":"2016-12-13 23:23:19.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"},{"updated":"2016-12-26 12:07:21.000000000","updated_by":{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},"reviewer":{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},"state":"REVIEWER"},{"updated":"2017-01-03 16:39:22.000000000","updated_by":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"reviewer":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"state":"REVIEWER"},{"updated":"2017-01-05 12:30:11.000000000","updated_by":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"reviewer":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"state":"REVIEWER"}],"messages":[{"id":"982d71dd6ee78313ec1515adcf5a931df81e4f30","author":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"date":"2016-10-29 21:28:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"891f010c543856e22ba6da826bde116c764267d5","author":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"date":"2016-10-29 21:30:14.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"b14fb797edbde1b67835546ec9950379e2f01a45","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-10-29 21:30:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-keystone-specs-docs-ubuntu-xenial http://docs-draft.openstack.org/24/391624/1/check/gate-keystone-specs-docs-ubuntu-xenial/2fcba62//doc/build/html/ : SUCCESS in 2m 03s\n- gate-keystone-specs-python27-ubuntu-xenial http://logs.openstack.org/24/391624/1/check/gate-keystone-specs-python27-ubuntu-xenial/bc5336e/ : SUCCESS in 1m 32s","accounts_in_message":[],"_revision_number":1},{"id":"d5166a7743a0e938f7c478255412860bcb3973ba","author":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"date":"2016-10-31 02:37:05.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f9be5e6f477e6c8e740947815e3ad3a17729d6d6","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-10-31 02:39:32.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-keystone-specs-docs-ubuntu-xenial http://docs-draft.openstack.org/24/391624/2/check/gate-keystone-specs-docs-ubuntu-xenial/07dc508//doc/build/html/ : SUCCESS in 2m 16s\n- gate-keystone-specs-python27-ubuntu-xenial http://logs.openstack.org/24/391624/2/check/gate-keystone-specs-python27-ubuntu-xenial/671458d/ : SUCCESS in 1m 32s","accounts_in_message":[],"_revision_number":2},{"id":"a6878031258e0a30b204b49b62803a8d559c5953","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2016-10-31 14:29:51.000000000","message":"Patch Set 2:\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"4304740fa358d6b8d2c4212c674ce4dd323c6697","author":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"date":"2016-10-31 14:39:54.000000000","message":"Patch Set 2:\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"ef9beeea98afa9c9a3bfd415e52b1f7eb68a2e46","author":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"date":"2016-11-01 16:22:58.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"b902dfa51c8a0275a0fb3fa0ab715f5b5dc22b71","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-11-01 16:27:55.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-keystone-specs-docs-ubuntu-xenial http://docs-draft.openstack.org/24/391624/3/check/gate-keystone-specs-docs-ubuntu-xenial/cda329b//doc/build/html/ : SUCCESS in 2m 59s\n- gate-keystone-specs-python27-ubuntu-xenial http://logs.openstack.org/24/391624/3/check/gate-keystone-specs-python27-ubuntu-xenial/4d3a487/ : SUCCESS in 1m 43s","accounts_in_message":[],"_revision_number":3},{"id":"f75f0d822ccd65b0b4099d5562370d71c22a5e95","author":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"date":"2016-11-01 19:32:06.000000000","message":"Patch Set 3:\n\nKnown issues:  does not cover endpoint specific policy.  This can be done as a follow on patch.","accounts_in_message":[],"_revision_number":3},{"id":"80fab18d4657d64b342e5c345f79b44bdd8b6997","author":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"date":"2016-11-02 03:26:55.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"6fd1d95610b6cfda9a9614a4ec6a5ea6f100bcae","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-11-02 03:29:52.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-keystone-specs-docs-ubuntu-xenial http://docs-draft.openstack.org/24/391624/4/check/gate-keystone-specs-docs-ubuntu-xenial/054124e//doc/build/html/ : SUCCESS in 1m 37s\n- gate-keystone-specs-python27-ubuntu-xenial http://logs.openstack.org/24/391624/4/check/gate-keystone-specs-python27-ubuntu-xenial/758e6c0/ : SUCCESS in 1m 39s","accounts_in_message":[],"_revision_number":4},{"id":"907726b5749e4203eb933afba9868865f1c24a63","author":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"date":"2016-11-02 13:40:46.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"61f8ff3c269aaf57be5d0250eac403f8693473ca","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-11-02 13:47:50.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-keystone-specs-docs-ubuntu-xenial http://docs-draft.openstack.org/24/391624/5/check/gate-keystone-specs-docs-ubuntu-xenial/9f4f82c//doc/build/html/ : SUCCESS in 2m 21s\n- gate-keystone-specs-python27-ubuntu-xenial http://logs.openstack.org/24/391624/5/check/gate-keystone-specs-python27-ubuntu-xenial/18d9b5a/ : SUCCESS in 3m 04s","accounts_in_message":[],"_revision_number":5},{"id":"379939deda8a61207c836e7539eecc662cfc7981","author":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"date":"2016-11-03 13:06:12.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"34670c1627f12504e054db5596c4a4bf5eef53ae","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-11-03 13:09:35.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-keystone-specs-docs-ubuntu-xenial http://docs-draft.openstack.org/24/391624/6/check/gate-keystone-specs-docs-ubuntu-xenial/3f3e87c//doc/build/html/ : SUCCESS in 1m 45s\n- gate-keystone-specs-python27-ubuntu-xenial http://logs.openstack.org/24/391624/6/check/gate-keystone-specs-python27-ubuntu-xenial/4a1fdf7/ : SUCCESS in 2m 29s","accounts_in_message":[],"_revision_number":6},{"id":"76689741c67eb453e8f28a81ca922e4853f13e56","author":{"_account_id":9237,"name":"Kevin Fox","email":"kevin@efox.cc","username":"kfox1111"},"date":"2016-11-03 20:27:24.000000000","message":"Patch Set 6:\n\n(3 comments)\n\nGenerally looks very good. A few comments/questions inline.","accounts_in_message":[],"_revision_number":6},{"id":"c5a018f33fd0f28820299ee1357486c4b514e996","author":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"date":"2016-11-04 02:39:12.000000000","message":"Patch Set 6:\n\n(3 comments)\n\nTHanks for the comments.  Some responses inline.\n\nLet me know if you have a solution to dealing with the parsing issue inside of middleware.","accounts_in_message":[],"_revision_number":6},{"id":"75db706707d3abb1ca5e1101a5be7c3cc7c6deba","author":{"_account_id":12542,"name":"Johannes Grassler","email":"jgr-launchpad@btw23.de","username":"jgrassler"},"date":"2016-11-07 13:40:06.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"d49c7420ed28de29dbaca0b67f189fb4739e22ab","author":{"_account_id":5707,"name":"Henry Nash","email":"henryn@linux.vnet.ibm.com","username":"henry-nash"},"date":"2016-11-07 14:20:52.000000000","message":"Patch Set 6: Code-Review+2\n\n(2 comments)\n\nI like this. Well documented. Nice job.","accounts_in_message":[],"_revision_number":6},{"id":"9d2a088df97544603554ee87feb9abaa4f99c759","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2016-11-08 22:48:51.000000000","message":"Patch Set 6: Code-Review-1\n\n(22 comments)\n\nI like the idea of solving this problem, but I have some questions inline. It seems as though we are pulling both the role check and the scope check into keystone.\n\nI\u0027m also concerned about the duplication of policy at the service endpoint and in keystone. If one gets out of sync with the other how are the conflicts resolved? I could be misunderstanding something but it seems as though it might make policy harder to manage for deployers because they now have to manage it in two places (according to how I understand the spec).\n\nI\u0027m looking forward to discussing this more!","accounts_in_message":[],"_revision_number":6},{"id":"22a2705a9b8d1e5b7afa28ff73b780688c116c72","author":{"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},"date":"2016-11-09 08:30:09.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"125a4caa040b2ee223e322c6824d115685eba23d","author":{"_account_id":17645,"name":"Shan Guo","email":"guoshan.jolie@gmail.com","username":"guoshan"},"date":"2016-11-10 07:37:23.000000000","message":"Patch Set 6: Code-Review-1\n\n(5 comments)\n\nSome nits inside.","accounts_in_message":[],"_revision_number":6},{"id":"c73e9ddc1f4cbb05d69c7351c59c6580c309aaa2","author":{"_account_id":10608,"name":"Matthew Edmonds","email":"edmondsw@us.ibm.com","username":"edmondsw"},"date":"2016-11-10 19:22:15.000000000","message":"Patch Set 6: Code-Review-1\n\n(45 comments)","accounts_in_message":[],"_revision_number":6},{"id":"b43a854722ee5e2dd431705820df53428a3b928c","author":{"_account_id":6482,"name":"Steve Martinelli","email":"s.martinelli@gmail.com","username":"stevemar"},"date":"2016-11-14 12:02:20.000000000","message":"Patch Set 6: Code-Review-1\n\n(1 comment)\n\nwith the new initiative [1] lance is taking up around this topic, i see no reason to quickly approve this and get it backlogged. I\u0027d rather have an organized conversation on the topic and flesh things out.\n\n[1] https://etherpad.openstack.org/p/keystone-policy-meeting","accounts_in_message":[],"_revision_number":6},{"id":"d0557f0767955410fd629bf7b7748638ddc4731f","author":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"date":"2016-11-16 20:59:12.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"11db67fa4305338652fa4b5f9508d96c841bfb08","author":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"date":"2016-11-16 20:59:45.000000000","message":"Patch Set 7: Workflow-1\n\nMOdified in place to address previous comments (not all) but this is about to undergo a substantial revision","accounts_in_message":[],"_revision_number":7},{"id":"d65d073cf6d6645e228a7d08a1c3914b7a03645e","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-11-16 21:01:47.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-keystone-specs-docs-ubuntu-xenial http://docs-draft.openstack.org/24/391624/7/check/gate-keystone-specs-docs-ubuntu-xenial/fa624c4//doc/build/html/ : SUCCESS in 2m 20s\n- gate-keystone-specs-python27-ubuntu-xenial http://logs.openstack.org/24/391624/7/check/gate-keystone-specs-python27-ubuntu-xenial/73cf630/ : FAILURE in 1m 41s","accounts_in_message":[],"_revision_number":7},{"id":"d67bfe7db428d53a74c4c81d8fd49f422bb89dd8","author":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"date":"2016-11-17 15:06:43.000000000","message":"Patch Set 6:\n\n(24 comments)\n\nmajor changes after this","accounts_in_message":[],"_revision_number":6},{"id":"addea014aa251185fb3a13d8b9ff84dbfabf9ea6","author":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"date":"2016-11-17 15:07:02.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"748fec5c08c6995073651c9516dfc0b39d612814","author":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"date":"2016-11-17 15:09:02.000000000","message":"Patch Set 8:\n\nLatest version adds in the \"is_process\" check which I expect to be the dominant pattern.  If there is sufficient feedback that indicates I should remove the check that goes along with the token validation, I will do so.","accounts_in_message":[],"_revision_number":8},{"id":"e1b53de369752ceafc907e8449323425936a99ac","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-11-17 15:12:27.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-keystone-specs-docs-ubuntu-xenial http://docs-draft.openstack.org/24/391624/8/check/gate-keystone-specs-docs-ubuntu-xenial/8d894bd//doc/build/html/ : SUCCESS in 2m 29s\n- gate-keystone-specs-python27-ubuntu-xenial http://logs.openstack.org/24/391624/8/check/gate-keystone-specs-python27-ubuntu-xenial/597d9ce/ : SUCCESS in 3m 10s","accounts_in_message":[],"_revision_number":8},{"id":"cfde0fb17df0c8d2113fa64a0fc105e9aeaa5ca2","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2016-11-17 22:33:09.000000000","message":"Patch Set 6:\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"4736d75ceeb953ed8c66e66cb1c2d22f61ef2b5c","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2016-11-17 23:49:36.000000000","message":"Patch Set 8: Code-Review-1\n\n(11 comments)\n\nThis is coming along, but I still have some questions inline. I think we should continue to iterate on this until the next policy meeting.","accounts_in_message":[],"_revision_number":8},{"id":"0a51fe2eab4df5fb2e6d8c0bd6b3f3e9e7b47802","author":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"date":"2016-11-18 20:09:21.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"d58966cd02b077e1b8b054db21704ebecb35b576","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-11-18 20:13:01.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-keystone-specs-docs-ubuntu-xenial http://docs-draft.openstack.org/24/391624/9/check/gate-keystone-specs-docs-ubuntu-xenial/402c50e//doc/build/html/ : SUCCESS in 2m 27s\n- gate-keystone-specs-python27-ubuntu-xenial http://logs.openstack.org/24/391624/9/check/gate-keystone-specs-python27-ubuntu-xenial/9829900/ : SUCCESS in 2m 00s","accounts_in_message":[],"_revision_number":9},{"id":"978d238a602981c17355294f4850c04b68ed64e0","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2016-11-21 21:56:39.000000000","message":"Patch Set 9: Code-Review-1\n\nI still have a few unaddressed comments from patch set 8.","accounts_in_message":[],"_revision_number":9},{"id":"2a37838eda409b973aea342226bd1f0c9e90c91c","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2016-11-22 19:35:16.000000000","message":"Patch Set 9:\n\nAs discussed in the keystone meeting today - we should break this into smaller, bite-sized specs.","accounts_in_message":[],"_revision_number":9},{"id":"6a816ac6018af1b7d44b445aabcc13f601b6fabd","author":{"_account_id":6534,"name":"David Chadwick","email":"d.w.chadwick@truetrust.co.uk","username":"d-w-chadwick"},"date":"2016-11-23 15:55:48.000000000","message":"Patch Set 9:\n\n(3 comments)\n\nHere are a few comments for your consideration","accounts_in_message":[],"_revision_number":9},{"id":"be77369ec35e3541266ea5196978c13006159637","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2016-11-23 20:00:00.000000000","message":"Patch Set 9:\n\n(3 comments)","accounts_in_message":[],"_revision_number":9},{"id":"3f9c7fae5b414de43dec2173294a5a1060cc1230","author":{"_account_id":11022,"name":"Rodrigo Duarte Sousa","email":"rodrigodsousa@gmail.com","username":"rodrigods"},"date":"2016-11-24 13:15:36.000000000","message":"Patch Set 9: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":9},{"id":"d82d40e6ffb347af07e3f4db382eb94a8aaaacc3","author":{"_account_id":11022,"name":"Rodrigo Duarte Sousa","email":"rodrigodsousa@gmail.com","username":"rodrigods"},"date":"2016-11-24 13:15:57.000000000","message":"Patch Set 9:\n\nlooks good, need to address other reviewer\u0027s comments","accounts_in_message":[],"_revision_number":9},{"id":"9ba35927bb751ea8747737843f1f5981d68ff287","author":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"date":"2016-11-27 18:53:55.000000000","message":"Patch Set 10: Published edit on patch set 9","accounts_in_message":[],"_revision_number":10},{"id":"d30ffbf25a6e01fec81ac90bac1a15f90b87e50e","author":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"date":"2016-11-27 18:54:31.000000000","message":"Patch Set 10:\n\nSome of my in line comments were dropped when I made the edit.","accounts_in_message":[],"_revision_number":10},{"id":"b7171f33ade42dade137438248a75b99c15374f0","author":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"date":"2016-11-27 18:54:57.000000000","message":"Patch Set 9:\n\n(6 comments)","accounts_in_message":[],"_revision_number":9},{"id":"518dcb8ccdf17e121db4a804f43a1b9703deb2a9","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-11-27 18:56:23.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-keystone-specs-docs-ubuntu-xenial http://docs-draft.openstack.org/24/391624/10/check/gate-keystone-specs-docs-ubuntu-xenial/b820632//doc/build/html/ : SUCCESS in 2m 17s\n- gate-keystone-specs-python27-ubuntu-xenial http://logs.openstack.org/24/391624/10/check/gate-keystone-specs-python27-ubuntu-xenial/fb03a62/ : FAILURE in 2m 06s","accounts_in_message":[],"_revision_number":10},{"id":"ec4a618b93261daa7111274223934dbfc79322a3","author":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"date":"2016-11-28 22:02:56.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"5e7ff29ccf5690821b2aed961e93bce19d0ed792","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-11-28 22:06:12.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-keystone-specs-docs-ubuntu-xenial http://docs-draft.openstack.org/24/391624/11/check/gate-keystone-specs-docs-ubuntu-xenial/ba86304//doc/build/html/ : SUCCESS in 2m 21s\n- gate-keystone-specs-python27-ubuntu-xenial http://logs.openstack.org/24/391624/11/check/gate-keystone-specs-python27-ubuntu-xenial/bd27bd0/ : SUCCESS in 2m 33s","accounts_in_message":[],"_revision_number":11},{"id":"2b7bf96a301ca086769c4e5436b4f9546b5cc4f6","author":{"_account_id":7725,"name":"David Stanek","email":"dstanek@dstanek.com","username":"dstanek"},"date":"2016-11-30 14:40:16.000000000","message":"Patch Set 11:\n\n(15 comments)\n\nDoes this mean a deployer need to be familiar with all of the URL patterns a service uses? Any reason we can\u0027t use the same target names used in the policy file?","accounts_in_message":[],"_revision_number":11},{"id":"3be4816d1bc0df8f769aabcae04fdc7e68959ca1","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2016-11-30 14:46:52.000000000","message":"Patch Set 9:\n\n(6 comments)","accounts_in_message":[],"_revision_number":9},{"id":"1f771f1fe56e45b5022693d7b92f76ffcefb00d5","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2016-11-30 14:53:07.000000000","message":"Patch Set 11:\n\n(5 comments)\n\nPorted my comments from patch set 9.","accounts_in_message":[],"_revision_number":11},{"id":"911bb570b25eb36adc7f669d9f9fac66a07e49ad","author":{"_account_id":7725,"name":"David Stanek","email":"dstanek@dstanek.com","username":"dstanek"},"date":"2016-11-30 15:04:42.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"8e681386c62ec6e9a71a864165351c6e1d4ea424","author":{"_account_id":7725,"name":"David Stanek","email":"dstanek@dstanek.com","username":"dstanek"},"date":"2016-11-30 15:26:17.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"4534e29078702ee365e1591ded0cadab77ab0a8a","author":{"_account_id":7725,"name":"David Stanek","email":"dstanek@dstanek.com","username":"dstanek"},"date":"2016-11-30 15:31:25.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"4ae309366bb560853a280a4e6f4277a4a4096761","author":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"date":"2016-11-30 23:23:16.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"120301484c519e30cdefda1465283e824a70191e","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-11-30 23:27:15.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-keystone-specs-docs-ubuntu-xenial http://docs-draft.openstack.org/24/391624/12/check/gate-keystone-specs-docs-ubuntu-xenial/237f29b//doc/build/html/ : SUCCESS in 2m 44s\n- gate-keystone-specs-python27-ubuntu-xenial http://logs.openstack.org/24/391624/12/check/gate-keystone-specs-python27-ubuntu-xenial/266ea77/ : SUCCESS in 3m 07s","accounts_in_message":[],"_revision_number":12},{"id":"0b7f62f1912bd8307a97c526573065175642fd9a","author":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"date":"2016-12-01 03:09:13.000000000","message":"Uploaded patch set 13: Patch Set 12 was rebased.","accounts_in_message":[],"_revision_number":13},{"id":"4446938c0836b11631dc92f1f73a4ca009239749","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-12-01 03:12:01.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-keystone-specs-docs-ubuntu-xenial http://docs-draft.openstack.org/24/391624/13/check/gate-keystone-specs-docs-ubuntu-xenial/fd02554//doc/build/html/ : SUCCESS in 1m 41s\n- gate-keystone-specs-python27-ubuntu-xenial http://logs.openstack.org/24/391624/13/check/gate-keystone-specs-python27-ubuntu-xenial/8019abf/ : SUCCESS in 1m 41s","accounts_in_message":[],"_revision_number":13},{"id":"76cee8839b7abcd516e8aacc6bac7d2b4a7d98e9","author":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"date":"2016-12-01 03:54:49.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"9cf17a644918525ce959801fe7d970903ea257f3","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-12-01 03:57:31.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-keystone-specs-docs-ubuntu-xenial http://docs-draft.openstack.org/24/391624/14/check/gate-keystone-specs-docs-ubuntu-xenial/2eefe97//doc/build/html/ : SUCCESS in 1m 42s\n- gate-keystone-specs-python27-ubuntu-xenial http://logs.openstack.org/24/391624/14/check/gate-keystone-specs-python27-ubuntu-xenial/24556c5/ : SUCCESS in 2m 30s","accounts_in_message":[],"_revision_number":14},{"id":"23f9310deddf7573785cb2d5dc6b8e9b193bb297","author":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"date":"2016-12-02 00:42:22.000000000","message":"Patch Set 14:\n\n(11 comments)","accounts_in_message":[],"_revision_number":14},{"id":"cacbed6e356eec330fd7cf22bcd38b8895a7ce96","author":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"date":"2016-12-02 00:56:19.000000000","message":"Patch Set 14:\n\n(9 comments)","accounts_in_message":[],"_revision_number":14},{"id":"fc022ede0a84d073b9535b7ae6de703f9bd5f1a4","author":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"date":"2016-12-02 01:02:26.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"0ccb9ac18c578bd5426fc9862466e48fd4c3cc9c","author":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"date":"2016-12-02 14:57:58.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"8b54a3d0561e1c8cc81eb39a9f4143a5d37e2dec","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-12-02 15:04:30.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-keystone-specs-docs-ubuntu-xenial http://docs-draft.openstack.org/24/391624/15/check/gate-keystone-specs-docs-ubuntu-xenial/7e487c0//doc/build/html/ : SUCCESS in 1m 59s\n- gate-keystone-specs-python27-ubuntu-xenial http://logs.openstack.org/24/391624/15/check/gate-keystone-specs-python27-ubuntu-xenial/37efcd0/ : SUCCESS in 3m 12s","accounts_in_message":[],"_revision_number":15},{"id":"cdfe5d6bb980bf2f7d493c2b5a47e074de75c199","author":{"_account_id":6482,"name":"Steve Martinelli","email":"s.martinelli@gmail.com","username":"stevemar"},"date":"2016-12-06 20:21:51.000000000","message":"Patch Set 15: Code-Review+2\n\n(7 comments)\n\noverall i like the intention of this work, thanks for placing it in the ongoing directory. i like the fact that it is backwards compatible and central to keystone.","accounts_in_message":[],"_revision_number":15},{"id":"794a092e520035e34f4b69943ff7ea26a2b81534","author":{"_account_id":6482,"name":"Steve Martinelli","email":"s.martinelli@gmail.com","username":"stevemar"},"date":"2016-12-06 20:29:51.000000000","message":"Topic changed from token-verify-role-check to bp/token-verify-role-check)","accounts_in_message":[],"_revision_number":15},{"id":"6bc0eb7aa89dc807092995ab2101f06861106ce2","author":{"_account_id":6482,"name":"Steve Martinelli","email":"s.martinelli@gmail.com","username":"stevemar"},"date":"2016-12-06 20:29:58.000000000","message":"Topic changed from bp/token-verify-role-check) to bp/token-verify-role-check","accounts_in_message":[],"_revision_number":15},{"id":"42e7aad3c21cd06744ea2e26a53141b36b061e94","author":{"_account_id":6482,"name":"Steve Martinelli","email":"s.martinelli@gmail.com","username":"stevemar"},"date":"2016-12-06 20:37:46.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"12f602ea134f2bbf6b78ab919bc0d7b2bcfafcce","author":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"date":"2016-12-06 20:58:50.000000000","message":"Patch Set 15: Code-Review+2\n\nNeeds followup cleanups (marked by stevemar)","accounts_in_message":[],"_revision_number":15},{"id":"562b1dffe48f0c67bdf4ab75360a32ba2672db41","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2016-12-06 23:11:53.000000000","message":"Patch Set 15: Workflow+1\n\n(13 comments)","accounts_in_message":[],"_revision_number":15},{"id":"1d64a44c5405faf64b48f7146d3adb1824d6885d","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2016-12-06 23:11:59.000000000","message":"Patch Set 15: Workflow-1","accounts_in_message":[],"_revision_number":15},{"id":"193cc8ce0adc575f68a0bc66fd8bca03c996d9de","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2016-12-06 23:12:20.000000000","message":"Patch Set 15: -Workflow\n\nAH - OMG, I clearly can\u0027t mouse properly!!!!","accounts_in_message":[],"_revision_number":15},{"id":"24544d9f9e6fcc553b928de570ce557d5c37f5c8","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2016-12-06 23:14:34.000000000","message":"Patch Set 15: Code-Review-1\n\nOk - sorry for the sporadic voting, I got a little click happy...\n\nI have a few more questions related specifically to how policy is today and how this spec plans to support it. I also ported several of David\u0027s comments from patch set 11.","accounts_in_message":[],"_revision_number":15},{"id":"5595e67f42d8045e7996aae413fb59fdc588b539","author":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"date":"2016-12-07 19:04:12.000000000","message":"Patch Set 15:\n\n(20 comments)","accounts_in_message":[],"_revision_number":15},{"id":"5bb2cda59f6db8452ce31155b6c751c7a2917224","author":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"date":"2016-12-07 19:04:52.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"b8f5d12fd2848a1aaa1b58ab90f0776e41cb0a2c","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-12-07 19:08:02.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-keystone-specs-docs-ubuntu-xenial http://docs-draft.openstack.org/24/391624/16/check/gate-keystone-specs-docs-ubuntu-xenial/d7ccc4b//doc/build/html/ : SUCCESS in 1m 48s\n- gate-keystone-specs-python27-ubuntu-xenial http://logs.openstack.org/24/391624/16/check/gate-keystone-specs-python27-ubuntu-xenial/4107097/ : SUCCESS in 2m 49s","accounts_in_message":[],"_revision_number":16},{"id":"29f873b62082ba77f7b9a2a5fb1564c31ecf6cee","author":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"date":"2016-12-07 19:23:40.000000000","message":"Uploaded patch set 17: Commit message was updated.","accounts_in_message":[],"_revision_number":17},{"id":"46ab4f61066661c724e0675b40fc279c98809f64","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-12-07 19:29:39.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-keystone-specs-docs-ubuntu-xenial http://docs-draft.openstack.org/24/391624/17/check/gate-keystone-specs-docs-ubuntu-xenial/7740623//doc/build/html/ : SUCCESS in 2m 36s\n- gate-keystone-specs-python27-ubuntu-xenial http://logs.openstack.org/24/391624/17/check/gate-keystone-specs-python27-ubuntu-xenial/d087856/ : SUCCESS in 1m 49s","accounts_in_message":[],"_revision_number":17},{"id":"0e3f36ea1572e3f345fe1dcd89b61b5a5372c966","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2016-12-07 21:48:48.000000000","message":"Patch Set 15:\n\n(3 comments)","accounts_in_message":[],"_revision_number":15},{"id":"78e84a7263115a4eeaef4648e64a3e5327aaa716","author":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"date":"2016-12-08 02:07:47.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"df80cb2c45e63a7f628f24f1cc1ffe1961634f25","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-12-08 02:11:13.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-keystone-specs-docs-ubuntu-xenial http://docs-draft.openstack.org/24/391624/18/check/gate-keystone-specs-docs-ubuntu-xenial/7bc726f//doc/build/html/ : SUCCESS in 1m 42s\n- gate-keystone-specs-python27-ubuntu-xenial http://logs.openstack.org/24/391624/18/check/gate-keystone-specs-python27-ubuntu-xenial/b77936f/ : SUCCESS in 1m 41s","accounts_in_message":[],"_revision_number":18},{"id":"ae01ebd68d524d5e270bfee3a7ec8a89ea181e4a","author":{"_account_id":6482,"name":"Steve Martinelli","email":"s.martinelli@gmail.com","username":"stevemar"},"date":"2016-12-08 03:39:20.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"bd6571352f0557fd855a37dd7fbd29b3be850c20","author":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"date":"2016-12-08 15:41:57.000000000","message":"Patch Set 18:\n\n(15 comments)","accounts_in_message":[],"_revision_number":18},{"id":"200b1c2219e10dbcc511da346e78a781e236e99c","author":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"date":"2016-12-08 16:08:11.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"93ffd2c14845791ab7f1026bf9ed8583285d2894","author":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"date":"2016-12-08 16:08:17.000000000","message":"Patch Set 18:\n\n(13 comments)","accounts_in_message":[],"_revision_number":18},{"id":"ce1984219810ed1f7b87fd170e173d565975a355","author":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"date":"2016-12-08 16:08:41.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"c4e3ac6bbf7d65d05ad32fa5dbe7d36c44c443e6","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-12-08 16:13:40.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-keystone-specs-docs-ubuntu-xenial http://docs-draft.openstack.org/24/391624/19/check/gate-keystone-specs-docs-ubuntu-xenial/ed44fc4//doc/build/html/ : SUCCESS in 4m 03s\n- gate-keystone-specs-python27-ubuntu-xenial http://logs.openstack.org/24/391624/19/check/gate-keystone-specs-python27-ubuntu-xenial/fde288c/ : SUCCESS in 2m 24s","accounts_in_message":[],"_revision_number":19},{"id":"4e96e9cf917d26a16422c3300b85201389f5affe","author":{"_account_id":10608,"name":"Matthew Edmonds","email":"edmondsw@us.ibm.com","username":"edmondsw"},"date":"2016-12-08 22:44:30.000000000","message":"Patch Set 19: Code-Review-1\n\n(19 comments)\n\nhaven\u0027t gotten through the whole spec, but here\u0027s a batch of comments to start looking at.","accounts_in_message":[],"_revision_number":19},{"id":"f6d7ec2fa6022ce050384ac455c091826498b5ed","author":{"_account_id":10608,"name":"Matthew Edmonds","email":"edmondsw@us.ibm.com","username":"edmondsw"},"date":"2016-12-09 13:57:24.000000000","message":"Patch Set 19:\n\n(38 comments)\n\nfurther comments","accounts_in_message":[],"_revision_number":19},{"id":"a160b0fc7e9e103c516c28a504fbb75149831220","author":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"date":"2016-12-09 22:06:09.000000000","message":"Uploaded patch set 20.","accounts_in_message":[],"_revision_number":20},{"id":"42a4fb198466f36017e0064044b05617e11d183c","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-12-09 22:10:40.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-keystone-specs-docs-ubuntu-xenial http://docs-draft.openstack.org/24/391624/20/check/gate-keystone-specs-docs-ubuntu-xenial/c9107fa//doc/build/html/ : SUCCESS in 1m 48s\n- gate-keystone-specs-python27-ubuntu-xenial http://logs.openstack.org/24/391624/20/check/gate-keystone-specs-python27-ubuntu-xenial/7b6cf79/ : SUCCESS in 3m 05s","accounts_in_message":[],"_revision_number":20},{"id":"650e721b0a9a0958700c9e5239d4c8b29f28c6cc","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2016-12-12 17:11:19.000000000","message":"Patch Set 20: Code-Review-1\n\n(14 comments)","accounts_in_message":[],"_revision_number":20},{"id":"01eddcfe57128b0e1bfe9a51e1dbd56a1f843b35","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2016-12-12 22:35:33.000000000","message":"Patch Set 20:\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"12df4c83b162db67ce6369992800cf54a8f4bb06","author":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"date":"2016-12-12 23:09:56.000000000","message":"Patch Set 20:\n\n(15 comments)","accounts_in_message":[],"_revision_number":20},{"id":"f3e5625c4f4c5594ec06c20a1c3eb0abca126d5f","author":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"date":"2016-12-12 23:10:08.000000000","message":"Uploaded patch set 21.","accounts_in_message":[],"_revision_number":21},{"id":"0c448b57d672b02d6c83350d3ed06b842f57f41f","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-12-12 23:15:30.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-keystone-specs-docs-ubuntu-xenial http://docs-draft.openstack.org/24/391624/21/check/gate-keystone-specs-docs-ubuntu-xenial/258e8d3//doc/build/html/ : SUCCESS in 3m 19s\n- gate-keystone-specs-python27-ubuntu-xenial http://logs.openstack.org/24/391624/21/check/gate-keystone-specs-python27-ubuntu-xenial/e608b87/ : SUCCESS in 3m 54s","accounts_in_message":[],"_revision_number":21},{"id":"fafb7f0cf74406246b1b5adbd199cf63b6aad80a","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2016-12-13 00:03:28.000000000","message":"Patch Set 20:\n\n(2 comments)","accounts_in_message":[],"_revision_number":20},{"id":"f1aa1838d8dab5a2e725df503b36ad5c3dab0443","author":{"_account_id":7725,"name":"David Stanek","email":"dstanek@dstanek.com","username":"dstanek"},"date":"2016-12-13 00:28:53.000000000","message":"Patch Set 20:\n\n(3 comments)","accounts_in_message":[],"_revision_number":20},{"id":"3b656ccc07513b507c479e2611c6b90d2663bde1","author":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"date":"2016-12-13 15:10:20.000000000","message":"Patch Set 20:\n\n(2 comments)","accounts_in_message":[],"_revision_number":20},{"id":"9ed690af4a30c75f6d87c6c7c5accac16cc8d480","author":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"date":"2016-12-13 18:20:03.000000000","message":"Patch Set 21: Code-Review+2\n\nSince this allows current workflow as is, and is ongoing (big change). I think it is the right direction.","accounts_in_message":[],"_revision_number":21},{"id":"7a3f0bf328780861cb3987dbb0491a8d8b702aa6","author":{"_account_id":6482,"name":"Steve Martinelli","email":"s.martinelli@gmail.com","username":"stevemar"},"date":"2016-12-13 19:00:56.000000000","message":"Patch Set 21: Code-Review+2","accounts_in_message":[],"_revision_number":21},{"id":"beb7243559881c594124a2d0e11494f230780bc5","author":{"_account_id":11022,"name":"Rodrigo Duarte Sousa","email":"rodrigodsousa@gmail.com","username":"rodrigods"},"date":"2016-12-13 23:18:24.000000000","message":"Patch Set 21: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":21},{"id":"08f73fbc7e40fde2094e1d1eb46eb8a578f158cb","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-12-13 23:18:37.000000000","message":"Patch Set 21: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":21},{"id":"d6f578e6acb04beab03705563a0d314755c0e334","date":"2016-12-13 23:23:19.000000000","message":"Change has been successfully merged into the git repository by Jenkins","accounts_in_message":[],"_revision_number":21},{"id":"12b1d50c7c9eab99517fd9a7c3b5f15dce0adf9e","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-12-13 23:23:19.000000000","message":"Patch Set 21: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-keystone-specs-docs-ubuntu-xenial http://docs-draft.openstack.org/24/391624/21/gate/gate-keystone-specs-docs-ubuntu-xenial/65fb429//doc/build/html/ : SUCCESS in 4m 03s\n- gate-keystone-specs-python27-ubuntu-xenial http://logs.openstack.org/24/391624/21/gate/gate-keystone-specs-python27-ubuntu-xenial/0ecc80f/ : SUCCESS in 2m 46s","accounts_in_message":[],"_revision_number":21},{"id":"297d5745866a210feef6c1fd3a4c242a5d16a528","author":{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},"date":"2016-12-26 12:07:21.000000000","message":"Patch Set 21:\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"63858b7cae452af73e00409ba4a07a74f89aede7","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2017-01-03 16:39:22.000000000","message":"Patch Set 21:\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"7141458be5ff653c674a2dd0098759c5ce402172","author":{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},"date":"2017-01-03 19:28:13.000000000","message":"Patch Set 21:\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"058e0db48431d4051eef9d8fda01fc9849cd49c3","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2017-01-05 12:30:11.000000000","message":"Patch Set 21:\n\n(15 comments)\n\nTotally late to the party here, but adding some comments on what I see. Basically same feedback as summit discussions, I don\u0027t really get this one.\n\nI am just not sure I understand what problems this is trying to solve. I mean its a cool, it just doesn\u0027t seem to address the usability issues our users are facing with policy. But I suspect I am miss-understanding something here.","accounts_in_message":[],"_revision_number":21},{"id":"0da85fe5d8db3efc3abb09e873ffc002e7fa987d","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2017-01-05 16:58:13.000000000","message":"Patch Set 21:\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"e014f718e4d418ce21ffb31adf1934ee99d3114d","author":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"date":"2021-09-24 20:29:41.000000000","message":"Patch Set 21:\n\n(55 comments)","accounts_in_message":[],"_revision_number":21}],"current_revision_number":21,"current_revision":"79f3b56ffac5957b0223ef789aa9b096b88ee727","revisions":{"bbe3755c8a203558363650b4ac4a43d2c578ebde":{"kind":"REWORK","_number":1,"created":"2016-10-29 21:28:30.000000000","uploader":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"ref":"refs/changes/24/391624/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone-specs","ref":"refs/changes/24/391624/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/24/391624/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/24/391624/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/24/391624/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone-specs refs/changes/24/391624/1"}}},"commit":{"parents":[{"commit":"fa7b18ad94400c85f45ec71eddd873e9fe2b6a71","subject":"Merge \"Changed the home-page link\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/fa7b18ad94400c85f45ec71eddd873e9fe2b6a71"}]}],"author":{"name":"Adam Young","email":"ayoung@redhat.com","date":"2016-10-29 21:24:09.000000000","tz":-240},"committer":{"name":"Adam Young","email":"ayoung@redhat.com","date":"2016-10-29 21:28:20.000000000","tz":-240},"subject":"Token Verify Role Check","message":"Token Verify Role Check\n\nAPIImpact:  This change allows for additional parameters to verify tokens.\n\nSecurityImpact: This change allows a greater degree of RBAC control, which\nshould increase security\n\nChange-Id: I2487fe09091e9b054a91a9853b99f3dc14fbeebb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/bbe3755c8a203558363650b4ac4a43d2c578ebde"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/bbe3755c8a203558363650b4ac4a43d2c578ebde"}]},"branch":"refs/heads/master"},"c6c10b414949ebc22e092fc22ec34afe72cd05d2":{"kind":"REWORK","_number":2,"created":"2016-10-31 02:37:05.000000000","uploader":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"ref":"refs/changes/24/391624/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone-specs","ref":"refs/changes/24/391624/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/24/391624/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/24/391624/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/24/391624/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone-specs refs/changes/24/391624/2"}}},"commit":{"parents":[{"commit":"fa7b18ad94400c85f45ec71eddd873e9fe2b6a71","subject":"Merge \"Changed the home-page link\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/fa7b18ad94400c85f45ec71eddd873e9fe2b6a71"}]}],"author":{"name":"Adam Young","email":"ayoung@redhat.com","date":"2016-10-29 21:24:09.000000000","tz":-240},"committer":{"name":"Adam Young","email":"ayoung@redhat.com","date":"2016-10-31 02:36:57.000000000","tz":-240},"subject":"Token Verify Role Check","message":"Token Verify Role Check\n\nAPIImpact:  This change allows for additional parameters to verify tokens.\n\nSecurityImpact: This change allows a greater degree of RBAC control, which\nshould increase security\n\nChange-Id: I2487fe09091e9b054a91a9853b99f3dc14fbeebb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/c6c10b414949ebc22e092fc22ec34afe72cd05d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/c6c10b414949ebc22e092fc22ec34afe72cd05d2"}]},"branch":"refs/heads/master"},"117e3c1a8f134d81380ad10c91f3a9c5c57ed2db":{"kind":"REWORK","_number":3,"created":"2016-11-01 16:22:58.000000000","uploader":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"ref":"refs/changes/24/391624/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone-specs","ref":"refs/changes/24/391624/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/24/391624/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/24/391624/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/24/391624/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone-specs refs/changes/24/391624/3"}}},"commit":{"parents":[{"commit":"fa7b18ad94400c85f45ec71eddd873e9fe2b6a71","subject":"Merge \"Changed the home-page link\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/fa7b18ad94400c85f45ec71eddd873e9fe2b6a71"}]}],"author":{"name":"Adam Young","email":"ayoung@redhat.com","date":"2016-10-29 21:24:09.000000000","tz":-240},"committer":{"name":"Adam Young","email":"ayoung@redhat.com","date":"2016-11-01 16:22:44.000000000","tz":-240},"subject":"Token Verify Role Check","message":"Token Verify Role Check\n\nAPIImpact:  This change allows for additional parameters to verify tokens.\n\nSecurityImpact: This change allows a greater degree of RBAC control, which\nshould increase security\n\nChange-Id: I2487fe09091e9b054a91a9853b99f3dc14fbeebb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/117e3c1a8f134d81380ad10c91f3a9c5c57ed2db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/117e3c1a8f134d81380ad10c91f3a9c5c57ed2db"}]},"branch":"refs/heads/master"},"756f6e1244886f02bbe5d5628361cc016985f17f":{"kind":"REWORK","_number":4,"created":"2016-11-02 03:26:55.000000000","uploader":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"ref":"refs/changes/24/391624/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone-specs","ref":"refs/changes/24/391624/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/24/391624/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/24/391624/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/24/391624/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone-specs refs/changes/24/391624/4"}}},"commit":{"parents":[{"commit":"fa7b18ad94400c85f45ec71eddd873e9fe2b6a71","subject":"Merge \"Changed the home-page link\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/fa7b18ad94400c85f45ec71eddd873e9fe2b6a71"}]}],"author":{"name":"Adam Young","email":"ayoung@redhat.com","date":"2016-10-29 21:24:09.000000000","tz":-240},"committer":{"name":"Adam Young","email":"ayoung@redhat.com","date":"2016-11-02 03:26:41.000000000","tz":-240},"subject":"Token Verify Role Check","message":"Token Verify Role Check\n\nAPIImpact:  This change allows for additional parameters to verify tokens.\n\nSecurityImpact: This change allows a greater degree of RBAC control, which\nshould increase security\n\nChange-Id: I2487fe09091e9b054a91a9853b99f3dc14fbeebb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/756f6e1244886f02bbe5d5628361cc016985f17f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/756f6e1244886f02bbe5d5628361cc016985f17f"}]},"branch":"refs/heads/master"},"ebe8e3d8cc2482e123c771f39eabe4e1c942c589":{"kind":"REWORK","_number":5,"created":"2016-11-02 13:40:46.000000000","uploader":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"ref":"refs/changes/24/391624/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone-specs","ref":"refs/changes/24/391624/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/24/391624/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/24/391624/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/24/391624/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone-specs refs/changes/24/391624/5"}}},"commit":{"parents":[{"commit":"fa7b18ad94400c85f45ec71eddd873e9fe2b6a71","subject":"Merge \"Changed the home-page link\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/fa7b18ad94400c85f45ec71eddd873e9fe2b6a71"}]}],"author":{"name":"Adam Young","email":"ayoung@redhat.com","date":"2016-10-29 21:24:09.000000000","tz":-240},"committer":{"name":"Adam Young","email":"ayoung@redhat.com","date":"2016-11-02 13:40:40.000000000","tz":-240},"subject":"Token Verify Role Check","message":"Token Verify Role Check\n\nAPIImpact:  This change allows for additional parameters to verify tokens.\n\nSecurityImpact: This change allows a greater degree of RBAC control, which\nshould increase security\n\nChange-Id: I2487fe09091e9b054a91a9853b99f3dc14fbeebb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/ebe8e3d8cc2482e123c771f39eabe4e1c942c589"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/ebe8e3d8cc2482e123c771f39eabe4e1c942c589"}]},"branch":"refs/heads/master"},"fd6287da8685d778f623a705fd335c48f52469d6":{"kind":"REWORK","_number":6,"created":"2016-11-03 13:06:12.000000000","uploader":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"ref":"refs/changes/24/391624/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone-specs","ref":"refs/changes/24/391624/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/24/391624/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/24/391624/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/24/391624/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone-specs refs/changes/24/391624/6"}}},"commit":{"parents":[{"commit":"fa7b18ad94400c85f45ec71eddd873e9fe2b6a71","subject":"Merge \"Changed the home-page link\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/fa7b18ad94400c85f45ec71eddd873e9fe2b6a71"}]}],"author":{"name":"Adam Young","email":"ayoung@redhat.com","date":"2016-10-29 21:24:09.000000000","tz":-240},"committer":{"name":"Adam Young","email":"ayoung@redhat.com","date":"2016-11-03 13:05:14.000000000","tz":-240},"subject":"Token Verify Role Check","message":"Token Verify Role Check\n\nVerify the roles for the API during the token\nvalidation process.\n\nAPIImpact:  addis parameters to verify tokens.\n\nSecurityImpact: This change allows a greater\n degree of RBAC control, which should increase\n security\n\nChange-Id: I2487fe09091e9b054a91a9853b99f3dc14fbeebb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/fd6287da8685d778f623a705fd335c48f52469d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/fd6287da8685d778f623a705fd335c48f52469d6"}]},"branch":"refs/heads/master"},"ef5ff2832c8dc8b618ef453b889d95f13c8a578e":{"kind":"REWORK","_number":7,"created":"2016-11-16 20:59:12.000000000","uploader":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"ref":"refs/changes/24/391624/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone-specs","ref":"refs/changes/24/391624/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/24/391624/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/24/391624/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/24/391624/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone-specs refs/changes/24/391624/7"}}},"commit":{"parents":[{"commit":"fa7b18ad94400c85f45ec71eddd873e9fe2b6a71","subject":"Merge \"Changed the home-page link\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/fa7b18ad94400c85f45ec71eddd873e9fe2b6a71"}]}],"author":{"name":"Adam Young","email":"ayoung@redhat.com","date":"2016-10-29 21:24:09.000000000","tz":-240},"committer":{"name":"Adam Young","email":"ayoung@redhat.com","date":"2016-11-16 20:59:05.000000000","tz":-300},"subject":"Token Verify Role Check","message":"Token Verify Role Check\n\nVerify the roles for the API during the token\nvalidation process.\n\nAPIImpact:  addis parameters to verify tokens.\n\nSecurityImpact: This change allows a greater\n degree of RBAC control, which should increase\n security\n\nChange-Id: I2487fe09091e9b054a91a9853b99f3dc14fbeebb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/ef5ff2832c8dc8b618ef453b889d95f13c8a578e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/ef5ff2832c8dc8b618ef453b889d95f13c8a578e"}]},"branch":"refs/heads/master"},"5f7aac4802098980b5959765a57298f66b2c4d53":{"kind":"REWORK","_number":8,"created":"2016-11-17 15:07:02.000000000","uploader":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"ref":"refs/changes/24/391624/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone-specs","ref":"refs/changes/24/391624/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/24/391624/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/24/391624/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/24/391624/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone-specs refs/changes/24/391624/8"}}},"commit":{"parents":[{"commit":"fa7b18ad94400c85f45ec71eddd873e9fe2b6a71","subject":"Merge \"Changed the home-page link\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/fa7b18ad94400c85f45ec71eddd873e9fe2b6a71"}]}],"author":{"name":"Adam Young","email":"ayoung@redhat.com","date":"2016-10-29 21:24:09.000000000","tz":-240},"committer":{"name":"Adam Young","email":"ayoung@redhat.com","date":"2016-11-17 15:06:55.000000000","tz":-300},"subject":"Token Verify Role Check","message":"Token Verify Role Check\n\nVerify the roles for the API during the token\nvalidation process.\n\nAPIImpact:  addis parameters to verify tokens.\n\nSecurityImpact: This change allows a greater\n degree of RBAC control, which should increase\n security\n\nChange-Id: I2487fe09091e9b054a91a9853b99f3dc14fbeebb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/5f7aac4802098980b5959765a57298f66b2c4d53"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/5f7aac4802098980b5959765a57298f66b2c4d53"}]},"branch":"refs/heads/master"},"28d39871c68c758a33c8da61e62f8ae47a53eb6a":{"kind":"REWORK","_number":9,"created":"2016-11-18 20:09:21.000000000","uploader":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"ref":"refs/changes/24/391624/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone-specs","ref":"refs/changes/24/391624/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/24/391624/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/24/391624/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/24/391624/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone-specs refs/changes/24/391624/9"}}},"commit":{"parents":[{"commit":"fa7b18ad94400c85f45ec71eddd873e9fe2b6a71","subject":"Merge \"Changed the home-page link\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/fa7b18ad94400c85f45ec71eddd873e9fe2b6a71"}]}],"author":{"name":"Adam Young","email":"ayoung@redhat.com","date":"2016-10-29 21:24:09.000000000","tz":-240},"committer":{"name":"Adam Young","email":"ayoung@redhat.com","date":"2016-11-18 20:09:14.000000000","tz":-300},"subject":"Token Verify Role Check","message":"Token Verify Role Check\n\nVerify the roles for the API during the token\nvalidation process.\n\nAPIImpact:  addis parameters to verify tokens.\n\nSecurityImpact: This change allows a greater\n degree of RBAC control, which should increase\n security\n\nChange-Id: I2487fe09091e9b054a91a9853b99f3dc14fbeebb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/28d39871c68c758a33c8da61e62f8ae47a53eb6a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/28d39871c68c758a33c8da61e62f8ae47a53eb6a"}]},"branch":"refs/heads/master"},"f914a1e95b468df11cc5783a83b888e63e55e010":{"kind":"REWORK","_number":10,"created":"2016-11-27 18:53:55.000000000","uploader":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"ref":"refs/changes/24/391624/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone-specs","ref":"refs/changes/24/391624/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/24/391624/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/24/391624/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/24/391624/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone-specs refs/changes/24/391624/10"}}},"commit":{"parents":[{"commit":"fa7b18ad94400c85f45ec71eddd873e9fe2b6a71","subject":"Merge \"Changed the home-page link\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/fa7b18ad94400c85f45ec71eddd873e9fe2b6a71"}]}],"author":{"name":"Adam Young","email":"ayoung@redhat.com","date":"2016-10-29 21:24:09.000000000","tz":-240},"committer":{"name":"ayoung","email":"ayoung@redhat.com","date":"2016-11-27 18:53:38.000000000","tz":0},"subject":"Token Verify Role Check","message":"Token Verify Role Check\n\nVerify the roles for the API during the token\nvalidation process.\n\nAPIImpact:  addis parameters to verify tokens.\n\nSecurityImpact: This change allows a greater\n degree of RBAC control, which should increase\n security\n\nChange-Id: I2487fe09091e9b054a91a9853b99f3dc14fbeebb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/f914a1e95b468df11cc5783a83b888e63e55e010"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/f914a1e95b468df11cc5783a83b888e63e55e010"}]},"branch":"refs/heads/master"},"1e14332661785f8cc9ffeccad6680e86294d1b91":{"kind":"REWORK","_number":11,"created":"2016-11-28 22:02:56.000000000","uploader":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"ref":"refs/changes/24/391624/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone-specs","ref":"refs/changes/24/391624/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/24/391624/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/24/391624/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/24/391624/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone-specs refs/changes/24/391624/11"}}},"commit":{"parents":[{"commit":"fa7b18ad94400c85f45ec71eddd873e9fe2b6a71","subject":"Merge \"Changed the home-page link\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/fa7b18ad94400c85f45ec71eddd873e9fe2b6a71"}]}],"author":{"name":"Adam Young","email":"ayoung@redhat.com","date":"2016-10-29 21:24:09.000000000","tz":-240},"committer":{"name":"Adam Young","email":"ayoung@redhat.com","date":"2016-11-28 22:01:45.000000000","tz":-300},"subject":"Role Check Check from Middleware","message":"Role Check Check from Middleware\n\nVerify the roles for the API during the\nkeystone middleware call, after token\nvalidation but prior to handing control\nover to the remote service.\n\nAPIImpact:  addis parameters to verify tokens.\n\nSecurityImpact: This change allows a greater\n degree of RBAC control, which should increase\n security\n\nChange-Id: I2487fe09091e9b054a91a9853b99f3dc14fbeebb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/1e14332661785f8cc9ffeccad6680e86294d1b91"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/1e14332661785f8cc9ffeccad6680e86294d1b91"}]},"branch":"refs/heads/master"},"de1a34f8941ff2d91f71e3284d48ea22ee588fa4":{"kind":"REWORK","_number":12,"created":"2016-11-30 23:23:16.000000000","uploader":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"ref":"refs/changes/24/391624/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone-specs","ref":"refs/changes/24/391624/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/24/391624/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/24/391624/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/24/391624/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone-specs refs/changes/24/391624/12"}}},"commit":{"parents":[{"commit":"fa7b18ad94400c85f45ec71eddd873e9fe2b6a71","subject":"Merge \"Changed the home-page link\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/fa7b18ad94400c85f45ec71eddd873e9fe2b6a71"}]}],"author":{"name":"Adam Young","email":"ayoung@redhat.com","date":"2016-10-29 21:24:09.000000000","tz":-240},"committer":{"name":"Adam Young","email":"ayoung@redhat.com","date":"2016-11-30 23:23:10.000000000","tz":-300},"subject":"Role Check Check from Middleware","message":"Role Check Check from Middleware\n\nVerify the roles for the API during the\nkeystone middleware call, after token\nvalidation but prior to handing control\nover to the remote service.\n\nAPIImpact:  addis parameters to verify tokens.\n\nSecurityImpact: This change allows a greater\n degree of RBAC control, which should increase\n security\n\nChange-Id: I2487fe09091e9b054a91a9853b99f3dc14fbeebb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/de1a34f8941ff2d91f71e3284d48ea22ee588fa4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/de1a34f8941ff2d91f71e3284d48ea22ee588fa4"}]},"branch":"refs/heads/master"},"82bf6fd35076eecce091fe562bbd7ed189fad871":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2016-12-01 03:09:13.000000000","uploader":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"ref":"refs/changes/24/391624/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone-specs","ref":"refs/changes/24/391624/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/24/391624/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/24/391624/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/24/391624/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone-specs refs/changes/24/391624/13"}}},"commit":{"parents":[{"commit":"550a24894f289bf00bd260f96f93afa2be6e510e","subject":"Fix python version i to 2.7 for docs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/550a24894f289bf00bd260f96f93afa2be6e510e"}]}],"author":{"name":"Adam Young","email":"ayoung@redhat.com","date":"2016-10-29 21:24:09.000000000","tz":-240},"committer":{"name":"Adam Young","email":"ayoung@redhat.com","date":"2016-12-01 03:08:50.000000000","tz":-300},"subject":"Role Check Check from Middleware","message":"Role Check Check from Middleware\n\nVerify the roles for the API during the\nkeystone middleware call, after token\nvalidation but prior to handing control\nover to the remote service.\n\nAPIImpact:  addis parameters to verify tokens.\n\nSecurityImpact: This change allows a greater\n degree of RBAC control, which should increase\n security\n\nChange-Id: I2487fe09091e9b054a91a9853b99f3dc14fbeebb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/82bf6fd35076eecce091fe562bbd7ed189fad871"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/82bf6fd35076eecce091fe562bbd7ed189fad871"}]},"branch":"refs/heads/master"},"fdb50926eeaa962d553efe3d32f878c62aeb8e69":{"kind":"REWORK","_number":14,"created":"2016-12-01 03:54:49.000000000","uploader":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"ref":"refs/changes/24/391624/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone-specs","ref":"refs/changes/24/391624/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/24/391624/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/24/391624/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/24/391624/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone-specs refs/changes/24/391624/14"}}},"commit":{"parents":[{"commit":"18852ff5162da0d6d58f231ce15212570264abb1","subject":"Fix python version to 2.7 for docs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/18852ff5162da0d6d58f231ce15212570264abb1"}]}],"author":{"name":"Adam Young","email":"ayoung@redhat.com","date":"2016-10-29 21:24:09.000000000","tz":-240},"committer":{"name":"Adam Young","email":"ayoung@redhat.com","date":"2016-12-01 03:54:40.000000000","tz":-300},"subject":"Role Check Check from Middleware","message":"Role Check Check from Middleware\n\nVerify the roles for the API during the\nkeystone middleware call, after token\nvalidation but prior to handing control\nover to the remote service.\n\nAPIImpact:  addis parameters to verify tokens.\n\nSecurityImpact: This change allows a greater\n degree of RBAC control, which should increase\n security\n\nChange-Id: I2487fe09091e9b054a91a9853b99f3dc14fbeebb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/fdb50926eeaa962d553efe3d32f878c62aeb8e69"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/fdb50926eeaa962d553efe3d32f878c62aeb8e69"}]},"branch":"refs/heads/master"},"d81a23c542e54c464bdfb862e0d485c2e59e353b":{"kind":"REWORK","_number":15,"created":"2016-12-02 14:57:58.000000000","uploader":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"ref":"refs/changes/24/391624/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone-specs","ref":"refs/changes/24/391624/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/24/391624/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/24/391624/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/24/391624/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone-specs refs/changes/24/391624/15"}}},"commit":{"parents":[{"commit":"603d1b52d31f7505cd1017f94b77cac7bb13a44a","subject":"Fix python version to 2.7 for docs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/603d1b52d31f7505cd1017f94b77cac7bb13a44a"}]}],"author":{"name":"Adam Young","email":"ayoung@redhat.com","date":"2016-10-29 21:24:09.000000000","tz":-240},"committer":{"name":"Adam Young","email":"ayoung@redhat.com","date":"2016-12-02 14:57:33.000000000","tz":-300},"subject":"Role Check Check from Middleware","message":"Role Check Check from Middleware\n\nVerify the roles for the API during the\nkeystone middleware call, after token\nvalidation but prior to handing control\nover to the remote service.\n\nAPIImpact:  adds new APIS for managing access\n            rules.  Adds RBAC check in all APIs.\n\nSecurityImpact: This change allows a greater\n degree of RBAC control, which should increase\n security\n\nChange-Id: I2487fe09091e9b054a91a9853b99f3dc14fbeebb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/d81a23c542e54c464bdfb862e0d485c2e59e353b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/d81a23c542e54c464bdfb862e0d485c2e59e353b"}]},"branch":"refs/heads/master"},"9d0f0cc56f62294330eac138446a1100ac002a5f":{"kind":"REWORK","_number":16,"created":"2016-12-07 19:04:52.000000000","uploader":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"ref":"refs/changes/24/391624/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone-specs","ref":"refs/changes/24/391624/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/24/391624/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/24/391624/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/24/391624/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone-specs refs/changes/24/391624/16"}}},"commit":{"parents":[{"commit":"603d1b52d31f7505cd1017f94b77cac7bb13a44a","subject":"Fix python version to 2.7 for docs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/603d1b52d31f7505cd1017f94b77cac7bb13a44a"}]}],"author":{"name":"Adam Young","email":"ayoung@redhat.com","date":"2016-10-29 21:24:09.000000000","tz":-240},"committer":{"name":"Adam Young","email":"ayoung@redhat.com","date":"2016-12-07 19:04:46.000000000","tz":-300},"subject":"Role Check Check from Middleware","message":"Role Check Check from Middleware\n\nVerify the roles for the API during the\nkeystone middleware call, after token\nvalidation but prior to handing control\nover to the remote service.\n\nAPIImpact:  adds new APIS for managing access\n            rules.  Adds RBAC check in all APIs.\n\nSecurityImpact: This change allows a greater\n degree of RBAC control, which should increase\n security\n\nChange-Id: I2487fe09091e9b054a91a9853b99f3dc14fbeebb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/9d0f0cc56f62294330eac138446a1100ac002a5f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/9d0f0cc56f62294330eac138446a1100ac002a5f"}]},"branch":"refs/heads/master"},"cfa7cf21503d995017d06e7c38447e888b2c4e6c":{"kind":"NO_CODE_CHANGE","_number":17,"created":"2016-12-07 19:23:40.000000000","uploader":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"ref":"refs/changes/24/391624/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone-specs","ref":"refs/changes/24/391624/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/24/391624/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/24/391624/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/24/391624/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone-specs refs/changes/24/391624/17"}}},"commit":{"parents":[{"commit":"603d1b52d31f7505cd1017f94b77cac7bb13a44a","subject":"Fix python version to 2.7 for docs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/603d1b52d31f7505cd1017f94b77cac7bb13a44a"}]}],"author":{"name":"Adam Young","email":"ayoung@redhat.com","date":"2016-10-29 21:24:09.000000000","tz":-240},"committer":{"name":"Adam Young","email":"ayoung@redhat.com","date":"2016-12-07 19:23:31.000000000","tz":-300},"subject":"Role Check from Middleware","message":"Role Check from Middleware\n\nVerify the roles for the API during the\nkeystone middleware call, after token\nvalidation but prior to handing control\nover to the remote service.\n\nAPIImpact:  adds new APIS for managing access\n            rules.  Adds RBAC check in all APIs.\n\nSecurityImpact: This change allows a greater\n degree of RBAC control, which should increase\n security\n\nChange-Id: I2487fe09091e9b054a91a9853b99f3dc14fbeebb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/cfa7cf21503d995017d06e7c38447e888b2c4e6c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/cfa7cf21503d995017d06e7c38447e888b2c4e6c"}]},"branch":"refs/heads/master"},"4e49f8ae0cbf36dd923c543c0041d6316b76fccb":{"kind":"REWORK","_number":18,"created":"2016-12-08 02:07:47.000000000","uploader":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"ref":"refs/changes/24/391624/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone-specs","ref":"refs/changes/24/391624/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/24/391624/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/24/391624/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/24/391624/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone-specs refs/changes/24/391624/18"}}},"commit":{"parents":[{"commit":"603d1b52d31f7505cd1017f94b77cac7bb13a44a","subject":"Fix python version to 2.7 for docs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/603d1b52d31f7505cd1017f94b77cac7bb13a44a"}]}],"author":{"name":"Adam Young","email":"ayoung@redhat.com","date":"2016-10-29 21:24:09.000000000","tz":-240},"committer":{"name":"Adam Young","email":"ayoung@redhat.com","date":"2016-12-08 02:07:39.000000000","tz":-300},"subject":"Role Check from Middleware","message":"Role Check from Middleware\n\nVerify the roles for the API during the\nkeystone middleware call, after token\nvalidation but prior to handing control\nover to the remote service.\n\nAPIImpact:  adds new APIS for managing access\n            rules.  Adds RBAC check in all APIs.\n\nSecurityImpact: This change allows a greater\n degree of RBAC control, which should increase\n security\n\nChange-Id: I2487fe09091e9b054a91a9853b99f3dc14fbeebb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/4e49f8ae0cbf36dd923c543c0041d6316b76fccb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/4e49f8ae0cbf36dd923c543c0041d6316b76fccb"}]},"branch":"refs/heads/master"},"65ba4ab026691a5d8600de63a73f31e66c6be34d":{"kind":"REWORK","_number":19,"created":"2016-12-08 16:08:11.000000000","uploader":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"ref":"refs/changes/24/391624/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone-specs","ref":"refs/changes/24/391624/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/24/391624/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/24/391624/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/24/391624/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone-specs refs/changes/24/391624/19"}}},"commit":{"parents":[{"commit":"603d1b52d31f7505cd1017f94b77cac7bb13a44a","subject":"Fix python version to 2.7 for docs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/603d1b52d31f7505cd1017f94b77cac7bb13a44a"}]}],"author":{"name":"Adam Young","email":"ayoung@redhat.com","date":"2016-10-29 21:24:09.000000000","tz":-240},"committer":{"name":"Adam Young","email":"ayoung@redhat.com","date":"2016-12-08 16:08:04.000000000","tz":-300},"subject":"Role Check from Middleware","message":"Role Check from Middleware\n\nVerify the roles for the API during the\nkeystone middleware call, after token\nvalidation but prior to handing control\nover to the remote service.\n\nAPIImpact:  adds new APIS for managing access\n            rules.  Adds RBAC check in all APIs.\n\nSecurityImpact: This change allows a greater\n degree of RBAC control, which should increase\n security\n\nChange-Id: I2487fe09091e9b054a91a9853b99f3dc14fbeebb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/65ba4ab026691a5d8600de63a73f31e66c6be34d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/65ba4ab026691a5d8600de63a73f31e66c6be34d"}]},"branch":"refs/heads/master"},"92533be8072af57a09c7b8515b20c021b6ac9b95":{"kind":"REWORK","_number":20,"created":"2016-12-09 22:06:09.000000000","uploader":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"ref":"refs/changes/24/391624/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone-specs","ref":"refs/changes/24/391624/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/24/391624/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/24/391624/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/24/391624/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone-specs refs/changes/24/391624/20"}}},"commit":{"parents":[{"commit":"fe41f03389a2f9a8337a61c3ba5fbe8d40b0d633","subject":"Fix a spelling mistake in a PCI-DSS spec","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/fe41f03389a2f9a8337a61c3ba5fbe8d40b0d633"}]}],"author":{"name":"Adam Young","email":"ayoung@redhat.com","date":"2016-10-29 21:24:09.000000000","tz":-240},"committer":{"name":"Adam Young","email":"ayoung@redhat.com","date":"2016-12-09 22:04:39.000000000","tz":-300},"subject":"Role Check from Middleware","message":"Role Check from Middleware\n\nVerify the roles for the API during the\nkeystone middleware call, after token\nvalidation but prior to handing control\nover to the remote service.\n\nAPIImpact:  adds new APIS for managing access\n            rules.  Adds RBAC check in all APIs.\n\nSecurityImpact: This change allows a greater\n degree of RBAC control, which should increase\n security\n\nChange-Id: I2487fe09091e9b054a91a9853b99f3dc14fbeebb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/92533be8072af57a09c7b8515b20c021b6ac9b95"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/92533be8072af57a09c7b8515b20c021b6ac9b95"}]},"branch":"refs/heads/master"},"79f3b56ffac5957b0223ef789aa9b096b88ee727":{"kind":"REWORK","_number":21,"created":"2016-12-12 23:10:08.000000000","uploader":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"ref":"refs/changes/24/391624/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone-specs","ref":"refs/changes/24/391624/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/24/391624/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/24/391624/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/24/391624/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone-specs refs/changes/24/391624/21"}}},"commit":{"parents":[{"commit":"fe41f03389a2f9a8337a61c3ba5fbe8d40b0d633","subject":"Fix a spelling mistake in a PCI-DSS spec","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/fe41f03389a2f9a8337a61c3ba5fbe8d40b0d633"}]}],"author":{"name":"Adam Young","email":"ayoung@redhat.com","date":"2016-10-29 21:24:09.000000000","tz":-240},"committer":{"name":"Adam Young","email":"ayoung@redhat.com","date":"2016-12-12 23:10:02.000000000","tz":-300},"subject":"Role Check from Middleware","message":"Role Check from Middleware\n\nVerify the roles for the API during the\nkeystone middleware call, after token\nvalidation but prior to handing control\nover to the remote service.\n\nAPIImpact:  adds new APIS for managing access\n            rules.  Adds RBAC check in all APIs.\n\nSecurityImpact: This change allows a greater\n degree of RBAC control, which should increase\n security\n\nChange-Id: I2487fe09091e9b054a91a9853b99f3dc14fbeebb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/79f3b56ffac5957b0223ef789aa9b096b88ee727"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/79f3b56ffac5957b0223ef789aa9b096b88ee727"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
