)]}'
{"id":"openstack%2Fkeystone-specs~456974","triplet_id":"openstack%2Fkeystone-specs~master~Ia84b788839fd82df04416249b8b30cba929c5ec0","project":"openstack/keystone-specs","branch":"master","topic":"452198","hashtags":[],"change_id":"Ia84b788839fd82df04416249b8b30cba929c5ec0","subject":"Role Check on Body Key","status":"ABANDONED","created":"2017-04-14 23:24:19.000000000","updated":"2020-05-19 17:10:45.000000000","total_comment_count":14,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"d017f09c8a0582874048b72a31493965a6065b42","_number":456974,"virtual_id_number":456974,"owner":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":10608,"name":"Matthew Edmonds","email":"edmondsw@us.ibm.com","username":"edmondsw"},{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"date":"2019-04-09 13:47:12.000000000","_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":13063,"name":"WEI CHEN","email":"dave.jungler@gmail.com","username":"wei.d.chen"},{"date":"2019-04-07 20:11:18.000000000","_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},{"value":1,"date":"2019-04-07 20:27:54.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"_account_id":7725,"name":"David Stanek","email":"dstanek@dstanek.com","username":"dstanek"},{"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"},{"_account_id":11022,"name":"Rodrigo Duarte Sousa","email":"rodrigodsousa@gmail.com","username":"rodrigods"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10608,"name":"Matthew Edmonds","email":"edmondsw@us.ibm.com","username":"edmondsw"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13063,"name":"WEI CHEN","email":"dave.jungler@gmail.com","username":"wei.d.chen"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7725,"name":"David Stanek","email":"dstanek@dstanek.com","username":"dstanek"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11022,"name":"Rodrigo Duarte Sousa","email":"rodrigodsousa@gmail.com","username":"rodrigods"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":10608,"name":"Matthew Edmonds","email":"edmondsw@us.ibm.com","username":"edmondsw"},{"value":0,"date":"2019-04-07 20:10:30.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"date":"2019-04-24 00:19:37.000000000","_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"date":"2019-04-07 20:10:30.000000000","_account_id":13063,"name":"WEI CHEN","email":"dave.jungler@gmail.com","username":"wei.d.chen"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2019-04-07 20:10:30.000000000","_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"date":"2019-04-07 20:10:30.000000000","_account_id":7725,"name":"David Stanek","email":"dstanek@dstanek.com","username":"dstanek"},{"date":"2019-04-07 20:10:30.000000000","_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"},{"date":"2019-04-07 20:10:30.000000000","_account_id":11022,"name":"Rodrigo Duarte Sousa","email":"rodrigodsousa@gmail.com","username":"rodrigods"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":7725,"name":"David Stanek","email":"dstanek@dstanek.com","username":"dstanek"},{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"_account_id":10608,"name":"Matthew Edmonds","email":"edmondsw@us.ibm.com","username":"edmondsw"},{"_account_id":11022,"name":"Rodrigo Duarte Sousa","email":"rodrigodsousa@gmail.com","username":"rodrigods"},{"_account_id":13063,"name":"WEI CHEN","email":"dave.jungler@gmail.com","username":"wei.d.chen"},{"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"},{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2017-05-31 17:06:33.000000000","updated_by":{"_account_id":10608,"name":"Matthew Edmonds","email":"edmondsw@us.ibm.com","username":"edmondsw"},"reviewer":{"_account_id":10608,"name":"Matthew Edmonds","email":"edmondsw@us.ibm.com","username":"edmondsw"},"state":"REVIEWER"},{"updated":"2019-04-07 20:10:30.000000000","updated_by":{"_account_id":7725,"name":"David Stanek","email":"dstanek@dstanek.com","username":"dstanek"},"reviewer":{"_account_id":7725,"name":"David Stanek","email":"dstanek@dstanek.com","username":"dstanek"},"state":"REVIEWER"},{"updated":"2019-04-07 20:10:30.000000000","updated_by":{"_account_id":11022,"name":"Rodrigo Duarte Sousa","email":"rodrigodsousa@gmail.com","username":"rodrigods"},"reviewer":{"_account_id":11022,"name":"Rodrigo Duarte Sousa","email":"rodrigodsousa@gmail.com","username":"rodrigods"},"state":"REVIEWER"},{"updated":"2019-04-07 20:10:30.000000000","updated_by":{"_account_id":13063,"name":"WEI CHEN","email":"dave.jungler@gmail.com","username":"wei.d.chen"},"reviewer":{"_account_id":13063,"name":"WEI CHEN","email":"dave.jungler@gmail.com","username":"wei.d.chen"},"state":"REVIEWER"},{"updated":"2019-04-07 20:10:30.000000000","updated_by":{"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"},"reviewer":{"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"},"state":"REVIEWER"},{"updated":"2019-04-07 20:10:30.000000000","updated_by":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"reviewer":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"state":"REVIEWER"},{"updated":"2019-04-07 20:10:30.000000000","updated_by":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"reviewer":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"state":"REVIEWER"},{"updated":"2019-04-07 20:27:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-04-09 13:47:12.000000000","updated_by":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"reviewer":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"state":"REVIEWER"},{"updated":"2019-04-24 00:19:37.000000000","updated_by":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"reviewer":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"state":"REVIEWER"}],"messages":[{"id":"da88dfb521a7300bf612d385ca6be05ffdaf8677","author":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"date":"2017-04-14 23:24:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9b7a172d047308c9c973db477e94b1e46cc37928","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-04-14 23:29:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-keystone-specs-docs-ubuntu-xenial http://docs-draft.openstack.org/74/456974/1/check/gate-keystone-specs-docs-ubuntu-xenial/44e699c//doc/build/html/ : SUCCESS in 5m 03s\n- gate-keystone-specs-python27-ubuntu-xenial http://logs.openstack.org/74/456974/1/check/gate-keystone-specs-python27-ubuntu-xenial/41fde29/ : SUCCESS in 2m 40s","accounts_in_message":[],"_revision_number":1},{"id":"c11a81a5e719f552a8ac8f48dd517b10087881d2","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-05-12 19:56:43.000000000","message":"Patch Set 1: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"65cf37f178d69bcdd8f39ffd6bc09cec17878d7a","author":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"date":"2017-05-16 19:01:33.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":1},{"id":"2849a56e46b8a3f43697b1136ab25c1194ccefa0","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-05-16 19:06:16.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-keystone-specs-docs-ubuntu-xenial http://logs.openstack.org/74/456974/2/check/gate-keystone-specs-docs-ubuntu-xenial/c38d059/ : FAILURE in 1m 43s\n- gate-keystone-specs-python27-ubuntu-xenial http://logs.openstack.org/74/456974/2/check/gate-keystone-specs-python27-ubuntu-xenial/4f7c063/ : SUCCESS in 3m 44s","accounts_in_message":[],"_revision_number":2},{"id":"abb5dffb324e0770665eb24062ee08efe4b0c805","author":{"_account_id":10608,"name":"Matthew Edmonds","email":"edmondsw@us.ibm.com","username":"edmondsw"},"date":"2017-05-31 17:06:33.000000000","message":"Patch Set 2: Code-Review-1\n\n(9 comments)","accounts_in_message":[],"_revision_number":2},{"id":"a7bafc1beb9d296dc6f3f4bff9357ad84a75ff51","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2017-05-31 19:49:42.000000000","message":"Patch Set 2:\n\nI have comments on the previous spec that might change the scope of this work.\n\nhttps://review.openstack.org/#/c/452198/8","accounts_in_message":[],"_revision_number":2},{"id":"3f291e5cf7094d15e0e4ca9d94c5cec20f2122ed","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2017-06-09 20:19:17.000000000","message":"Patch Set 2: Code-Review-2\n\nAs of pike-2 keystone has entered specification freeze [0]. If you\u0027d like to apply for an exception, please submit a request to the openstack-dev mailing list tagged with [keystone] and propose a discussion during keystone\u0027s weekly team meeting [1].\n\nPlacing a procedural -2 until Queens opens for development or an exception is granted.\n\n\n[0] https://releases.openstack.org/pike/schedule.html\n[1] https://etherpad.openstack.org/p/keystone-weekly-meeting","accounts_in_message":[],"_revision_number":2},{"id":"a7550804da46e5350ce68ada6821f74d7d0a8a07","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2017-12-18 03:16:18.000000000","message":"Removed reviewer Jenkins with the following votes:\n\n* Verified-1 by Jenkins (3)\n","accounts_in_message":[],"_revision_number":2},{"id":"8a6ce81862920b0082832b4ddd63d23ed1b169cc","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2018-08-21 11:48:41.000000000","message":"Patch Set 2: -Code-Review","accounts_in_message":[],"_revision_number":2},{"id":"4a6327c8671657e6ec09f2a6bb4e33c66ebdcf7b","author":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"date":"2019-04-07 20:10:11.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"ad296720fc717ed5b5c3aabbfa17b28e96dba2fc","author":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"date":"2019-04-07 20:11:18.000000000","message":"Patch Set 3:\n\nI\u0027ve repurposed this spec from the RBAC in middleware approach.  It applies equally as well to the App-creds approach.","accounts_in_message":[],"_revision_number":3},{"id":"134b3c9e4fe0062d8ee17d4b6de3541b590ce182","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-07 20:27:54.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/74/456974/3/check/openstack-tox-docs/1531729/html/ : SUCCESS in 4m 49s","accounts_in_message":[],"_revision_number":3},{"id":"3422fdb3e8f86241c424a2984e59bd505b3990da","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-04-08 21:04:18.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"9fd9f1690b434cc5c6b479d543d5d72f2934d23d","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2019-04-09 13:46:40.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"9356cc512a9ba2c2e571a93692071d27e5bc8c15","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2019-04-09 13:47:12.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"efbf0aed4ff0f9b961e4cd9f828826e9ae2569b0","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-04-24 00:19:37.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"5039ee7568e3c1305617b840c559a9b4147b3dd2","author":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"date":"2020-05-19 17:10:45.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"a5e8ad68fe22cea9039dfd8e00b75a845272d52c","revisions":{"fb8acd66c0875e1b5766448e4187252aee96ba86":{"kind":"REWORK","_number":1,"created":"2017-04-14 23:24:19.000000000","uploader":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"ref":"refs/changes/74/456974/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone-specs","ref":"refs/changes/74/456974/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/74/456974/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/74/456974/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/74/456974/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone-specs refs/changes/74/456974/1"}}},"commit":{"parents":[{"commit":"32ac556bc24c53349d4c18dd49c06e2250496f93","subject":"Commit to RBAC in middleware in Pike release","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/32ac556bc24c53349d4c18dd49c06e2250496f93"}]}],"author":{"name":"Adam Young","email":"ayoung@redhat.com","date":"2017-04-14 23:24:12.000000000","tz":-240},"committer":{"name":"Adam Young","email":"ayoung@redhat.com","date":"2017-04-14 23:24:12.000000000","tz":-240},"subject":"Role Check on Body Key","message":"Role Check on Body Key\n\nChange-Id: Ia84b788839fd82df04416249b8b30cba929c5ec0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/fb8acd66c0875e1b5766448e4187252aee96ba86"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/fb8acd66c0875e1b5766448e4187252aee96ba86"}]},"branch":"refs/heads/master"},"2ff7b52ca8fe88b9db9a6ed98d464b40bf66a86d":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2017-05-16 19:01:33.000000000","uploader":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"ref":"refs/changes/74/456974/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone-specs","ref":"refs/changes/74/456974/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/74/456974/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/74/456974/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/74/456974/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone-specs refs/changes/74/456974/2"}}},"commit":{"parents":[{"commit":"b1e1874865d4cf483b19d2d9946fc45d7ad41461","subject":"Commit to RBAC in middleware in Pike release","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/b1e1874865d4cf483b19d2d9946fc45d7ad41461"}]}],"author":{"name":"Adam Young","email":"ayoung@redhat.com","date":"2017-04-14 23:24:12.000000000","tz":-240},"committer":{"name":"ayoung","email":"ayoung@redhat.com","date":"2017-05-16 19:01:33.000000000","tz":0},"subject":"Role Check on Body Key","message":"Role Check on Body Key\n\nChange-Id: Ia84b788839fd82df04416249b8b30cba929c5ec0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/2ff7b52ca8fe88b9db9a6ed98d464b40bf66a86d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/2ff7b52ca8fe88b9db9a6ed98d464b40bf66a86d"}]},"branch":"refs/heads/master"},"a5e8ad68fe22cea9039dfd8e00b75a845272d52c":{"kind":"REWORK","_number":3,"created":"2019-04-07 20:10:11.000000000","uploader":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"ref":"refs/changes/74/456974/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone-specs","ref":"refs/changes/74/456974/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/74/456974/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/74/456974/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/74/456974/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone-specs refs/changes/74/456974/3"}}},"commit":{"parents":[{"commit":"e46f639d0af718d4fdd82eb8e90c220ae08722c5","subject":"Replace openstack.org git:// URLs with https://","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/e46f639d0af718d4fdd82eb8e90c220ae08722c5"}]}],"author":{"name":"Adam Young","email":"ayoung@redhat.com","date":"2017-04-14 23:24:12.000000000","tz":-240},"committer":{"name":"Adam Young","email":"ayoung@redhat.com","date":"2019-04-07 20:09:59.000000000","tz":-240},"subject":"Role Check on Body Key","message":"Role Check on Body Key\n\nChange-Id: Ia84b788839fd82df04416249b8b30cba929c5ec0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/a5e8ad68fe22cea9039dfd8e00b75a845272d52c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/a5e8ad68fe22cea9039dfd8e00b75a845272d52c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
