)]}'
{"id":"openstack%2Fkeystone-specs~575144","triplet_id":"openstack%2Fkeystone-specs~master~Ie5cfd122554ccb06be3a7b165209c6b9c3f453db","project":"openstack/keystone-specs","branch":"master","topic":"575144","hashtags":[],"change_id":"Ie5cfd122554ccb06be3a7b165209c6b9c3f453db","subject":"Add role implication note to basic-default-roles","status":"MERGED","created":"2018-06-13 15:14:04.000000000","updated":"2019-04-26 21:49:56.000000000","submitted":"2019-04-26 21:49:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"575144-1556315396659-99a4d632","meta_rev_id":"d8b734b8a2c266754412679ee4d7555af3c1c349","_number":575144,"virtual_id_number":575144,"owner":{"_account_id":11589,"name":"Harry Rybacki","email":"hrybacki@redhat.com","username":"hrybacki"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":11589,"name":"Harry Rybacki","email":"hrybacki@redhat.com","username":"hrybacki"},{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":2,"date":"2019-04-26 21:49:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11022,"name":"Rodrigo Duarte Sousa","email":"rodrigodsousa@gmail.com","username":"rodrigods"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"all":[{"value":2,"date":"2019-04-24 01:46:47.000000000","_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":1,"date":"2019-04-24 18:04:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11589,"name":"Harry Rybacki","email":"hrybacki@redhat.com","username":"hrybacki"},{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-04-26 21:42:32.000000000","_account_id":11022,"name":"Rodrigo Duarte Sousa","email":"rodrigodsousa@gmail.com","username":"rodrigods"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11022,"name":"Rodrigo Duarte Sousa","email":"rodrigodsousa@gmail.com","username":"rodrigods"},"all":[{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":11589,"name":"Harry Rybacki","email":"hrybacki@redhat.com","username":"hrybacki"},{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-04-26 21:42:32.000000000","_account_id":11022,"name":"Rodrigo Duarte Sousa","email":"rodrigodsousa@gmail.com","username":"rodrigods"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"_account_id":11022,"name":"Rodrigo Duarte Sousa","email":"rodrigodsousa@gmail.com","username":"rodrigods"},{"_account_id":11589,"name":"Harry Rybacki","email":"hrybacki@redhat.com","username":"hrybacki"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-08-21 10:40:44.000000000","updated_by":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"reviewer":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"state":"REVIEWER"},{"updated":"2019-04-24 01:46:47.000000000","updated_by":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"reviewer":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"state":"REVIEWER"},{"updated":"2019-04-26 21:42:32.000000000","updated_by":{"_account_id":11022,"name":"Rodrigo Duarte Sousa","email":"rodrigodsousa@gmail.com","username":"rodrigods"},"reviewer":{"_account_id":11022,"name":"Rodrigo Duarte Sousa","email":"rodrigodsousa@gmail.com","username":"rodrigods"},"state":"REVIEWER"},{"updated":"2019-04-26 21:49:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9355a506cad52ab290f6296d9cb6d0f606ba22b2","author":{"_account_id":11589,"name":"Harry Rybacki","email":"hrybacki@redhat.com","username":"hrybacki"},"date":"2018-06-13 15:14:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4346c4919942bfdf1a95ff65f27818e328bc6a41","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-06-13 16:51:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/44/575144/1/check/build-openstack-sphinx-docs/89dc0f4/html/ : SUCCESS in 5m 43s","accounts_in_message":[],"_revision_number":1},{"id":"69b723d8e397d0fd569f6bdcf9bf0e5567e595bb","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2018-08-21 10:40:44.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ce126c88708664d2eb463b49ee67476e60d070b7","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-04-24 01:46:34.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"069a3835f34e468cd4f2a80206ec6254ce078499","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-04-24 01:46:47.000000000","message":"Patch Set 2: Code-Review+2\n\nrebased","accounts_in_message":[],"_revision_number":2},{"id":"a987fdfa82b23d8fb78b3c5519b84443c7374215","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 01:54:52.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/44/575144/2/check/openstack-tox-docs/ac12273/html/ : SUCCESS in 4m 42s","accounts_in_message":[],"_revision_number":2},{"id":"250f3aafdec1d19c8d566967fc55e7ed6abd48b4","author":{"_account_id":11589,"name":"Harry Rybacki","email":"hrybacki@redhat.com","username":"hrybacki"},"date":"2019-04-24 18:04:51.000000000","message":"Patch Set 2: Code-Review+1\n\nGood addition.","accounts_in_message":[],"_revision_number":2},{"id":"3568a7adfe47e803dff497bd89f3af5378562300","author":{"_account_id":11022,"name":"Rodrigo Duarte Sousa","email":"rodrigodsousa@gmail.com","username":"rodrigods"},"date":"2019-04-26 21:42:32.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"94ca984419226d9a34c00425784279a8901dacb4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-26 21:42:45.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c21bda3522affee6405116bfdd47d321a590f2c8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-26 21:49:56.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"920caf51d6344b33cfc591f0704ea0b6a47d8b3a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-26 21:49:56.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/44/575144/2/gate/openstack-tox-docs/fa1a16a/html/ : SUCCESS in 5m 53s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"fa3479f3a3ed2c864196313f3b76efe83ca327a8","revisions":{"ecfbaad3ef45e968c2020b1e925f141f06f23e0c":{"kind":"REWORK","_number":1,"created":"2018-06-13 15:14:04.000000000","uploader":{"_account_id":11589,"name":"Harry Rybacki","email":"hrybacki@redhat.com","username":"hrybacki"},"ref":"refs/changes/44/575144/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone-specs","ref":"refs/changes/44/575144/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/44/575144/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/44/575144/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/44/575144/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone-specs refs/changes/44/575144/1"}}},"commit":{"parents":[{"commit":"0dd3d3f6e62a13e10a084cebb936129a0ef1ffc6","subject":"Follow-up -- replace \u0027auditor\u0027 role with \u0027reader\u0027","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/0dd3d3f6e62a13e10a084cebb936129a0ef1ffc6"}]}],"author":{"name":"Harry Rybacki","email":"hrybacki@redhat.com","date":"2018-06-13 15:12:18.000000000","tz":-240},"committer":{"name":"Harry Rybacki","email":"hrybacki@redhat.com","date":"2018-06-13 15:12:18.000000000","tz":-240},"subject":"Add role implication note to basic-default-roles","message":"Add role implication note to basic-default-roles\n\nExpanding upon the Risk Mitigation section of the spec. Note that\nrole implications will be created admin-\u003emember-\u003ereader regardless of\nwhether or not a new role was created during bootstrap.\n\nChange-Id: Ie5cfd122554ccb06be3a7b165209c6b9c3f453db\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/ecfbaad3ef45e968c2020b1e925f141f06f23e0c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/ecfbaad3ef45e968c2020b1e925f141f06f23e0c"}]},"branch":"refs/heads/master"},"fa3479f3a3ed2c864196313f3b76efe83ca327a8":{"kind":"REWORK","_number":2,"created":"2019-04-24 01:46:34.000000000","uploader":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"ref":"refs/changes/44/575144/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone-specs","ref":"refs/changes/44/575144/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/44/575144/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/44/575144/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/44/575144/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone-specs refs/changes/44/575144/2"}}},"commit":{"parents":[{"commit":"357c3706aa952bf4c8e69ac0e5e92d6d93a425d8","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/357c3706aa952bf4c8e69ac0e5e92d6d93a425d8"}]}],"author":{"name":"Harry Rybacki","email":"hrybacki@redhat.com","date":"2018-06-13 15:12:18.000000000","tz":-240},"committer":{"name":"Colleen Murphy","email":"colleen@gazlene.net","date":"2019-04-24 01:46:27.000000000","tz":-420},"subject":"Add role implication note to basic-default-roles","message":"Add role implication note to basic-default-roles\n\nExpanding upon the Risk Mitigation section of the spec. Note that\nrole implications will be created admin-\u003emember-\u003ereader regardless of\nwhether or not a new role was created during bootstrap.\n\nChange-Id: Ie5cfd122554ccb06be3a7b165209c6b9c3f453db\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/fa3479f3a3ed2c864196313f3b76efe83ca327a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/fa3479f3a3ed2c864196313f3b76efe83ca327a8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
