)]}'
{"id":"openstack%2Fkeystone-specs~657129","triplet_id":"openstack%2Fkeystone-specs~master~I3e1ab025bb998b14c0a71854b9109d9f29b25ee9","project":"openstack/keystone-specs","branch":"master","hashtags":[],"change_id":"I3e1ab025bb998b14c0a71854b9109d9f29b25ee9","subject":"Move the request-helpers spec for keystonemiddleware to attic","status":"MERGED","created":"2019-05-04 16:24:15.000000000","updated":"2019-05-13 17:31:57.000000000","submitted":"2019-05-13 17:31:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"657129-1557768717310-d3510572","meta_rev_id":"4f42438bd0b7fb17658a9e93fda19765118195e8","_number":657129,"virtual_id_number":657129,"owner":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":2,"date":"2019-05-13 17:31:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"all":[{"value":2,"date":"2019-05-04 17:59:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":2,"date":"2019-05-13 17:25:58.000000000","_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"all":[{"value":0,"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":1,"date":"2019-05-13 17:25:58.000000000","_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-04 17:59:30.000000000","updated_by":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"reviewer":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"state":"REVIEWER"},{"updated":"2019-05-13 17:25:58.000000000","updated_by":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"reviewer":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"state":"REVIEWER"},{"updated":"2019-05-13 17:31:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9ee4f4ef6e6b36d2e9e6231384287aae90b81ebb","author":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"date":"2019-05-04 16:24:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7378d718e9df7abe8141fcf4881f3db15f1a26ba","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-04 16:29:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/29/657129/1/check/openstack-tox-docs/7c5556c/ : FAILURE in 4m 46s","accounts_in_message":[],"_revision_number":1},{"id":"2e6e4f6ca2125527e10572d77fc5611e0a4937c2","author":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"date":"2019-05-04 16:47:01.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5fcfbbdab18942197317bc14baa811eb5c2b163f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-04 16:56:11.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/29/657129/2/check/openstack-tox-docs/a2fc170/html/ : SUCCESS in 4m 12s","accounts_in_message":[],"_revision_number":2},{"id":"a0955ff7e19e3e8ea995484ec26a1493a4e4eaed","author":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"date":"2019-05-04 17:59:30.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"11056c3379e51f2352e3d45a73286b2f589a894b","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-05-13 17:25:58.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"37f9ea7c8eae510b6d7ce39d157fd9a98c4cdc24","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-13 17:26:10.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"df1fe1a3b2db2e0d594d44233406e211d85483f6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-13 17:31:57.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/29/657129/2/gate/openstack-tox-docs/2a2f308/html/ : SUCCESS in 5m 07s","accounts_in_message":[],"_revision_number":2},{"id":"4f42438bd0b7fb17658a9e93fda19765118195e8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-13 17:31:57.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d41918f3e9a0c82e38371b60276864b33be8d8a6","revisions":{"17476dccd5ad6f142d3a2ea82d22fc6d8881b116":{"kind":"REWORK","_number":1,"created":"2019-05-04 16:24:15.000000000","uploader":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"ref":"refs/changes/29/657129/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone-specs","ref":"refs/changes/29/657129/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/29/657129/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/29/657129/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/29/657129/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone-specs refs/changes/29/657129/1"}}},"commit":{"parents":[{"commit":"256c23c9d0c0368860d2b3ff98b8fddcfb4b4c23","subject":"Move endpoint-enforcement-middleware spec to attic","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/256c23c9d0c0368860d2b3ff98b8fddcfb4b4c23"}]}],"author":{"name":"morgan fainberg","email":"morgan.fainberg@gmail.com","date":"2019-05-04 16:19:15.000000000","tz":-420},"committer":{"name":"morgan fainberg","email":"morgan.fainberg@gmail.com","date":"2019-05-04 16:19:15.000000000","tz":-420},"subject":"Move the request-helpers spec for keystonemiddleware to attic","message":"Move the request-helpers spec for keystonemiddleware to attic\n\nMove the request-helpers backlog spec for keystonemiddleware to the\nattic. At the Denver PTG (2019) we discussed this spec. We are in a\nvery different space from where we were at the time of proposal, and\nif there is a desire to revisit this specific specification it can\nbe brought back from the attic.\n\nChange-Id: I3e1ab025bb998b14c0a71854b9109d9f29b25ee9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/17476dccd5ad6f142d3a2ea82d22fc6d8881b116"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/17476dccd5ad6f142d3a2ea82d22fc6d8881b116"}]},"branch":"refs/heads/master"},"d41918f3e9a0c82e38371b60276864b33be8d8a6":{"kind":"REWORK","_number":2,"created":"2019-05-04 16:47:01.000000000","uploader":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"ref":"refs/changes/29/657129/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone-specs","ref":"refs/changes/29/657129/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/29/657129/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/29/657129/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/29/657129/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone-specs refs/changes/29/657129/2"}}},"commit":{"parents":[{"commit":"256c23c9d0c0368860d2b3ff98b8fddcfb4b4c23","subject":"Move endpoint-enforcement-middleware spec to attic","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/256c23c9d0c0368860d2b3ff98b8fddcfb4b4c23"}]}],"author":{"name":"morgan fainberg","email":"morgan.fainberg@gmail.com","date":"2019-05-04 16:19:15.000000000","tz":-420},"committer":{"name":"morgan fainberg","email":"morgan.fainberg@gmail.com","date":"2019-05-04 16:46:46.000000000","tz":-420},"subject":"Move the request-helpers spec for keystonemiddleware to attic","message":"Move the request-helpers spec for keystonemiddleware to attic\n\nMove the request-helpers backlog spec for keystonemiddleware to the\nattic. At the Denver PTG (2019) we discussed this spec. We are in a\nvery different space from where we were at the time of proposal, and\nif there is a desire to revisit this specific specification it can\nbe brought back from the attic.\n\nChange-Id: I3e1ab025bb998b14c0a71854b9109d9f29b25ee9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/d41918f3e9a0c82e38371b60276864b33be8d8a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/d41918f3e9a0c82e38371b60276864b33be8d8a6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
