)]}'
{"id":"openstack%2Fkeystone-specs~657133","triplet_id":"openstack%2Fkeystone-specs~master~I238b16a30f131bf9d6a754c4dda48ac8e83a51b0","project":"openstack/keystone-specs","branch":"master","hashtags":[],"change_id":"I238b16a30f131bf9d6a754c4dda48ac8e83a51b0","subject":"Move \u0027functional testing\u0027 spec to attic","status":"MERGED","created":"2019-05-04 16:33:29.000000000","updated":"2019-05-13 17:44:16.000000000","submitted":"2019-05-13 17:44:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"657133-1557769456255-72f24b91","meta_rev_id":"eef27783000be9a911ecd24b619ca3cb3c7ba712","_number":657133,"virtual_id_number":657133,"owner":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":2,"date":"2019-05-13 17:44:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"all":[{"value":2,"date":"2019-05-04 17:58:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":2,"date":"2019-05-13 17:29:34.000000000","_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"all":[{"value":0,"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":1,"date":"2019-05-13 17:29:34.000000000","_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-04 17:58:43.000000000","updated_by":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"reviewer":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"state":"REVIEWER"},{"updated":"2019-05-13 17:29:34.000000000","updated_by":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"reviewer":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"state":"REVIEWER"},{"updated":"2019-05-13 17:44:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"629fa20ed43bc4f0d917a0370a274a27bb4c8439","author":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"date":"2019-05-04 16:33:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1c87ebb2fa2f0c46e46adee79b445a7f7413cb0b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-04 16:40:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/33/657133/1/check/openstack-tox-docs/6727ac3/html/ : SUCCESS in 4m 56s","accounts_in_message":[],"_revision_number":1},{"id":"2509aba851510ea0c2dbe225a55041b592e8f4b9","author":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"date":"2019-05-04 17:58:43.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"24da90d3d3e1eab3fcde7a68df16d4f8c40b8343","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-05-13 17:29:34.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e359d82d0124aac7481d3a47e4dd099b209fdd49","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-13 17:29:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3981d66261607b53fb73077683116303c4fce7ca","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-13 17:44:16.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"eef27783000be9a911ecd24b619ca3cb3c7ba712","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-13 17:44:16.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/33/657133/1/gate/openstack-tox-docs/9b15c83/html/ : SUCCESS in 5m 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"60e79f7c5a396d42a66547f8803dfbd14b0fd5dc","revisions":{"60e79f7c5a396d42a66547f8803dfbd14b0fd5dc":{"kind":"REWORK","_number":1,"created":"2019-05-04 16:33:29.000000000","uploader":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"ref":"refs/changes/33/657133/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone-specs","ref":"refs/changes/33/657133/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/33/657133/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/33/657133/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/33/657133/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone-specs refs/changes/33/657133/1"}}},"commit":{"parents":[{"commit":"fa3479f3a3ed2c864196313f3b76efe83ca327a8","subject":"Add role implication note to basic-default-roles","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/fa3479f3a3ed2c864196313f3b76efe83ca327a8"}]}],"author":{"name":"morgan fainberg","email":"morgan.fainberg@gmail.com","date":"2019-05-04 16:31:44.000000000","tz":-420},"committer":{"name":"morgan fainberg","email":"morgan.fainberg@gmail.com","date":"2019-05-04 16:31:44.000000000","tz":-420},"subject":"Move \u0027functional testing\u0027 spec to attic","message":"Move \u0027functional testing\u0027 spec to attic\n\nAs discussed at the Denver (2019) PTG, this spec is not super useful as\nproposed. We have started work to improve our testing in a number of\nways. If the specific use-case of functional testing as proposed in the\nspec receives interest the spec can be retrieved from the attic.\n\nChange-Id: I238b16a30f131bf9d6a754c4dda48ac8e83a51b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/60e79f7c5a396d42a66547f8803dfbd14b0fd5dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/60e79f7c5a396d42a66547f8803dfbd14b0fd5dc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
