)]}'
{"id":"openstack%2Fkeystone-specs~661784","triplet_id":"openstack%2Fkeystone-specs~master~I58b1b57febaea7f8bc75bf9e737e2d42f7823cf6","project":"openstack/keystone-specs","branch":"master","topic":"access-rules","hashtags":[],"change_id":"I58b1b57febaea7f8bc75bf9e737e2d42f7823cf6","subject":"Update access rules spec with decisions from PTG","status":"MERGED","created":"2019-05-28 15:25:14.000000000","updated":"2019-07-02 17:38:38.000000000","submitted":"2019-07-02 17:38:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":10,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"661784-1562089118616-3639b714","meta_rev_id":"8b5c09314ea3bde9c2145f3e3e37cfb26f588a4f","_number":661784,"virtual_id_number":661784,"owner":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-06-26 17:47:40.000000000","_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"value":2,"date":"2019-07-02 17:38:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"all":[{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":2,"date":"2019-06-26 15:23:19.000000000","_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":2,"date":"2019-07-02 17:32:22.000000000","_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"all":[{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":1,"date":"2019-07-02 17:32:22.000000000","_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-26 15:23:19.000000000","updated_by":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"reviewer":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"state":"REVIEWER"},{"updated":"2019-07-02 17:32:22.000000000","updated_by":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"reviewer":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"state":"REVIEWER"},{"updated":"2019-07-02 17:38:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1435f5d47c6bdab07ca53f5238eb2aba41dbba3e","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-05-28 15:25:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"47e05b2b117f24c2ac81260ca67764a61d57c7b8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-28 15:36:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/84/661784/1/check/openstack-tox-docs/fd2bb15/html/ : SUCCESS in 5m 32s","accounts_in_message":[],"_revision_number":1},{"id":"27db41aae32bcaa1e7c71e5805c5996b7e2eed24","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-05-28 15:41:53.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"953f772d3ebecb8638aa01a8969789400659a860","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-28 15:51:46.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/84/661784/2/check/openstack-tox-docs/5bb6b5e/html/ : SUCCESS in 4m 50s","accounts_in_message":[],"_revision_number":2},{"id":"4a00082d079ba6caeb6a652ff94e01a6eb15351b","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2019-05-29 13:42:51.000000000","message":"Patch Set 2: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"9460b20e671e8c8f42372d2ead6f7024d841cfb4","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-05-29 17:14:33.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"140df632fb3a71dced37fb5d0d25c3665681bd1b","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-05-29 17:45:28.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"73d1281dc74ab199b3acba1eab47442cc70dbb2b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-29 17:57:11.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/84/661784/3/check/openstack-tox-docs/fa1b4dd/html/ : SUCCESS in 4m 19s","accounts_in_message":[],"_revision_number":3},{"id":"290911e7d9e66dcac475b005dcfe2b971ad11f90","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-06-05 20:15:51.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"20d3abb5925401f5cc0f6c45648e5d9df040b3d7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-05 20:26:12.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/84/661784/4/check/openstack-tox-docs/c2b23e7/html/ : SUCCESS in 4m 26s","accounts_in_message":[],"_revision_number":4},{"id":"51da9f3d110a9635b8cb07259daee934b8a8574b","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2019-06-26 15:23:19.000000000","message":"Patch Set 4: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"3dc75a3fac57c0ac7f5d44485678719ecef361b6","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2019-06-26 15:24:04.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"e25b675d472b415cc18268429094a890e51a1fd3","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-06-26 17:47:40.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"88eaa0225f53f8bcc8673ef5c9be2daa5b03c730","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2019-06-26 19:14:58.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"e69431ba5c2dba28dd233ad7769c6ac91ab2b92d","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2019-07-02 17:32:22.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"8d84725b7a6830e434119bc41f060a54288533b2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-02 17:32:33.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"a00bec1418ced3d6b56481d1f19b2fb11cf55506","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-02 17:38:38.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/84/661784/4/gate/openstack-tox-docs/b5caa75/html/ : SUCCESS in 4m 14s","accounts_in_message":[],"_revision_number":4},{"id":"2cb5df499f3eb8967f025d906461f08a87563dea","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-02 17:38:38.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"94df8711bc838cb28512e1ad3fc9b8d8480cd87d","revisions":{"5ea9619e34d9b08f2d07a8fac7620e65638f295a":{"kind":"REWORK","_number":1,"created":"2019-05-28 15:25:14.000000000","uploader":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"ref":"refs/changes/84/661784/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone-specs","ref":"refs/changes/84/661784/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/84/661784/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/84/661784/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/84/661784/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone-specs refs/changes/84/661784/1"}}},"commit":{"parents":[{"commit":"7837c2031eebaa060c960542c9c5b1d5bc05064d","subject":"Merge \"Repropose federated attributes in the user API for Train\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/7837c2031eebaa060c960542c9c5b1d5bc05064d"}]}],"author":{"name":"Colleen Murphy","email":"colleen@gazlene.net","date":"2019-05-28 15:24:40.000000000","tz":-420},"committer":{"name":"Colleen Murphy","email":"colleen@gazlene.net","date":"2019-05-28 15:24:40.000000000","tz":-420},"subject":"Update access rules spec with decisions from PTG","message":"Update access rules spec with decisions from PTG\n\nChange-Id: I58b1b57febaea7f8bc75bf9e737e2d42f7823cf6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/5ea9619e34d9b08f2d07a8fac7620e65638f295a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/5ea9619e34d9b08f2d07a8fac7620e65638f295a"}]},"branch":"refs/heads/master"},"84dd128c1af24b1897b838f47560bb8272ef0515":{"kind":"REWORK","_number":2,"created":"2019-05-28 15:41:53.000000000","uploader":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"ref":"refs/changes/84/661784/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone-specs","ref":"refs/changes/84/661784/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/84/661784/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/84/661784/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/84/661784/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone-specs refs/changes/84/661784/2"}}},"commit":{"parents":[{"commit":"7837c2031eebaa060c960542c9c5b1d5bc05064d","subject":"Merge \"Repropose federated attributes in the user API for Train\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/7837c2031eebaa060c960542c9c5b1d5bc05064d"}]}],"author":{"name":"Colleen Murphy","email":"colleen@gazlene.net","date":"2019-05-28 15:24:40.000000000","tz":-420},"committer":{"name":"Colleen Murphy","email":"colleen@gazlene.net","date":"2019-05-28 15:41:38.000000000","tz":-420},"subject":"Update access rules spec with decisions from PTG","message":"Update access rules spec with decisions from PTG\n\nChange-Id: I58b1b57febaea7f8bc75bf9e737e2d42f7823cf6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/84dd128c1af24b1897b838f47560bb8272ef0515"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/84dd128c1af24b1897b838f47560bb8272ef0515"}]},"branch":"refs/heads/master"},"bf460ff65ff165f8cf40f97a26b9128c7066f4f6":{"kind":"REWORK","_number":3,"created":"2019-05-29 17:45:28.000000000","uploader":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"ref":"refs/changes/84/661784/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone-specs","ref":"refs/changes/84/661784/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/84/661784/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/84/661784/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/84/661784/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone-specs refs/changes/84/661784/3"}}},"commit":{"parents":[{"commit":"7837c2031eebaa060c960542c9c5b1d5bc05064d","subject":"Merge \"Repropose federated attributes in the user API for Train\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/7837c2031eebaa060c960542c9c5b1d5bc05064d"}]}],"author":{"name":"Colleen Murphy","email":"colleen@gazlene.net","date":"2019-05-28 15:24:40.000000000","tz":-420},"committer":{"name":"Colleen Murphy","email":"colleen@gazlene.net","date":"2019-05-29 17:45:12.000000000","tz":-420},"subject":"Update access rules spec with decisions from PTG","message":"Update access rules spec with decisions from PTG\n\nChange-Id: I58b1b57febaea7f8bc75bf9e737e2d42f7823cf6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/bf460ff65ff165f8cf40f97a26b9128c7066f4f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/bf460ff65ff165f8cf40f97a26b9128c7066f4f6"}]},"branch":"refs/heads/master"},"94df8711bc838cb28512e1ad3fc9b8d8480cd87d":{"kind":"REWORK","_number":4,"created":"2019-06-05 20:15:51.000000000","uploader":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"ref":"refs/changes/84/661784/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone-specs","ref":"refs/changes/84/661784/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/84/661784/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/84/661784/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/84/661784/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone-specs refs/changes/84/661784/4"}}},"commit":{"parents":[{"commit":"7837c2031eebaa060c960542c9c5b1d5bc05064d","subject":"Merge \"Repropose federated attributes in the user API for Train\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/7837c2031eebaa060c960542c9c5b1d5bc05064d"}]}],"author":{"name":"Colleen Murphy","email":"colleen@gazlene.net","date":"2019-05-28 15:24:40.000000000","tz":-420},"committer":{"name":"Colleen Murphy","email":"colleen@gazlene.net","date":"2019-06-05 20:15:40.000000000","tz":-420},"subject":"Update access rules spec with decisions from PTG","message":"Update access rules spec with decisions from PTG\n\nChange-Id: I58b1b57febaea7f8bc75bf9e737e2d42f7823cf6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/94df8711bc838cb28512e1ad3fc9b8d8480cd87d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/94df8711bc838cb28512e1ad3fc9b8d8480cd87d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
