)]}'
{"id":"openstack%2Fkeystone-specs~672164","triplet_id":"openstack%2Fkeystone-specs~master~I1c24cece2c64c9453698280cc365ac150d2474a4","project":"openstack/keystone-specs","branch":"master","topic":"python3-train","hashtags":[],"change_id":"I1c24cece2c64c9453698280cc365ac150d2474a4","subject":"Correct style errors","status":"MERGED","created":"2019-07-22 20:01:05.000000000","updated":"2019-07-22 20:30:48.000000000","submitted":"2019-07-22 20:30:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"672164-1563827448865-31ef7618","meta_rev_id":"879dfb878321059b9d7a7b7fd032bf426bef35ff","_number":672164,"virtual_id_number":672164,"owner":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},{"value":2,"date":"2019-07-22 20:30:48.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"all":[{"value":2,"date":"2019-07-22 20:22:17.000000000","_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"all":[{"value":1,"date":"2019-07-22 20:24:54.000000000","_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-22 20:24:54.000000000","updated_by":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"reviewer":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"state":"REVIEWER"},{"updated":"2019-07-22 20:30:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9921699edc92639c8d5e1454ca63e62ef897e2d0","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-07-22 20:01:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3bd3ff0d6ec4dc1b85262d9a41f8c8b1d68bbe75","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-22 20:10:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/64/672164/1/check/openstack-tox-docs/eab14a4/html/ : SUCCESS in 3m 09s","accounts_in_message":[],"_revision_number":1},{"id":"441cb2dcfd2dfd635fcfa792c47cdd473b585da1","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-07-22 20:11:40.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"77ada96dd3164584eb39a9fc5f3d6e810d96db98","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-22 20:20:58.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/64/672164/2/check/openstack-tox-docs/a815843/html/ : SUCCESS in 7m 05s","accounts_in_message":[],"_revision_number":2},{"id":"8055c0eabc23c95463c207cb2bb21745dc9f6310","author":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"date":"2019-07-22 20:22:17.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"ede845b792f19bdf751d9cc141818ee5226f0db5","author":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"date":"2019-07-22 20:24:54.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"5a8a8db76f27e32f5cd35f377261b777a161d86c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-22 20:25:05.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"86b7fc399f92579a516bdb32a16ed0d0185f3bbd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-22 20:30:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"d3249cf1221fcf03425172c27bba4a3b47850fee","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-22 20:30:48.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/64/672164/2/gate/openstack-tox-docs/245b1bb/html/ : SUCCESS in 3m 48s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"4f1bf7ddefe6486e37f9bff463a8335a2fa90c06","revisions":{"5c1f64597c3ab3eb28b9ccb69576849e085b4e53":{"kind":"REWORK","_number":1,"created":"2019-07-22 20:01:05.000000000","uploader":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"ref":"refs/changes/64/672164/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone-specs","ref":"refs/changes/64/672164/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/64/672164/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/64/672164/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/64/672164/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone-specs refs/changes/64/672164/1"}}},"commit":{"parents":[{"commit":"1e9ae94efb948c92a17a34b2700b99e1ea79b35a","subject":"Use upper-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/1e9ae94efb948c92a17a34b2700b99e1ea79b35a"}]}],"author":{"name":"Colleen Murphy","email":"colleen.murphy@suse.de","date":"2019-07-22 19:58:15.000000000","tz":-420},"committer":{"name":"Colleen Murphy","email":"colleen.murphy@suse.de","date":"2019-07-22 20:01:00.000000000","tz":-420},"subject":"Correct style errors","message":"Correct style errors\n\nThese style errors weren\u0027t caught before the specs merged because the\nlinter jobs weren\u0027t being run when only RST files were changed. Correct\nthem now so that a later patch can update the jobs.\n\nChange-Id: I1c24cece2c64c9453698280cc365ac150d2474a4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/5c1f64597c3ab3eb28b9ccb69576849e085b4e53"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/5c1f64597c3ab3eb28b9ccb69576849e085b4e53"}]},"branch":"refs/heads/master"},"4f1bf7ddefe6486e37f9bff463a8335a2fa90c06":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-07-22 20:11:40.000000000","uploader":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"ref":"refs/changes/64/672164/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone-specs","ref":"refs/changes/64/672164/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/64/672164/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/64/672164/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/64/672164/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone-specs refs/changes/64/672164/2"}}},"commit":{"parents":[{"commit":"5aab0c140e7a067b884e0e8ffe1b62ad2e543c55","subject":"Merge \"Expiring Group Membership Through Mapping Rules\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/5aab0c140e7a067b884e0e8ffe1b62ad2e543c55"}]}],"author":{"name":"Colleen Murphy","email":"colleen.murphy@suse.de","date":"2019-07-22 19:58:15.000000000","tz":-420},"committer":{"name":"Colleen Murphy","email":"colleen.murphy@suse.de","date":"2019-07-22 20:11:36.000000000","tz":-420},"subject":"Correct style errors","message":"Correct style errors\n\nThese style errors weren\u0027t caught before the specs merged because the\nlinter jobs weren\u0027t being run when only RST files were changed. Correct\nthem now so that a later patch can update the jobs.\n\nChange-Id: I1c24cece2c64c9453698280cc365ac150d2474a4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/4f1bf7ddefe6486e37f9bff463a8335a2fa90c06"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/4f1bf7ddefe6486e37f9bff463a8335a2fa90c06"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
