)]}'
{"id":"openstack%2Fkeystone-specs~692345","triplet_id":"openstack%2Fkeystone-specs~master~Ia4b92e1b34344bf79626058264f35a328d55c6c5","project":"openstack/keystone-specs","branch":"master","hashtags":[],"change_id":"Ia4b92e1b34344bf79626058264f35a328d55c6c5","subject":"Switch to Ussuri jobs","status":"ABANDONED","created":"2019-10-31 11:34:58.000000000","updated":"2020-01-16 02:22:46.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"34fe4b13cad240ab4512b8570b6eb566cc516dec","_number":692345,"virtual_id_number":692345,"owner":{"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2019-10-31 11:45:14.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"all":[{"value":-1,"date":"2019-12-05 11:33:22.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-31 11:45:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-12-05 11:33:22.000000000","updated_by":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"reviewer":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"state":"REVIEWER"}],"messages":[{"id":"bc0a10737bd068064916d23384afbf477b425172","author":{"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},"date":"2019-10-31 11:34:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d0dc26ebbb9d9fd3c9a10029bc48ab732829eb8e","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2019-10-31 11:39:05.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"db59e3319bf93c80389a80265b1c309ee8bfc520","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-31 11:45:14.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8627e3e7020143949840df4da48d23e6 : SUCCESS in 4m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/600cea9e920a41fcbce19af6e7de6b1e : FAILURE in 2m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/433d1a1c1bf14ee5924096d5a2aa0a4e : SUCCESS in 3m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5018bc9675a34caeabfc06bffe5da41c : SUCCESS in 3m 19s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/20050f953cec4d3a98920983f6ecd425 : SUCCESS in 2m 51s","accounts_in_message":[],"_revision_number":1},{"id":"d43cc74023adcfbbfa0befe6f5f12b2eac45bcf3","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2019-12-04 05:49:32.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"3c2e12647ebda489b5941333c0bce0bbf84998aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-04 05:54:16.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66eb843208df4cfb8db7830a922b212d : SUCCESS in 4m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6936cdbfed124b53b069c85fc0046271 : FAILURE in 1m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/50f50de60ebe444fa9d4fbf44ec559a3 : SUCCESS in 2m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ef6420e3b0c345fca3db6fd1fbf4c93e : SUCCESS in 3m 12s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7098a2fd19ee43ceb108b358add77655 : SUCCESS in 2m 33s","accounts_in_message":[],"_revision_number":1},{"id":"7ba64e25543ec0f21288ebfebae3b2663c4744b6","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2019-12-05 11:33:22.000000000","message":"Patch Set 1: Code-Review-1\n\nkeystone-specs doesn\u0027t run pep8 [1]. We need to see why Zuul is running pep8 here.\n\n[1] https://github.com/openstack/keystone-specs/blob/master/tox.ini#L3","accounts_in_message":[],"_revision_number":1},{"id":"34fe4b13cad240ab4512b8570b6eb566cc516dec","author":{"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},"date":"2020-01-16 02:22:46.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c0f60913607a90400a13e53b4788365ddbbd074f","revisions":{"c0f60913607a90400a13e53b4788365ddbbd074f":{"kind":"REWORK","_number":1,"created":"2019-10-31 11:34:58.000000000","uploader":{"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},"ref":"refs/changes/45/692345/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone-specs","ref":"refs/changes/45/692345/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/45/692345/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/45/692345/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/45/692345/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone-specs refs/changes/45/692345/1"}}},"commit":{"parents":[{"commit":"b90f7c87021fb3eb9cb093a5af9ef9670e0af347","subject":"Set up for Ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/b90f7c87021fb3eb9cb093a5af9ef9670e0af347"}]}],"author":{"name":"Q.hongtao","email":"qihongtao@inspur.com","date":"2019-10-31 11:33:47.000000000","tz":480},"committer":{"name":"Q.hongtao","email":"qihongtao@inspur.com","date":"2019-10-31 11:34:04.000000000","tz":480},"subject":"Switch to Ussuri jobs","message":"Switch to Ussuri jobs\n\nChange-Id: Ia4b92e1b34344bf79626058264f35a328d55c6c5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/c0f60913607a90400a13e53b4788365ddbbd074f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/c0f60913607a90400a13e53b4788365ddbbd074f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
