)]}'
{"id":"openstack%2Fkeystone-specs~698279","triplet_id":"openstack%2Fkeystone-specs~master~I85579947f31fb74047b33529ad8918787ac7ed34","project":"openstack/keystone-specs","branch":"master","topic":"698279","hashtags":[],"change_id":"I85579947f31fb74047b33529ad8918787ac7ed34","subject":"Alembic Migrations Specification","status":"ABANDONED","created":"2019-12-10 15:46:02.000000000","updated":"2024-11-22 15:17:36.000000000","total_comment_count":100,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"90677f1f1f98dbbf42dad6f126d2311c0c03f8b6","_number":698279,"virtual_id_number":698279,"owner":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-01-20 08:00:57.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":0,"date":"2020-05-27 15:41:13.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":-1,"date":"2020-01-20 07:57:45.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-17 20:05:34.000000000","updated_by":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"reviewer":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"state":"REVIEWER"},{"updated":"2020-01-14 20:27:27.000000000","updated_by":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"reviewer":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"state":"REVIEWER"},{"updated":"2020-01-20 08:00:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-05-27 15:41:13.000000000","updated_by":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"reviewer":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"state":"REVIEWER"}],"messages":[{"id":"fd709182efe5250e9256b305e830ee71c2b303d4","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2019-12-10 15:46:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"02d9d12c61c3174be9e86d8c6ec5ad64f78b0e57","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-10 15:52:17.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27776654e2774dc2a72713e05e69ba2d : FAILURE in 4m 12s","accounts_in_message":[],"_revision_number":1},{"id":"b7c6441092562bcd536271a13a21fb7cb13be10c","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-12-10 18:02:09.000000000","message":"Patch Set 1: Code-Review-1\n\n(11 comments)","accounts_in_message":[],"_revision_number":1},{"id":"be1f9c356e0ba0b0b8558e55a7823079308cce3c","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2019-12-11 11:20:58.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9735fb1da0be167de8d9ca0d64214fa70431efad","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2019-12-11 11:21:24.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"8b63a5ec67c449bd29740136253165bf1f71e009","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-11 11:25:23.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3906536737894685808b432c0fcf244d : FAILURE in 3m 21s","accounts_in_message":[],"_revision_number":2},{"id":"0af8bd00bd46d554818a1c8103dd31aca341f691","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2019-12-11 16:48:27.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"a6ebcab7d5c5e32441d0534dba40da586115e7f5","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2019-12-11 16:55:00.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"e8a93ad6d2f9a26d80fd639892d64ff8d1b9cbcd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-11 17:01:19.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1cebcec0c7b04082acdbd4993b9a20d3 : FAILURE in 4m 45s","accounts_in_message":[],"_revision_number":4},{"id":"241cd854ce4bcf9b3da18ec660107a55acd35d8a","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2019-12-11 17:02:52.000000000","message":"Patch Set 1:\n\n(10 comments)","accounts_in_message":[],"_revision_number":1},{"id":"3d65cb039cb11f6797dbf84b4c344ab95ea5b9c5","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2019-12-12 06:18:50.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"327cd297bc241103e8aba65c385d7d6d1006170f","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2019-12-12 06:20:04.000000000","message":"Uploaded patch set 6: Commit message was updated.","accounts_in_message":[],"_revision_number":6},{"id":"0e6250281f1e1019f5055960e670ac79d54482cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 06:31:02.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4585359563264758a34c8971d57ed67b : SUCCESS in 9m 28s","accounts_in_message":[],"_revision_number":6},{"id":"34da1abb3bc9b587f255b2eab272fa521265fd72","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2019-12-12 07:12:20.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"504f0bcb1935bee7d549fcd317b66d455e9dfbcb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 07:21:24.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2c951f042d141dd83cbaedc25030fab : SUCCESS in 4m 46s","accounts_in_message":[],"_revision_number":7},{"id":"d2c104255d9793e4b747568d86084b525a8a5320","author":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"date":"2019-12-17 19:52:28.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nCommenting on older patchset for a specific comment. Other comments in newer patchset.","accounts_in_message":[],"_revision_number":1},{"id":"041abfda4c15efac6399e2026e4109b7662ec768","author":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"date":"2019-12-17 20:05:34.000000000","message":"Patch Set 7: Code-Review-1\n\n(3 comments)\n\nDid a quick pass at the spec, mostly focusing on the Problem Description section.","accounts_in_message":[],"_revision_number":7},{"id":"6ba059072b42e9cd804bdc0df620b928e05b4d0b","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-12-23 18:51:16.000000000","message":"Patch Set 7: Code-Review-1\n\n(9 comments)\n\nThe old alembic spec is still in the backlog, it needs to be moved/removed as part of this spec proposal.","accounts_in_message":[],"_revision_number":7},{"id":"8716ae40501279972787cb08d017fc1468217d89","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2020-01-11 13:28:51.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"c34c36d5eb0f41539195786049805e93531e6088","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-11 13:37:52.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91a9c1ad19c143dba78e82c8bbef691e : SUCCESS in 4m 19s","accounts_in_message":[],"_revision_number":8},{"id":"19c6489c23d1e639fdfea9be3124ead6db1f176e","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2020-01-11 13:58:06.000000000","message":"Patch Set 7:\n\n(11 comments)","accounts_in_message":[],"_revision_number":7},{"id":"55c16d148fd7ce099f98f308db76f3db09205589","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2020-01-12 18:17:24.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"213a45a06a8a69a6969c946792034e29db01fb7a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-12 18:21:02.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/904dd4356efd47d5bd082f9226cc9562 : FAILURE in 3m 16s","accounts_in_message":[],"_revision_number":9},{"id":"3d819d69a04b4ba497139f34889e0f3d686e54a7","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2020-01-12 18:22:24.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"c2c7a11416051bd7272992f698debe1c3c2bf7fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-12 18:27:19.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e28186b2dee34ea6b5eaa5e40e0abcbe : SUCCESS in 4m 37s","accounts_in_message":[],"_revision_number":10},{"id":"99e8a005611be115337d8b9532cf82cdeeee3000","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2020-01-13 05:45:48.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"1052b6bf2496ef0b5a65515e32b2dbf2cc67ab0b","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2020-01-13 05:48:19.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"2d8cdaa140dc32aca566255bae91e8b26aaa35c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-13 05:54:10.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/54b6e41f6b7d4ccc83dfd624a8cb9b9e : SUCCESS in 5m 00s","accounts_in_message":[],"_revision_number":11},{"id":"03e02568efa589a567ad8760d48a92386add8536","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2020-01-14 20:27:27.000000000","message":"Patch Set 11: Code-Review-1\n\n(18 comments)\n\nLooks ok to me so far, got a few wording nits and a couple questions.","accounts_in_message":[],"_revision_number":11},{"id":"8e0d1390dfa4edd01f23d1bda4d49f43e8c60f70","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2020-01-17 07:19:09.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"242e563040df42d8abf658080c06b94ef31b6f0a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-17 07:24:39.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79a59a583311413b8d161c21a04128a3 : SUCCESS in 5m 09s","accounts_in_message":[],"_revision_number":12},{"id":"c7939e4d4efcc6c3ba7493444ccf2578b4af6273","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2020-01-17 07:28:44.000000000","message":"Patch Set 11:\n\n(15 comments)","accounts_in_message":[],"_revision_number":11},{"id":"e67ff0c53755adac60df4929079a188620e03636","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2020-01-17 10:46:39.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"276d24c6b9812c09ebc96f6567604e151bfe3942","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2020-01-17 21:24:41.000000000","message":"Patch Set 12: Code-Review-1\n\n(10 comments)\n\nAs mentioned previously, the old alembic spec is still in the backlog, it needs to be moved/removed as part of this spec proposal.","accounts_in_message":[],"_revision_number":12},{"id":"806a600693dc37a9ce6057f6f50c99cddf4fb8ae","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2020-01-20 07:31:32.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"1bc1aa2add561cb492aeedcc2b75cf7df16953a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-20 07:38:11.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/89dd17abf7a54850b69ed1414812afd7 : SUCCESS in 6m 07s","accounts_in_message":[],"_revision_number":13},{"id":"865027e2164e7cde3dd578d940ec0e300e698596","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2020-01-20 07:48:52.000000000","message":"Patch Set 12:\n\n(10 comments)","accounts_in_message":[],"_revision_number":12},{"id":"f3edc219402492fc639fd233774e768085fdd362","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2020-01-20 07:54:49.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"fb1e26f73320a5e5c02987e13535d238e20b9f1e","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2020-01-20 07:57:45.000000000","message":"Patch Set 14: Workflow-1\n\nOn hold till we add rolling upgrade CI before merging this.","accounts_in_message":[],"_revision_number":14},{"id":"45a2efbad4125c054cbbfe27b13298c67485c375","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-20 08:00:57.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/774c016654ab4dc78dd5030f298161db : SUCCESS in 5m 03s","accounts_in_message":[],"_revision_number":14},{"id":"0f9f2b526dc8460517a1bddc76a2ebdb1d414389","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2020-03-03 16:35:17.000000000","message":"Patch Set 14: Code-Review-2\n\nProcedural -2, as we\u0027re past spec freeze for ussuri: https://releases.openstack.org/ussuri/schedule.html\n\nWill remove -2 when the new cycle opens.","accounts_in_message":[],"_revision_number":14},{"id":"a1377f5e7749eca8b909a3bba8edaa49a8a3262d","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2020-05-27 15:41:13.000000000","message":"Patch Set 14: -Code-Review","accounts_in_message":[],"_revision_number":14},{"id":"90677f1f1f98dbbf42dad6f126d2311c0c03f8b6","tag":"autogenerated:gerrit:abandon","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2024-11-22 15:17:36.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":14}],"current_revision_number":14,"current_revision":"a6a33aa0d0cf117817680b6bd1947b7860bb4395","revisions":{"6f6b1c5107e34f1f6d52ebecc744dac06d114d3c":{"kind":"REWORK","_number":1,"created":"2019-12-10 15:46:02.000000000","uploader":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"ref":"refs/changes/79/698279/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone-specs","ref":"refs/changes/79/698279/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/79/698279/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/79/698279/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/79/698279/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone-specs refs/changes/79/698279/1"}}},"commit":{"parents":[{"commit":"b90f7c87021fb3eb9cb093a5af9ef9670e0af347","subject":"Set up for Ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/b90f7c87021fb3eb9cb093a5af9ef9670e0af347"}]}],"author":{"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","date":"2019-12-10 15:44:57.000000000","tz":330},"committer":{"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","date":"2019-12-10 15:45:55.000000000","tz":330},"subject":"Alembic Migrations Specification","message":"Alembic Migrations Specification\n\nChange-Id: I85579947f31fb74047b33529ad8918787ac7ed34\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/6f6b1c5107e34f1f6d52ebecc744dac06d114d3c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/6f6b1c5107e34f1f6d52ebecc744dac06d114d3c"}]},"branch":"refs/heads/master"},"a4733a65659a9eec41d778be02fd9d1497a80b57":{"kind":"REWORK","_number":2,"created":"2019-12-11 11:20:58.000000000","uploader":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"ref":"refs/changes/79/698279/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone-specs","ref":"refs/changes/79/698279/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/79/698279/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/79/698279/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/79/698279/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone-specs refs/changes/79/698279/2"}}},"commit":{"parents":[{"commit":"b90f7c87021fb3eb9cb093a5af9ef9670e0af347","subject":"Set up for Ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/b90f7c87021fb3eb9cb093a5af9ef9670e0af347"}]}],"author":{"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","date":"2019-12-10 15:44:57.000000000","tz":330},"committer":{"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","date":"2019-12-11 11:20:39.000000000","tz":330},"subject":"Alembic Migrations Specification","message":"Alembic Migrations Specification\n\nChange-Id: I85579947f31fb74047b33529ad8918787ac7ed34\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/a4733a65659a9eec41d778be02fd9d1497a80b57"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/a4733a65659a9eec41d778be02fd9d1497a80b57"}]},"branch":"refs/heads/master"},"512e2c6f556c1418af962cd273060fe8162bed6b":{"kind":"REWORK","_number":3,"created":"2019-12-11 16:48:27.000000000","uploader":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"ref":"refs/changes/79/698279/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone-specs","ref":"refs/changes/79/698279/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/79/698279/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/79/698279/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/79/698279/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone-specs refs/changes/79/698279/3"}}},"commit":{"parents":[{"commit":"b90f7c87021fb3eb9cb093a5af9ef9670e0af347","subject":"Set up for Ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/b90f7c87021fb3eb9cb093a5af9ef9670e0af347"}]}],"author":{"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","date":"2019-12-10 15:44:57.000000000","tz":330},"committer":{"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","date":"2019-12-11 16:48:04.000000000","tz":330},"subject":"Alembic Migrations Specification","message":"Alembic Migrations Specification\n\nChange-Id: I85579947f31fb74047b33529ad8918787ac7ed34\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/512e2c6f556c1418af962cd273060fe8162bed6b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/512e2c6f556c1418af962cd273060fe8162bed6b"}]},"branch":"refs/heads/master"},"26e9699053b6b1765256a99b8199ae49e1d37043":{"kind":"REWORK","_number":4,"created":"2019-12-11 16:55:00.000000000","uploader":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"ref":"refs/changes/79/698279/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone-specs","ref":"refs/changes/79/698279/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/79/698279/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/79/698279/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/79/698279/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone-specs refs/changes/79/698279/4"}}},"commit":{"parents":[{"commit":"b90f7c87021fb3eb9cb093a5af9ef9670e0af347","subject":"Set up for Ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/b90f7c87021fb3eb9cb093a5af9ef9670e0af347"}]}],"author":{"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","date":"2019-12-10 15:44:57.000000000","tz":330},"committer":{"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","date":"2019-12-11 16:54:37.000000000","tz":330},"subject":"Alembic Migrations Specification","message":"Alembic Migrations Specification\n\nChange-Id: I85579947f31fb74047b33529ad8918787ac7ed34\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/26e9699053b6b1765256a99b8199ae49e1d37043"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/26e9699053b6b1765256a99b8199ae49e1d37043"}]},"branch":"refs/heads/master"},"6d8e6469b7a70ef5eeda1299765bf6a920162b61":{"kind":"REWORK","_number":5,"created":"2019-12-12 06:18:50.000000000","uploader":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"ref":"refs/changes/79/698279/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone-specs","ref":"refs/changes/79/698279/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/79/698279/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/79/698279/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/79/698279/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone-specs refs/changes/79/698279/5"}}},"commit":{"parents":[{"commit":"b90f7c87021fb3eb9cb093a5af9ef9670e0af347","subject":"Set up for Ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/b90f7c87021fb3eb9cb093a5af9ef9670e0af347"}]}],"author":{"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","date":"2019-12-10 15:44:57.000000000","tz":330},"committer":{"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","date":"2019-12-12 06:18:07.000000000","tz":330},"subject":"Alembic Migrations Specification Change-Id: I85579947f31fb74047b33529ad8918787ac7ed34","message":"Alembic Migrations Specification\nChange-Id: I85579947f31fb74047b33529ad8918787ac7ed34\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/6d8e6469b7a70ef5eeda1299765bf6a920162b61"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/6d8e6469b7a70ef5eeda1299765bf6a920162b61"}]},"branch":"refs/heads/master"},"761b333ec84a9a10910fa1d783e2df37129fd626":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2019-12-12 06:20:04.000000000","uploader":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"ref":"refs/changes/79/698279/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone-specs","ref":"refs/changes/79/698279/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/79/698279/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/79/698279/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/79/698279/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone-specs refs/changes/79/698279/6"}}},"commit":{"parents":[{"commit":"b90f7c87021fb3eb9cb093a5af9ef9670e0af347","subject":"Set up for Ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/b90f7c87021fb3eb9cb093a5af9ef9670e0af347"}]}],"author":{"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","date":"2019-12-10 15:44:57.000000000","tz":330},"committer":{"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","date":"2019-12-12 06:19:35.000000000","tz":330},"subject":"Alembic Migrations Specification","message":"Alembic Migrations Specification\n\nChange-Id: I85579947f31fb74047b33529ad8918787ac7ed34\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/761b333ec84a9a10910fa1d783e2df37129fd626"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/761b333ec84a9a10910fa1d783e2df37129fd626"}]},"branch":"refs/heads/master"},"46aaf88ed1478bca0802f7f56f5600a678f6c185":{"kind":"REWORK","_number":7,"created":"2019-12-12 07:12:20.000000000","uploader":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"ref":"refs/changes/79/698279/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone-specs","ref":"refs/changes/79/698279/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/79/698279/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/79/698279/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/79/698279/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone-specs refs/changes/79/698279/7"}}},"commit":{"parents":[{"commit":"b90f7c87021fb3eb9cb093a5af9ef9670e0af347","subject":"Set up for Ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/b90f7c87021fb3eb9cb093a5af9ef9670e0af347"}]}],"author":{"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","date":"2019-12-10 15:44:57.000000000","tz":330},"committer":{"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","date":"2019-12-12 07:12:01.000000000","tz":330},"subject":"Alembic Migrations Specification","message":"Alembic Migrations Specification\n\nChange-Id: I85579947f31fb74047b33529ad8918787ac7ed34\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/46aaf88ed1478bca0802f7f56f5600a678f6c185"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/46aaf88ed1478bca0802f7f56f5600a678f6c185"}]},"branch":"refs/heads/master"},"cb2d031815161128eee5aee23659e908c5ec921c":{"kind":"REWORK","_number":8,"created":"2020-01-11 13:28:51.000000000","uploader":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"ref":"refs/changes/79/698279/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone-specs","ref":"refs/changes/79/698279/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/79/698279/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/79/698279/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/79/698279/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone-specs refs/changes/79/698279/8"}}},"commit":{"parents":[{"commit":"b90f7c87021fb3eb9cb093a5af9ef9670e0af347","subject":"Set up for Ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/b90f7c87021fb3eb9cb093a5af9ef9670e0af347"}]}],"author":{"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","date":"2019-12-10 15:44:57.000000000","tz":330},"committer":{"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","date":"2020-01-11 13:28:28.000000000","tz":330},"subject":"Alembic Migrations Specification","message":"Alembic Migrations Specification\n\nChange-Id: I85579947f31fb74047b33529ad8918787ac7ed34\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/cb2d031815161128eee5aee23659e908c5ec921c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/cb2d031815161128eee5aee23659e908c5ec921c"}]},"branch":"refs/heads/master"},"b991e896198e8a628068291996842c102ce6814a":{"kind":"REWORK","_number":9,"created":"2020-01-12 18:17:24.000000000","uploader":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"ref":"refs/changes/79/698279/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone-specs","ref":"refs/changes/79/698279/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/79/698279/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/79/698279/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/79/698279/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone-specs refs/changes/79/698279/9"}}},"commit":{"parents":[{"commit":"b90f7c87021fb3eb9cb093a5af9ef9670e0af347","subject":"Set up for Ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/b90f7c87021fb3eb9cb093a5af9ef9670e0af347"}]}],"author":{"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","date":"2019-12-10 15:44:57.000000000","tz":330},"committer":{"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","date":"2020-01-12 18:16:54.000000000","tz":330},"subject":"Alembic Migrations Specification","message":"Alembic Migrations Specification\n\nChange-Id: I85579947f31fb74047b33529ad8918787ac7ed34\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/b991e896198e8a628068291996842c102ce6814a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/b991e896198e8a628068291996842c102ce6814a"}]},"branch":"refs/heads/master"},"c2aef905f3bfd57a3e6a080d0536da1b6250c074":{"kind":"REWORK","_number":10,"created":"2020-01-12 18:22:24.000000000","uploader":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"ref":"refs/changes/79/698279/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone-specs","ref":"refs/changes/79/698279/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/79/698279/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/79/698279/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/79/698279/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone-specs refs/changes/79/698279/10"}}},"commit":{"parents":[{"commit":"b90f7c87021fb3eb9cb093a5af9ef9670e0af347","subject":"Set up for Ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/b90f7c87021fb3eb9cb093a5af9ef9670e0af347"}]}],"author":{"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","date":"2019-12-10 15:44:57.000000000","tz":330},"committer":{"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","date":"2020-01-12 18:21:42.000000000","tz":330},"subject":"Alembic Migrations Specification","message":"Alembic Migrations Specification\n\nChange-Id: I85579947f31fb74047b33529ad8918787ac7ed34\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/c2aef905f3bfd57a3e6a080d0536da1b6250c074"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/c2aef905f3bfd57a3e6a080d0536da1b6250c074"}]},"branch":"refs/heads/master"},"4537327d1520faccb23427a4020195381862f689":{"kind":"REWORK","_number":11,"created":"2020-01-13 05:48:19.000000000","uploader":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"ref":"refs/changes/79/698279/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone-specs","ref":"refs/changes/79/698279/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/79/698279/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/79/698279/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/79/698279/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone-specs refs/changes/79/698279/11"}}},"commit":{"parents":[{"commit":"b90f7c87021fb3eb9cb093a5af9ef9670e0af347","subject":"Set up for Ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/b90f7c87021fb3eb9cb093a5af9ef9670e0af347"}]}],"author":{"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","date":"2019-12-10 15:44:57.000000000","tz":330},"committer":{"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","date":"2020-01-13 05:45:18.000000000","tz":330},"subject":"Alembic Migrations Specification","message":"Alembic Migrations Specification\n\nChange-Id: I85579947f31fb74047b33529ad8918787ac7ed34\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/4537327d1520faccb23427a4020195381862f689"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/4537327d1520faccb23427a4020195381862f689"}]},"branch":"refs/heads/master"},"1860f6da5de5aaedf4ac0d5985169cab2bcef193":{"kind":"REWORK","_number":12,"created":"2020-01-17 07:19:09.000000000","uploader":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"ref":"refs/changes/79/698279/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone-specs","ref":"refs/changes/79/698279/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/79/698279/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/79/698279/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/79/698279/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone-specs refs/changes/79/698279/12"}}},"commit":{"parents":[{"commit":"b90f7c87021fb3eb9cb093a5af9ef9670e0af347","subject":"Set up for Ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/b90f7c87021fb3eb9cb093a5af9ef9670e0af347"}]}],"author":{"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","date":"2019-12-10 15:44:57.000000000","tz":330},"committer":{"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","date":"2020-01-14 13:51:53.000000000","tz":330},"subject":"Alembic Migrations Specification","message":"Alembic Migrations Specification\n\nChange-Id: I85579947f31fb74047b33529ad8918787ac7ed34\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/1860f6da5de5aaedf4ac0d5985169cab2bcef193"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/1860f6da5de5aaedf4ac0d5985169cab2bcef193"}]},"branch":"refs/heads/master"},"bd7c4a0ba11a39a03c72f3f9850cdbd50453e7ce":{"kind":"REWORK","_number":13,"created":"2020-01-20 07:31:32.000000000","uploader":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"ref":"refs/changes/79/698279/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone-specs","ref":"refs/changes/79/698279/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/79/698279/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/79/698279/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/79/698279/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone-specs refs/changes/79/698279/13"}}},"commit":{"parents":[{"commit":"b90f7c87021fb3eb9cb093a5af9ef9670e0af347","subject":"Set up for Ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/b90f7c87021fb3eb9cb093a5af9ef9670e0af347"}]}],"author":{"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","date":"2019-12-10 15:44:57.000000000","tz":330},"committer":{"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","date":"2020-01-17 14:04:14.000000000","tz":330},"subject":"Alembic Migrations Specification","message":"Alembic Migrations Specification\n\nChange-Id: I85579947f31fb74047b33529ad8918787ac7ed34\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/bd7c4a0ba11a39a03c72f3f9850cdbd50453e7ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/bd7c4a0ba11a39a03c72f3f9850cdbd50453e7ce"}]},"branch":"refs/heads/master"},"a6a33aa0d0cf117817680b6bd1947b7860bb4395":{"kind":"REWORK","_number":14,"created":"2020-01-20 07:54:49.000000000","uploader":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"ref":"refs/changes/79/698279/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone-specs","ref":"refs/changes/79/698279/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/79/698279/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/79/698279/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/79/698279/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone-specs refs/changes/79/698279/14"}}},"commit":{"parents":[{"commit":"b90f7c87021fb3eb9cb093a5af9ef9670e0af347","subject":"Set up for Ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/b90f7c87021fb3eb9cb093a5af9ef9670e0af347"}]}],"author":{"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","date":"2019-12-10 15:44:57.000000000","tz":330},"committer":{"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","date":"2020-01-17 14:27:34.000000000","tz":330},"subject":"Alembic Migrations Specification","message":"Alembic Migrations Specification\n\nChange-Id: I85579947f31fb74047b33529ad8918787ac7ed34\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/a6a33aa0d0cf117817680b6bd1947b7860bb4395"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/a6a33aa0d0cf117817680b6bd1947b7860bb4395"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Workflow\u003dMIN"],"failing_atoms":["label:Workflow\u003dMAX"],"atom_explanations":{}}}]}
